)]}'
{"id":"openstack%2Fswift~885206","triplet_id":"openstack%2Fswift~master~I94e8cfd154aa058d91255efc87776224a919f572","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-01-18 06:42:09.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-01-18 05:26:07.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I94e8cfd154aa058d91255efc87776224a919f572","subject":"tests: Fix probe test when encryption is enabled","status":"MERGED","created":"2023-06-05 04:07:20.000000000","updated":"2024-01-18 06:43:28.000000000","submitted":"2024-01-18 06:42:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"885206","meta_rev_id":"a3e9d8dd62fa03a640f47703223090c1c3c40b63","_number":885206,"virtual_id_number":885206,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-18 06:42:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-18 05:26:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-18 05:26:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-05 05:07:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-12 15:28:53.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2024-01-17 14:40:51.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"d43e60c86952eda299339938c79da9aa811075f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-05 04:07:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18b7961d03f7a9d548c134aa976eb75ff9d9cbb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 05:07:58.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aac7293b797c43b1a10399bc0bf56475\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94f92595323e434f9e179e03cb488695 : SUCCESS in 8m 37s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/dd8ea6a3a88e40f496a8146cd80e53f7 : SUCCESS in 52m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b68cf94af05a423e921838d7f2dd3cee : SUCCESS in 59m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46de71a67abd47b29e2c2c9375a68f41 : FAILURE in 4m 03s","accounts_in_message":[],"_revision_number":1},{"id":"d17b72347de7ecb82d407e073cc22ee33e947448","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 05:27:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21d1a63027ec4e8ea77aeca06df219a1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c52004cfd4b649fcb593ad6236e16a1a : SUCCESS in 1h 16m 54s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8c148aa422044dc59f03a99355221201 : SUCCESS in 1h 04m 31s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3a39a0d653b849da9d35eccebb2e2e7a : SUCCESS in 18m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/561a74aca4f540a9b8db129c453733d6 : SUCCESS in 16m 52s","accounts_in_message":[],"_revision_number":1},{"id":"a722daa341f0bc8fcddacd3c6d53672f3cb0c209","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-05 06:09:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7b0cbd054b601ecf777709dae1dfa6206ab777db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 06:32:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3614face83e48d899951ff6bb4b5e40\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/be2f9e3ed98d4279bf17745dbe34214e : SUCCESS in 15m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4a2e70bb4d054129a771cc8e709a1ce9 : SUCCESS in 15m 39s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac5a445a051b4022820d252c1dfc5ae5 : SUCCESS in 22m 41s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/369ff2dd3c694e279c11a66e1fee4eca : SUCCESS in 18m 35s","accounts_in_message":[],"_revision_number":2},{"id":"2122f45fbdceae6874e507040fba3e0940396d50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 07:21:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a46560ac94ed46649914ff898ebf6b72\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/800e3fee40c54435b462328ea0ac1aa7 : SUCCESS in 7m 41s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3e9bc7d8398346a7a2d6e8978ebd0462 : SUCCESS in 56m 40s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/17eb761c0668499ba9505244263ea284 : SUCCESS in 58m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77d2870275b24c01b8b647b0bfc29500 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":2},{"id":"c70d286a5ba426b5e1cdb9380dc2337cdb47bde6","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-06-12 15:28:53.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6448c9500c16593067bbbe7331f617f290d847eb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-13 00:06:00.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"74fc72c3192f8b0f4e63776666af2195983c90c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-13 00:06:28.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"872d06c350c0f847f12bcb91c125e159cbd6ef80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 01:13:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f53d3badf369470c97a6724885634d00\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a563fb208f3b49b98563e13097706a8a : SUCCESS in 10m 11s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b40f370cb1db40fe9f23ebe4f717494f : SUCCESS in 52m 01s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4630981b64f1453f86f639021c149868 : SUCCESS in 56m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a708b7af8ed45539a5bb67d99bd7977 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":3},{"id":"9cdf821d93abbe4ea37a9010b211da8686af8c38","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 01:24:17.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6378074c7d954a40b53ad75f0c0f7fbb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d11ed8dc73074a4fae84c291a54e7573 : NODE_FAILURE Node request 300-0021447735 failed in 0s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fca723db1aeb4e699394508a664d90d4 : NODE_FAILURE Node request 300-0021447736 failed in 0s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/bc70a1a216a14b83951fb5f8cb1989c0 : NODE_FAILURE Node request 300-0021447737 failed in 0s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d2da677f3f11459fbbed08fab46382ad : SUCCESS in 1h 15m 07s","accounts_in_message":[],"_revision_number":3},{"id":"5d6e060dcaa9aa5872efa3960c27c00c7425f081","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-01-17 14:40:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a9ea3a2589dc53c23be70a6e60ae97d72cf2ede4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 14:41:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cf2eb03a1a8eb84774af25d2a3ccdcf5fcdc8036","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 16:24:48.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/35082fdc35c144b280ebf6f9871bc14f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c3166a53eea4c85a40d290737966ead : SUCCESS in 9m 37s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5afd2b82bc244bf4bf62332d01a13d27 : SUCCESS in 1h 04m 05s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4beb4dc1c53b4f27a645729bf478afdf : SUCCESS in 26m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3750910fa3ad486d8d9a0b7729bd84d4 : SUCCESS in 18m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/eda1f223e166461ea0c39c7b3908310f : FAILURE in 17m 46s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/eec374ec12f449c09d38dba632e057b5 : FAILURE in 17m 28s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c57e0fa669a74debb342a06b8d3e8761 : FAILURE in 21m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6a1ed6117d2f47a7b0129b55b691fcc8 : SUCCESS in 21m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/83beef9755854c6ea272f69fd046cee4 : SUCCESS in 18m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0a411d64d31b4b369c06545d769ac66d : SUCCESS in 19m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0cbce71204064421b3af15cd3da0482d : SUCCESS in 20m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4927681643024e06ad1c3e92ec355d45 : SUCCESS in 16m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f848a35a59884586bdec66faa9051290 : SUCCESS in 18m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5694f0c42c5e41a1aacb762af8c386c0 : SUCCESS in 16m 41s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d907305c4e074d99933843962f83f079 : SUCCESS in 9m 05s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b41acdabf8664174bd6e120d549d3156 : SUCCESS in 1h 01m 36s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c67d91ee83af4e50a0228e3c7e33d324 : SUCCESS in 1h 09m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/befd906d4cf9440d8181a960f3351104 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":3},{"id":"b7c0d79dee79923ab9caa58c4c5a017ab7967435","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-17 16:54:33.000000000","message":"Patch Set 3: Code-Review-1 Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f0b5e1cd0e9f965af2d7c8a5442bcf6996606fc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-17 18:19:26.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1, Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"496c6fcf6ee6541413b8ef4825c74ce8628f15e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 18:42:58.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a2c641091f6419b9ded39caa8a3213f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/506a804f48f04e9797e335092e4be634 : FAILURE in 19m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/22615aed605841c89c9827dd3ec1d4a6 : FAILURE in 18m 29s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/af1d162359c64063bc990369a4a7ec51 : SUCCESS in 22m 27s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/44e6f477c1924d02a5af6cafaca2e775 : SUCCESS in 20m 11s","accounts_in_message":[],"_revision_number":4},{"id":"63e229f3d168292f83dc050e4db8b732c0f84635","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 19:47:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b2e463e91ec49ef8d5e3bb7e8861cda\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09cef4d7f4b44f34a4f1013ce9f77986 : SUCCESS in 9m 23s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/be1fbbaba2dd43e594159a260869c6f8 : SUCCESS in 1h 01m 38s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/282906b407af4171b5e59ebd809c4ade : SUCCESS in 1h 01m 23s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c241aaa9885241d594214dd02b8db64b : SUCCESS in 1h 09m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b0fccb882ff457ba57488859748a5bf : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":4},{"id":"06d3f7bbca816336031298eed2f58cfbdfba23ff","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-18 05:26:07.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"b369395d0c28b4509a432172eae54973a933e054","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 05:26:37.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"88c4f304a4a7ee0b7096cdaa40443ac1bf10e9c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 06:42:09.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40c48fff8bc54428ba9fddbc4d29410b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0150a4d64da049f1994c8f6d6a8117df : SUCCESS in 8m 39s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e366d9bc89f44b41b9da2dc6daae0aaa : SUCCESS in 1h 02m 59s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5ea1c48ea14842a2bc2d75417f5498ac : SUCCESS in 19m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/824347f3d2e44eea905387f4dabbd040 : SUCCESS in 16m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bb563f47f8d44dc698e94f847fd9f0cb : SUCCESS in 16m 56s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6dffd3b2db964be8b92ae01017a4ecab : SUCCESS in 16m 37s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ecc119dd4bed4919bfc742c028a03987 : SUCCESS in 17m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9b73cade4c6e42fdb0e350fbf6a4fcf1 : SUCCESS in 22m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b4770739610a4ee9831914b33baa864a : SUCCESS in 19m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/af79e88f8ec74d7c8bd7a0b5d301a163 : SUCCESS in 19m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/049112faba664f07a2138b771055ee34 : SUCCESS in 20m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/af0b557731a64776882a40edc6c0948f : SUCCESS in 15m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b65052225df04b96b10f93a658cede59 : SUCCESS in 15m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b642b813386e4631bef82941812973d9 : SUCCESS in 7m 18s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f44d549f90544e30b6b30e4eae06c366 : SUCCESS in 7m 15s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c45d5b3b2ff34e1e8da015217e189154 : SUCCESS in 54m 36s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ce50391e051a4991be3413137018adcc : SUCCESS in 1h 04m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd2a2b9da5f54cbf98f185b8beaf3676 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":4},{"id":"5f8db1a8adaed2a399d625dadb32b0bb507edbdc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 06:42:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"a3e9d8dd62fa03a640f47703223090c1c3c40b63","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 06:43:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e38917bc36944a49bd3116aab238696\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35fafc8a66a04bf5b2f091153665171c : SUCCESS in 1m 00s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/2beda4d436d94e1e90472d2cb4ac507c : FAILURE in 26s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/8ea6138ff82144ffac3bd9b35785373b : FAILURE in 31s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7e3925aa9c004a0add738f2ed14c466c92a3f88d","revisions":{"68676aac7d29238184534d7c6e82c81594b79223":{"kind":"REWORK","_number":1,"created":"2023-06-05 04:07:20.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/06/885206/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/885206/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/885206/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/885206/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/885206/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/885206/1"}}},"commit":{"parents":[{"commit":"eb3fe260d51d021674b2bccc4b9d1b4cc0757450","subject":"Merge \"memcached: add method and time_spent to exception logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb3fe260d51d021674b2bccc4b9d1b4cc0757450"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-05 04:06:13.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-05 04:06:16.000000000","tz":-420},"subject":"tests: Fix probe test when encryption is enabled","message":"tests: Fix probe test when encryption is enabled\n\nChange-Id: I94e8cfd154aa058d91255efc87776224a919f572\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/68676aac7d29238184534d7c6e82c81594b79223"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/68676aac7d29238184534d7c6e82c81594b79223"}]},"branch":"refs/heads/master"},"75495b9f35d8c6232d0ccb8b268cd7754b6210ce":{"kind":"REWORK","_number":2,"created":"2023-06-05 06:09:21.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/06/885206/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/885206/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/885206/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/885206/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/885206/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/885206/2"}}},"commit":{"parents":[{"commit":"eb3fe260d51d021674b2bccc4b9d1b4cc0757450","subject":"Merge \"memcached: add method and time_spent to exception logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb3fe260d51d021674b2bccc4b9d1b4cc0757450"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-05 04:06:13.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-05 06:08:48.000000000","tz":-420},"subject":"tests: Fix probe test when encryption is enabled","message":"tests: Fix probe test when encryption is enabled\n\nChange-Id: I94e8cfd154aa058d91255efc87776224a919f572\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/75495b9f35d8c6232d0ccb8b268cd7754b6210ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/75495b9f35d8c6232d0ccb8b268cd7754b6210ce"}]},"branch":"refs/heads/master"},"fef86c5de14cf0dfb01c7e4ed5ba15f655c56f2d":{"kind":"REWORK","_number":3,"created":"2023-06-13 00:06:28.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/06/885206/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/885206/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/885206/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/885206/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/885206/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/885206/3"}}},"commit":{"parents":[{"commit":"eb3fe260d51d021674b2bccc4b9d1b4cc0757450","subject":"Merge \"memcached: add method and time_spent to exception logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb3fe260d51d021674b2bccc4b9d1b4cc0757450"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-05 04:06:13.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-13 00:06:06.000000000","tz":-420},"subject":"tests: Fix probe test when encryption is enabled","message":"tests: Fix probe test when encryption is enabled\n\nChange-Id: I94e8cfd154aa058d91255efc87776224a919f572\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fef86c5de14cf0dfb01c7e4ed5ba15f655c56f2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fef86c5de14cf0dfb01c7e4ed5ba15f655c56f2d"}]},"branch":"refs/heads/master"},"7e3925aa9c004a0add738f2ed14c466c92a3f88d":{"kind":"REWORK","_number":4,"created":"2024-01-17 18:19:26.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/06/885206/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/885206/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/885206/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/885206/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/885206/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/885206/4"}}},"commit":{"parents":[{"commit":"2331c9abf2c4472d7274874cf9edc37cc764fa81","subject":"Merge \"tests: Switch get_v4_amz_date_header to take timedeltas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2331c9abf2c4472d7274874cf9edc37cc764fa81"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-05 04:06:13.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-17 18:19:08.000000000","tz":-480},"subject":"tests: Fix probe test when encryption is enabled","message":"tests: Fix probe test when encryption is enabled\n\nChange-Id: I94e8cfd154aa058d91255efc87776224a919f572\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e3925aa9c004a0add738f2ed14c466c92a3f88d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e3925aa9c004a0add738f2ed14c466c92a3f88d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
