)]}'
{"id":"openstack%2Fswift~885209","triplet_id":"openstack%2Fswift~master~Idf09ab11dd732e37b13c4b1ab7906a86d3752d1a","project":"openstack/swift","branch":"master","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-06-12 19:14:19.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-06-12 19:14:19.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Idf09ab11dd732e37b13c4b1ab7906a86d3752d1a","subject":"replicator: Try harder to continue syncing to other nodes","status":"NEW","created":"2023-06-05 06:05:11.000000000","updated":"2023-06-12 20:11:15.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"be9c92d8201a708beafd86acafe58fbfc9bc585a","_number":885209,"virtual_id_number":885209,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-06-12 20:11:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-05 06:26:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-05 07:28:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-12 15:23:08.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"73f4ca5bb6d4f2c0bca5dda8ddfa7084436d7eb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-05 06:05:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8d63c39a08009a62d9680746a2ba3385dcf5052","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 06:26:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a64b7f6fb9cc48099053922ae86386f0\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec3409e61ef74229bfd8e44dbf29d1c6 : SUCCESS in 15m 43s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7656b69d64d0479ea1789ba18da029d1 : SUCCESS in 16m 08s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/38a204e4b6914871ba28a8e6e6059f7a : SUCCESS in 18m 25s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a9676d55fe904e17a73a642fb30bed80 : SUCCESS in 19m 51s","accounts_in_message":[],"_revision_number":1},{"id":"341dd1fdecf2f2b4787ec4adbe073ab98c9c6a99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 07:28:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5be482db980c4c0e8e9bc1e816724483\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/488cd433f0f7435582f23ff582aefe01 : SUCCESS in 8m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/4ce7671dd3ff4304aaa94a8e5d29b25e : SUCCESS in 1h 07m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/440c5cccb0bc4e7a95d870e7dea4db66 : SUCCESS in 35m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d4fe59d86cf64b01812b6c91b299b9a1 : SUCCESS in 51m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8af0b0fe6bd145c19b7e38323754bdad : SUCCESS in 23m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aea8ca5bc1b84f5988fb1aac44520cd4 : SUCCESS in 15m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c29a79e26f85441d82d66241a675ef69 : SUCCESS in 22m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ea5f8588bbb84805b17787112bf949ac : SUCCESS in 8m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ebe737c5cbb24ccfac497e380d0d6e79 : SUCCESS in 5m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1dbd4187346a47eb940740572fa4fcfa : SUCCESS in 17m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/59505893f6e447658fef88ac11f11df4 : SUCCESS in 21m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/24f26b301aa040e09c5febffa7e5e92c : SUCCESS in 15m 22s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/56339f45f4cb40ad86f0c68b434f7fc8 : SUCCESS in 13m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3acceccaee504e7693a175ef8e9a2059 : SUCCESS in 17m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/771fcfbf294b4953ac67f2cf355f3663 : SUCCESS in 16m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b453c94bc84e4495b5183225c3ef52b9 : SUCCESS in 24m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8558e77357c54f528e610cb006949801 : SUCCESS in 11m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a18b0f72c0cb47d790465966b4bca0aa : SUCCESS in 13m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6b627bf601d74f4fb2f19277156cd328 : SUCCESS in 24m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cbd6a6bdceea47369631bd0ad840fdb7 : SUCCESS in 6m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/21f9acfd10ce4d328d452d3464d8b73c : SUCCESS in 24m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f1ad055037de4d26bc7f2271f63f06d4 : SUCCESS in 8m 00s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/115cca33000546a095bd5e570a73ed20 : SUCCESS in 59m 25s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4138510c69a347789bd4fa60cb5fd068 : SUCCESS in 1h 06m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/488ce59e29eb4b0f9bf430b4df89fd6a : SUCCESS in 35m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2b3499995c194ed5b9ee953921593810 : SUCCESS in 46m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8eee0e3f609d480e9d1491ed7104a293 : SUCCESS in 13m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c9fd90043da44028bf1711575c3ad1a : SUCCESS in 4m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/235e12bc2a1743319ea7cbb3f86466b3 : SUCCESS in 20m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/130f6305464446a8bb40222220f2a25a : SUCCESS in 54m 29s","accounts_in_message":[],"_revision_number":1},{"id":"e5b17829229b7852ccee18f8f7a74189ea0d3b17","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-06-12 15:23:08.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a977ba0df882e27247ed7ab27422d95ceeeaa393","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-12 19:14:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a113a2c3dd660c5bf022ec0e973587b9279f8290","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-12 19:14:24.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"07cb81572ce03eddb8d8cbd78c13d8114838754e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-12 19:35:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e300a72910ea47258972c0640c1cce74\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d724b10af5804d65a14edccf4bd718ff : SUCCESS in 17m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ba768be1890441a2831e8da008564bdd : SUCCESS in 15m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6917fdb8d1754d40a2ea58b9c4cf5537 : SUCCESS in 19m 48s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e796a5d676f42608a60502ab12b3c8b : SUCCESS in 17m 41s","accounts_in_message":[],"_revision_number":2},{"id":"be9c92d8201a708beafd86acafe58fbfc9bc585a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-12 20:11:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed3ec5252ebb44b2b773f11548221f75\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee4e8e8692f94695b0fe81f415add171 : SUCCESS in 8m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/c7267acde731404a9a327cedf524df28 : SUCCESS in 51m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/09d6cf7903024f059da8be2dfa2e4205 : SUCCESS in 36m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec56c5e188ac40c4a540f7a51bfd516f : SUCCESS in 45m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7b7b95bb7eeb4b968e166180dbc01132 : SUCCESS in 25m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5635637cc09742beb7813c6a6eee917d : SUCCESS in 22m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f2810ac160f49878ec1c6b470fff5a4 : SUCCESS in 18m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d28aea46bb28444887758d4d1eae31b2 : SUCCESS in 9m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/21940803d7614384b2cf5993a0cd9a3d : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/961779096f0344f3aec52bf55870c2c9 : SUCCESS in 21m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c74670d8309e46eb94741a2a6b370808 : SUCCESS in 22m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8ed4f94bb9454443b2d3f73531c931c4 : SUCCESS in 20m 46s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a816532c566543b8b697fff787b4c21a : SUCCESS in 22m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7b98a60d52704193973386656a58b898 : SUCCESS in 21m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c1d8dc7e383c4469b3b6a80660e76f34 : SUCCESS in 22m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aee72ce5e13842389a94454963b10a92 : SUCCESS in 21m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/75ca1e8fdb364806bf2e802ab74d8a62 : SUCCESS in 22m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/39deecbdf28f487796cc5a5cc3e682d4 : SUCCESS in 19m 42s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a2827866be6342c594920351c1a4207b : SUCCESS in 15m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/78acf132ab38471f9e310dd3cd5692d6 : SUCCESS in 7m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cc9d0f07fc77424fbc8efcf4e9c6a1c7 : SUCCESS in 16m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1b7143c3c5f9485ab0d567d2488b0e6b : SUCCESS in 8m 05s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/16340359602c4bad88f870e6bef9d73c : SUCCESS in 39m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8b3b0932bcb445e087c64c3cb232121c : SUCCESS in 51m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5f2b4652a295437b8e8adb1434751c07 : SUCCESS in 38m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ce662fc192c24274851831fd4287494d : SUCCESS in 33m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b456d99607c487ead848b1bc947b893 : SUCCESS in 16m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d357812e2d84110924cfb859a5f4455 : SUCCESS in 4m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7ff06c074270400d938808ecc5a9c500 : SUCCESS in 16m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0bd308ce71e44aebbf1660189a009706 : SUCCESS in 44m 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f119661010708224333950e12bc72e9dc935f858","revisions":{"2655987e8c12733b85cdbcf98a63e27972e50003":{"kind":"REWORK","_number":1,"created":"2023-06-05 06:05:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/09/885209/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/885209/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/885209/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/885209/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/885209/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/885209/1"}}},"commit":{"parents":[{"commit":"eb3fe260d51d021674b2bccc4b9d1b4cc0757450","subject":"Merge \"memcached: add method and time_spent to exception logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb3fe260d51d021674b2bccc4b9d1b4cc0757450"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-05 05:49:56.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-05 05:49:56.000000000","tz":-420},"subject":"replicator: Try harder to continue syncing to other nodes","message":"replicator: Try harder to continue syncing to other nodes\n\nThe post-rsync REPLICATE call doesn\u0027t have to succeed right away; if\nthere\u0027s a timeout or connection error or something, the hashes won\u0027t\nmatch on the next pass, either, so we\u0027ll get another opportunity.\n\nMeanwhile, having such an error bubble up out of sync() means that we\nmay exit the node loop early and not check all available primaries.\nThis can cause some flakiness in probe tests.\n\nChange-Id: Idf09ab11dd732e37b13c4b1ab7906a86d3752d1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2655987e8c12733b85cdbcf98a63e27972e50003"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2655987e8c12733b85cdbcf98a63e27972e50003"}]},"branch":"refs/heads/master"},"f119661010708224333950e12bc72e9dc935f858":{"kind":"REWORK","_number":2,"created":"2023-06-12 19:14:24.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/09/885209/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/885209/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/885209/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/885209/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/885209/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/885209/2"}}},"commit":{"parents":[{"commit":"eb3fe260d51d021674b2bccc4b9d1b4cc0757450","subject":"Merge \"memcached: add method and time_spent to exception logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb3fe260d51d021674b2bccc4b9d1b4cc0757450"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-05 05:49:56.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-12 19:13:45.000000000","tz":-420},"subject":"replicator: Try harder to continue syncing to other nodes","message":"replicator: Try harder to continue syncing to other nodes\n\nThe post-rsync REPLICATE call doesn\u0027t have to succeed right away; if\nthere\u0027s a timeout or connection error or something, the hashes won\u0027t\nmatch on the next pass, either, so we\u0027ll get another opportunity.\n\nMeanwhile, having such an error bubble up out of sync() means that we\nmay exit the node loop early and not check all available primaries.\nThis can cause some flakiness in probe tests.\n\nChange-Id: Idf09ab11dd732e37b13c4b1ab7906a86d3752d1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f119661010708224333950e12bc72e9dc935f858"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f119661010708224333950e12bc72e9dc935f858"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
