)]}'
{"id":"openstack%2Fswift~885319","triplet_id":"openstack%2Fswift~master~Ifec8d3a40f00141a73f6e50efe0b53b382ab2ef3","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-08-18 14:52:43.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-08-18 16:09:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifec8d3a40f00141a73f6e50efe0b53b382ab2ef3","subject":"proxy: Get rid of iter_nodes helper","status":"MERGED","created":"2023-06-05 22:50:52.000000000","updated":"2023-08-18 16:10:38.000000000","submitted":"2023-08-18 16:09:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"885319","meta_rev_id":"112bd61fcdcabc07165e7fa14a320d3eb0e30ec2","_number":885319,"virtual_id_number":885319,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-18 16:09:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2023-08-18 14:52:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2023-08-18 14:52:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-05 23:56:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-06 00:03:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-24 14:39:45.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2023-07-24 16:42:48.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"6b8384ae69b5010341b151e06d7e19b476dcbe56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-05 22:50:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14ae999440bcaa5a4a70a43690863137cddbee52","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 23:56:26.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a01ba6b1234b4bd8a5225694344dc725\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b0da89479b27494c80dbae7271dc9184 : FAILURE in 17m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/645a2cf5bf064af8a39e0a9b0ba8525e : FAILURE in 1h 01m 41s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3752dd07878245ab90903059ca63ef1e : SUCCESS in 22m 17s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/155d4d443cdc47799c222dbef03c298a : SUCCESS in 1h 03m 18s","accounts_in_message":[],"_revision_number":1},{"id":"c466c5098098e5f3775b6c9e1594bace833f583a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 00:03:35.000000000","message":"Patch Set 1: Verified-1\n\n(17 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c7c2092b0da045849cce59a2751b30e2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a2d52c4e8584509882e23751b15f713 : SUCCESS in 7m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/d1c3a4db3ddd4b038c00ecbcd80cb69d : SUCCESS in 38m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a6d056cf6cc043c4bac531cce62a59dc : SUCCESS in 34m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f14275c0ffc145d0adc1d7a460eac36b : SUCCESS in 51m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/85006c17ab2f40cf851f1218ed18f4d9 : SUCCESS in 18m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5294a7ad045341cd97bf940b98f99399 : SUCCESS in 23m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dfa4f7c2f5c142aaa7c2ec8e26577ff4 : SUCCESS in 19m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/595e275f4d3b48b28ed86727942e619b : SUCCESS in 9m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/982e36e9651d45d99da3571ce7422798 : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1c0833e8a99048aabb7b6963767170d4 : FAILURE in 23m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/92970e6c632c41c0a9c0d4ab2aecbed3 : FAILURE in 12m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d580dc0ce553440fb13bc6817d9a8024 : FAILURE in 21m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4df3dff792824310841c2bfa01b250ef : FAILURE in 15m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/71846e827b764eb69c3616cad4f43202 : SUCCESS in 17m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/006dee8754c34dbbaab4d8c549c1caae : SUCCESS in 16m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c93a025918c342df93f54ef753cc406c : SUCCESS in 17m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1f1aa80b5f254aa484a9a65bb6d8e434 : SUCCESS in 22m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a56aa2da0c22458391e6efc45fc936c3 : SUCCESS in 12m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1e87ef79e0ac4bcb991e751e16ada10e : SUCCESS in 15m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e7a788d1ad364a868f9b6b807ab126b8 : SUCCESS in 7m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3522f63fe0d84b219af62e2b74bd4b1d : SUCCESS in 16m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b44dd16d878e4a3faf8897264ec9b9bc : SUCCESS in 8m 10s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/628931d26e87476083a2e6b2ac19cad4 : SUCCESS in 59m 45s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/aff115508ad847a9aa5009b4cfa2b99c : SUCCESS in 56m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a4b7df4b4d3d423ea474a55a9f78b618 : SUCCESS in 36m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fb825aa1c1c14504acafeb4c7ad05cd7 : SUCCESS in 35m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f575d69c801432c8ae06a3d7d908103 : FAILURE in 18m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c4227ba71104161a402c7a7a8eb485e : FAILURE in 4m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/14ea4e8e27864e5598d173f2580968e3 : SUCCESS in 26m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/85c7d68db62442a59dad8b54189b8278 : SUCCESS in 1h 07m 03s\n\nWarning:\n  Comments left for invalid file swift/common/wsgi.py","accounts_in_message":[],"_revision_number":1},{"id":"c31abffa2844f7944e94bc00eaf0523ee5172fa4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-06 04:12:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"cde23a2725ab6e2f8e0772404a24aedc21ee1231","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 04:41:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10c40de423564d0481c72570be346ddb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9557159ed368434cb5e53bc11fa626e0 : SUCCESS in 17m 17s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/836252dabc6448a58f36047c95459a76 : SUCCESS in 18m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/55f1dd7705c94a5cbc3dea27dd3ff95c : SUCCESS in 22m 16s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a43c58d7eeab46878cc4c25b820cbf44 : SUCCESS in 22m 57s","accounts_in_message":[],"_revision_number":2},{"id":"08223af9133f153f8cf1a759657a598f3994aa82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 05:24:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e136764649024a6f91bc08e85a6ab40f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e73ebc5d1cc64c4eb29c1a5129fecfac : SUCCESS in 10m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/f99816abdc844754b86bc662441fd0ff : SUCCESS in 46m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ae888d1fa04e42b98b9ed1090d4f4126 : SUCCESS in 1h 01m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/779e90040a964a4c9ec6d58472c77b60 : SUCCESS in 51m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/631e543797604f268efbf84358d2f63d : SUCCESS in 18m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2895ab9eb2784d52b3198d1700161bc7 : SUCCESS in 20m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0c1995b7e1c945e5a475a5e3753fb4b8 : SUCCESS in 25m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7b6bc8fe2e4c457982a919cd247a521b : SUCCESS in 9m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/274cbfcedda84f0294b38e17031e5ef0 : SUCCESS in 6m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d3fa13a45dfc4c66a785f09d7122abfe : SUCCESS in 23m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/afd73c0218764f7b8324924ef79e9cba : SUCCESS in 21m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/26c4b6a00c714fd2999dccc443bdb4bb : SUCCESS in 14m 27s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8c1ba46963b84ca9806368a509c011e8 : SUCCESS in 22m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/66676de6344c407ba391102840c7aab3 : SUCCESS in 15m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ed28a37d46334c62bf96f6461dea3dfd : SUCCESS in 15m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fae11d232fc54f85ac0262c1e2b6189a : SUCCESS in 15m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c6b9d5f2dba74173b9e2743cf74e217a : SUCCESS in 11m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/47b99dc2420e4d2db9d261649656403c : SUCCESS in 13m 16s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a464b1adb661467db767b4ab092e787d : SUCCESS in 12m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/662b70a9a10648c7bce983eac84ed97f : SUCCESS in 7m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6d3cf731f51f408b93767e9a05ca691e : SUCCESS in 17m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e84e78bb94224973ba3696e3301225bc : SUCCESS in 7m 26s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3aa752dd22f5402e8725cd19f23ff358 : SUCCESS in 52m 54s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/74cc489553d24916affb8163da4496d4 : SUCCESS in 1h 10m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8bee70623e2f432e8fab81ff74cf1600 : SUCCESS in 45m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/873c48c676e5440798b0f58fc0df3ad1 : SUCCESS in 47m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a947acae60d4d95aa46b692e6843686 : SUCCESS in 16m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e35b6dd0ab7e41e38e4f5a7bea660640 : SUCCESS in 5m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7e80d24af1c14fc3bf13d24d0a446603 : SUCCESS in 21m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d2f85fcaca7408a926192071a0e567e : FAILURE in 45m 25s","accounts_in_message":[],"_revision_number":2},{"id":"d22aafb57df1612379849693bd1e11ee842ecf93","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-24 14:39:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aa28696b1acc9f5130f8060ab015a1ef47d4aaa6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-24 15:43:51.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"48064b15dec487c5816a686f70cad2739183d8ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-24 15:58:51.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"cfeb71103dfac126a38fd6a62b70785e062cf337","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-24 16:42:48.000000000","message":"Patch Set 4: Code-Review+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"726fa8aa10b4e5af5a643dad1af882bc0e9f65d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 16:57:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2a8b514e9f24ecd8879ac3c177220bf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b12549905b0e4e499dff1cb6e9ffd44e : SUCCESS in 7m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/21fb842e82d24711a70f625a9ea30c1c : SUCCESS in 50m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/578be7e392f34cbf84a1003dcc10e63b : SUCCESS in 50m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5284e72448014c7cafeddc9a9f16f327 : SUCCESS in 44m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9221aa6203454c329728ac6e1ad513ed : SUCCESS in 19m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cce6d8367e1240e9902b7cdf6481eca2 : SUCCESS in 17m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c290338fb6074219ab26c223c46bab09 : SUCCESS in 20m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e7bb776390974561a7183e626bf3169d : SUCCESS in 9m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3ff8cfebf4f7428caa3400332daa8127 : SUCCESS in 5m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d88f6dc0263c4df7b42d9954c0f05d56 : SUCCESS in 16m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/293be76290cc478e86db7d109a089c4e : SUCCESS in 16m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/08689f3ba4ac438c833616ceff5be846 : SUCCESS in 16m 31s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a82c520440904b219201157ae64903cc : SUCCESS in 16m 58s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1aafc6a2082d4393b78af3450227aed3 : SUCCESS in 23m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/94cff338ca774bd580b81aab0a37451b : SUCCESS in 16m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cdb563f1688549318baf74c20a053d36 : SUCCESS in 16m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/76678f08185444cb9374d359f66e648d : SUCCESS in 15m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6316f1c7ae404d8bbaa87934401e9d1b : SUCCESS in 14m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7c45e3e854634b84ae918181685d2c20 : SUCCESS in 17m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/359fcd35d36a427fb1a40ea95f506eeb : SUCCESS in 16m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5252c0aef80e40a5b05718817f0b8266 : SUCCESS in 6m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1be2b0c0786d4415a7f1b5ee473c21ac : SUCCESS in 18m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e6d7646c89bb41f887e52eb49cabdb6b : SUCCESS in 6m 51s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/838e1ecf44e84c3ab30e83443d6c8543 : SUCCESS in 56m 07s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ab5c5975b92741ecb3450a259e918f9d : SUCCESS in 55m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/314f148f0a8e4137bc2f5f3d71d0012c : SUCCESS in 45m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/15629e8d0e2d4c9fbe9b038bbc1b8e5e : SUCCESS in 36m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d14e3372fb145a4a73ab27465327171 : SUCCESS in 17m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66fbf1e48c3d43fca33dea426ac448b6 : SUCCESS in 5m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d7e908dccaa04ffb9c08de27f65f01de : SUCCESS in 20m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5efd806a8d1c40c79f1a0a015ce9345e : SUCCESS in 54m 04s","accounts_in_message":[],"_revision_number":4},{"id":"a29c9140a5a120fc121a8efb287a9e5cf8541bf2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 17:24:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db30b2ecc0cf4234b952d1aceb53bc7d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c6c41563b8af4487821d1e99afa6f586 : SUCCESS in 17m 38s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/acba856c0a6c495787a4c9d89188f8e9 : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ea58e7f346b4b91ad0999673203ca40 : SUCCESS in 23m 22s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f90586be52d44ded97b6933a108bb808 : SUCCESS in 1h 23m 29s","accounts_in_message":[],"_revision_number":4},{"id":"104d82c5a4d697dd6a88e932456c7761e214816a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-16 19:09:02.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"c76e2af35c0ad6e4f2668acad522356a7be4c6b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 20:07:52.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02c18de44ae648b69062502262a07cd9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4e33f1498d74e3cae33880fd940b66a : SUCCESS in 10m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/c5f7e4ca8173400c9d61d74e31e9ad31 : SUCCESS in 45m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b29a27cf2787443aacb20edf45cce273 : SUCCESS in 35m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/24c9bb3a23ae4f06ae3e4b7663145cf5 : SUCCESS in 47m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4695a29cd0be43f8a32273f933d2fe66 : SUCCESS in 20m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8b6d8f60fd7642908e6b6696b2b61f2a : SUCCESS in 19m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ba91528ac69742339320bae5c8cde9fc : SUCCESS in 23m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ed7dd0fbb89d4005a29a15e2c81d8d60 : SUCCESS in 8m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d9bbec37b34c4c61a6363fbfdb42ad56 : SUCCESS in 6m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2b4621252ce54207b1aeffbef09dac34 : SUCCESS in 20m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5fb7b8855bd749908a43d2187038bf3e : SUCCESS in 13m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8ab1e8e6a52341be93f13b8aeeab2b22 : SUCCESS in 19m 08s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1f2ae91c909b4fb1bac8f2f392f19533 : SUCCESS in 17m 32s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8beaec2762d043a1b9694df1037c9a4a : SUCCESS in 16m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/71c64d4930454a95b057ac150b884c4e : SUCCESS in 16m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3d512fe2327a48e9b8bf8e474208e12e : SUCCESS in 17m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/88abbb00c29a4d4c8d88fe590fcfbd64 : SUCCESS in 27m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b29fdcc617fb428ca6fc7c4ffa2ce943 : SUCCESS in 20m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/92ef044f4e934b7eb838303e1ce715c0 : SUCCESS in 16m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0a26923e13ad4d0db4572c5c2948a6a0 : SUCCESS in 15m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/eecccc0ae3f0488ea0e46aa593206921 : SUCCESS in 6m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f73df3145fd34d06923671d318dd34a9 : SUCCESS in 16m 35s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/da9a5b965b9b495fb0f019d7c9230faa : SUCCESS in 6m 50s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cc73b60ddd904fbda32986308ffbbd2f : SUCCESS in 53m 59s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8fccb1481efc4ec59f83a7866089bd1c : SUCCESS in 48m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f5100e210e4a448ea69d2e23ef5e37ea : SUCCESS in 47m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6cd2d1d3cbe74ea89666a8cb514675ba : SUCCESS in 53m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a2aa5d7044046e783617163c14be601 : SUCCESS in 18m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69db927988c842969d50e270b3e88e18 : SUCCESS in 6m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/68becfd6618948a4a47266f98351af63 : SUCCESS in 22m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f295eea196d84cf3b8d07467fe6df9f3 : SUCCESS in 44m 46s","accounts_in_message":[],"_revision_number":5},{"id":"ca3b731e40b84c0508eff1e5917eed2e79cb636d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 20:28:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4864f8776b77442494c8434106a671d9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/10d488c7dc374bd690e54999efe725cc : SUCCESS in 15m 19s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1ae9f1426de44752a60b58df3027966b : SUCCESS in 1h 01m 29s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e76f9fb22c86479fa3ec44e6815ff270 : SUCCESS in 19m 41s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/84de50bb571a48aa8a6bc289e5702c8c : SUCCESS in 1h 17m 52s","accounts_in_message":[],"_revision_number":5},{"id":"4d821abcd8131f431f0fcd7032225067cd36bc4a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-18 14:52:43.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d62eb8ddfeaa582171e8329549c947842e52e6d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 14:53:20.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b5e28035f9ec52f18ceb143deaef56832d8c65de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 16:09:11.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e635707c3e244a5f9af7288045ac6343\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb2d5ba8898649718e8214a72203a037 : SUCCESS in 10m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/48ba0e7fad3b4355a094bedd71dca380 : SUCCESS in 49m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/55b16c733c784f6dbf70599b506cf13c : SUCCESS in 57m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ecb1dda50c2f4b0b85479e01d547a314 : SUCCESS in 55m 03s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b4bb4811264a48bfabf4ea5fc6324b94 : SUCCESS in 9m 15s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/5fd6294121fd4061af93812d90d78132 : SUCCESS in 7m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fcd336c519154554bdec9d488adbde9a : SUCCESS in 16m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8ee70b756f2d4dfe909c3063dfeff1ab : SUCCESS in 17m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/57a7c8e694de4fa6a1d793979ad6d920 : SUCCESS in 15m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8fab12b7e67c48cb8aa989b0aefe9e69 : SUCCESS in 16m 57s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b35540aa8198455982da381927a13e7d : SUCCESS in 17m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5d7ab9f3730341708281c13ac51cf775 : SUCCESS in 17m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/06691cbcca1f4658b62bb2b7f93ec31c : SUCCESS in 19m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/200629c8cd3844dcbac57ecc3bb84bd0 : SUCCESS in 17m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/04dce9f8dd7f42c3a4824dab35a00b33 : SUCCESS in 20m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b76af5dbdfa64411af460bb3f9a16c10 : SUCCESS in 17m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3982548aaf2b48f08f8bc0e93f05127e : SUCCESS in 18m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/12f5bda1a9864dbcb1492a40265a97c0 : SUCCESS in 8m 45s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c4094f6ae9d241a5a3de2ce0840cdf19 : SUCCESS in 7m 00s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c527ef1ca7f14537b97605ccea68069b : SUCCESS in 52m 21s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/00ea5f92113f4bfbbb5f72401c0ee5ca : SUCCESS in 1h 04m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0e23c57c50a4405ea1cb910910cf85f5 : SUCCESS in 49m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/31db4264280643dd801d83629671acd3 : SUCCESS in 45m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce17b57381714e2aadddcf43cd8919b2 : SUCCESS in 25m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7ab55c761cc4f1ebaa7a52d79663306 : SUCCESS in 5m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/90ce06c7a89f46d2a47f6bc816b6c387 : SUCCESS in 1h 10m 43s","accounts_in_message":[],"_revision_number":5},{"id":"2e5da68b76490a751d199c746175222fe1361d92","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 16:09:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"112bd61fcdcabc07165e7fa14a320d3eb0e30ec2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 16:10:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1717c6171cf445eb98088c89266ba63b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b54bb1ad600438aa323e869c92fd767 : SUCCESS in 1m 06s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f778c4808f9f4c8ea491abe9530a5d22 : SUCCESS in 28s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/429674c0cab04b70bc5de63910963ee1 : SUCCESS in 32s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e5d730dc56c0c75c9f3a8b7f6119dcc638ccded0","revisions":{"f3b05f834caf8603bbb7731d430548a9fc5227b6":{"kind":"REWORK","_number":1,"created":"2023-06-05 22:50:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/19/885319/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/885319/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/885319/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/885319/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/885319/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/885319/1"}}},"commit":{"parents":[{"commit":"ff24428f417cbf0bc71ce54a6ec9528ff618485b","subject":"proxy: Bring back logging/metrics for get_*_info requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff24428f417cbf0bc71ce54a6ec9528ff618485b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-10 19:59:39.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-05 22:49:14.000000000","tz":-420},"subject":"proxy: Get rid of iter_nodes helper","message":"proxy: Get rid of iter_nodes helper\n\nAll it did was proxy through to NodeIter, anyway.\n\nChange-Id: Ifec8d3a40f00141a73f6e50efe0b53b382ab2ef3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f3b05f834caf8603bbb7731d430548a9fc5227b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f3b05f834caf8603bbb7731d430548a9fc5227b6"}]},"branch":"refs/heads/master"},"3b4be52be0787ef995086f015748f7fe01faf7f1":{"kind":"REWORK","_number":2,"created":"2023-06-06 04:12:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/19/885319/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/885319/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/885319/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/885319/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/885319/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/885319/2"}}},"commit":{"parents":[{"commit":"ff478f78f324691f8129f480f7df1e2a257ee7a0","subject":"proxy: Bring back logging/metrics for get_*_info requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff478f78f324691f8129f480f7df1e2a257ee7a0"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-10 19:59:39.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-06 02:50:45.000000000","tz":-420},"subject":"proxy: Get rid of iter_nodes helper","message":"proxy: Get rid of iter_nodes helper\n\nAll it did was proxy through to NodeIter, anyway.\n\nChange-Id: Ifec8d3a40f00141a73f6e50efe0b53b382ab2ef3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3b4be52be0787ef995086f015748f7fe01faf7f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3b4be52be0787ef995086f015748f7fe01faf7f1"}]},"branch":"refs/heads/master"},"bc4cf16cb805b65a9cde352a8aac02f5e57558da":{"kind":"REWORK","_number":3,"created":"2023-07-24 15:43:51.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/19/885319/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/885319/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/885319/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/885319/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/885319/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/885319/3"}}},"commit":{"parents":[{"commit":"88c9c3bc0b4a4f7f3c9c500e4f5962bdebab8dc3","subject":"proxy: Bring back logging/metrics for get_*_info requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/88c9c3bc0b4a4f7f3c9c500e4f5962bdebab8dc3"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-10 19:59:39.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-07-24 14:54:16.000000000","tz":60},"subject":"proxy: Get rid of iter_nodes helper","message":"proxy: Get rid of iter_nodes helper\n\nAll it did was proxy through to NodeIter, anyway.\n\nChange-Id: Ifec8d3a40f00141a73f6e50efe0b53b382ab2ef3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc4cf16cb805b65a9cde352a8aac02f5e57558da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc4cf16cb805b65a9cde352a8aac02f5e57558da"}]},"branch":"refs/heads/master"},"73473f0aba634b2bfcea95548c92b1e629fbe781":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-07-24 15:58:51.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/19/885319/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/885319/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/885319/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/885319/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/885319/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/885319/4"}}},"commit":{"parents":[{"commit":"d2c9c5002169723266f4100b6d6d2aa03f89c555","subject":"proxy: Bring back logging/metrics for get_*_info requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2c9c5002169723266f4100b6d6d2aa03f89c555"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-10 19:59:39.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-07-24 15:53:15.000000000","tz":60},"subject":"proxy: Get rid of iter_nodes helper","message":"proxy: Get rid of iter_nodes helper\n\nAll it did was proxy through to NodeIter, anyway.\n\nChange-Id: Ifec8d3a40f00141a73f6e50efe0b53b382ab2ef3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73473f0aba634b2bfcea95548c92b1e629fbe781"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73473f0aba634b2bfcea95548c92b1e629fbe781"}]},"branch":"refs/heads/master"},"e5d730dc56c0c75c9f3a8b7f6119dcc638ccded0":{"kind":"REWORK","_number":5,"created":"2023-08-16 19:09:02.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/19/885319/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/885319/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/885319/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/885319/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/885319/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/885319/5"}}},"commit":{"parents":[{"commit":"cc59929412f2fa0271c2940afaa3fa3c6e9aff99","subject":"Merge \"s3api: extra unit test assertion for statsd metric prefix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc59929412f2fa0271c2940afaa3fa3c6e9aff99"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-10 19:59:39.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-08-16 17:16:18.000000000","tz":-420},"subject":"proxy: Get rid of iter_nodes helper","message":"proxy: Get rid of iter_nodes helper\n\nAll it did was proxy through to NodeIter, anyway.\n\nChange-Id: Ifec8d3a40f00141a73f6e50efe0b53b382ab2ef3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e5d730dc56c0c75c9f3a8b7f6119dcc638ccded0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e5d730dc56c0c75c9f3a8b7f6119dcc638ccded0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
