)]}'
{"id":"openstack%2Fswift~886010","triplet_id":"openstack%2Fswift~master~Ic598715f69227f24ebf5617846a84e6bf1115a96","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-09-11 03:54:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic598715f69227f24ebf5617846a84e6bf1115a96","subject":"tests: Include filter with no _pipeline_final_app","status":"MERGED","created":"2023-06-13 17:54:54.000000000","updated":"2023-09-11 03:55:17.000000000","submitted":"2023-09-11 03:54:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886010","meta_rev_id":"6fbe10916f58bfd56d5f0a9387cbca8b9f67b446","_number":886010,"virtual_id_number":886010,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-11 03:54:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2023-09-11 02:49:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2023-09-11 02:49:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-13 18:56:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-11 02:49:06.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"0b09406f2ce9551be5ffbd42f92593e02cdaf3c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-13 17:54:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b6ccb58317e4a29d8086370a3adfa0931bbb37f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 18:56:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b629c9373b714aa38bd26c6691bea74a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33ee6af2070c49a98c97f5a1c3a7815f : SUCCESS in 6m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/36a0e64583574b7faafcb572bc8c87ad : SUCCESS in 17m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0e41038b6a8b41d1832cee489cc59dfd : SUCCESS in 17m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/037e1985d73d4ac3b408995aa5c2e437 : SUCCESS in 16m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2620194d368a4e66be8c79c6ed6e5781 : SUCCESS in 7m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d571ed232e0b4ddb927747d063370ae9 : SUCCESS in 4m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/23cb450130674fe3a53fdf8310c93b72 : FAILURE in 20m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/827f211799bf45de974a3f04563d4979 : FAILURE in 14m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0a7e0a00b74046b29c6748625e46e13b : FAILURE in 14m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b4eeb7ad324b44988c50a4af71e47fed : FAILURE in 18m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/230456966c9d48aba354dd07e3b1e74c : SUCCESS in 18m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/213a6be83ccd4bf1bb32004b8f397158 : SUCCESS in 15m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/674fa57fee624587b1aba4d9d40020f5 : SUCCESS in 15m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1d07b3e82def4b2abf844a5a3ee84465 : SUCCESS in 12m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d05c1711e35a4d53be67552655c29382 : SUCCESS in 13m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/42272b9b5cfe45f6bcad9b771dca110b : SUCCESS in 14m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ee04d1ead5cc46079659e515e7be8948 : SUCCESS in 38m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b6267d154c7d47258a307b3bd4692b68 : SUCCESS in 52m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aafb344ae35c4e2791f222f490ed6402 : FAILURE in 15m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c9d3cccfb9d4d14ad600d36d232c100 : SUCCESS in 3m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d70b0cdeb22d4b57ba5a8b2a3b66f4d3 : SUCCESS in 20m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"141bd9465f0a4e550c40ebd2932c7f64a2111461","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 19:29:11.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33f3386a48264b28970838db2998f26c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1ce46566955a4f4c93605f6ff50d8976 : FAILURE in 19m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e05e69c0106f49d0833e98fc214db4c0 : FAILURE in 15m 15s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/acd797fed2a1405c8d6eaf7a9415fbdc : TIMED_OUT in 1h 31m 33s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d8af10b5504b44548ff7eb013472df54 : SUCCESS in 18m 12s","accounts_in_message":[],"_revision_number":1},{"id":"0901975947cc6d9ca5b8ad02f157b8afea326b65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-23 17:58:34.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"920e364a2e475e479d035ce760d74123eedd3e3a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 18:22:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d1324f75f49448abec906aa2a78af1e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0021241e7b7c433c94f76eb83bbebd29 : SUCCESS in 15m 45s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c9345c16131c4360aef4cd6c385af5e4 : SUCCESS in 15m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/cbc9492645c44afe952b540531dfb898 : SUCCESS in 19m 15s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f884a60432d242c5a8a13e2244aacd47 : SUCCESS in 21m 01s","accounts_in_message":[],"_revision_number":2},{"id":"f05d92bc541fcfbc90e7031cabc184b88fbbf7c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 18:57:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24ac267d55034b4c950f50a4883af31d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/290ca0ad5e8a4010b80267ca58433de9 : SUCCESS in 9m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b78d8a9bc9140fd831090daaa7f48cd : SUCCESS in 23m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a91259802f9e431ea46ba03135a2b40b : SUCCESS in 24m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/96a762b7b4ae46a288602a2209176555 : SUCCESS in 25m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6161a993783b4ba9b796184692f163de : POST_FAILURE in 38m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7aba391204f246b08b8de066dd977086 : POST_FAILURE in 34m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/413bc7b27d2c4565b3806ae2d67a14ec : SUCCESS in 23m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/af1c0aa22cd9437d8b9301561e767326 : SUCCESS in 15m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2cf980fb7cf14c6baa600b9d363bf19b : SUCCESS in 15m 43s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b458556090ba4517877f5c5867d0ab9a : SUCCESS in 19m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7b75707a379a4373925cfa23d751713e : SUCCESS in 23m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fc6a13f4e31f4539a8753e4a102ed154 : SUCCESS in 16m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3471e910ad9b4c99a31f973d52dc3c51 : SUCCESS in 19m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ae092ffd019e49e8af0e44c73beb3fed : SUCCESS in 13m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7b804d75633c469ebde8fd62ba889a1d : SUCCESS in 23m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4345a1b247fd4f15a63d3f72ab4530c3 : SUCCESS in 16m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a1b735e7ce6d44288128c0a6fc837acc : SUCCESS in 53m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dc14d7ac5bdf41daa36649b60a58ce67 : SUCCESS in 51m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21cefb8ab84d465b8ae9eab589a126aa : SUCCESS in 21m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b61ed22a30424266845d562dc9d8cdab : SUCCESS in 4m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d613be0c331945f18bf0c87463ddaf0a : SUCCESS in 21m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"59cb3fa06d5e28a95af4062253fe57950f70ed83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-21 22:36:10.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"c455028d3f9bbea17c9e1703ddb1bd9f8c9edc38","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 22:59:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/344b8067c7b649ba86ed78e0d31e9b0f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/68f885dfedce44b687f0f6e705b09d6b : SUCCESS in 20m 43s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/93b01778f62742af89df607f71d84c43 : SUCCESS in 21m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4216e9e9c03349e69a149bc2af33cbad : SUCCESS in 22m 16s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/eef8b21c3ff142c0bd9492a23a89e00b : SUCCESS in 20m 48s","accounts_in_message":[],"_revision_number":3},{"id":"84f594d4bed40578958579f33d736cf067e9fb25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 23:38:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49685bd61db6472b95c64eefdc4765de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e093832f03aa4a6badb27393a7402bd7 : SUCCESS in 6m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ce27a51ec470486e9238c1e407b32420 : SUCCESS in 16m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/124ee47f998b4dcda6bf09fdfe2d0cad : SUCCESS in 23m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8525392699b844289a267578e0eedd66 : SUCCESS in 24m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/44362fe43f8c4d78af347c08a54711f6 : SUCCESS in 7m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5b359fe96d3c47dbb67b13c3ba4fd70e : SUCCESS in 5m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e4047621de6741e08b18c1752fabe06f : SUCCESS in 21m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4d33511823e84f5e9252602697bd981f : SUCCESS in 22m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/aef5e2d4234b4acabbae43fca9d646f6 : SUCCESS in 20m 32s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0c8bab41b6da432295fcab8522a397d4 : SUCCESS in 17m 23s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d29daa0c33c84a7cbefa0b71e1b549c4 : SUCCESS in 16m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0e12ab5f41d343688a6f34583c1ace13 : SUCCESS in 15m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d76381cd918141dc987e759f3755a36e : SUCCESS in 16m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ad6097d3720b4e02b906a0a4f45a8958 : SUCCESS in 16m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/430010615fd24a9081c6e2baae800f20 : SUCCESS in 14m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4b7c4502af0f45598655297f370923fd : SUCCESS in 14m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0e9ef4d5f8554885bcf7e16202a49964 : SUCCESS in 15m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a8f0417dd1ff4ac89f2a2e2eadec083e : SUCCESS in 56m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f29b9b83dff44b8cb4fab9f246f3fb7b : SUCCESS in 52m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6daaffd45f6a425cb841e183457a7f10 : SUCCESS in 17m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c680af9098a4989a5988b6b2cadfbca : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9b52944f64eb49549d902d425612ee74 : SUCCESS in 16m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"46dd0d747389699b65986a79f8457d98a51c81f9","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-09-11 02:49:06.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"79848e208ace0b5772a4fc7c7836d4e1d857f8d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 02:49:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"eb7da3fc23b55db0f176a4a088ed0f330bb6dc12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 03:54:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee1aa27a499c4205911b17cfb91add03\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d74a94ad39442e79822c9a4e1fd4e0a : SUCCESS in 11m 24s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/79f36974454142bcbb739f84789842bc : SUCCESS in 11m 30s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/21da2a2ae5eb45e2b65996bcf37dd2c3 : SUCCESS in 8m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/62d854c9c03c4e7c8c40347f930d46e6 : SUCCESS in 39m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fbf811136e6749cf913147a3f23195e5 : SUCCESS in 17m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6113b0222d2d4a56ad3d9cf08d25cc66 : SUCCESS in 15m 19s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2986e2b83c0e46029de1c27084d15491 : SUCCESS in 17m 46s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/12d33db4a2e843538fbe54b185a76234 : SUCCESS in 16m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8f09e3a901ae4609a561f825504aa161 : SUCCESS in 16m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b0cf9febc7454697a4cf3e93ef5decc5 : SUCCESS in 15m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a22f56f54e694acababce26af348f382 : SUCCESS in 15m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bece068a1f194cf7998b49ba89ce3bc7 : SUCCESS in 15m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9afc5b54ed844c98b3e4ea5eca9849be : SUCCESS in 15m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5d2b366191084f89972ac3b012937998 : SUCCESS in 22m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c827d0ecc2ee4d8fad0fdb55398e0747 : SUCCESS in 7m 32s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5e3024b470364a6fb2a231d3d9a8cf4d : SUCCESS in 8m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/03982cdbe116487c81108c9bebab81ea : SUCCESS in 1h 01m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9b2202588ef448ff9a9c76e8921b4706 : SUCCESS in 56m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10d098201dbf49db8bad68937167fecb : SUCCESS in 20m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc1a6456bc2b483288e80bf54ef03e87 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":3},{"id":"f370908fe0935944a1e495c5f281f058705f2b39","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 03:54:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"6fbe10916f58bfd56d5f0a9387cbca8b9f67b446","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 03:55:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a78f92dfcb424338a71e22f239e88ea8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddd6f866bc9245609e3061b9c7b870df : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/51ff96a7ceaf4e04805a53b3778db69c : SUCCESS in 39s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/0e8033d542874ef996cb2da302b43870 : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"25e643eb8debdc995cf4ab93ac5ca81edec782c8","revisions":{"000d2c7655e032b76919d0a3e5c301f86ddd3bcf":{"kind":"REWORK","_number":1,"created":"2023-06-13 17:54:54.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/10/886010/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/10/886010/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/886010/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/886010/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/886010/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/10/886010/1"}}},"commit":{"parents":[{"commit":"b67e53efa75cc142bb527551e70c5ae09b34d9e3","subject":"proxy: add new metrics to account/container_info cache for skip/miss","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b67e53efa75cc142bb527551e70c5ae09b34d9e3"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-13 17:49:15.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-13 17:49:19.000000000","tz":-420},"subject":"tests: Include filter with no _pipeline_final_app","message":"tests: Include filter with no _pipeline_final_app\n\nThirdy-party middlewares may compose themselves of multiple filters,\nsimilar to what we have done with encryption and versioned_writes.\nWe should assume they use the get_account_info / get_container_info\nAPIs, but we cannot assume that they have been updated to pass our\npipeline info through all their layers.\n\nChange-Id: Ic598715f69227f24ebf5617846a84e6bf1115a96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/000d2c7655e032b76919d0a3e5c301f86ddd3bcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/000d2c7655e032b76919d0a3e5c301f86ddd3bcf"}]},"branch":"refs/heads/master"},"6b9b99a46547ac152e9e3e84d2e61108514f82a7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-06-23 17:58:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/10/886010/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/10/886010/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/886010/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/886010/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/886010/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/10/886010/2"}}},"commit":{"parents":[{"commit":"ca3f1077063a5b1e28239b57f9f8ef22dc7901ce","subject":"tests: Fix replicator test for py311","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca3f1077063a5b1e28239b57f9f8ef22dc7901ce"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-13 17:49:15.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-23 17:57:16.000000000","tz":-420},"subject":"tests: Include filter with no _pipeline_final_app","message":"tests: Include filter with no _pipeline_final_app\n\nThirdy-party middlewares may compose themselves of multiple filters,\nsimilar to what we have done with encryption and versioned_writes.\nWe should assume they use the get_account_info / get_container_info\nAPIs, but we cannot assume that they have been updated to pass our\npipeline info through all their layers.\n\nChange-Id: Ic598715f69227f24ebf5617846a84e6bf1115a96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6b9b99a46547ac152e9e3e84d2e61108514f82a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6b9b99a46547ac152e9e3e84d2e61108514f82a7"}]},"branch":"refs/heads/master"},"25e643eb8debdc995cf4ab93ac5ca81edec782c8":{"kind":"REWORK","_number":3,"created":"2023-08-21 22:36:10.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/10/886010/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/10/886010/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/886010/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/886010/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/886010/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/10/886010/3"}}},"commit":{"parents":[{"commit":"1d742eee398934fc501b5ca9798993b5f6ace9ed","subject":"Merge \"tests: Pollute stderr less\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d742eee398934fc501b5ca9798993b5f6ace9ed"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-13 17:49:15.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-08-21 22:35:29.000000000","tz":-420},"subject":"tests: Include filter with no _pipeline_final_app","message":"tests: Include filter with no _pipeline_final_app\n\nThirdy-party middlewares may compose themselves of multiple filters,\nsimilar to what we have done with encryption and versioned_writes.\nWe should assume they use the get_account_info / get_container_info\nAPIs, but we cannot assume that they have been updated to pass our\npipeline info through all their layers.\n\nChange-Id: Ic598715f69227f24ebf5617846a84e6bf1115a96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/25e643eb8debdc995cf4ab93ac5ca81edec782c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/25e643eb8debdc995cf4ab93ac5ca81edec782c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
