)]}'
{"id":"openstack%2Fswift~886538","triplet_id":"openstack%2Fswift~master~I59add10d886134af86e1bddec2ba39931c1213cb","project":"openstack/swift","branch":"master","topic":"py311-support","attention_set":{},"removed_from_attention_set":{"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2023-06-23 00:55:22.000000000","reason":"\u003cGERRIT_ACCOUNT_597\u003e replied on the change","reason_account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-06-23 10:42:20.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-06-23 11:59:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I59add10d886134af86e1bddec2ba39931c1213cb","subject":"tests: Fix replicator test for py311","status":"MERGED","created":"2023-06-20 23:42:55.000000000","updated":"2023-06-23 12:00:39.000000000","submitted":"2023-06-23 11:59:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"886538-py311-support","meta_rev_id":"7b9bd79537ad382380f1fb72385fedd770c3a612","_number":886538,"virtual_id_number":886538,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-23 11:59:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2023-06-23 00:55:22.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2023-06-23 10:42:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2023-06-23 10:42:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-21 00:10:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-21 00:48:49.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"},{"updated":"2023-06-21 01:15:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-21 14:13:30.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2023-06-22 10:26:08.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"ee1eea579fb3dd60d5590bc0f93ecbf87f98335f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-20 23:42:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b411f52cb24b48ab89a3fa1ae0add34778905135","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 00:10:42.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39d4d081fbf843d59d5f75ee9fd14bba\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c5e42b5494a48188a8a5ca4222891d5 : FAILURE in 16m 51s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d9bbe88b431548a0b2cdeb3619b4075c : FAILURE in 16m 43s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b6f8d37090734229bb5b327b7d5c1f05 : SUCCESS in 21m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/dfa7d82cb9664c659c4025503ff3fd7b : SUCCESS in 23m 10s","accounts_in_message":[],"_revision_number":1},{"id":"0569c9286083fb99be7442710108a0504c5c12bd","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-06-21 00:48:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2a487933f4374a66da1f3eb631df64b8671d0b02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 01:15:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0448060c19074f63ba6dbdbd811b696c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/570290c90f794b7e9322106cc7ccedc8 : SUCCESS in 8m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a69b6d78ae3248838cb3437d93ee341f : SUCCESS in 21m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7bfafb1194b84775ae93160bf58bae0f : SUCCESS in 21m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b38bb3bb66164d61ad74cb165316848f : SUCCESS in 18m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1d65aaa7c0a0468ca3eb9f4966aaf893 : SUCCESS in 8m 43s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e58e5f74dcd14b17988467ed826a72e8 : SUCCESS in 5m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a7bd2edc907a4dfd8e3da0600aacba2e : FAILURE in 21m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f11cf42624344aafad27779bf105f099 : FAILURE in 20m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/49fc40e0d8db4bf1af61d05b8440dc5a : TIMED_OUT in 1h 31m 28s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6b5d4e38a2d24a889d1cdeb4eb66abb5 : FAILURE in 20m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a7ec4701a2c54245ad744e503759fb8b : SUCCESS in 18m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a0901ade508040c39ec50915fb6c0279 : SUCCESS in 13m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d5bc1b3d511f4b84a33ac4c1890cc8bd : SUCCESS in 21m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/adc9d0bd7af54bfb8ea84acd8f469596 : SUCCESS in 14m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/196301873cfa414bafa0e11b3a1b64a0 : SUCCESS in 24m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e200b8b42099460baeb22869415620c5 : SUCCESS in 15m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c8864fe419d24b3eaa738c0aa638916b : SUCCESS in 49m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cd7faaafe47e497aa8edc6dc004328ca : SUCCESS in 46m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70fd457aceb74e83ac8cf539c290d632 : FAILURE in 22m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab4e0d47631c4daab35a0dfdfe9e4b5a : SUCCESS in 5m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ef19ed97b8ed416091e6a76e5701023f : SUCCESS in 22m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6006d381790996747e6a3d900d4ca4858281b22b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-21 01:16:30.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"203e85471981386462bc0e2078166c05ef2f3cd1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 01:41:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69c9aa83fe6e4bfbb606022badd0e4ff\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/784099dabe984bedad507ba24d1b377a : SUCCESS in 18m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3b02e5fbbbb748ad937ea89ea7879edd : SUCCESS in 17m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5619256b048040a498295275dac36105 : SUCCESS in 20m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2d3769048e90401081192083356419ad : SUCCESS in 20m 04s","accounts_in_message":[],"_revision_number":2},{"id":"9f7a4b8d5a1be6f771fcab1ee1418298e955580b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 02:05:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0059d6eb862408fa0c69b4249662f18\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/478988f73a3c4e3299f18697198627f7 : SUCCESS in 8m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d263d7c9fa4640ce8238d2701431bc13 : SUCCESS in 21m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7faca2d820ef49bdbda2bcf40d6ec21f : SUCCESS in 21m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed173388c76641f6acbe0c1391317d4f : SUCCESS in 22m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fe5a24470c41446699e13f8485e79fc8 : SUCCESS in 8m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/83460b9993d74f2e8d89caf4fe8e0a44 : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2d9aefedc4bc4732bfa21346117aae38 : SUCCESS in 17m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8a27ff7d55a041d3867cd1fbfcc85d4f : SUCCESS in 16m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f34d8a67a00e4d16a35807f00ec152af : SUCCESS in 17m 19s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/854b85382e9a4621a4c08a0344551cb0 : SUCCESS in 14m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/95033a2724ee45c7be69d487df88b244 : SUCCESS in 22m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/82e075ce45a4447ea84c15d6f6d7de1a : SUCCESS in 17m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0d50da0eee0643929c14f877f0241027 : SUCCESS in 23m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/df84c4e9a8e24674bb3074f81fcc4446 : SUCCESS in 15m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/85dbffab620248e79c5725d9f99cb550 : SUCCESS in 18m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/03ba224069214b93b75f5fb2b0ce9cf3 : SUCCESS in 21m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/747ae72a4b4b420594e15958dbe6c11c : SUCCESS in 47m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/374470cb0d1f464cb6506b2fcb6f23f7 : SUCCESS in 45m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23714c9a2a854776b7ee30cf38636157 : SUCCESS in 21m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d87f969e169843f7b1789d2d7f1691ff : SUCCESS in 4m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b33f9e62bd044e27b400ca092b22024d : SUCCESS in 22m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"44c887b371b2fb20b701b4aa4152cc6956419f92","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-21 04:12:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1e0d69c90e628429e3e346c3ea767292a9f50efe","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-06-21 14:13:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4e391f6f4328ba221f256bf58549ba3f78f9fbdf","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-22 10:26:08.000000000","message":"Patch Set 2: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"65dbc314c9d8faa2e1436254ebd7f621f4752795","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-23 00:28:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"68ccd52b458e84a7b132503f76ae636ebf0c0a1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-23 00:28:27.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"b31a2268919f7d8db41011ea336e739acfccbba3","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-06-23 00:55:22.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0e4c76de34ed24484a49ca13bb9e2b6830368cb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 01:40:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cd9aedb5f1b44e6b5963bf2d2f28b1e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e1399ca641c48b0b94ad71b2e395b7c : SUCCESS in 8m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b178c4478ee74deb9d1431d38702d10c : SUCCESS in 23m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/84cfda99645d4253a6a49593c3f1534f : SUCCESS in 18m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b00a52a1e163445d953cc8951c0381fa : SUCCESS in 18m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9e399a30c5474c9aab982827ef6ab4ea : POST_FAILURE in 38m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b2363a3256214e7382d753af9174e927 : POST_FAILURE in 34m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9068134004b14f66b20a04d2d7de1c6b : SUCCESS in 21m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2b765433c2ae4401a5a3f7d6699ea938 : SUCCESS in 20m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/92a5c88664e946ff95ec4d7ec6a40e81 : SUCCESS in 16m 08s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e97c7ce0e79141cdacb89c0a64064937 : SUCCESS in 15m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3df6bdd36543423aa24dfcf078e0f2b5 : SUCCESS in 16m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/78945c263059446dbfddf565c1712ab6 : SUCCESS in 16m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0f6fb37663834be0a83a41d61a695ffb : SUCCESS in 16m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1ad579463d964758bc40e9302ebafe77 : SUCCESS in 15m 50s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/573ddb0a090447c5bf05001d6385fcb3 : SUCCESS in 15m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/706024d27ef24f74a9ffeee58f7a4545 : SUCCESS in 55m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f22cdc6598754cc8a70c0daf525a81c3 : SUCCESS in 51m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0b92a7d52f9842e4b8debbe9c7aaae7d : SUCCESS in 59m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aec8cbab0994492abdc859d8ed7a1e2b : SUCCESS in 14m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04cdaab7cdea413eb7fe89610dbcb825 : SUCCESS in 9m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3ac047915d19480a8505d8c265e5686d : SUCCESS in 19m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"578c73a17a99dc70905727f6092273458917246a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 01:52:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30d858b86b6f452781437f0949463bc6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a166e06330dc4534940e579d855dbeff : SUCCESS in 17m 15s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cd988592fa004eddacf95d3c8e7070c6 : SUCCESS in 1h 21m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/aa4471df558c4f509274f2bb4a596b5c : SUCCESS in 19m 31s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a0d93756921c406785698326b48a6c3d : SUCCESS in 20m 17s","accounts_in_message":[],"_revision_number":3},{"id":"971fa5b4a170cba9ece944eedb863350db41f9fc","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-23 10:42:20.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"61dc2995aec746ba34c9a4c11fa19157afbf0855","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 10:42:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e03a80401b2f2b70e9a1824919a337cc1de9085d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 11:59:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33671e524a3844a68dbb9f50300a397d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddefb494ce6a45fea16ae9e4b8b19a96 : SUCCESS in 7m 50s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/951f5316eb6d4de793d6a05c3456509d : POST_FAILURE in 38m 14s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/ca48053cb9b94438a52f3604708d1cff : POST_FAILURE in 35m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/067d9690d788485b9a5cf7266b4086f2 : SUCCESS in 32m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bb8c2006baa141af8770c4433e215b88 : SUCCESS in 27m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d79cff67f7cc4bd4836a45c4cff5de3f : SUCCESS in 26m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ef040e716f1c4394ba3a5b098fd2b843 : SUCCESS in 17m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/660da3e99c0d485ca2cdc5c0dfdccbea : SUCCESS in 26m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/626fb89e3f1e4e498a90482ce50a794e : SUCCESS in 16m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4f7043c57bb942228975bdfdbbc080cf : SUCCESS in 15m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/795ecbf1921944f0b829a5ae4e41dd3d : SUCCESS in 13m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a10949447136431d8403e810ee04ac9c : SUCCESS in 50m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/31bdf2360a774d5b9037385846cafa37 : SUCCESS in 13m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/89813b0a35b54344bdc675870001262e : SUCCESS in 8m 13s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/44b0a6ee28a24beaa9d0a6ab6db1468d : SUCCESS in 6m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ca23df9ec4ae443fbb87e7b1c88e65ad : SUCCESS in 40m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1ea6d5a7cf5d46869c9a2f4b8f926afb : SUCCESS in 49m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f829032e4c94f24bf6f483e4dc1b43d : SUCCESS in 15m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65bb1917870a4102b7905d4ba79512e6 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":3},{"id":"3925fdcca2ab9630923017a44339ce34b8660770","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 11:59:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"7b9bd79537ad382380f1fb72385fedd770c3a612","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 12:00:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d09f83b188c4c729775d4bde756f75c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8271ce03f5514505900182977206b38e : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/0c8edf0f35ca470f9229c4b374b47f23 : SUCCESS in 37s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/46b06d6f788d4ed8858de929fa253a14 : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ca3f1077063a5b1e28239b57f9f8ef22dc7901ce","revisions":{"043c8da461f8281e87dc88d8c6f436c0df76713a":{"kind":"REWORK","_number":1,"created":"2023-06-20 23:42:55.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/38/886538/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/886538/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/886538/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/886538/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/886538/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/886538/1"}}},"commit":{"parents":[{"commit":"e59de0fdfd66543228dc6d9ee37dda0c2f6b8229","subject":"Green GreenDBConnection.execute","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e59de0fdfd66543228dc6d9ee37dda0c2f6b8229"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-09 21:25:19.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-20 23:21:48.000000000","tz":-420},"subject":"tests: Fix replicator test for py311","message":"tests: Fix replicator test for py311\n\nshutil.rmtree started using os.rmdir in a way that made our tests fail.\n\nChange-Id: I59add10d886134af86e1bddec2ba39931c1213cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/043c8da461f8281e87dc88d8c6f436c0df76713a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/043c8da461f8281e87dc88d8c6f436c0df76713a"}]},"branch":"refs/heads/master"},"3e766fb2eeea68278f231bfe637c26c7d3fb9526":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-06-21 01:16:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/38/886538/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/886538/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/886538/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/886538/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/886538/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/886538/2"}}},"commit":{"parents":[{"commit":"c5a3354fea345566010420801e8d27326920db85","subject":"Green GreenDBConnection.execute","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5a3354fea345566010420801e8d27326920db85"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-09 21:25:19.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-21 01:12:29.000000000","tz":-420},"subject":"tests: Fix replicator test for py311","message":"tests: Fix replicator test for py311\n\nshutil.rmtree started using os.rmdir in a way that made our tests fail.\n\nChange-Id: I59add10d886134af86e1bddec2ba39931c1213cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e766fb2eeea68278f231bfe637c26c7d3fb9526"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e766fb2eeea68278f231bfe637c26c7d3fb9526"}]},"branch":"refs/heads/master"},"ca3f1077063a5b1e28239b57f9f8ef22dc7901ce":{"kind":"REWORK","_number":3,"created":"2023-06-23 00:28:27.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/38/886538/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/886538/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/886538/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/886538/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/886538/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/886538/3"}}},"commit":{"parents":[{"commit":"c5a3354fea345566010420801e8d27326920db85","subject":"Green GreenDBConnection.execute","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5a3354fea345566010420801e8d27326920db85"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-09 21:25:19.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-23 00:28:12.000000000","tz":-420},"subject":"tests: Fix replicator test for py311","message":"tests: Fix replicator test for py311\n\nshutil.rmtree started using os.rmdir in a way that made our tests fail.\n\nChange-Id: I59add10d886134af86e1bddec2ba39931c1213cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca3f1077063a5b1e28239b57f9f8ef22dc7901ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca3f1077063a5b1e28239b57f9f8ef22dc7901ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
