)]}'
{"id":"openstack%2Fswift~886633","triplet_id":"openstack%2Fswift~master~I8ed9b4ca5af90e9a64ec996725e59cba34f796a5","project":"openstack/swift","branch":"master","topic":"TOPIC-BRANCH","attention_set":{},"removed_from_attention_set":{"36116":{"account":{"_account_id":36116,"name":"Philippe SÉRAPHIN","email":"philippe.seraphin@infomaniak.com","username":"pseraf"},"last_update":"2023-07-07 20:06:08.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8ed9b4ca5af90e9a64ec996725e59cba34f796a5","subject":"In the case where we can\u0027t stat the device, an error search in the Kernel logs must also be carried out, and the device unmounted if necessary","status":"ABANDONED","created":"2023-06-21 12:37:37.000000000","updated":"2023-07-07 20:06:08.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d56f763329166c784e108d2d803796917e369f91","_number":886633,"virtual_id_number":886633,"owner":{"_account_id":36116,"name":"Philippe SÉRAPHIN","email":"philippe.seraphin@infomaniak.com","username":"pseraf"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-06-21 18:30:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-21 13:03:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-21 18:30:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cfff75705044698bdd3f99567cb166bbd731c960","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36116,"name":"Philippe SÉRAPHIN","email":"philippe.seraphin@infomaniak.com","username":"pseraf"},"date":"2023-06-21 12:37:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"78ed816886d1bce3325aa0db3387183b6a6fdbd7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 13:03:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1b8fb1a455b49c885ee5555f6d98c1b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/189f3748d2e44f8684162655364ec043 : SUCCESS in 19m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b57a3d85bd154e3f868b816b7d37d39d : SUCCESS in 18m 32s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1af2bfec784a4dd29366819b260c29e6 : SUCCESS in 21m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f62677579eab4e4e851cb2881ddfeeb9 : SUCCESS in 23m 02s","accounts_in_message":[],"_revision_number":1},{"id":"52aa646c65b1734d67714a3d318640f75e7658e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 18:30:01.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/78211c63ddf440dbbad6e6d53566f652\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82209938be784846855fb8beaf7c3b6b : SUCCESS in 10m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/55a6f81a412843ada83aff632c11b79c : SUCCESS in 1h 01m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4fd57a215b5e43c59b218ee0c6972ca4 : SUCCESS in 58m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/58332a42acb3443486df1889f246c7f8 : SUCCESS in 54m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/583548f1337a4d308d26d22ed7fa9910 : SUCCESS in 19m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fffdee9568804d6b8525365ca39d309b : SUCCESS in 17m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9a6872c6a62b4c6db71e8acc037b4f76 : SUCCESS in 18m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8e1165b52ce443ba97626b6ba779b5be : POST_FAILURE in 38m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0ed39701a85348e2919279ea4c8d2ea1 : POST_FAILURE in 35m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/56a49bee8b024a53b218744933007639 : SUCCESS in 18m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/345dd86cc03f4e12ab868a7fda7f4411 : SUCCESS in 16m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9e1f8d56d38a479687ce2d977f82e642 : SUCCESS in 17m 24s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/bffecf58b4da410a843b48637cc9fe24 : SUCCESS in 16m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/73518b3b6689449393aa6f1194d4e101 : SUCCESS in 29m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b38a4750f182456688982948ae384f6d : SUCCESS in 17m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a78c8c2e02b64f89a1b2f0068e1fcf52 : SUCCESS in 27m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1b13d702c70544fcafcc69018a3b989a : SUCCESS in 14m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/af643b0d312543588c38e84fdf2ab919 : SUCCESS in 18m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/36d309f12fc54b2fb03b6d91b24e38fb : SUCCESS in 18m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c2407112e24a4ac5977fee4e05b6329c : SUCCESS in 7m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/55b2c85e037f4b98a4bc17b788824eb3 : SUCCESS in 15m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/42684806f15a417eb195c09e9aded662 : SUCCESS in 7m 52s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c0529bcd12924a74867662ef5fff7461 : SUCCESS in 1h 08m 02s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9bb0ec91596c48d5a6e15a57f08fbca7 : SUCCESS in 1h 06m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a3cd5c0281444f1f87fb7159597aeb49 : SUCCESS in 55m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/18c0531b139b4432b470f3ba8c155802 : SUCCESS in 59m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5bb7bcd4dddc4bee80b7a45d935566a2 : SUCCESS in 19m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceb9a354801f44fbbd41bc6a8c2f6e9c : FAILURE in 4m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a377374c31c74a5987a3a2ecd2c6bcd4 : SUCCESS in 28m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ed1b84eb6be4443bd5f76c94a5d746c : SUCCESS in 55m 55s","accounts_in_message":[],"_revision_number":1},{"id":"d56f763329166c784e108d2d803796917e369f91","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-07-07 20:06:08.000000000","message":"Abandoned\n\nFix was merged as https://review.opendev.org/c/openstack/swift/+/887021","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"761863ada93b13a9cb1b1055942b7982fb9877c8","revisions":{"761863ada93b13a9cb1b1055942b7982fb9877c8":{"kind":"REWORK","_number":1,"created":"2023-06-21 12:37:37.000000000","uploader":{"_account_id":36116,"name":"Philippe SÉRAPHIN","email":"philippe.seraphin@infomaniak.com","username":"pseraf"},"ref":"refs/changes/33/886633/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/886633/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/886633/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/886633/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/886633/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/886633/1"}}},"commit":{"parents":[{"commit":"2343521f751388ba2e4158194fb8c2aabc28d08c","subject":"Merge \"proxy: print logs when write new shard range cache into memcache.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2343521f751388ba2e4158194fb8c2aabc28d08c"}]}],"author":{"name":"Philippe SERAPHIN","email":"philippe.seraphin@infomaniak.com","date":"2023-06-21 12:37:24.000000000","tz":120},"committer":{"name":"Philippe SERAPHIN","email":"philippe.seraphin@infomaniak.com","date":"2023-06-21 12:37:24.000000000","tz":120},"subject":"In the case where we can\u0027t stat the device, an error search in the Kernel logs must also be carried out, and the device unmounted if necessary","message":"In the case where we can\u0027t stat the device, an error search in the Kernel logs must also be carried out, and the device unmounted if necessary\n\nChange-Id: I8ed9b4ca5af90e9a64ec996725e59cba34f796a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/761863ada93b13a9cb1b1055942b7982fb9877c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/761863ada93b13a9cb1b1055942b7982fb9877c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
