)]}'
{"id":"openstack%2Fswift~886654","triplet_id":"openstack%2Fswift~master~I596415d17f77f48a6e8a63a61b734a8ca0865847","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-03-20 23:24:40.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I596415d17f77f48a6e8a63a61b734a8ca0865847","subject":"CI: Move probe tests to centos 9 stream","status":"MERGED","created":"2023-06-21 17:22:16.000000000","updated":"2024-03-21 01:48:34.000000000","submitted":"2024-03-21 01:46:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"886654","meta_rev_id":"cc9a290ca600a58ee927fb8ddb918963b6d69896","_number":886654,"virtual_id_number":886654,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-21 01:46:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2024-03-20 02:17:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-03-15 15:19:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2024-03-20 02:17:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-21 17:56:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-21 20:18:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-19 22:56:01.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2024-03-20 02:17:27.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"5e9f053a9da97c742c797f35b9ebb438690cb576","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-21 17:22:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"632b9511f8754a93501309686f952f9f098db816","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 17:56:53.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f8f79bcade343d8aee3817b5dd81f65\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/16b9d90f9c9f43d8b63577a1ae22b3de : NODE_FAILURE Node request 300-0021509804 failed in 0s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/96de8cbab2a14b7aa706fc869349c1d0 : NODE_FAILURE Node request 300-0021509805 failed in 0s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f33f9cac633426aa71b126e01512d05 : SUCCESS in 33m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2a4ee6d430754aed9102d2cd05f82ab2 : NODE_FAILURE Node request 300-0021509807 failed in 0s","accounts_in_message":[],"_revision_number":1},{"id":"8cb9c8f8cfa6dff5d6b902eba2c0f6e70b4f3975","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 20:18:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb8176d1082f40e3bbdbf5c1a971bf48\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c54cb3e3be7047c3bda9d5ad52bc3b0e : SUCCESS in 9m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/839d49a8a7a640309303b6ec5d11e929 : SUCCESS in 1h 05m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2f00e8a5c46f4e409ac9fa39040e186c : SUCCESS in 1h 00m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e98ab75b3335461a85abe5eea157e9b0 : SUCCESS in 1h 11m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c790ecd3e055479195b33dee42092f42 : SUCCESS in 24m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c01a454a6a3c4d99b0d09ecac3d67d6f : SUCCESS in 20m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ac34bc0ddfb463b99cf8ea8cdebd636 : SUCCESS in 17m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/375ec82607364bafbef06d7e184349bf : POST_FAILURE in 39m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/098a29c63ff54e93a52e06a3b65b248a : POST_FAILURE in 34m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8a770b53ab9443d38745eefeec321326 : SUCCESS in 17m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/550aa6e892cc4e2c819d662471a34f5a : SUCCESS in 17m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/487b827fdaa3492a83c1f9284e7d3098 : SUCCESS in 16m 42s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/52e936bad45d4ec588fa1d45b8a4c8c3 : SUCCESS in 16m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dbf72de3347349c5a587c0e7b7c9a7cd : SUCCESS in 18m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6c4d5082a71a4fa9b7f1542307712b6e : SUCCESS in 17m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0b55e453d2bf4d789becfc7ea2a0bad3 : SUCCESS in 18m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c2caa815c2db4ef7ad90098af31cff25 : SUCCESS in 14m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a4fb123d45f5463ca96cb357eb0360f2 : SUCCESS in 13m 42s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/660a93c997a24ec2a67901022070311d : SUCCESS in 18m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c651e002a6d84154a3e7fcb304f1ee2f : RETRY_LIMIT in 3m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c40ec59d03464701b21273f13a36034b : SUCCESS in 16m 48s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c7b5d551a9b542a7bc51f39fea3f5088 : SUCCESS in 7m 54s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/88c7dd8dab0c46bdb9c84ee62b994b8a : SUCCESS in 52m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/783da6df3b9a4a17bee6d9e60f604fc6 : RETRY_LIMIT in 3m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c2ce99b18b5d4c88aaa34fa88402e14f : SUCCESS in 57m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2a251ac2221c40f883cbc5d1ad57a07b : SUCCESS in 50m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/659c67a2978b47a7bc9d7d666aa6230b : SUCCESS in 19m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81a69b223e4242bf9570db603809b5d9 : SUCCESS in 5m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3d7f0676b0cd472f97ae87c0eaf24d54 : SUCCESS in 21m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43338347fc514db5b0ce464c80cd2bd2 : SUCCESS in 50m 12s","accounts_in_message":[],"_revision_number":1},{"id":"00d2f6eedf75b0fea797a5aae041abfd9b4e9d36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-12 22:11:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"cf05f3e8364a5d7b363ef2e972f2a28056ebe961","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 22:31:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a68a243bc59046e9999c4a392e599ae1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d93e26e4bd14d409cdad367714f4b4e : SUCCESS in 15m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/771bfe61015741ceae1dc1aa5a34a36d : SUCCESS in 16m 47s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f2085adc8ed4bf390dbd2dc599bc3c5 : SUCCESS in 19m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/be8f942aabc5487eb641e57215a379cb : SUCCESS in 18m 01s","accounts_in_message":[],"_revision_number":2},{"id":"db323f401b756f3b20e1dca6d226458fd0a1a06e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 23:24:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2da9b7e367a744e8b192e7a16e7e488c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2db722cbb65c429a891148b6cf74cbeb : SUCCESS in 8m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/b20124f45d364c41a21a9ececbe63cbd : SUCCESS in 43m 47s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9f8a0a2329c441cc844967c7c36372d8 : SUCCESS in 1h 00m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/55a504df53d84d80b7f1f39e5407e98d : SUCCESS in 36m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/038dada2a50a4730bdca30a1138879fe : SUCCESS in 1h 05m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f360ba5f0cc8451fb99159f7197f59c5 : SUCCESS in 21m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d54cedda9cc4f7aabc8a6727da16a89 : SUCCESS in 20m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/073a78da45f24f849af28850b48e702a : SUCCESS in 22m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/23d643604ec947098330d14379428235 : SUCCESS in 11m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/14028c793b72471eb138e889397a5899 : SUCCESS in 4m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bcb7eef5a196473fbbf594431e8f36e4 : SUCCESS in 20m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/569e6734d92f415192e2e010024f4261 : SUCCESS in 18m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f27b6b9e30b5401ea331c0e3c99d4963 : SUCCESS in 20m 35s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3d61936cc95c491b884009f032dbe6e8 : SUCCESS in 18m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0ed127afeaf44cb79528eca9d8218779 : SUCCESS in 21m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4b9552a89eba4ea8a7ad1c5d742fa77c : SUCCESS in 22m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c75e1b4199394109a3f4be21426d52c7 : SUCCESS in 22m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0160ca4eac9146519f0f8b009ba9d73a : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a6f4ad44a17c4ec787c8add218478d8e : SUCCESS in 19m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a09431cc826441678fd7d8d111274664 : SUCCESS in 19m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1f22c5af76f94633bd3f6678d07b3841 : RETRY_LIMIT in 3m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ce1e8d5f9c1c48079c1d3f49470d2d68 : SUCCESS in 16m 24s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8f4a8ae709dc4cbdb3344c1916d782f7 : SUCCESS in 5m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/93294f46cd894b5ab5f35e5c28803ce1 : RETRY_LIMIT in 3m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3ca92b3743d04f15a52498bc60f38b14 : SUCCESS in 51m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8e6f4c518e6944f6954074b366c9b7e3 : SUCCESS in 52m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4214230152a41d5afdb3ca8eb5d84be : SUCCESS in 14m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcdfa4383bab4f1a95f6b34b5871f890 : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6a33a29a25f94654bd481c00f241187b : SUCCESS in 23m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/19b6434744c74d6bab810b89e3b78340 : SUCCESS in 58m 57s","accounts_in_message":[],"_revision_number":2},{"id":"dc13117d2e5742eef43cf521dcaeb2462d5b8293","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-14 20:58:49.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a7a8ec821d980b3522e8925cd9f417f1429f4261","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-14 21:18:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c0e0267555005c79f689696ab51be486fecfc65f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 21:41:58.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58591c2dc24e4965a4a3fe589291b94f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c493e9b2667a4aca80a6eff80c2ca040 : SUCCESS in 14m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7c76333f335947898043c8da23ebfc09 : SUCCESS in 15m 15s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f7137d72618c47d3a582dc890a611703 : SUCCESS in 22m 27s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/32c1de0012f44db4beea5115bca113c9 : SUCCESS in 18m 06s","accounts_in_message":[],"_revision_number":4},{"id":"0e60dba31f3cb587c8e1fb6a36b1d55efa3d5c17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 22:29:00.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee2abd327b414998aec516509710c58f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/793278f484b94cd8813661115dab5927 : SUCCESS in 11m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/c1dd4ff665af428dbb25532470be13cb : SUCCESS in 56m 25s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f9f32a8989c647208fb3eb3a066cece4 : SUCCESS in 56m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/92d1687e152b4751842d0e24be4bb2d2 : SUCCESS in 56m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8559c719e57449ebc1b88873616e035 : SUCCESS in 59m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f406f7cb0c6f422db93b2c8805f64022 : SUCCESS in 17m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/56bed26315b846f9b2666d3b25e7d179 : SUCCESS in 19m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9a785e6e97af4d87837a2e1136315521 : SUCCESS in 19m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e1373e9ac07547cd827787b68f16dfad : SUCCESS in 9m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9094ade536af4a3abbc531882390d4aa : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/737d25cb93594a6d9d059169d071ab8f : SUCCESS in 17m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/34081c58312843688781456e40efcbc0 : SUCCESS in 17m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c820d559fb7245fd8bad44a8391e4aa3 : SUCCESS in 18m 21s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/70ce32cc84a54e6aba1eb3635b3ed530 : SUCCESS in 16m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/479f914d881547c684d88e2879450bea : SUCCESS in 17m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fdc756134f1945109fbe80e15481027b : SUCCESS in 19m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a65de7fbef2c4af7b43641101b3abfde : SUCCESS in 19m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e989441a721f4fb08d3615a2cd5abb36 : SUCCESS in 14m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3a029364f5ca42ab813ea8389476b5aa : SUCCESS in 16m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/88daf7c6735b4ccbb699cac900358f43 : SUCCESS in 13m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a0797d1ef88d46deaa6c82deaf51d8fe : FAILURE in 8m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/35fd200d646548edbb713340a5961152 : SUCCESS in 12m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/aaa132184574431c997e4251cd3a361f : SUCCESS in 7m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3473be9f850348d194c4dc9d959aa06d : SUCCESS in 48m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6555d891d9e7453ab3d648d66ab834cb : SUCCESS in 35m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/24833e41a46640ac8aeadc45fcfdd2fe : SUCCESS in 36m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d8fa2756c5846a2828a39d6511a2e90 : SUCCESS in 16m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ac84f827921458b989740525c45edef : SUCCESS in 5m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/61d2e8f11aaf491094d88d3eba75f7e3 : SUCCESS in 25m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c30e71d6c564e78897de5596d0cbdea : SUCCESS in 59m 34s","accounts_in_message":[],"_revision_number":4},{"id":"e8e2a1f4477544ff7f880709562824ec25b9dd8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-15 02:14:18.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"063bcf82f31d29c32305e35b1b436928c2396a58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-15 03:28:59.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"3adc9dcc1280dd9e6c9227ba84984ec314e85462","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-15 03:46:01.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"7703a108292aeaf0831d5528a1497df1fbbe169a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 05:01:33.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e20173835d4d4667ba8b90a481096220\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0c25aafdc614016945f14fbd126fcf2 : SUCCESS in 8m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/414ffa19609d414c852f58a8ace5fddc : SUCCESS in 1h 04m 38s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f2b0c97aa2b346be836ed4b2a647e355 : SUCCESS in 38m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fc1c3a12c35f4139b8d2a77cae99a4d7 : SUCCESS in 38m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dfe1dc09060c4fa0942094bd3df7a02e : SUCCESS in 1h 04m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2285a293c6114564a32a19aee2286e18 : SUCCESS in 21m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/867867c35b7a4431ae53481c1f74eff4 : SUCCESS in 21m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f4cf252fc96a40f596a53bcc4f07820f : SUCCESS in 23m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/df1d406c8c31430d9faa35d58be5ba61 : SUCCESS in 9m 43s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a19da9c989be4343b70603db9a2ead2e : SUCCESS in 6m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cd67a27506bb4ecdaf10ff7924f818ea : SUCCESS in 18m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cdc64e16204945c28180513fc0dbe97a : SUCCESS in 18m 17s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c87414698b4647b59a894dd742dc5975 : SUCCESS in 14m 46s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/be071911f8c749b597e8e7292095e5e7 : SUCCESS in 18m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ad044e5604de40e39acb7d7d7f69ada7 : SUCCESS in 17m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4623164d15b14dd285fa18ce8a2d614e : SUCCESS in 20m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7d592f8f37ba4aaea1961f689f4b0005 : SUCCESS in 17m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1225ac062af04966a2075fbca39bd036 : SUCCESS in 16m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b5db933e8f804ca2936e49ef44ec28d2 : SUCCESS in 16m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ed268cd535ef48b28f42bf1ee8137d09 : SUCCESS in 16m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a883335253e94652882dd27a7957a8af : FAILURE in 6m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/168c7e6e42874cff95f98fc78b1a45ed : SUCCESS in 16m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2e776eb91f0c4b86b7a13c2ae49b9339 : SUCCESS in 6m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/dfe3943afc2646f9a29bdec1e67b7f38 : FAILURE in 58m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8204020f4ccb4c10b8e8d0384a907a20 : SUCCESS in 51m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/225dcee308874694873300a15480c546 : SUCCESS in 43m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a13337c9296e4ceda34f55083129a443 : SUCCESS in 21m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/142d1127cb224573b07d64c80c423836 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7583e1342cb24f349325c848236b6918 : SUCCESS in 19m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f43db091e64d4781beb5f1e7e207a992 : SUCCESS in 41m 35s","accounts_in_message":[],"_revision_number":7},{"id":"0f30bf0a784d5ff44f8e1a80e9badf60d7f900aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 05:19:10.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86ef209f081a40a8bba034c585c6d36f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/40f2506182934540964420ed9ce8967f : TIMED_OUT in 1h 30m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/99199430aac349aa9a5459554b4de447 : SUCCESS in 1h 23m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e6096b50f11641d8963fd66839c0b3b9 : SUCCESS in 23m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7257977467034359a1488c19a60d2629 : SUCCESS in 1h 23m 31s","accounts_in_message":[],"_revision_number":7},{"id":"51abc1cb2696d75b5e7320a28a73b18ed6c5a6d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-15 15:10:17.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"694c4d64ad0eb7e814093f4473d051542c2e4f6a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-15 15:19:37.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"21a0ec0f758d17ec95434aef3ff1d4f03b20bb16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 16:09:11.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b447cafafe34acb804d2c582e233c79\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64cb2c1b2c1d4e169ae9ea69d2a63bcf : SUCCESS in 9m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/23a311354f94480aad8c6b55433348e1 : SUCCESS in 52m 18s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9af4d39c4d5b4ceab50c13c1ec88223f : SUCCESS in 51m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/de5e31dc91e643d89e2726116c1daf88 : SUCCESS in 45m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0daf536578f641b0b0565b4b6b32b3a0 : SUCCESS in 50m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2fb5eb4689f64580a1243911c1b448eb : SUCCESS in 21m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8c1d00a004044c3aa3f5cbf8fb007a1d : SUCCESS in 27m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/55efee95fdef4c60a68751063559763c : SUCCESS in 22m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c97b7e27fea1482191540210a2ec7e39 : SUCCESS in 9m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9600db4ebb7140ebb12cf6d573e22b3b : SUCCESS in 5m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b6ce6aa062d54103ac376495a63f408d : SUCCESS in 20m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/132d4c92aefa49e884768fbbe87239c1 : SUCCESS in 19m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3c0cc96dde8345aba287d59a565c4c15 : SUCCESS in 16m 34s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ebbe362470c44ad2bba8df0616bbcb8c : SUCCESS in 20m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e39df2e70d29412da01c2b4951aa5c36 : SUCCESS in 19m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b6008c4fc10e4c1692f45ccdda6551ee : SUCCESS in 18m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/854b0027e6004c97960b3baa2113d966 : SUCCESS in 19m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/882ddaa711f748cea70aac55e1cb7f53 : SUCCESS in 16m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a45a00e743164eaf8f825305f03c222b : SUCCESS in 18m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2df6e3ea3a2f4c26b4a4b0dc146abb9a : SUCCESS in 17m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6c583fc63b364063ae18f8d423ba9877 : SUCCESS in 6m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/81dc16abd70246d2a06c7899c0f4df1e : SUCCESS in 14m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/20273d409b6a4dba9728622804e443ad : SUCCESS in 6m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/17f4312b8a3446c0af42a9a7e1d7dc0c : SUCCESS in 53m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/52e4cf07c13649fbaa737a032d372673 : SUCCESS in 47m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/16ae3c1559604ae59c63b3ead0dd3878 : SUCCESS in 48m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c86a8d44d16496c96d8ce851d1be580 : SUCCESS in 19m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88aac17ac76e47948955143d20857233 : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/da68ba1159b0420aab802cdd7485a90a : SUCCESS in 20m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/28e6c79a727d448d9ca962f26b02831d : FAILURE in 45m 46s","accounts_in_message":[],"_revision_number":8},{"id":"48f786f0e989cb7c9050589ebacce43455faf216","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 16:14:57.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/329bb2e95d504c32b964d4e710f36ede\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4f659204076460390a0d295777d7a27 : SUCCESS in 19m 22s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/92ebc468827944ca8cc544d7a78f7cf2 : SUCCESS in 1h 01m 43s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a9e72b8c7b564779bed57715f8e0b53b : SUCCESS in 22m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/05b2fb93c2314a7db1b507e94e60e2ca : SUCCESS in 22m 34s","accounts_in_message":[],"_revision_number":8},{"id":"25c9f5d71648708bd37024e306909bff87379bc7","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-03-19 22:56:01.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"93df08183efe3c0ac552f670fd075c39eeb89dc5","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-03-19 22:56:09.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7dee97e452f97929744dabb5a8f39046eeb43a62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 00:07:14.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d56a87107df64c8784f5b16f6c80ecac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2cb3c2b671349be9f5d89f0a2748f42 : SUCCESS in 8m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/7c0415ad92b745429ff056a72c1cba02 : SUCCESS in 1h 05m 30s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/fecdc0946f37407ea97814f180ffd276 : SUCCESS in 1h 00m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a3cb14742d114c3c8a395f9a3cdedf7a : SUCCESS in 49m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5214fc70a944a80920867cd773c67fe : SUCCESS in 53m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cd978249d632486d926d8a4e62e1b1fe : SUCCESS in 17m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/681d1144b5234be885765befafc07556 : SUCCESS in 20m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d08549cb9c624024842384d5f92eedd3 : SUCCESS in 21m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ded7d35ec86743aeac05abcaafeeb0eb : SUCCESS in 9m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/539f4909e98f4510aa2c7c3a6d9deafc : SUCCESS in 6m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a5caeee27502476a879591cc8d9a941f : SUCCESS in 15m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5fedc6da2111467caaf8fa22512fe95c : SUCCESS in 20m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/01d8053f02b74f8d9e33aaca5512ce56 : SUCCESS in 16m 17s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9708805367c742b0aeded3a9681e6593 : SUCCESS in 17m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f21bbe6b19d24847abd5538c8488f63a : SUCCESS in 21m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7e21688d119344f2a75ab83440ae82d7 : SUCCESS in 16m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3c81aa2c4257444cafe4bd6b04bf0fb1 : SUCCESS in 18m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ccbdfcc059ce45d0a0cc06e248cc9509 : SUCCESS in 20m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ad30ed4c7af8477b9402579c0a9bfa69 : SUCCESS in 17m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f35e598b9878483aba4d82a12e75f41b : SUCCESS in 18m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fafc6659dc5a42e49e6cfde341f02878 : SUCCESS in 6m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0fe6d0319b29434a9ecd401c340a61b4 : SUCCESS in 16m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a8042c2d0e904222a1c94c5743293059 : SUCCESS in 6m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/63bb1425f37e440c8d3bda2806191396 : SUCCESS in 1h 03m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/79d7f249ea7c4e8e8c5e23cc9071750d : SUCCESS in 55m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/48e318f9815848e28981f61f65b75c72 : SUCCESS in 47m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54b38dd2077f402ba3bec829729bc9cc : SUCCESS in 19m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a845dc12106846c2b9c2c4fd6a51c137 : SUCCESS in 4m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0ca545a9c10a4a3f961598186710600e : SUCCESS in 19m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cdb892e710da4f3086262cbe65975d70 : SUCCESS in 59m 10s","accounts_in_message":[],"_revision_number":8},{"id":"a90d523846c60b46607e427d97f35968b383be58","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 00:19:40.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2b17c3530d2462089c2432111740fd3\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a1f9600cbe2d4bdfad981df8eb6b0437 : SUCCESS in 14m 46s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5dafb48f1f524243801317780d90f3ff : SUCCESS in 1h 04m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/30f1aa6f7a434775b8b5d9006ba2690d : SUCCESS in 24m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/313e562f23154e749f7348be394880af : SUCCESS in 1h 21m 48s","accounts_in_message":[],"_revision_number":8},{"id":"efae712c2d8f585cacb3b31e74933b555263f339","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-03-20 02:17:27.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8960ed3195546618444de0fbef027e04b8214bf9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 02:18:31.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"97a27b202df567a4868accfd0c05e2d8c0ab69b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 03:52:44.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/12a61055f73543a9b843929471efdeef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce73cc82fcd14680b8594c2f61204674 : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/23c12b3b420848788ea3d80b49dc8d69 : SUCCESS in 56m 21s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7784ffe772a447b1854ecd989f1cac4a : SUCCESS in 55m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c6623ce5c4484f6db07a7ed86b5ea44d : SUCCESS in 42m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/23137c42ebcf4c1b93d2541e02cc9505 : SUCCESS in 55m 09s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e8a7a01255634ab9bca1acec041c358f : SUCCESS in 12m 08s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/66680af5144941aaba2acd78ce2e91d2 : SUCCESS in 6m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/98233efe29e649f7b7541f30d0c65fda : SUCCESS in 18m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e2f67bab3976477581c19db1c8aae3c2 : SUCCESS in 19m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ad76bcd228a84e56a2888f1d1ef549af : SUCCESS in 18m 21s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8abd85f8bc804e13979a9401abcfe00b : SUCCESS in 17m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f1059fdb0d56458c84d424186d2f1538 : SUCCESS in 17m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7eb2119148f4422d9013a768e5a406bc : SUCCESS in 18m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aaf6704546504a07bfcc4cec6b3cff32 : SUCCESS in 18m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/26dc5fdf4e3044eaa7c1ee73bd268da7 : SUCCESS in 16m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/da2bea0cf4d34b12a5313ac1e849028c : SUCCESS in 16m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cf01a7654f634393a9b84b8485f633a3 : SUCCESS in 15m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5ccd89a46093442dbb598e848461d478 : SUCCESS in 8m 09s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ab8fa29f38ef45f6b2091d9567732be5 : SUCCESS in 9m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d47d465f56ec47b0b5692aeb602428a8 : SUCCESS in 1h 02m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/83f9a54b5ac84b8fa33d8dfe2b289e34 : SUCCESS in 52m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f79ab8cbf6be4f7ab8a13d4065d5a0c5 : SUCCESS in 57m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1969e41a38c449893ca5a9fddf22299 : SUCCESS in 19m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/910033687dec47728ff742affecdd2e8 : SUCCESS in 4m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70ecb2eaf19a47b19397c19315cf6f9a : FAILURE in 57m 16s","accounts_in_message":[],"_revision_number":8},{"id":"52a0bfd8fb19029450f3f8c0287463b960bba8db","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-20 23:24:40.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3e946a4833d85598868a3f52d3bf27f2f9fa3eaa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 00:20:17.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23e3ffeaea7a43f2a698e124c10f151d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/44b8d6f6af3743bfa72796c99f497555 : SUCCESS in 16m 29s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/488c23ff06784bf680018c2501442bc8 : SUCCESS in 16m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c2c58cd37e4049059e869166257644f6 : SUCCESS in 22m 24s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/47771731016c4432b315e40467c488df : SUCCESS in 53m 41s","accounts_in_message":[],"_revision_number":8},{"id":"dc357004ff6b56d61fc41821b42f004eb1dce34e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 00:35:59.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fb37b7fa1fe4eb9a26f8ebda222f73e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a3abe0446d44bd6bef9405a4e9b4037 : SUCCESS in 6m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/03f90542540b4d84b6614d03d75a392f : SUCCESS in 42m 02s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3522e1b3ee424cb6af988b518a9d84b6 : SUCCESS in 40m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/754ed67b191d4aa3b75066b3df19db39 : SUCCESS in 51m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5615f12d1389467ab0c1296507a00975 : SUCCESS in 56m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/523e2cf4112343b6a9d79ff23b2517b9 : SUCCESS in 18m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aa759206df544dfe8cbe485feeb6da7c : SUCCESS in 15m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6ea858f4876c45a5b9d4e3325941a1c0 : SUCCESS in 20m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/54c753c778e84bf1bf32d94716e4a87d : RETRY_LIMIT in 2m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/955b002867d54d7486cb3cf9661cf9ad : RETRY_LIMIT in 3m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/945fcffddceb44cf8a19789238a38b65 : SUCCESS in 21m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/da2f8642f4584f56a220e1cde50976f9 : SUCCESS in 17m 22s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ed2d0cc919d44e20a8137657cfdeabf9 : SUCCESS in 18m 45s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/eaf05eb6d41c4ed781d950341a741408 : SUCCESS in 18m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3dd2be7a087d41b4a6df86d75a28a206 : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9c4a10e7ceda40e7b2079e5c6516b1e3 : SUCCESS in 19m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/053caa99c0714b63919abe2a68be17a3 : SUCCESS in 15m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5d7467e6cd78485484cffe8bfdaf54e0 : SUCCESS in 16m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2d024bea50a4486b85f478e2bf8df808 : SUCCESS in 15m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5c9ffae47c3d4557bf7fe1c31cc94745 : SUCCESS in 16m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0c5a05627e0a403ba9254e8e65217bd6 : SUCCESS in 13m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7ff300dc352f4d2a95e2407be62ce387 : SUCCESS in 14m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/74b67c65d5a54f7ab8151b14bb0fc7ce : SUCCESS in 7m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c2527c51cce44fd806278d56a8e38d1 : SUCCESS in 1h 02m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c0fabeb7d01b4756a2df559b56d00108 : SUCCESS in 50m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/44048f8ca59146da86df89038c583089 : SUCCESS in 49m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93e7c5623eb540af8f29c034d779fe32 : SUCCESS in 20m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef813f6cd1da48e5ab8d4e11cdf67f8b : SUCCESS in 3m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/62962243d7d445ffb334848945fa6d08 : SUCCESS in 23m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/99bb166f40134c9095e7c7969e495197 : SUCCESS in 43m 10s","accounts_in_message":[],"_revision_number":8},{"id":"34bce57b5c1ea6d03990a3fa2489a8e98ec9df1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 00:36:38.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"fe3e5411d6d5896c41be84f9702f4e9501814dd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 01:46:32.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2970e8edd8dd4c09b32ead055f4fca61\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aae0dfc7c93d4c02b913a4e4b1f7630d : SUCCESS in 9m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/43e2ff17d2214be09e55b60468e26682 : SUCCESS in 1h 04m 04s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8a53ed7e8ff94c20b8f519b9e937d038 : SUCCESS in 1h 04m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b00a8ba456f44f94893c5c32903fe1f6 : SUCCESS in 48m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2aa1e2f61c914ed8852635fed4ea6b4a : SUCCESS in 40m 43s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/7003d540e6d34a819c49e7b3b13c5789 : RETRY_LIMIT in 3m 35s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/d61dc862d4334706817b4e413aa6b147 : RETRY_LIMIT in 3m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e4ef8fc472d54351ae21d62270f398de : SUCCESS in 19m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/26e3d0cd9b05412f8256798aa8950501 : SUCCESS in 16m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/59e18204bb334f3a900b55a77d4ae794 : SUCCESS in 17m 19s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/982562d10a0749ba8666b32944da1c15 : SUCCESS in 17m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3e32790a6c6e4e8c9187d26bcd19419b : SUCCESS in 16m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/19f1bb20a3a149739907cfe652c28a76 : SUCCESS in 18m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/166301b36fba4d25afc6273ce38bde96 : SUCCESS in 18m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/83e800998be345f095c57d9a6b7c76dc : SUCCESS in 14m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/350c7a7e48d04dd8b2a655d197afd357 : SUCCESS in 16m 07s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f3179b118d944a9d8e92c6c903e28740 : SUCCESS in 16m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8488703c0d154cec92f91cbcf1da7ee2 : SUCCESS in 5m 09s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/20868b24c5ae4ba584c5fa59ffb1b94e : SUCCESS in 6m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae6f0c6b241d4d55871b923716c593eb : SUCCESS in 52m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/53becae979464389b782af29a1bb50d1 : SUCCESS in 37m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d55daa069ac5457291238508adca35aa : SUCCESS in 39m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a8a05514e9144b2aeabbbe4f7bcc210 : SUCCESS in 19m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1b826386f644fb3ae2ea00497be811a : SUCCESS in 3m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d1e6226a3634762b1f5f756b6a4d94c : SUCCESS in 37m 09s","accounts_in_message":[],"_revision_number":8},{"id":"bec699fd38ea0dee20a4fbf8a78091aad7ff67db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 01:46:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"cc9a290ca600a58ee927fb8ddb918963b6d69896","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 01:48:34.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe70a743c6ae4ebd8e9773556b440c8b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a2fe8faa807425faee48521de29b405 : SUCCESS in 48s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f1986daf24f9408ca876fd9762ace7d1 : SUCCESS in 35s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/ad9ec21acfb1405da80a878c88c1d561 : SUCCESS in 32s (non-voting)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"1c03d586431862c2772cb1b532aefc6cf2e671dc","revisions":{"312839d13f75f507c605ce65d4df8bf5cfcc6111":{"kind":"REWORK","_number":1,"created":"2023-06-21 17:22:16.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/54/886654/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/886654/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/886654/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/886654/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/886654/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/886654/1"}}},"commit":{"parents":[{"commit":"2343521f751388ba2e4158194fb8c2aabc28d08c","subject":"Merge \"proxy: print logs when write new shard range cache into memcache.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2343521f751388ba2e4158194fb8c2aabc28d08c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-21 17:21:32.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-21 17:21:32.000000000","tz":-420},"subject":"CI: Move py3 probe tests to centos 9 stream","message":"CI: Move py3 probe tests to centos 9 stream\n\nChange-Id: I596415d17f77f48a6e8a63a61b734a8ca0865847\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/312839d13f75f507c605ce65d4df8bf5cfcc6111"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/312839d13f75f507c605ce65d4df8bf5cfcc6111"}]},"branch":"refs/heads/master"},"8e3a122b00cb2c6381462a618082b84d0aa4fc13":{"kind":"REWORK","_number":2,"created":"2024-03-12 22:11:06.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/54/886654/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/886654/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/886654/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/886654/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/886654/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/886654/2"}}},"commit":{"parents":[{"commit":"5f95b6b26db1c86dd9c5276c4124dd80041df75f","subject":"AUTHORS/CHANGELOG for 2.33.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f95b6b26db1c86dd9c5276c4124dd80041df75f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-21 17:21:32.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-12 22:10:19.000000000","tz":-420},"subject":"CI: Move probe tests to centos 9 stream","message":"CI: Move probe tests to centos 9 stream\n\nChange-Id: I596415d17f77f48a6e8a63a61b734a8ca0865847\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e3a122b00cb2c6381462a618082b84d0aa4fc13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e3a122b00cb2c6381462a618082b84d0aa4fc13"}]},"branch":"refs/heads/master"},"e40a0bd90337e7721f312356567afd5884ba8da5":{"kind":"REWORK","_number":3,"created":"2024-03-14 20:58:49.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/54/886654/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/886654/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/886654/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/886654/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/886654/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/886654/3"}}},"commit":{"parents":[{"commit":"5f95b6b26db1c86dd9c5276c4124dd80041df75f","subject":"AUTHORS/CHANGELOG for 2.33.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f95b6b26db1c86dd9c5276c4124dd80041df75f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-21 17:21:32.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-14 20:58:42.000000000","tz":-420},"subject":"CI: Move probe tests to centos 9 stream","message":"CI: Move probe tests to centos 9 stream\n\nChange-Id: I596415d17f77f48a6e8a63a61b734a8ca0865847\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e40a0bd90337e7721f312356567afd5884ba8da5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e40a0bd90337e7721f312356567afd5884ba8da5"}]},"branch":"refs/heads/master"},"f2b0a53ccef26b0acbd47b8dfd8112ec0f14f2ef":{"kind":"REWORK","_number":4,"created":"2024-03-14 21:18:24.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/54/886654/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/886654/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/886654/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/886654/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/886654/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/886654/4"}}},"commit":{"parents":[{"commit":"5f95b6b26db1c86dd9c5276c4124dd80041df75f","subject":"AUTHORS/CHANGELOG for 2.33.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f95b6b26db1c86dd9c5276c4124dd80041df75f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-21 17:21:32.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-14 21:18:16.000000000","tz":-420},"subject":"CI: Move probe tests to centos 9 stream","message":"CI: Move probe tests to centos 9 stream\n\nChange-Id: I596415d17f77f48a6e8a63a61b734a8ca0865847\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2b0a53ccef26b0acbd47b8dfd8112ec0f14f2ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2b0a53ccef26b0acbd47b8dfd8112ec0f14f2ef"}]},"branch":"refs/heads/master"},"6f5c6afc75479dfc68fde56f32e4eb8a63632e01":{"kind":"REWORK","_number":5,"created":"2024-03-15 02:14:18.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/54/886654/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/886654/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/886654/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/886654/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/886654/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/886654/5"}}},"commit":{"parents":[{"commit":"b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7","subject":"Merge \"s3api test for zero byte mpu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-21 17:21:32.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-15 02:14:10.000000000","tz":-420},"subject":"CI: Move probe tests to centos 9 stream","message":"CI: Move probe tests to centos 9 stream\n\nUpdate geckodriver, as firefox has been updated to 115+\n\nChange-Id: I596415d17f77f48a6e8a63a61b734a8ca0865847\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f5c6afc75479dfc68fde56f32e4eb8a63632e01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f5c6afc75479dfc68fde56f32e4eb8a63632e01"}]},"branch":"refs/heads/master"},"d476dc7b16eefc99c737c731869b4c8515eac7e1":{"kind":"REWORK","_number":6,"created":"2024-03-15 03:28:59.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/54/886654/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/886654/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/886654/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/886654/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/886654/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/886654/6"}}},"commit":{"parents":[{"commit":"d1696f55c2791692a23389e1b3e028e8141482b8","subject":"tests: Update CORS geckodriver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1696f55c2791692a23389e1b3e028e8141482b8"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-21 17:21:32.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-15 03:27:55.000000000","tz":-420},"subject":"CI: Move probe tests to centos 9 stream","message":"CI: Move probe tests to centos 9 stream\n\nUpdate geckodriver, as firefox has been updated to 115+\n\nChange-Id: I596415d17f77f48a6e8a63a61b734a8ca0865847\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d476dc7b16eefc99c737c731869b4c8515eac7e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d476dc7b16eefc99c737c731869b4c8515eac7e1"}]},"branch":"refs/heads/master"},"78f45517318aaece86f84cb07d8089b8c7fc20bf":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-03-15 03:46:01.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/54/886654/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/886654/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/886654/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/886654/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/886654/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/886654/7"}}},"commit":{"parents":[{"commit":"af15ad53fb7818cdf14c92523fd6799bfabc12ec","subject":"tests: Update CORS geckodriver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af15ad53fb7818cdf14c92523fd6799bfabc12ec"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-21 17:21:32.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-15 03:45:47.000000000","tz":-420},"subject":"CI: Move probe tests to centos 9 stream","message":"CI: Move probe tests to centos 9 stream\n\nUpdate geckodriver, as firefox has been updated to 115+\n\nChange-Id: I596415d17f77f48a6e8a63a61b734a8ca0865847\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78f45517318aaece86f84cb07d8089b8c7fc20bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78f45517318aaece86f84cb07d8089b8c7fc20bf"}]},"branch":"refs/heads/master"},"1c03d586431862c2772cb1b532aefc6cf2e671dc":{"kind":"REWORK","_number":8,"created":"2024-03-15 15:10:17.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/54/886654/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/886654/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/886654/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/886654/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/886654/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/886654/8"}}},"commit":{"parents":[{"commit":"af15ad53fb7818cdf14c92523fd6799bfabc12ec","subject":"tests: Update CORS geckodriver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af15ad53fb7818cdf14c92523fd6799bfabc12ec"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-21 17:21:32.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-15 15:09:25.000000000","tz":-420},"subject":"CI: Move probe tests to centos 9 stream","message":"CI: Move probe tests to centos 9 stream\n\nPin selenium to 3.x for now, until we can run down the issues with 4.x\n\nChange-Id: I596415d17f77f48a6e8a63a61b734a8ca0865847\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c03d586431862c2772cb1b532aefc6cf2e671dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c03d586431862c2772cb1b532aefc6cf2e671dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
