)]}'
{"id":"openstack%2Fswift~886694","triplet_id":"openstack%2Fswift~master~I2a5f3e3d841d52d013b481118f56f7a1f6ab3f98","project":"openstack/swift","branch":"master","topic":"zanata/translations","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-06-28 21:26:07.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I2a5f3e3d841d52d013b481118f56f7a1f6ab3f98","subject":"Imported Translations from Zanata","status":"MERGED","created":"2023-06-22 03:37:36.000000000","updated":"2023-06-29 01:28:04.000000000","submitted":"2023-06-29 01:27:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886694-zanata/translations","meta_rev_id":"3159f75dc9ed7e355a1eb20288a2ce60a9a10053","_number":886694,"virtual_id_number":886694,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-29 01:27:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-28 21:26:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-28 21:26:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-22 04:32:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-22 05:10:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-23 18:07:06.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"b4af4ae1a57a46755d34c7da787aa39e45241169","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-06-22 03:37:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8885ffac49247031560ae74043a7c0cba0b7ad85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 04:32:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9a4d8d93eab4932932d0a7e5685f498\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fcc82e46fa094d90a143dd9a409f6ae1 : SUCCESS in 14m 44s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/19a6e491d3144c7e86257abe4cc94872 : SUCCESS in 17m 24s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3513e3055de04f5d8e7d7359cfaefc69 : SUCCESS in 52m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d42964f8a0f343db8151cef967e08c17 : SUCCESS in 19m 10s","accounts_in_message":[],"_revision_number":1},{"id":"2749f1032f0903e74f181fe8aa4a68b0a5a2aa25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 05:10:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b1dfa48c8594077958bcce262c24fb7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e1438bbc18444c0a9ffdcbb4123e01e : SUCCESS in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba6d4294ecd644379feaf8557fd4eaac : SUCCESS in 5m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/8766153fe15d49409541193b703f3be9 : SUCCESS in 46m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9cbccc39c0114d638f677e674b6b3bad : SUCCESS in 45m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/153effd06ef94daa8568d7cb648c6568 : SUCCESS in 1h 22m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/73f06f354130441fb7abf3490f10d881 : SUCCESS in 30m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9647346ab8a14a3d89627143aaaafd0b : SUCCESS in 29m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e6f3f232e01444d8aa86556f38745d33 : SUCCESS in 17m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f9c0c02cacea4fbaa602fa0db34a90f8 : POST_FAILURE in 38m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c731d4290cb047db9124b3f8438b2e9d : POST_FAILURE in 36m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/49200c1e67cb49c09c0d5df88768c33b : SUCCESS in 13m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/77a0eba701a04818a0d456e7851ca1fd : SUCCESS in 16m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7fa41971e3604bfb812e1ca019b06977 : SUCCESS in 17m 30s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ede589d617184703a68a399048126045 : SUCCESS in 24m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3dfd3e7b00db4977bb0b30ab9853b22a : SUCCESS in 19m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1567f1282db74c748c774d562bfa4c34 : SUCCESS in 35m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6b56eb73d79f4a1e956f7deac5826d8e : SUCCESS in 20m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2c8fc185407f484a84cbba3ff8900a69 : SUCCESS in 13m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c8e09ac4a4e34294ba00b2a76af54f76 : SUCCESS in 15m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9e6c4302d11942e5bfba6dc57631c371 : SUCCESS in 20m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ab6a6de6588049bfbe99807255e94b4d : SUCCESS in 8m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a6deffbf0f144240bba93a37e7aaf100 : SUCCESS in 19m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cb9caa52d50948a393775c98bf8defab : SUCCESS in 11m 29s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/df2224fa82cf49cfa495fe9ebb6e9d17 : SUCCESS in 58m 45s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a3f78f33808f491381199ed4738051e3 : SUCCESS in 54m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d039356fb704490983ce8f315140505e : SUCCESS in 53m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/42b8772e1aa54f80a01ec51f677bd1c3 : SUCCESS in 54m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53560b6bc5734519b52b97e65e63b3b3 : SUCCESS in 25m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3568b7d3e86b49a5bf9231d09f765d90 : SUCCESS in 5m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/02fda9157dea481b9bfa98862c6588bb : SUCCESS in 19m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/24dbaf5cda684f20b2358e09f1353c47 : FAILURE in 42m 32s","accounts_in_message":[],"_revision_number":1},{"id":"ea3447a3aa57b1e624c9e9c834c2022b13eb78e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-06-23 03:14:08.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a5185c655bdb91a4f3d122d1ec1ed080803df9d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 03:38:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86c2e34464f542b791785b3d9128caca\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/104f5070871b4578bef57d0f2a7a3da1 : SUCCESS in 16m 20s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/92eddce53df749b7b4d011fe0e3d07f7 : SUCCESS in 17m 19s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/49643a2bb729417d87abd3ccbcc00341 : SUCCESS in 20m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fe191e30d5314dd299fcb5939fd6f9bc : SUCCESS in 21m 08s","accounts_in_message":[],"_revision_number":2},{"id":"28c8cd9475122a55018ee496139705cbf3c73736","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 05:00:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/43e736828e9749ac9717f8dbf561c00f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2be8c281eff74c908000442dfdb54c71 : SUCCESS in 9m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23d319b42f14431ca1760f302e202815 : SUCCESS in 8m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/83e5cd9d1d7a4953b114b374fa6bae34 : SUCCESS in 1h 02m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/93619a8d34a4418a9fea12ced47cb114 : SUCCESS in 1h 12m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ffd88ea9c6dc41d9b24cd37659403ce0 : SUCCESS in 1h 04m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9b02636be27348fdbf832a9f417f55d7 : SUCCESS in 19m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/51387be1cc88429183995d4f4520b8bf : SUCCESS in 26m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b2dac1d5238c4b4bbf59e637945d6af1 : SUCCESS in 19m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/52ce105433d348fd8b7efef55cce35ed : POST_FAILURE in 37m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6733bbd7aa6149b1a642a24123b0982f : POST_FAILURE in 34m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/febc3cef21be4ad3b61a3cc06ffa2e6d : SUCCESS in 19m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d6860e36fa54412d8661246b22f3bfc8 : SUCCESS in 14m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d8daee5466234236a24c3fd86e38d108 : SUCCESS in 25m 15s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e83b714ba55c480da76200be3847f715 : SUCCESS in 17m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5272b106fa0746e69bdf63f6c911d418 : SUCCESS in 26m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5c40f9e7156d491893c9d2517bf7346a : SUCCESS in 25m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1ef1ba247f7545a8aa2218e25819437e : SUCCESS in 28m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/adc197bbda4441feb749c3d3a9eacde9 : SUCCESS in 13m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/69b9eee58b7d4b6397132ccaf1a7eaab : SUCCESS in 26m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e5a37a0b848d401d887f1782532162da : SUCCESS in 15m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/052f6eb2cc9f43acabfaec298326e95b : SUCCESS in 6m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a6c9e2d9e10c44eeb62848e409186845 : SUCCESS in 17m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1a0d69583a9a4d58a7db8165acd51521 : SUCCESS in 7m 23s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bd7206414e0649d0b72857c5c7263152 : FAILURE in 55m 11s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e6022acd45a44a2da3f9de3b6e5d5d8f : SUCCESS in 1h 19m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a30ea3e537724c6196767f0df5756820 : SUCCESS in 44m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d24eab09c0fc4644831a4ab727829612 : SUCCESS in 1h 00m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a76d2f8a333402fb6ac0c5f05bab576 : SUCCESS in 25m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af1a458802a241e08238deb99ccf832e : SUCCESS in 4m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cac8a14771fe4d8aa4ceb86942f4bd52 : SUCCESS in 31m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7de9b5e78c1a482eb2464dd3abf54997 : SUCCESS in 49m 36s","accounts_in_message":[],"_revision_number":2},{"id":"4b48a059ebda2acb98747f1ce8ed8287b9cee823","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-23 18:07:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"124906dd97e1fcb88f6147d36aa4ac586eaa7f39","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 19:18:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b7ec2d09c77423dae526ee71e0b8017\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d7ca0ebc84814c56aab8668b69e213ab : SUCCESS in 14m 58s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9ca146aad8a744aa85af63aed01ff375 : SUCCESS in 15m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e7198b2a4fe45b18e79843f526e44f3 : SUCCESS in 19m 21s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/11f26db41a694d2797f5d0d03dbb523b : SUCCESS in 1h 09m 08s","accounts_in_message":[],"_revision_number":2},{"id":"f3086d7348e22025511437a6028dc5d4830e1927","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 21:23:48.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c95f021d537b4db49b3317fd496c2214\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8acf1b58bb4c4adba10ced0aa7369261 : SUCCESS in 8m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/855f1f6a05cb45eaa23f75e1217229f3 : SUCCESS in 4m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/996d3aeed4b043f6b15d877a9277eefc : SUCCESS in 44m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1c0bccdbfc3146c6a9a696f7a381469b : SUCCESS in 34m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/392bd6652e934c9ba6b0c3158b06919c : SUCCESS in 49m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9eb33cd350664b798124c70cc3feff90 : SUCCESS in 17m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/08bc4d72e5a54def92522173da237536 : SUCCESS in 21m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fb021200f3f64392b63d77845f83d742 : SUCCESS in 20m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/085fd54b01394fd6b9add667d542d7a2 : POST_FAILURE in 38m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7db2615753b54f9eab85a2cc67657688 : POST_FAILURE in 35m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e85d56bddd6b4dceac6ab1d45975066a : SUCCESS in 17m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1c28207207ce4f29b16da90c3c5ca7c3 : SUCCESS in 17m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4b6bf103932e4b0ba0c0defd62eefafb : SUCCESS in 12m 26s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/442c4f3797cc44fcbf43e6e357de71ec : SUCCESS in 14m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9c9179c2056c4776a8b07df9a65ebb7f : SUCCESS in 14m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9a437a8d42084c79bb73866989921bb6 : SUCCESS in 20m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fc683ebe1f734c3580d52ec407da6f1d : SUCCESS in 26m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/639173869a554ae494bf49adda28724f : SUCCESS in 11m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e19bef2ff7434feb9aacbc3149571be4 : SUCCESS in 15m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e88031798a634b6aacf58d210996011a : SUCCESS in 15m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c279e7f4e2a24e0aa266536e4e43be76 : SUCCESS in 6m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cdcd517297ce4a268c50e651c46b43fc : SUCCESS in 18m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4dcecf8ef0944d4fb7dfab02deb2c52e : SUCCESS in 7m 59s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c0c3457552c045c2813f1f6f8c0cde21 : SUCCESS in 54m 34s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/37cab333bff64293a1cb159ac0ed6227 : SUCCESS in 56m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/725dd0ea8c6844a78900ca30a17d7ccb : SUCCESS in 52m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/10dbb87ba24946169ae1aa613233cede : TIMED_OUT in 1h 35m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02006d9784594dc8b101937b3a979049 : SUCCESS in 18m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/443fd1ec4a3e4a7396ea2b5756b14645 : SUCCESS in 7m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f29a782186ad4c4ca1922a759c2efc12 : SUCCESS in 18m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc956125a344435eabc73797d365c078 : TIMED_OUT in 3h 08m 21s","accounts_in_message":[],"_revision_number":2},{"id":"24743c5033615463ae17860eb088afc93e54e4a7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-23 21:57:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8af8febbffd7b3770586fb116819bff483a8a50c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 22:21:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed4937f34219454582e4001baeba5821\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e9b20348e7a34147a43782ad6180bc22 : SUCCESS in 15m 33s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/63a116f0dae9465eb82471ce470209ef : SUCCESS in 15m 32s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1da41027dfe34185b64d50bd123a18c8 : SUCCESS in 19m 42s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a336a6c3b2d94388a8394bfd37d54c7b : SUCCESS in 18m 24s","accounts_in_message":[],"_revision_number":2},{"id":"7aebc5ab3d700f975c93e480b0f2659a10297070","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 23:16:11.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3b3083ff5208429683f18c9d6a49343f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f124f83f7cd846b48fcef941f85fadd1 : SUCCESS in 8m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e48931125f1e45629d8a9d47739a58fe : SUCCESS in 5m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/1fe184bbe596499b9c21198d26e826ff : SUCCESS in 57m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f0c48b20682345daaea05fe9b128912c : SUCCESS in 40m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c2d97e9b0e9e4b8facdf1759aea54178 : SUCCESS in 1h 04m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/88cff3610f3b4d39ba834e0bc1d0920f : SUCCESS in 18m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e9e28a8704414e44871d3b7b0edda1ac : SUCCESS in 1h 09m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/711b4c3de0a648d98e928d3902189c86 : SUCCESS in 16m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5582bf83f84d4792964405a01fba820e : SUCCESS in 21m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/16f14f714c9f4762af71623fd7a4387a : SUCCESS in 18m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1d85b0825d654f2884bfbd6e0baf2dd1 : SUCCESS in 17m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6827b5f9004e42e7a0c538fdf9cdfc6a : SUCCESS in 28m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d458774dc0c443c88b41c8fe51ab1ed2 : SUCCESS in 15m 04s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/da9225b9802f4094a69cc663822cab91 : SUCCESS in 13m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3cab2cb8e21245a9bfd8e6715e45eebe : SUCCESS in 18m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d2033153f3804a58bef0600f6ca1bd34 : SUCCESS in 17m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c0275c8df8d14e6299a4aaa6b3fd0f24 : SUCCESS in 1h 11m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cc245d1873b1400c8641390853b52b6a : SUCCESS in 14m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/32a30569321242089506716ff03b64f1 : SUCCESS in 11m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5cfcf6f36de9451a9d1782deb43e743f : SUCCESS in 15m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7988f6ac45294e9b8dca933d14091ceb : SUCCESS in 6m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/45ed9b3ed57049b080605f285378dc61 : SUCCESS in 16m 46s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b3f0a75588af47d78d9bdd0dc718ed01 : SUCCESS in 7m 56s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6ea57361210d4e3e84ea0c246b94fdc5 : SUCCESS in 57m 26s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/14e424aed8834073bf2db6fecbf612e2 : FAILURE in 53m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5479671c7d9249c694b6a7f3ac17e088 : SUCCESS in 49m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8a83245454a74a6e8824321f25e972ed : SUCCESS in 48m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98f48f6a923e4bd08694471d24a073c8 : SUCCESS in 20m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fceed48b1985411cb9d0a7a603fc86c1 : SUCCESS in 5m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/97a7d0a691074933bc368edce430639b : SUCCESS in 19m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/40881a2ce5cb42ceb7c801f863a96fab : SUCCESS in 52m 22s","accounts_in_message":[],"_revision_number":2},{"id":"f6f91a3f4266c167fa66b9e1bd566b3c72a92da9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-06-24 03:33:31.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"054b03702a5776ac59a2e365e355e9f715192a92","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-24 03:54:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94176719cb8c47f1a7830675550c9c42\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/08b53919f31842318fe00322fd1d9ee5 : SUCCESS in 15m 45s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3c890bc6976b4da1ba032eb81bcfb6f9 : SUCCESS in 15m 53s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/bbc637d02b6a468ea2eb3c27e54ad1c6 : SUCCESS in 18m 25s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5369a74b795d484f9752760c1498175f : SUCCESS in 19m 10s","accounts_in_message":[],"_revision_number":3},{"id":"86bfd65957d936dbe3fe66e8b27cf745fa267491","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-24 05:08:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a9ada01795084941a3f5572a813ea1f2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55158066d6db4b5899ef500af8b5a9bf : SUCCESS in 9m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/853274ca17df4999b82434c5e356f75d : SUCCESS in 5m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/c5c14f6169164cde9a00b953a78a0d9e : FAILURE in 21m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/314bbc5c7797466c9e7e9881afb7c3a8 : SUCCESS in 44m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d04229d2fab242d291abfbf922ed7e83 : SUCCESS in 52m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f0a8cc24ffe040a9a8c16a2b9cce0d8e : SUCCESS in 16m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5b17af3a93e84a88a95114393c4836b5 : SUCCESS in 16m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/40c5f8abcdd84c27ab3d57481c429a5d : SUCCESS in 17m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7640b1eaab774aef9d5b8f8c530a8780 : SUCCESS in 8m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6eae7fe9aed947efa2da7f7fbb45f8dc : SUCCESS in 5m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b3bc094d74814f7aae7b7b4689c1267e : SUCCESS in 16m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4b55114bd8574590b69a31fa042bbc4d : SUCCESS in 15m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a8c91b853c574da295ebdff3e94e7349 : TIMED_OUT in 1h 31m 21s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/bbb534e3977c4b8a8b5ddec84da88002 : SUCCESS in 17m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c616e4364955418d9a7cb7419af20d0c : SUCCESS in 1h 18m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/96400c87f376487dbb7e157f64b16584 : SUCCESS in 18m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1e28ec5ae39b403ea11a360891dba115 : SUCCESS in 17m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3d493d07ccb5467689dc58f751a79d09 : SUCCESS in 15m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c392f6cfdbc84aff8cf4b11648bc661b : SUCCESS in 15m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d14ad206dcef4f32ac7ef46f6d24832d : SUCCESS in 17m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/189b1ad6b28f49e2b6708907c67b14d9 : SUCCESS in 7m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a0c52183a3214db484e8ce442e6aa08f : SUCCESS in 18m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4675fabb1ffb4b998c7ac296cf2f649a : SUCCESS in 8m 34s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bfd7df2890b240058926d77f522b868c : SUCCESS in 1h 10m 14s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/2a8cf7d87ab4488a8d986a151d254bd9 : SUCCESS in 55m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aba9c3155f4848e2a35be1c12098b1fc : SUCCESS in 47m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d86d9616ac094743b36a0187fe2f7431 : SUCCESS in 47m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e00c7b83f1684f25bbe1181ce1d422bc : SUCCESS in 18m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe5522a9869947be862e38716015868a : SUCCESS in 5m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/12ea06051f994efd9d118fb667889ea2 : SUCCESS in 25m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/14ff0b060f01491d8d9e2cfb8f1eba6c : SUCCESS in 46m 37s","accounts_in_message":[],"_revision_number":3},{"id":"50fa63834e17c189cca1c6e940bd332b7fcef200","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-06-27 02:24:20.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"9c4e513c97699e83f949c242a5f576338cccbd7f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 03:07:19.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e6a34938ccb41d8a4a91c66184511c5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b62c506d66674f519347a17949f01a20 : SUCCESS in 40m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/29936c8c341d48a283e0f1041c61ed06 : NODE_FAILURE Node request 300-0021550154 failed in 0s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/551e91c5506844e9879a8064b324f9dd : NODE_FAILURE Node request 300-0021550155 failed in 0s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ce87770e4c744bb0b2847568ee024799 : SUCCESS in 37m 47s","accounts_in_message":[],"_revision_number":4},{"id":"f049500a5d0c5574245892a365ffb702c2551db4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 04:20:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b93eb3dc42949edb6f769bf5bcbed57\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9863f1f6c67400a97a9d8d4c5504769 : SUCCESS in 9m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e381db54d6441a7864dd3224fa222b1 : SUCCESS in 5m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/a1599e7ea33946878848da19735d197e : SUCCESS in 1h 11m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/86e3a0c11b174fefbc574d1626111f81 : SUCCESS in 1h 16m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d2d94a53b90b484e9e300ab337dbd65b : SUCCESS in 1h 03m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6cff60f1c5fd4eba8f93f03c1f88007a : SUCCESS in 21m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df8523873b0645ee907b76ac4d33136a : SUCCESS in 18m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/29264031575040aea98a503a073beecd : SUCCESS in 19m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c24133afd8314a37a1739e9d68171889 : SUCCESS in 10m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f8c29b76217e47c49562f45944250e8f : SUCCESS in 6m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2520be06f8d44faebaabea0c3604a3c5 : SUCCESS in 18m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/da13f51696b5475a859c031eb30d6057 : SUCCESS in 16m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d6b52b39ec734824b0f81ac23e81cfc8 : SUCCESS in 17m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e5fc27fb4d644a928ed0fae222e0f50c : SUCCESS in 17m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/57912e22530248f29e7cc9caa422e824 : SUCCESS in 17m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/10465b45eb0f448d95a57c583262571b : SUCCESS in 17m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2fee0efa7ab1451cb1b21571b8cf330d : SUCCESS in 17m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7724d79470f74c12a09fee9212488d36 : SUCCESS in 15m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e2018bd3c59d4eaa90d23ca5c872b496 : SUCCESS in 14m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/945e8eb42b92455ca9fa28eedc8ab274 : SUCCESS in 19m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4aad79cbbb8d40b581e86b54ab5e5ce3 : SUCCESS in 7m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5f63803bb01c4f379f8663d76d9f8420 : SUCCESS in 18m 57s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b556309ba23043d792e9d8ac66d2c8b6 : SUCCESS in 6m 59s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a4fd552930ca49b79fc4e9a80d370265 : SUCCESS in 1h 03m 54s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b876e908f13e4314918ad3619093187c : SUCCESS in 1h 07m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bf0d7f27235248a7af9721e7de717847 : SUCCESS in 1h 03m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3af42d977dc940cfab041170fef3a1b6 : SUCCESS in 49m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7bb0cb9882c54c7481bc426728192d53 : SUCCESS in 22m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1eeb0758695345d2a67951d049065711 : SUCCESS in 5m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3668823589d044cba5e7ba3728251b98 : SUCCESS in 22m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a12a2ec4355e44d282b43a5045f9d875 : SUCCESS in 1h 12m 53s","accounts_in_message":[],"_revision_number":4},{"id":"9205e7685939f6f0ff71f996b76d77ef551f9f55","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-27 18:09:51.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"dcab951db37f466dbdd45e92ad3c4dd8cc89adca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 18:10:29.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"82285c795e7777768bd59b243b21109c5c7c3da5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 21:07:57.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/666ec72492b24f12880ffe9271101dd6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ab9e91cb757414f9e59574df3508305 : SUCCESS in 9m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/048e655ef7d94f98bdfa158534bcc9dd : SUCCESS in 6m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/68c4ef2ef2194fd2a22d233ee9e61b3e : FAILURE in 53m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/57730f6580e04e338872d80e94c7e131 : SUCCESS in 59m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/22afe4d7db3c40e88ef78aa2d46a430c : SUCCESS in 47m 13s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f12ef4b6c1af4cae99ec152197e16ef9 : SUCCESS in 8m 37s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/83f1edc6653343599840ac21c85136f5 : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b018314252d14206b5bb7627a2d30db8 : SUCCESS in 27m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8302b793f2244287bb015caee4cf2f21 : SUCCESS in 13m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ba202462eeb4442ea217f6772dacb0f2 : SUCCESS in 19m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d90e9005d00547eea2c534a9ddd11b98 : SUCCESS in 18m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5799c49016ce4aafb8f1d924752675b6 : SUCCESS in 17m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f1745c7b2ed04c8da8f675f378834c41 : SUCCESS in 17m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9b799ea7a5ce4630a478b2a619fdd1e5 : SUCCESS in 14m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9ae2eed4d0dc4990bbb9a1233e7fe65d : SUCCESS in 15m 50s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f907c1bbc77a431b843e84f9ea29f716 : SUCCESS in 19m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4ca90b0bda4e4c3dada7a17881073c56 : SUCCESS in 20m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/467237cebd934ffb803826479fb88e3f : SUCCESS in 8m 12s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a5a17b7a8f5b44f1b1c09e2c126184b9 : SUCCESS in 9m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b5983292c93d4c6185d74f9a0d85ff0c : TIMED_OUT in 2h 01m 58s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/611e69c9c281465daa5074da0f4aba74 : FAILURE in 1h 14m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6d61c83daed14a12a05c1459edb65215 : SUCCESS in 1h 05m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6962898cd0d04d48b891f87c5fd55e5f : SUCCESS in 39m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2be65c70b82419fb829ca8d14e9218a : SUCCESS in 15m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a54bb802852428ba2c56b2cbdd69f14 : SUCCESS in 4m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f61b1f27a3442b3b18e278ff4032409 : SUCCESS in 51m 50s","accounts_in_message":[],"_revision_number":4},{"id":"eb75ffbb898c7dbf504376171855186cf6ed70fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-06-28 02:13:45.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":5},{"id":"2dfff2da637642e6b1f221ddbc32c9717ae83c31","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 02:51:17.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ff8d5647ad142b684f5a0d49458dbae\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d8804b0b6dc7455f9c43a96cb37d7fec : NODE_FAILURE Node request 300-0021561091 failed in 0s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2f6882b03243459d9417a7ebb08d43cc : NODE_FAILURE Node request 300-0021561092 failed in 0s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9f243e7dcaac48aaa5b6f76e4166e585 : SUCCESS in 35m 10s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/cec95e6142524559a65a39996cd122ed : NODE_FAILURE Node request 300-0021561094 failed in 0s","accounts_in_message":[],"_revision_number":5},{"id":"1a3205ca1dcb5267476be4ebf2c12160b5700ce2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 04:24:09.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d677d63ff9db4112b3a8a6e2ca25dc9f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/996405689633435fb8327e633e39db82 : SUCCESS in 9m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5dc9f0d34c79425a8053828c01f51d70 : SUCCESS in 4m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/ae017cdd545f4af1b40d997464abe80e : SUCCESS in 1h 06m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/eba2329ae34042679c441659ae313e0b : SUCCESS in 44m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/309e7103eac6483a9f908f8d2dc1bc50 : SUCCESS in 51m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/544604f3e457409c840897c89d1563cf : SUCCESS in 16m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b6bc3c4a04346dca16e058ea232022d : SUCCESS in 15m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/74bd8f9d4aaf48e19fabcd9b0459a317 : SUCCESS in 21m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/77c4ebbba1ce43eb8b5f95706f7694e3 : SUCCESS in 7m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/30e1e35533c14197a99cbc3ebcf2758f : SUCCESS in 5m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7746b0223d6b4668874ea15e9a3db9ae : SUCCESS in 26m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a9a86e9fb8c347ceba40a3a5024960a2 : SUCCESS in 12m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1b1758322ba740baacab71677ddae31e : SUCCESS in 12m 57s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c3e37e76f19e486da1ad253eda53d810 : SUCCESS in 13m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/34ce343cb2ca4b03af13cd3bff5f1046 : SUCCESS in 15m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/63ec2ad5c6f14d8f8958fb004e2382d7 : SUCCESS in 17m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f4b69c2d4d444f13bfcc23cb3f653f77 : SUCCESS in 20m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3205c28cdc0143bd97f8d851088d4428 : SUCCESS in 19m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/290cae99d4c94454b2d928bd20e5421a : SUCCESS in 23m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5f6441a665cc4aca80adf980aa08de47 : SUCCESS in 22m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3f8023d9e04f438d8e7e515887d5b77c : SUCCESS in 7m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fbdf7610825e45a5abc2a83891470eb8 : SUCCESS in 18m 11s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/75da344b52364fe2bfd4a02eafa2278f : SUCCESS in 7m 56s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4df4508682da434199593b2b7202af52 : SUCCESS in 57m 40s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/865a1a511506411ba0d181985b4c4288 : SUCCESS in 1h 08m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/60f8886d5df443c3ba66d46df2338f49 : SUCCESS in 1h 05m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ed4ce894c5964d3cb22e2bf6454ca45c : SUCCESS in 1h 12m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e4374f080164147969c9e731ae9ef6e : SUCCESS in 20m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cfb8bb1aea34c01a1a6fa0a956135d5 : SUCCESS in 5m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2072a511d59640c996bfb91482ce3e1f : SUCCESS in 33m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ccb1c8ad92844758e0f718a1ae426c4 : SUCCESS in 1h 17m 15s","accounts_in_message":[],"_revision_number":5},{"id":"d77268ffd360087cc6f37611d7abcd497942dd5b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-28 21:26:07.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"f303915a1630b204924e94b6365d195e1b498ec9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 21:26:43.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"d01ab55dbeab99bc29c6c73e9b0769cdcc17ea7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 01:27:03.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fdb4c0c02834ba4b6efff4b3da3f6cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d04618633836497498c1fa0168a0bb78 : SUCCESS in 9m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ae113535b4646f0a189fbb4302a3f79 : SUCCESS in 4m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/b27dc3578d32411fad7cc8ae84c0830c : SUCCESS in 1h 02m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3aa636f31e064677b887c1cd13a0ba2d : SUCCESS in 58m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7547f41d4cf44da18be43252a9dfaaa3 : SUCCESS in 1h 04m 16s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/004d5f646de5401784f0bb737cb0f3bb : SUCCESS in 9m 49s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/06a4d6e18a814c8f8991c579193c1253 : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3770151b583043ceaa0ee3bf578bbeea : SUCCESS in 17m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/938548110f664686b9336c9ae4717c25 : SUCCESS in 15m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/03e99dd97e8b42068522f679960b8057 : SUCCESS in 12m 35s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0baf9f847f074f84b91fb19ecf4cba36 : SUCCESS in 14m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bb722e3a2fdb4fe0b76894144b9cad45 : SUCCESS in 16m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3d1cb2f54f804946947ed10fb64f2a51 : SUCCESS in 16m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/951ed9ac04d4449f8dbd67ff5ec71621 : SUCCESS in 17m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a3da98e618854c1cb6feab6be53bbb31 : SUCCESS in 15m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6691eba50c994db89b5429fda90c6944 : SUCCESS in 15m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c9936438ab2846e9a8520ead12ee0e2d : SUCCESS in 15m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/98f660092dab4b0f8e00250e02c51d8d : SUCCESS in 7m 02s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5776457576394d78b635f8f26fb71d73 : SUCCESS in 7m 59s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b76a675c461b4927831ee02eb7d45fb0 : SUCCESS in 55m 09s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f90ba4beba64420689c8380f8293ea13 : SUCCESS in 1h 04m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/757a31b0983748f8b922e57b64c16a3a : SUCCESS in 54m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2fc394ecec3f43d4bc05378ed25d9e41 : SUCCESS in 56m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc7bd7adbd394745bdde26a59fbfb298 : SUCCESS in 14m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cd762ee386a4501bc95856e4374ae2c : SUCCESS in 4m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba5bba64dbb4438693f6e7607e3b1366 : SUCCESS in 1h 11m 51s","accounts_in_message":[],"_revision_number":5},{"id":"2bf6af343056a34de82126f8ac7dca441ada1ce2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 01:27:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"3159f75dc9ed7e355a1eb20288a2ce60a9a10053","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 01:28:04.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b9ac194ee0c42aaba21ff65d6d2e295\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64ae69ae43b54cbfa91adad85e766917 : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a32f0186ee964ac59e29deca92b06dae : SUCCESS in 41s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/aeba4d7a3c9b4914bd76922fd8fa1d34 : SUCCESS in 30s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/2943e7dc78bc4d3cbe7008a1f21470b7 : SUCCESS in 30s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5ef169b924895b383aca321a5c816db13d5d9f88","revisions":{"0c88a737d5a7676c67d6fd863346fdf49feb7ca5":{"kind":"REWORK","_number":1,"created":"2023-06-22 03:37:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/94/886694/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/886694/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886694/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886694/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886694/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/886694/1"}}},"commit":{"parents":[{"commit":"2343521f751388ba2e4158194fb8c2aabc28d08c","subject":"Merge \"proxy: print logs when write new shard range cache into memcache.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2343521f751388ba2e4158194fb8c2aabc28d08c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-06-22 03:37:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-06-22 03:37:31.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I2a5f3e3d841d52d013b481118f56f7a1f6ab3f98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c88a737d5a7676c67d6fd863346fdf49feb7ca5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c88a737d5a7676c67d6fd863346fdf49feb7ca5"}]},"branch":"refs/heads/master"},"162a5ef0f9f338bdb1b06c03e71ded79be0a405d":{"kind":"REWORK","_number":2,"created":"2023-06-23 03:14:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/94/886694/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/886694/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886694/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886694/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886694/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/886694/2"}}},"commit":{"parents":[{"commit":"c5a3354fea345566010420801e8d27326920db85","subject":"Green GreenDBConnection.execute","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5a3354fea345566010420801e8d27326920db85"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-06-23 03:14:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-06-23 03:14:06.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I2a5f3e3d841d52d013b481118f56f7a1f6ab3f98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/162a5ef0f9f338bdb1b06c03e71ded79be0a405d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/162a5ef0f9f338bdb1b06c03e71ded79be0a405d"}]},"branch":"refs/heads/master"},"95f22bbc3fdf15a62762de032d44381520d46fcf":{"kind":"REWORK","_number":3,"created":"2023-06-24 03:33:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/94/886694/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/886694/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886694/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886694/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886694/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/886694/3"}}},"commit":{"parents":[{"commit":"d952951347deec0e7ea5cd7d4d013759a16fbb59","subject":"Merge \"move test_GET_pipeline to BaseTestObjectController\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d952951347deec0e7ea5cd7d4d013759a16fbb59"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-06-24 03:33:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-06-24 03:33:00.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I2a5f3e3d841d52d013b481118f56f7a1f6ab3f98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/95f22bbc3fdf15a62762de032d44381520d46fcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/95f22bbc3fdf15a62762de032d44381520d46fcf"}]},"branch":"refs/heads/master"},"827832ddc5e0558315a3b6a42c1b9c67e2ceba94":{"kind":"REWORK","_number":4,"created":"2023-06-27 02:24:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/94/886694/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/886694/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886694/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886694/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886694/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/886694/4"}}},"commit":{"parents":[{"commit":"7b1f7ee857b0d7019b3e6e17a75d38ebbf650c15","subject":"Merge \"py3: Quiet RemoteDisconnected tracebacks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b1f7ee857b0d7019b3e6e17a75d38ebbf650c15"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-06-27 02:24:18.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-06-27 02:24:18.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I2a5f3e3d841d52d013b481118f56f7a1f6ab3f98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/827832ddc5e0558315a3b6a42c1b9c67e2ceba94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/827832ddc5e0558315a3b6a42c1b9c67e2ceba94"}]},"branch":"refs/heads/master"},"5ef169b924895b383aca321a5c816db13d5d9f88":{"kind":"REWORK","_number":5,"created":"2023-06-28 02:13:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/94/886694/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/886694/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886694/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886694/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886694/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/886694/5"}}},"commit":{"parents":[{"commit":"39127fee5be6eb7733938d31e82bb6553ea9cbeb","subject":"Merge \"sharder: Fix completed metric emission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39127fee5be6eb7733938d31e82bb6553ea9cbeb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-06-28 02:13:42.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-06-28 02:13:42.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I2a5f3e3d841d52d013b481118f56f7a1f6ab3f98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ef169b924895b383aca321a5c816db13d5d9f88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ef169b924895b383aca321a5c816db13d5d9f88"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
