)]}'
{"id":"openstack%2Fswift~886704","triplet_id":"openstack%2Fswift~master~I732d0984c84a12650b18652ea5051d23be4853e3","project":"openstack/swift","branch":"master","topic":"TOPIC-BRANCH","attention_set":{},"removed_from_attention_set":{"36116":{"account":{"_account_id":36116,"name":"Philippe SÉRAPHIN","email":"philippe.seraphin@infomaniak.com","username":"pseraf"},"last_update":"2023-06-27 08:00:41.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I732d0984c84a12650b18652ea5051d23be4853e3","subject":"Adjust column width","status":"ABANDONED","created":"2023-06-22 04:45:46.000000000","updated":"2023-06-27 08:00:41.000000000","total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"937884d5cae70718961098a5f3725bff5f7e7750","_number":886704,"virtual_id_number":886704,"owner":{"_account_id":36116,"name":"Philippe SÉRAPHIN","email":"philippe.seraphin@infomaniak.com","username":"pseraf"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-06-22 06:09:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":-1,"date":"2023-06-22 05:24:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-22 05:24:46.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2023-06-22 06:00:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-22 06:09:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a9251cec93d50e38f1aec0902d70cb8ede43d9a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36116,"name":"Philippe SÉRAPHIN","email":"philippe.seraphin@infomaniak.com","username":"pseraf"},"date":"2023-06-22 04:45:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ca022141904d82c1309c4802bfcbfdb997affae","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-06-22 05:24:46.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"64bfa5c3ad500e8959327f977d5e151f9044ba5c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 06:00:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5775794580d441e2b5822bc51266ff70\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a6e250be2def45fc8e1a1d370db4cb07 : SUCCESS in 14m 06s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/316323b0e1494a5185e4e9d5f723e9a8 : SUCCESS in 14m 38s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2adcd7deb194467aafe1d17bcb1f4ffc : SUCCESS in 1h 11m 32s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/95f51786ee5d41aa9b4ea3bb172a7cb4 : SUCCESS in 22m 20s","accounts_in_message":[],"_revision_number":1},{"id":"33f0816a6f1a944a53a8609fd712edcb827b70d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 06:09:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f977c1d9c544423fade6550bd4a47e57\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bcef667f5f748d2b88c4f75c26e0a8f : SUCCESS in 9m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/60443b93447c459093d82120807304f2 : SUCCESS in 55m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/06d8c80eca5949b2a486a4af8f172edd : SUCCESS in 1h 16m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8511a36272cd460eaf0ab76ab27acce1 : SUCCESS in 1h 07m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e66cd5d0f76e4aa299ca6d41a44c9a76 : SUCCESS in 19m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0814f85490b641649858686c83c5d1ca : SUCCESS in 20m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d3eea4615944ab48477806213198a6a : SUCCESS in 15m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fa6cdc87f3c442a58c51d344f74426e7 : POST_FAILURE in 39m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5e8a360750074c17b8db82d881475365 : POST_FAILURE in 58m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b16d78e2cb3d48dd9080dd6e429d4271 : SUCCESS in 19m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c41937ca088f4088949cf6405409c850 : SUCCESS in 12m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7a3036f9a791451cb8b9315411ab9cd7 : SUCCESS in 14m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7e50bf958c4349bf93957650f609fabc : SUCCESS in 12m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/859a782531284ce99b76025e4bd2400c : SUCCESS in 17m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7badd3cfc3134c1ea5c11489ba03f72b : SUCCESS in 19m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/227826281f5c4606be1fc5048201369c : SUCCESS in 20m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8ce3ee3c79dd49e4ae196fa3f43ea836 : SUCCESS in 12m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6cd7a128144b43968060cf5373e43a42 : SUCCESS in 12m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4ce5625d545b4931b773a9632ccf9cb0 : SUCCESS in 18m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cf0d1ae4d6034a339f89869c372c664f : SUCCESS in 8m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c260be104029492682593413a8abe8d0 : FAILURE in 19m 59s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3d5f80942cfd4f32b1385997a8ccd469 : SUCCESS in 8m 54s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/498f8d3072014965ab0b0c66105db275 : SUCCESS in 56m 56s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f68c9ff48ffe42e39ae2f4b8480f0006 : SUCCESS in 1h 02m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a89f4352fb814c9fbe09417a401b87f2 : SUCCESS in 49m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d017d6b9355e410ea374e4f8ffb290df : SUCCESS in 52m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d574a4897ec4546beba0da84ee11763 : SUCCESS in 20m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e12245157ba54553a16b4ee8c2f8bf89 : SUCCESS in 5m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5aedffac629a44a0a4adcab1d4751a2c : SUCCESS in 26m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe5460cb4c0349238bb1f4462708b018 : FAILURE in 47m 09s","accounts_in_message":[],"_revision_number":1},{"id":"937884d5cae70718961098a5f3725bff5f7e7750","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36116,"name":"Philippe SÉRAPHIN","email":"philippe.seraphin@infomaniak.com","username":"pseraf"},"date":"2023-06-27 08:00:41.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6da6b17f6e947a3a4ecde14afd2efb038d300ed3","revisions":{"6da6b17f6e947a3a4ecde14afd2efb038d300ed3":{"kind":"REWORK","_number":1,"created":"2023-06-22 04:45:46.000000000","uploader":{"_account_id":36116,"name":"Philippe SÉRAPHIN","email":"philippe.seraphin@infomaniak.com","username":"pseraf"},"ref":"refs/changes/04/886704/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/04/886704/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/886704/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/886704/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/886704/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/04/886704/1"}}},"commit":{"parents":[{"commit":"2343521f751388ba2e4158194fb8c2aabc28d08c","subject":"Merge \"proxy: print logs when write new shard range cache into memcache.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2343521f751388ba2e4158194fb8c2aabc28d08c"}]}],"author":{"name":"Philippe SERAPHIN","email":"philippe.seraphin@infomaniak.com","date":"2023-06-22 04:45:26.000000000","tz":120},"committer":{"name":"Philippe SERAPHIN","email":"philippe.seraphin@infomaniak.com","date":"2023-06-22 04:45:26.000000000","tz":120},"subject":"Adjust column width","message":"Adjust column width\n\nChange-Id: I732d0984c84a12650b18652ea5051d23be4853e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6da6b17f6e947a3a4ecde14afd2efb038d300ed3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6da6b17f6e947a3a4ecde14afd2efb038d300ed3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
