)]}'
{"id":"openstack%2Fswift~886780","triplet_id":"openstack%2Fswift~master~I0fb1f764722d73db6d3b79acc128f37f51499d35","project":"openstack/swift","branch":"master","topic":"disconnect-tracebacks","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2023-06-26 11:23:39.000000000","reason":"Change was submitted"},"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2023-06-23 17:05:23.000000000","reason":"\u003cGERRIT_ACCOUNT_597\u003e replied on the change","reason_account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-06-26 08:29:41.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-06-26 11:23:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0fb1f764722d73db6d3b79acc128f37f51499d35","subject":"py3: Quiet RemoteDisconnected tracebacks","status":"MERGED","created":"2023-06-22 18:39:52.000000000","updated":"2023-06-26 11:24:50.000000000","submitted":"2023-06-26 11:23:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"886780-disconnect-tracebacks","meta_rev_id":"7903c113c57f77fe5975cc154b3fe4a292cac14f","_number":886780,"virtual_id_number":886780,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-26 11:23:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":2,"date":"2023-06-23 17:05:23.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2023-06-26 08:29:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":1,"date":"2023-06-26 07:50:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2023-06-26 08:29:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-22 18:58:57.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2023-06-22 19:02:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-22 20:19:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-22 21:11:39.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2023-06-23 13:58:30.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2023-06-26 07:50:07.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"9a748df53afebe8aeceda07f4df38f2dd0caf8b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-22 18:39:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18ac9bf5cd850ed1fd0d132bf5b7d88aa2bbcc91","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-22 18:47:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dd6f7a23edf0a80f0e9fd4dfaaf1cb908b9e265e","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-06-22 18:58:57.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"013410a1891c895187175415c9e1f3bc4be7a2bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 19:02:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d920269bb04345628f5662623823df35\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3fd6d4f631134a2398c8b49aed84b582 : SUCCESS in 15m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ff771636cd7a49a197f303b22f52ca85 : SUCCESS in 18m 30s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/80e5e1d779b84c2b881ee164a30e4422 : SUCCESS in 21m 44s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/043335b196634bbca77589ad9a8e745e : SUCCESS in 18m 52s","accounts_in_message":[],"_revision_number":1},{"id":"10221863d11bfce0d0031aaeba955d15b955a36c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 20:19:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa4cd19aec744bdf8507d1072b865ddd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bdc0a5e704145b9abdc29779eb97553 : SUCCESS in 8m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/442faa0b8b4446d7b052a80a129c5188 : SUCCESS in 50m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/801c34ad028e4ff583a18961b2c8db4d : SUCCESS in 49m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01be3faa2b904b878d3fd20876e0f6ae : SUCCESS in 1h 04m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5f1b572bceff47e79e0b12e7b47ffe3d : SUCCESS in 18m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a52f7e7142074655b5b831d557349ce2 : SUCCESS in 26m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad069c2648024ae290859a4208cf1489 : SUCCESS in 36m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fecf47c1eaee4b1ca5228c32d7b79c9b : POST_FAILURE in 38m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/966b372ea0414a42b5728ecb3956d2f0 : POST_FAILURE in 35m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cd80350bc54f4ea6aa2c1c75aa7cd323 : SUCCESS in 24m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/09e9142538c746d9988f4eac1a147873 : SUCCESS in 20m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/14d44c44f54844e985d5e405b8e11b2d : SUCCESS in 17m 40s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/975a17a78281462dbf3f533e729afa46 : SUCCESS in 16m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2386b5d142ab4b6b802e355c86d539e1 : SUCCESS in 29m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/09f1ff21b8974c17aef223cba26f7854 : SUCCESS in 27m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0da9ccccdc114db09910363afb981495 : SUCCESS in 26m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/10e98e1341d94c758aa95bf490d6af70 : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fe8e41c284554cb483f79dbea8c13c9a : SUCCESS in 23m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cff0358502ae427ba66a0ecbe4d66ee7 : SUCCESS in 18m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9e3ad629662b45bdaaa4b8ce893b57a2 : SUCCESS in 6m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/313d6986de04401388c307c133b2e243 : SUCCESS in 23m 03s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/187b79380d5d4c9193eefc608c123782 : SUCCESS in 9m 02s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/060e133136a340abb2cfc788caad0492 : SUCCESS in 1h 00m 55s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0f0d73793a944e4daba1dd0fd6c14032 : SUCCESS in 1h 08m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/53080eb28e614b8eb69b419095f68c09 : SUCCESS in 55m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cb0db2b4fcc742cc80214bf0e3575f95 : SUCCESS in 1h 04m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24ee46abed3a4e62a347c447f14aef03 : SUCCESS in 21m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e8fef4e94aa43d78977ca2d5e84bd12 : SUCCESS in 5m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/65ccac82836b42c496344d73bcb971d1 : SUCCESS in 23m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ae365bbd10d467184e791dcf1b1c457 : SUCCESS in 59m 30s","accounts_in_message":[],"_revision_number":1},{"id":"bd049c52ec1ef89d82565861b492a0a4b4feb882","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-22 20:46:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"4de205844b7aa1d6ad20b9e14da1c867d94129ba","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-06-22 21:11:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6c2ca8f2b3572e2e34ade96faefcb7e69a50aac5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 21:44:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b639145d973d4c1d951e567811159731\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b106a436242943af8fac5c8943aecb75 : SUCCESS in 12m 58s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ef70a09b775a4501ba1efa1b1ba061d0 : SUCCESS in 56m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f5b8a1f5ae1b40349ddbc444c498d958 : SUCCESS in 20m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3ed8cf268af549839c5ac68b542b49b6 : SUCCESS in 23m 12s","accounts_in_message":[],"_revision_number":2},{"id":"851e6584f17c13c9f0956cbf408dd86bf9e1ab1e","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-06-22 22:03:58.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"726a394a4401492004b99bd72045b8cdce8a9030","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 22:31:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4fe162442fd49da89c7e2044f5ffea1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b0fd6bccd724b758ccd38dfbbad6120 : SUCCESS in 8m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/fb740809f621409e8e74dfb2956b2b2d : SUCCESS in 48m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2f058256e58643adb7af063f7692b32c : SUCCESS in 35m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61ac0cab73da4870a2e20a2130c040ba : SUCCESS in 54m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/176a7354040f46bd8dd755495bdd113b : SUCCESS in 21m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/96a91da5af9b47d38e1dc4218d986d97 : SUCCESS in 17m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ca484dab23d743acaafff39212da78a0 : SUCCESS in 20m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5e0ad933afce42d3b280cf2916f276d5 : POST_FAILURE in 38m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c31d071161e746548e4062dff907d086 : POST_FAILURE in 34m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ffce16da63dc45e7b8dc97069c111efb : SUCCESS in 17m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d40d39ffd5ef4bffbf2c0cbce2567738 : FAILURE in 20m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5d2eab59d2a04ba3a7352f663cda4b08 : SUCCESS in 19m 06s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c7bb86c495fc41f381acb465fa645ae8 : SUCCESS in 14m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7181e538f6d64d0d904d4bf8642ef153 : SUCCESS in 15m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/54ce134c8a334a4e9a1a0dcaeebf06f2 : SUCCESS in 15m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e78db4ee49184f0da8fb90b5717b65cf : SUCCESS in 16m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4f09d655a07b40cea1d6f01949f2457e : SUCCESS in 14m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a44c642ee97c4166bbf8bef7b8efb4d1 : SUCCESS in 14m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9d31cd2d212f431f97f4989da48942e7 : SUCCESS in 13m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/65748416e56a40cc92a0a448cc259150 : SUCCESS in 6m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a177426d37464c99995a9caebed11e38 : FAILURE in 1h 29m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/339a23724120417aab94e6bf3ef0e14c : SUCCESS in 7m 42s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9c5af29f22044885920615d9d4af22ea : SUCCESS in 54m 06s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ada3336e2c6b42bfac4eef4bab90e3a2 : SUCCESS in 1h 14m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/292030cc00014b91a8492d4383975b0a : SUCCESS in 38m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/181b9bf603c048b8b8db977c86e23dc8 : SUCCESS in 38m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/047d2c0267cd4e9c8b84baa4c3d3d96a : SUCCESS in 18m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47a6baa281d7432aab627527405349fd : SUCCESS in 5m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/72fedb69976548b18c3493e66c7fef2b : SUCCESS in 20m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce0d84c387b54725983aabe3ae49c880 : SUCCESS in 49m 56s","accounts_in_message":[],"_revision_number":2},{"id":"a65bdacf0224670415a22889feab9a3985e806a6","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-23 13:58:30.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fb0108a8d88be7ddb58af39a8f94c9dc5a401519","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-23 16:41:46.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5a5f63ff1014b607a8d22ff64f0433520e6790b2","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-06-23 17:05:23.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"98290e920fce2274df31062efa8be12505347e8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 18:15:18.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96c562e6756343a49997c479a0b2ebbd\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3473dc29e98f4f8393835417b3a9f95a : SUCCESS in 17m 38s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/246a841851fa44e98980538f9bfa9bfc : SUCCESS in 20m 06s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/39d76befece248e1ade87215654fab11 : TIMED_OUT in 1h 30m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a2d39e688820449b9687ca78a321263d : SUCCESS in 21m 58s","accounts_in_message":[],"_revision_number":3},{"id":"0accbda885f280f93c1e10303cad4568c35d3954","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 18:16:50.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/460ea7d3218144d1b86bface5ad03d91\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3185494fbdbc4afb967d8f7675ddd03a : SUCCESS in 9m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/81fed0ff927d413c9ae724df16d151a7 : FAILURE in 32m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/21d64b3375064086827e56d98741ce76 : SUCCESS in 50m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d69f533a67a41a69a3edb4bf9ff1034 : SUCCESS in 52m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d61904ad8e044c63b9f734a10eb40799 : SUCCESS in 17m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d32926c8a6141548e5e4db385a46c52 : SUCCESS in 18m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e46f2a1ce6b947568538d6f3c9a2e551 : SUCCESS in 17m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/13d610868bc44c1cb9ceb3a417c757d5 : POST_FAILURE in 38m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5aad68ec44f042fd91e9308c1bbcdcb1 : POST_FAILURE in 34m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a9a5ca808ced44538bac1048c3c7e0a4 : SUCCESS in 25m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/78569e05d6064443a02539fcd9b6c835 : SUCCESS in 16m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6a5c5130e7594edfbd1b971fdc355a9e : SUCCESS in 18m 59s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d01c83c5a914410e9f0a8a6648c865eb : SUCCESS in 15m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/72f5a0f53794472eb387c752dbeacc49 : SUCCESS in 28m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/09e1fc9c2d7245278b369ce524c6079b : SUCCESS in 15m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c914a7ca0e854ad6a9733c077f41b70a : SUCCESS in 18m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/845bc95e8a074f3d8d991dc8e3285bcf : TIMED_OUT in 1h 31m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2ab3930ae2dc45adb5fac0275a53105d : SUCCESS in 14m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/93c47b384486493c9b2243f813e06086 : SUCCESS in 22m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7675152f6f1d44c58699190002a4eacf : SUCCESS in 6m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4115208db4444f25ae05c5da7db85e52 : SUCCESS in 28m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f65da1e123c540b2a96ed6d19e3f83bf : SUCCESS in 7m 19s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f809c57600854243af82742fa2e1ad9d : SUCCESS in 48m 52s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b641ba7b5f6542108b77dbf542906b50 : SUCCESS in 57m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1841147cd3634e2ab6ccd363b0e4038e : SUCCESS in 53m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d442880ece3d45288e8fbee74a0b6c77 : SUCCESS in 41m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e37537cb61924f449754809cf1b0e947 : SUCCESS in 16m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbc436b1103e450691a7b4f3b4c95b91 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ce21d414b9124b90b7e3835d54cab369 : SUCCESS in 21m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f78c6414d0604b83a5ac6b9f89ab4d86 : SUCCESS in 51m 00s","accounts_in_message":[],"_revision_number":3},{"id":"ba8d530adfdec334c88a51115665db92a01f8f5e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-23 19:33:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"049ba4133db25fe90744cc23efb0654e9ddfe109","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 19:59:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea31bf6b380f4e68b76105216788392f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a041db4a10f6411e8008294d5f34db92 : SUCCESS in 16m 31s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3cbb8e4dfb1b43219fcff3b3af8e4ac0 : SUCCESS in 16m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a9ffceadc56140baa0bb1dfefffa33e8 : SUCCESS in 22m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3828e35bbc144b68a35696c337f0a35d : SUCCESS in 18m 59s","accounts_in_message":[],"_revision_number":3},{"id":"82e63429b961fe9453192c9e703d0449956f9334","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 21:08:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a09fc94f7ccd46ef81a6421b906dba88\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51adfe59158a4d19b4a295e0ebd36a4f : SUCCESS in 9m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/5593eaa92b0447ddb190901a3f7d98e1 : SUCCESS in 1h 00m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/59ebdd1a2b0446d68a1c80c977d7903d : SUCCESS in 53m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d31d27b0d2441549ef2aed107791d01 : SUCCESS in 1h 02m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d49bb52be896475ab6be68cc6b6b0dd3 : SUCCESS in 18m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8120476502ec45fab1e57932678da13d : SUCCESS in 24m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5b0f6989bb184eb39d8b66992290aa44 : SUCCESS in 16m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2c164b6f2756453b8070f71bdde857ca : POST_FAILURE in 37m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/337c058236f94424aee5a49a2e6f5be4 : POST_FAILURE in 35m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/86fe6b1459674b5a92be67effe05e417 : SUCCESS in 24m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/82ffb72d05534fe8a82950049ac90d31 : SUCCESS in 23m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5e2d28010ba44455a35453a706c90d2e : SUCCESS in 23m 18s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/085d72a4f8744fdb9c0d7abbf22b53a3 : SUCCESS in 15m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7c4e1dc83bf84cafbba8fc790bbb61a3 : SUCCESS in 26m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/05a266ce6356448dab14d9b63c80194d : SUCCESS in 17m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/db1044ae689149859fe7834b3ee43348 : SUCCESS in 25m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/be3a4e4b44e84ef89e0bba95d5d3b545 : SUCCESS in 14m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/95630ca1040744e7b16e7e973973f8f1 : SUCCESS in 15m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4d3f474a40e641b1887c5d0c667ffe27 : SUCCESS in 17m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2b0ac00cc8f24d07a07151b059b3e8a4 : SUCCESS in 8m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a2a2a41927ee4ce3b6b4238856f8c7aa : SUCCESS in 16m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a3ba75e12077494aa7a06d81390ee3f9 : SUCCESS in 8m 10s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d09b1262f66845ca9cab9e78f85bb17e : SUCCESS in 52m 35s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4510f710bf284785b9ce67b24b6bbc4c : SUCCESS in 1h 04m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1a17217d20ff4023b76874357fb439c6 : SUCCESS in 47m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2e0cb8289c1b46d291cc4fa90b11d91d : SUCCESS in 50m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d47b48fa817430ba2207bfd39a928fb : SUCCESS in 16m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfcc11ac6c414b85a05a3727d9dd7c0d : SUCCESS in 4m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/72451804b99242ea9325e199e5ecbfdc : SUCCESS in 19m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67387deb06c249a7ba8d4b04388480dd : SUCCESS in 1h 02m 56s","accounts_in_message":[],"_revision_number":3},{"id":"4c5bb69f4af6ca872d76ad48e8ae117691440c59","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-06-26 07:50:07.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c9481021f80cb58fdc77849142f87f9ca248d6f3","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-26 08:29:41.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3a90b32b0698823057128af91360745fb52e2055","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 08:30:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f9ef46d6fa5f7bf8dae40a05176a2691aa22b463","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 11:23:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bfe415aaab945f0a07ae29c1d6d859b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6adf1fcbf2614e29bbeff0cb80910c84 : SUCCESS in 9m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/987fae5c8f4342bebf7cc3dbe13ecfb9 : SUCCESS in 1h 08m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c98dd67e6b584df9802de9d211bc2b5c : SUCCESS in 54m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/489e021c29ab4b8f9d68c31a9f0d1f66 : SUCCESS in 1h 09m 03s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/3082b779573c427fad5887c4578487f4 : SUCCESS in 10m 06s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/fd3aef2264734ed19bb11c383c989bf2 : SUCCESS in 6m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5018e665778b426bbff23388b6029254 : SUCCESS in 19m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7acbc9e00597408ab4049151903a3887 : SUCCESS in 31m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e13caf2fcae3403996f2093746d0f3d2 : SUCCESS in 14m 33s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/91ada5bb064743249dc9dfcd19d02018 : SUCCESS in 30m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/89fa813e174445dea8bb1b871722545e : SUCCESS in 22m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/984786e23f334ea79987aca128398867 : SUCCESS in 18m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aaac8c761d384db0a7a21d96964d9902 : SUCCESS in 21m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ad8536a0aca24f89b890d74354e937fc : SUCCESS in 14m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9ced7fb2da5d4d869abc89e2b54f1fa5 : SUCCESS in 21m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9281979d82664572850c2bee1ea0b4e9 : SUCCESS in 15m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/339648501c1a441293c1c612133862fe : SUCCESS in 8m 00s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/043356a289944773a9e4998a34a98f80 : SUCCESS in 8m 16s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/057e5856f1484360a227c8d14d8b358a : SUCCESS in 1h 03m 51s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/64501b852fdc4dfd9166c633dad794d3 : SUCCESS in 1h 04m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4a9e885fd0d9455683d2b587a96af60e : SUCCESS in 53m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c9151d89c9da49fcb8cb3f5e94987835 : SUCCESS in 52m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2068c17d27174b1eb6133c9380ecc367 : SUCCESS in 21m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/818ef9b88df64a3d98e7cdae1f824e47 : SUCCESS in 6m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce38632147294488943be4537e43da0a : SUCCESS in 1h 00m 21s","accounts_in_message":[],"_revision_number":3},{"id":"0f028f5c2e9c17f67346e708d1a41eaa7997609f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 11:23:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"7903c113c57f77fe5975cc154b3fe4a292cac14f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 11:24:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8aa9de44f2444d4c9328164f489b8a1d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f35733babdf24af6b3ae8151f0237c5c : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/86fb4eac86f5411d9f06def2e5eed111 : SUCCESS in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/7673565a25704e549ed2f54c5bf4c433 : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"483e17d5b4f9c1c679332f055791721be66b0b02","revisions":{"bf046ee08bd067b214df0cc30de835b1b8ff43b2":{"kind":"REWORK","_number":1,"created":"2023-06-22 18:39:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/80/886780/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/886780/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/886780/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/886780/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/886780/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/886780/1"}}},"commit":{"parents":[{"commit":"2343521f751388ba2e4158194fb8c2aabc28d08c","subject":"Merge \"proxy: print logs when write new shard range cache into memcache.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2343521f751388ba2e4158194fb8c2aabc28d08c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-22 18:38:50.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-22 18:38:50.000000000","tz":-420},"subject":"py3: Quiet RemoteDisconnected tracebacks","message":"py3: Quiet RemoteDisconnected tracebacks\n\nChange-Id: I0fb1f764722d73db6d3b79acc128f37f51499d35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bf046ee08bd067b214df0cc30de835b1b8ff43b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bf046ee08bd067b214df0cc30de835b1b8ff43b2"}]},"branch":"refs/heads/master"},"c51e22520fd9692f25c9300b3d7926aeb077be66":{"kind":"REWORK","_number":2,"created":"2023-06-22 20:46:06.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/80/886780/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/886780/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/886780/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/886780/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/886780/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/886780/2"}}},"commit":{"parents":[{"commit":"2343521f751388ba2e4158194fb8c2aabc28d08c","subject":"Merge \"proxy: print logs when write new shard range cache into memcache.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2343521f751388ba2e4158194fb8c2aabc28d08c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-22 18:38:50.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-22 20:42:02.000000000","tz":-420},"subject":"py3: Quiet RemoteDisconnected tracebacks","message":"py3: Quiet RemoteDisconnected tracebacks\n\nRemoteDisconnected from both BadStatusLine and ConnectionResetError\n(which in turn eventually inherits from OSError). We want to make\nsure it gets handled as a BadStatusLine, as it doesn\u0027t get its errno\nset and would otherwise get the default traceback handling.\n\nChange-Id: I0fb1f764722d73db6d3b79acc128f37f51499d35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c51e22520fd9692f25c9300b3d7926aeb077be66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c51e22520fd9692f25c9300b3d7926aeb077be66"}]},"branch":"refs/heads/master"},"483e17d5b4f9c1c679332f055791721be66b0b02":{"kind":"REWORK","_number":3,"created":"2023-06-23 16:41:46.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/80/886780/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/886780/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/886780/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/886780/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/886780/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/886780/3"}}},"commit":{"parents":[{"commit":"2343521f751388ba2e4158194fb8c2aabc28d08c","subject":"Merge \"proxy: print logs when write new shard range cache into memcache.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2343521f751388ba2e4158194fb8c2aabc28d08c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-22 18:38:50.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-23 16:40:39.000000000","tz":-420},"subject":"py3: Quiet RemoteDisconnected tracebacks","message":"py3: Quiet RemoteDisconnected tracebacks\n\nRemoteDisconnected from both BadStatusLine and ConnectionResetError\n(which in turn eventually inherits from OSError). We want to make\nsure it gets handled as a BadStatusLine, as it doesn\u0027t get its errno\nset and would otherwise get the default traceback handling.\n\nChange-Id: I0fb1f764722d73db6d3b79acc128f37f51499d35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/483e17d5b4f9c1c679332f055791721be66b0b02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/483e17d5b4f9c1c679332f055791721be66b0b02"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
