)]}'
{"id":"openstack%2Fswift~886802","triplet_id":"openstack%2Fswift~master~If36b428ac24b05a77a7fcd324256e6357d2ee410","project":"openstack/swift","branch":"master","topic":"TOPIC-BRANCH","attention_set":{},"removed_from_attention_set":{"36116":{"account":{"_account_id":36116,"name":"Philippe SÉRAPHIN","email":"philippe.seraphin@infomaniak.com","username":"pseraf"},"last_update":"2023-06-27 07:59:53.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If36b428ac24b05a77a7fcd324256e6357d2ee410","subject":"Adjust column width","status":"ABANDONED","created":"2023-06-23 04:46:35.000000000","updated":"2023-06-27 07:59:53.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8e5e8ff12d787a2896c5d9ec13f4d43b0a2bb6cb","_number":886802,"virtual_id_number":886802,"owner":{"_account_id":36116,"name":"Philippe SÉRAPHIN","email":"philippe.seraphin@infomaniak.com","username":"pseraf"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-06-23 07:02:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-23 05:08:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-23 07:02:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-27 07:53:23.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"state":"CC"}],"messages":[{"id":"e8d8635f260d64ec3a7d2ef78f4fbdac73691924","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36116,"name":"Philippe SÉRAPHIN","email":"philippe.seraphin@infomaniak.com","username":"pseraf"},"date":"2023-06-23 04:46:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"550e96d653b2f2e01cc98ed0114802866e6609a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 05:08:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/631cc7ef59da4eff822e5a3280d270f4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/15f00ab471a74e808260264de03da6df : SUCCESS in 14m 18s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0c25f81589d04cfd8660e0a6175d0c36 : SUCCESS in 15m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/cf9b1b21f6ec4e4f98d0c6d92d74da7b : SUCCESS in 19m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e56f684ab6044bd8ab530daafb683ce0 : SUCCESS in 17m 36s","accounts_in_message":[],"_revision_number":1},{"id":"858bfe77bccfdf612e11eaa4e6f184517be601cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 07:02:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/243adc5267dd487099364d78c4aabc4e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16dd3c91f269475f880f35d55fef7a94 : SUCCESS in 12m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/28addbdc53ca4e6d9573199375444dfc : SUCCESS in 45m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6f0da472aaf349478dad81713c9e9196 : TIMED_OUT in 2h 08m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d5256d31e0e456d8123d84ee2141341 : SUCCESS in 44m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2bd8085a7f384bf1ab76a34b9bc99888 : SUCCESS in 36m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c90c5a27a30d4e94b4dfb50855b2b3e5 : SUCCESS in 23m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/550e9a3bf0c145e6b51198711d14d10f : SUCCESS in 17m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/129b76aec5774c2eae06098972e1f7bd : POST_FAILURE in 40m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/29968f39ac8a4653841a317ee3c85cf7 : POST_FAILURE in 35m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/968d7e7394964b998aa765b1fafc6cba : SUCCESS in 20m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/03e574338f864aa79d37fe935de14f99 : SUCCESS in 15m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/54490af9faca475b829307feca6f6a04 : SUCCESS in 13m 57s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/27ce4a3fd39c4308b64a4d131d77fcbe : SUCCESS in 15m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5d60e31460114a53b8c6d2509af7c87c : SUCCESS in 16m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/801fedd65edd4eeaaba9434ed007a2ba : SUCCESS in 16m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e3c85076b65c41aca93c353f00851885 : SUCCESS in 17m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f32e630e6f0d489ca0850ac9782dbedc : SUCCESS in 12m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5b17e0c315764a4c82bb5d25a95d5ed2 : SUCCESS in 13m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ae49f2ec6340410b96e8d13ff63bd293 : SUCCESS in 14m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/176db74a6af947bc84deb7062fd3b87d : SUCCESS in 8m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ba32890143cd489db2f691c91f24420a : SUCCESS in 16m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4e39789926494c0aa89279ddbd96f604 : SUCCESS in 11m 47s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/485863c640e74452a38fc2e11f0e4127 : SUCCESS in 1h 03m 09s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a7dcf23ee3b74b138eb57677a179626e : SUCCESS in 52m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2dc03e899b9647e99d7f3b89e69e38d4 : SUCCESS in 41m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6b72235ae7934e9abf42ea309bbba480 : SUCCESS in 40m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1bc462033d374ff9add559e0af7c2195 : SUCCESS in 17m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb279bf7cc1841b9b486fef7f349ca6a : SUCCESS in 5m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/63e66540a3ad4719a004b31d10f61bb8 : SUCCESS in 27m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c4f9e25fa7044f98d66608797759688 : SUCCESS in 1h 06m 15s","accounts_in_message":[],"_revision_number":1},{"id":"71b269337e531ac00b32217139a78cc4f7871d1e","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2023-06-27 07:53:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8e5e8ff12d787a2896c5d9ec13f4d43b0a2bb6cb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36116,"name":"Philippe SÉRAPHIN","email":"philippe.seraphin@infomaniak.com","username":"pseraf"},"date":"2023-06-27 07:59:53.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"159ffe655d6f5f8b4e313d510e5218482cc3688a","revisions":{"159ffe655d6f5f8b4e313d510e5218482cc3688a":{"kind":"REWORK","_number":1,"created":"2023-06-23 04:46:35.000000000","uploader":{"_account_id":36116,"name":"Philippe SÉRAPHIN","email":"philippe.seraphin@infomaniak.com","username":"pseraf"},"ref":"refs/changes/02/886802/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/02/886802/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/886802/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/886802/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/886802/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/02/886802/1"}}},"commit":{"parents":[{"commit":"e65ca2f84ad2ee2968a7660b4eeb722130284603","subject":"Add Major:Minor detection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e65ca2f84ad2ee2968a7660b4eeb722130284603"}]}],"author":{"name":"Philippe SERAPHIN","email":"philippe.seraphin@infomaniak.com","date":"2023-06-23 04:46:11.000000000","tz":120},"committer":{"name":"Philippe SERAPHIN","email":"philippe.seraphin@infomaniak.com","date":"2023-06-23 04:46:11.000000000","tz":120},"subject":"Adjust column width","message":"Adjust column width\n\nChange-Id: If36b428ac24b05a77a7fcd324256e6357d2ee410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/159ffe655d6f5f8b4e313d510e5218482cc3688a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/159ffe655d6f5f8b4e313d510e5218482cc3688a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
