)]}'
{"id":"openstack%2Fswift~886823","triplet_id":"openstack%2Fswift~master~I31ed36d32682469e3c5ca8bf9a2b383568d63c72","project":"openstack/swift","branch":"master","topic":"unify-get","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2023-07-24 22:04:28.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-07-24 22:04:28.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-07-24 22:04:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I31ed36d32682469e3c5ca8bf9a2b383568d63c72","subject":"proxy: remove client_chunk_size and skip_bytes from GetOrHeadHandler","status":"MERGED","created":"2023-06-23 13:42:30.000000000","updated":"2023-07-24 22:05:41.000000000","submitted":"2023-07-24 22:04:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":8,"has_review_started":true,"submission_id":"886823-unify-get","meta_rev_id":"849941da05f8390008cedd94dbf23654eaec69bd","_number":886823,"virtual_id_number":886823,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-24 22:04:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2023-07-24 14:18:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-07-24 16:18:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2023-07-24 14:18:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-23 14:02:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-23 15:03:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-24 04:02:07.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2023-07-24 14:13:17.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2023-07-24 16:18:45.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"REVIEWER"}],"messages":[{"id":"33745399ba6d863e407b7f30ab37020eb8b3f31c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-23 13:42:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e05f2c49ecaea194618b28021cf4085aad79d11a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 14:02:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42f1bcd8b0314750bbfb2df4de0c02b8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b0aa3ddd8061461aac95df97b62f38c5 : SUCCESS in 15m 20s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/547363fa51f343ddbe8aa2579b0e7eb1 : SUCCESS in 14m 10s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b8f3b104f550489a84e8049db4da2d3b : SUCCESS in 18m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1d1803c39cd048298333ce96a4624566 : SUCCESS in 19m 03s","accounts_in_message":[],"_revision_number":1},{"id":"6a9eca2f937da058c1063829daae20bea82cff4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 15:03:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e45eaeb0df194faeb306426060a24fcc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f709b10dc3974f368cde9186eb620e1c : SUCCESS in 9m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/0703bae56cbf4f93a29a287682fd91eb : SUCCESS in 1h 03m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/41ac5599fd6b498d9884f0c80dd9cbe4 : SUCCESS in 55m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e8ef2f54eed4411a2e0140a8c1d24fe : SUCCESS in 1h 02m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5141cf92648a4d139ac13e5b4f4e215d : SUCCESS in 28m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d4c4e107d5e44332a586fa8903c79058 : SUCCESS in 28m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b840fb7420fc4aeca47d0032e4120007 : SUCCESS in 30m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4dfa025625124038beb810c7934a6248 : POST_FAILURE in 38m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/aa1a00fc42874d65a2a048e09bc723cd : POST_FAILURE in 34m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/02b18136a9f2485cae85019a294e2843 : SUCCESS in 27m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8a811dbb6beb4392832c4ddca5c6d7d6 : SUCCESS in 28m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fea2ccb2b814442fa115e1c6ba3108c6 : SUCCESS in 28m 17s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e1a9092bc2b348b58da2607cca54874d : SUCCESS in 29m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f24e8a05a32a458da812816483204faf : SUCCESS in 29m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e7f4a963f4084a5a82a2e6cf19661ba6 : SUCCESS in 16m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/58dc499e2d024aa6be1c48c8906ed773 : SUCCESS in 17m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6ffe6e57cc234f73a5913a900696d7ec : SUCCESS in 13m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/abe4730db1fd4e268fcc524aee1bac67 : SUCCESS in 13m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0f1b50da0ba64462a8a32d5a787a0a7d : SUCCESS in 14m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d74eeede3f2a4988a7a9196ac04f8e7f : SUCCESS in 13m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8066cbcc18f540b48faa12c0ffce4d71 : SUCCESS in 1h 10m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a14665c85dac45f3a75f771a7add9d60 : SUCCESS in 7m 19s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9faafb6b1f7446d29d976552096886bf : SUCCESS in 45m 22s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/20c730c64d84495da76bfab369f3f52e : SUCCESS in 1h 04m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1adcd3b9cccd4eefbdf7c8d408ff9076 : SUCCESS in 55m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b67a95e68ba2407e9d6492d67a809edf : SUCCESS in 1h 00m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3dd0c68af8cd4aca841d79b8f1c44375 : SUCCESS in 18m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e13addd53abd4f4590141e16db9ffd09 : SUCCESS in 4m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/00b2fd9c680849fe8cf30460ee16898a : SUCCESS in 19m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d103903423848cfb416316a27cf39cf : SUCCESS in 1h 01m 54s","accounts_in_message":[],"_revision_number":1},{"id":"63143f645ec367497bf70e1c03711d8a2fd12b34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-26 09:26:45.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"8e1162722fb007befa187cd04507a0158d6ee6b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 09:56:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9129738d38f4ac887f0539003bfcae9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/66e243b22ed947239a4ee5ad24b38fdb : SUCCESS in 21m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/26dc4573d9af4597aca8c252d576cec7 : SUCCESS in 19m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/461113e489ba41499b4dbeda7ea13e54 : SUCCESS in 27m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4529587a355a4664a901b080eb2e0d17 : SUCCESS in 26m 51s","accounts_in_message":[],"_revision_number":2},{"id":"2212ef5f7eb30c87d2aaed613d57a2973e02fcf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-26 11:12:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4336a790fef3a8884d9c64e26546af9146e30b3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 11:34:22.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9be278eb894c49579d9e27025813d2f6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6daabe9b202d493883a1735494a23b4c : SUCCESS in 17m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aa8164d7f0004b6895f2ba8a04c2b9d7 : NODE_FAILURE Node request 300-0021543717 failed in 0s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4a4e87ae88740089a25148bff9f39b8 : NODE_FAILURE Node request 300-0021543718 failed in 0s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/74b43cab8fb148529f0a21e8ee475563 : NODE_FAILURE Node request 300-0021543719 failed in 0s","accounts_in_message":[],"_revision_number":3},{"id":"efd6e5029dc086faad29a31c01d3ffef0d07b6fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 14:56:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cceaebd08664534a868e93ca768cab3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9549fd42974542a1ae11b29cc4ff2e97 : SUCCESS in 9m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/ba3f2e23db114f6ba4bbed98edae4d5b : SUCCESS in 1h 09m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b75cdf5adeff4f5f90d6a650055ec485 : SUCCESS in 1h 08m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/231e432e33a9417ca36754ee3e9d6ccc : SUCCESS in 1h 22m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f59ea9166ca84adba4e7997113b8b1bd : SUCCESS in 17m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4ba4ad20f7be4edabfb56df593024fb1 : SUCCESS in 15m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3866c7d0afa44a3bace830ad910c890a : SUCCESS in 20m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c4ffd419e17f4b88a239d4e7067550c3 : SUCCESS in 8m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/08b06375a3c34da08d037bdbadbe1e87 : SUCCESS in 5m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/40221de3ade64efc8acfd540c105f033 : SUCCESS in 17m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7b7972d7839847e2912faa44c7d5ac98 : SUCCESS in 16m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/efeab2e5134f40b0ac1c1e571da80eed : SUCCESS in 19m 47s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/530e5c16213d4a909bc8789f75f4157b : SUCCESS in 16m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c1d04bc240e2493ba28e59c4099a82da : SUCCESS in 16m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c7d5a8c73622416a990c7ed4b42a4911 : SUCCESS in 20m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/47e63910de884834830492d619cb163f : SUCCESS in 19m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d250f37e5ad145498b1d86cf20a8167c : SUCCESS in 17m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fb941aa1325742f3958b0751490391e3 : SUCCESS in 13m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/394841d0ff004c448f1a54f31c85cef7 : SUCCESS in 15m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d2fcff15bf714f17835230a9e5f73955 : SUCCESS in 6m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8c02ee2e5ead4450a05f987551a8a16e : SUCCESS in 16m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4b5eb16b8e99486c9f81f84b8a3d7814 : SUCCESS in 7m 47s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2463bf4255bd475eb274b820fc0a1fc5 : SUCCESS in 58m 12s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f7eac2d077704182b59f903dd613a0a9 : SUCCESS in 1h 08m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f7eb4094864a47c5936a9969971d77d3 : SUCCESS in 49m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/28cc1348f3dc404498777d61d49fe2be : SUCCESS in 54m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7ab8bc03695473487c82ef4e980dee8 : SUCCESS in 28m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e3cbee33b1e4f178c4140ddccaa2cf2 : SUCCESS in 4m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2ce8e73d06684ce7807a4808e0f2dc3a : SUCCESS in 29m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/888481a104624421b50f577c3baf3d7a : SUCCESS in 1h 03m 03s","accounts_in_message":[],"_revision_number":3},{"id":"28db277736ec55ac487a1fd72064f37a933c8e88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-27 17:04:28.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"847c490a1634ea72df8f02bc492e23c01f47b1e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 18:08:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce0a52b7d4444a23a6d46c1d74c2c77e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c9b08cd91344e7d9522d422cdddcb6c : SUCCESS in 14m 40s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/848605bdf9c549ebb7494ab45c0e33eb : SUCCESS in 16m 08s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd62ed59701b4130ae7d2fcb678b8a12 : SUCCESS in 17m 49s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7b91c38842584ac9832ae451cacf1769 : SUCCESS in 1h 01m 42s","accounts_in_message":[],"_revision_number":4},{"id":"c3b1f1ea0a2ddc303ad858825118dd184275cb6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 19:54:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/985339a886d54498b9a76353a9070f2b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fa39bf6f12f4ac88e5d903aca970775 : SUCCESS in 9m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/901cbae0cf084aa7b9e8b0ecd795d0ae : SUCCESS in 1h 09m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8c0c2f63928c46d3bcb7c6763eccaae7 : SUCCESS in 35m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d81706fd4c924990836798e20a8278d2 : SUCCESS in 47m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1cb86d442deb479981d142d26f872741 : SUCCESS in 30m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/02b1a06b1e4b466f804f3887b5360905 : SUCCESS in 19m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4967d77255e94cdcb64990d269b570aa : SUCCESS in 21m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4bd5dbab7eb84741bf823d491d55e0c2 : SUCCESS in 10m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d4221ac8db0c4d6a86501ba1262d21fa : SUCCESS in 6m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2d0844bcebdb4ca0824e3ecf507f74dc : SUCCESS in 18m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/075e0e74143b4161b876bc730b21774a : SUCCESS in 16m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/adddeee16a81485dae8246201d5ee9d0 : SUCCESS in 18m 31s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/04aea02a4ba94c23afe6e3f89364a073 : SUCCESS in 28m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6d03939147b24cddb62dd9b21dfb615c : SUCCESS in 28m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a5fb76bf4c43463aac1268fbe74786c1 : SUCCESS in 23m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a5c1a4889fc744168978aab22da207bb : SUCCESS in 16m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9864b2d97f1140e1a1d9d7a95914d7c5 : SUCCESS in 12m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a3b925f8b6ed435da0c71b97544cb0b9 : SUCCESS in 14m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4de9e7891fe3438d8d16d93da440fc23 : SUCCESS in 16m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c870c85ae3644d128ca75184735f459a : SUCCESS in 6m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/518a8209aa3847d597a51573db55af08 : SUCCESS in 13m 42s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/57188605449840c7ac7dbc7edde7b3b4 : SUCCESS in 9m 53s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f8045ab2b2044018bb82a50dbe31aa6a : SUCCESS in 42m 42s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1b2b98537b1b489589c13b5794925adb : SUCCESS in 1h 05m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d32554d5bbfc4b24a14d763248c4fe8a : SUCCESS in 58m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9f8090c4d18742bc9fc81117ba64266b : SUCCESS in 39m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ffd2a88ab604ab291bfd73a64fee1eb : SUCCESS in 37m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3681d1aa9eb14f2599a11e717da06c9a : SUCCESS in 5m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d3246898806b4fbe971fad890a40a9f2 : SUCCESS in 30m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7fc6851dbdf9459698f2681d569fe809 : SUCCESS in 1h 25m 26s","accounts_in_message":[],"_revision_number":4},{"id":"c9451c5ab12c47e5fa25b966397447994f82006e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-10 15:25:10.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"00b412cd9b919e14998919dbd6ebd58e48282e40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 16:46:47.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e6e1dcc5b6f4e9185ea93d50a85152a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a2a1be4d37c4359b4cf0180472ec998 : SUCCESS in 8m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/343cb0eca7164c4ebb23315e910dfebb : SUCCESS in 41m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/04801e233862454d86d12adff23a5fa7 : SUCCESS in 1h 00m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/789fd3d474da432e9829188447b5ec16 : SUCCESS in 1h 03m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4de9d63c981245dcbf6486aed55a6704 : SUCCESS in 18m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/59a7db29ad2d47d6a5964ddff8ed527b : SUCCESS in 24m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d64ded4db30d479ead6244114655c691 : SUCCESS in 21m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/32bc71759e1c404f87f0ec468ed9fde4 : SUCCESS in 8m 40s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f5eca8a172534f338e50d8c7242d326c : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8b7a3479292f45b3b7602f7d753e3f10 : SUCCESS in 19m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7afcd26dbd9848b3b1d582195fb63406 : SUCCESS in 18m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4b4896599b9441e78f780535e78b89fb : SUCCESS in 23m 03s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/fc5c356ddcfb40d2bdbbcc66c1bcd1df : SUCCESS in 16m 07s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/37b22da8660e4b19878375c666ca4a53 : SUCCESS in 18m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f72ebde117054e8b98c3e24e97568845 : SUCCESS in 14m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ac8a7a7d70b64ee08c4ca27382c66470 : SUCCESS in 18m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c10a989e06d44d3897b3e58264c9453c : SUCCESS in 16m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ec714a8a550e46689dae8da4caf84226 : SUCCESS in 15m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/da52ee48b3e64309bf8712ba110c3665 : SUCCESS in 15m 09s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/930a8d106d514ee486c78eccc4a96702 : SUCCESS in 16m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/08e1be678d844b93898496f26fe78fc2 : SUCCESS in 7m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/daa8440fb36249cb87597db4d5f52fb7 : SUCCESS in 16m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e0435ab437ac4f38b790f69cab934ece : SUCCESS in 7m 34s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4326aa2eaa1c48a1888fdec4c2ca7d56 : SUCCESS in 58m 00s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/2cd9f1dd2e5d407ea9993f7a88547015 : SUCCESS in 1h 10m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5ef8237ac21747d7846ae68bfa9e06ac : SUCCESS in 58m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/26724eecaf43498fb3ed622c7dadb1d2 : SUCCESS in 42m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa15a3a748634131aede4ac3a6ec18af : SUCCESS in 12m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae3326adad1f49f8a510cab0bc9280ee : SUCCESS in 5m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e38e801dc0834964bc68a936c3a32b08 : SUCCESS in 19m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c9cf377dd22d4b7cb6686bdd54363d90 : SUCCESS in 1h 06m 37s","accounts_in_message":[],"_revision_number":5},{"id":"ba5064619af6285a891b2d817dde41e2e16a4474","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 16:51:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a8f4ce68a374081a93923127ff6bc7d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/090a7d83507a407784b23274d5526311 : SUCCESS in 19m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3a2ed264631b44618239e36eef6aa075 : SUCCESS in 1h 24m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0df1e6b5c11b4a4d86d20555d0f6fb45 : SUCCESS in 21m 20s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7bdd0f214e1d4012a7db6661a308288c : SUCCESS in 18m 56s","accounts_in_message":[],"_revision_number":5},{"id":"b477cf5a7948f0d59a94511db295e6d31c972650","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-07-24 04:02:07.000000000","message":"Patch Set 5: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"06c84f8dcb130d22bac1130d9f218139032624ad","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-07-24 14:13:17.000000000","message":"Patch Set 5: Code-Review+2\n\n(9 comments)","accounts_in_message":[],"_revision_number":5},{"id":"83d7837c5ac053b8d3172dabe48465a963ce3375","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-07-24 14:13:54.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"e773cdc8d5205c20545e13126ca9794b78892aac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-07-24 14:15:23.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"23af43e45f1b15075f607d8ecfb54bd25886e2a1","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-07-24 14:15:41.000000000","message":"Patch Set 6: Code-Review-1 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"bde796187a5b2d9e7317ff5cf725730e5e4dc411","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-07-24 14:18:29.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4c621ef226546552630998f3e7d51fdd350e0a16","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 14:55:28.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/040045fc75ea4293b5ee1ef3ee460747\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b4b39876987c4dd4933e69df58a961b3 : SUCCESS in 16m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a54a57c7de244082a59e5ab31e70aa25 : SUCCESS in 38m 39s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/35e666c7604f454082680ae6943dfcfd : SUCCESS in 18m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d0b8bd6c1bb849d3b1f87baa9a608316 : SUCCESS in 38m 03s","accounts_in_message":[],"_revision_number":7},{"id":"e89272f1ad580f518315e5b3ebc001c234ef1bff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 15:29:39.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/877796629a5a4038b77a8627163a467a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/deefddea506d499a88cfdf873dad4553 : SUCCESS in 8m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/ccb23a139afe4fc79f0acfa5d7dff638 : SUCCESS in 1h 08m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/59360dc90a7b464789b9cdaf91ee4ae7 : SUCCESS in 1h 01m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a604cbc0e435487cbeffc3cb4e0bf8f2 : SUCCESS in 46m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/081fca0d60534728b34385fbcdebafd8 : SUCCESS in 23m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/636b87807939489a960d9d2a7fc2292f : SUCCESS in 20m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bfe87c7f9cc344afa98c4e8ffba818ef : SUCCESS in 13m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0e0ed07faa3b42c3b1666e75dab1f58d : SUCCESS in 8m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/58bb31e9ac264079a44aaba06fe2e7d8 : SUCCESS in 5m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/51efc506a8c348c9b8b0305df89131ba : SUCCESS in 17m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7601dd185a574dca8895a048331477cd : SUCCESS in 24m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/85db9aac9b4d417c8f7b9e4299c6ca47 : SUCCESS in 16m 29s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5ef64ac8a0c74e50b5bcf06083665390 : SUCCESS in 13m 59s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/15881952d7584ec5b0cd74e8726ecafd : SUCCESS in 15m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c9ce29728c2940be888fd566ca7cfac1 : SUCCESS in 24m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c596c389d1f24717a16dd826daf34f08 : SUCCESS in 25m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7a1dc38438224cbcbb3204cf68f16690 : SUCCESS in 29m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/db5c6d44a1644618aae284dc5e93f4f8 : SUCCESS in 15m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/27dd32f0dbd349b4bd9b12a4322daf9c : SUCCESS in 22m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6718647a73de452b94b1efccf01f94f2 : SUCCESS in 15m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/783defd15ecb48d78ccfcc4c78f5220a : SUCCESS in 7m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/46a7c8958ad44b62928a842ebeefeb77 : SUCCESS in 22m 35s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/dea110c76ae747f6b743272b498d66d2 : SUCCESS in 9m 27s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a35a83e9c1b241dc8873be425cfe1a4b : SUCCESS in 1h 06m 42s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5ddda243fd9b456db3691076a1ff66c8 : SUCCESS in 59m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/57f2dd3c8ec54ac6b9bc89b3000a70a2 : SUCCESS in 50m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/24d9df9a0dc046a69352c7a6dea7a9b2 : SUCCESS in 58m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d417d5a1ab984ba9b9761bb1f6b2eb0f : SUCCESS in 19m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ccbdbd4d16548e9a36c13555c7d4a77 : SUCCESS in 5m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9fe68b2d9db74d349fade011bb10c461 : SUCCESS in 28m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/95a4ed9c412b499bb5320e5879ca9c62 : SUCCESS in 54m 17s","accounts_in_message":[],"_revision_number":7},{"id":"742e13b8a6ff965257b1a98d082b40bd7371e97d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 15:30:20.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"cace721111f7342b0aecd0ace59085dd17157370","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-24 15:37:08.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4a91b379ed154c1c2fef5072f364386555537982","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-07-24 16:18:45.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"7b37e6349386388019169e1efeb4826c2cb25780","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 22:04:28.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5b2f51c0a344fcf952620ec37335b30\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6b4e460f97448868756b59c3b2fcb51 : SUCCESS in 8m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/617e26f185ed430fb00c7583d45cb774 : SUCCESS in 51m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f5c12c628fad4d3ea949f1a2049cb936 : SUCCESS in 44m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3a463957d2b843aeb70d05c1bcfbd3be : SUCCESS in 56m 04s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/1b861e79a3d049728169373e94c9e5d4 : SUCCESS in 8m 03s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/41c93daeb61441e0bba2887991a895a3 : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/40fa55e2598e4950aaccb457e96bba10 : SUCCESS in 27m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5fe47651212745159a041b761c2e428c : SUCCESS in 17m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bc454ff4ad5b4006bbac12666e7af603 : SUCCESS in 21m 27s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5aa5e45c995a4e28919a9a5746cef43a : SUCCESS in 25m 49s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8e47dc0146e64a1da2bf430356b1d33b : SUCCESS in 23m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/78bda96fdd2b4521850d38996f4bf92c : SUCCESS in 18m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3b010a92a6054c15936e5f9205174151 : SUCCESS in 14m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c00d0747aa144863af784f3f8804a61f : SUCCESS in 18m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8992c70efec748e79587c84136fb9e90 : SUCCESS in 14m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b5c214a521b84c7c930bfebc98cf43ce : SUCCESS in 14m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/27bdb7dc48af4f829cd941ac0c587f04 : SUCCESS in 12m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0a23081af4cb45e3a04c76ed91485bd6 : SUCCESS in 5m 56s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/68e231ba2700476f80e30be3c2046d05 : SUCCESS in 6m 40s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/093e50cda173473ea83bd38c1031b0f8 : SUCCESS in 42m 43s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/97d5fb18873a4276942d8d01e584809c : SUCCESS in 54m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/42bead6114044ea29935736e3ec2c5bc : SUCCESS in 54m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5d06134ab77a43ebba9e2131dbb27907 : SUCCESS in 38m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a603c11430754604a9b95d039ab3e27f : SUCCESS in 28m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/179e216752a543eb8a923910024de6d3 : SUCCESS in 4m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/efaf7efc17924ecc81f0fc1b5b2c0515 : SUCCESS in 53m 28s","accounts_in_message":[],"_revision_number":7},{"id":"2d8639fc4e7017d2ad1ab929723abfc6152eeecd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 22:04:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"849941da05f8390008cedd94dbf23654eaec69bd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 22:05:41.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4c03350fec84e27a70c63472a556bc9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b77925f991ef458bb58b216f681654b4 : SUCCESS in 52s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f2e07a211c6c4ee39687ca18495e4a43 : SUCCESS in 35s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/67bbb8b56d104416a9065f4420afe9dd : SUCCESS in 29s (non-voting)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"369a72c4cf48cf8637e508155cbef8471c69a39a","revisions":{"67738885294be92ea2d7cff87c8298bb594da214":{"kind":"REWORK","_number":1,"created":"2023-06-23 13:42:30.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/23/886823/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/886823/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/886823/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/886823/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/886823/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/886823/1"}}},"commit":{"parents":[{"commit":"b8dc7b3d353462adb16911f757918bf55ab99277","subject":"unify duplicate code in replication, EC GET paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b8dc7b3d353462adb16911f757918bf55ab99277"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-23 12:21:02.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-23 13:40:36.000000000","tz":60},"subject":"proxy: remove client_chunk_size and skip_bytes from GetOrHeadHandler","message":"proxy: remove client_chunk_size and skip_bytes from GetOrHeadHandler\n\nThe client_chunk_size attribute was introduced into GetOrHeadHandler\nfor EC support [1]. It was only ever not None for an\nECObjectController. The ECObjectController stopped using\nGetOrHeadHandler when the ECFragGetter class was introduced [2], but\nthe EC specific code was not expunged from GetOrHeadHandler. In [3]\nthe ECFragGetter client_chunk_size was renamed to fragment_size to\nbetter reflect what it represented.\n\nThe skip_bytes attribute was similarly introduced for EC support. It\nis only ever non-zero if client_chunk_size is an int. For EC,\nskip_bytes is used to undo the effect of expanding the backend\nrange(s) to fetch whole fragments: the range(s) of decoded bytes\nreturned to the client may need to be narrower than the backend\nranges. There is no equivalent requirement for replicated GETs.\n\nThe elimination of client_chunk_size and skip_bytes simplifies the\nyielding of chunks from the GetOrHeadHandler response iter.\n\nRelated-Change:\n[1] I9c13c03616489f8eab7dcd7c5f21237ed4cb6fd2\n[2] I0dc5644a84ededee753e449e053e6b1786fdcf32\n[3] Ie1efaab3bd0510275d534b5c023cb73c98bec90d\n\nChange-Id: I31ed36d32682469e3c5ca8bf9a2b383568d63c72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67738885294be92ea2d7cff87c8298bb594da214"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67738885294be92ea2d7cff87c8298bb594da214"}]},"branch":"refs/heads/master"},"9d4c76250addafd7de3989a666045ac00be98d09":{"kind":"REWORK","_number":2,"created":"2023-06-26 09:26:45.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/23/886823/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/886823/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/886823/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/886823/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/886823/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/886823/2"}}},"commit":{"parents":[{"commit":"e290d47c435627082b3825b67e27ea3a3c99f01f","subject":"unify duplicate code in replication, EC GET paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e290d47c435627082b3825b67e27ea3a3c99f01f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-23 12:21:02.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-26 09:17:03.000000000","tz":60},"subject":"proxy: remove client_chunk_size and skip_bytes from GetOrHeadHandler","message":"proxy: remove client_chunk_size and skip_bytes from GetOrHeadHandler\n\nThe client_chunk_size attribute was introduced into GetOrHeadHandler\nfor EC support [1]. It was only ever not None for an\nECObjectController. The ECObjectController stopped using\nGetOrHeadHandler when the ECFragGetter class was introduced [2], but\nthe EC specific code was not expunged from GetOrHeadHandler. In [3]\nthe ECFragGetter client_chunk_size was renamed to fragment_size to\nbetter reflect what it represented.\n\nThe skip_bytes attribute was similarly introduced for EC support. It\nis only ever non-zero if client_chunk_size is an int. For EC,\nskip_bytes is used to undo the effect of expanding the backend\nrange(s) to fetch whole fragments: the range(s) of decoded bytes\nreturned to the client may need to be narrower than the backend\nranges. There is no equivalent requirement for replicated GETs.\n\nThe elimination of client_chunk_size and skip_bytes simplifies the\nyielding of chunks from the GetOrHeadHandler response iter.\n\nRelated-Change:\n[1] I9c13c03616489f8eab7dcd7c5f21237ed4cb6fd2\n[2] I0dc5644a84ededee753e449e053e6b1786fdcf32\n[3] Ie1efaab3bd0510275d534b5c023cb73c98bec90d\n\nChange-Id: I31ed36d32682469e3c5ca8bf9a2b383568d63c72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d4c76250addafd7de3989a666045ac00be98d09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d4c76250addafd7de3989a666045ac00be98d09"}]},"branch":"refs/heads/master"},"db947cf34d24914608c4febabc09b143cf021954":{"kind":"REWORK","_number":3,"created":"2023-06-26 11:12:36.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/23/886823/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/886823/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/886823/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/886823/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/886823/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/886823/3"}}},"commit":{"parents":[{"commit":"68e48c8684aa6a7f4673b1f8750b06d1e8109ebd","subject":"proxy: error limit the correct node when resuming GETs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/68e48c8684aa6a7f4673b1f8750b06d1e8109ebd"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-23 12:21:02.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-26 11:04:39.000000000","tz":60},"subject":"proxy: remove client_chunk_size and skip_bytes from GetOrHeadHandler","message":"proxy: remove client_chunk_size and skip_bytes from GetOrHeadHandler\n\nThe client_chunk_size attribute was introduced into GetOrHeadHandler\nfor EC support [1]. It was only ever not None for an\nECObjectController. The ECObjectController stopped using\nGetOrHeadHandler when the ECFragGetter class was introduced [2], but\nthe EC specific code was not expunged from GetOrHeadHandler. In [3]\nthe ECFragGetter client_chunk_size was renamed to fragment_size to\nbetter reflect what it represented.\n\nThe skip_bytes attribute was similarly introduced for EC support. It\nis only ever non-zero if client_chunk_size is an int. For EC,\nskip_bytes is used to undo the effect of expanding the backend\nrange(s) to fetch whole fragments: the range(s) of decoded bytes\nreturned to the client may need to be narrower than the backend\nranges. There is no equivalent requirement for replicated GETs.\n\nThe elimination of client_chunk_size and skip_bytes simplifies the\nyielding of chunks from the GetOrHeadHandler response iter.\n\nRelated-Change:\n[1] I9c13c03616489f8eab7dcd7c5f21237ed4cb6fd2\n[2] I0dc5644a84ededee753e449e053e6b1786fdcf32\n[3] Ie1efaab3bd0510275d534b5c023cb73c98bec90d\n\nChange-Id: I31ed36d32682469e3c5ca8bf9a2b383568d63c72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/db947cf34d24914608c4febabc09b143cf021954"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/db947cf34d24914608c4febabc09b143cf021954"}]},"branch":"refs/heads/master"},"684b28042d1eb8e135e1e8ec6adf64f4a807f3ca":{"kind":"REWORK","_number":4,"created":"2023-06-27 17:04:28.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/23/886823/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/886823/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/886823/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/886823/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/886823/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/886823/4"}}},"commit":{"parents":[{"commit":"68e48c8684aa6a7f4673b1f8750b06d1e8109ebd","subject":"proxy: error limit the correct node when resuming GETs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/68e48c8684aa6a7f4673b1f8750b06d1e8109ebd"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-23 12:21:02.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-27 17:04:12.000000000","tz":60},"subject":"proxy: remove client_chunk_size and skip_bytes from GetOrHeadHandler","message":"proxy: remove client_chunk_size and skip_bytes from GetOrHeadHandler\n\nThe client_chunk_size attribute was introduced into GetOrHeadHandler\nfor EC support [1]. It was only ever not None for an\nECObjectController. The ECObjectController stopped using\nGetOrHeadHandler when the ECFragGetter class was introduced [2], but\nthe EC specific code was not expunged from GetOrHeadHandler. In [3]\nthe ECFragGetter client_chunk_size was renamed to fragment_size to\nbetter reflect what it represented.\n\nThe skip_bytes attribute was similarly introduced for EC support. It\nis only ever non-zero if client_chunk_size is an int. For EC,\nskip_bytes is used to undo the effect of expanding the backend\nrange(s) to fetch whole fragments: the range(s) of decoded bytes\nreturned to the client may need to be narrower than the backend\nranges. There is no equivalent requirement for replicated GETs.\n\nThe elimination of client_chunk_size and skip_bytes simplifies the\nyielding of chunks from the GetOrHeadHandler response iter.\n\nRelated-Change:\n[1] I9c13c03616489f8eab7dcd7c5f21237ed4cb6fd2\n[2] I0dc5644a84ededee753e449e053e6b1786fdcf32\n[3] Ie1efaab3bd0510275d534b5c023cb73c98bec90d\n\nChange-Id: I31ed36d32682469e3c5ca8bf9a2b383568d63c72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/684b28042d1eb8e135e1e8ec6adf64f4a807f3ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/684b28042d1eb8e135e1e8ec6adf64f4a807f3ca"}]},"branch":"refs/heads/master"},"5bd21ba912f42ee6893ec5614609344da9e19f95":{"kind":"REWORK","_number":5,"created":"2023-07-10 15:25:10.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/23/886823/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/886823/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/886823/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/886823/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/886823/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/886823/5"}}},"commit":{"parents":[{"commit":"1b7cf29476a2f77ceab1a082100b0c7e7cb8dc93","subject":"Object-server: backfill unit test coverage for keep_cache_private","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b7cf29476a2f77ceab1a082100b0c7e7cb8dc93"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-23 12:21:02.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-07-10 15:19:41.000000000","tz":60},"subject":"proxy: remove client_chunk_size and skip_bytes from GetOrHeadHandler","message":"proxy: remove client_chunk_size and skip_bytes from GetOrHeadHandler\n\nThe client_chunk_size attribute was introduced into GetOrHeadHandler\nfor EC support [1]. It was only ever not None for an\nECObjectController. The ECObjectController stopped using\nGetOrHeadHandler when the ECFragGetter class was introduced [2], but\nthe EC specific code was not expunged from GetOrHeadHandler. In [3]\nthe ECFragGetter client_chunk_size was renamed to fragment_size to\nbetter reflect what it represented.\n\nThe skip_bytes attribute was similarly introduced for EC support. It\nis only ever non-zero if client_chunk_size is an int. For EC,\nskip_bytes is used to undo the effect of expanding the backend\nrange(s) to fetch whole fragments: the range(s) of decoded bytes\nreturned to the client may need to be narrower than the backend\nranges. There is no equivalent requirement for replicated GETs.\n\nThe elimination of client_chunk_size and skip_bytes simplifies the\nyielding of chunks from the GetOrHeadHandler response iter.\n\nRelated-Change:\n[1] I9c13c03616489f8eab7dcd7c5f21237ed4cb6fd2\n[2] I0dc5644a84ededee753e449e053e6b1786fdcf32\n[3] Ie1efaab3bd0510275d534b5c023cb73c98bec90d\n\nChange-Id: I31ed36d32682469e3c5ca8bf9a2b383568d63c72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5bd21ba912f42ee6893ec5614609344da9e19f95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5bd21ba912f42ee6893ec5614609344da9e19f95"}]},"branch":"refs/heads/master"},"7417b538092e8212bb21d2470cc9945dae9ce9e5":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2023-07-24 14:13:54.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/23/886823/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/886823/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/886823/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/886823/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/886823/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/886823/6"}}},"commit":{"parents":[{"commit":"1b7cf29476a2f77ceab1a082100b0c7e7cb8dc93","subject":"Object-server: backfill unit test coverage for keep_cache_private","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b7cf29476a2f77ceab1a082100b0c7e7cb8dc93"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-23 12:21:02.000000000","tz":60},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-07-24 14:13:54.000000000","tz":0},"subject":"proxy: remove client_chunk_size and skip_bytes from GetOrHeadHandler","message":"proxy: remove client_chunk_size and skip_bytes from GetOrHeadHandler\n\nThe client_chunk_size attribute was introduced into GetOrHeadHandler\nfor EC support [1]. It was only ever not None for an\nECObjectController. The ECObjectController stopped using\nGetOrHeadHandler for Object GET when the ECFragGetter class was introduced [2], but the EC specific code was not expunged from GetOrHeadHandler. In [3] the ECFragGetter client_chunk_size was renamed to fragment_size to better reflect what it represented.\n\nThe skip_bytes attribute was similarly introduced for EC support. It\nis only ever non-zero if client_chunk_size is an int. For EC,\nskip_bytes is used to undo the effect of expanding the backend\nrange(s) to fetch whole fragments: the range(s) of decoded bytes\nreturned to the client may need to be narrower than the backend\nranges. There is no equivalent requirement for replicated GETs.\n\nThe elimination of client_chunk_size and skip_bytes simplifies the\nyielding of chunks from the GetOrHeadHandler response iter.\n\nRelated-Change:\n[1] I9c13c03616489f8eab7dcd7c5f21237ed4cb6fd2\n[2] I0dc5644a84ededee753e449e053e6b1786fdcf32\n[3] Ie1efaab3bd0510275d534b5c023cb73c98bec90d\n\nChange-Id: I31ed36d32682469e3c5ca8bf9a2b383568d63c72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7417b538092e8212bb21d2470cc9945dae9ce9e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7417b538092e8212bb21d2470cc9945dae9ce9e5"}]},"branch":"refs/heads/master","description":"Edit commit message"},"369a72c4cf48cf8637e508155cbef8471c69a39a":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2023-07-24 14:15:23.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/23/886823/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/886823/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/886823/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/886823/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/886823/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/886823/7"}}},"commit":{"parents":[{"commit":"1b7cf29476a2f77ceab1a082100b0c7e7cb8dc93","subject":"Object-server: backfill unit test coverage for keep_cache_private","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b7cf29476a2f77ceab1a082100b0c7e7cb8dc93"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-23 12:21:02.000000000","tz":60},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-07-24 14:15:12.000000000","tz":-300},"subject":"proxy: remove client_chunk_size and skip_bytes from GetOrHeadHandler","message":"proxy: remove client_chunk_size and skip_bytes from GetOrHeadHandler\n\nThe client_chunk_size attribute was introduced into GetOrHeadHandler\nfor EC support [1]. It was only ever not None for an\nECObjectController. The ECObjectController stopped using\nGetOrHeadHandler for Object GET when the ECFragGetter class was\nintroduced [2], but the EC specific code was not expunged from\nGetOrHeadHandler. In [3] the ECFragGetter client_chunk_size was renamed\nto fragment_size to better reflect what it represented.\n\nThe skip_bytes attribute was similarly introduced for EC support. It\nis only ever non-zero if client_chunk_size is an int. For EC,\nskip_bytes is used to undo the effect of expanding the backend\nrange(s) to fetch whole fragments: the range(s) of decoded bytes\nreturned to the client may need to be narrower than the backend\nranges. There is no equivalent requirement for replicated GETs.\n\nThe elimination of client_chunk_size and skip_bytes simplifies the\nyielding of chunks from the GetOrHeadHandler response iter.\n\nRelated-Change:\n[1] I9c13c03616489f8eab7dcd7c5f21237ed4cb6fd2\n[2] I0dc5644a84ededee753e449e053e6b1786fdcf32\n[3] Ie1efaab3bd0510275d534b5c023cb73c98bec90d\n\nChange-Id: I31ed36d32682469e3c5ca8bf9a2b383568d63c72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/369a72c4cf48cf8637e508155cbef8471c69a39a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/369a72c4cf48cf8637e508155cbef8471c69a39a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
