)]}'
{"id":"openstack%2Fswift~886864","triplet_id":"openstack%2Fswift~master~Ibd405b79820cedaf9cb02906b8c13c8d34f4d7ec","project":"openstack/swift","branch":"master","topic":"unify-get","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-06-27 04:42:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibd405b79820cedaf9cb02906b8c13c8d34f4d7ec","subject":"proxy: error limit the correct node when resuming GETs","status":"MERGED","created":"2023-06-23 17:23:49.000000000","updated":"2023-06-27 04:44:17.000000000","submitted":"2023-06-27 04:42:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":8,"has_review_started":true,"submission_id":"886864-unify-get","meta_rev_id":"55c75a489905a440147bd265cb2ce501ac913910","_number":886864,"virtual_id_number":886864,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-27 04:42:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2023-06-27 02:49:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2023-06-27 02:49:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-23 17:44:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-23 18:40:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-27 02:49:23.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"eb239fd6a26465e943e6405e225b82fe13182ba3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-23 17:23:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60d4c80b0ade29c36ba921cc02fea5d757160d23","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-23 17:28:33.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3f237f1abd9d31722622cf47238bf677b5a31415","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 17:44:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d953885b0f634f45878c09efba16e123\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/71533df2a0344c76a64ff327a9021f9c : SUCCESS in 14m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/086483c1751d4bbcb3d9845be42f7cef : SUCCESS in 14m 34s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e6ad8ec8d6bd41e9ab27fa2121a8209b : SUCCESS in 18m 05s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/888194c439644996a9cc92a880cae1cf : SUCCESS in 19m 39s","accounts_in_message":[],"_revision_number":1},{"id":"2e5adbe47f9c86cf7cf0672391c8a3aed9f18c91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 18:40:59.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c2e9a43c40b4c9b8e65d419c616e832\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d7de09f73864683b4c49e153e656055 : SUCCESS in 8m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/78d3572a57784b10a7aa321f8ce3caeb : SUCCESS in 46m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/109338ea75874a0f9b9bb1cbf5e0182d : SUCCESS in 42m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31bb9f6f589047ee85729d9bb2286e8b : SUCCESS in 49m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0174be42ce96417d95b3f1464ac97ab1 : SUCCESS in 19m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d0b60bd2a3c44e31852f3e56c51ada82 : SUCCESS in 17m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/02a7bc52d0b34699aad15e8b15d28066 : SUCCESS in 21m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a4e666d1302444afbe16d1a37b6eae07 : POST_FAILURE in 37m 33s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/17a647cf0bbe4a0b9f630bc85e04c74e : POST_FAILURE in 35m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/270676c703784e538560496cce8b7d04 : SUCCESS in 29m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a3317772601c43dfa1d2c4040364df14 : SUCCESS in 16m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7beb5642b1674b0fbe2de6fd693537b2 : SUCCESS in 13m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/64712ff5f6374a5da3079705f994d7c8 : SUCCESS in 18m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/886be15bbda248419231f4984fce8a2f : SUCCESS in 21m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/64332da55aeb4953ac7e37f637634387 : SUCCESS in 16m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/15241b48323c4c6691c8e233ea56f4b2 : SUCCESS in 19m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3068bba799bf481f8130060028a3ad41 : SUCCESS in 12m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/66942e15fa7c47b2ac88fbe8b908cd07 : SUCCESS in 14m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/929c9c4c37f24f648914c01ab7ace2f6 : SUCCESS in 15m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f39ece44568b4d8bbf034fd33b73a7eb : SUCCESS in 6m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c3175806159444d4beb6157512de0871 : SUCCESS in 20m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bac43b321cda4d8ca987cca9832217ae : SUCCESS in 7m 24s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/833cc84626aa4aa69151fd89cac9b9d0 : SUCCESS in 59m 32s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/2dd7dcf53a254619850cd35be3e088ac : SUCCESS in 1h 03m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/23e58e3e46754a89ac99c23d87aa65a2 : SUCCESS in 1h 00m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fe9d2ccb174e433fb7f59a698665da5c : SUCCESS in 55m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8762f2c8f4b24c9bade95c35b8bf1a15 : SUCCESS in 16m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73253a0e41d94abaacf73ae89a007541 : FAILURE in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/33f9be9e6901462bbe166ff9d7655891 : SUCCESS in 22m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41eaed6d168e4e05912510cd4b696b70 : SUCCESS in 1h 11m 36s","accounts_in_message":[],"_revision_number":1},{"id":"4d585e2a4a434fbe268e6115dcb61ab17eba31d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-26 11:12:36.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b3209f436545fd11596d977342e4d3d327e45da8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 11:36:59.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f83f8dd34a24461980e605ce8d819066\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f1fcabde8fb1495da8bd37a68af42387 : NODE_FAILURE Node request 300-0021543712 failed in 0s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b954b521f54e44febd1bf035cd049d72 : NODE_FAILURE Node request 300-0021543713 failed in 0s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3b420a4988474c22b9c0e16cbefbb27f : SUCCESS in 20m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/740d03dced9b42338a483ad28f6bbd8f : SUCCESS in 20m 03s","accounts_in_message":[],"_revision_number":2},{"id":"cce019baee5264eca436964d9d629cc47cb58539","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 15:37:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4872315aaea34b62830da3e3785418dd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2b9e735276c4683a5f90e6107e09178 : SUCCESS in 8m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/33dab5509739476ea031fb8cf511d9bf : SUCCESS in 1h 16m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a644a42a331a4b0686647c181b78ff16 : SUCCESS in 56m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ebbda35305124b99b88a8363d425556d : SUCCESS in 1h 02m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7d20cd20bb2b44fd8852dcaa4a7338ec : SUCCESS in 29m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/77a2f4217c224586a346970b96d094e3 : SUCCESS in 17m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/024ddeb1e5b84dac862e1673e0b3173b : SUCCESS in 18m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8cc6154dc0fa41ec8459eb0dfe14d7a2 : SUCCESS in 9m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/75392f79237b41c7b5c8197396ca897a : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5adbfaba4e84417eaa454ff631fd6de2 : SUCCESS in 24m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/87087f413a1c42eda237724c6391f047 : SUCCESS in 18m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/11af53b115f3479eac4a3c250f6e89c6 : SUCCESS in 21m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/11ed1b877ba840f6add79a9982b5a22b : SUCCESS in 17m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/50009ba6bbc149b1abbb32ddd3eee44e : SUCCESS in 23m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2274015d07f544da9c3f1d613f783df8 : SUCCESS in 20m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e047a3ea09c94b289b3d9d6973e42036 : SUCCESS in 26m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a9d287f942384a91aa9dc5f71ae63663 : SUCCESS in 14m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/66d6973a30cb47c0aacf28958329d2bd : SUCCESS in 23m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/08254a7a198b4029a7fcdbf99c176535 : SUCCESS in 21m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9b26b7a39ecf4218b9bb1f06e8ab864b : SUCCESS in 6m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/712d5b5649f24bd991fe4d9e6c8f69c7 : SUCCESS in 18m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f204662e2c384a02b021d6c06b47a878 : SUCCESS in 8m 31s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6ae66e0d2b9b416e8d2f2bff8728c49f : SUCCESS in 55m 17s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b259bf7bea2f4db1b8583c3e7c85f704 : SUCCESS in 58m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3635f1474f9f4828ad0e4c54fec4777c : SUCCESS in 52m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c877b49bcea44a528a1cae3ea265bcfd : SUCCESS in 54m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77a155901f3f4452b1c26341acae121d : SUCCESS in 20m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b43ad1d0a2f4ba4a265e8f8df68276e : SUCCESS in 4m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6bcb52d1a72e4efeb3424e635273e87a : SUCCESS in 38m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5602c41982024e10a6194d9c5430d84e : SUCCESS in 57m 03s","accounts_in_message":[],"_revision_number":2},{"id":"7c07033c3b1d29c5a04933e3d30a2bd4383cfaa4","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-06-27 02:49:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":2},{"id":"96d386db96f06e8a72c985039fd52d719dc75f9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 02:50:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b53f81f585220ab5d04fc09070caff4e9a605516","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 04:42:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d8a97acbb7744fe9fed7ec86b810288\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36dd493012fe4e05ae16b2cb4dcf0e48 : SUCCESS in 8m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/de975cab7aa74ce99ab5d743b4064dc8 : SUCCESS in 1h 24m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7ca71cdbb1f44dd7b8e7c8fe8c9946ca : SUCCESS in 46m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9efdf507f11149eebe6c1b25c30e861a : SUCCESS in 59m 04s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/051acd85711947bda5b694afc64f2d07 : SUCCESS in 7m 39s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/ca9f194268ae44f2aa803a9576deb483 : SUCCESS in 7m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/36803928285e408991e952bfeffbb115 : SUCCESS in 17m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4ad1f18d5c8b47c39e74c0ba8cafdfa3 : SUCCESS in 24m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7b3929f768fa4d4583c81806b60795d0 : SUCCESS in 15m 03s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/80c8159d9a1d47df98985dcd882ef07c : SUCCESS in 16m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f3713ca9e095491c8fd22080bafd897a : SUCCESS in 15m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/133fcc53add3461588df997c0022aedc : SUCCESS in 18m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fec63667571b447e8ceac28c2302c334 : SUCCESS in 18m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7c2144279c5e4ec58759ba9c45de38dd : SUCCESS in 15m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fb52ff6cd1bd4c64ac992ab2fb146d37 : SUCCESS in 14m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4032b57a74de4c7b852dc08b1c25c535 : SUCCESS in 18m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2778bde1e6264946aaf553ff44a0e11b : SUCCESS in 7m 02s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3dee5d402f704ee79a39100fd422c7b3 : SUCCESS in 7m 35s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ca7be7213e2145949c0baa4c04cf0e23 : SUCCESS in 51m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/34ff903d09f24f5f9ca77dc88d611b78 : SUCCESS in 1h 07m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5586538561e9449181696442b18a406c : SUCCESS in 42m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/65afe085563e4227917513e5d2d651b0 : SUCCESS in 58m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6cd0b7c9a73144c8ab5470f5ca7fc7d5 : SUCCESS in 19m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13b54795dca64ab6aee2b682645bf0f3 : SUCCESS in 4m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f03bd660f3444311ac1035d187c004eb : SUCCESS in 49m 05s","accounts_in_message":[],"_revision_number":2},{"id":"a46b996e661bd253545dc39da4ae700bbae1039a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 04:42:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"55c75a489905a440147bd265cb2ce501ac913910","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 04:44:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c58a1bc30524144a4620a618b22eea1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71dc680ab5fc44229f7e4774f4567062 : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/52c8e3a94d014ff7ac737f0f4fe47d2c : SUCCESS in 39s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/9ffc23295af64b0ab3ca961813a7657e : SUCCESS in 31s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"68e48c8684aa6a7f4673b1f8750b06d1e8109ebd","revisions":{"594253fe03e19beb8d65c77b5e36e0eb3ef5dc40":{"kind":"REWORK","_number":1,"created":"2023-06-23 17:23:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/64/886864/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/64/886864/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/886864/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/886864/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/886864/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/64/886864/1"}}},"commit":{"parents":[{"commit":"b8dc7b3d353462adb16911f757918bf55ab99277","subject":"unify duplicate code in replication, EC GET paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b8dc7b3d353462adb16911f757918bf55ab99277"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-23 17:19:30.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-23 17:23:37.000000000","tz":60},"subject":"proxy: error limit the correct node when resuming GETs","message":"proxy: error limit the correct node when resuming GETs\n\nPreviously, when resuming a get using a new source node, the new\nsource node would be error limited rather than the old node. This\npatch adds a helper method to unify all cases when an old node should\nbe error limited, and ensures that the old rather than new node is\nlimited.\n\nChange-Id: Ibd405b79820cedaf9cb02906b8c13c8d34f4d7ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/594253fe03e19beb8d65c77b5e36e0eb3ef5dc40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/594253fe03e19beb8d65c77b5e36e0eb3ef5dc40"}]},"branch":"refs/heads/master"},"68e48c8684aa6a7f4673b1f8750b06d1e8109ebd":{"kind":"REWORK","_number":2,"created":"2023-06-26 11:12:36.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/64/886864/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/64/886864/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/886864/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/886864/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/886864/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/64/886864/2"}}},"commit":{"parents":[{"commit":"e290d47c435627082b3825b67e27ea3a3c99f01f","subject":"unify duplicate code in replication, EC GET paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e290d47c435627082b3825b67e27ea3a3c99f01f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-23 17:19:30.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-26 11:04:33.000000000","tz":60},"subject":"proxy: error limit the correct node when resuming GETs","message":"proxy: error limit the correct node when resuming GETs\n\nIn some cases, when resuming a get using a new source node, the new\nsource node would be error limited rather than the old node. One case\nwas previously fixed [1], but other erroneous cases remain.\n\nThis patch adds a helper method to unify handling of all cases when an\nold node should be error limited, and ensures that the old rather than\nnew node is limited.\n\n[1] Related-Change: I02f9c0ffd879423211b87f7ac5acde762933f47d\nChange-Id: Ibd405b79820cedaf9cb02906b8c13c8d34f4d7ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/68e48c8684aa6a7f4673b1f8750b06d1e8109ebd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/68e48c8684aa6a7f4673b1f8750b06d1e8109ebd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
