)]}'
{"id":"openstack%2Fswift~886868","triplet_id":"openstack%2Fswift~master~I02591038a089ce43684f0245665349e40381d75a","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2024-06-27 21:46:36.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-06-27 17:33:15.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I02591038a089ce43684f0245665349e40381d75a","subject":"obj: Close replicator/reconstructor connections","status":"MERGED","created":"2023-06-23 19:09:01.000000000","updated":"2024-06-27 21:48:09.000000000","submitted":"2024-06-27 21:46:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886868","meta_rev_id":"d12f8fe9904caba65379920f56151655204d4b45","_number":886868,"virtual_id_number":886868,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-27 21:46:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-27 20:39:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-27 20:39:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}],"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-23 19:18:17.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"},{"updated":"2023-06-23 20:15:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-23 20:54:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f4aae4a72983105314f8084f93ee34133e76b6a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-23 19:09:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32200ede4cb928a48bc212c3d5ebe4c9502a60ce","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-06-23 19:18:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4a0dcca25084c25c0086fdb7a1267ed604645db0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 20:15:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a8d2c6833aa4b8abad7c14c59e935a5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e3761c85814b441880cc12cdd15e7388 : SUCCESS in 14m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a9bd8257c6884db3ab889380f2d2eac7 : SUCCESS in 15m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6d44e2862d8f45c68a0826036760a078 : SUCCESS in 21m 48s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b10f2f3508b9468b9968bc61279d64f0 : SUCCESS in 1h 03m 51s","accounts_in_message":[],"_revision_number":1},{"id":"f9018f238839d29f0224f6db5604926dd720f549","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 20:54:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18116c38f207468caad2ed3e479c124b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94ca29eefeef4f0688b283b50872d9d0 : SUCCESS in 7m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/03378bfd26e64e069036b6e3b2514499 : SUCCESS in 47m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6e7c7fc98aeb493eb70ff8857f84f7a6 : SUCCESS in 41m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eadaf2f2a8904e63a365f2df7fc772be : SUCCESS in 52m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4e4c372b94c4421098465cca05ca352c : SUCCESS in 17m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/48389d6962a04887b4549351f449f4ec : SUCCESS in 22m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cd8a54038a944c2bbf00a7c099abdbe3 : SUCCESS in 19m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/996f036b289342ada9831e29f317c0d3 : POST_FAILURE in 37m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a3321add76f7465aa9920fec387c7759 : POST_FAILURE in 35m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9485e44660e1449dac99cde5d89a3297 : SUCCESS in 17m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dd2b2263ae08424ba621b1ac68290737 : SUCCESS in 15m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/98394728ccd54a2a902258c21ad0c6a9 : SUCCESS in 19m 22s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e0e1296e92d64b328c5d7e2bc64f90e0 : SUCCESS in 38m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8fb3e593e9ba4c38b6d38be74de67d17 : SUCCESS in 16m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/863285b1c1ce40e29aaa066e5d471cf4 : SUCCESS in 16m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d9852eacc6164b97ac4037c6014715fe : SUCCESS in 17m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5058c3996f2740adab3a39d461520980 : SUCCESS in 19m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/317446f5f22845b0b8208c954c051eb7 : SUCCESS in 24m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8788aafd51dd49d68c44097d3ac07770 : SUCCESS in 20m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3d02746363ba45eca0b53bb78cee7917 : SUCCESS in 7m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d6e64fbd74b44f2b952a4d225a27722a : SUCCESS in 18m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b7d1f8a40ff343668eab92659f2572f4 : SUCCESS in 8m 27s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6b322959a1b94e8a990490d4d49b17d1 : SUCCESS in 1h 03m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a4c90e91c1a64424bb1af93656da637b : SUCCESS in 1h 03m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dc3a00c32e3042fc84a1a2d441e47048 : SUCCESS in 51m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/48e5e5f803e0441bb0cbb79747491527 : SUCCESS in 51m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15b9925e90134dd3b3e408dd502f826e : SUCCESS in 22m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a302bc3145d04cb59f2491c91816fa80 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2a96b787083b4edaa774194f895f53f5 : SUCCESS in 21m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6fa38e2075549d4bffeca38440372b9 : SUCCESS in 54m 13s","accounts_in_message":[],"_revision_number":1},{"id":"a3ae2fa507d3206d1d53db0a5f77a53f95d976ed","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-27 17:33:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7c24eecd7c11690db74be1aac570588f4f4ca6c6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-27 20:39:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a5f245fb6897e88ab42001727c90704dc2f22391","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 20:40:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2d48269e88e825be14e8b1b5a8cc24ef432cc473","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 21:46:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5689218d39f44c8856814e73d544294\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2626b1926006495880c7cad9335fcff7 : SUCCESS in 10m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/fda19eede0d34392bd116bfeb35afecc : SUCCESS in 1h 06m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/26198575858e4c58a90115d72acb0f7e : SUCCESS in 52m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c03bc2dcf23544468927fc3f58fc290d : SUCCESS in 1h 00m 49s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c4526b6f1fd14af396806be3e8a24426 : SUCCESS in 10m 53s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/495440844ded4356acd3d2106ce308b6 : SUCCESS in 4m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/501a0653b8cc449385f24fceb22e8792 : SUCCESS in 19m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/eca4eeecb8b4479a9f5363f18cac6aaf : SUCCESS in 18m 12s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0e7b84670c674468901c2678b238a362 : SUCCESS in 19m 47s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9f57ab460cc1485189b88765841111b1 : SUCCESS in 17m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ed79f074f00a43f89f74366ebd4b75b1 : SUCCESS in 23m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9b98ae9f2ff543219177fefb81e36912 : SUCCESS in 21m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d6f100b3fb8645d9847333003f8c40d9 : SUCCESS in 18m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bf64e28f57ec4200a4392d0b29651756 : SUCCESS in 18m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/71bca19268354a7790d40a73e2ad6d14 : SUCCESS in 17m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e67d0c327ce6418a9eb80acc1becc00e : SUCCESS in 15m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/18c065eb09e94893ac8388913150f67c : SUCCESS in 17m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a3dfcc87b0fc48108ce42298669be949 : SUCCESS in 6m 18s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/63a715e86db54b1b9e7e130d19eba919 : SUCCESS in 7m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/af8e08be5591478ebff6e4fa523b8df5 : SUCCESS in 57m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e31f46c7c3e642f09279a6334afaa48a : SUCCESS in 54m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bae518a2bafc4141b643f3ece082e6d6 : SUCCESS in 53m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0646a1556484319aaad06627b03d8d6 : SUCCESS in 16m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/288d9035f72c4996b4494c2906a88ff0 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ef879ad58e9847a985660c1b0a58a267 : SUCCESS in 15m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f13d32d54c5406ebfe1453f2dbca4c7 : SUCCESS in 51m 09s","accounts_in_message":[],"_revision_number":1},{"id":"3143cc6521b22163fad92d167b87a2224f03b600","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 21:46:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d12f8fe9904caba65379920f56151655204d4b45","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 21:48:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f039b1bf49bc4d199e1ee4ee37983a82\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ead3d58d09241d69b32baf407232fb6 : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f99d31d87bd54aecba8c9fa2aa803d39 : SUCCESS in 36s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/da3e50f475dc4f56b716254585db0537 : SUCCESS in 1m 13s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fa088eb72d17aa37e72ce479411a48ea70e4d92e","revisions":{"fa088eb72d17aa37e72ce479411a48ea70e4d92e":{"kind":"REWORK","_number":1,"created":"2023-06-23 19:09:01.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/68/886868/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/886868/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/886868/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/886868/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/886868/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/886868/1"}}},"commit":{"parents":[{"commit":"ca3f1077063a5b1e28239b57f9f8ef22dc7901ce","subject":"tests: Fix replicator test for py311","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca3f1077063a5b1e28239b57f9f8ef22dc7901ce"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-23 19:08:28.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-23 19:08:28.000000000","tz":-420},"subject":"obj: Close replicator/reconstructor connections","message":"obj: Close replicator/reconstructor connections\n\nChange-Id: I02591038a089ce43684f0245665349e40381d75a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fa088eb72d17aa37e72ce479411a48ea70e4d92e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fa088eb72d17aa37e72ce479411a48ea70e4d92e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
