)]}'
{"id":"openstack%2Fswift~886994","triplet_id":"openstack%2Fswift~master~I76ea042ef4b3f5cc1caa4774e35d6191f4ca548e","project":"openstack/swift","branch":"master","topic":"unify-get","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2023-08-08 01:33:44.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-08-08 01:33:44.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-08-08 01:33:44.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-08-07 16:54:47.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I76ea042ef4b3f5cc1caa4774e35d6191f4ca548e","subject":"proxy: encapsulate Getter resp, node and parts_iter","status":"MERGED","created":"2023-06-26 17:08:53.000000000","updated":"2023-08-08 01:34:47.000000000","submitted":"2023-08-08 01:33:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":95,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"886994-unify-get","meta_rev_id":"a0f6dca2d81d204fc05d0f05d32d264a77b8a9f1","_number":886994,"virtual_id_number":886994,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-08 01:33:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-07 16:54:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-07 16:54:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-26 18:33:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-26 18:36:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-10 19:33:07.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2023-07-20 01:27:44.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2023-08-02 23:18:45.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2023-08-03 00:08:47.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2023-08-04 23:06:46.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"1d5353c190c201f5cf2d720877690b77516b77ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-26 17:08:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"748fa18df8529e3a792ddb180d058c162ca63e27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-26 17:09:32.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1e5b56b4263a200a9fc94f85ce7dfe3ee95118d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-26 17:20:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2e31a6020637719b18931c0e41dcdafe95806537","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-26 17:20:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"de6418312f71a91e5e121aca41810591cc6ec1a9","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-26 17:25:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"37a837589d38d0e4df033ae32eba62ca61df6996","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 18:33:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97584fd17e2c4960ac5b2a92a579bc98\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a66d4b55b76a43a68f3b096574ac9ffa : SUCCESS in 18m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6edf9c1d0cb742bd8095de7c80b7ce6c : SUCCESS in 17m 51s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f50ebb24f2234da2a04d8fb08301e72b : SUCCESS in 1h 00m 52s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f048b906fa5a4499882e78b8cd11f156 : SUCCESS in 21m 33s","accounts_in_message":[],"_revision_number":3},{"id":"5c27d7bb1f49d404b95f8c22977651b6e931fb26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 18:36:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c01185e1d09341b493a414a7aa909600\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1008a951855f4a278eed814e85fc9bab : SUCCESS in 8m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/9b6d200279424f47a911e205d7d4d682 : SUCCESS in 53m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cd858f123fb843ceb69346b86cafce23 : SUCCESS in 39m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71432ea821844d6d97854791dae5ba8b : SUCCESS in 57m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad43e2da9e4e49c095ef82e036fcc733 : SUCCESS in 23m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a826344625ce4426b67b5e58aa7d26c2 : SUCCESS in 23m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/46bef63136264ebaaf3b92e85200ddbf : SUCCESS in 24m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/67b22864af5549098bda5ef6d5a08f9a : SUCCESS in 8m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/be3c7436e17c4ed2a99ac620d99f62d1 : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bfb0a6820dad4c519954432f4a1553df : SUCCESS in 13m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5a283e3ac36640a8bd5f93a81c6e0ab8 : SUCCESS in 15m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c746798ed650458c9ea95b08b83fb6a0 : SUCCESS in 42m 12s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/23eb1c671a2f484d8e45f1b890d932fd : SUCCESS in 14m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c285073016db48698ba834427b418d9e : SUCCESS in 14m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5de3d6c77896497587ce0ffa2bfa48b0 : SUCCESS in 21m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c7e4f26217fe4bbe8ca0c4007b926878 : SUCCESS in 14m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3a2a0a8414ec4bdb88af669482cab166 : SUCCESS in 14m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/db33c3020d324a07b229762b584ee92f : SUCCESS in 16m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2fcd18931d9e422098bae04e5c15e71b : SUCCESS in 12m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ad1f28926f4b4ea986219ee593c05bf9 : SUCCESS in 6m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/39c944e5e10a40aea3b1d7eb3c61bbd0 : SUCCESS in 19m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/59d2b710892446588bbdc198eb88e539 : SUCCESS in 7m 29s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d3a9b739ad42438da5b49910cb1db384 : SUCCESS in 1h 01m 05s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3642c5fa600c42f5962f6ae2c93ee1da : SUCCESS in 1h 05m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/29e9fee51d6a49e1af0342f1fbc11b35 : SUCCESS in 56m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/60927584a0064a63b0f563d558eb4cb2 : SUCCESS in 53m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5328b4bbd5ba41a78692898f4178ed28 : SUCCESS in 22m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a25c7c5740e746bcb0cc1c158f4692ff : SUCCESS in 5m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9e6c9ba4259f464680c81b39c43224c7 : SUCCESS in 20m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9338483db0404526b52c5a741c8e043e : SUCCESS in 1h 07m 53s","accounts_in_message":[],"_revision_number":3},{"id":"c54759b64cac4331e9eb9a8fe9bec0832891d8cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-27 14:21:05.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"6d4ff338f0f660551ef8b30333805e62a762cd41","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 14:43:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19d29ae4951f4561a912118544c126b8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a23016e1e634be9a6da40de0613df29 : SUCCESS in 15m 12s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/04697617cbb04364b85c820805491fd6 : SUCCESS in 15m 38s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/fb57d8bd02a1424aac50124c57a08c28 : SUCCESS in 19m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d466e942a81a441090f851dbcadc2b82 : SUCCESS in 21m 45s","accounts_in_message":[],"_revision_number":4},{"id":"71bc6f6f5345edfede76af36e29bb8382c3dc938","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 16:04:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/999c21c6039f4a8297ce166d44cd0810\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20d8af3b875a416fbb4c406cd51b8b58 : SUCCESS in 9m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/651bf1809b8042dba70331f02b3ee4f0 : SUCCESS in 54m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7cfc26e100114b9690fc1bf075ce5e43 : SUCCESS in 42m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f734304a83ee4da4bf40dd6b717daceb : SUCCESS in 44m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b48b47056b0c4ccfa372086322cbd788 : SUCCESS in 18m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a095a0b679074146a43bd772a82e94c3 : SUCCESS in 17m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/80d0c31c1e184e34bfb7018ab114c558 : SUCCESS in 20m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4577e81d822244a69dcf01cd5fc500c3 : SUCCESS in 9m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ac9d32cf787341db92874183da427e38 : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bd2da7f5a742419783f10a1496dd8242 : SUCCESS in 14m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cfb1b25c26714206a4f1079d92c414fd : SUCCESS in 14m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/06c562ef57544175bf2916e1e8af30e1 : SUCCESS in 16m 42s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/82a83929f37045118247d40ef7513de0 : SUCCESS in 13m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dff3aea5c49d401da369544096a980b7 : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f3f5d3045c5845d19be70adcb16b0991 : SUCCESS in 15m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8afcb107f31b4efa80f075653c8cd31f : SUCCESS in 15m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/aade847098df40699cae66d3c52ec91e : SUCCESS in 13m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2750d370321940e9a40824f860e2b378 : SUCCESS in 16m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5a7ac39ea6db484fbae1591d6a2f1535 : SUCCESS in 34m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6bb05a2eeae2427487ddd235077f0473 : SUCCESS in 8m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ac2585acd344481398eae1e26f431e31 : SUCCESS in 15m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/09a60dcc997240fe9b738c7ffaaea26d : SUCCESS in 6m 40s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7f3ca5114363477ea8f20bde86c80702 : SUCCESS in 1h 17m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4b3a4afd15494ebeb8f387d62442fc7a : SUCCESS in 57m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5cd89764e8194ea8be78287969a978ca : SUCCESS in 47m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/22a61a1a28494b8f8eb213b0c6566e0a : SUCCESS in 41m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7138992912504bdbbde87c5c16e86bf7 : SUCCESS in 37m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1be76353155b46058311d6b3e14b4c17 : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cd2d7d7f060d4d2db6de86994e9aa5c1 : SUCCESS in 20m 49s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/516b676f30fb47dfacc697624cb54842 : SUCCESS in 1h 11m 06s","accounts_in_message":[],"_revision_number":4},{"id":"2822a0e665f5460f016375738dc94733e7669905","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-27 17:04:28.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"9c4175d994d99503264b3c986e080c3fd3644cf0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 18:15:20.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fccae86bf5241abac05aa8d88674649\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7960678a20c34de9999a711750e944ff : SUCCESS in 1h 05m 48s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3eabc6e6089143ff979c66b72d733c65 : SUCCESS in 14m 09s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0f20b0a612fa43cf8dbd1edce6fc63ea : SUCCESS in 1h 08m 42s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7a1a7815f6de49b48924fc8e2cec4013 : SUCCESS in 23m 18s","accounts_in_message":[],"_revision_number":5},{"id":"d47c481c47cfcf9cf64a7b567f1c59765396aaae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 19:53:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1cb9a08d94f4bef89351180cda723df\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe701c5681354f3daede738dccd8d5db : SUCCESS in 8m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/9efc763d340f4c06bb140e130576fc86 : SUCCESS in 45m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/daf9f15b933648f1ab3821f9d6eaab50 : SUCCESS in 53m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e90d198f1f24853bbc77e094b0f2695 : SUCCESS in 45m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a03cb8b6a4d946ba984e1351ca35498c : SUCCESS in 15m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4a78c01a4eff47bb90f3a04fa2cf94f1 : SUCCESS in 16m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/affa41da20404b1384ac384a539712e2 : SUCCESS in 19m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5deb2aec12904fefa579607c39fe30d8 : SUCCESS in 8m 46s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bb9ac9aa6386435dbfdcee25550d3895 : SUCCESS in 5m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b96a21d8bfee441eb0d1362f350ac36f : SUCCESS in 20m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b2791986073e48e7879d11e0af19ec19 : SUCCESS in 18m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d1ae952593864f2e907841605028b72e : SUCCESS in 22m 02s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/54bccac755114b4f8f43d7f4d84aa7b3 : SUCCESS in 17m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6a9d37b41fce40a38e63380f5fa1272d : SUCCESS in 14m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/88f5fd1982244600b9475fad2ef8ec74 : SUCCESS in 14m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/901466b497914e159541337584bf34e0 : SUCCESS in 19m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/69871ddb22cc4e2c850afad1c4096521 : SUCCESS in 14m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/371df99115b34812942eb9ab2a281144 : SUCCESS in 18m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c0cc40f78e8d46bf85fbe4b5df559842 : SUCCESS in 17m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/30592ece1e5f4b5898a65c4680b73c67 : SUCCESS in 8m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1b906a7cd80347b9a3b3912f562ac64a : SUCCESS in 15m 21s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1f82e85547484e5fb45e32ea5dc8fbf6 : SUCCESS in 7m 00s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5c0515629f5b483ca6f22f7b25467bf3 : SUCCESS in 55m 26s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4f3c2dd2d9f340ada2e325de80262156 : SUCCESS in 1h 08m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/be93d07f18664fb78b5b359b3d9f7e8e : SUCCESS in 38m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/48108cfc8b83483d96082f8ac2a3365f : SUCCESS in 40m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e4d335606594f028c889abee5e54f52 : SUCCESS in 19m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10c5ce3c1e8d41f49aa5938e07420002 : SUCCESS in 4m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b52d2f0e7e3c415eaa52f31b9574655d : SUCCESS in 19m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/15ab8565356d4e848657ae1f892245b8 : SUCCESS in 1h 13m 57s","accounts_in_message":[],"_revision_number":5},{"id":"0eb01099f577aebd6105a711e5ce3e2cb8391151","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-10 15:25:10.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"9f0258c6614b0f99ed9935531adf2f4ef2691b6e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 15:52:39.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e884f80c59ee47ffb32a0a065960485b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/83f2f984de1b409392b0849c83d8a8ac : SUCCESS in 15m 54s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f0ceaefc98a64b11a76d6b844318b837 : SUCCESS in 16m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7fbb319b2202403395261a05034eeecb : SUCCESS in 24m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8eed53262cb9472a97570b2130f7a6a9 : SUCCESS in 18m 07s","accounts_in_message":[],"_revision_number":6},{"id":"f86a7b7bb6952694ada735cf756ae779dc70b532","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 16:44:47.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a3b018b0e029473fa2cf06cf4c78d569\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76fcd71fdc8e40a2b0f8fcee40a43081 : SUCCESS in 9m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/d30faf730a224741b40f77a337ed93cb : SUCCESS in 43m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c793fedc0087427bb56d9209d33cce53 : SUCCESS in 48m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1bde5a7712a1467189466f17d1596b19 : SUCCESS in 47m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bdae1259f391440890d1769633ab06c5 : SUCCESS in 17m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5cf94f1eafd948489d6f7d32abd6862d : SUCCESS in 15m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eee6f98b032e49a0923d9a3b3fb29e88 : SUCCESS in 18m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3b8647dcb2ef413db0987d708b3cc4d1 : SUCCESS in 8m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0e2b1117d12d456092083941aaf8245f : SUCCESS in 5m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cdc541e86b384cf1911852c94e63ae5b : SUCCESS in 20m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b59ab7e58cd043d9b643b5e9689aabe2 : SUCCESS in 15m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4e4cdb3e350b40eabf7727798e6d20f7 : SUCCESS in 15m 14s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/976d273ae73e455a8c067999d2b8bb63 : SUCCESS in 17m 25s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3a902f0325254fbb8c007d33b42cadae : POST_FAILURE in 16m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/49d71bc1b3db415a9303353aced27191 : SUCCESS in 15m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/29e0946f9eef44b4889bd1833a05b421 : SUCCESS in 15m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/911ca029c5614500b7d4d59dc7f6868e : SUCCESS in 19m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/dbfb8514f13745fcac92dcaf4dc7cd2c : SUCCESS in 14m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/688fa1ae43df4514883405c1e3a859f2 : SUCCESS in 26m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c7a03225f5ba49488ae72b1f017b1a93 : SUCCESS in 12m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a44c9bf82add46d3a32dfb7a6675abe9 : SUCCESS in 7m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/205c826b2c20450a99edd3c094f154a7 : SUCCESS in 16m 27s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a2dde752ff1d4c828ae7c981a5add947 : SUCCESS in 7m 20s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f8d755d1d1dc4f7394f6ab97045c0afa : SUCCESS in 42m 23s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5027146804bd4703ab5e67f47d6a4392 : SUCCESS in 57m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/85ca7799ed0a4fdf82367505c7041060 : SUCCESS in 1h 04m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c4899a0e6389451bb3e9d8561cca9b6f : SUCCESS in 54m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f47f3980d1d474c94a3a0d4a0fd448c : SUCCESS in 17m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d025a024db924d179953adb78a9b9dd6 : SUCCESS in 5m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/847ecf80c820427888aad357694e27c8 : SUCCESS in 20m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4dc1226becd8442290afbc40d16b1d8e : SUCCESS in 48m 16s","accounts_in_message":[],"_revision_number":6},{"id":"5e1ff3c066cf6b9a5cbb26dbf66325f7445ca560","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-07-10 19:33:07.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"8492e4f96c6c5a0da0a4260cd8b43368e21c6dc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-11 11:45:39.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"37590707c32e4c0ab09ad8d17f2b1d3b0cc5b0c5","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-11 11:46:32.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"02e32c508d956df6ea9bd81b67c1fc31238a23a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 12:08:44.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37f233fa980849c88190f9455608ba37\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4d4b1245d5ed4dae891cb78a4670ae60 : SUCCESS in 16m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8a1fc4768ea64cc194bcb3aa4b84c237 : SUCCESS in 17m 02s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f7ac1c5cc4ab434284bb56ef6ea43399 : SUCCESS in 22m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/74bea0948b674a67be116eab567ff0bf : SUCCESS in 17m 18s","accounts_in_message":[],"_revision_number":7},{"id":"faf02c27d3dbbbd4fc9836950826f84fbc77f739","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 12:55:35.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f679780d09744979abbc4b5ad37716a6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d4e3e3872e84a54beea8caa185a29b8 : SUCCESS in 7m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/0a9b25820b764bb19d4647ab982aef9b : SUCCESS in 46m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fbefa2da46ee40d68caa0e4f7c363cdc : SUCCESS in 33m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb356d0b18ad4292a31168999f872802 : SUCCESS in 51m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5e2c6a346645482ab3d6e211fbaae9ef : SUCCESS in 13m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/101513cf7eba4141bb96ec10105bb304 : SUCCESS in 14m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8077e7ae329c4a2a95112105fc92cd6e : SUCCESS in 18m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d872cab1f0914e0b8b05cab1d45b459d : SUCCESS in 9m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8e7be80021ab494982cac239b7dc5f02 : SUCCESS in 7m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3f97a10081eb4369b8f686784c9aacf5 : SUCCESS in 16m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b54601e742564361a0de10f56e93cd7a : SUCCESS in 18m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/78a81056969a416787720f46d6a2ebc3 : SUCCESS in 11m 53s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/36a10f4cd06c48afa78147dc5d191091 : SUCCESS in 12m 30s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2526ed82fa4443ca9d7aaa46016376ed : SUCCESS in 16m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7ed0f88e73ff49d380bb653882b8176d : SUCCESS in 17m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e232644a11aa4fcc93de47c042199bb5 : SUCCESS in 17m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0eb28000af9e4b2fa5c7278428208c5a : SUCCESS in 16m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/814132777cea419d949cc94db05f1de0 : SUCCESS in 14m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3a19200f156f4f5bb5f5cc855496fc10 : SUCCESS in 15m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/068c59c2dc1743bf87ae35ef880ac6a7 : SUCCESS in 13m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a6c1201cb31c4d20af654c6b5ee789a8 : SUCCESS in 6m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/96e321e5e23c49c48475a32cb63a276d : SUCCESS in 16m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/651b4924c06e4554898fe212987a5424 : SUCCESS in 7m 19s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0bb7fbd5cb3442fe916af670129529db : SUCCESS in 48m 49s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/cf5dd455875146519782ab5ac0a782b9 : SUCCESS in 58m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/af7c1de6fa574f7b9d95111a90a80a23 : SUCCESS in 38m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2a14ad5d61d54032ad0cf2cbc5440fa4 : SUCCESS in 49m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a73d316da3364217b3c7aca53df285e6 : SUCCESS in 14m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b777f6f2d564838a46887072e3520bb : SUCCESS in 4m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8b879b191e0b4405a3ec70131659e778 : SUCCESS in 21m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77ebcb0631da4bed85442ad895cc4865 : SUCCESS in 1h 01m 30s","accounts_in_message":[],"_revision_number":7},{"id":"a2dcf68cae3c37d6e9d00c7da9a78336ba5cae91","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-07-20 01:27:44.000000000","message":"Patch Set 7: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"b822356f59131f49ba4031ac24534a0b74406751","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-20 15:38:51.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"d71b21ff0bd3a95e8a6d815be5850a12b55df429","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 16:43:24.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4cf473ca87e4f15868d208e3fa063f6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d0a7c33bcea43ea92a04dccab168a17 : SUCCESS in 17m 23s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7f2f6c7aed3243d985440b035df53c78 : SUCCESS in 17m 37s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d6f3beb0894543739d9c038fcda0714c : SUCCESS in 18m 52s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/666ed37d7e794f6b82afcbeaeb210024 : SUCCESS in 1h 02m 47s","accounts_in_message":[],"_revision_number":8},{"id":"73e6cb899652ea8370d381993ee90d39fd0d93fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 16:56:21.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/026f6da93990497f9a4332291e55f003\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec62cf414b054fbd81821beb36a8ff34 : SUCCESS in 8m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/fddb7fcf6b5b4c47be2476280be4a054 : SUCCESS in 1h 11m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3bea414c97c14096b50bfd173a36485d : SUCCESS in 51m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/acc8052d08ab43c1bb309cc368022bf9 : SUCCESS in 1h 03m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cf81b9a241d94f80b973aa09d4ca7fae : SUCCESS in 17m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d3e702a1ee2c4c619eed8217b7940911 : SUCCESS in 20m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1b28b2d5dee443979111700e5cd37d6c : SUCCESS in 26m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bef5a795fc0b430cb8bb3f5dda818d03 : SUCCESS in 10m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/991d5e7fe1714182ae0791f40862f9c2 : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0b65510aa7d44e46a23ea41695cc5a0f : SUCCESS in 16m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8bbf8188fbac433ab248b82549c82e52 : SUCCESS in 16m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/508da2688bcd4b9d8bbcb26f1382dbcf : SUCCESS in 17m 10s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/17fd08c9de40456f97f079ea267a30d0 : SUCCESS in 22m 52s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5bc3102ef3744f1baf6334c5b9850f43 : SUCCESS in 18m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d5b9d7eecf9e4e43901ef92c0ad4e103 : SUCCESS in 19m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/17863e0178e24ec5ad5f80d3033ac1e4 : SUCCESS in 22m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4fa714ebcfcb4219b394bd519ea5d5a9 : SUCCESS in 18m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bd1099b16841409896b5c807173ab58b : SUCCESS in 23m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/35d5b9f8552643d29f5f9d3347fd773f : SUCCESS in 17m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fff5a08f72aa4242ba58dfacac9aed78 : SUCCESS in 16m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ea55b792979f4f10960b1060fff60811 : SUCCESS in 7m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dc5298e3057449e9962268bf27a83d3b : SUCCESS in 22m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/62f895565b09412ea273c96d1efcf534 : SUCCESS in 10m 50s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/98609c524e164709864bf635d272bf6e : SUCCESS in 51m 16s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3e694c79201949f38d5a4ae36c1c7bdd : SUCCESS in 1h 05m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f9bb12f8b105477f86cc8a32c99cf20e : SUCCESS in 57m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b824394b605c44beab4f1d7a8d5bef9e : SUCCESS in 57m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e286c1acda344fa4aaa3c29f78d1b3fe : SUCCESS in 19m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c198e413000b4dd684d7bb4edc8addc1 : SUCCESS in 7m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/49bdec357d5a486e875404844fee1af6 : SUCCESS in 19m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e9817f30dc044a4869d35ac86b4f869 : SUCCESS in 41m 25s","accounts_in_message":[],"_revision_number":8},{"id":"01795e90ce1c728444ac8b3c153245a4a2243754","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-07-24 03:39:06.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"41e4cc1492df336d18461abb065eb625bc43bdff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-07-24 03:39:38.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"ff3f030f45d2b59d50d396901fd770ea6dc237a3","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-07-24 03:40:19.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"fb5a25fb004844a1a7ca687e043add974a70c277","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 04:53:15.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9be990dfd55749618c2e0b29517e13f6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dee5ecdf98ae4d59b31e09a8bd92e3a8 : SUCCESS in 9m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/d9098c772caa46729dc0313e5efa0a50 : SUCCESS in 46m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2c8ed1044cd74df4af96118cc45ab27b : SUCCESS in 41m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bc65487a4f4a4247ba2c0d80e25c2b80 : SUCCESS in 49m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f3390446d2d241db884318ff6a27bd0c : SUCCESS in 15m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6099a1b98c004b5c9bd91f7e6d025093 : SUCCESS in 19m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c42199f2cae74626898cf71a4a21b3fc : SUCCESS in 26m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/87c7a7370b864cee850e7f71a72c7fcd : SUCCESS in 8m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/596a8098bea74b4badadb4f65aecb217 : SUCCESS in 6m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8f95b70732ef430995edc428063404c4 : SUCCESS in 15m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fe6a7ed9b8b34bd9a57a6af5ba1a40b5 : SUCCESS in 23m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4189659582d040909b23271121365128 : SUCCESS in 26m 32s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d7380490030f4cd2b7a113b08d5912e8 : SUCCESS in 16m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/01b6e57e94f6477b93ecb41dafab8c35 : SUCCESS in 26m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7aa2855b6c5a415394748919e2f101fe : SUCCESS in 24m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c12ba7f9d51140c196c2a36ba5eb85da : SUCCESS in 25m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7c0f98566de04e48974a852aa8821756 : SUCCESS in 17m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/717814f82346472b982c17a444f544bb : SUCCESS in 17m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/64fd28d6b07e4684b3cef9187aed3e6c : SUCCESS in 23m 42s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0c5ea522e46c43ba8ceb1d388279cf77 : SUCCESS in 25m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1e75272429d8487da31d1e16bc257e51 : SUCCESS in 7m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/980a4c97dfb74516a95d3858d08f9ae5 : SUCCESS in 19m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/16475c8599434a109d833ab4d6d0dcf3 : SUCCESS in 8m 36s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0ac11e6c3e0b400fb002b8c962a8ec67 : SUCCESS in 59m 11s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f8b57fb456894a12b17d5d7886a45567 : SUCCESS in 1h 06m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0563356686074a60a17b80923ba9253a : SUCCESS in 57m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4c3db92346594985ad3ab3eed1fb63e1 : SUCCESS in 58m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31d5c5acd80b43068c52198f1a6e0f63 : SUCCESS in 15m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0004bc665c3941449ed8b0eac465ffef : SUCCESS in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6e703dd3b5dc4ebfafa93315d93711f5 : SUCCESS in 29m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3dd5b85535e946c7b8aabba0ae32e461 : SUCCESS in 46m 56s","accounts_in_message":[],"_revision_number":9},{"id":"a9ae62451185265422f0409c0e3388cf147841ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 04:57:33.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be5db78028684d4399ed81138a110e3a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/38f697e024da432e80078407d13d1c1e : SUCCESS in 13m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/73d0e6c48c594591aabe877480719588 : SUCCESS in 14m 12s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e0066d3d7db2439ea5387bfb81b18f38 : SUCCESS in 17m 22s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b6e893e735f04a8fb86694338cc46099 : SUCCESS in 16m 58s","accounts_in_message":[],"_revision_number":9},{"id":"0810d60e21348cd03bd5ca9234452b64e34bea7d","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-07-24 05:13:21.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"577bc14e76f7668f0d511c808d922dc298df5183","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-24 09:06:38.000000000","message":"Patch Set 9:\n\n(7 comments)","accounts_in_message":[],"_revision_number":9},{"id":"23ef34c4fd70d5da2a7940e9c847b2fa5119f785","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-07-25 00:44:40.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"3427996f7a7e408a6736d10eb500c8ec12cf9738","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-07-25 00:45:24.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"d0991d026c2042c10726d1b4623f841c653dad64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 02:03:00.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/739b735e3ebe45b492a8b3d3d3069db4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83b8855368e6417a8d05aff8f066fa4d : SUCCESS in 8m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/50bae4b69e1c434e9e2a565031fd6192 : SUCCESS in 1h 03m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/58e5b0191e594a5994955ec043f17967 : SUCCESS in 37m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9dd34bca575a44c0abaa5e9866b4c355 : SUCCESS in 1h 09m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0db0748d533d4940a1b218a38c318cc5 : SUCCESS in 22m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3e2f116b617b4cfda059c96cb0f1306e : SUCCESS in 17m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8cc7c168247f47bf8d32a9d821b0bda5 : SUCCESS in 20m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3e52bd05572946d8b9a05c1f6f224ec9 : SUCCESS in 9m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a7644cd46d0d41a096ff6d735bb6b498 : SUCCESS in 4m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/08842454877e4a089d5f7628587d8ef1 : SUCCESS in 18m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/50e4718e74ce46f28f7a930623d8e1d6 : SUCCESS in 16m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e7cb2422044f4a98aa149a54310d0f8f : SUCCESS in 16m 11s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/67ba8458251c4c8f993d45e047ad61ab : SUCCESS in 17m 15s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/66db74a82cce42ae9a240f6343ca8893 : SUCCESS in 18m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6c5ba00c096643b0b1dfcee90eb8d5d9 : SUCCESS in 18m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c06ccec4ae8b4407bc86f80efd17c05c : SUCCESS in 17m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/771f728b480349aab5150bd19c7d9a8c : SUCCESS in 19m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/de45a33cc2cf4ba4bedacc04b0ff7ed1 : SUCCESS in 10m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d50ba68c77ef4142bb6ba1cce8f9a014 : SUCCESS in 15m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/551151b0ca964a34bae7e7daa5d24cab : SUCCESS in 12m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1eff0d4fadb54ba48626911fdff6ad9f : SUCCESS in 7m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/477ea8443a954d3d8f15e72f901ba705 : SUCCESS in 12m 19s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/33fb3951eafa48799eb232fb7c1f685c : SUCCESS in 8m 32s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9d8b194816534f71b75a27b863c890f9 : FAILURE in 40m 32s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9c4c0dde33b3464193c0015b8c3e09c0 : SUCCESS in 1h 00m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5d1bab497a594f4cb85d3e6f0f38acde : SUCCESS in 1h 02m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d0621bc2398740429188aa1d9aff7818 : SUCCESS in 33m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df0ecf5536d34e5b8b2e98a7cbbf60ce : SUCCESS in 12m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1b225e259c34214b1ea584d1be8ca6c : SUCCESS in 4m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/28d5b041f98d48038b8e3a715b101fd5 : SUCCESS in 17m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ce893af3428497c914bb66afdb867d9 : SUCCESS in 1h 12m 27s","accounts_in_message":[],"_revision_number":10},{"id":"8a841401ece1313d15732f91f92ef3de20276ceb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 02:12:29.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01627a9245884a3d8d8b359be7c51a1c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8994ac764b1b493f939daf0e5d3fd823 : SUCCESS in 14m 43s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d8091cad7d3947a4b1904ebbce6fd682 : SUCCESS in 15m 08s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/28936e3ab3cf4a608b7bd38e405b9631 : SUCCESS in 17m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ef654a6ee08b4721908f57b4caebddcf : SUCCESS in 1h 25m 59s","accounts_in_message":[],"_revision_number":10},{"id":"842b13556369c9eaf87bd19e5be221791e4583e7","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-25 09:42:10.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"3386c6bafc28eb32b073c1fa79c59ca616661c90","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 11:14:54.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1dc3a38d038b4e71bf77de7b4660a8a4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/93f43a71c0a84b2f84fe96e1a4dd7795 : SUCCESS in 14m 31s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ea94e3e353274a03bb2f7f316c4713e7 : TIMED_OUT in 1h 31m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/841c65951d82430ebf0c6f8520b7a2b7 : SUCCESS in 20m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8d7977901c5f4ea6b8b42bf680e0d5e0 : SUCCESS in 1h 17m 01s","accounts_in_message":[],"_revision_number":10},{"id":"c84526d5469b730243e523c562882586e8d5abb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 11:33:18.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/705cff33def44ac5af8125622c646638\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/265e4398563e47b7a9a710b70d146dd9 : SUCCESS in 9m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/cdcdd46bcf434a18ba0c9c369b6cdef8 : SUCCESS in 47m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f40ce69ab24c436b9f83333314b36329 : SUCCESS in 49m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6bcdf2b665914e5fbaee6fb02643584f : SUCCESS in 1h 21m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8b11bc0d5e744ce0932e01ebba83d00a : SUCCESS in 37m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1e82eb3b043047e4a75e740b00eb1155 : SUCCESS in 18m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bfbd53f5ebc044d0b332483f7864ff7f : SUCCESS in 23m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cfde3fd5fcd74ea88c2016ec93a1bed3 : SUCCESS in 9m 13s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7b1274eb14c3475eb6d72c09c02efab2 : SUCCESS in 6m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a894155e63f34dc1b2f58e1ebe77c900 : SUCCESS in 19m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/476bd843385f4759a2254a3add3a7e6e : SUCCESS in 20m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a1068875d6504b54ad2f4db778ca406f : SUCCESS in 12m 24s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/670322b20089481dbda68e90261a777c : SUCCESS in 25m 39s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c76b750a14ac4244a0b87ab6cd414ede : POST_FAILURE in 23m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b569a7d76352403595a9d360e77a4d62 : SUCCESS in 19m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b56600e46cbb45fe8c473890b6c5e51c : SUCCESS in 19m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c0f47ab48f48416f9906a0c08abd933a : SUCCESS in 21m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bae639d4fcc74ed4b77bb1f5b42efefa : SUCCESS in 14m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fec41ca82f73422f911798659912ffc7 : SUCCESS in 14m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/db548fd7b2ca4b9d9550fee265688755 : SUCCESS in 16m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d6bd2d8d628f4c19a95d25eeeef74221 : SUCCESS in 6m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/005b90b9de204b239b3cf6fb133d6e42 : SUCCESS in 18m 04s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5ea1d61fff3a425aadb12b1a3ac036f2 : SUCCESS in 6m 59s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/73c8756c35074aa6bd7a0ef292471f5e : SUCCESS in 55m 06s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/bf40fcea12b64dabbea90ada53f4fa8c : SUCCESS in 57m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7f765c732a3142d586db3571b0ddc313 : SUCCESS in 46m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ac9050f7ea4d4defa357252c6fad0ae0 : SUCCESS in 37m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b41a3789a4a641c9afb2dfef63652be4 : TIMED_OUT in 40m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfc77a62ad344367a6de7b348b4ec1cc : SUCCESS in 4m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c6fba34ff0ae415491da05528263d614 : SUCCESS in 21m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f02ef8b1087347f7be4d7df530f422a2 : SUCCESS in 1h 08m 14s","accounts_in_message":[],"_revision_number":10},{"id":"1c023e59ba271dccdc190c6e06b019b0ee0d9c3b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-26 12:45:46.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"a46cb89386ad0fd7ae732cbdcd6bc3ec6410ea5d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 15:05:42.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aec01b7bbfa14a12a8a749a7dbda3044\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/746d14a39eed46b0899ab18b8db274ca : TIMED_OUT in 1h 31m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/265e2ca171b04840afc033ea86cc9829 : SUCCESS in 21m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c2f0b954bf147529863311b61f4ee3c : TIMED_OUT in 1h 31m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1ba488f26eac4847bffa553a41511646 : TIMED_OUT in 1h 30m 53s","accounts_in_message":[],"_revision_number":10},{"id":"c8171beecf2991f2bfb0a2f2d6a5bbe682589f2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 15:21:03.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/646c9747bcf7450689b506baa1d71714\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/494858f67528454484c507f85d9f0597 : SUCCESS in 9m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/054186bbaf634ff39d8e2e99390b9240 : FAILURE in 53m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/335be7fadab142c3834850bdc270aca6 : SUCCESS in 49m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2bab4571208a417e8e27ebb1a36292a6 : SUCCESS in 1h 18m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dc9709d1aa4349d08d32e7166a5843e1 : SUCCESS in 18m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/60a39158112944bb90ca4c4772733bee : SUCCESS in 18m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c77dcd43ccd44c7397d9735906fde9f4 : SUCCESS in 27m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f7ac29963f3c41d392e668ea36ae482b : SUCCESS in 8m 13s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/32253dee43094ba8b169da0e9e5c712b : SUCCESS in 7m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3e52ddbebbc744249a4701a1f1433a32 : SUCCESS in 19m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5449e5b8c21947a49237e5690aeb7d3e : SUCCESS in 19m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1005db0fd33847269efc1cd389169bd7 : SUCCESS in 24m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e922bde863c14ae0941baacbc37ee3b5 : SUCCESS in 16m 55s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ec00a27f074c4b4f9b9e421268b3d38b : SUCCESS in 23m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/df689d4597a94eebb7d86a83629fd97a : SUCCESS in 18m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5d5b384cb920429e8721c129bb2a63a1 : SUCCESS in 17m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6c9798e2379a4e1b842d8c5a4974dd5f : SUCCESS in 24m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fa541264162943a397e150d85e8d02b6 : SUCCESS in 14m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/772e7b7c4ad84886ba1e02e172c62971 : SUCCESS in 17m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ef3030a64d81467b9e7787ea022bf85e : SUCCESS in 16m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/51258d18336b45fab18b4fa2994b04d8 : SUCCESS in 8m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6df8f50471f14dbf8e29808244df8e03 : SUCCESS in 17m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bfce355a190342989af15027bc125d30 : SUCCESS in 7m 48s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d2d94411abf54078aee5f9ba98482eb5 : SUCCESS in 55m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0c16ddd2a02040e494fac33012646e9b : FAILURE in 1h 22m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2a690a34c30d4c0d9a6e2b949e66d329 : SUCCESS in 51m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a97290aaebd5412a9cfa58f93e9d120f : SUCCESS in 59m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/239a051b877544d2960690bc816033be : SUCCESS in 18m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/606a0ac531474555b48b8acd9f6ac14b : SUCCESS in 4m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/20e51ddb9a0344cebade65c13e69deb0 : SUCCESS in 28m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67b294839863489ebaa80f027d131376 : SUCCESS in 42m 58s","accounts_in_message":[],"_revision_number":10},{"id":"d96024b8d8688bc77b8ae217c4de2600c85cfe8b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-02 15:52:24.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"f491d9e587a19e8e4bb554893a5e9fd601f72d22","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 16:20:34.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a73fb45348349edb3bde54ff3f323ff\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/792c8db48c694dab950b48e7aae00408 : SUCCESS in 21m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3b606145dad34545904886193f8fdb69 : SUCCESS in 21m 25s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2e8b62644479463a8739107552c61656 : SUCCESS in 26m 35s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4eddd12a2763437092afc48d719e846b : SUCCESS in 26m 26s","accounts_in_message":[],"_revision_number":10},{"id":"ef02bbabf5778fddb473689453aa454622763420","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 19:52:55.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4168826125b6436fb2497bc6632e091d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cae5f3ced1f6417b915064e9d6c3ed95 : SUCCESS in 10m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/d5dc2de0086e42869f7950eaf6cf1b73 : SUCCESS in 44m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/58a737c9fd994bdcb11829eed8aea2ca : SUCCESS in 1h 07m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c425dbbf12f241fb83ae3097f0601279 : SUCCESS in 1h 07m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a7948f08b59b4a9885b4687517a23565 : SUCCESS in 19m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/67c054eb2dac426caea4a911265837de : SUCCESS in 18m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2d38eca6e93e4921beca2970f73a673e : SUCCESS in 18m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c8f5a4dd67c54e5fa0d88e0f24ce7176 : SUCCESS in 10m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/121fe2e7748e480db53699a028b2e25f : SUCCESS in 6m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9b531f116d2d4af0bb3b56c22eafa35a : SUCCESS in 21m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d9d815658ef64c51a51a7abf334ddfbb : SUCCESS in 29m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/465ae9c660f049f89a2d8f6a331cffb9 : SUCCESS in 13m 37s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/712629d69d074fa79e10f35fadec9c3d : SUCCESS in 14m 17s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3bb70fc83e4b48c099d1f95f2299aa1b : SUCCESS in 19m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/35e96c8aa1cc4954ac9977e9131d3aa8 : SUCCESS in 16m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6c51a197ac004e8b9e952149590eb734 : SUCCESS in 20m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e7ae161f87f14d3c82627495af9b1b9a : SUCCESS in 22m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/584d5aac2fac489ca213f2a6a54f65b4 : SUCCESS in 15m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0a692024b959403ab2e8006c91b8c3dd : SUCCESS in 14m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7e687d91614f456dadb1dab3fe36f854 : SUCCESS in 21m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/426b744ce8214de8b6f29093c22d1bb6 : SUCCESS in 7m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e130cb6616b74c1cbef689604de91aee : SUCCESS in 15m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/416322ec19724a4aa174e36b3e623150 : SUCCESS in 7m 35s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a1fea27f605f4ccb81acd42cebb813da : SUCCESS in 1h 14m 34s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c3626f507a4947c9961b3b34381a66e5 : SUCCESS in 1h 02m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0f9c50f8b83543998e6ccaf482540c20 : SUCCESS in 38m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/60ece0394b864220bbb3fe493cbde07c : SUCCESS in 41m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b6b2d54733f45128629f2c9510f8456 : SUCCESS in 17m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c10a4644c8034393a6f3000ba4124ee0 : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bd7b470207bb43b4bd43d4f2af1fe3f7 : SUCCESS in 20m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b4e633ced97e46fa8939f5b3f7164795 : SUCCESS in 41m 02s","accounts_in_message":[],"_revision_number":10},{"id":"5a92cc66b6bfbf3296d59ee42661c6a198290d40","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-08-02 23:18:45.000000000","message":"Patch Set 10: Code-Review+2\n\n(22 comments)","accounts_in_message":[],"_revision_number":10},{"id":"d11af0a5e183ea69c64fbbc2d2e51f1dd953fe36","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-03 00:08:47.000000000","message":"Patch Set 10:\n\n(10 comments)","accounts_in_message":[],"_revision_number":10},{"id":"4822ffe128687175c01c6b8e33ae1663e3f26398","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-03 15:34:18.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"41261eb80509e21305ccedf5a7eb0c5d78c1be11","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-03 15:34:27.000000000","message":"Patch Set 10:\n\n(26 comments)","accounts_in_message":[],"_revision_number":10},{"id":"467cbba6105e33be678fde174efeb5cb200dc0f1","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-03 15:57:00.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"599048dafdb9839169743aed0880b46c8a48954c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 16:58:42.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43e2b9d91b464ecfb5b705b750de6432\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fcbd1369f02401d9bdc3960359c2e16 : SUCCESS in 9m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/920fb60a5656416fa3fc4803675bf6fc : SUCCESS in 41m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/72dd7f4ea6724a1dab5174c3ecaeedf4 : SUCCESS in 53m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ae9f8b579e44f02adeeb1075a8041b8 : SUCCESS in 1h 16m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3b0d697659744907b9538cfb01d63a2d : SUCCESS in 17m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8f9521a302f84c65a5dfc7d9df140c8f : SUCCESS in 17m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/30c407db301746bf9e9612f991e608e9 : SUCCESS in 20m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/721fde57cc474d598d0938c884122db6 : SUCCESS in 10m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/de948e1147024dae956ce103a27f452a : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d60eac0801a74397a1baba06dc6ce9db : SUCCESS in 19m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/405690ac32d7442dbf45deb19b068201 : SUCCESS in 16m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/05eec7512a244bf08aeb9f3d272c3d72 : SUCCESS in 15m 51s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6aa4c3ac39384018862343453ffe1c68 : SUCCESS in 17m 28s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2bbe16c0d5b347f4b7a00546947a1e4e : SUCCESS in 16m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8f655192360b4fb582a832cf7ab73c14 : SUCCESS in 17m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b39ef30499784127ae9d8524143efe4c : SUCCESS in 17m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6489e1bb840f450baa16d60324bb656f : SUCCESS in 17m 17s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b4dd63421c99422795dcb4972e5714c4 : SUCCESS in 15m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8be8c854b14c4a00bde8374894a796b5 : SUCCESS in 17m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c709cfbd05044231bbbdde6284077b44 : SUCCESS in 18m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c3efe01726a34d188733ca4df8929085 : SUCCESS in 9m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2ef2ce90529c43b7a805fddbcd96ac80 : SUCCESS in 20m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/dd945c4f6ca04ee48bf137049297f024 : SUCCESS in 10m 03s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f6be4040f324407ab63307c6c1abb714 : SUCCESS in 59m 55s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4a1fd8a33c5d43c5840741ff671bacb1 : SUCCESS in 1h 15m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c20d44dcfad7479bb47924f95e0144de : SUCCESS in 55m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e9e5f8bc126e46e0852e384592767266 : SUCCESS in 1h 08m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/246ef8f44bf14760968e066442fff0ad : SUCCESS in 18m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78091302192d4c208e07359ee6344fbd : SUCCESS in 5m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/496fb173b5f74e64ba7ec790959c669e : SUCCESS in 24m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37e5e5362c184c8eb78491745c6ab282 : SUCCESS in 1h 14m 54s","accounts_in_message":[],"_revision_number":11},{"id":"df6d6f9ee2db09d75387ea1238478bfc071b9f03","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 17:07:51.000000000","message":"Patch Set 11:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6765f977d81422a9f0e81b2a9ba632a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4faa7d8aefd44ff0aad1fbaa41de5fbd : TIMED_OUT in 1h 31m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c6b6a0b70d70438f91bac360ea8415ec : TIMED_OUT in 1h 30m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/279e485d2c624b99b08acfd54cb1b9eb : TIMED_OUT in 1h 30m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2a5c622383b64af0a39dbb1a7daff458 : TIMED_OUT in 1h 31m 00s","accounts_in_message":[],"_revision_number":11},{"id":"db011ae97a4f9158bfe27793cd03676d9a76b571","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-04 14:05:26.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"c0bc85310891dc8ee0e0e59da8c8591e363b956e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-04 23:06:46.000000000","message":"Patch Set 11: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":11},{"id":"9d57c2364d9400d9d5c5142154b546a9d65ceffd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-07 10:19:42.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"f7b7822205aed73cd20a86a00be8dc2dce372485","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 11:52:45.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33da1e7870234235aec088dc132bb7d3\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f8d444ef4e3447829a40501236c5d606 : SUCCESS in 14m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/517e18e9f7bd4d419d5eaca5c0a46f1a : TIMED_OUT in 1h 30m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/fb1a5f8c4bb94003adfa5e950cb08a1f : SUCCESS in 18m 45s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d2171e06e70a4b65a452de56f89cfb79 : SUCCESS in 1h 11m 44s","accounts_in_message":[],"_revision_number":12},{"id":"2066b4ab4f643a5604d5eacb16fe3245cef1327d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 11:57:00.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/95b593054dc44d449db9c79586e642eb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/663d494dcf904ef099249c2259fa65a7 : SUCCESS in 7m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/493cbea750b04d5687ee5155d47c04d9 : SUCCESS in 48m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b568878eb72e4542bdb591eac1500236 : SUCCESS in 59m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7564dec8ea934daaadccc3402bd7513d : SUCCESS in 1h 12m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ff87e793a57a47148d6290a3522f512f : SUCCESS in 19m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/68014f5ba76348888ef69162ad3dc005 : SUCCESS in 19m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0f7df4feb7894d18b10933cd81258ae6 : SUCCESS in 21m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f7596f6233c04c78bb1df517fe7ca80b : SUCCESS in 8m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/14f7604422854038849631d8eb5d2ff6 : SUCCESS in 5m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/69ae815506c54794a6c2ea302cace3d1 : SUCCESS in 19m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7eb59b2abcab46499c76da7b1a825d0e : SUCCESS in 16m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1db2bf418f3248078da90c3e127ba10a : SUCCESS in 15m 42s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2e02748cd5ff4c7e9d03d200cca5c8bb : FAILURE in 41m 06s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b4e286cfe9c046ed98c53e364ce21271 : SUCCESS in 18m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f58885a5fc7b4ddb985de576673d3ee1 : SUCCESS in 18m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bdf82ae322084cf481567458d641c4bb : SUCCESS in 17m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/88e36ac5fdeb484fb1901e75555395ea : SUCCESS in 21m 31s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8e2f2be4c6d949c680165154a8d89aaf : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0199692cab984c679ea7f2d8176803fb : SUCCESS in 15m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d9009bc9e7ec4ea1a5622fe5bff02d8e : SUCCESS in 14m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/45a958aa96a64b96ac42a5cfe1b05e86 : SUCCESS in 7m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f07117214ea1485eb04b2c822d5eb69c : SUCCESS in 14m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/31301ef1d4b149a487142f90a0c53ad9 : SUCCESS in 8m 00s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cb9c8623f340449da665e1727ad2072e : SUCCESS in 1h 09m 13s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b3f1ff21e97b4af9b75c9b556e93c7fc : SUCCESS in 1h 05m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1c5ab717833b48199aa07140bd457dc4 : SUCCESS in 56m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ad329823fd9443c6bfc02b1d49d1ba21 : SUCCESS in 38m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e66d9ec8d6c4485baaf8ce67731ec2c9 : SUCCESS in 28m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6abe8f686a3a4c3299f3fe91adfeae5e : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cee8d1e247134f7285202bdc1bf48920 : SUCCESS in 28m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6799dfbcf12e407e88776b4a24781c1e : SUCCESS in 54m 22s","accounts_in_message":[],"_revision_number":12},{"id":"59318e8108c3429c6ff1557baeaaf5ec59ee937e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-07 14:15:11.000000000","message":"Patch Set 12:\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"dc586e74e46a45bc15743948c2f24260422d03d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 14:38:07.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cdca515f21441d0b50fd55137f3b7f0\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6c9aa3708a784f85a8cf1345ae78d621 : SUCCESS in 16m 48s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f006baaefbe1445498afb64debbdc477 : SUCCESS in 17m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/11e109c0bd6e441db0aa10c98148cf00 : SUCCESS in 20m 20s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7b52f37118944b9880cb8df3b4b9c7ea : SUCCESS in 21m 56s","accounts_in_message":[],"_revision_number":12},{"id":"e46ff24935ecef1238e0b70e1fbbaf259384a4c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 15:45:52.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5483e0697a5c48e6b39d224ef39c5346\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3ba21ab4dbd4d05a785efdd1634166a : SUCCESS in 8m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/a4e70e609bad45cdac28815cef13fb9f : SUCCESS in 49m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/985baa1ad28e4a84a5b2884019dd60ed : SUCCESS in 43m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e4117b2eadc47208ebc9c232e657e72 : SUCCESS in 1h 12m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b85a13f8edac4e33a80ef1d61b2b0f2a : SUCCESS in 18m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/13e6eacd7884421e800071a3dc700bf8 : SUCCESS in 16m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c1d7c335829044129d9071dd4e6ae047 : SUCCESS in 19m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0f7ab4345d8a444a8c7007d194337803 : SUCCESS in 10m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4b071641f59d43f7b0761fcb1174b06d : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/916ea67c43d34dfb8f8443b1a54cbe01 : SUCCESS in 17m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a000bf1fd0c04cfcaeb7aa1d5ee2448b : SUCCESS in 22m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9ef51c8bd659418b829da3d324438462 : SUCCESS in 24m 03s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/00537f3b6ee64fa9be5641c28518aa69 : SUCCESS in 15m 48s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/24fa3fd1b8d343bab3e649c034a45eb9 : SUCCESS in 15m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/279196fb0a0141ec96afb81e6a8767cd : SUCCESS in 16m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a8138a4a72a44fc7a43e1627e1c14272 : SUCCESS in 23m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2f2f6295f1d745c18077bb7afb732202 : SUCCESS in 14m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8568f40c83674b0e91fa8c5c6dd82526 : SUCCESS in 13m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e1dbb53988d44975909182e991cb725a : SUCCESS in 12m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fd3a60d0a3314387904c3256d5255901 : SUCCESS in 16m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d849a4068f48480291264cb4153dfe1b : SUCCESS in 6m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f5b68dbbfb084bb3bda351e747ede5e2 : SUCCESS in 14m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9b837459df0e4d2e8684b32adea96eb1 : SUCCESS in 6m 56s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f969c1695d3242b69f6b7da9664180b5 : SUCCESS in 44m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6cde03b3ff5c40e691e263972dacd3a9 : SUCCESS in 1h 00m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/30e2fa53429c485c8f078ad8413ec5e2 : SUCCESS in 38m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/29359c0c40954df0af3d48cc40968e9a : SUCCESS in 46m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24f992557e6246e4a4346d1d2d5c3e8b : SUCCESS in 18m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51e495c439864bfe9b737f7665927aa2 : SUCCESS in 4m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d0f1d4faf6d5479994be50f83adf94ec : SUCCESS in 26m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3aef7a3243254b9496990e9544cdd4f6 : SUCCESS in 1h 10m 08s","accounts_in_message":[],"_revision_number":12},{"id":"526feab51127cb85aa7fa03e7a7d65071b46a8a3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-07 16:54:47.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"f3bac283fdbcda64a545675e70be70b336221d7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 16:55:23.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"a9c72ac9b75b474ba2c3284364e269e3f13f3641","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 21:28:53.000000000","message":"Patch Set 12: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e278222eaa6d4dffb4bc63b93bd4cea6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef1f0fe3266342ac9055ee5f69eb49b4 : SUCCESS in 9m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/14f717b551f64c79b24b4c3c9bdc34a3 : SUCCESS in 59m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0fcc15399a53487993243fb13e061c68 : SUCCESS in 52m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/801ef819fb1246e2af3aa4c06097bb0c : SUCCESS in 44m 39s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/395c443d6fd64b3a9370ebc3c986777c : SUCCESS in 8m 34s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/ab92ea9793324e1183a3c6b34e2684b1 : SUCCESS in 5m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/da55d96ecfeb45d9ba4bfec4f487ab92 : SUCCESS in 16m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e7f29409217b48fe9e7a2309d8c27c02 : SUCCESS in 15m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/707c4a67bf7c4b518aaa99a1b0c17619 : SUCCESS in 24m 58s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/21a7f522828c41c9875723582950b6e0 : SUCCESS in 13m 04s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f0fca296399b47f6bbbeff0e0620d7dc : SUCCESS in 26m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9894bff34aea4ad5bc4935b6c0cab6e0 : SUCCESS in 14m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6a6733fba0864c69931a8bec86e85bd6 : SUCCESS in 16m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f2232c9d018646388833f7c954791074 : SUCCESS in 15m 13s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6d8f934328aa49e1a094936d3ca6015a : SUCCESS in 29m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5439e3d136dd4c3f80297ee0511a99e3 : SUCCESS in 56m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2319e5551bfe48c393b2a56627349e79 : SUCCESS in 11m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1b9ce7ce918f444daa151577d07226f9 : SUCCESS in 5m 48s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/700307f9fff0410dacb6c6eb8d904028 : SUCCESS in 7m 40s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/908359f596da46fa8da45dbcccad6abe : FAILURE in 1h 14m 47s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/acaee2b9044e42babb44f82946fbcac6 : SUCCESS in 1h 00m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2941b71830fb401da519dcfccf37117c : SUCCESS in 33m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3b75209a6d1f407fa747e213a6d3adde : SUCCESS in 55m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/375a172927ae4f6f96055588505ed3f8 : SUCCESS in 16m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/664f55b9cd1343df837717530f943669 : SUCCESS in 4m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d94bc2b654a485e911b1fd57ea02daf : SUCCESS in 59m 17s","accounts_in_message":[],"_revision_number":12},{"id":"ae45fff7f682933e4cd23d5686b81cfd69c31a3f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-07 22:47:44.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"912a56dfd369de2d9059ef99b70adb592c2119da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 23:09:07.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b5bd414e9c042aaab54c9b1e8495b34\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9570fc239ad34841aa3230b3bc7f1350 : SUCCESS in 20m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7e56e5192ce2474da63cfcb75f3a3704 : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d5d331303d3948dda43d0f6331d2735b : SUCCESS in 18m 31s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/577052150d0e4d1b98042f8d013d4a4f : SUCCESS in 19m 12s","accounts_in_message":[],"_revision_number":12},{"id":"e40d529ae7f24c15879110498fa879ffd117c8e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 00:02:37.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/465f568602544f7a8e7a9fa108c7bc90\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9edd9dd4ded647c8a657772a8a529d25 : SUCCESS in 9m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/8df7bde23e734f589dfa7c80a6bdb12d : SUCCESS in 53m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f3955e4926674ac29458e9984fd1ac6b : SUCCESS in 47m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31bcf5451efc430e824a0638e07459d8 : SUCCESS in 57m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/39c78c7ec3994406b6fc69b7e97ed1d1 : SUCCESS in 29m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f7f504602b044e60b8fae1dff5f78936 : SUCCESS in 19m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/49de75a6bddd448dadfdaf19ace6e00e : SUCCESS in 27m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a59a7e5c01644571bc22ca9c76a72314 : SUCCESS in 9m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/016ec8c13b9d4e19bc1a1cd3befd2dc7 : SUCCESS in 6m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/73fffbeb50744944bb68d98561fcf87a : SUCCESS in 16m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f3c9626ecc4f4e248a32c38fc006fa11 : SUCCESS in 15m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b9a0711acbda49b69623c27e5a12f0f4 : SUCCESS in 16m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/63da702fc5b34b81b9d05e6ba396068a : SUCCESS in 15m 51s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/23fde44032154cc1807f9046be858269 : SUCCESS in 16m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c0062e2afb434ec4baad87afacab1859 : SUCCESS in 17m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1bfa058938114efea597fe2b0da002a3 : SUCCESS in 17m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/400810ff7f6c4735ba3aaec2d0d19a1d : SUCCESS in 17m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e2708e93ba7446158cc75c17bb71c9cb : SUCCESS in 16m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a25dbd90646b471b885b4f3fa0284412 : SUCCESS in 15m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d5305394a1e6410bb5730bf0449894cb : SUCCESS in 15m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2f49ac400cf944218f4559f7ae830c2e : SUCCESS in 6m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/877a3bcdcec84943b132a5864937d3c9 : SUCCESS in 16m 33s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7b7ce4259456470f81d121f6ad68e93c : SUCCESS in 7m 40s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/40939ed4c20d4caca76b5a9641681d02 : SUCCESS in 1h 09m 11s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/77ce54e1fbd3452e957c82bb09ad1c82 : SUCCESS in 1h 02m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/487b92858cde45baa1299d264e9bcd1c : SUCCESS in 56m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b03e04f3d69143858cb889b624ac562e : SUCCESS in 51m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fdf1dcd56486455f92995953d885cbf1 : SUCCESS in 33m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/715ff1a549fe4044b27ec51e221ee8a5 : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/664376d2b0dd40819bf91fecd9385717 : SUCCESS in 30m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/44066d3630ec47f2b1cd667e74f86397 : SUCCESS in 1h 03m 36s","accounts_in_message":[],"_revision_number":12},{"id":"f36d90d24e94a00e9ad9e2d316a9c960651fc817","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 00:03:15.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"68c7c986d3e463921883bf60ed0e969b373b343a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 01:33:43.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a290593be3e34992ab98083ec2ade8f1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/206832fc7b5e44e3b7c330a41404b0f4 : SUCCESS in 8m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/4310cacb486e4f3394c9be07b894859a : SUCCESS in 57m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d69a0e0baf9f4eb6b384208e40b1b3cd : SUCCESS in 59m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2a361e349cb4df090e7ce564b49ac9d : SUCCESS in 1h 04m 28s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d1980e4135a24b10a00204d2a2e0b15e : SUCCESS in 8m 52s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/3bcdb86fa36e4f86a686ba1fef85c91c : SUCCESS in 6m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5a875132f43d46ec8cee756b4a2c02a4 : SUCCESS in 17m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/decc3f0e606942efb699ded18adce438 : SUCCESS in 17m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1e1199fa73554482a1cce0bd97e6aacc : SUCCESS in 11m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4b52de3b8ff740c4bd29175533f0a8bd : SUCCESS in 15m 59s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/76d695b4e22c4298b6ce45fd2ba94a77 : SUCCESS in 18m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7cc0f96132394bdebfdbf134e6946b8b : SUCCESS in 19m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cba33164c7ae475a8b1ff2546ea1c2cd : SUCCESS in 19m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/822d65ba5bb84e6abd22efcf309d6863 : SUCCESS in 20m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3d3e486ded784628a8ed2dcad6ffb2eb : SUCCESS in 13m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/bab08d6a7f454839aa2a1706a2f66f92 : SUCCESS in 12m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/582f851c1d4d4700b6c127666daa194a : SUCCESS in 18m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6341ad7724754a8689c51dd44478c9c5 : SUCCESS in 6m 27s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f742271548aa4c71ae7537d35a25ea24 : SUCCESS in 8m 08s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6a5e368da15f4404a4b6e3d1a4430397 : SUCCESS in 40m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5e83386726814811988e6c1981afe5e1 : SUCCESS in 52m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a8ab83175d7a4d8e800cefc3bf5d0324 : SUCCESS in 35m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/100697992e204a5cafef7ee4fd49ab1d : SUCCESS in 43m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a787cdbee0bb48b493651434512c1676 : SUCCESS in 13m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e77fa1070fb42d58bd61a558d724a05 : SUCCESS in 4m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c003ccb099b94d3eb768cc80c5816e66 : SUCCESS in 40m 38s","accounts_in_message":[],"_revision_number":12},{"id":"66a671ae383e2037af8779b5a53bb2ef850ef212","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 01:33:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"a0f6dca2d81d204fc05d0f05d32d264a77b8a9f1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 01:34:47.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61af1ccd1112454fb87bd2158e17bb7c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db01cb5849bc48ce8225dfc0e495be46 : SUCCESS in 39s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/7de930bb6020484080894ecc98c357f8 : SUCCESS in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/94af30b1544d4331bbc7150a11c06906 : SUCCESS in 29s (non-voting)","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"f8a6dee622e47cec7111007069dd4ca74b56b49c","revisions":{"a14216c7485efb4114fb9020c77cd39c53bb8c62":{"kind":"REWORK","_number":1,"created":"2023-06-26 17:08:53.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/94/886994/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/886994/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/886994/1"}}},"commit":{"parents":[{"commit":"db947cf34d24914608c4febabc09b143cf021954","subject":"proxy: remove client_chunk_size and skip_bytes from GetOrHeadHandler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/db947cf34d24914608c4febabc09b143cf021954"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-26 17:02:27.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-26 17:02:27.000000000","tz":60},"subject":"proxy: encapsulate Getter resp, node and parts_iter","message":"proxy: encapsulate Getter resp, node and parts_iter\n\nThe proxy GetterBase manages a set of attributes related to the\nbackend source for a GET response: source (a response object), node\nand source_parts_iter. These attributes are always updated together so\nbenefit from encapuslation, along with some helper methods to simplify\nthe GET paths.\n\nChange-Id: I76ea042ef4b3f5cc1caa4774e35d6191f4ca548e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a14216c7485efb4114fb9020c77cd39c53bb8c62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a14216c7485efb4114fb9020c77cd39c53bb8c62"}]},"branch":"refs/heads/master"},"8476a721dbec6c28d9faf7b5c392d00d18200034":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-06-26 17:09:32.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/94/886994/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/886994/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/886994/2"}}},"commit":{"parents":[{"commit":"db947cf34d24914608c4febabc09b143cf021954","subject":"proxy: remove client_chunk_size and skip_bytes from GetOrHeadHandler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/db947cf34d24914608c4febabc09b143cf021954"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-26 17:02:27.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-26 17:09:18.000000000","tz":60},"subject":"proxy: encapsulate Getter resp, node and parts_iter","message":"proxy: encapsulate Getter resp, node and parts_iter\n\nThe proxy GetterBase manages a set of attributes related to the\nbackend source for a GET response: source (a response object), node\nand source_parts_iter. These attributes are always updated together so\nbenefit from encapuslation, along with some helper methods to simplify\nthe GET paths.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I76ea042ef4b3f5cc1caa4774e35d6191f4ca548e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8476a721dbec6c28d9faf7b5c392d00d18200034"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8476a721dbec6c28d9faf7b5c392d00d18200034"}]},"branch":"refs/heads/master"},"ac2efb883b0fbf624aed302229057c2c389c1fbd":{"kind":"REWORK","_number":3,"created":"2023-06-26 17:20:41.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/94/886994/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/886994/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/886994/3"}}},"commit":{"parents":[{"commit":"db947cf34d24914608c4febabc09b143cf021954","subject":"proxy: remove client_chunk_size and skip_bytes from GetOrHeadHandler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/db947cf34d24914608c4febabc09b143cf021954"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-26 17:02:27.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-26 17:20:30.000000000","tz":60},"subject":"proxy: encapsulate Getter resp, node and parts_iter","message":"proxy: encapsulate Getter resp, node and parts_iter\n\nThe proxy GetterBase manages a set of attributes related to the\nbackend source for a GET response: source (a response object), node\nand source_parts_iter. These attributes are always updated together so\nbenefit from encapuslation, along with some helper methods to simplify\nthe GET paths.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I76ea042ef4b3f5cc1caa4774e35d6191f4ca548e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac2efb883b0fbf624aed302229057c2c389c1fbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac2efb883b0fbf624aed302229057c2c389c1fbd"}]},"branch":"refs/heads/master"},"7de5df797fb854984d45091276d536073088a121":{"kind":"REWORK","_number":4,"created":"2023-06-27 14:21:05.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/94/886994/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/886994/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/886994/4"}}},"commit":{"parents":[{"commit":"db947cf34d24914608c4febabc09b143cf021954","subject":"proxy: remove client_chunk_size and skip_bytes from GetOrHeadHandler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/db947cf34d24914608c4febabc09b143cf021954"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-26 17:02:27.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-27 14:20:53.000000000","tz":60},"subject":"proxy: encapsulate Getter resp, node and parts_iter","message":"proxy: encapsulate Getter resp, node and parts_iter\n\nThe proxy GetterBase manages a set of attributes related to the\nbackend source for a GET response: source (a response object), node\nand source_parts_iter. These attributes are always updated together so\nbenefit from encapuslation, along with some helper methods to simplify\nthe GET paths.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I76ea042ef4b3f5cc1caa4774e35d6191f4ca548e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7de5df797fb854984d45091276d536073088a121"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7de5df797fb854984d45091276d536073088a121"}]},"branch":"refs/heads/master"},"a33c04e7917114ea0149221287c326d0fb5dfb8e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-06-27 17:04:28.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/94/886994/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/886994/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/886994/5"}}},"commit":{"parents":[{"commit":"684b28042d1eb8e135e1e8ec6adf64f4a807f3ca","subject":"proxy: remove client_chunk_size and skip_bytes from GetOrHeadHandler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/684b28042d1eb8e135e1e8ec6adf64f4a807f3ca"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-26 17:02:27.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-27 17:04:16.000000000","tz":60},"subject":"proxy: encapsulate Getter resp, node and parts_iter","message":"proxy: encapsulate Getter resp, node and parts_iter\n\nThe proxy GetterBase manages a set of attributes related to the\nbackend source for a GET response: source (a response object), node\nand source_parts_iter. These attributes are always updated together so\nbenefit from encapuslation, along with some helper methods to simplify\nthe GET paths.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I76ea042ef4b3f5cc1caa4774e35d6191f4ca548e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a33c04e7917114ea0149221287c326d0fb5dfb8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a33c04e7917114ea0149221287c326d0fb5dfb8e"}]},"branch":"refs/heads/master"},"456f539b16053240e3d2a2f874a13738f8a39b4c":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-07-10 15:25:10.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/94/886994/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/886994/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/886994/6"}}},"commit":{"parents":[{"commit":"5bd21ba912f42ee6893ec5614609344da9e19f95","subject":"proxy: remove client_chunk_size and skip_bytes from GetOrHeadHandler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5bd21ba912f42ee6893ec5614609344da9e19f95"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-26 17:02:27.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-07-10 15:20:00.000000000","tz":60},"subject":"proxy: encapsulate Getter resp, node and parts_iter","message":"proxy: encapsulate Getter resp, node and parts_iter\n\nThe proxy GetterBase manages a set of attributes related to the\nbackend source for a GET response: source (a response object), node\nand source_parts_iter. These attributes are always updated together so\nbenefit from encapuslation, along with some helper methods to simplify\nthe GET paths.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I76ea042ef4b3f5cc1caa4774e35d6191f4ca548e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/456f539b16053240e3d2a2f874a13738f8a39b4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/456f539b16053240e3d2a2f874a13738f8a39b4c"}]},"branch":"refs/heads/master"},"2934c2cfa29116668241ba818d11a0f255cf6bcf":{"kind":"REWORK","_number":7,"created":"2023-07-11 11:45:39.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/94/886994/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/886994/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/886994/7"}}},"commit":{"parents":[{"commit":"5bd21ba912f42ee6893ec5614609344da9e19f95","subject":"proxy: remove client_chunk_size and skip_bytes from GetOrHeadHandler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5bd21ba912f42ee6893ec5614609344da9e19f95"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-26 17:02:27.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-07-11 11:43:47.000000000","tz":60},"subject":"proxy: encapsulate Getter resp, node and parts_iter","message":"proxy: encapsulate Getter resp, node and parts_iter\n\nThe proxy GetterBase manages a set of attributes related to the\nbackend source for a GET response: source (a response object), node\nand source_parts_iter. These attributes are always updated together so\nbenefit from encapuslation, along with some helper methods to simplify\nthe GET paths.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I76ea042ef4b3f5cc1caa4774e35d6191f4ca548e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2934c2cfa29116668241ba818d11a0f255cf6bcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2934c2cfa29116668241ba818d11a0f255cf6bcf"}]},"branch":"refs/heads/master"},"46e0cc0a11f5afeb121907582c8184deec3e6ddd":{"kind":"REWORK","_number":8,"created":"2023-07-20 15:38:51.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/94/886994/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/886994/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/886994/8"}}},"commit":{"parents":[{"commit":"5bd21ba912f42ee6893ec5614609344da9e19f95","subject":"proxy: remove client_chunk_size and skip_bytes from GetOrHeadHandler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5bd21ba912f42ee6893ec5614609344da9e19f95"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-26 17:02:27.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-07-20 15:35:18.000000000","tz":60},"subject":"proxy: encapsulate Getter resp, node and parts_iter","message":"proxy: encapsulate Getter resp, node and parts_iter\n\nThe proxy GetterBase manages a set of attributes related to the\nbackend source for a GET response: source (a response object), node\nand source_parts_iter. These attributes are always updated together so\nbenefit from encapsulation, along with some helper methods to simplify\nthe GET paths.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I76ea042ef4b3f5cc1caa4774e35d6191f4ca548e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46e0cc0a11f5afeb121907582c8184deec3e6ddd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46e0cc0a11f5afeb121907582c8184deec3e6ddd"}]},"branch":"refs/heads/master"},"4de0d9ea86d4cfeb44f8c9f9ae495dd9a9bb226b":{"kind":"REWORK","_number":9,"created":"2023-07-24 03:39:38.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/94/886994/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/886994/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/886994/9"}}},"commit":{"parents":[{"commit":"5bd21ba912f42ee6893ec5614609344da9e19f95","subject":"proxy: remove client_chunk_size and skip_bytes from GetOrHeadHandler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5bd21ba912f42ee6893ec5614609344da9e19f95"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-26 17:02:27.000000000","tz":60},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-07-24 03:31:36.000000000","tz":600},"subject":"proxy: encapsulate Getter resp, node and parts_iter","message":"proxy: encapsulate Getter resp, node and parts_iter\n\nThe proxy GetterBase manages a set of attributes related to the\nbackend source for a GET response: source (a response object), node\nand source_parts_iter. These attributes are always updated together so\nbenefit from encapsulation, along with some helper methods to simplify\nthe GET paths.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I76ea042ef4b3f5cc1caa4774e35d6191f4ca548e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4de0d9ea86d4cfeb44f8c9f9ae495dd9a9bb226b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4de0d9ea86d4cfeb44f8c9f9ae495dd9a9bb226b"}]},"branch":"refs/heads/master"},"1d7d58dd4c3db09d53cd90bfa49855f83ba5ae6e":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2023-07-25 00:44:40.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/94/886994/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/886994/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/886994/10"}}},"commit":{"parents":[{"commit":"8b900becd5bed2e0c0555cc55bcaef372910b88b","subject":"Merge \"proxy: remove client_chunk_size and skip_bytes from GetOrHeadHandler\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b900becd5bed2e0c0555cc55bcaef372910b88b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-26 17:02:27.000000000","tz":60},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-07-25 00:44:22.000000000","tz":600},"subject":"proxy: encapsulate Getter resp, node and parts_iter","message":"proxy: encapsulate Getter resp, node and parts_iter\n\nThe proxy GetterBase manages a set of attributes related to the\nbackend source for a GET response: source (a response object), node\nand source_parts_iter. These attributes are always updated together so\nbenefit from encapsulation, along with some helper methods to simplify\nthe GET paths.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I76ea042ef4b3f5cc1caa4774e35d6191f4ca548e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d7d58dd4c3db09d53cd90bfa49855f83ba5ae6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d7d58dd4c3db09d53cd90bfa49855f83ba5ae6e"}]},"branch":"refs/heads/master"},"24d2976f63191f778bcf351d601876bdf988dbc4":{"kind":"REWORK","_number":11,"created":"2023-08-03 15:34:18.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/94/886994/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/886994/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/886994/11"}}},"commit":{"parents":[{"commit":"8b900becd5bed2e0c0555cc55bcaef372910b88b","subject":"Merge \"proxy: remove client_chunk_size and skip_bytes from GetOrHeadHandler\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b900becd5bed2e0c0555cc55bcaef372910b88b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-26 17:02:27.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-08-03 15:30:29.000000000","tz":60},"subject":"proxy: encapsulate Getter resp, node and parts_iter","message":"proxy: encapsulate Getter resp, node and parts_iter\n\nThe proxy GetterBase manages a set of attributes related to the\nbackend source for a GET response: source (a response object), node\nand source_parts_iter. These attributes are always updated together so\nbenefit from encapsulation, along with some helper methods to simplify\nthe GET paths.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I76ea042ef4b3f5cc1caa4774e35d6191f4ca548e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24d2976f63191f778bcf351d601876bdf988dbc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24d2976f63191f778bcf351d601876bdf988dbc4"}]},"branch":"refs/heads/master"},"f8a6dee622e47cec7111007069dd4ca74b56b49c":{"kind":"REWORK","_number":12,"created":"2023-08-07 10:19:42.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/94/886994/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/886994/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/886994/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/886994/12"}}},"commit":{"parents":[{"commit":"8b900becd5bed2e0c0555cc55bcaef372910b88b","subject":"Merge \"proxy: remove client_chunk_size and skip_bytes from GetOrHeadHandler\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b900becd5bed2e0c0555cc55bcaef372910b88b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-26 17:02:27.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-08-07 10:17:17.000000000","tz":60},"subject":"proxy: encapsulate Getter resp, node and parts_iter","message":"proxy: encapsulate Getter resp, node and parts_iter\n\nThe proxy GetterBase manages a set of attributes related to the\nbackend source for a GET response: source (a response object), node\nand source_parts_iter. These attributes are always updated together so\nbenefit from encapsulation, along with some helper methods to simplify\nthe GET paths.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I76ea042ef4b3f5cc1caa4774e35d6191f4ca548e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8a6dee622e47cec7111007069dd4ca74b56b49c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8a6dee622e47cec7111007069dd4ca74b56b49c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
