)]}'
{"id":"openstack%2Fswift~887230","triplet_id":"openstack%2Fswift~master~I7fc923ad0ef37459b7a76ce360dd7f320053d3f7","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-08-31 16:54:26.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-09-05 22:58:55.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I7fc923ad0ef37459b7a76ce360dd7f320053d3f7","subject":"slo: 500 if we can\u0027t load the manifest","status":"MERGED","created":"2023-06-28 20:22:43.000000000","updated":"2023-09-06 01:12:06.000000000","submitted":"2023-09-06 01:11:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"887230","meta_rev_id":"933b9b51341f3ccc3962ba089784fff26b5789a7","_number":887230,"virtual_id_number":887230,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-06 01:11:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2023-08-31 16:54:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2023-08-31 16:54:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-28 21:27:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-28 22:01:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-10 19:27:26.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"c49a0156c578696d903888a25113a2df166ba7ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-28 20:22:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b57abb96828788978d7b3ff798b46f520d5c29b3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-28 20:30:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af1ff60966cc4bb37854f5201fd5f346b1fee078","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 21:27:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/442e57d0b4754c719e36768c51c0c512\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3696589425a045ceb43ec4c29df6841e : SUCCESS in 17m 24s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a1477ec39a71467a8a85ac79e19f7332 : SUCCESS in 19m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f3228acb05824b768cad7797b39447ff : SUCCESS in 24m 42s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f13ad9b942174ea09e65588200113bea : SUCCESS in 1h 03m 11s","accounts_in_message":[],"_revision_number":1},{"id":"764e4a0f0bc78c5a96952ea449ec5fe3d3b50d8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 22:01:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4397ab0de23d49a199db315cd533e431\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccbf046894814c2db2d29b86eb65811a : SUCCESS in 8m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/5bfee57d26154f1d913a9959f50bcc2c : SUCCESS in 51m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a51e28ac859a4973ad22b8656f5bb877 : SUCCESS in 43m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f6b6fd2ed3824405a847fd3455b6bd68 : SUCCESS in 53m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/393c8df36c6f4e65b2a3cef998d4ee94 : SUCCESS in 27m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ba2a3396186e4f0b86439c4fbf666de4 : SUCCESS in 26m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/339cb6227de34a1e9e986cc3616fd543 : SUCCESS in 27m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ccba23d92d0543c8aca51e7ac04d30a6 : SUCCESS in 8m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c8ac9eae5d8247d4a93155d096ca855e : SUCCESS in 5m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7a0bafbb4d9b4e12a72a7e80de0d66a6 : SUCCESS in 28m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1a10df20112944feba72a4e396ea931c : SUCCESS in 13m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7cb2988a56bb45acb6ad76372a429276 : SUCCESS in 12m 30s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/72d9b687cbca475bb282b7f49d53bd83 : SUCCESS in 14m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e17367eda4e044f6aa25226e1786527e : SUCCESS in 15m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b949f87bcd554aca97328eb47cf7976c : SUCCESS in 14m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6f032701b20c4afe97a87b66f2239b2d : SUCCESS in 16m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/29560f8757b741f39e0c30c6b6cd0b31 : SUCCESS in 12m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/375e407f9b344e5fbcc61db8a084e829 : SUCCESS in 15m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a4c4ed420ab2494ca9ca3fe886ecec6c : SUCCESS in 16m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dc525b7829aa4eefa2e2751950011e4a : SUCCESS in 6m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/630852ccd94a45d9bd2cd7b9f04473a3 : SUCCESS in 17m 16s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e7d1894d23254ce0aaf3cbd2d28ff484 : SUCCESS in 7m 20s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/70cc08c304e44b078d1c4becb585ebe7 : SUCCESS in 1h 01m 16s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a603b01e9c9b4acd9b26627acf925397 : SUCCESS in 1h 05m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4dae3fc56320415e87c08fb598c33acf : SUCCESS in 48m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c99759dc0b1945b5a438a5b83001e186 : SUCCESS in 53m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a6daeaf19514dbc926b90d6260826fd : SUCCESS in 26m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4dabbbdf9ce4af78b0fa95cfe40ca57 : SUCCESS in 6m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/59261268f6b54d1f8c24fe4fe72fedff : SUCCESS in 22m 49s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2fae97d5ea14765a706eb08c89c13c0 : SUCCESS in 58m 33s","accounts_in_message":[],"_revision_number":1},{"id":"c3ed5919f568fa5814938a284f0f4e11699449ab","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-07-10 19:27:26.000000000","message":"Patch Set 1: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"028d95f2835c67f5b572315103c85fe26b8b9a84","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-07-11 00:20:22.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"050b22e4b051faba2a98d250c33bda1faac7f307","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-08-31 16:54:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"abe136d8a9c1e5ff6cb23d7056287e157fa973df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 16:55:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aaa5d71673c04be57c9763764230bf811bda3b39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 19:30:16.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed3b1372cd664f00b41d1407980364b2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1511d0c0e6284df6a66d6e48ed2e25aa : SUCCESS in 10m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/2ff883916a44469f9a3107a7847d03cb : SUCCESS in 43m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/80933c756d5d4c76a29ccc075d74ace1 : FAILURE in 37m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1242e6eb87c145d3ac4881e337d2a511 : SUCCESS in 59m 34s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6441013b10ec417a8998500cb466f3bb : SUCCESS in 10m 44s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/8c3c64f939604a8fbbf22af8f4030c8b : SUCCESS in 7m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/edcc9ccb907e468988be6e51f0ddc6d7 : SUCCESS in 17m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a965aff66ff542f198609a1a188340d7 : SUCCESS in 18m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/683a38a872724da990c7fd70d8f3caa3 : SUCCESS in 17m 33s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/966fbdb2c7374957b116dbfef3284204 : SUCCESS in 17m 30s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/fa1b7c1b4cbd48c59da0a38d651bdd06 : SUCCESS in 17m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3feff82d092649bf8694ebbfbc5e7914 : SUCCESS in 20m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c36c7ecee745419db8e333a6c26d2854 : SUCCESS in 21m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7a5b5edd70c641dfb0793068b278aa3f : SUCCESS in 16m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b1ec6d4a4a794a4ca82b21a66ee7f718 : SUCCESS in 15m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/485b1ea74431477b8b8774263a1b0d1f : SUCCESS in 17m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5df343f6596c49adb9745a6b995f77eb : SUCCESS in 16m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/312c025a45c44c72a9a0385ed30c5caf : SUCCESS in 8m 27s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d76ccb9a6e3b47448a295ad7943828a4 : SUCCESS in 8m 50s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9be7cacad06e4c8f9c8d96fa7df97791 : SUCCESS in 46m 21s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b365fa64bdf34f48a29cae4e32146592 : SUCCESS in 1h 01m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dc5552752db544e2a1659074c229f66f : SUCCESS in 46m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/489c54d243354aea974888ff6733c0d4 : SUCCESS in 39m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f213f6f31a044e1ea99f3bf115fbef8b : SUCCESS in 15m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0171d888c5f64903ba66faf26368ef3e : SUCCESS in 5m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b65699fb87041d8a661fdda3880ed5a : SUCCESS in 1h 08m 49s","accounts_in_message":[],"_revision_number":1},{"id":"8f177d4cdf223c3efbe6432cbf977294f40be602","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-05 15:11:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a0b1122d79c53228b9fb4d7a8d1671c9720187c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 16:15:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/030a18d1a2fa4bda98fb6b2f9077ffa0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c936c1f628c44f14a022831c9bb96482 : SUCCESS in 10m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/6e84249758d94f18950d3af02696d8d2 : SUCCESS in 51m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7a864a3be2df4c6989748c784ea86705 : SUCCESS in 54m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00f821ea31234c9a8fa08e7ea3b23e66 : SUCCESS in 1h 01m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f2f893373b6a4894815eb33b662af21f : SUCCESS in 18m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4600c8208e394a90912ef671de5e2c17 : SUCCESS in 21m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f0064b37774b4b42988ffb079e0875de : SUCCESS in 25m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/51f7e31d42b04ad9903a03af1089fe17 : SUCCESS in 10m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e5b00e9b8d0c4fa39731e3f83b7cfe8d : SUCCESS in 5m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a6e78f5ad9c94a019e017beffe022d51 : SUCCESS in 17m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/62780c5917d6412abf3e9cd92b1a44bb : SUCCESS in 26m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e087a442f8874bd2a534b98e2c533f28 : SUCCESS in 14m 19s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f0a66cf72e5d406f95f769b5a4144412 : SUCCESS in 18m 05s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/18861fdec9354c268305791a329c15fd : SUCCESS in 17m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7e26dd1779dd42438a9d995354e774e0 : SUCCESS in 17m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/88733edc3f3c436cbcc1ee20c4811258 : SUCCESS in 23m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f3cb69542e6740d3a1674430e09637d1 : SUCCESS in 20m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/db0e4e3847ce4e6aa5cc21a6c95a0496 : SUCCESS in 14m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7bad9f16a026429ebac0789495c1159a : SUCCESS in 15m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d77f406735ed4dd7aca38cd6f9ddc6fb : SUCCESS in 13m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e41fda6e928e441781e11df7ab7c1654 : SUCCESS in 7m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5ef88ed60fe5424ab8c025a4cf76a3b8 : SUCCESS in 15m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b1f1475678ff484ab464175f21374f61 : SUCCESS in 7m 45s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b1a428ba4c63435fb2298b75c7ccac2e : SUCCESS in 46m 12s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8081398b973e4267ac05602b856dba46 : SUCCESS in 56m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c76e7338eb3d429a989c7222d271f80e : SUCCESS in 37m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3fcb509076254b97a1a18b071bd28c63 : SUCCESS in 39m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6105c13bf03490899a83cbe6249d978 : SUCCESS in 14m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d921ae415eb44f42a0433b6491b17db7 : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cee3b087f90a4e90acfa2d60547057a3 : SUCCESS in 18m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3840947de34484f973b95f6bb2afd3f : SUCCESS in 46m 20s","accounts_in_message":[],"_revision_number":1},{"id":"98ab846c2f6994cb622f91c3bce786cf1e1bec89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 16:16:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bf2b1027b7dbad60ba8744cfb6f2433178ea2a19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 16:33:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4912e1e936f744e4b91002331889fa47\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/517710f982704e35b8efbd33c6b9c079 : SUCCESS in 14m 48s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/51a0fc4a68e54250b6b68ed2a59eb381 : SUCCESS in 14m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d88449af22334c499b23b7b3d61399f1 : SUCCESS in 17m 37s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/463728f02a9f422da3c616fd503c8bb4 : SUCCESS in 1h 20m 09s","accounts_in_message":[],"_revision_number":1},{"id":"45a10cfe7c326d1a9c60ff2608c9172b8594e413","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 18:17:01.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/299c8c2d58b14b1ea6cf1e16ccf95a46\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43c615d457df495487406a3b1ffd93d0 : SUCCESS in 7m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/75d4e803a01740339d11eaaa8f161d55 : SUCCESS in 49m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/91726c4de3ea451d9b9edc8c30140198 : SUCCESS in 39m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1571e2c1ad434b5d9b33bb1bd0661f46 : SUCCESS in 1h 04m 24s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f3c64bc349944772ab26bdfa32dfccd1 : SUCCESS in 9m 00s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/8e5ecf9bb33047ee855a5ce595a35e86 : SUCCESS in 6m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5c5d0b71994f438f8a8f0e7ce0ce2b26 : SUCCESS in 19m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/30b9e0a2088e4be7a4f2a53c58726d35 : SUCCESS in 15m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ced6fbf933aa4332a9363c024f111b22 : SUCCESS in 11m 18s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/86f8e897d3374d91a1df429a255e31e6 : SUCCESS in 12m 09s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1b3f5357971a4eeba8e57a022e5e535a : POST_FAILURE in 18m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e8ba1cfe7075430b904129f4895ba7c8 : SUCCESS in 18m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5921f4ed39c140a1a3cb4147d46d1c36 : SUCCESS in 17m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aeaef4d57ae74ec3b384463ed77746d8 : SUCCESS in 14m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0992600fab90451d80fdcbfbd18cf579 : SUCCESS in 15m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1c765d7c4b2a45439b4471a0a37065de : SUCCESS in 14m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cda040ef07f94ff2835b842be5c7ea36 : SUCCESS in 15m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9b03a87a62a944368bc41a7457782480 : SUCCESS in 6m 53s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3a7efc992ecb4f838072aabbc6bf1841 : SUCCESS in 7m 27s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f1ef2eeb19d94c688ab4d8364c4dea45 : SUCCESS in 1h 04m 10s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/af2db4e1edda4edabb41fe0bfa2c4b74 : SUCCESS in 59m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/435b10db67f84dd6b7377d6183abeda6 : SUCCESS in 47m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1800af17f4c04a5d841d48e13739d7e5 : SUCCESS in 45m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/992cbdbace2a409a925e00b7931fb1f9 : SUCCESS in 17m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce9602e0fa524e3a8491611fc1eb37ed : SUCCESS in 4m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9147fcc672f24a07b07d679d97423383 : SUCCESS in 44m 16s","accounts_in_message":[],"_revision_number":1},{"id":"54b14d7519128a818e69bcd7867840fa6b568b06","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-05 22:58:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"03647b6bcd2252c1b79af874bfdc2c6126dbe1b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 23:16:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/594b4a8c4a6e4aa4bae0479a37703a21\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/404fed7b33e5473fb50a5254c41d346a : SUCCESS in 13m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f4e80a1841e94c9eb13dfc114e55bf6f : SUCCESS in 14m 24s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0187158ceb2c4be9baf47bbc0d2128e1 : SUCCESS in 16m 48s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a8ee02f564f94b30abec66f2aa1cbdcf : SUCCESS in 16m 29s","accounts_in_message":[],"_revision_number":1},{"id":"b213dd0e2339d1230130ef7ea5ed8766d722b777","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 00:02:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be12927cd23642e08fd7c801bcaa734a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4488bd42fcab451f8a6bbe559a47df49 : SUCCESS in 8m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/22ffead117cc4a94ac6877adae523734 : SUCCESS in 43m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/50c483d25edb4bf39a4f87ba34a3d834 : SUCCESS in 52m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0931faf03c0649b69b720b0bb1a139aa : SUCCESS in 44m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aab256130eb7422a86dbfe1bb2b5b696 : SUCCESS in 19m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5a63c8fd407e42e5be277b6b1179645d : SUCCESS in 18m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed9750f991ef4f4687c9583bbcecc0ed : SUCCESS in 19m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c3be1565a61a4a50b44ec8f67b1685af : SUCCESS in 10m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/07a91e965e77488aa3afbf3bfc90b400 : SUCCESS in 8m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/80c96a702a4848d194ad976fe1402eae : SUCCESS in 15m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9c28aab2277647578e4510d049eb45a7 : SUCCESS in 16m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/add4b9ec861e4edb9b2d469510d9294c : SUCCESS in 16m 35s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/cd0cbc19a50d40d39cabe5d9f7bb5d06 : SUCCESS in 16m 33s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/baef59a615f4441b94dcb3058a926339 : SUCCESS in 15m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ece2c75c873d497695a96cab0f67f362 : SUCCESS in 17m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/26737eef625449599c1b49c674ccbbf7 : SUCCESS in 13m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/acea673b653a4e8d945701b274842462 : SUCCESS in 20m 13s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c5ca19171a204338b70d3b41b84a378b : SUCCESS in 15m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/42c2df0808094e36979a98c6069220ec : SUCCESS in 12m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3a0d0422a02a4d4b81034477b2da9d19 : SUCCESS in 20m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/31f415f708d24de0bdeb5a5d712ce4de : SUCCESS in 6m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/26883d32c40c4c1396c4ea91c129f084 : SUCCESS in 16m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3fde15f769734a40bd161004d6f0426f : SUCCESS in 9m 19s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a9bcb86f292b4dc8a4fdf659726e471f : SUCCESS in 57m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a2e5b864e38e4bf781ca86ecf1deddff : SUCCESS in 46m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0f0beb29d2a64525ae7ac63181b74f8a : SUCCESS in 43m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fafcdb4564164a6cb8cc89ed4ee1add0 : SUCCESS in 36m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c55a5d5321a2411fac8d4faa9beca17a : SUCCESS in 18m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1b4078a38764ed89b54ab0eef8561ed : SUCCESS in 4m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9de75597d93a4aed86d274c0b89b089a : SUCCESS in 21m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef9dbe16e6bd47588b5763b895ac1e57 : SUCCESS in 45m 25s","accounts_in_message":[],"_revision_number":1},{"id":"8bfb0d6afe0460423c2705b518577c1a26b3f8a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 00:03:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d0fd43247279e9c29b16233dd1f660b618518040","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 01:11:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5505d268cc9648648d426c93745c3f39\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cd6eab400044c2abb7880e4852252fc : SUCCESS in 7m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/bf366047762241a0b8bfc8f6f5a46a19 : SUCCESS in 41m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d21ab7feee9045b584cdb7955b1e95cb : SUCCESS in 47m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/09a7c7babad348c7b1404f275d566dcc : SUCCESS in 1h 01m 03s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d28d72432ba74c878051a648ac2d12c5 : SUCCESS in 9m 57s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/247568f2d24f414fadc42ca59b107217 : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/01a8bfea80ab47e3ae68a4838dde39f4 : SUCCESS in 16m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5120ace667cd4fa5904ec767d629eecb : SUCCESS in 15m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0727e72144c040988df397c9d2f20aa4 : SUCCESS in 14m 59s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7071ff50dfa94f04b0653d3b251ce0ba : SUCCESS in 11m 27s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/70ecb27e56284c2b874bfac5f22a2084 : SUCCESS in 15m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/09b37a40e8e84294851d17865cb23124 : SUCCESS in 16m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d95209eb5fb44d54bffc02e6348c918e : SUCCESS in 17m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/28ae8a91224d4d7282e36be02d1130bd : SUCCESS in 17m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/43c61ee596f54c31b50a879bc2751c19 : SUCCESS in 13m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/80ff8389c3ac40268d65fc6e7ba13884 : SUCCESS in 11m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cb90f71bee844c95af762687f36b1826 : SUCCESS in 15m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e65214b46d7a4c09a230da9d0507ca61 : SUCCESS in 6m 03s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/96ed7abd05a64c5093f6ce7fabf12dbb : SUCCESS in 6m 59s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4640d447ece3495ba49dd860b81d3a15 : SUCCESS in 49m 41s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/fa61eff795d34a76910f26b398ae445a : SUCCESS in 46m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2a316fa6d7b7437d9b17f6e83e77cd89 : SUCCESS in 40m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6de55e570fa247c9be51f367911700a4 : SUCCESS in 39m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3dabc77c47c42669c25bdfee51cec3d : SUCCESS in 13m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a7cd0f5e8e641ab9f1a54e212720c1b : SUCCESS in 4m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f35dc2316c3d4e6ca2dc951db9b9ab25 : SUCCESS in 39m 00s","accounts_in_message":[],"_revision_number":1},{"id":"cc5047b714a165ea365e683e4131b169f229eacc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 01:11:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"933b9b51341f3ccc3962ba089784fff26b5789a7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 01:12:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67e2bc74559d483e928f066094943a91\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e27e6cbca74143299362220ab767b18d : SUCCESS in 43s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/61051461a6a64e97a341ba66b0e34f2f : SUCCESS in 30s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/e0688ae16d2a4380a41a7b3e818fc515 : SUCCESS in 45s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6e5b5a659aca10789a53816a775a566719707478","revisions":{"6e5b5a659aca10789a53816a775a566719707478":{"kind":"REWORK","_number":1,"created":"2023-06-28 20:22:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/30/887230/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/30/887230/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/887230/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/887230/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/887230/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/30/887230/1"}}},"commit":{"parents":[{"commit":"39127fee5be6eb7733938d31e82bb6553ea9cbeb","subject":"Merge \"sharder: Fix completed metric emission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39127fee5be6eb7733938d31e82bb6553ea9cbeb"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-28 20:04:47.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-28 20:05:26.000000000","tz":-420},"subject":"slo: 500 if we can\u0027t load the manifest","message":"slo: 500 if we can\u0027t load the manifest\n\nWhen the proxy app is trying to send back an object and hits an error\n(maybe a timeout, maybe an EC decode error) *after* it has sent headers\nand started streaming data, it just stops sending data, expecting\nclients to notice the discrepency in Content-Length and retry.\n\nWhen that happened in SLO while reading a manifest, previously we\u0027d just\nassume the manifest is empty and send back an empty response. This would\ncause confusion for users (who\u0027d think we lost data or soemthing) and\nwas clearly wrong.\n\nNow, return a 500 to the client. Retrying is perfectly reasonable.\n\nChange-Id: I7fc923ad0ef37459b7a76ce360dd7f320053d3f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e5b5a659aca10789a53816a775a566719707478"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e5b5a659aca10789a53816a775a566719707478"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
