)]}'
{"id":"openstack%2Fswift~887235","triplet_id":"openstack%2Fswift~master~I31f622f6ebfa255d4c52089ca182cbf665c40a30","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-09-05 22:58:17.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_15343\u003e by clicking the attention icon","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I31f622f6ebfa255d4c52089ca182cbf665c40a30","subject":"proxy: Pop ignore-range backend header on resume","status":"MERGED","created":"2023-06-28 23:07:40.000000000","updated":"2023-09-05 22:58:17.000000000","submitted":"2023-07-24 06:12:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"887235","meta_rev_id":"88728712cbdcdb9d633ebbe2d87c34d4963d5fda","_number":887235,"virtual_id_number":887235,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-24 06:12:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2023-07-24 05:04:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2023-07-12 11:20:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":1,"date":"2023-07-06 14:32:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2023-07-24 05:04:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-28 23:27:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-29 00:23:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-06 14:32:59.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"REVIEWER"},{"updated":"2023-07-10 12:29:52.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2023-07-10 16:36:10.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2023-07-24 05:04:13.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"a158d51aeff4cffcd5d9aeb20237275c9a0da8af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-28 23:07:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ade3cd577b692be02b6089c3fdd253905828aa2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 23:27:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9d99924b2a14cf498aa00194170d4b7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/daea94619be349e991177a00efa2ab7c : SUCCESS in 17m 31s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/71064393f93547e88cf20ce0a17d29ea : SUCCESS in 15m 23s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/94b88345944546b28888f22555cf41e3 : SUCCESS in 19m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b3a99924392641e686135903641358a4 : SUCCESS in 17m 09s","accounts_in_message":[],"_revision_number":1},{"id":"18375a188985b78d51c886f305f52d9a7b3205f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 00:23:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eeffc1ab6b9245588f11e789504ee272\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c069aadcc9624bd6811ef1e2639e4211 : SUCCESS in 8m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/8a0c42d144a649468ace1da41f0a2bea : SUCCESS in 41m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f2cc9ab38d0c4b0bb07c40f2c499aec1 : SUCCESS in 39m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/36d79418c7654b9390039caf1540e610 : SUCCESS in 43m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/49809a956f84473cbb102bd71db69180 : SUCCESS in 17m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c1ea7216d0494a8eba9694ab6fa67865 : SUCCESS in 19m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/45a0db463a3f441c9ea0773cdf0fa0ab : SUCCESS in 19m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a9cf1c40bc6e4838860a52acb320035d : SUCCESS in 7m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6751e96e48b845de84b422f4f58ba8d2 : SUCCESS in 4m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7ddd280beb924ce9ae17eadac2f43bbf : SUCCESS in 17m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/80e1aa48dcb446d48aab87aca07a7693 : SUCCESS in 17m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f2be2f386717481483eab670cf519bc1 : SUCCESS in 18m 02s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/37eb4b833b19431f9c0397a1ec0bb847 : SUCCESS in 18m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/83bb5246f0744e459f6779c72aa1f803 : SUCCESS in 17m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ca17a1c054ac4fefaa6fcc66aa91d54a : SUCCESS in 17m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d0834dde2d7a4abe8697e301069d64ea : SUCCESS in 15m 17s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/326d141a8140411fbd5a18fec785076f : SUCCESS in 11m 06s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3c76ca76eeca4517990c800eacae7409 : SUCCESS in 12m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/99e68125971849f5b4c7a88983ec279e : SUCCESS in 12m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/17181484dac34dab9ecce3e9bbc608b0 : SUCCESS in 6m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/aa1a33cf12c9477c8adc92c64d54b3d5 : SUCCESS in 17m 03s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c310b80c491e4b65bcda2ab0cce7e2e8 : SUCCESS in 7m 20s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a87ed3c877df402c9b8d8139c81d3e92 : SUCCESS in 1h 06m 15s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ad1b571ab8f4441fa7c86c0d6e79c746 : SUCCESS in 1h 03m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/76baad33abcc4ccfa6bf80a1febea4a9 : SUCCESS in 1h 03m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e5694f2d14f3473cbe36f6c632ef920a : SUCCESS in 37m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0b6238aec714d40908af84bc56975b3 : SUCCESS in 12m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b856a7b3f46149f5a0747c61de4e1335 : SUCCESS in 4m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b3c69d5cb4724864bf2990b01a7437f8 : SUCCESS in 20m 43s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30a58b0d341843cc983a41dc6ed11a24 : SUCCESS in 39m 32s","accounts_in_message":[],"_revision_number":1},{"id":"66432bb99ef72f64d536bd2f646d0e35a49e960b","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-07-06 14:32:59.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9610dcf9625827cec60cb78c330603d65a151cd0","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-10 12:29:52.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5c1f3c84f32a856e621330d8ae33d2f1613a8995","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-10 16:36:10.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2e54d8c9fc4c5ffb6e53e507d1c63cfe6d6b3205","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-12 11:20:35.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c75bceea6f47e2bcf031b1e51cbf9a812956d5b8","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-07-24 05:04:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bab4f6c7170529afedea6a12063a468ee4d7dd48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 05:04:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"06bf194a0f973c77c85b56c3339afe8c7933d95e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 06:12:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb13a093815e4d8caa9ba5c495fd577d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f95304e43fb4c088768e85cc07d964d : SUCCESS in 9m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/4cbad65558d24f33a22daf2d796a25c2 : SUCCESS in 57m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e6adbe2cc865466c9d5e02ccf19ca118 : SUCCESS in 55m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5be1ceee32b435184a89dbc74830528 : SUCCESS in 59m 31s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/01f5214775b542de8aad7eab0072e820 : SUCCESS in 10m 28s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/3ad8239a0d29489da280223dba91db4d : SUCCESS in 9m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c75d353bd06145d788278a1d89195e72 : SUCCESS in 17m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5e6ae93c67ef436fab603d15baab4b43 : SUCCESS in 17m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/536e285881314fc59c491c7277af0175 : SUCCESS in 12m 35s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e1a340668f8d4013842cbe0d9d911d1d : SUCCESS in 17m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/943eb1081a924d1394947196526844b7 : SUCCESS in 11m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0de7608f5d8b4951bd769618fac84870 : SUCCESS in 23m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8c40d6efd9f848aca3d66fbd698aea36 : SUCCESS in 25m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2f9922e7441943b6937f4c3b76a6e937 : SUCCESS in 24m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/68734f3a04ee48e98eaae1a2570a4fa2 : SUCCESS in 20m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d6c3f9ea060e47968891d0b2101a07fd : SUCCESS in 13m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d206bb93b36c4129b857990d33166a9c : SUCCESS in 24m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ced22ec788054d1688f1f0f819279174 : SUCCESS in 6m 49s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7ad761693d794c4dadc67359518c1b32 : SUCCESS in 7m 05s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/23e63f9dc0564fd59fc2a13fc599f90a : SUCCESS in 58m 28s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/bf0d0f8ca91641d8b923118baed3a434 : SUCCESS in 1h 07m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4e0e001e30354e5783df4c6ffb07018f : SUCCESS in 44m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/448008c71f3044c8a7d0ba17fb632473 : SUCCESS in 53m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/677fda8dae904160b260ae5a0c2b58de : SUCCESS in 17m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55a4b8c394cd4b66a0d5bc90edb1afb8 : SUCCESS in 5m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72c132116409443e9711b8e231cb8084 : SUCCESS in 1h 01m 14s","accounts_in_message":[],"_revision_number":1},{"id":"3455995f907f7bb81e4c0b68d7a91f53b5a11429","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 06:12:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"184ed0bb9a9c8990525100e46811e7b454a8a86a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 06:13:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd9aee0e15944a7096bd354e7dd5e243\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/574971e6a84d4baab049487afc39bf48 : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/7d60b18a4ec54138a82ee206f9958b4c : SUCCESS in 28s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/bb503f54d53146c0b05a69a698452c74 : SUCCESS in 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dc3a6f3c0471e3a852a212c435711784401fc23a","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-17 16:14:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"20757612c309631fcdd729a03e14ca71d1c55790","revisions":{"20757612c309631fcdd729a03e14ca71d1c55790":{"kind":"REWORK","_number":1,"created":"2023-06-28 23:07:40.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/887235/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/887235/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/887235/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/887235/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/887235/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/887235/1"}}},"commit":{"parents":[{"commit":"39127fee5be6eb7733938d31e82bb6553ea9cbeb","subject":"Merge \"sharder: Fix completed metric emission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39127fee5be6eb7733938d31e82bb6553ea9cbeb"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-28 22:50:50.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-28 22:53:02.000000000","tz":-420},"subject":"proxy: Pop ignore-range backend header on resume","message":"proxy: Pop ignore-range backend header on resume\n\nThe proxy has a class of \"recoverable\" timeouts; when they trip, it\ntries to resume from another node, with a range offset that accounts\nfor the bytes already served.\n\nIf we set the X-Backend-Ignore-Range-If-Metadata-Present header (like\nwe do in SLO), it may cause the object servers to return the *whole*\nobject. When that happens on the initial GET, the proxy is smart enough\nto still track the appropriate range to fetch when resuming -- but then\nit would send the ignore-range header so the new object-server would\nsend the whole object.\n\nThis would cause problems in one of two ways:\n\n    For replicated policies, too many bytes would be yielded out of\n    the proxy app. proxy-logging would notice, truncate the response,\n    and log a warning about it.\n\n    For EC policies, at least one frag iter would have the wrong\n    length, causing a decode error from pyeclib.\n\nNow, pop off that header, if present.\n\nChange-Id: I31f622f6ebfa255d4c52089ca182cbf665c40a30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20757612c309631fcdd729a03e14ca71d1c55790"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20757612c309631fcdd729a03e14ca71d1c55790"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
