)]}'
{"id":"openstack%2Fswift~887267","triplet_id":"openstack%2Fswift~master~I4976b3ee24e4ec498c66359f391813261d42c495","project":"openstack/swift","branch":"master","topic":"p-s3api-metrics","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-07-05 22:48:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4976b3ee24e4ec498c66359f391813261d42c495","subject":"s3api: emit metrics for error responses","status":"MERGED","created":"2023-06-29 11:28:18.000000000","updated":"2023-07-05 22:48:43.000000000","submitted":"2023-07-05 22:48:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"887267-p-s3api-metrics","meta_rev_id":"68607f2a8711c547cce8fa214ef68b3ef8cea6a7","_number":887267,"virtual_id_number":887267,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-05 22:48:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2023-07-04 04:06:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-07-05 21:34:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-29 12:38:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-29 13:40:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-03 08:21:20.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2023-07-05 21:34:39.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"82553e6f26835df042813037f1580a6d0b9f26ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-29 11:28:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e10bf6b439cf240e00674b96b719d173f96bf89","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 12:38:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11574c4b6af54bae9114c94c1ac9ebdb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c79fa49919344a8d9c6b52f75eb5aaaf : SUCCESS in 14m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f28b1be32d6b442fbe133da9925ebbf6 : SUCCESS in 16m 51s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c92e56f0bcf4a96b39c257064a8023b : SUCCESS in 20m 44s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e41819a94e24ed291ef944778516f3f : SUCCESS in 1h 07m 14s","accounts_in_message":[],"_revision_number":1},{"id":"dc2074620ef9ee5d44ba28203ac258ed6bd0b05f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 13:40:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c7410306a0c4a739f19fba63a536da2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59837cb41c7c4d6eb0ba20daee61ceec : SUCCESS in 10m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/83ece77b5f2d4b59af02317dfdb2bb5d : SUCCESS in 1h 08m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7ea73092671940c2b5af2cbf53309154 : SUCCESS in 1h 05m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d354cac189a42b590efb8ccc53ab2ce : SUCCESS in 1h 02m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/879ed0f123634b6a9b7053d4d26257e9 : SUCCESS in 21m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e0a4c0e95d094e46a94b1712b9cd60dd : SUCCESS in 19m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1d7e30d8c80a4caa938ea0338a6d226e : SUCCESS in 17m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/87505136b9b74221803846ce9aa7c555 : SUCCESS in 9m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a078beca5e4c40b994281f88e6c911d1 : SUCCESS in 5m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/90f8f006ce87436fba2dade2f24d1006 : SUCCESS in 24m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c906c3c9c3cb421f990ecf4e7e6e0c1e : SUCCESS in 20m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d2bc7a0b5e95457fb02e3d7a37e5efad : SUCCESS in 20m 55s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/eeb25d96452b495dbc896ee5169705a1 : SUCCESS in 18m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fbf380f0b231427fad0db6c858a63cc8 : SUCCESS in 19m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fee6e2ff6ab94cf38178b39910a44cb4 : SUCCESS in 14m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ac415cd6c92c4f9c9ed7bc42b6d37c40 : SUCCESS in 15m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3d17c1f03ae34e28bb315ab2f26d513a : SUCCESS in 14m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5ecd8a3c81c749f3b30010b146066fd2 : SUCCESS in 15m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f37581d2c83440f9bcaacd6cde799699 : SUCCESS in 14m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5754aafaf8a74956ba6cba122a2ef9bc : SUCCESS in 6m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/25b785e785fb43499c90c1c07037be64 : SUCCESS in 14m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/902aea0aa8454b9e85a72f52fbfec5d6 : SUCCESS in 6m 34s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/93df07c666b04b7b90d5e645347aed75 : SUCCESS in 52m 58s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8372537cd4a84c9daad3fcb5eeb32a41 : SUCCESS in 59m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/88fd7aaa39d44a99813a05c89b828d5a : TIMED_OUT in 1h 34m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8fbac5a643c44ac3aeab425933b93b5b : SUCCESS in 51m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a9b46e250864c128685456f3002b236 : SUCCESS in 22m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0318eca7bec04ecea3c812678d5d91c2 : SUCCESS in 4m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2ae5414ace414d8d89b0e9f60f114fdc : SUCCESS in 18m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64771689c9b84aa8ad20634e67bb2be4 : SUCCESS in 1h 00m 57s","accounts_in_message":[],"_revision_number":1},{"id":"61f2a94c9591f6f508b073c4704d6f9f2e71e2b8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-30 16:36:26.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8811225505666b7eaafdbf95c1585971d6dc9775","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 17:37:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b870c35a916b4efca969a3dbe3665908\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ef40ec8d917045c2a8071006c6ee0793 : SUCCESS in 17m 25s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7740800ac8014256a42d710036a6ac60 : SUCCESS in 18m 37s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c64cbad4369d4026ae566ab5534e5605 : SUCCESS in 59m 23s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d632b4ae5e8d480bad1a21c43fbb4207 : SUCCESS in 20m 09s","accounts_in_message":[],"_revision_number":1},{"id":"268e43d3abc2fef8de4bc2e315b9f41859c4c56f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 17:55:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d7ff2dadb514cebab263b0974b17950\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44b93446dbb2428e850bb78271389985 : SUCCESS in 8m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/49a4ad2062b940fd9a582934d1392d70 : SUCCESS in 42m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b365bf751f4a4fd3ab5f50f205eb6ae6 : SUCCESS in 42m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52008764a0704a469327f7d591f8bbd4 : SUCCESS in 1h 00m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d370041215594afbbb039ce872f1332d : SUCCESS in 17m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b40b14a0e98f4bb3a2a7f6535c43716a : SUCCESS in 28m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f2364f771634461ebfc11886ddd9308f : SUCCESS in 31m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b626631db4ef4ea2b846273a8041343f : SUCCESS in 8m 46s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/acd50f36e4a14772a8e579bd4b46ebf5 : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/baedfa8663fc435e8e2d4cbcaf34a77d : SUCCESS in 17m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1cee35b1de87439a97517e65d3ccf56a : SUCCESS in 31m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3bd6ac21b18a4842a2017b7d01d55966 : SUCCESS in 14m 01s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ad5416dfc9ba497587b67df16c00dee5 : SUCCESS in 17m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/de821d2557a44618a7b818ecedc73581 : SUCCESS in 17m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6d85fc46ec3c4328ad96315bb891164e : SUCCESS in 15m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f3e98c29057d4136a24c07612fcdc4fc : SUCCESS in 19m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5ad03affb4864eed9825f654dfc520b6 : SUCCESS in 12m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1308e2b964a1479396bab7de350895a4 : SUCCESS in 15m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/61458ff85f4f404fbe91cf219a90aea3 : SUCCESS in 13m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/01ed084d40884c5390e6beb75809e8cf : SUCCESS in 6m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c9d97fd32a354d8499d7e4aa89c05a68 : SUCCESS in 15m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ecbbc5af369c464e8e0df92276189cfa : SUCCESS in 8m 22s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/64f5e47110d04ba6bca17792e158eb5c : SUCCESS in 55m 29s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e2960d2df9514272a9de4b8470804376 : SUCCESS in 1h 07m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b3b2db13ea09430e8714b01436823397 : SUCCESS in 39m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/201c30234aa4487fb262de0995376d07 : SUCCESS in 1h 10m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f166ff5587f4607a91980db1dd2a739 : SUCCESS in 14m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13576986462a4c64bae5c300399336dc : SUCCESS in 4m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/52aa189f8eb64a9a89d2031d37910e5c : SUCCESS in 24m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf30a58b33c64bb297e622f971b69d12 : SUCCESS in 42m 47s","accounts_in_message":[],"_revision_number":1},{"id":"eb9f4f2f33eac3a829d533be58605cdea0549eed","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-07-03 08:21:20.000000000","message":"Patch Set 1: Code-Review+1\n\n(4 comments)\n\nFrom an initial look, its looking awesome Al. Ill bring this up on a vsaio with statsd attached just to confirm.\n\nThere are alot of repeating reasons which, with metrics, could cause issues if we error one... so maybe we\u0027d want to think about constanting them, at least for ones we keep repeating.. although I\u0027m not convinced it\u0027s a blocker.","accounts_in_message":[],"_revision_number":1},{"id":"47b6e5d37f95319f4f26a14441cb90fc3a0c3672","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-07-04 04:06:00.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3662fb2a4a1575b5c595f40b98db316da64931d6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-07-05 21:34:39.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"43473bde960e6dee9fc42cba36a748d31d1f5433","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 21:35:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cb2b7d75c2eb599b6c116b6e40298846ad1e3400","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 22:48:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69b377068ac549f6a98c9cbeb5825ddd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d90b8ebc115345b5a432c44a45502898 : SUCCESS in 9m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/4046f236a3de49e6ab9794b17a565c5c : SUCCESS in 1h 05m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3642d3419b76484f81c2b1328f0ff819 : SUCCESS in 55m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa86e90ae6cb4772bffe113e6f91b298 : SUCCESS in 57m 38s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/823c9b6a166d4b6abb6dfca261c4aeaf : SUCCESS in 7m 47s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/e7eeb1cb76914bab9a2fb1e0ceda7787 : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6f2c2418558d4c0da0528561865ca7e5 : SUCCESS in 12m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/96e3ef088e9e4399850786fa770d3836 : SUCCESS in 15m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/59af7848036b4c778171e863b430bf95 : SUCCESS in 28m 30s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e1101f02969e42bb98b65e321b49df53 : SUCCESS in 15m 27s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/7b475257cf794743904369d8c07df4a5 : SUCCESS in 12m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6078935845924b44a20068507eb1ab18 : SUCCESS in 28m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/99222935f1844b56bfccca5a82697516 : SUCCESS in 17m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/80bcfd13b1924200ba22dbd5b800028f : SUCCESS in 17m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/63e5732b663c4b2180d96c0b33c89087 : SUCCESS in 11m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fbebb73c958a4ea39428a6a04ea7cba4 : SUCCESS in 12m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/431b683c52ed4f2cbc7fc8509926c7b0 : SUCCESS in 18m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9ae876b4f0724c07bd573644fba786d1 : SUCCESS in 6m 39s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/adf746b47f884a399c92a5ac06709495 : SUCCESS in 7m 46s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2a7489685cb743a4be79138c4dba2b85 : SUCCESS in 1h 01m 43s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/93257207a1a34b899312dfe75ae6bcd8 : SUCCESS in 1h 04m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6ea55bf3dc284635b2a676c3799e9aa1 : SUCCESS in 58m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/64f2f32143b84f6a96e173ed51fbf5a1 : SUCCESS in 48m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6040446adf2b448784fcf4c567523354 : SUCCESS in 17m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/234613307acf4b789c5caa53141dbce7 : SUCCESS in 5m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d35451caa653428bb239bac4e67bdc65 : SUCCESS in 49m 26s","accounts_in_message":[],"_revision_number":1},{"id":"68607f2a8711c547cce8fa214ef68b3ef8cea6a7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 22:48:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f9af0b70b7ce22014e5756d3bdf863f929f6c71b","revisions":{"f9af0b70b7ce22014e5756d3bdf863f929f6c71b":{"kind":"REWORK","_number":1,"created":"2023-06-29 11:28:18.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/67/887267/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/887267/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/887267/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/887267/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/887267/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/887267/1"}}},"commit":{"parents":[{"commit":"11296f1642e1b1948cec5c98dccb2504c6cac4be","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11296f1642e1b1948cec5c98dccb2504c6cac4be"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-29 09:01:53.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-29 11:27:57.000000000","tz":60},"subject":"s3api: emit metrics for error responses","message":"s3api: emit metrics for error responses\n\nUsers sometimes ask why their request received a 403\nresponse. Sometimes s3api will include a reason in the response body,\nbut client code may not make this visible to the user. To provide some\nother insights, this patch adds statsd metrics when error responses,\nsuch as 403, are returned from the s3api middleware.\n\nThe new metrics have the form:\n\n  s3api.\u003cstatus_int\u003e.\u003cerror_class\u003e[.reason]\n\nFor example:\n\ns3api.403.SignatureDoesNotMatch\ns3api.403.RequestTimeTooSkewed\ns3api.403.AccessDenied.invalid_date\ns3api.400.InvalidBucketName\n\nChange-Id: I4976b3ee24e4ec498c66359f391813261d42c495\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9af0b70b7ce22014e5756d3bdf863f929f6c71b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9af0b70b7ce22014e5756d3bdf863f929f6c71b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
