)]}'
{"id":"openstack%2Fswift~887365","triplet_id":"openstack%2Fswift~master~Ie63fdd8f3febe441963bc5755175878433a8c4c7","project":"openstack/swift","branch":"master","topic":"p-fix-test-source","hashtags":[],"change_id":"Ie63fdd8f3febe441963bc5755175878433a8c4c7","subject":"Rename TestSource which is not a TestCase","status":"MERGED","created":"2023-06-30 10:32:33.000000000","updated":"2023-06-30 12:55:48.000000000","submitted":"2023-06-30 12:54:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887365-p-fix-test-source","meta_rev_id":"718dc2c3c995bd31ffecbef13b120b028af60532","_number":887365,"virtual_id_number":887365,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-30 12:54:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2023-06-30 11:54:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2023-06-30 11:54:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-30 11:18:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"76f31998c3626e41718fd0e3038c0aa458a9d875","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-30 10:32:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e4c7a371f067668ceed67e6f7cef4cd1bd91ca7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 11:18:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6a2ea03a0db4dcc86d0544f69648e84\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/722577764e0f4e0e909bf74714ef0653 : SUCCESS in 8m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e510f6d018f8476bbaff593b7aaaf095 : SUCCESS in 15m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8e5c0e1e07e943a8a9b392001df78f73 : SUCCESS in 26m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/99590041550442aeabd1469bd7fcccfe : SUCCESS in 20m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/53cac5a88b654f2eb5de1fa5bceb12a4 : SUCCESS in 9m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e254ce0846dd42639f6f9786f60f8509 : SUCCESS in 5m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9698fa0b7c2a43fb9a802cda8cba17af : SUCCESS in 17m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/79a51959794f4c908452c55daf08b158 : SUCCESS in 18m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7accafc6621f45a49fe9fbfbcd7b8699 : SUCCESS in 15m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6924ec3be097478b8517429529d1787a : SUCCESS in 17m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/74ed7c1f544a451983ac74454226f36b : SUCCESS in 16m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2d10fc3151b34ddfbac50f6f9e4f76b9 : SUCCESS in 17m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3ef69a61bb294a5cb7c0d35164c2d7f4 : SUCCESS in 17m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8d74a617fa744488a875ccafc87e4e29 : SUCCESS in 12m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5ffb10453ac04bc8996d5865c0b2a4d4 : SUCCESS in 18m 07s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/72c0aa168be94ed493aabc0235403fc8 : SUCCESS in 14m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/44386fe397e34376a0ac549821c9ad10 : SUCCESS in 32m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cff3f69a0e6a477180b8d83c66fb07e7 : SUCCESS in 44m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7169a9029b24ff48560d9f1e8bac5f9 : SUCCESS in 20m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0534574bba44a459ecca5b88b905afb : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b9ca208fc8b84833994873775a197856 : SUCCESS in 26m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3ab94782aafcedc2de64a338be9c0030d2a13772","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 11:45:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/257aa9c5e2c34897abea736fa52c9b7d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f41613c1b464bbca36c4558b9086f0a : SUCCESS in 21m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ff98dac1e0a4405ca9a2daca1577132a : SUCCESS in 18m 15s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/57b2ffa048064f75b0e903ff44aec915 : SUCCESS in 25m 07s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e2dec5a03f864d319c7d2081948aedba : SUCCESS in 1h 10m 53s","accounts_in_message":[],"_revision_number":1},{"id":"24379e5a6dd730643ce902ffc7ff88847178013f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-30 11:54:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3e849a54c8976dc796aaea108d3ce5b0f1edc843","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 11:54:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9578446551cd33c90a095ca76238adb5e5f7b669","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 12:54:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca0b0785459745db9469cf05e2a77776\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/412b7fe5c7cd4c2dbdd0516aa25a632e : SUCCESS in 7m 44s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/19bd27a8b6e145be9dc7dd5d21b26d69 : SUCCESS in 8m 46s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/62d8970e2e6c44368cdd1047aed4994d : SUCCESS in 5m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4709e4dbcba04552ab45cac1135da1ca : SUCCESS in 27m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c749ecc614834ad0a6926c765ed174be : SUCCESS in 15m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8f962d32439b4311a2711e17887b5994 : SUCCESS in 30m 11s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/cea51e7914d843f38e5a702af37f3d9e : SUCCESS in 14m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b9e9e067483e42d9a83df7b724cc552a : SUCCESS in 15m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/75df7222d4fc4bb58b179a1afd4e6ed1 : SUCCESS in 15m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b1ee6647a7894a8ca0aff1e48c188667 : SUCCESS in 15m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7d7526dd97744583b9716b1afa391d7d : SUCCESS in 24m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3b312e4e9243426ea7f673074915e2f8 : SUCCESS in 15m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/804b24b50d5e434daa0dd8ab161540ec : SUCCESS in 14m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/02884aa64fc0438ba2201cf19b2ba6ae : SUCCESS in 8m 14s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f3b4912dd5b9453f81dc9ed9da421d5a : SUCCESS in 9m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/85f4832f18474ba0a95ba429979e8bb4 : SUCCESS in 52m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4610eb279e1c40b3a7b764fd8e7ad3b5 : SUCCESS in 36m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eed77b139a924e6baf506f12dcc71d48 : SUCCESS in 17m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17c00d47c646444cbbef9d4bdb85228d : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"1f0f8e285f09e07df1c9636f2df099b1c8add249","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 12:54:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"718dc2c3c995bd31ffecbef13b120b028af60532","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 12:55:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07f3827f823941b8a7378a79045a789e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9027cccc54134508be42fb9af0d66e50 : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/1e48008f4bb144969063765e28e1b420 : SUCCESS in 34s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/07cc43796e0340ada75af4ead59a288d : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a82d9c58b6035dc3a69d4274919084d6ec6950c7","revisions":{"a82d9c58b6035dc3a69d4274919084d6ec6950c7":{"kind":"REWORK","_number":1,"created":"2023-06-30 10:32:33.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/65/887365/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/65/887365/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/887365/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/887365/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/887365/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/65/887365/1"}}},"commit":{"parents":[{"commit":"11296f1642e1b1948cec5c98dccb2504c6cac4be","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11296f1642e1b1948cec5c98dccb2504c6cac4be"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-30 10:27:26.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-30 10:31:09.000000000","tz":60},"subject":"Rename TestSource which is not a TestCase","message":"Rename TestSource which is not a TestCase\n\nHaving a class named TestSource that is not a TestCase causes pytest\nwarnings:\n\n  \"PytestCollectionWarning: cannot collect test class \u0027TestSource\u0027\"\n\nChange-Id: Ie63fdd8f3febe441963bc5755175878433a8c4c7\nRelated-Change: Ie8e0560f90207eede1992c8665e4d8571a505580\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a82d9c58b6035dc3a69d4274919084d6ec6950c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a82d9c58b6035dc3a69d4274919084d6ec6950c7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
