)]}'
{"id":"openstack%2Fswift~887910","triplet_id":"openstack%2Fswift~master~I035ef2dcc0d0e09337bd2e742baeff1fb62bf018","project":"openstack/swift","branch":"master","topic":"keep_cache_tests","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2023-07-07 17:04:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I035ef2dcc0d0e09337bd2e742baeff1fb62bf018","subject":"Object-server: backfill unit test coverage for keep_cache_private","status":"MERGED","created":"2023-07-07 05:39:56.000000000","updated":"2023-07-07 17:05:41.000000000","submitted":"2023-07-07 17:04:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887910-keep_cache_tests","meta_rev_id":"0353cc9e224e708007df8579f7ba555dccabcfc4","_number":887910,"virtual_id_number":887910,"owner":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-07 17:04:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-07-07 15:56:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-07-07 15:56:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-07 06:04:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-07 07:07:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-07 15:56:57.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"1b5254b9dac9561415abf22d662c35116f603412","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-07-07 05:39:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"288c16b4fc806359f9c235ec0446a248295b0076","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 06:04:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b4eedd4eb7d4f15b26b61aee7a4f712\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/98d8b768a77b4d76b9ad3315fbacd192 : SUCCESS in 15m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/576c721505bb4257a54168de847b62f8 : SUCCESS in 15m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9bd59ba4ca2d4896baf8285f90f5516f : SUCCESS in 21m 10s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/01590acae0fa4f90b8e58234e2e6bdf9 : SUCCESS in 21m 05s","accounts_in_message":[],"_revision_number":1},{"id":"0705956207b1ab59a58d95d4e70a2a95f1c424b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 07:07:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a2fc64c8df44a70ac730f7f3dd05f86\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ba003e3a71444f88c4be2386f404693 : SUCCESS in 9m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2eeb1aee058b4da4aeeb05ca95aad7c5 : SUCCESS in 18m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/57f977afc59043528cfdde3d96d677a1 : SUCCESS in 20m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/107b784f0b37425f990b94f18cec5a08 : SUCCESS in 18m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/936a1a710d444a13b2ae678a432c637b : SUCCESS in 8m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/737092df1e8d47d6ae60dfaa6d171360 : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/93a5108f865148218992df64cbabf8a4 : SUCCESS in 19m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/892ca52cb41d424895599da35c54b350 : SUCCESS in 20m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0518f60425e54605b93a56ea91fa1212 : SUCCESS in 20m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1ad2950d662d48b9a303fcb60c8229ef : SUCCESS in 19m 10s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e9dcaeeb68514e3b8140048de66797c5 : SUCCESS in 19m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/046afce700964736ac5f189a894025b0 : SUCCESS in 19m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4b9b395ebf6e43489afec6b4a91140bd : SUCCESS in 22m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c8bda004e2e54034bcf5dd15c70eb69a : SUCCESS in 24m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bfcd8d2f469e441aaf3800c50e454a7f : SUCCESS in 19m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b2431707908d4716ad4eda7e3fb83080 : SUCCESS in 19m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4a3a4580871540738bdbab3a9111f871 : SUCCESS in 19m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1a7473b0d1374279888c8533f96d613d : SUCCESS in 57m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/944501a4af8c4b16b79a486f47da652d : SUCCESS in 1h 23m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a3cb80a6c6f46c9948d6eb46c895cc1 : SUCCESS in 13m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cf6b28245f041e58cd524b8b6927741 : SUCCESS in 4m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/987b64d64a784466b6874273f83c33c4 : SUCCESS in 15m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e8294eac306d401a3c0adeb8e5708d4ea7de09a0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-07-07 15:56:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2666fe6e5c7f2af5f872e6ed351cecee7b8e32a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 15:57:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0c63626a2edd8da5ccc8d3c8adf799895c870b97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 17:04:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2aa8ec287cce4bda8fa19e1f5829bd72\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d394124f4de4dacb7c9c6818c4a708f : SUCCESS in 9m 03s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ce05a40e82664a56a90289aa4897174f : SUCCESS in 9m 10s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/414d6f74879c498cabb431a52993dd55 : SUCCESS in 10m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d9ba5070ddef4e029e15fa34f4e374cf : SUCCESS in 17m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/068bf31374414be783915dc57b50b54c : SUCCESS in 17m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b35215010d96480a9bf1d503b0c93a5b : SUCCESS in 23m 39s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/79b40b8677d048fbac563e00e096d7c6 : SUCCESS in 13m 26s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/217d7582d3784e0ab36b6507ad511742 : SUCCESS in 19m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ac4af51fb6e444f59131389d20b72269 : SUCCESS in 17m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5babde90c0c040a798611698ba5f3e72 : SUCCESS in 19m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cf315de5edcf4b1f910d86b673122498 : SUCCESS in 19m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1d2cd65d360446b9a4df487e21f109fb : SUCCESS in 19m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d8d02709eb0b4a3585235569c47bd2c8 : SUCCESS in 13m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/94028a4e26ee4472839a816ee3a4c47a : SUCCESS in 16m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/72c8ed4d95f54fed83a89d8edf922d01 : SUCCESS in 7m 44s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8897d43af6fa4690b490d74fd0405d21 : SUCCESS in 11m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b1dbaf0bdda343e2abaeac9458d946ef : SUCCESS in 36m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7935f106483749e990f1360b9c13e97c : SUCCESS in 1h 06m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3e9b3c6db0a4ee0acc845733eb6ef2b : SUCCESS in 19m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fda1f7a614e74cbca1a200f53e8f4211 : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":1},{"id":"a48a808410084a9356d73f7d24b717051ecb741c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 17:04:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0353cc9e224e708007df8579f7ba555dccabcfc4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 17:05:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9caa39e206bf4b3e994b11e5f9b9c362\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/912e679c74a040578206c93f352a9531 : SUCCESS in 43s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/fee9cd6f13164c439eb3a32c61f58b73 : SUCCESS in 29s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/440e966c43c94d288219013d087b3497 : SUCCESS in 29s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b7cf29476a2f77ceab1a082100b0c7e7cb8dc93","revisions":{"1b7cf29476a2f77ceab1a082100b0c7e7cb8dc93":{"kind":"REWORK","_number":1,"created":"2023-07-07 05:39:56.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/10/887910/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/10/887910/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/887910/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/887910/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/887910/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/10/887910/1"}}},"commit":{"parents":[{"commit":"671dbc8862a36954065b7b3f90627c8664b9c8f1","subject":"Merge \"docs: Format metrics in fixed-width font, not italics\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/671dbc8862a36954065b7b3f90627c8664b9c8f1"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-07-07 05:36:10.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-07-07 05:36:10.000000000","tz":-420},"subject":"Object-server: backfill unit test coverage for keep_cache_private","message":"Object-server: backfill unit test coverage for keep_cache_private\n\nChange-Id: I035ef2dcc0d0e09337bd2e742baeff1fb62bf018\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b7cf29476a2f77ceab1a082100b0c7e7cb8dc93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b7cf29476a2f77ceab1a082100b0c7e7cb8dc93"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
