)]}'
{"id":"openstack%2Fswift~888229","triplet_id":"openstack%2Fswift~master~Icdc74856f7beb318bfd84b39b5776d0d531a74b9","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-07-24 06:12:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icdc74856f7beb318bfd84b39b5776d0d531a74b9","subject":"Add more unit tests for ranged GETs","status":"MERGED","created":"2023-07-12 11:10:18.000000000","updated":"2023-07-24 06:14:19.000000000","submitted":"2023-07-24 06:12:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888229","meta_rev_id":"5933a4ca461a9121981932821b53cf4be700a49c","_number":888229,"virtual_id_number":888229,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-24 06:12:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2023-07-24 05:04:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2023-07-24 05:04:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-12 11:33:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-12 12:24:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-24 05:04:45.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"489d41ea1fcc347e8bebdb62b85753dcab5ebf65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-12 11:10:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4313baff2e01772dbe7867a4d7dd175281119343","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 11:33:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08bd5f52436b4cb0bf6d1b97e6f5f1d0\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/91efd7f271264e8892428b90106dc6cc : SUCCESS in 15m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ff0f2698f4414b7681ac462ceb420645 : SUCCESS in 18m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/696a659b39494d86baeddff2f7dc8cf0 : SUCCESS in 22m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bbeceff7f9d241919f3a37e74ed41d67 : SUCCESS in 21m 55s","accounts_in_message":[],"_revision_number":1},{"id":"542545e239c614ad86cab7c2b76b3e2781e11be9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 12:24:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7288d05753b4bf1b40e0023dfd93c66\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38453940c88b4062977a7bc4e4c4d815 : SUCCESS in 8m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f8a43e9c1d78464f9fe30258d1972b4a : SUCCESS in 18m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fcf4ae60648b41c9ba95075aae508ce9 : SUCCESS in 18m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/32e9c81c56fe44f6a33aadd949b11bb0 : SUCCESS in 24m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/18e24350b1524f11afacff0ad5fa7182 : SUCCESS in 9m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4a735b07f27e4ff2b903bfd6ef337a98 : SUCCESS in 5m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bfa75574854c474dbb6d6c94bf554215 : SUCCESS in 19m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0b92427b38aa4405a8093ee9808807a6 : SUCCESS in 20m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/07841c517c6049559ca17c91149d477d : SUCCESS in 15m 00s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3b6cebb1aa0040dbb967f007fa374679 : SUCCESS in 17m 19s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/db0d828f84114fcd9484c9dc9700d221 : SUCCESS in 20m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b2ed3a7ff20b432491f7b3fbcd9e8773 : SUCCESS in 15m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/adf2ec78457f46d5b89f1b064ad8d508 : SUCCESS in 19m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/df463b5962ca478498639f3cefd765f7 : SUCCESS in 17m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/29cc7dc96a6c4f2ba9e2a260705e4ffc : SUCCESS in 15m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f10485adbbde4944bceddaa5a668ba64 : SUCCESS in 16m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1cabce18c41844c881d026ff647032e1 : SUCCESS in 13m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/07cdf49bdff64df88fc2695f7c2f71be : SUCCESS in 55m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3d5bbb6a7290442eb1e1fe7a298dea09 : SUCCESS in 1h 07m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7aa17cad59d749b5a43393857ff370a6 : SUCCESS in 17m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d9b8e802ab54ca882a19102dbd149c8 : SUCCESS in 6m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7e2da99d3ced4dd886e577b99eac7013 : SUCCESS in 21m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0c2a43e4fb14b998a9e0918ccd47af49fb5e0b8e","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-07-24 05:04:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d370bacee73add9d1c6d9684f588309e46d76284","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 05:05:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"46c1219bb980c710f40691ef19151d11ae5177aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 06:12:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca55b613f3d342cbbd94ac031cc8150a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38cdec43ad5e49a282ba5c43ad04251a : SUCCESS in 10m 38s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e9b57707821f40e79f79b893ea330db4 : SUCCESS in 9m 14s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/40bbe2be14a0422a83eb9f359c0fb513 : SUCCESS in 6m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a11150a93b704fe5891e25089baedb00 : SUCCESS in 14m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1f0bf39be48e4d5fa15938a22cfe2db6 : SUCCESS in 17m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8b2afa79a5024ef99f08fec662aafb47 : SUCCESS in 11m 20s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/36beaaba89524f1da8f6fbb55b39cd25 : SUCCESS in 16m 40s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/61d47b5f03df4167ad441fdf00da2fe9 : SUCCESS in 17m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c17bf1fba2f345a788f82bd14cdabdb2 : SUCCESS in 14m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ea75bca6a15d4d8fa63bb3f209fe6c2a : SUCCESS in 18m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cf6a12a2036e44bd8281c4684ff281a4 : SUCCESS in 14m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c3710b2dea7a4a9088bef08f3f1e1f80 : SUCCESS in 16m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c0528793bdd24257be6fa0bc7ed391e8 : SUCCESS in 14m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/006b4f6907454fadb936f316d0d5359d : SUCCESS in 15m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/33dec0ed047f46008be99f4e3e66456b : SUCCESS in 6m 59s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/efcf1b94d2e7418f9b4b48706c80fe5d : SUCCESS in 8m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a8b811725a974e74bd6b273e0cab5625 : SUCCESS in 50m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ba1cee1817d74cec9ca61f9fd49f2627 : SUCCESS in 38m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9595e3895f9647238cc471ea6ef69bd5 : SUCCESS in 16m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da2e312751c44c66b2124cfb4be20719 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":1},{"id":"7eba3ebc33584422f3c527115faaf0f1580e17d6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 06:12:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5933a4ca461a9121981932821b53cf4be700a49c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 06:14:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9946c1b12c347f691f6e47f36dfa518\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce42372bbdf34aa8804472ffa3ac73e8 : SUCCESS in 52s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e3c6b479e4aa4f0d84e7a8202f448e4f : SUCCESS in 27s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/033ed6b581934c41a74b533052174a89 : SUCCESS in 38s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"771d605d0314374b575ca018929db7c534729636","revisions":{"771d605d0314374b575ca018929db7c534729636":{"kind":"REWORK","_number":1,"created":"2023-07-12 11:10:18.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/29/888229/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/888229/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/888229/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/888229/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/888229/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/888229/1"}}},"commit":{"parents":[{"commit":"20757612c309631fcdd729a03e14ca71d1c55790","subject":"proxy: Pop ignore-range backend header on resume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20757612c309631fcdd729a03e14ca71d1c55790"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-07-12 11:07:36.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-07-12 11:08:30.000000000","tz":60},"subject":"Add more unit tests for ranged GETs","message":"Add more unit tests for ranged GETs\n\nChange-Id: Icdc74856f7beb318bfd84b39b5776d0d531a74b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/771d605d0314374b575ca018929db7c534729636"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/771d605d0314374b575ca018929db7c534729636"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
