)]}'
{"id":"openstack%2Fswift~888575","triplet_id":"openstack%2Fswift~master~I43f79de3f0603b9fab8bf6f7b4c3b1892a8919b3","project":"openstack/swift","branch":"master","topic":"container_get_ns","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2023-08-04 10:40:44.000000000","reason":"Change was submitted"},"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2023-08-04 10:40:44.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-08-04 10:40:44.000000000","reason":"Change was submitted"},"32761":{"account":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"last_update":"2023-08-04 10:40:44.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-08-04 10:40:44.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-08-04 10:40:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I43f79de3f0603b9fab8bf6f7b4c3b1892a8919b3","subject":"container-server: use LIMIT in get_own_shard_range() query","status":"MERGED","created":"2023-07-14 13:39:01.000000000","updated":"2023-08-04 10:41:53.000000000","submitted":"2023-08-04 10:40:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":38,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888575-container_get_ns","meta_rev_id":"b48e7dda1617f434896688bf1795c15e641b1efa","_number":888575,"virtual_id_number":888575,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-04 10:40:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":2,"date":"2023-08-04 08:11:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-04 09:31:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":2,"date":"2023-08-03 23:47:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":1,"date":"2023-08-04 08:11:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-14 14:02:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-14 14:54:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-14 16:36:05.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2023-07-17 16:41:22.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2023-07-18 22:10:18.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"CC"},{"updated":"2023-07-19 21:54:20.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2023-07-24 04:41:09.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2023-07-28 01:44:09.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-08-02 22:26:25.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2023-08-04 09:31:39.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"488e1baf10d0c2d31788d6db594e486515315e20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-14 13:39:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d99e6f4caf68d253a56dce29eda375a65c118ac3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-14 14:02:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54c290ee1a7c4979b1662b691f2ade01\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1ebecbd87f3b4a1ea58d4fac24404c6d : SUCCESS in 16m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/43cc4a9b09f3489f85fb67161430f66e : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/38900cef84a8429aac9c27db0381a6e5 : SUCCESS in 22m 19s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0f6c5310d89548caab9eebce34324b07 : SUCCESS in 18m 59s","accounts_in_message":[],"_revision_number":1},{"id":"aba8f33e43a511e637b6bb7866dd0bfa57c2d738","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-14 14:54:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a71839756b24a748dab07ae2463199b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/188f07bf34df4fa9bef25aae4e614b4f : SUCCESS in 7m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/6ad1208ceb8044f4a3beda8515d9eda9 : SUCCESS in 38m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bf50660d5400469c8f0b70b5fafae389 : SUCCESS in 54m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/796a76734b084966a6c67fd56bd38f32 : SUCCESS in 58m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/03ee0056220948cc9ef374208a4cbd1f : SUCCESS in 21m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e273cee081a74df79a141b66bf51edd5 : SUCCESS in 20m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b6eb00840fcc4969b46fdca4d802ec27 : SUCCESS in 21m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/87b4c885ad7d443f95827bdc92217d1c : SUCCESS in 9m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1de330f7e585442582dc6c5111bfb001 : SUCCESS in 5m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a4564bc40bc448b899e6a6888fd9dc4b : FAILURE in 14m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/361442087a934e6aaec0a3d79abb1853 : SUCCESS in 15m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8f6f74130b0445f68bb9f13fc334f256 : SUCCESS in 13m 02s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4120189ec33b471a83387657e7b86856 : SUCCESS in 15m 34s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/6cf56d26855944c999b0ed3d5a52d0b5 : FAILURE in 56m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/af996da016a24a1db808c1555102024e : SUCCESS in 15m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/41d90bd4bc11416f8761cfeb283fb368 : SUCCESS in 13m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f2bad054aadc4dbea3b8a2e5c4c4958b : SUCCESS in 1h 07m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/db3d2109edc14fd69b4708d81a0332bd : SUCCESS in 14m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/19ede1204d74492b8fc54c7751b4fd91 : SUCCESS in 13m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7ac26f560a2b40069c53bb5048aee18c : SUCCESS in 14m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/87debf20b5c0400a883f0b9f5e7bce88 : SUCCESS in 7m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b3dc280295ab44908ce5bab2aec92b0c : SUCCESS in 1h 07m 19s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c2e83124982f40cf8a4c7d71db054440 : SUCCESS in 7m 27s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f469f0eda051416f816262d7de831265 : SUCCESS in 1h 01m 58s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/50176b695eab461f8c58a08a13716f37 : SUCCESS in 59m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2c53d7ea77b348e58cbe2bc20fc38184 : SUCCESS in 1h 00m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6330631975e14b709e697cdb990364df : SUCCESS in 59m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6cb8fea85458420d95f90575d39b8da1 : FAILURE in 18m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29a725c78d134472af18c37d2ce5f7fa : SUCCESS in 5m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/547f3d542c8241e99e6b9f1ad8877c04 : SUCCESS in 20m 41s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7cdbae43b0914279bce2d1b7b382190a : SUCCESS in 1h 07m 50s","accounts_in_message":[],"_revision_number":1},{"id":"17c93c0394cd1eaa5ce2df7c1b4bb0e166e24729","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-07-14 16:36:05.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1c86b4800fa56b0bc06afe4f839f3ea6bd6ba35c","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-07-17 16:41:22.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1d8a556acb27a083e3480018b2cec97b8ca1df61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-17 18:03:36.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"dd3f258859f1f708696d014bd6d80f790971d556","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 18:31:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d75de4171ecc464e8c48195e1dad62a7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/677583f81ea4492cb0bb7752028db4e6 : SUCCESS in 16m 04s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a90e4104854a43bda8f1969ee39ba1a1 : SUCCESS in 17m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/58fc56fee1fb4c72b0a25acd57a68e80 : SUCCESS in 24m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9042574b3a794a13ba5cdc5f0db24b97 : SUCCESS in 20m 48s","accounts_in_message":[],"_revision_number":2},{"id":"9c4e0b93c0fab0099cd57a599f6078dc34c9bbef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 19:10:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/073dea4f5dea4e9da527259d118cb92c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca92b13dd9614c0bac26c04df1cc5164 : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/1dfe59a8d5274b87b274843decddb0d2 : SUCCESS in 39m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6db17f7ce0fa4936bcfbaebfcd1cca2d : SUCCESS in 44m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28043482c6324afeba2ea62f148b4e90 : SUCCESS in 41m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dda6346e5f194ec8920e00c002f39b31 : SUCCESS in 18m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5a3c25070bc54bdcaa52a4b832febe0e : SUCCESS in 14m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/08afd266d9054285a9478b30b9d2129e : SUCCESS in 18m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3cc07e363f6d4d978d931c9806f89dbf : SUCCESS in 8m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b099906ef5b04d87a989d62423df694f : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9c82003e487848d291dadbbe69721080 : SUCCESS in 17m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/255ea64bafe44b54a8360661f4825692 : SUCCESS in 16m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b6da4c9bc6124fdeb56d960f33e10b9b : SUCCESS in 15m 16s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ebe3dbc391344f039557c1ae50b02fca : SUCCESS in 33m 53s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/dc106bdcf5804f4c85f0b15768af4004 : SUCCESS in 17m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cd17683144164f1ca0914734e1531065 : SUCCESS in 14m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e9b82e458cd349cfa9e5a95cac46e610 : SUCCESS in 20m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a77f945876334afe9515f798dabe707f : SUCCESS in 31m 31s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1e94454ae101458086de6870600e9937 : SUCCESS in 14m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6da83606611f4708b230dada6d8112f6 : SUCCESS in 11m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b0ce046157ff44a5aace4a4eff7632de : SUCCESS in 14m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/395fd01a04ac47968d8162a7b6bdb5bd : SUCCESS in 6m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7ceaca5d63044986a85875529f34f0d8 : SUCCESS in 16m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9f1c2fa6805d40079cbe496075a31601 : SUCCESS in 7m 36s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f9a9ba21a4714380affe6626fd922e2d : FAILURE in 57m 09s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0e54d2dec30c426183c7ce6fde86d76e : SUCCESS in 1h 00m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bccfb54bac25463894545f2412873797 : SUCCESS in 50m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3c1ec52b6623430c828e832e58611994 : SUCCESS in 55m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4910570554b74ed38750a792b79fbedc : SUCCESS in 17m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eb0e0a560424955893b7cf321fdb661 : SUCCESS in 5m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d3dde5de39274852b79a8ded2f758c67 : SUCCESS in 17m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cfbc05ea61fa4a339ca4a64005df00a4 : SUCCESS in 48m 25s","accounts_in_message":[],"_revision_number":2},{"id":"99d87c35388c8086ff8193c79e3f6672fde6782d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-07-18 13:51:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"32a8d61df9a2cfd02b3c21f1bfc4328d40792306","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-18 16:20:07.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a5faf60a1d06eb0a401538672b8cdcb0f00b92fd","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-07-18 22:10:18.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"86398ed4661c3b8df1d0f665e87b68daac9307e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-19 14:00:51.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"30ad2ee7c910225ebe6bebb2529c4567401c0352","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 14:26:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/099f48db39ec403c8b26152072b39e53\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0e1850417d06488487e701f258420ef1 : SUCCESS in 18m 04s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6b200ed672f24471910920ed3216ef40 : SUCCESS in 19m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c14b0307ab25480c91aac953efd63604 : SUCCESS in 22m 16s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/39f61cf6bcdb4b69b590edb7837fd937 : SUCCESS in 22m 54s","accounts_in_message":[],"_revision_number":3},{"id":"492c9539144fd95fcf50a743e0426d82cda8baf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 15:23:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f3007f48efd4185a96f30bdbc3245cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46d7687fe80d4ad1adcf57f0dcf137b4 : SUCCESS in 11m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/1d5c660c90714699b63f9f59bbb8c775 : SUCCESS in 1h 08m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9c38fc32eb3c4a8cb58a2874d27c388f : SUCCESS in 1h 01m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4755e15c7a1e4df0907e56de5c89c552 : SUCCESS in 1h 10m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b9980ef8689d476089fdcd2e54789d68 : SUCCESS in 17m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a24f69a21fd746f0b6826459aa363c50 : SUCCESS in 17m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fa29898a6e234a12a6c92de1f7d9a682 : SUCCESS in 18m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9e9f0a474a22430db1faaad5464dc3a7 : SUCCESS in 9m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c384568339a14abdbeb53aa8da2596b5 : SUCCESS in 6m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/76d94e8b626a4c04b12a4a4180434756 : SUCCESS in 17m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/237315139b5448458f786f34ca78d33c : SUCCESS in 13m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4282373727514b2d8079636326b89256 : SUCCESS in 16m 40s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/bec20ceb07be4c5eafc04baa3da6f087 : SUCCESS in 18m 19s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1098e478728446fd93ec517226ed8a62 : SUCCESS in 17m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/738a39cf3aaf426cb682f04bc08f5a87 : SUCCESS in 19m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/62a2b7dc1d9b44a89e29437d6f3700ef : SUCCESS in 18m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/04c9f90c3e9d4a61aa11d10ed5b996e3 : SUCCESS in 16m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/aca1743240aa4fc2a0fd2cb64c262c70 : SUCCESS in 17m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/19368360254b441fa4914e0f8fc6b2c6 : SUCCESS in 16m 07s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e5878cbd836e4fd0b8607983d2625fd1 : SUCCESS in 36m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d4a387c8b5e64c4492a9e9c8eac8033e : SUCCESS in 8m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/096d6e3a54404d1cac48b0e57da8a1a3 : SUCCESS in 18m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e71911411e3e4569bb014fb4956b94ea : SUCCESS in 8m 12s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2e2612cdcf4d44bda10859e08b9a7f4d : SUCCESS in 59m 49s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/eb774fe7caf04b0e850da642ea90a352 : SUCCESS in 1h 02m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bfdecfa90f7a4eaba8b6139cbd017046 : SUCCESS in 50m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/37db89f169914d33ae0f217562580012 : SUCCESS in 1h 04m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9610da56e0b4382a5d6542610c85846 : SUCCESS in 20m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faddcf7a75fa47d3a7911e8de25e5117 : SUCCESS in 5m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/abbc65137a244f709d3e83215e2c59cd : SUCCESS in 20m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bcbed8c4db2847fa9fef7aff6135dbd6 : SUCCESS in 55m 19s","accounts_in_message":[],"_revision_number":3},{"id":"e65747142627e653c416c047eef9aebaa02b1501","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-19 17:52:49.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"be3fe873caf70f9d428a46f30cf2044eaae09e50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 18:14:41.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/082a84a3eae14778a33ae762b0c54fb6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0a84adf461714d8aa5416d781e4230a3 : FAILURE in 20m 43s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/51a88fba6ba24dff83e852dc489aefe4 : FAILURE in 20m 46s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/905c96e2cbd2434b89dcc9384f40da19 : SUCCESS in 20m 49s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3a690229e9aa4eef8810aee6b63e84da : SUCCESS in 19m 37s","accounts_in_message":[],"_revision_number":4},{"id":"ff59197fd12ca45d3f422839f28c6fa42c669870","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 19:05:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c577574957f9495c9b61e7d574d3cbec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8cf318432fe40cb94e608a35b7abba8 : SUCCESS in 9m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/e0eab685f2d14947b3a1a48f9a2bf3b1 : SUCCESS in 55m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/67922f84c51a48a5af1955027a7168e3 : SUCCESS in 43m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/adf2f32ea4bf4f3da15372b67e5ee130 : SUCCESS in 47m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2438cea9e7004e82bc5b3d034aff0fbe : SUCCESS in 13m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/537a19cc3f7145289a39de14df6f3657 : SUCCESS in 15m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/78615dbe2a6c4de2a222f33dc121496e : SUCCESS in 23m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ceaba4dfbf7d4b319df6ccbbb98f8471 : SUCCESS in 7m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b64d52cc7d1744ed93ccdec7675ea669 : SUCCESS in 5m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/03a972b23f1c41f787f64e6a44b2251a : SUCCESS in 15m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d3ab5aedde6a41ca956a80cc521f3823 : FAILURE in 20m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3ad3c709de284ee191128a54ac8326c7 : FAILURE in 23m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0f58af953d3542ca84d19f9ef13f8c98 : FAILURE in 38m 59s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1ae4356aef6d4ebda9d9d2e8594b1fff : FAILURE in 24m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/24031bd64aa54bffa9e9f8a95dd784b5 : SUCCESS in 14m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bb2f68405cb849688064f96ff224ff60 : SUCCESS in 24m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3fef3fa090fe4245a79990480b892701 : SUCCESS in 16m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f50245965ece46cf950e81d503f4887a : SUCCESS in 15m 50s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d2504d4afdeb40448e0312337bb20f31 : SUCCESS in 22m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a4c50f5e29b14844aa726ccc6d4fcad8 : SUCCESS in 25m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/900fac99887440e78f4f182893515b6d : SUCCESS in 6m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/679313fafc6f463db5da2407fd7adf72 : SUCCESS in 18m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/75f3c2502ee84a15a509e1d94ccbeff8 : SUCCESS in 8m 00s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/221968150c8947c8828fa0d8bb3a54ae : SUCCESS in 55m 25s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b405cd3a54da43c2bd40361a42b16d18 : FAILURE in 1h 07m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b5f8f2b3e61b4d8b976b30faa64379fb : SUCCESS in 48m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6a93344d57ec4371a346ec0839fbdb27 : SUCCESS in 42m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87a625d2d8e14af0b26be1998fcb15bd : SUCCESS in 25m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/309553b43fad4f14afadec26aceabfde : SUCCESS in 4m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2f07bd1e36754f39b2f1a2eb151d7874 : SUCCESS in 20m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/571851f2a5bf4b87889961cdc6714e05 : SUCCESS in 55m 31s","accounts_in_message":[],"_revision_number":4},{"id":"7e56f8c1ca34c6c8e9524018670cad89af0bdd54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-07-19 21:46:27.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"ddbf7c8b0f9c59685937fc4e45c5606167307cbe","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-07-19 21:54:20.000000000","message":"Patch Set 5: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"a645674a416c50c275a891c1cad3fbe4e9d1e597","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 22:10:55.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1acc7266f93a49b79b5ac8880d2d2032\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/27013e0590e04854a52935be24221b20 : SUCCESS in 21m 35s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/61df4a1ea8d34112963d75ed78be84a2 : SUCCESS in 15m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/857dada3cee54a2898c064f6619bdf0d : SUCCESS in 22m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b9a638f8f334d1d8979a5a5a68c3b06 : SUCCESS in 23m 27s","accounts_in_message":[],"_revision_number":5},{"id":"7b22a0778761e62f5bf9f30e83b55cfbf00a9f4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 22:45:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4328ee25ab9e488698bd6e64fc6015b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43b1bcc72c16476f90f72417c23bbc9f : SUCCESS in 8m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/5e653f0fb3f6423ca3af7368bbfd564e : SUCCESS in 41m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/45d985ed5ab145758be53185a614d52e : SUCCESS in 42m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c799e92cc6d4ffea11834efd07265fd : SUCCESS in 46m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e958ad8b1c114c1aa34a2ae0e02bfb9a : SUCCESS in 19m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9abbcc6c0c9c4c9186d67ffd3adb9864 : SUCCESS in 18m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9bc16ab5fec04764bd38ca1ec050bf0e : SUCCESS in 19m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/05f808ca56a742a690ac74d0ddbc1797 : SUCCESS in 8m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/658e2ed68c1f4b369d4fd887d7d52dd5 : SUCCESS in 6m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7c794f5526c8410fba6b63ed39c4572c : SUCCESS in 16m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/528d2b0046a54ab8acfea5dedb396ce2 : SUCCESS in 16m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4124807a234647869ff9ad4408021196 : SUCCESS in 13m 06s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5a6b450519004248b6c452b528f32014 : SUCCESS in 15m 44s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3e0bcb0278f94fcf978c150d3943e3b2 : SUCCESS in 20m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a7f1765b21764ac0891b112ee7eb5ebd : SUCCESS in 15m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2d68489ef22e4412a6ac428112aded95 : SUCCESS in 15m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5be27f2e3986403b8c293a17178d9eff : SUCCESS in 16m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7737b1e499d642fc8a650b1b27fc1738 : SUCCESS in 15m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c2e25a0039f847fd953220eba1a9c05a : SUCCESS in 13m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/105dbb77d0ee4804bd0098451aa74fa0 : SUCCESS in 14m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6943602cc6f04cb08233dbf7c853c1e4 : SUCCESS in 6m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5660dcf94ef04b1fa9b1696bed990afe : SUCCESS in 17m 46s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4bbc39db44374fdd9398f8c448ac0f5f : SUCCESS in 7m 04s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4771c67999d94d8088d8f68006097159 : SUCCESS in 57m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/19f2fabe579b4c6ca94e40e499331092 : SUCCESS in 46m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4ec78519060c4dcfa681dfff3a20471e : SUCCESS in 45m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2cf3e5e92e224e13b148de9de54622ab : SUCCESS in 32m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79e31fc835904001b8c56bc07b6d1104 : SUCCESS in 19m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b0161fc5a074f308f8956449da576c1 : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/628ca33991354d80ae84cce7364e83ec : SUCCESS in 23m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1128b97253cc4664a0e6ed5bed7214c6 : SUCCESS in 43m 33s","accounts_in_message":[],"_revision_number":5},{"id":"52c65bb3a9d255ddb6b0a31c5567d9cc063454ad","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-20 11:15:30.000000000","message":"Patch Set 5:\n\n(8 comments)","accounts_in_message":[],"_revision_number":5},{"id":"3e71fb9be37b313360b69f081968b040a91606ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-20 13:22:58.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"8b5c626bcc66504fb7d8e025601f4233e2cf1fff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 14:55:51.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a750c8c83c648e2b55e857ae4a22418\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b26c4fd099e24a0ba1b514020d89a0b5 : SUCCESS in 1h 14m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ad536ba9cb5640cbb4e3f40da7776b4a : SUCCESS in 19m 16s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ee36465e2f5744f0b1b2290f701c9049 : TIMED_OUT in 1h 30m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca8b7278b48c41fb8eb67c5c90153e22 : SUCCESS in 22m 14s","accounts_in_message":[],"_revision_number":6},{"id":"9e70771179eaade1a690fe45561bd367b5852843","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 15:23:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/daf3607084204b5fa1124706685e1bfa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddb749f333b94aed9e3580f06e74819d : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/4460fdfb8d5d439797ec16eea75bd559 : SUCCESS in 45m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ba474d4cea754ff9a0867a8e17ea7768 : SUCCESS in 45m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c437b2acf46424e9bf6c76d791bb899 : SUCCESS in 1h 01m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c3ede053b7764ddea530bcddee224777 : SUCCESS in 24m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7e39e56c88094c96bd9e50b0eb2425a3 : SUCCESS in 19m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f718ef7d57cf49a99b5af93f383c7d9b : SUCCESS in 17m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/81ac90295e8249aea129d17fbdd6105b : SUCCESS in 9m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/dbc3fd6ea2fa4d9297380608fa5fc47e : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4a90e89e89ba4e9992efe25d65bef40d : SUCCESS in 18m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ace0bb9f648e425db130ba5a68c82967 : SUCCESS in 15m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/be237b657e2c4b148fc2886d747a3fbf : SUCCESS in 16m 44s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/fcb407fae5bc4964bcc752caeba0f848 : SUCCESS in 16m 02s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9bb7e6d15bb74444995732bca5a3a421 : SUCCESS in 15m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b8ed9d147c2b467ca58da9a13fed1332 : SUCCESS in 17m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1900edca23b143d6ac3bb36ae2ae178a : SUCCESS in 16m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f5013b22a57f4e76a3541c94216cd55e : SUCCESS in 39m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1e16844d99034ccd8f70b1c8442253d9 : SUCCESS in 17m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/84ad2edd555f4124bfbd7a9317b68710 : SUCCESS in 16m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5e9897b1d3bd444a98a3b461b24d0bbe : SUCCESS in 16m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/940784e78b434909b84ee0b1e7692eb4 : SUCCESS in 8m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c213aebde70e46e0b137163f872109e1 : SUCCESS in 19m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b9c6e9a836f5476f98244dea2b44ff53 : SUCCESS in 9m 18s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/deb14fa35853402ba8c7d0d085aa94f2 : SUCCESS in 1h 20m 42s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f9c3acdd6c31454a99928f5c7a89677f : SUCCESS in 1h 09m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e004c7e87bf9496496e852fae447ec68 : SUCCESS in 43m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/be0bf3713b954e97b374d9dfecf1bc2f : SUCCESS in 49m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c805fa651d3b4937b18e163bcf85441b : SUCCESS in 14m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4110ca3a01ff47c2a346c2803c41fa0f : SUCCESS in 5m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4059939e18f342cea4b46b3db202673c : SUCCESS in 19m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ecb000b8883f446caef269a0e0e96d21 : SUCCESS in 50m 15s","accounts_in_message":[],"_revision_number":6},{"id":"57bec01eb6d22cf8b9252aef32b8f85ce970af20","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-07-20 18:26:36.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5de660a87c0422a1c05691cdb008f3b6e8b4c5cc","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-07-24 04:41:09.000000000","message":"Patch Set 6: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"9dac66c6e2f3361234d49610fd211c46870e24be","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-24 10:17:26.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"6802c43f90330aa2096183b70d89f8e993a49bd6","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-07-25 00:22:13.000000000","message":"Patch Set 6: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"b437c8c09a9047463c8e313ff7b17983d6ffa7f6","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-07-28 01:44:09.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"d5611916f01e5f58e34444db2840348dbac54bb6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-02 22:26:25.000000000","message":"Patch Set 6: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"c856f5fdd472d821f683e2b7cfaea2406f042d51","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-03 13:11:48.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"bcd70131b61b25baaba1e235b35e2062cf9adc59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-03 13:31:15.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"c0d3d8753fba546a60e72c326fd99cc0c8759562","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 14:47:08.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1dc1451a7a97479291345a925cb65d55\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a127ecf1b06c4b849360a3db3b756a78 : SUCCESS in 9m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/d6820218efbf4f67a711a04231fc44fb : SUCCESS in 1h 03m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/aed35fd00f55461fbe97d7eeaf7a567e : SUCCESS in 52m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e294aa9a89ac4928b8e56882ae75bb09 : FAILURE in 1h 02m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/71fb6e13c2cc4a5495b0ab5ff0cbaa80 : SUCCESS in 20m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1bee5fa39dc9482695219292bce6ba3b : SUCCESS in 18m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c044ed80d8624b9d8b56a7965b564944 : SUCCESS in 19m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b840cc2ac0c24858836261d4e1fc2739 : SUCCESS in 9m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/655a991e33c14d17ac94627b725ab650 : SUCCESS in 6m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/984b8653afec4e1c8cf8cde2f5618112 : SUCCESS in 22m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f09895548e384478a8ecb79c559ef903 : SUCCESS in 18m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/654ebfa3612b4a1bb209a0019a7a4164 : SUCCESS in 15m 29s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/452f1466d61f4e9baccbbe19ea24329e : SUCCESS in 21m 39s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3addbf4efbb844dcbd861f58aafa0d7c : SUCCESS in 14m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b6655bd209b14c92bead258ac818125f : SUCCESS in 22m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8d194bc35f3d4ccd8a6e6e9d3b5375a6 : SUCCESS in 17m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0c452b8ff83d446e8c08b961bb8ac913 : SUCCESS in 23m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e99967cda1f84addba65ee9655b98e9a : SUCCESS in 14m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3c75c3cab3b941a3b7514e60151009bb : SUCCESS in 15m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d1e357eb244b427695193414eff6d59c : SUCCESS in 15m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ba20b53b8710437195be782aec1d529a : SUCCESS in 7m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/786126e52a024330ac75f3b2c1b6e719 : SUCCESS in 18m 03s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0844984ecb5b46e685baaff3bd9fe10f : SUCCESS in 6m 47s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/03f56c8ca5b445448818e62b677dfe2f : SUCCESS in 1h 09m 01s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0d82ce55d9f14d29bbf45db09f4d31d9 : SUCCESS in 1h 06m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0e535d4e3d1641b196921354567dd5db : SUCCESS in 1h 02m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0bfb5f5196b44b8baafd3c7ee2145ae3 : SUCCESS in 1h 00m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c09617adddc04b71a917602a894147df : SUCCESS in 18m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb043b0326ea4d1090ae01becfda84f9 : SUCCESS in 5m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/96273be2fde64a80a1cbe79eb133bac7 : SUCCESS in 26m 02s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe245a08b9f94b5d9e7488fe99e2f8a3 : SUCCESS in 1h 05m 00s","accounts_in_message":[],"_revision_number":7},{"id":"cc850a9043d62397bc8864e0b989a12cb4f399da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 15:00:55.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6a56d6699e846b8a1a11a76c4d68820\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7abe74f689b34ee8a7452d29ddb8e64f : SUCCESS in 18m 22s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2e46e14a08e44b4fad1679732dc7532a : SUCCESS in 16m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/91f2feb1c6a04801842aa7b1e4181217 : SUCCESS in 1h 27m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e16159103ba04d3b95414162f9f72405 : SUCCESS in 19m 54s","accounts_in_message":[],"_revision_number":7},{"id":"0d3d4322fb0b205b0f30f17209d2f8df5b346140","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-03 17:20:47.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"002f24901ddbf4e5164b0343c2fc011bf3ce3b57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 18:45:21.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/322e31fef0504127a08d912300cb709a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5037ad097cf64f099f2d4985c19df004 : SUCCESS in 8m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/a8088de137a3470eb93777d8768a7112 : SUCCESS in 58m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bbe6f42378a645e4aae6099f061bfb66 : SUCCESS in 51m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ce6d63c165c460eb0b33b53e93e383f : SUCCESS in 58m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/34f58916a18c4b5ea6508b340f455763 : SUCCESS in 18m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c8b0f0a89f55411fac586a7546c6727b : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ab2f93bada49486eb17bb68a9cc40bdd : SUCCESS in 18m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8f87aa09f52f42d6abeb8f940c822cca : SUCCESS in 7m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3c327db864de4ece80b9c2a7e0437b0e : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a909f63ad1154520822019900abc5265 : SUCCESS in 22m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/be3409ffad594edb9ea82310b37867bd : SUCCESS in 18m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bdbf2f3577344ee6bca7f8853fff08d1 : SUCCESS in 20m 45s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2dc978b56e8b4d049b857b5f9f10ef5b : SUCCESS in 20m 46s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/4385c7aaa8a6499384c9d5f4d54a7409 : SUCCESS in 17m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/401e3065214e49449ff267a4fd76c44f : SUCCESS in 16m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1082e355e228457d9dd2cc1f9566f39d : SUCCESS in 21m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b9d5639d331245a999364c864523891d : SUCCESS in 24m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ff9981e9c67b42e8846d0d479fda7194 : SUCCESS in 18m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/09fdb2d79a5d4fce9927f9f393bdb9c9 : SUCCESS in 14m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cc90144b32ab47b19e503e058ad4607a : SUCCESS in 14m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4db52687cbff4df596c3acd79c038782 : SUCCESS in 6m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c09766f70dac460397feed67139ad25f : SUCCESS in 16m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bb9f0b73ed1b4b21aa39fa36d4ce2ab4 : SUCCESS in 7m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a5c62c250ebc4b13bfdea045628e384c : SUCCESS in 49m 04s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ff195ce732724389b7fe9c8eb0abaed1 : SUCCESS in 56m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/66c5f661c72b4a9780292e2d54b3ff4e : SUCCESS in 1h 07m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/027c7e9137de41ffaf989174f9c31b02 : SUCCESS in 1h 17m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0467a039412f4d56ad3938bcfa77bc58 : SUCCESS in 20m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5bd9c3dfb694cfc8fc65b8f30792672 : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2b795a69e9074562bd6b88491171e58b : SUCCESS in 20m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a05943993a3428296ec509444ab9fb4 : SUCCESS in 47m 06s","accounts_in_message":[],"_revision_number":7},{"id":"de02d36022d7920f14b767e1fc06308f54dee3f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 18:46:08.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"15cd0a27bc540a030c4bcd8c94664df766215890","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 18:54:11.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e90a11ff5f2543be8240be8515796a9d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/104c3ca91be84ddebe5b98c86ee8add6 : SUCCESS in 14m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9d778ec78b054458ac7179cc477338bd : SUCCESS in 15m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c26697f4ebf84f74a303500b9cdf56bd : SUCCESS in 19m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6e8f01a82c82480eb8df2dac7441a5d9 : TIMED_OUT in 1h 31m 14s","accounts_in_message":[],"_revision_number":7},{"id":"87b4006b111ce3f06bc95f1b8540e92c6647021f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-08-03 23:47:46.000000000","message":"Uploaded patch set 8: New patch set was added with same tree, parent tree, and commit message as Patch Set 7.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Workflow+1\n","accounts_in_message":[],"_revision_number":8},{"id":"0e2245b38dfb97a5fbff0322f5ead673a34df47a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 00:10:42.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a5957bd5f224570ada990ed7f910f99\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/793b9742e56648389fe8b63abecdfc22 : SUCCESS in 17m 28s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/84fd8b11bf4846f29e4f524ba2e92401 : SUCCESS in 17m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6eecd7921cdf4484a8f0bedfbfef36ad : SUCCESS in 20m 17s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/115009dc5125425c91c562cac4d91d28 : SUCCESS in 21m 41s","accounts_in_message":[],"_revision_number":8},{"id":"58c803eddeb28c1c43e158ce179e85b245344577","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 01:21:56.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8dad3b68f3bf4e7ab12413e6e5c2215f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d270ccbf63e4cd4889e9b3138e24c5b : SUCCESS in 8m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/31e220302d9b404e986c1029aef2bc9c : SUCCESS in 1h 04m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/24ed95984e264297b76ff1f441314247 : SUCCESS in 55m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1c8a5169b8f94fcc8c726e4fc035303b : FAILURE in 1h 33m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9bb3a05405d84061b5d27651b295bc42 : SUCCESS in 16m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f0a1bd4c96940ad9e9302cdff360c9b : SUCCESS in 55m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/adba415a22774a35b993e33370729f6b : FAILURE in 1h 20m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fe0c51c80d724464b117aa3b3142bed0 : SUCCESS in 8m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9576bf98ce23495581f2ef469786642d : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0e7e7232551f4939bc2b169f05023a8b : SUCCESS in 44m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0a83999ab4c9473ca1924b1ca42d8ab8 : SUCCESS in 37m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/388186245ca4467cbd16a33433e5f3a7 : SUCCESS in 14m 58s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/48ffac9dddbf42a2bc6d18a1d14c4b2b : SUCCESS in 37m 13s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ca62a7e0e9ba4e0ba28b75fa91ce00f0 : SUCCESS in 15m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f5e6926a313b49ac97fb39494ca14b06 : SUCCESS in 33m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0349160035ed4dc0b70ceeeb1a7e4a62 : SUCCESS in 21m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/261fac9c44e245da8f5ea1ac70142337 : SUCCESS in 29m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5a9e5afe0de7439ebf4d4fcfe46433ad : SUCCESS in 13m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d2c598cb27d54f42a6b97019eb357d9f : SUCCESS in 24m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e631012edae04614a486e3e969203ccc : SUCCESS in 11m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7f0a9ef8af3a4e2ca88b45a40d9210f9 : SUCCESS in 5m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0595603fc8ba4180a986bcbdfeff55c1 : SUCCESS in 14m 06s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9c4c5a496bf04d3bb5ad72f30f379b5d : SUCCESS in 5m 52s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/112c8a6511cc47e4bc908f631726ed1a : SUCCESS in 40m 41s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e2b5f0bf5214ae49cc523c8eeadda30 : SUCCESS in 49m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cce52643f0b149829578ef77750f4713 : SUCCESS in 53m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/01e4ffe45d1f4098abc87d98a5a10779 : SUCCESS in 49m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dec43f27ddb14e30b7122f9cef06fcdf : SUCCESS in 18m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27e5f1a455434a08b2b2109e89eb07ca : SUCCESS in 4m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f016b59a32ed42d6895acbb4cf3f8509 : SUCCESS in 19m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82574a3947c447cab595c633b91492e2 : SUCCESS in 1h 03m 10s","accounts_in_message":[],"_revision_number":8},{"id":"f0e6245d0635c9cfbef22e12ed7c13205ebf21f1","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-04 08:11:42.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2b713b0685094d50cab1a03803d6bfba316db4c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 09:23:50.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/139b0fcfd7af429294d7869b7a61698d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/047f711a6295472f860f5e4501018bd7 : SUCCESS in 8m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/6b3f7c589ccf46cdbb40e1835cf5fab9 : SUCCESS in 1h 02m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5ec00980eaf342a49fef60ef541ce543 : SUCCESS in 36m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e146804b26614430acd51ad981c4058b : SUCCESS in 47m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/73cf672606a2466ba6d75a5e4efeb601 : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a8cafabe6fde473fa805554109bcc7f7 : SUCCESS in 18m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b4f1996ed6a644a6b11706bb5f1ecebc : SUCCESS in 19m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8413839686644501b8e72fd551e5e2bd : SUCCESS in 8m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c8ee9a48e9914a86b4204c95233c6606 : SUCCESS in 5m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9b61ae671324497a945bd607a8a907db : SUCCESS in 15m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fac8997a2bd04baaaace1c0fbf37397f : SUCCESS in 16m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/54afcf77a5fb4b8ea318fc238e4ae734 : SUCCESS in 16m 40s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/225d8d277c114de084bda6eef192e012 : SUCCESS in 12m 50s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f9b4f4343b32413da390476b45c62d80 : SUCCESS in 17m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cd709576cb654af5a26efab41903f9d5 : SUCCESS in 18m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/943cf58e953040e78c79fdc59f109a70 : SUCCESS in 17m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/671499081e8e402ca7d2e346c30d92a1 : SUCCESS in 20m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c3ebf760c8ed40adb8abe6350ab1b6a9 : SUCCESS in 14m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2bb6fce88b21412e97c95a581c0a660a : SUCCESS in 18m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/86c9e6be123a443da2d52eebbfcbb8cd : SUCCESS in 12m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d900483fe8bd4153a65cef830899f1f3 : SUCCESS in 8m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/78ea7f69548f4ee496572cba784250ed : SUCCESS in 17m 27s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/66f46ca7de984493b7b2cc94092a8b57 : SUCCESS in 7m 36s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4b8bfe9b014142daa18ea689b69228ed : SUCCESS in 54m 16s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/639c35d7bc9845ecb74ee8d458ebbaf7 : SUCCESS in 49m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9d16ed2bbdcb4886b536191556d2cd55 : SUCCESS in 45m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ee3ebbc500a34f4ca2cc8baaed45079d : SUCCESS in 39m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6d9ef9539d9412194953c13bb8e7499 : SUCCESS in 17m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76436faa2852458b8fdb967b89d7e64f : SUCCESS in 5m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ec58b5065cc34931afacd34cf66e9172 : SUCCESS in 20m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/58532a7d2b6146b6a662178ed1612c9d : SUCCESS in 1h 00m 36s","accounts_in_message":[],"_revision_number":8},{"id":"6bc28f587f5fe45e6f0dc39ea6bacac217d5f570","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 09:24:24.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"7b7ecff928d77a20c9e97fecfd47d76e69b82393","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-08-04 09:31:39.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"d08971b26de6029ed75a141d683db2dd07247bbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 10:40:44.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd9e01574a1f4bacb49a75a21ed7cf8d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72bfef7d1bb941b385f058150549e807 : SUCCESS in 8m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/2498ce518fa846d7a953e8ab637cd6db : SUCCESS in 55m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1165121ccbd4411ab4698e8c900200b2 : SUCCESS in 1h 01m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d92a2d6b38ec4f5dbd14623f65be476a : SUCCESS in 1h 00m 53s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/acf56616deea44acb1700de77c6bcf36 : SUCCESS in 8m 00s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/e4af4d4880a04caf941322ab87bc1cc2 : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3f26ef784de84c9dbb98b682e5056bba : SUCCESS in 17m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4ade712c544b451c9afae377c34724d9 : SUCCESS in 15m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f14ab261a8f94e35a28de55ade5cae55 : SUCCESS in 15m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/cc5d6d3b3ff247418929d0f6bb496b05 : SUCCESS in 17m 43s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ebca2d47c0774962a0e14fcbdb62ba07 : SUCCESS in 16m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/60ac3ebb9f884da595d10f96b0a8d105 : SUCCESS in 16m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/25a4bef809cc46c4a8d415e8804de092 : SUCCESS in 21m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/898f6de639fb4409868d44aa547fbc6a : SUCCESS in 22m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9fa8cf6f69ad48dc8e527062bd78c598 : SUCCESS in 16m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/63d014cd8b764983b16e5b92a4125074 : SUCCESS in 17m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c7d70e37c0bb40c1a63aff3b8b0b4ba8 : SUCCESS in 11m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4be60bf97441426f82280dbf54854918 : SUCCESS in 7m 00s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1a2d06d0370946099588b9dd0b0aef8d : SUCCESS in 6m 32s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/13ab16dc157648c0835a3c515f00ddb7 : SUCCESS in 54m 53s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a5f8ac79b2ed46fd821fdb5b5d3ca186 : SUCCESS in 1h 01m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/05f410366a5d462cbf1242e5234d3a48 : SUCCESS in 51m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5d069096e2bb433d8507bfe420b4ace3 : SUCCESS in 54m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/662ec9f2df45447d886353ed9c81d592 : SUCCESS in 22m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c96894e01ff543ae9fb5a247d72e2b5f : SUCCESS in 4m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/57c00164540c48688660cdbf45ff6031 : SUCCESS in 1h 05m 28s","accounts_in_message":[],"_revision_number":8},{"id":"38b31ef6ef164227ff2769f087170de84337c1b5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 10:40:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"b48e7dda1617f434896688bf1795c15e641b1efa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 10:41:53.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32f042e086744112b19177928be2c837\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2aa32cabb6c4bb5a5a336d78c29d294 : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/9fe5566d829144fb948dc393d5def5b9 : SUCCESS in 30s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/a5f83959f1964ace85c3bfe609c54b5e : SUCCESS in 31s (non-voting)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"6e8d82c97e28026e03f71bf38ccbcd2f1c606afa","revisions":{"197a29d78b23f679b17d371a132503a042689582":{"kind":"REWORK","_number":1,"created":"2023-07-14 13:39:01.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/75/888575/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/888575/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/888575/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/888575/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/888575/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/888575/1"}}},"commit":{"parents":[{"commit":"19bd1158555e59729fa755dc080782906b80c9a4","subject":"container-server: do [end_]marker filtering in SQL query","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/19bd1158555e59729fa755dc080782906b80c9a4"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-07-14 13:24:33.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-07-14 13:34:36.000000000","tz":60},"subject":"container-server: use LIMIT in get_own_shard_range() query","message":"container-server: use LIMIT in get_own_shard_range() query\n\nThe get_own_shard_range() is called during every container GET or HEAD\nrequest (as well as at other times). The method delegates to\nget_shard_ranges() which queries the DB for shard ranges. Although the\nquery has conditions to only select a single shard range name, and the\nmethod will return only the first matching shard range, the query had\nno LIMIT. Adding a LIMIT of 1 significantly reduces execution time\nwhen the DB has many shard range rows.\n\nOn the author\u0027s machine, using a DB with ~12000 shard ranges this\npatch reduces the get_own_shard_range() execution time by a factor of\napproximately 100.\n\nChange-Id: I43f79de3f0603b9fab8bf6f7b4c3b1892a8919b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/197a29d78b23f679b17d371a132503a042689582"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/197a29d78b23f679b17d371a132503a042689582"}]},"branch":"refs/heads/master"},"f029965b7ad6b209268cbbdf4fc99d69237f6f38":{"kind":"REWORK","_number":2,"created":"2023-07-17 18:03:36.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/75/888575/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/888575/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/888575/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/888575/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/888575/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/888575/2"}}},"commit":{"parents":[{"commit":"19bd1158555e59729fa755dc080782906b80c9a4","subject":"container-server: do [end_]marker filtering in SQL query","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/19bd1158555e59729fa755dc080782906b80c9a4"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-07-14 13:24:33.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-07-17 17:59:23.000000000","tz":60},"subject":"container-server: use LIMIT in get_own_shard_range() query","message":"container-server: use LIMIT in get_own_shard_range() query\n\nThe get_own_shard_range() method is called during every container GET\nor HEAD request (as well as at other times). The method delegates to\nget_shard_ranges() which queries the DB for shard ranges. Although the\nquery has conditions to only select a single shard range name, and the\nmethod will return only the first matching shard range, the query had\nno LIMIT. Adding a LIMIT of 1 significantly reduces execution time\nwhen the DB has many shard range rows.\n\nOn the author\u0027s machine, using a DB with ~12000 shard ranges this\npatch reduces the get_own_shard_range() execution time by a factor of\napproximately 100.\n\nChange-Id: I43f79de3f0603b9fab8bf6f7b4c3b1892a8919b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f029965b7ad6b209268cbbdf4fc99d69237f6f38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f029965b7ad6b209268cbbdf4fc99d69237f6f38"}]},"branch":"refs/heads/master"},"dba39c3b660ad5a215ba32832114d794ee0217e9":{"kind":"REWORK","_number":3,"created":"2023-07-19 14:00:51.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/75/888575/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/888575/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/888575/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/888575/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/888575/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/888575/3"}}},"commit":{"parents":[{"commit":"05ee709d9611a52c760a3e3748484b30e98add7e","subject":"container-server: do [end_]marker filtering in SQL query","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05ee709d9611a52c760a3e3748484b30e98add7e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-07-14 13:24:33.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-07-19 13:56:54.000000000","tz":60},"subject":"container-server: use LIMIT in get_own_shard_range() query","message":"container-server: use LIMIT in get_own_shard_range() query\n\nThe get_own_shard_range() method is called during every container GET\nor HEAD request (as well as at other times). The method delegates to\nget_shard_ranges() which queries the DB for shard ranges. Although the\nquery has conditions to only select a single shard range name, and the\nmethod will return only the first matching shard range, the query had\nno LIMIT. Adding a LIMIT of 1 significantly reduces execution time\nwhen the DB has many shard range rows.\n\nOn the author\u0027s machine, using a DB with ~12000 shard ranges this\npatch reduces the get_own_shard_range() execution time by a factor of\napproximately 100.\n\nChange-Id: I43f79de3f0603b9fab8bf6f7b4c3b1892a8919b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dba39c3b660ad5a215ba32832114d794ee0217e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dba39c3b660ad5a215ba32832114d794ee0217e9"}]},"branch":"refs/heads/master"},"6c43739c78b911c00c69a42bd70becd4de62da7b":{"kind":"REWORK","_number":4,"created":"2023-07-19 17:52:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/75/888575/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/888575/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/888575/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/888575/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/888575/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/888575/4"}}},"commit":{"parents":[{"commit":"05ee709d9611a52c760a3e3748484b30e98add7e","subject":"container-server: do [end_]marker filtering in SQL query","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05ee709d9611a52c760a3e3748484b30e98add7e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-07-14 13:24:33.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-07-19 17:51:58.000000000","tz":60},"subject":"container-server: use LIMIT in get_own_shard_range() query","message":"container-server: use LIMIT in get_own_shard_range() query\n\nThe get_own_shard_range() method is called during every container GET\nor HEAD request (as well as at other times). The method delegates to\nget_shard_ranges() which queries the DB for shard ranges. Although the\nquery has conditions to only select a single shard range name, and the\nmethod will return only the first matching shard range, the query had\nno LIMIT. Adding a LIMIT of 1 significantly reduces execution time\nwhen the DB has many shard range rows.\n\nOn the author\u0027s machine, using a DB with ~12000 shard ranges this\npatch reduces the get_own_shard_range() execution time by a factor of\napproximately 100.\n\nChange-Id: I43f79de3f0603b9fab8bf6f7b4c3b1892a8919b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c43739c78b911c00c69a42bd70becd4de62da7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c43739c78b911c00c69a42bd70becd4de62da7b"}]},"branch":"refs/heads/master"},"bbde27ce22f5b4cbf707a493e69eca98178d81fb":{"kind":"REWORK","_number":5,"created":"2023-07-19 21:46:27.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/75/888575/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/888575/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/888575/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/888575/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/888575/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/888575/5"}}},"commit":{"parents":[{"commit":"05ee709d9611a52c760a3e3748484b30e98add7e","subject":"container-server: do [end_]marker filtering in SQL query","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05ee709d9611a52c760a3e3748484b30e98add7e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-07-14 13:24:33.000000000","tz":60},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-07-19 21:45:22.000000000","tz":-420},"subject":"container-server: use LIMIT in get_own_shard_range() query","message":"container-server: use LIMIT in get_own_shard_range() query\n\nThe get_own_shard_range() method is called during every container GET\nor HEAD request (as well as at other times). The method delegates to\nget_shard_ranges() which queries the DB for shard ranges. Although the\nquery has conditions to only select a single shard range name, and the\nmethod will return only the first matching shard range, the query had\nno LIMIT. Adding a LIMIT of 1 significantly reduces execution time\nwhen the DB has many shard range rows.\n\nOn the author\u0027s machine, using a DB with ~12000 shard ranges this\npatch reduces the get_own_shard_range() execution time by a factor of\napproximately 100.\n\nChange-Id: I43f79de3f0603b9fab8bf6f7b4c3b1892a8919b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bbde27ce22f5b4cbf707a493e69eca98178d81fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bbde27ce22f5b4cbf707a493e69eca98178d81fb"}]},"branch":"refs/heads/master"},"0a70315e2bdbb235eabb08dbda96f35a2a848700":{"kind":"REWORK","_number":6,"created":"2023-07-20 13:22:58.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/75/888575/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/888575/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/888575/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/888575/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/888575/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/888575/6"}}},"commit":{"parents":[{"commit":"f7d6d5806edafc79f9ee6ad1abaddfbd462787c6","subject":"container-server: do [end_]marker filtering in SQL query","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7d6d5806edafc79f9ee6ad1abaddfbd462787c6"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-07-14 13:24:33.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-07-20 13:04:18.000000000","tz":60},"subject":"container-server: use LIMIT in get_own_shard_range() query","message":"container-server: use LIMIT in get_own_shard_range() query\n\nThe get_own_shard_range() method is called during every container GET\nor HEAD request (as well as at other times). The method delegates to\nget_shard_ranges() which queries the DB for shard ranges. Although the\nquery has conditions to only select a single shard range name, and the\nmethod will return only the first matching shard range, the query had\nno LIMIT. Adding a LIMIT of 1 significantly reduces execution time\nwhen the DB has many shard range rows.\n\nOn the author\u0027s machine, using a DB with ~12000 shard ranges this\npatch reduces the get_own_shard_range() execution time by a factor of\napproximately 100.\n\nChange-Id: I43f79de3f0603b9fab8bf6f7b4c3b1892a8919b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a70315e2bdbb235eabb08dbda96f35a2a848700"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a70315e2bdbb235eabb08dbda96f35a2a848700"}]},"branch":"refs/heads/master"},"e9e0c4a95f06dadf43e3ba6d18c29adb39105319":{"kind":"REWORK","_number":7,"created":"2023-08-03 13:31:15.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/75/888575/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/888575/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/888575/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/888575/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/888575/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/888575/7"}}},"commit":{"parents":[{"commit":"55cd675f8b7b7f1d46d463141f86e242f8e5f7a6","subject":"Merge \"proxy: add new metrics to account/container_info cache for skip/miss\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55cd675f8b7b7f1d46d463141f86e242f8e5f7a6"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-07-14 13:24:33.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-08-03 13:03:46.000000000","tz":60},"subject":"container-server: use LIMIT in get_own_shard_range() query","message":"container-server: use LIMIT in get_own_shard_range() query\n\nThe get_own_shard_range() method is called during every container GET\nor HEAD request (as well as at other times). The method delegates to\nget_shard_ranges() which queries the DB for shard ranges. Although the\nquery has conditions to only select a single shard range name, and the\nmethod will return only the first matching shard range, the query had\nno LIMIT. Adding a LIMIT of 1 significantly reduces execution time\nwhen the DB has many shard range rows.\n\nOn the author\u0027s machine, using a DB with ~12000 shard ranges this\npatch reduces the get_own_shard_range() execution time by a factor of\napproximately 100.\n\nChange-Id: I43f79de3f0603b9fab8bf6f7b4c3b1892a8919b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9e0c4a95f06dadf43e3ba6d18c29adb39105319"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9e0c4a95f06dadf43e3ba6d18c29adb39105319"}]},"branch":"refs/heads/master"},"6e8d82c97e28026e03f71bf38ccbcd2f1c606afa":{"kind":"NO_CHANGE","_number":8,"created":"2023-08-03 23:47:46.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/75/888575/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/888575/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/888575/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/888575/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/888575/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/888575/8"}}},"commit":{"parents":[{"commit":"55cd675f8b7b7f1d46d463141f86e242f8e5f7a6","subject":"Merge \"proxy: add new metrics to account/container_info cache for skip/miss\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55cd675f8b7b7f1d46d463141f86e242f8e5f7a6"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-07-14 13:24:33.000000000","tz":60},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-08-03 18:48:31.000000000","tz":-420},"subject":"container-server: use LIMIT in get_own_shard_range() query","message":"container-server: use LIMIT in get_own_shard_range() query\n\nThe get_own_shard_range() method is called during every container GET\nor HEAD request (as well as at other times). The method delegates to\nget_shard_ranges() which queries the DB for shard ranges. Although the\nquery has conditions to only select a single shard range name, and the\nmethod will return only the first matching shard range, the query had\nno LIMIT. Adding a LIMIT of 1 significantly reduces execution time\nwhen the DB has many shard range rows.\n\nOn the author\u0027s machine, using a DB with ~12000 shard ranges this\npatch reduces the get_own_shard_range() execution time by a factor of\napproximately 100.\n\nChange-Id: I43f79de3f0603b9fab8bf6f7b4c3b1892a8919b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e8d82c97e28026e03f71bf38ccbcd2f1c606afa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e8d82c97e28026e03f71bf38ccbcd2f1c606afa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
