)]}'
{"id":"openstack%2Fswift~888704","triplet_id":"openstack%2Fswift~stable%2Fvictoria~I05fdf0e1b00d54f5364d02c59019a4c54eb061ca","project":"openstack/swift","branch":"stable/victoria","topic":"gate-fix-cap-virtualenv","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-08-01 09:21:34.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-08-01 10:32:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I05fdf0e1b00d54f5364d02c59019a4c54eb061ca","subject":"[stable-only] Cap virtualenv/setuptools","status":"MERGED","created":"2023-07-17 15:04:36.000000000","updated":"2023-08-01 10:33:51.000000000","submitted":"2023-08-01 10:32:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888704-gate-fix-cap-virtualenv","meta_rev_id":"18ae92dbfce96a077931494cda35c18db9167e1c","_number":888704,"virtual_id_number":888704,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-01 10:32:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2023-08-01 09:23:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-07-31 21:44:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2023-08-01 09:23:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-17 16:20:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-31 21:44:19.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"78b3ecf4d9bdbf3bf6d249f48a06b1fc478c515e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-07-17 15:04:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9a3fc58c1c9b46096a830febc66eb1297495757","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 16:20:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/26e004df2b774447acd5c5c7f58a768c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7348d3e2d4e44827afce8e5fc7f8f6fa : SUCCESS in 9m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d60dc1998a0428ebc534763565f0488 : SUCCESS in 5m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/2ff01bb46f9c4db991749a41968f3a68 : SUCCESS in 1h 10m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fb9f9c1fbd4e4ca5aeb5d82949c75db2 : SUCCESS in 50m 46s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a83a9f7a53b944baa20de88d7c5cf6bf : FAILURE in 3m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ffc4467a1cff43f1af2f0c89a3c6817b : FAILURE in 3m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1f1eed1dac0343ea99caa4535f416b92 : SUCCESS in 13m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8110251612084d869eca5bd493d4648d : SUCCESS in 10m 17s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b07cfbe0fb8242ec9ab7d5b528760ed1 : SUCCESS in 11m 20s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fc2ec85d12e64d26bd6297ce99852067 : SUCCESS in 13m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e36bc627ed434b8289c42a9b955a1ec1 : SUCCESS in 17m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/768abdec55d444acb576e31ed5686334 : SUCCESS in 15m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/45321b5d6beb4b45ad3fbb3db1504c56 : SUCCESS in 17m 33s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1a7b5bba02be4ac8b81c501c425a1b50 : SUCCESS in 15m 19s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/442f47ebe5fe473c9aff5708bb019f06 : SUCCESS in 15m 57s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f8565d2c31f34566923e12bf2033f08d : SUCCESS in 18m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0ba993503fe74f1badb049b00ed431f8 : FAILURE in 6m 31s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/31d119c2b8f449ccbe9ca2c8fbd68692 : SUCCESS in 44m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3d336dfbbec04ce48f9cfee98345f296 : SUCCESS in 32m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/99b001c75e2746198721e2b89353140a : SUCCESS in 27m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b717b9fa5590418c877f1016ef26674e : FAILURE in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd9c419d4c6e4f40961b33dbf7577b4a : FAILURE in 3m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0d610af8692b4b6eaff7d0277c04c724 : RETRY_LIMIT in 7m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f7ec326c6500450a9c8846d4b9ec5167 : SUCCESS in 55m 31s","accounts_in_message":[],"_revision_number":1},{"id":"2b75fa6e910f2cfdd18631dc504857aab6df470d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-07-31 18:19:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"149c17e1ce3570916d4f46d7656ebaa738ac7837","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-07-31 19:01:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2aa475bb31364ea3b0094f838d1255c0988e5165","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 20:01:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1efb790fd43412ca33ccf91e7e45b89\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9a8e00c0f89439cbbaffb245e9f593f : SUCCESS in 7m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/40837793de8c4f7284f592cccb73f202 : SUCCESS in 2m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5cc1023de8eb4e20aa700d0bc3e6dc74 : SUCCESS in 4m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/4815f83ac58649ea8aa61b04e17c5328 : SUCCESS in 41m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/58c04df5c45c4ba28669a667c1843061 : SUCCESS in 53m 44s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/47b0022d5cab4c4daa312f5f3008eaff : FAILURE in 3m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7b78979861ac4840b0247213c8216dec : FAILURE in 3m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/94c4c17bfb78428c978b26096551521f : SUCCESS in 10m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/33d83b10ffc74c4c8f21baee86b675fb : SUCCESS in 13m 03s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dc090de0b2b44cbbbd8306bd496551b2 : SUCCESS in 14m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/43c0566fcdac4d31ba0b23827b7c888c : SUCCESS in 11m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5cedb9c4921747599cf7ab261607f439 : SUCCESS in 16m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/955030e2879c4ceca454b541c3e9f17a : SUCCESS in 16m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ba047815fcc34a63935721c2d1e15a6e : SUCCESS in 17m 10s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b01ddd558fa642cd9d16faac96ae127a : SUCCESS in 16m 33s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b3d972dfe3ef479b9e9703e04bda7ec4 : SUCCESS in 16m 52s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/cfd10f70436b4dc7b5698962ef7a722f : SUCCESS in 18m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8ca62282c07f436bad574d1177f5a1d9 : FAILURE in 6m 50s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3ce047f1ae8249f880ae4a3daa580995 : SUCCESS in 42m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a8f5b12bea5b413e99038f7822439f91 : SUCCESS in 43m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a0d0f9c5d4fd499985b89973963b7b50 : SUCCESS in 32m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/51e5e9ea1dea48daa8385e516e20acda : SUCCESS in 13m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc5f1bdc60f949d88bc68f8beb810705 : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1d687aae236c47449930ce262c109dcc : RETRY_LIMIT in 6m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa0f33ed2d9f42d2b1c35fa8491d2396 : SUCCESS in 46m 27s","accounts_in_message":[],"_revision_number":3},{"id":"f0cf3c60c071c27ad7e33105e731f41e4ace3658","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-07-31 21:44:19.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"22e18564b6ff504b63f7bc1a30a219148c52c2e2","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-08-01 09:21:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7498cbbe472041e2fac0e3a4f5cc9ac6bbf5178c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-08-01 09:23:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"20b69998fa8f5f850e94c47c0b4bf112023f2a92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 09:24:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"540a0be2107642f3a05d60a9d50e48e0d395a8e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 10:32:02.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95200cd80ea948aca46c21d6e4a7534c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78567c9e09b848cfaf138463d6a00cf7 : SUCCESS in 9m 29s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2ea1c633ef564640b9cd5f86c834363d : SUCCESS in 3m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba576e397e124202aac1f823ff4a3254 : SUCCESS in 5m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/562fff0b77fc448c9d10759591d393fd : SUCCESS in 1h 07m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d897855e242447eeb2f4dc3ac2684a8a : SUCCESS in 43m 19s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/844793daa44b493289c8055765793a51 : POST_FAILURE in 5m 20s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/0a885c6dafc040189919b175e25a8511 : POST_FAILURE in 5m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7f171ba2cb6d442fbee5b42be9d1205e : SUCCESS in 12m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2bcc2aadb3ee4732896fcf4e98de4b3c : SUCCESS in 11m 21s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6fddb25e3cd44182a1bf0577730e998c : SUCCESS in 11m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0b8f2843463347af8e51b78ccc66b071 : SUCCESS in 10m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/00f7a88872164d71beb29bcf42e7ffad : SUCCESS in 16m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/33143823529f4928a9cf10866bf53fe3 : SUCCESS in 17m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/064f595b99ba4c7a8e7ed02f4cf5a5e1 : SUCCESS in 16m 59s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/27f01d4ff4024e5c8a61ecc283ec6f3b : SUCCESS in 17m 06s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d7b10c2e9ec24988b6d510b64ded4433 : SUCCESS in 18m 55s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/46308832a0904dfeac2c8db16c63b3b5 : SUCCESS in 41m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3412df474b734cf8ba5dcb040c814ac9 : SUCCESS in 42m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7aac7e12a802434d996a7a6536266caf : SUCCESS in 35m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1918544d14e84360a7f8a69d1055f927 : SUCCESS in 13m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e26b96a98f243bc8c22ea940d60beb9 : SUCCESS in 7m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d196bee0c834964893afa2f395be0d4 : SUCCESS in 50m 35s","accounts_in_message":[],"_revision_number":3},{"id":"6600234fb10801605aa4158518f11d9fbfa6b5ab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 10:32:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"18ae92dbfce96a077931494cda35c18db9167e1c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 10:33:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a5ff392cd60497198a6078ade14fced\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dad39d323534fcd8a75e3651cd4ab9b : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81adfaf3a6944cf1a56c69f548678e52 : SUCCESS in 39s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/858ca1c4824843019989c41db5b61e64 : FAILURE in 24s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/079dc2c5158d414697d150280284cbd9 : FAILURE in 27s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f80dce1f1921ec7812893bb22243b86aa7356f93","revisions":{"6c6b59d7542401ccb04c48dcd416af77867a5b67":{"kind":"REWORK","_number":1,"created":"2023-07-17 15:04:36.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/04/888704/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/04/888704/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/888704/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/888704/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/888704/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/04/888704/1"}}},"commit":{"parents":[{"commit":"069865e2b8c80000d0df197a33ec34510b54002b","subject":"Merge \"s3api: Prevent XXE injections\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/069865e2b8c80000d0df197a33ec34510b54002b"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-07-17 15:03:19.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-07-17 15:04:11.000000000","tz":120},"subject":"[stable-only] Cap virtualenv/setuptools","message":"[stable-only] Cap virtualenv/setuptools\n\nWith latest virtualenv release (20.24.0) and its bundled setuptools\nversion, stable/victoria gates started to fail [1], because setuptools\nis not compatible with the pbr version (5.5.0) which is in victoria\nbranch\u0027s upper constraint [2]. pbr\u0027s next release (5.5.1) removed the\nfailing code, this is why the change is only needed in stable/victoria.\n\nThis patch caps virtualenv to \u003c20.24.0 to fix stable/victoria\u0027s gate,\nand also caps tox for the \u0027inner\u0027 virtualenv, otherwise tox\u003e4 would be\ninstalled, that would introduce failures (those failures are fixed in\nnewer stable branches, but it would need several backports and \u0027outer\u0027\ntox is capped \u003c4 anyway, so no worth dealing with it here).\n\n[1] AttributeError: module \u0027setuptools.command.easy_install\u0027 has no attribute \u0027get_script_header\u0027\n[2] https://opendev.org/openstack/requirements/src/commit/636b6b3bde4b68ef27ac6b524cb888152eddf517/upper-constraints.txt#L44\n\nChange-Id: I05fdf0e1b00d54f5364d02c59019a4c54eb061ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c6b59d7542401ccb04c48dcd416af77867a5b67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c6b59d7542401ccb04c48dcd416af77867a5b67"}]},"branch":"refs/heads/stable/victoria"},"d93dd50f7cb01b34eda5510a3f9639c90cfc8e68":{"kind":"REWORK","_number":2,"created":"2023-07-31 18:19:41.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/04/888704/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/04/888704/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/888704/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/888704/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/888704/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/04/888704/2"}}},"commit":{"parents":[{"commit":"069865e2b8c80000d0df197a33ec34510b54002b","subject":"Merge \"s3api: Prevent XXE injections\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/069865e2b8c80000d0df197a33ec34510b54002b"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-07-17 15:03:19.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-07-31 18:18:09.000000000","tz":-420},"subject":"[stable-only] Cap virtualenv/setuptools","message":"[stable-only] Cap virtualenv/setuptools\n\nWith latest virtualenv release (20.24.0) and its bundled setuptools\nversion, stable/victoria gates started to fail [1], because setuptools\nis not compatible with the pbr version (5.5.0) which is in victoria\nbranch\u0027s upper constraint [2]. pbr\u0027s next release (5.5.1) removed the\nfailing code, this is why the change is only needed in stable/victoria.\n\nThis patch caps virtualenv to \u003c20.24.0 to fix stable/victoria\u0027s gate,\nand also caps tox for the \u0027inner\u0027 virtualenv, otherwise tox\u003e4 would be\ninstalled, that would introduce failures (those failures are fixed in\nnewer stable branches, but it would need several backports and \u0027outer\u0027\ntox is capped \u003c4 anyway, so no worth dealing with it here).\n\nAlso pin the pep8 and lower-constraints jobs to bionic nodes, which\nseem better positioned to deal with py27.\n\n[1] AttributeError: module \u0027setuptools.command.easy_install\u0027 has no attribute \u0027get_script_header\u0027\n[2] https://opendev.org/openstack/requirements/src/commit/636b6b3bde4b68ef27ac6b524cb888152eddf517/upper-constraints.txt#L44\n\nChange-Id: I05fdf0e1b00d54f5364d02c59019a4c54eb061ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d93dd50f7cb01b34eda5510a3f9639c90cfc8e68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d93dd50f7cb01b34eda5510a3f9639c90cfc8e68"}]},"branch":"refs/heads/stable/victoria"},"f80dce1f1921ec7812893bb22243b86aa7356f93":{"kind":"REWORK","_number":3,"created":"2023-07-31 19:01:46.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/04/888704/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/04/888704/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/888704/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/888704/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/888704/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/04/888704/3"}}},"commit":{"parents":[{"commit":"069865e2b8c80000d0df197a33ec34510b54002b","subject":"Merge \"s3api: Prevent XXE injections\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/069865e2b8c80000d0df197a33ec34510b54002b"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-07-17 15:03:19.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-07-31 19:00:00.000000000","tz":-420},"subject":"[stable-only] Cap virtualenv/setuptools","message":"[stable-only] Cap virtualenv/setuptools\n\nWith latest virtualenv release (20.24.0) and its bundled setuptools\nversion, stable/victoria gates started to fail [1], because setuptools\nis not compatible with the pbr version (5.5.0) which is in victoria\nbranch\u0027s upper constraint [2]. pbr\u0027s next release (5.5.1) removed the\nfailing code, this is why the change is only needed in stable/victoria.\n\nThis patch caps virtualenv to \u003c20.24.0 to fix stable/victoria\u0027s gate,\nand also caps tox for the \u0027inner\u0027 virtualenv, otherwise tox\u003e4 would be\ninstalled, that would introduce failures (those failures are fixed in\nnewer stable branches, but it would need several backports and \u0027outer\u0027\ntox is capped \u003c4 anyway, so no worth dealing with it here).\n\nAlso:\n- Pin the pep8 and lower-constraints jobs to bionic nodes, which seem\n  better positioned to deal with py27.\n- Pin pyOpenSSL\u003d\u003d21.0.0 in lower-constraints.txt, as that was the last\n  version to support py27.\n\n[1] AttributeError: module \u0027setuptools.command.easy_install\u0027 has no attribute \u0027get_script_header\u0027\n[2] https://opendev.org/openstack/requirements/src/commit/636b6b3bde4b68ef27ac6b524cb888152eddf517/upper-constraints.txt#L44\n\nChange-Id: I05fdf0e1b00d54f5364d02c59019a4c54eb061ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f80dce1f1921ec7812893bb22243b86aa7356f93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f80dce1f1921ec7812893bb22243b86aa7356f93"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
