)]}'
{"id":"openstack%2Fswift~888774","triplet_id":"openstack%2Fswift~master~Ia89485bae8f92f4f3d9f5375cab8ff08f70a11a7","project":"openstack/swift","branch":"master","topic":"p-s3api-stats-mutation","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-07-20 21:48:40.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-07-20 15:40:26.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}},"hashtags":[],"change_id":"Ia89485bae8f92f4f3d9f5375cab8ff08f70a11a7","subject":"s3api: fix statsd prefix mutation","status":"MERGED","created":"2023-07-18 14:31:46.000000000","updated":"2023-07-20 21:50:30.000000000","submitted":"2023-07-20 21:48:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"888774-p-s3api-stats-mutation","meta_rev_id":"ef63002a6c61abf536536853e7ab8a7b1e513f03","_number":888774,"virtual_id_number":888774,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-20 21:48:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2023-07-20 15:40:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2023-07-20 15:40:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-18 15:29:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-18 15:59:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-19 15:13:22.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2023-07-20 15:40:26.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"aeb41a185431bbc8a5d4b28748319cf4fbde1d47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-18 14:31:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62e4e0e1d83738df51e8af9d20fbfd2dddcc4f66","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-18 14:41:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8fa8318177e50ce79f91dd013faaa6f84b0fbbdb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-18 15:29:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22169b7c8ba0487f9ea4aba59211d13e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2368d201f8134fc0b7c4c68da223c89a : SUCCESS in 12m 48s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f2ede15eccc641b29cef6368c770837d : SUCCESS in 15m 06s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0399e939fdf14a8fabf793a91300a434 : SUCCESS in 16m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e82d7bfbf22647b59509e4234fe5ab5a : SUCCESS in 56m 04s","accounts_in_message":[],"_revision_number":1},{"id":"fc103c95e70c0a66db6f7c4c125c5aeed568e8e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-18 15:59:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b518c0f7a6f94ac8aacc553db8d79a45\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e93014a0c97472a8cf0ba6e575eb27f : SUCCESS in 8m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/7f5e80e0a2d64fe4a65924e2f7357f96 : SUCCESS in 39m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a5ed1796a94a41d588a16701f2be9954 : SUCCESS in 38m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c1f32aecf46414f8f773a82c8a56c08 : SUCCESS in 1h 13m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8a3f7c17f98041b2bbc5da413ab3ddea : SUCCESS in 15m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7db05d77e8c54421863cefca1f06d6a5 : SUCCESS in 24m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7be10a7dfb3e4e77ad904ce2f94b23a5 : SUCCESS in 26m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/377ba0976fe74d629b65aeb1a16d3d7b : SUCCESS in 7m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a411a08ec5334bdd8240a7433241af13 : SUCCESS in 5m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/22260a3043d0477ea6e96c47615491dc : SUCCESS in 24m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/08b56536ffe44b93b376ca8aa0005671 : SUCCESS in 25m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e3c708276d9247bcaef93d6f185fd8fb : SUCCESS in 24m 35s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/18c3f6d965514096bc065e172cd2b456 : SUCCESS in 26m 47s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9f252b6839bb4c1e8d3307429ca2012b : SUCCESS in 23m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/36cd5ac00303434f82f11f8cbb7f91f4 : SUCCESS in 17m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2b091f486cd545b38f797cf21f775ed7 : SUCCESS in 15m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ed32e7ed9aff48398c921530b83649ba : SUCCESS in 16m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9c64c34a268f4664baf16d3342673f4b : SUCCESS in 11m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/13c56ef2e0824c2eb8654fc59f080167 : SUCCESS in 25m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/881c8a00adb24bce892f464a740e4fa8 : SUCCESS in 26m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d87f00b167564461a55ab5594b56ff79 : SUCCESS in 6m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/955cc3db842742c888e6a69d71c27456 : SUCCESS in 16m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c0bfc5a3f21b41d99094982969162bea : SUCCESS in 6m 22s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b012c9b7d20d46bca657f866d9c54e40 : SUCCESS in 41m 47s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/54235b9cdfd5492e9e9cb51962e21029 : SUCCESS in 1h 20m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/de60b42a43024ac38542510843e29189 : SUCCESS in 37m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2fd65107b60b4e2badb1939c15cbce15 : SUCCESS in 37m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c031df8054274fb68675b33f8aff7eeb : SUCCESS in 19m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aa25934d2fa4f5092ef128cf87263f5 : SUCCESS in 4m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ebe25012fe884e6986b093b6fab02141 : SUCCESS in 19m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b2f5317917940cb91bb44604bdb64ce : SUCCESS in 1h 10m 28s","accounts_in_message":[],"_revision_number":1},{"id":"d11dbe187ef9b714a388e45521ad94df5ef41678","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-07-19 15:13:22.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"babd8868322f7ad035f4aa842d62be5ffe7f0c1e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-20 09:25:28.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b9ff1d5a315c99d5035e45c9a2293e9d906117c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-20 09:25:46.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"3894a2ddbf38d8d76ac60eb918bddfaabb94270b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 10:54:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96eb28b45a5e4eddbab7279da61e9491\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/692b77603a22402aa51a24b9d9be97bf : SUCCESS in 1h 07m 40s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ed9b445490524a3fa1915a4aa7f315d9 : SUCCESS in 1h 16m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b287f599335e48c0b2552ce9ac69560c : SUCCESS in 1h 25m 33s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4c1729ea9ad04bc5a91aef06e438be10 : SUCCESS in 1h 26m 29s","accounts_in_message":[],"_revision_number":2},{"id":"ad45d4af1a641858e3c508240a51215de196fb4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 12:19:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e859bff467c4c3eaf3fe873fee3f386\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35181cc9b6524579a91e60c30c85fc48 : SUCCESS in 9m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/a690203342634b25bf88e9c71e0e64cb : FAILURE in 55m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9ac75f9b957d4841ba7cd7f8b24fc6f7 : SUCCESS in 1h 01m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e314cfeba5b460eae07ccf4f2709aa9 : SUCCESS in 1h 09m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/09055520ad844c22b4c1f6400ddbedf1 : SUCCESS in 18m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bac3945122254e1e85d32fe0c51a0231 : SUCCESS in 21m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2957bacc7a474b99959e01f325b3e023 : SUCCESS in 20m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e2eb0e123774469e80db21773e508d56 : SUCCESS in 9m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ec73b3dc430f404cbde2f6f4b80f67d7 : SUCCESS in 6m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/757fab5f3b9e4e00926805eeb809c841 : SUCCESS in 15m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5878e9bb0bc24d2a93737b28aebe6f92 : SUCCESS in 17m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/867269333a5b43d6af808516290817c9 : SUCCESS in 16m 10s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2982e6dbe6c3473fa2090361c27d301b : SUCCESS in 18m 09s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/354dc64a223c4b8598be439a9039ff28 : SUCCESS in 16m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/23608a12ddf14f96b9ec39fcb7dfa702 : SUCCESS in 22m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/115472f8d1b942af95edf7d5a62675db : SUCCESS in 20m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f7ff592825c44271b0718c4718e3a748 : SUCCESS in 15m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/73c17fb7f7234942953d0b6e3893e1e2 : SUCCESS in 17m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/158c1d1a71374a198b6164b8a659e5ff : SUCCESS in 14m 07s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e2e0296103ce4d01b209375abd726f31 : SUCCESS in 16m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cc218343f648463ea690c66052e608fa : SUCCESS in 7m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/78bc1e02ef1542a8af0ac41d56ee2a57 : SUCCESS in 19m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b7511e4346d24b7c99a21cf7b76ac5df : SUCCESS in 7m 17s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1213fc6129c448678a12e3ea1ee57e29 : SUCCESS in 1h 18m 52s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9dc00b591ddd4b5485fb25afae06f877 : FAILURE in 1h 09m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/92e398e964b9446384e1f806d185f675 : SUCCESS in 59m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7d0da640962b4fd892bdd9880040d7d5 : SUCCESS in 55m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/742c887cf5e643e0be830662ed93caf6 : SUCCESS in 17m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0bba6cb8d6a428096dbec4892bf923e : SUCCESS in 4m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3d004e51ffac49bd9dee7037061d00d5 : SUCCESS in 30m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/40167b54ea764562b223d7745c02a16d : SUCCESS in 52m 41s","accounts_in_message":[],"_revision_number":2},{"id":"762e7f8db97243338186945a26fccefcc01e0f7c","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-07-20 15:40:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"44820f2cc41d67f31e439ce32717d8f303b2afa1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 17:30:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b44202cedec474aabd8826fddf82a66\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04be7616f29a4b4bb9d1df9936c2735e : SUCCESS in 7m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/cda5d70fa1234dbfbcaef6c2f9c4dc2c : SUCCESS in 40m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bcbcd73ef95b451293fae9516dfdbdfe : SUCCESS in 43m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/35ced802a16e4160ba1a1401fdbcc93a : SUCCESS in 53m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f25f2f58c3ea485db86cc338b452ec93 : SUCCESS in 19m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9df0758486454a609261ff6dcb157be5 : SUCCESS in 26m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ba6b25742c2e44589c9f316745911bab : SUCCESS in 20m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8f4e8166cbb14c56ab666ac7f3f08ad8 : SUCCESS in 10m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b68a21e63a3f4dc19917672a8b29381c : SUCCESS in 6m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ab1d08371c294a64aa322736057ff8df : SUCCESS in 17m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cd1fc5da6d534ea6bf486da1b86407d0 : SUCCESS in 25m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/35718367679647af8f0e1f6d816c50d0 : SUCCESS in 16m 19s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ee98614d385944999005a5eeb49ed786 : SUCCESS in 14m 58s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b0f224fef24f4fc69cce9be7c0b2ce04 : SUCCESS in 15m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3e411ba72a6a4da59c4d12ced2b1df2f : SUCCESS in 17m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fb69a503280d448eb6a1adef2b23a53e : SUCCESS in 18m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f7fa8bb825b140a1a5f0559c69fabe9c : SUCCESS in 15m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ac6df35ea0f14188a444cc484e33e8cc : SUCCESS in 15m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/04fcb15d93f44c2280438d519b740d4e : SUCCESS in 15m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e4e50a8eae52416e98b124a7d1b0ed3d : SUCCESS in 15m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ce3e7094312340e88c2acf65fe4e9c5c : SUCCESS in 6m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/67f28329ef184eccb21592b009251a45 : SUCCESS in 19m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a4812466d2b44f10be0cc22f22a2de7b : SUCCESS in 7m 20s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ca5e25271f7c44beb7cfccc8914cc94c : SUCCESS in 50m 07s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f21d26537dec48a992485627cda81f3a : SUCCESS in 1h 01m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f74b1884ce7642178d92ef4725525c95 : SUCCESS in 54m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6dee59c972d94b9d8159ec8cd8e295f0 : SUCCESS in 38m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/05297092c22a4680b1670acd68bd12c6 : SUCCESS in 16m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1e8057edca5486f83f1c3fddc629e81 : SUCCESS in 5m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3bcd2ef119cd4b809f1c39f09b71cc5f : SUCCESS in 19m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9120395ede6f46c7ae6ad9de4c243258 : SUCCESS in 54m 58s","accounts_in_message":[],"_revision_number":2},{"id":"a3fea543d101dda64a42b609425ad97b4dc983da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 17:31:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8e1f26a0488bc8b9c582d908f1d4da6e5a0e03b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 21:48:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f251ca97f2d44f21ac10156434f2f1e3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75d9fa467f7444f09a82e019ac0199c9 : SUCCESS in 7m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/11208c01578e4d61895275fa2248c905 : SUCCESS in 1h 34m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dc5c29c526b14841b15b8d2d6f3acc78 : SUCCESS in 45m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8baeee73f1914b158d2abfa27c71dc9a : SUCCESS in 52m 20s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/913bb65a0abb48be85f7dab684930f13 : SUCCESS in 8m 10s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/90b5d85dfb6d4c1782c13e5c4d7f6fd1 : SUCCESS in 5m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d3c286cd8c3b46d894b0b1e7b3a820b4 : SUCCESS in 16m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4c4d93d663544f52bd4f48003cde3310 : SUCCESS in 17m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f9f2ab4ca06f4edfaacddc1db2ab6c61 : SUCCESS in 15m 21s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8f55fcaa9ca5417b9dcf44054f118da4 : SUCCESS in 26m 23s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/67f3dd6e86a448d5a086e4565e686dfd : SUCCESS in 15m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5e5106f7b07349b2989097ecadd5f948 : SUCCESS in 16m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eeec4be050f44c1da008aa9e4be833ca : SUCCESS in 17m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/90656723a18e4510ba0760c71242a5be : SUCCESS in 17m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2d7df5db152542a3ac3d12abeb3da30b : SUCCESS in 14m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ea9331bf4cd14fce91fdc46ed522aa56 : SUCCESS in 25m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/23545ac9f9bd45fea054eead598be2c5 : SUCCESS in 13m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/313fc2281e474a9bb8e808455cd339c1 : SUCCESS in 8m 04s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/378e62b8568b4ae4b5722b7b676df92c : SUCCESS in 6m 43s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/45487f988d63424daa5826cc583fc058 : SUCCESS in 1h 11m 28s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/42f0a0eb73da4b91964032a5ab3e5a73 : SUCCESS in 1h 03m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3cfd34c4ff1f456a86388b3324aa2c97 : SUCCESS in 39m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/920b298428fd45d7ac08410a8a19a0c3 : SUCCESS in 56m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/738e1849093b466ebc062d6afdb17487 : SUCCESS in 29m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1560eab118244ad6a5a000fd65c4a537 : SUCCESS in 5m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25548da796e8477db56d1e3e6c696e9b : SUCCESS in 50m 41s","accounts_in_message":[],"_revision_number":2},{"id":"a9d9f57d262d85cf9559d20f6a36f96d6e659777","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 21:48:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ef63002a6c61abf536536853e7ab8a7b1e513f03","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 21:50:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a149fb57c3046c8b968e5f75206a817\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3701d814ae6c4b01837a90b94d18a09e : SUCCESS in 48s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/7491139714da4022b9f05c8c09344460 : SUCCESS in 1m 31s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/ad511d4196c148d2a2ac5fc6757ed79b : SUCCESS in 31s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bdbe8ce9f821a6809f19d7aefa8c1e9a7d1328de","revisions":{"af3cf63d80508e78015c4ea08c5b3133e78162ed":{"kind":"REWORK","_number":1,"created":"2023-07-18 14:31:46.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/74/888774/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/888774/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/888774/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/888774/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/888774/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/888774/1"}}},"commit":{"parents":[{"commit":"7cc8a0172916410f75ee8fde17050957677c45b2","subject":"Merge \"Object-server: keep SLO manifest files in page cache.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7cc8a0172916410f75ee8fde17050957677c45b2"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-07-18 14:00:25.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-07-18 14:25:31.000000000","tz":60},"subject":"s3api: fix statsd prefix mutation","message":"s3api: fix statsd prefix mutation\n\nThe \u0027log_route\u0027 argument of utils.get_logger() determines which global\nLogger instance is wrapped by the returned LogAdapter. Most middlewares\n(s3api being the exception) explicity set \u0027log_route\u0027 to equal the\nmiddleware \u0027brief\u0027 name e.g. \u0027bulk\u0027, \u0027tempauth\u0027 etc. However, the\ns3api middleware sets \u0027log_route\u0027 to be the config \u0027log_name\u0027, if that\nkey is found in config.\n\nWhen a proxy pipeline is instantiated via wsgi.run_wsgi(), all\nmiddlewares and the proxy app are passed a default conf with\n\u0027\"log_name\": \"proxy-server\"\u0027. As a result, the s3api middleware calls\nget_logger() with log_route\u003d\u0027proxy-server\u0027 and its LogAdapter\ntherefore shares the same Logger instance used by proxy-server app\n(and any other middleware that similarly fails to explicitly\ndifferentiate \u0027log_route)\u0027.\n\nEach Logger instance has a StatsdClient instance bound to it by\nget_logger(). The Related-Change added statsd metrics to the s3api\nmiddleware and sets \u0027s3api\u0027 as the \u0027statsd_tail_prefix\u0027 when calling\nget_logger(). This had the unintended effect of replacing the shared\nLogger instance\u0027s StatsdClient with one that has prefix \u0027s3api\u0027, such\nthat stats emitted by the proxy app (e.g. memcache shard range\nhit/miss stats) would be erroneously prefixed with \u0027s3api\u0027.\n\nThis patch modifies the s3api middleware logger instantiation to\nexplictly set log_route\u003d\u0027s3api\u0027, so that the s3api middleware\nLogAdapter now wraps a unique global Logger instance, with a unique\nStatsdClient instance bound to it.\n\nThe \u0027server\u0027 attribute of the middleware\u0027s LogAdapter, which may be\nincluded in log lines by the \"%(server)s\" format element, is not\naffected by this change. Its value is derived from the config\n\u0027log_name\u0027 or the \u0027name\u0027 argument passed to get_logger().\n\nChange-Id: Ia89485bae8f92f4f3d9f5375cab8ff08f70a11a7\nRelated-Change: I4976b3ee24e4ec498c66359f391813261d42c495\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af3cf63d80508e78015c4ea08c5b3133e78162ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af3cf63d80508e78015c4ea08c5b3133e78162ed"}]},"branch":"refs/heads/master"},"bdbe8ce9f821a6809f19d7aefa8c1e9a7d1328de":{"kind":"REWORK","_number":2,"created":"2023-07-20 09:25:46.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/74/888774/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/888774/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/888774/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/888774/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/888774/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/888774/2"}}},"commit":{"parents":[{"commit":"7cc8a0172916410f75ee8fde17050957677c45b2","subject":"Merge \"Object-server: keep SLO manifest files in page cache.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7cc8a0172916410f75ee8fde17050957677c45b2"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-07-18 14:00:25.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-07-20 08:56:09.000000000","tz":60},"subject":"s3api: fix statsd prefix mutation","message":"s3api: fix statsd prefix mutation\n\nThe \u0027log_route\u0027 argument of utils.get_logger() determines which global\nLogger instance is wrapped by the returned LogAdapter. Most middlewares\n(s3api being the exception) explicity set \u0027log_route\u0027 to equal the\nmiddleware \u0027brief\u0027 name e.g. \u0027bulk\u0027, \u0027tempauth\u0027 etc. However, the\ns3api middleware sets \u0027log_route\u0027 to be the config \u0027log_name\u0027, if that\nkey is found in config.\n\nWhen a proxy pipeline is instantiated via wsgi.run_wsgi(), all\nmiddlewares and the proxy app are passed a default conf with\n\u0027\"log_name\": \"proxy-server\"\u0027. As a result, the s3api middleware calls\nget_logger() with log_route\u003d\u0027proxy-server\u0027 and its LogAdapter\ntherefore shares the same Logger instance used by proxy-server app\n(and any other middleware that similarly fails to explicitly\ndifferentiate \u0027log_route)\u0027.\n\nEach Logger instance has a StatsdClient instance bound to it by\nget_logger(). The Related-Change added statsd metrics to the s3api\nmiddleware and sets \u0027s3api\u0027 as the \u0027statsd_tail_prefix\u0027 when calling\nget_logger(). This had the unintended effect of replacing the shared\nLogger instance\u0027s StatsdClient with one that has prefix \u0027s3api\u0027, such\nthat stats emitted by the proxy app (e.g. memcache shard range\nhit/miss stats) would be erroneously prefixed with \u0027s3api\u0027.\n\nThis patch modifies the s3api middleware logger instantiation to\nexplictly set log_route\u003d\u0027s3api\u0027, so that the s3api middleware\nLogAdapter now wraps a unique global Logger instance, with a unique\nStatsdClient instance bound to it.\n\nThe \u0027server\u0027 attribute of the middleware\u0027s LogAdapter, which may be\nincluded in log lines by the \"%(server)s\" format element, is not\naffected by this change. Its value is derived from the config\n\u0027log_name\u0027 or the \u0027name\u0027 argument passed to get_logger().\n\nChange-Id: Ia89485bae8f92f4f3d9f5375cab8ff08f70a11a7\nRelated-Change: I4976b3ee24e4ec498c66359f391813261d42c495\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bdbe8ce9f821a6809f19d7aefa8c1e9a7d1328de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bdbe8ce9f821a6809f19d7aefa8c1e9a7d1328de"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
