)]}'
{"id":"openstack%2Fswift~889146","triplet_id":"openstack%2Fswift~master~I7b9907645d7357824e197fa213f0ec68283ee72b","project":"openstack/swift","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"I7b9907645d7357824e197fa213f0ec68283ee72b","subject":"Imported Translations from Zanata","status":"MERGED","created":"2023-07-21 04:14:50.000000000","updated":"2023-07-31 19:21:30.000000000","submitted":"2023-07-31 19:20:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889146-zanata/translations","meta_rev_id":"0f7257817625214409d662e75f785f9726d235c0","_number":889146,"virtual_id_number":889146,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-31 19:20:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-07-31 19:06:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-07-31 19:06:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-21 04:42:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-31 19:06:52.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"24539ef25e19de5ff0277bd750421949f5542b52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-07-21 04:14:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88687c149833ec38d7189598a635bb0118c04b18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-21 04:42:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03ee5194637040448d6b129d1ed638ac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8af0d27d8fa14004a25ec980c0e376d5 : SUCCESS in 8m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6a33f7576a24510be02884296f3c5e0 : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":1},{"id":"6c815d71a5e49b17dcc4da96c57343651d562725","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-07-22 03:59:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ef5d936c1859b0406e44be4abfe021f51517b0a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-22 04:13:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be7304a861a7408e98ce9e9ba48ad311\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3861c0c31099494ab333178a5553cf08 : SUCCESS in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55fa78b3b1de431588b7645fd40efb48 : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":2},{"id":"f3771a3dcfecb5136819b504774f1f39678dacdb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-07-25 03:57:59.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"0458934eb44cd7236b6afa70db6460eece6fc837","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 04:06:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/143ed636fd104a34b55bab7a59c49257\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e0648d3ba644171a37b66fd5929ae1f : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c85a87d4382444eaed0a073997a1522 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":3},{"id":"5b45cf68398c0a85992598e42f259f2c956864b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-07-27 03:58:30.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"8987e50ec8d7d1ea78dc70ebdb69b2b350a5ba79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 04:08:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c591070e29748afb06b2cf4622a2dca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bda6a9bd417c4022a79eb464f9bbedc8 : SUCCESS in 9m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7727dc9814534b1aad4f51cacc300d77 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":4},{"id":"bc61908d96bdad7a93ffc7fb4f79d19b4aef70d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-07-28 04:12:02.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"eb6321b8719efd370e8dfd19119ad2bdff1cd03a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 04:35:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/731219f49b9d4d96a79009e0515569da\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4903aea358642668734f50671fccd19 : SUCCESS in 10m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b7eca7d548f4754a451d0b1bbfcb0d8 : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":5},{"id":"1dc73c9c77211a61adabad76a58d53d2982cccef","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-07-31 19:06:52.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"6e079f8901f674ab3149a96774abaff8ef16ace1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 19:07:14.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"d2703dd21344cb66e90186ae24d79c9add9c84e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 19:20:29.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbeaf01569674d12888ed7e151217dc5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/721d3258c0f240f4aa358dd375fb1e56 : SUCCESS in 8m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf885fc2d3884ff8a391d5e33247bd8b : SUCCESS in 5m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9067ec82cebd44d4a6cb05580393a2d0 : SUCCESS in 7m 32s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4ad2b93986494cdab836fa77f1409de7 : SUCCESS in 7m 44s","accounts_in_message":[],"_revision_number":5},{"id":"dd438afb29f868019dc3d50d54f37352c9e4f7ce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 19:20:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"0f7257817625214409d662e75f785f9726d235c0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 19:21:30.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c10f174907014dc9b21e5aeae3f86269\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7abed41367d9405baba75a842fd3b315 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b64a6ae8f1444c0ab807dccd07056828 : SUCCESS in 36s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/8b8f8e8a32724f6681306bebb6b52794 : FAILURE in 28s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/27ad1ceb6cd54e8db13074fa9ce4261d : FAILURE in 29s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"8a5baf8e1316f39b169b3cb60cd243994f335a3b","revisions":{"2ae2cefe1d31a69d65bf26ef21de3d0f482a76f1":{"kind":"REWORK","_number":1,"created":"2023-07-21 04:14:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/46/889146/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/46/889146/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/889146/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/889146/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/889146/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/46/889146/1"}}},"commit":{"parents":[{"commit":"bdbe8ce9f821a6809f19d7aefa8c1e9a7d1328de","subject":"s3api: fix statsd prefix mutation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bdbe8ce9f821a6809f19d7aefa8c1e9a7d1328de"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-07-21 04:14:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-07-21 04:14:48.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I7b9907645d7357824e197fa213f0ec68283ee72b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ae2cefe1d31a69d65bf26ef21de3d0f482a76f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ae2cefe1d31a69d65bf26ef21de3d0f482a76f1"}]},"branch":"refs/heads/master"},"2d7da9914eae590929d867354e55bbb756143d6d":{"kind":"REWORK","_number":2,"created":"2023-07-22 03:59:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/46/889146/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/46/889146/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/889146/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/889146/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/889146/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/46/889146/2"}}},"commit":{"parents":[{"commit":"1f007fd86bae35d20861755fb11440a222035681","subject":"Merge \"Add a docstring for utils.Namespace\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f007fd86bae35d20861755fb11440a222035681"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-07-22 03:59:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-07-22 03:59:37.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I7b9907645d7357824e197fa213f0ec68283ee72b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d7da9914eae590929d867354e55bbb756143d6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d7da9914eae590929d867354e55bbb756143d6d"}]},"branch":"refs/heads/master"},"694c9fb7aa8bbfbccd35a3403750e591107e01fd":{"kind":"REWORK","_number":3,"created":"2023-07-25 03:57:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/46/889146/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/46/889146/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/889146/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/889146/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/889146/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/46/889146/3"}}},"commit":{"parents":[{"commit":"8b900becd5bed2e0c0555cc55bcaef372910b88b","subject":"Merge \"proxy: remove client_chunk_size and skip_bytes from GetOrHeadHandler\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b900becd5bed2e0c0555cc55bcaef372910b88b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-07-25 03:57:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-07-25 03:57:57.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I7b9907645d7357824e197fa213f0ec68283ee72b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/694c9fb7aa8bbfbccd35a3403750e591107e01fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/694c9fb7aa8bbfbccd35a3403750e591107e01fd"}]},"branch":"refs/heads/master"},"ee8ae73cc72969b7affb3b1f7d7a66c92edcf84c":{"kind":"REWORK","_number":4,"created":"2023-07-27 03:58:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/46/889146/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/46/889146/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/889146/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/889146/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/889146/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/46/889146/4"}}},"commit":{"parents":[{"commit":"167aff4b84b3f52b9c7148392e27b0d4e79767f0","subject":"Merge \"versioning: Only list versions container if it seems to exist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/167aff4b84b3f52b9c7148392e27b0d4e79767f0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-07-27 03:58:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-07-27 03:58:28.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I7b9907645d7357824e197fa213f0ec68283ee72b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee8ae73cc72969b7affb3b1f7d7a66c92edcf84c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee8ae73cc72969b7affb3b1f7d7a66c92edcf84c"}]},"branch":"refs/heads/master"},"8a5baf8e1316f39b169b3cb60cd243994f335a3b":{"kind":"REWORK","_number":5,"created":"2023-07-28 04:12:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/46/889146/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/46/889146/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/889146/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/889146/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/889146/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/46/889146/5"}}},"commit":{"parents":[{"commit":"167aff4b84b3f52b9c7148392e27b0d4e79767f0","subject":"Merge \"versioning: Only list versions container if it seems to exist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/167aff4b84b3f52b9c7148392e27b0d4e79767f0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-07-28 04:11:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-07-28 04:11:58.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I7b9907645d7357824e197fa213f0ec68283ee72b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8a5baf8e1316f39b169b3cb60cd243994f335a3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8a5baf8e1316f39b169b3cb60cd243994f335a3b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
