)]}'
{"id":"openstack%2Fswift~889785","triplet_id":"openstack%2Fswift~master~Idb4020fdeee87a9164312dc9647ab0820b098ff8","project":"openstack/swift","branch":"master","topic":"fix-manifest-test","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-08-18 15:38:17.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-08-18 14:44:20.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2023-08-18 15:38:17.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-08-18 15:38:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idb4020fdeee87a9164312dc9647ab0820b098ff8","subject":"tests: fix FakeSwift HEAD with query param","status":"MERGED","created":"2023-07-27 01:06:36.000000000","updated":"2023-08-18 18:56:41.000000000","submitted":"2023-08-18 15:38:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":169,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"889785-fix-manifest-test","meta_rev_id":"d80919990f3a62c1686c949fde260a40d57a1b79","_number":889785,"virtual_id_number":889785,"owner":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-18 15:38:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2023-08-18 13:57:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2023-08-18 14:44:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":1,"date":"2023-08-18 14:44:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-27 01:25:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-27 03:06:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-27 15:59:13.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2023-08-04 15:50:31.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2023-08-07 19:21:11.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"c071f9c373d8b10a86d57a21569e5463a90f2da8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-07-27 01:06:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc293cbbc19141ed32934171158d6e618ceb6f50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 01:25:38.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90da7f1b1d564e0d842de1f8ff2bb11d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6edc05b03cec4fb79cdf1f90fd3bc6b8 : FAILURE in 14m 34s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/862049191f3b4123bb864cb216f06836 : FAILURE in 14m 53s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b999be6a81e24775adce8cf74e15e94c : SUCCESS in 17m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/16183f93912f4622b94200c2346a7efa : SUCCESS in 17m 45s","accounts_in_message":[],"_revision_number":1},{"id":"2d56bf02573d9fba8d9c2e04bfedad93cbc705a0","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-07-27 01:37:34.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"72ac50773fe4301c69ddfe2e24627970f6994311","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-07-27 01:39:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"220de696adefaa2097c3c5e64c7a13d2928f156d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-07-27 01:49:41.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"9b94fec7202675cd9168023276d2292d5d4e2bf2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 03:04:30.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc6403cc5e8f48dbaf50510e299a9170\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8172fd8e620e417c955ad8831fb7f7c6 : FAILURE in 14m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a23a912bfa324d69a1a3964b300672e9 : FAILURE in 14m 22s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/211d817a90334c03a1793aa96d2436b4 : SUCCESS in 1h 13m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9c587ae7efbe4300aa250e536d4c14a5 : SUCCESS in 1h 07m 07s","accounts_in_message":[],"_revision_number":3},{"id":"e3bfcae9a5b383b0c329b96335d21308f6d018a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 03:06:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/93117121aec5427a86c5f85c4dcd15b5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07f8624ba1704f879441aede512bb015 : SUCCESS in 9m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7cfc29ef3bdf4bf9bc54fa277297556a : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1ae27a7b5509473f9fa7d3cba3d63605 : SUCCESS in 18m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8ebe2207dd1f4dcbb3d52dc69c3f4170 : SUCCESS in 20m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9f0114a2b9924b4fbf054b4fe42ceba3 : SUCCESS in 7m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/23062f6c3e274891b5c60f5709992ba3 : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/938d3eb3c3de448d8dcd13609ac8052a : FAILURE in 16m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b6c8e1bc197649c89a0d72a35a06a1cb : FAILURE in 17m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/19fd9f6bff014055a6d46f5423d81805 : FAILURE in 21m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e4e8f03640194ed383d294ec6a158cde : FAILURE in 21m 09s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/67164b9e0787434bb46aa6fb9241d504 : FAILURE in 14m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8140aa80654140a29f4a869fd108d15a : SUCCESS in 23m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7a3a336909724dfe81e97b43699fdeb0 : SUCCESS in 18m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/43cb940405844227bf277880bef9360c : SUCCESS in 25m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a87a8cab4ac547b887363e523405baf7 : SUCCESS in 11m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7a57a7e4df124d67a925cce99c97a6d0 : SUCCESS in 17m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/290e6cdcf61e4db897f7ffc845344504 : SUCCESS in 15m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2bd49061adaa4ac1b3d7ac6bfbd861ea : SUCCESS in 58m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4d09b2ec45cf4a57b7f1b8057b162cb0 : SUCCESS in 58m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9fa4947722ed442ba6cad6c5db00a443 : FAILURE in 20m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e505be355bd40d6bea7a12ac1f0e845 : SUCCESS in 5m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/78e01b1563264b508b2402958e70bb4f : SUCCESS in 34m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9e005fc15a1922737eb0a63f1ad4aee3b89c5076","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-07-27 15:59:13.000000000","message":"Patch Set 3: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":3},{"id":"7e3a7060f16473a62392a11b0155324da50cbc91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-07-28 18:54:17.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"03073df645b3f1696d8e753a48264349646fc346","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-07-28 19:06:11.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"0f616ec25a74c13e9789ed2d6a3830eabe517469","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 20:06:01.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e593b56090864451a18c324e7e9e7c7f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/952b9d70e800423d9092cc63f9f43601 : SUCCESS in 10m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c203f87e62684b2d9ca47906ca6980bc : SUCCESS in 18m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2204ea1a1043418287596c9b52a37477 : SUCCESS in 34m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9c188ef12c524f78ae9ad7cb178f4b76 : SUCCESS in 18m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/35e54af79390463d9c3c1844c8abe755 : SUCCESS in 9m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5590a0034b56489f9a60d587d53b4165 : SUCCESS in 5m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/25419750147745498526a38899d15168 : FAILURE in 33m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e4ef991e6cf84e288c8a6e681700a2e0 : FAILURE in 17m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d79ff92f427f421292e9ba56f47ef2a0 : FAILURE in 13m 23s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3a933d73fed34b32afc3817d86cbc92f : FAILURE in 17m 12s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/76c332c589fd44a7a9fd9ebe16ff648e : FAILURE in 16m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0c60edc5107846adb9e97d1addf8f5d6 : SUCCESS in 16m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bfc13b005257455694891729b580756d : SUCCESS in 15m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/982ca25510cf45ec9597d85083b42aa8 : SUCCESS in 17m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/05a90e88121440c8bb85454a0c34e7c8 : SUCCESS in 16m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/629f430bd22d4cea9878d6496d02c8fa : SUCCESS in 31m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d6a142a5b1654b599a5ea75d66da2387 : SUCCESS in 15m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4008132b81dd440eb8cf3631d5e71ea7 : SUCCESS in 51m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d4389674688e43b88c14f4e5d0864eba : SUCCESS in 53m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/153f29396de7488abbe631cd0b5a1347 : FAILURE in 18m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b82087d13a44349a148e4601b73e15e : SUCCESS in 5m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/91e476cb80f245a58347f7272e9175c2 : SUCCESS in 22m 15s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"28d7d76954a599bf034e20097e0a669f3037bbc7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 20:23:05.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ae10793e54349b18be74489dbb16b05\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0018bd89771641d7b471bf8148c1262d : FAILURE in 13m 40s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8431cd6b0867409cbca4f080994ef889 : FAILURE in 13m 34s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3b848ef1f51147529f59a49766cc8a9d : SUCCESS in 15m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a16caa70c4144c059caab752fd821ccf : SUCCESS in 1h 15m 14s","accounts_in_message":[],"_revision_number":5},{"id":"4bf1291e34b0b85390bdd27f08ecb5906a01f73f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-07-31 19:29:04.000000000","message":"Patch Set 5: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":5},{"id":"7333f99ff13eaf0f536a7d50361af667c1c0183a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-01 17:43:24.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"899ed91cab49945d614b5e510721485f64eb572c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 18:58:55.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c6feb7ae60a406c82eac8b39d38fd69\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80907eef2d734d9eaab013e0ccec493a : SUCCESS in 8m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/d5ffca8cc9dd44229c010f5356e70cdb : SUCCESS in 58m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f7c519688f334ac88862ad85d62574a6 : SUCCESS in 35m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f4a72af84b944ebdae62b76ae55428f3 : SUCCESS in 48m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9403485cf2c9470ea5a5118e335bb2b1 : SUCCESS in 31m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f20877ecec8c47458de37e96b5e1876a : SUCCESS in 38m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/805bf1fbcdef4e769c641f37503b807c : SUCCESS in 17m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/54c092bf7120400db79cb63a1ea4e64d : SUCCESS in 8m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e0ed2fcca9234065bfe5ec4d032ac5e0 : SUCCESS in 5m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/495c27acf8364a4aaf165c0561282a09 : SUCCESS in 33m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1e1df1e74611434b925687229fd78156 : SUCCESS in 11m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d8d1f18dfb06447bb6fedb1797929b59 : SUCCESS in 13m 30s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b3411dbc8a9b44aa92ba9291be7cb786 : SUCCESS in 29m 27s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8cd50c1d709a40a59e72f654e5fb227b : SUCCESS in 15m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8ad10bceeefc4ff1b9cd1a5e0157f4a0 : SUCCESS in 16m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c6ea47803472424cb4e172a9a4f2b1ac : SUCCESS in 16m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/01c5a97698334c0ab3f7e227948be7d3 : SUCCESS in 15m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/188da5701df146b0940c0e09318e4927 : SUCCESS in 32m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a037e9342dae48c49c0ede0c476da5d4 : SUCCESS in 15m 16s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ef691e6546d64ce5829a72d22565cac7 : SUCCESS in 16m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fa18db22c3024085822d1efae290cb13 : SUCCESS in 7m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bf802893e9564b899bb4225e717fe249 : SUCCESS in 14m 26s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5bf0bcb8ae984086a5796dc353961d8f : SUCCESS in 7m 46s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/eebfb6a5923544268b067ebddb17b745 : SUCCESS in 54m 19s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7d0ce9eef8a2431b848e6e1e705ea028 : SUCCESS in 1h 08m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a212e5ae64ea40be84d68a6934e6d63f : SUCCESS in 53m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/922b8bd78b0d4aeba03a46e6c7d6aa8c : SUCCESS in 48m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e700c07c27d4984af7238f17478ad51 : SUCCESS in 17m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d374d6b4c89b407f83cd385863171130 : SUCCESS in 4m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/922cfefdb1ce443fbdef288b1c14dc93 : SUCCESS in 21m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe9411193b1d455ea728847ea80420db : SUCCESS in 36m 26s","accounts_in_message":[],"_revision_number":6},{"id":"625e02c1cd58c3cb34311c15b2e38a1f3d60dc3a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 19:03:41.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2938fbfc6d2743e792918a5c251f0b84\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/abf74954cde84e898e0fdab6ca191079 : SUCCESS in 13m 41s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0a358d7482824daab14d127fcc736d31 : SUCCESS in 50m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b651281bc5a446e5bc14753733d61d28 : SUCCESS in 1h 18m 37s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a2bef6f3405944a081dcf4a1a84c97da : SUCCESS in 53m 24s","accounts_in_message":[],"_revision_number":6},{"id":"52e4824e676f43109380004ab57f4e27c29c00ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-01 20:15:16.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"3b4ce218986e4a1ccd5227831798b458a984b850","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 21:06:41.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a835d792c7804bbc8f12177de2289d16\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d0d569c0e3e4a5a86718e701429a510 : SUCCESS in 8m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ab62777fd8754c21acd048e6f51b4b40 : SUCCESS in 20m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b5dd04c47a514425a21666b2a3437862 : SUCCESS in 18m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0ba3f68736cd4fbeb4ef645821f32a06 : SUCCESS in 16m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2bc954686f894d33a20b3805da9d7d1c : SUCCESS in 8m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7759213c685c45f4af2f207ad120599a : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5f2b72a509cd4aa9b4a240e8b1998a1c : SUCCESS in 14m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3acc814ab6834d82aa04c5f5b57e9629 : SUCCESS in 29m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5ed3e775aff5421cbc34af8acef8acb3 : SUCCESS in 20m 27s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/93a18d524fd8442bb72e6089af0b25d3 : SUCCESS in 31m 39s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/4c86af9446e549fb83ec211cf3489e82 : SUCCESS in 17m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/55f66d497331484db128a7bf5925f3cf : SUCCESS in 18m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/343726ccf29d484bbab775b3e5f6e5d3 : SUCCESS in 16m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/388bd57312874909a70225e224c478b0 : SUCCESS in 33m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/acb5fb4d424c4180bb8f157673da0a4a : SUCCESS in 12m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/50e75d126d15412ea9db211c29820e8a : SUCCESS in 15m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8bfe2756acdd4e37b6322a3770dafeba : SUCCESS in 28m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7f6c1ff877b64bc49743025e72a1d884 : SUCCESS in 47m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b8a1e1ce177f4e018265bd3e0c4ed078 : SUCCESS in 46m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3f2917763a342969a1ad486afa88e36 : SUCCESS in 16m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e8d997f84114a12b0bbf550bf4217e6 : SUCCESS in 5m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9d66e734c90e439d8ebf3c2397aee0d9 : SUCCESS in 24m 45s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ada0456aaba40cf7f00bf8857bd25dcd98486c5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-01 21:22:03.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"3b2df76e4687edc0e33bd3bbd9d01547a80ac0b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-01 21:28:53.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"f92931010a0859739cd62f30b96e5d73caa99934","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 22:14:51.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a493918d810741aba23e79cfbe1ec727\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45d721f6ee554b34b7e7fb4bc411e39d : SUCCESS in 10m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d358bbfcf093438996411b0df97902ef : SUCCESS in 24m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/40d60cc09ebd484e9528020438b7ce65 : SUCCESS in 18m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/10edbaf0476949789762d7aec4fcf90d : SUCCESS in 15m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/14441e9505b94b7e80ee1578e62cbe52 : SUCCESS in 8m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e5a295d264e34c33b65cf5585bf9a2e4 : SUCCESS in 5m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3035065a17c64a5d9743ed47ba40dd24 : SUCCESS in 15m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ee3965f048f147a6b767ae683a139a9a : SUCCESS in 12m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8b27f3c334df4d62a488c80358b47cac : SUCCESS in 15m 46s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/91f0e5e06a524e3cbf2c18c59ad68ae3 : SUCCESS in 15m 45s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/057abada292746bebeffbf59d7214e5f : SUCCESS in 16m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8b6d2eb6451144eda5b240cfddcc3908 : SUCCESS in 15m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/82a8376c955f4c8c9036bb5702f4af29 : SUCCESS in 17m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0c8ef67fef0f410ba11be32202bd943a : SUCCESS in 18m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/12571cad87c7400e934bfcac76d7fa5e : SUCCESS in 12m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c76daa1d3ab243d29a6b61b2aa1765af : SUCCESS in 12m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6b026b24e9014e519cf95244b4740e42 : SUCCESS in 16m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f1064aa722e143928598f15f6702880e : SUCCESS in 45m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/413c380191514aae93fa4cfb08378f7e : SUCCESS in 36m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25a4052084fe49b984e2e8cd636ba92c : SUCCESS in 16m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8534bf43cf6f4c3eb53eeae73e496da3 : SUCCESS in 5m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c852f9a6c241470882bf78a135978dcf : SUCCESS in 17m 53s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"a538eb68f43d361e1f0ffd44eac268c29fc6e57d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 23:01:45.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5b89cd247f9478388157ce8b4daa682\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/def09eab26c545909197ff0a4880fd8c : TIMED_OUT in 1h 31m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/247905dfd55c4d31a2587eb84f9ac8a3 : TIMED_OUT in 1h 30m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/26f671fc83254daf9c6bfb9cf911bd03 : TIMED_OUT in 1h 30m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6d43c58dd2d949328bf5776948665560 : SUCCESS in 22m 44s","accounts_in_message":[],"_revision_number":9},{"id":"d1481c0267d8a1b741259bcc0949ab5a6c01ff9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-02 21:31:30.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"5bef8fed663e8b04f603e89f1db0c5f0323e96b6","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-08-02 21:53:53.000000000","message":"Patch Set 9: Code-Review+1\n\n(12 comments)","accounts_in_message":[],"_revision_number":9},{"id":"6aecad845257425abffb379fa98da7b5bcf2ca3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 21:57:05.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ec1af3e894f4c15b153934592e09e72\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c045d79a06f4c3fb2ca11b7ce029b16 : SUCCESS in 24m 15s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eea4392ad4b64e36b1396febc0aae8cd : SUCCESS in 20m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2a56dd968b5e4c968826ddc0182146a5 : SUCCESS in 22m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fa1dc9fbb86947b88581b8d765e9ebe4 : SUCCESS in 24m 21s","accounts_in_message":[],"_revision_number":10},{"id":"0592f9aab2ade5f983e3c5f61c70faf5b2702afe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 22:17:20.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e556e7cd0b2a4226b87d990e07d2cf55\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b65e5f7b22324a2abc07d53ae6aae21d : SUCCESS in 9m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bc0d985c81504fcf8c6548f298172a47 : SUCCESS in 24m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7d4a1c286508467eb62c4f9f18f4f129 : SUCCESS in 24m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1a8a94e1fc794922ac5927f53af15aad : SUCCESS in 19m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/90ce20771ad04184816e14a9d1b40309 : SUCCESS in 8m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/aba7f8724dde415ca715ba2215bd7140 : SUCCESS in 5m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/30485c16ffd24e8e892276ed0c8ea3e5 : SUCCESS in 24m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7bad8b68cae34b0db8fa40c8bed5db32 : SUCCESS in 23m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e5f00d6cd35a430d86072d8fe116722f : SUCCESS in 15m 27s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b1e5a74b9a524e46bd2668c1eb1558d6 : SUCCESS in 15m 40s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b1ebc15f70cb455a8e03c78ef08d5226 : SUCCESS in 16m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cfddcf78e94d41078dcfd8ff06acddcd : SUCCESS in 17m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/26046c0d6f12491a8d9123b2ba2960b8 : SUCCESS in 13m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1ab09530294b409db134c8809c3552d5 : SUCCESS in 24m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/02dd696a6d6e42d385c70abb59bba541 : SUCCESS in 13m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c4037254fd8e4cccadd81d80703d222c : SUCCESS in 17m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6099680ee91846459f4fba80efe83575 : SUCCESS in 15m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3666b1248c9d421ca9d8dbb8fb7a91df : SUCCESS in 44m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0221dc45b0b0442eb1649e5dd5afe126 : SUCCESS in 41m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9da9ee5831f400683067ba13ec96a6f : SUCCESS in 18m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b392438a96f4752ba5f3e358b60e96e : SUCCESS in 4m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/03415fcb1d074d90afdcdf46b388ceee : SUCCESS in 26m 01s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"949f4a47bb9d0fe2f50c1b862e3bfbf967b10b0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-02 22:33:58.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"91c3ca202f150f3ca9dbc1c50d0a53506024638c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-02 22:35:06.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"e2f1d5f22292c65cc062e4a0bbbaeaa1dc485ab6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 23:22:24.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4be613327834d1a9cfd3411e006bf41\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/563c28628513425eb3660ee2ca990365 : SUCCESS in 8m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3d01dd27e76e488d9ed73a2316cfb57a : SUCCESS in 18m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/31fb91af96e541dea1c5d6e6d9ec7ef8 : SUCCESS in 17m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5dceb427367c415ab0c3a143c30e5985 : SUCCESS in 22m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/04b5d79d68764d558d28e8748d991342 : SUCCESS in 7m 43s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/748bf7595a054271950381bb56e28e22 : SUCCESS in 5m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/15a06fa87a824a58b78e89721491eb0d : SUCCESS in 20m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/791416a875c248c7863fc844aafabcd6 : SUCCESS in 18m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e900be91186c4a48a97c3c3d482000f1 : SUCCESS in 18m 37s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b7ab23bca1804b7e9d4561640dacefc8 : SUCCESS in 19m 09s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/979d0f2f1e564079a1a000d8a0dc1e40 : SUCCESS in 16m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5c886e2cccf44bf6aac3cb9b1579b31b : SUCCESS in 15m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fb036a2dc574433d8510ddb2631971d2 : SUCCESS in 13m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/85f6a054441746f7842cc5513656a023 : SUCCESS in 16m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c4eb20af3d9d4571b5a226ddb5d8baed : SUCCESS in 11m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f586c10deb104cca9d89d27ba8858d55 : SUCCESS in 14m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/953f0a05682a47d397700a1de85c99b2 : SUCCESS in 15m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9ae578fb14c44fe591e0b319176bc760 : SUCCESS in 45m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d455da121f374dc4bd8495438c0a6b7c : SUCCESS in 32m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e735a3d3cde495bb7b26134b04aa07a : SUCCESS in 17m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8378b3ff4a9240069a1e7564e83a7a89 : SUCCESS in 4m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8a571e6ab5b24dd38c62291055d24455 : SUCCESS in 26m 42s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"728b3e435374c12a4bfde74841bfc4997947917f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 00:05:34.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9bd397a86364ff3ad8a8a63fbb7849c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/84e2713f504f4f948ec715066df918ff : SUCCESS in 1h 28m 51s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/885c6689734f4cd38748bf6060d2e915 : SUCCESS in 14m 52s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4405522134604f31853a0a38e9bc0e7d : SUCCESS in 1h 00m 10s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b0c05335486b45c3856d64079e921e9b : SUCCESS in 16m 49s","accounts_in_message":[],"_revision_number":12},{"id":"e6640278ca2ac8efcbd9192fecaee927bdc87bc3","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-04 15:50:31.000000000","message":"Patch Set 12: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":12},{"id":"52e0a6fecdc42eadf1ff8fea8a38def7d423a05c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-07 16:17:10.000000000","message":"Patch Set 13: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"edffb8325fa467068edf76b7eabe8b298d16efbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-07 16:17:21.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"15dabaf043000db8b8470caa3f77c5c934ed11f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-07 16:29:02.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"cc6c44008f7b43d4dafc3d46bf617d48a89a6aeb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-07 16:59:19.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"9badf99c10acf4b927ef08514997d794b91b0e92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 18:04:50.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52890e49e7924df4a91e91bccaa32d6e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2a32818747f4ffb82bf3b357c881569 : SUCCESS in 9m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f38f7c4edde84ecf86635b51ec374f9f : SUCCESS in 19m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b1d250ad58fe4837b3213df7e6647e12 : SUCCESS in 20m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c2d5f631cee54c389a378f8309aa7219 : SUCCESS in 19m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/13961b2d11ae403ea557872e511acc0e : SUCCESS in 9m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5d0198de8fa241a6bea651b1a11f29b5 : SUCCESS in 6m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0b317a2875ec48d9bf18a2290cbe0eea : SUCCESS in 21m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6283f38a5ece40ca9f1fd4bd352d00bf : SUCCESS in 22m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/30fa59c1aa49449fae30269a0d004331 : SUCCESS in 14m 30s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/680bf1c373834577ae1637c72e1fab72 : SUCCESS in 19m 44s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0f935b2b958b4d27b96cf0a47d6a63eb : SUCCESS in 15m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a552ded869604e619c6010db8323af88 : SUCCESS in 20m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b3a016ebb1ab45cb8d91779c3e9ed2bf : SUCCESS in 18m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f1c9008e9e9c49d5b86c7f06f8824f3a : SUCCESS in 17m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7d425c9584e84b3a93223e30b3f6439c : SUCCESS in 15m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e6c71dc4a2f947f88001823c0c56b430 : SUCCESS in 14m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bc71365d590f47299efdf5c815e05a60 : SUCCESS in 19m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/22e3f3e3d86142b7866f6837bf9e1525 : SUCCESS in 45m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ff829bdbc82246d9a151eb1d79b5b2b5 : SUCCESS in 57m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e84ef513aa64b65bb10a51129c4b6cf : SUCCESS in 21m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d919f2832f4847fe8f601af6dd934aa0 : SUCCESS in 5m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/65788693ca4046a6a9d2ce86768c35e8 : SUCCESS in 24m 38s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"19c7b4cf21040925ae6bad34107c2a12ab3308d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 18:32:02.000000000","message":"Patch Set 16:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64c002ad5ba846ada1afd9dd5e71ae75\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/52af13888e4c43249d9ff03e25946789 : SUCCESS in 13m 17s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9b5e2eec5531497eaba5d993239972a9 : TIMED_OUT in 1h 30m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3a69fba2ef0942558871df52b1dc9c60 : SUCCESS in 18m 36s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/14b70c6c2b63427d9d34319979d880f6 : TIMED_OUT in 1h 30m 50s","accounts_in_message":[],"_revision_number":16},{"id":"2a25d8a41d8d48b1376d4dfa1402f36ee902778e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-07 19:21:11.000000000","message":"Patch Set 16:\n\n(9 comments)","accounts_in_message":[],"_revision_number":16},{"id":"12cebf0bc1258aeaecb62bb6db3fc00634272169","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-09 17:15:10.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"130c4b889151a8de14ead2ba60c90f280c290d2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 17:37:11.000000000","message":"Patch Set 17:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4f1dcf0e177465f86b26c4bf8f77e18\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6bc1baa08b6c4ed28e133325e4641f0d : FAILURE in 6m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e0a024e923414ea489c540cef8d8a55f : FAILURE in 7m 32s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d7d6f81b60cb47a4b582d69a494e5cf1 : SUCCESS in 20m 58s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2255f4e0f1884f12944fbcc0317107ff : SUCCESS in 19m 57s","accounts_in_message":[],"_revision_number":17},{"id":"81a0d9c4167de042294abad526143a95c528f95a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 18:15:06.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b232c92f2cb480f8a009813dbe50aa7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c74e81369abd47538521672bbff63da3 : SUCCESS in 8m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1cd7239a382e48869bd6923aaa1dda38 : SUCCESS in 17m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/228efb67d22c453ab2b31f5b1a79cb7c : SUCCESS in 26m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ace0746820d4214a6dd33db28d7ee87 : SUCCESS in 13m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/928c1dbb6a0449348808e573d047012b : SUCCESS in 9m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e09d9878bb0e431eaf1d7d9683f6cc68 : SUCCESS in 6m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7c5caae0360c420a9251f12ce4121ed8 : FAILURE in 7m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0a0dd9a31ad24c97a420594ba4edaaa9 : FAILURE in 6m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c355b59ba1e64270b87b5a7b2d877c48 : FAILURE in 8m 24s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4f12d39767894c488df40bf8a39944e8 : FAILURE in 6m 59s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f82dc12d0c72470790d5a606e3c169cc : FAILURE in 8m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6d0446d9a4124434a9f4d573cbd0d0ff : SUCCESS in 16m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0950cca020ba4c4190e5aab944da94ba : SUCCESS in 18m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b4447c0c33414662a38b023d956a0032 : SUCCESS in 18m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/51404b7e775944d7b65695e623a542a7 : SUCCESS in 14m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/58ce308c3bc54bbeab7422701a0fedff : SUCCESS in 13m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ed581a8168d64976b990d1d56238f3a8 : SUCCESS in 14m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7d52aee27b2c436da394e7f50f0bbcfc : SUCCESS in 48m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/499382db51274bbba7ab743af208dfd6 : SUCCESS in 52m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4e011ece8c84656a45e89eaf01536ac : FAILURE in 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19be32a560be45e08bb2558f93472d64 : SUCCESS in 5m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fde631f12fcf40f09a85eb1f49e389b2 : SUCCESS in 19m 29s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"56a9b706514cf4ee90f69839c43d301bbfac03a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-09 19:55:57.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"8c058c43f8ed97e6e23c77e357bba79840dbfc6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-09 19:56:08.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.","accounts_in_message":[],"_revision_number":19},{"id":"fd5d334d6632ac8c9d414398dc6d63914609f023","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 20:51:49.000000000","message":"Patch Set 19:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00ccc64b15234e2d8da2e797aa81c347\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b35da1bd2744498c8bc789b2e9ab9474 : FAILURE in 5m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b626622984cb4034a1260653df315ac3 : FAILURE in 6m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2b5b6f4bb8c94cb884ae525475416021 : SUCCESS in 17m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a59b21ffbf0c48ebbb97d5b4d5e7ce8b : SUCCESS in 49m 20s","accounts_in_message":[],"_revision_number":19},{"id":"f21c640487f1a0ee4108ce3f3c496be6247f186d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 21:05:45.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0712d2c2cae9498db5c14209dffffb08\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a117f729b454964ac4f60f660247f0d : SUCCESS in 8m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f89af4a144a34e8d95bb9ac89c28307e : SUCCESS in 18m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/64f4c4c4fbaf457c9c334ffbbf6221c1 : SUCCESS in 27m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/116b08cdac014adc879c45550039a901 : SUCCESS in 26m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cfda9d6f14604effac3b29702b199af7 : SUCCESS in 7m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a3fd9834be944604b313f9822ddc7740 : SUCCESS in 5m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/681287222c3d4ecbb86f714b9c72fc70 : FAILURE in 7m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4bfec6b0eba14b2cad8fe5e0f245c527 : FAILURE in 6m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fffc5b02bd8644ca83dee69ccfbcd171 : FAILURE in 10m 06s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1e20a8e6d2e14c469a973d68ec2bd3f4 : FAILURE in 9m 23s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/525b6e80239742aba005a5395165530c : FAILURE in 18m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/87e8509ac7684349b1bb457e3a267f32 : SUCCESS in 14m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9b0bc56586f146aa9e60135ccb8d4703 : SUCCESS in 18m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d6a97f1ba42749949bf753770395761f : SUCCESS in 19m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/436e0aa7f8944426b9580d17453300dd : SUCCESS in 13m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/db99c52983b2493c88f257f1fe4dfbd4 : SUCCESS in 15m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ce4e2f66f3154d5da3514bbff5505521 : SUCCESS in 14m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/197d95ee4b9e42df94e3a681e68aca7f : SUCCESS in 57m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0bab51c4e1a744daa7dbd545a8c1ae5f : SUCCESS in 56m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6e19ec9d76545d7a9eebe4ef09d33f9 : FAILURE in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/988144254cb94305a249cabc9a824e4f : SUCCESS in 5m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/47ec6d1692c842ebb183943713945597 : SUCCESS in 22m 29s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"274049cb91139d05bd1166fad9a2a9c70be0d818","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-10 20:04:47.000000000","message":"Patch Set 20: Patch Set 19 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"5f0781ebeea18ab3c21e94e1770146cecf57bfa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 20:52:40.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9a30a6c9a67c4c06abc9c46e648bb7b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83cf3c185a2143cc9c95cc1804b41c03 : SUCCESS in 9m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/58b7f44f850c4eb4aaf996cfba577cd1 : SUCCESS in 22m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5257f7f3f2724f9dba281672b91677e3 : SUCCESS in 18m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/88cc00e51a46496ead5137e626655cc0 : SUCCESS in 19m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b3684d94a0ec42bf954cc30800062e9e : SUCCESS in 9m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9911e7fc281742ea97066382b6e69dd3 : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/751fdbe057cb4dc4a120097dfbc11daf : FAILURE in 6m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7a29e8be648644e585bf5b0c09feb501 : FAILURE in 10m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/470a5430e96e45f486d9bd8c7e8353a3 : FAILURE in 17m 53s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8963520228f74d1c9797d5b89fbcfa0e : FAILURE in 7m 32s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/150f55cf563f498ab0a298904e1fd7c5 : FAILURE in 7m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4235a713f888473b96531e7ae469e6d0 : SUCCESS in 14m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d72c9a677c514d7a82a06d387b63e153 : SUCCESS in 26m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ce5cdc0263a64b8f9fd94f0b6faccc3a : SUCCESS in 17m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/51b863ab03cd4636a4b778caaef814b1 : SUCCESS in 14m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/335beb93be4a4349b8f1fdb39be3e18f : SUCCESS in 12m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bab612b8a3204b96a3a61af0ec83276f : SUCCESS in 15m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/181d679636044ffaa6c7cf641a3116a9 : SUCCESS in 35m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ab1f236e73ae4550b000e4c86b78590b : SUCCESS in 46m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a28e2ee6fd247fb82761301d88908c9 : FAILURE in 8m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/791bcca56ae949348f98890563fdc81a : SUCCESS in 5m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/64c72e86bfcc4f21a5c5b02845c2a9e0 : SUCCESS in 21m 37s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"169488e91bffff5f3512468cc64424297df3b904","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 21:20:04.000000000","message":"Patch Set 20:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c759c92b2f6e48988bb2a708831b2d36\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dce20a7a0b494c708e69823188c33c1c : FAILURE in 7m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/67682a5c76d4441ea8d7f18b5e20e1af : FAILURE in 15m 14s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/647960e24f36445d8bf8a29e6b1a2ac8 : SUCCESS in 20m 37s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/46fb6016a63941b3b471f190409bb662 : SUCCESS in 1h 13m 31s","accounts_in_message":[],"_revision_number":20},{"id":"8496fd3653ddee61411ffe8216bd7048a3302c56","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-08-11 23:06:47.000000000","message":"Patch Set 20: Code-Review-1 Workflow-1\n\n(29 comments)","accounts_in_message":[],"_revision_number":20},{"id":"4d920da95bcdee604e34ec1fece1f887dfe03148","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-14 21:36:44.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"158fafed1d67394cbcfb89f2f34fc39c3c2b838f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-08-16 02:04:22.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":21},{"id":"f4078bb14ab95cd98bce6552a82d7e6c1ac84469","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 03:45:58.000000000","message":"Patch Set 21:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f651f76f14a84bf1a235c2ebb0a4a53b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e2748dda62244f3592d724dbfed11d0a : TIMED_OUT in 1h 30m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/688bb6a48cd4455aa0e9735aa7b557c8 : SUCCESS in 1h 29m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6c32540d61414a659ae22a916648ff4e : SUCCESS in 29m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a5b11ffcde7d4b7ebb19af8581ee94a4 : SUCCESS in 23m 35s","accounts_in_message":[],"_revision_number":21},{"id":"f8c9db3a2a692daac2ada1fd650f2d0f8e31add4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 04:07:51.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06b05dcc791f435cac037c7e1672417c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/267e4408b6e54d3aa399a7d82b43a3a9 : SUCCESS in 9m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/93523727e1774526947cadcd922501c4 : SUCCESS in 18m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4a9651e013bc42f99b519880804dfdc1 : SUCCESS in 19m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/786771b1047d44b483b949e4fe9e2d68 : SUCCESS in 18m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1e23c545d89844d6b31dbf0f0da701ee : SUCCESS in 9m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/35dc4f7a3059473da315a5cce8856178 : SUCCESS in 6m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b4b01fda7fd9487cbaccce4533fe0ce2 : SUCCESS in 20m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9f316cd6713d426d96d14ebda644daaa : SUCCESS in 13m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8ae9e38cbabd414ca19856ac3d00849b : SUCCESS in 15m 06s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a1280f544b114b2f83dea5df3432b63f : SUCCESS in 26m 15s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c4dca4faee6643f0a3b96f7706e8f976 : SUCCESS in 18m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4b4df7d81f96435cacb8648a82b6e6a9 : SUCCESS in 18m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/71ac00b944a345f29d13330cb5b2927b : SUCCESS in 19m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7e1248baf04f4f509efa1d94fd3ff3d2 : SUCCESS in 16m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fad9f559e6054d43afb65f6632537797 : SUCCESS in 24m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/abded9b6139445b182263c7bf2cc536e : SUCCESS in 16m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d57b0fbc13104de58ca7fe911d872789 : SUCCESS in 16m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cd5b0efa72284aac8d735094200102f2 : SUCCESS in 1h 18m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ca5887aa301046feb9e0d6148f608434 : SUCCESS in 1h 06m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3be00665644a4dd5aaab40fa87e31a53 : SUCCESS in 17m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86f17116cdb746d0adfdf54025f06368 : SUCCESS in 5m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/62b3c83425314d578108e128b8d3dba4 : SUCCESS in 24m 23s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"65b3b7e14d83de81ae0b09ede8fe7733f08f1e30","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-16 14:08:36.000000000","message":"Patch Set 21: Code-Review-1\n\n(16 comments)","accounts_in_message":[],"_revision_number":21},{"id":"8a34db7d2bc27f2d923b94b9bc5dcb8fdfef879a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-17 15:18:38.000000000","message":"Patch Set 22: Patch Set 21 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"4b65e3cfa85efd30d966dbc2cd3979aa83437bcc","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-17 15:33:25.000000000","message":"Patch Set 22:\n\n(34 comments)","accounts_in_message":[],"_revision_number":22},{"id":"ee4dff462e4f1cd176ea0502171b5663d1e20fa1","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-17 15:40:16.000000000","message":"Patch Set 22:\n\n(3 comments)","accounts_in_message":[],"_revision_number":22},{"id":"8d387f71f40a3cb38fd552584158cb9a3d10249e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-17 15:46:15.000000000","message":"Patch Set 22: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":22},{"id":"1f3d507fa73530e7d319edd8f441fa3f0eaddaeb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-17 15:47:24.000000000","message":"Patch Set 23: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":23},{"id":"a1e736453c6ce5efd682c1f63239fd9b35f84d06","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-17 15:50:26.000000000","message":"Patch Set 23:\n\n(5 comments)","accounts_in_message":[],"_revision_number":23},{"id":"6d126173b059488f74e9a4c1af7f3160e9379a6b","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-17 15:51:19.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"945f5a7ceffc5aff6ae31a71a11df35c3b5d4ce1","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-17 15:51:32.000000000","message":"Patch Set 23:\n\n(6 comments)","accounts_in_message":[],"_revision_number":23},{"id":"27eb180f2de11012dc7d26a71820c1fafef2a8ee","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-17 15:53:17.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"2910257755168db4d4eb59b2e75bba445761d57a","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-17 15:54:45.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"fdf66b151ce839f13a60eb377180a18dedbc8729","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-17 15:55:27.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"cb5585c52e400fb69d054339101532a25cdfcb2a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 17:58:32.000000000","message":"Patch Set 24:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a362fb4de27409695ffbe73904af0ad\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/74c9a00f272b4916b4fc480991b2e359 : FAILURE in 23m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/727c0d6c15ae45e4a8ea21c4d67528ca : FAILURE in 24m 08s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b11eb1c782894d288042897fb1747d8c : SUCCESS in 25m 48s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/34151cf70d754ae2abf406cb6bf5bf6b : SUCCESS in 1h 13m 44s","accounts_in_message":[],"_revision_number":24},{"id":"67d464560ac7530cfccb713d1a5851ba36ad9a20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 19:01:27.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/516ae8a7b0f64d0aadcad563c6fd9cb2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44c362a5caa34d9c8f68e32e584b0ef0 : SUCCESS in 8m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/67a8f5c2b40b473495283b5b16cf916e : SUCCESS in 18m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a6d980a48e1e489b9d655dd6ec2f4b8b : SUCCESS in 19m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9b4f82e0c59e4450bcc85c98282b7472 : SUCCESS in 22m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c0c2346afcbd43a288cd8e23520c0c14 : SUCCESS in 9m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/96b0ba3285fa47abba4f495863d6fad6 : SUCCESS in 6m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8548e4aa164648d690e76a081b234e12 : FAILURE in 19m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fe914e5a64434a00904e1d2e65f2eddb : FAILURE in 15m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/56d1c1d6eab0466d9526c7bc9fcb30d9 : FAILURE in 16m 11s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/85a1b3c5e19d4ac696161c85dd963648 : FAILURE in 25m 01s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b5b719abe6a9436e848c7c931fdb73d9 : FAILURE in 20m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2d3b0b5db2e14bee9cb0312eac884f58 : SUCCESS in 17m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/60e43eb407144cd0b289785c85efd5c1 : SUCCESS in 16m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/035aa556190f440a831c08b8ae182901 : SUCCESS in 16m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7149756fed914ca69b1f3261f5634126 : SUCCESS in 14m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/93c5fded20ab401d8adeeec19365a335 : SUCCESS in 13m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ab64b6fdb99d46d0abaaf0f6ded2a0c8 : SUCCESS in 15m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d5b185a79fd148d7b96a632d8844610f : SUCCESS in 43m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fb54bd65f3e44a3ea696f29f6c929e1f : SUCCESS in 50m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e2b26880a6c45819dbd8694035c6df1 : FAILURE in 20m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36f26b90bce24e649c20f779dc2faf58 : SUCCESS in 4m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/94756e543be34fea9a3606e9a5b69bbe : SUCCESS in 21m 48s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"925cf8555bb8ab6b2802c3e0bbda504fd610ef31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-17 22:35:52.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":25},{"id":"da85e0f897f2b62abf8b19eeae8caa1ce445a604","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 22:56:00.000000000","message":"Patch Set 25:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05edac3562b142f391e0ba54f6041709\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d54a3453b35d471ebba01f1ff9612a15 : FAILURE in 13m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aaf24bfebb264a0395112c32a0dad45f : FAILURE in 15m 16s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/189339c68b85487383528af7ff79d56c : SUCCESS in 18m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4614945d9703473d9651be1c591f85ef : SUCCESS in 19m 06s","accounts_in_message":[],"_revision_number":25},{"id":"77adf792b9bf252b33c016b1ab73df0619911ef8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 23:14:04.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad63bff322f0492baefcb113c3de2c7b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8c54c0cd1714e878cdffc2bdbd1678d : SUCCESS in 8m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1c20111810eb4c2aa535626f8b0c0c8e : SUCCESS in 18m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a02ca5f99fa24a6a98df4ff6cf5a0ccb : SUCCESS in 18m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f90ebaef93bc4a349ce74fb83c5a4dd8 : SUCCESS in 18m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/54d63aa9106c4259b89792e6aa1b36a0 : SUCCESS in 8m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/12bc1cbaf0d548c9ab72df9b1e236bdc : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3f3144ada9ea4e8c9ba4c50830a5aef8 : FAILURE in 12m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/96bcd28698a94ad19369c8ce3b93cd47 : FAILURE in 12m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b6dbdcb9d35547589ad929960174dd6e : FAILURE in 13m 15s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/45d3008f14df4f4294baf3bf0297d197 : FAILURE in 12m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/4f3f91a6f0ac410584b6b299d2bf6714 : FAILURE in 13m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a7017dc7905640f084ec5563bad1ec6f : SUCCESS in 12m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/af39ded92bde48c9b4d52f02c2d8a487 : SUCCESS in 14m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/08a6f1873fbb41cfadeeff8a3e3dc90d : SUCCESS in 16m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/176b344c94ec478ebf21f1ac39d8b801 : SUCCESS in 14m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/74328190fb114ed8a19178b88a8a2fff : SUCCESS in 12m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3ad4a9717b254f51a8c698ee719a779f : SUCCESS in 14m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fe2f88c1622f453d82fa987e0e694561 : SUCCESS in 35m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9279fde4b42a4551ac4cc6e26cf88da0 : SUCCESS in 37m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/435f7df038424d0ca2b00b8726cab72c : FAILURE in 16m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a63fd9a80da43a7bd32e0c44b01309c : SUCCESS in 4m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c0ca35452ee9462fb8981b891a54d8e5 : SUCCESS in 18m 41s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"0112396c316305ea5724de9c4de93c18e5cd12bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-17 23:22:32.000000000","message":"Uploaded patch set 26.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":26},{"id":"303a608f46064a25c21bbbc80f8a7e3a464d1acb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-17 23:23:44.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"dd3a69cbf99d854921c94ca03e6aba5b2c28ffd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-17 23:27:40.000000000","message":"Patch Set 28: Commit message was updated.","accounts_in_message":[],"_revision_number":28},{"id":"6d3666d20ce6b2dfad292ee9fec6f264a0a70e0a","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-17 23:34:59.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"97fab8d4ecb06c84ae85252086bcac10b8ef8be8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 23:46:59.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb4d8f68f6804fd5bd1ebd42f9635c08\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/206e1f24adda495281af9ba5ef2a9c56 : SUCCESS in 14m 10s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9c74c7387c7b4619ac7a2b6227dc6949 : SUCCESS in 13m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a78dec5c612a4c1aa8efda5213206063 : SUCCESS in 18m 21s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b2e2a6c4c63458cbd7c7f8aaf7a764f : SUCCESS in 16m 09s","accounts_in_message":[],"_revision_number":28},{"id":"0f6f214fe68c259fd16b0db830091542e1628cb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-17 23:51:02.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"a45b426586ed632da445b8701e7b3c2bbe45d13b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 00:11:36.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b80f9ca277084075a758a4494398fda6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/62d1bfae24a747d39577db14ad971ccb : SUCCESS in 14m 04s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ea32f250bf354299822775408f15c73d : SUCCESS in 14m 23s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c53a89ada82a410cbe82d09431db1453 : SUCCESS in 19m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0374626fef38462c8c69886a3d5c113b : SUCCESS in 17m 21s","accounts_in_message":[],"_revision_number":29},{"id":"8725f3bbcd89f679b02a0a85dfecbacb29d18925","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 00:52:29.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cb9c27d88a84bc0a2c2b53a97a96b38\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03a32613f76d48389eddaa619916da0a : SUCCESS in 7m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/18d3ecdcd82d48aaaea6b8cadd3a1f67 : SUCCESS in 18m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f49e4a8899e64a4ca421e96d7d5954a4 : SUCCESS in 19m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cfd5abd8d8404627b58ba61be987e1a3 : SUCCESS in 18m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5a9a06a14be5489fb5923ab77816421c : SUCCESS in 9m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d318dc7314854834a62652d48d56838c : SUCCESS in 4m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2920dbb05cc1497bba002b5b6a40734a : SUCCESS in 16m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8e311ff534a24e01a0eef69db730cd21 : SUCCESS in 17m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5cd40e3d567247bca092ef622f1baf45 : SUCCESS in 14m 52s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/11a9505451664efba808fe60f164bd1c : SUCCESS in 17m 35s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8239ecef6253456492fc11d98bb45ed3 : SUCCESS in 16m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9620f070f9ca4fd7b3b18c6e5315da2a : SUCCESS in 19m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bc058d9647c4495db5657be81f28095e : SUCCESS in 20m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e90ea152d5744e31816f70c8ef51452d : SUCCESS in 17m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1f96fcfd823344698241eb609abcd721 : SUCCESS in 17m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ec594da9485c4ce68c4267b14bb15a40 : SUCCESS in 14m 26s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cd579db2aad843b5b5a69d8c5ca2d4d4 : SUCCESS in 16m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5a7288b405fa40048115bc740fe02742 : SUCCESS in 51m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/463a7fdf84b946138c62c8f405ed76ce : SUCCESS in 55m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/71e8164856bc4f5fb6c1989866df1c29 : SUCCESS in 19m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ace8fcc019b43b1ae6c70c9b0329ada : SUCCESS in 5m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a95b82868426439e84db56f54f3805f8 : SUCCESS in 21m 01s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"7000b51a8605e7222c1aaa8c7a8820c43055b44b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-18 12:34:38.000000000","message":"Patch Set 29:\n\n(3 comments)","accounts_in_message":[],"_revision_number":29},{"id":"bf6d6c7069f266c55936d37ee7ab1912a1f9ff23","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-18 12:34:43.000000000","message":"Patch Set 29: Code-Review-1","accounts_in_message":[],"_revision_number":29},{"id":"d5e6e2b5ea73e293398f5dc2095e09215a454f06","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-18 12:37:01.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"cade6eeac02896758e9368280da7b07992fbed0c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-18 13:13:35.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"836ad36880e471a1b17ce9daaf2497e377a64610","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-18 13:48:39.000000000","message":"Uploaded patch set 30.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":30},{"id":"90b1fb1c26f56f482d5f5c1c7e573ff032c10028","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-18 13:52:12.000000000","message":"Patch Set 30:\n\n(4 comments)","accounts_in_message":[],"_revision_number":30},{"id":"f410a94043e6ba6f09117890c5c8b86d39bf2556","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-18 13:57:07.000000000","message":"Patch Set 30: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"11e1253a5eacc780d9b70180b1be9ed96b83b250","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-08-18 14:44:20.000000000","message":"Patch Set 30: Code-Review+2 Workflow+1\n\n(8 comments)","accounts_in_message":[],"_revision_number":30},{"id":"21f54f78a8788148aad6786b9422dc4bddccec77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 14:45:41.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7709d6432c234e28b4df3b2f91815c1a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d00f5f092ef49deaeccd3a20e39b177 : SUCCESS in 9m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/419140de2bb843feb752b52d283b7d24 : SUCCESS in 17m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c80dfaa4b8974fa48f62025a84d3fd32 : SUCCESS in 15m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3df7b4b5153d4ea5b8982b51767e4ff3 : SUCCESS in 20m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/145ea8475c18471fb4ab52705d2b643f : SUCCESS in 8m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f3dd4913db9d429bbc0f8ceec11978ef : SUCCESS in 5m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b156a1ec3b9549e78174746b5233db69 : SUCCESS in 16m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/853261ce97574dac834e7ea2cad8935c : SUCCESS in 18m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7c994ac81e804d4dae622683ebeff282 : SUCCESS in 15m 11s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ab72bde94ac942b4ab77e28265f9611b : SUCCESS in 22m 44s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1216d11c92884d15ba799bc42446b94c : SUCCESS in 24m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9e5621af3a694cf098ddb1b362b6e346 : SUCCESS in 16m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3f3c5907134a4201a9b6ceff8ae58769 : SUCCESS in 17m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/770db9f20cc241bfa083561f87094781 : SUCCESS in 17m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9ec3dcc7ca2040779ef6d5da355c2cb6 : SUCCESS in 13m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/226adf7e07b24c0389b2d0483400ebab : SUCCESS in 15m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2c1581b5b6b54a59bef23b65e0763131 : SUCCESS in 14m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aad2b25b37764abfabd6cec2c6bcf633 : SUCCESS in 51m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/601d76d2231e4f1084bee3ff042ee9bb : SUCCESS in 43m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da8fb51a35f94472bd340296e997a255 : SUCCESS in 16m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0ff1a649be44764a9ca3051ea5bfada : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b28f8e1166284e5eb85263cb6d65a9b4 : SUCCESS in 21m 26s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"8adecc7757846326c049a6a88395d0a68d4f770d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 14:46:06.000000000","message":"Patch Set 30: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":30},{"id":"b938b07219ea57d586d0303426d0111d6efe33d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 15:02:10.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0064c333b909475780853c3ce600cbaa\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/197602e1c4164b268516d9a5ac080060 : SUCCESS in 57m 12s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2a1b0ccec0db4706a4d119d1fa295b27 : SUCCESS in 1h 11m 25s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6edf733aeba745f4a59dd9105716bff2 : SUCCESS in 27m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b66195f652949158c9b0838c6b8b9e1 : SUCCESS in 27m 54s","accounts_in_message":[],"_revision_number":30},{"id":"0e09b9ebcf47feabc20f316065c745e69b6254a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 15:38:17.000000000","message":"Patch Set 30: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2be6763089e4030bc2dd709b25dcdbf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64100c792912463082e65f95f996ad96 : SUCCESS in 7m 44s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/7fae0c66e7e345089ceda2a2d6cde3e8 : SUCCESS in 11m 50s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/48ff4246bb8947b1b2f634a984a87335 : SUCCESS in 6m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e22f61c23a644f218f54c515e6fa0f24 : SUCCESS in 24m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/80ae0c51da4243d9a6927383297d2155 : SUCCESS in 17m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c2cc31f720ca461aa32dfef773b4cf9c : SUCCESS in 15m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/af5984cb023a45c0b4e3aa80e9bdf488 : SUCCESS in 25m 27s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0d3db08dee904ab7aff7ef00cf4ebaf8 : SUCCESS in 16m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/63f84c4526924d18bb47f2a67c1f60a3 : SUCCESS in 17m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9a5eea77002b4d24b3ae9afd1ea2e0fa : SUCCESS in 18m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c86c4bc234b94c94b5b94a338d4768b6 : SUCCESS in 16m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/34c0ecfafacb4a70a5906856d4de75d5 : SUCCESS in 17m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2115ded78b134d1f8a88cd81e954a96a : SUCCESS in 15m 01s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e597a8e4255641a9ae4a8adb5e7a51aa : SUCCESS in 16m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7595dcc3d9554834a04c3c588d664f18 : SUCCESS in 6m 44s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b827b4b1aa024f1a9da023cbe61a21e0 : SUCCESS in 7m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d13d84c6c0824734b067451053c5a3ba : SUCCESS in 47m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9c54782af7a546deae908535705ee1a7 : SUCCESS in 50m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5b4e094de6c4b3abf5530585348c038 : SUCCESS in 14m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7912dce8980148e5aed5e10fcc697dac : SUCCESS in 7m 01s","accounts_in_message":[],"_revision_number":30},{"id":"8ba840b86eba0e82065c31ed99702dd8295a16b2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 15:38:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":30},{"id":"7864aa83f36c67be065d1fdf68fd0a1e1722d665","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 15:39:19.000000000","message":"Patch Set 30:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a1381f8a6814da98f0fbe7a53857ec0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/045e9a56e05145a5afb76c0e98428b4d : SUCCESS in 43s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/cb8787f90d29476fa47ece81b2060850 : SUCCESS in 34s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/e1590888f64140b0be08aa3988bd01f4 : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"d80919990f3a62c1686c949fde260a40d57a1b79","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-18 18:56:41.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30}],"current_revision_number":30,"current_revision":"dab7192e1ed13ee8cf62970762cbb78edbd4ed17","revisions":{"28117d7598d714acec877ec11557fba4111eeb0a":{"kind":"REWORK","_number":1,"created":"2023-07-27 01:06:36.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/85/889785/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/889785/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/889785/1"}}},"commit":{"parents":[{"commit":"740636b8e42bf30fb020e40d492c981162686aef","subject":"Merge \"Add more unit tests for ranged GETs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/740636b8e42bf30fb020e40d492c981162686aef"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-24 17:45:21.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-27 01:06:11.000000000","tz":-420},"subject":"Improve coverage for FakeSwift tests","message":"Improve coverage for FakeSwift tests\n\nChange-Id: Idb4020fdeee87a9164312dc9647ab0820b098ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28117d7598d714acec877ec11557fba4111eeb0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28117d7598d714acec877ec11557fba4111eeb0a"}]},"branch":"refs/heads/master"},"c2a728a5a480939af18e207eca3bf39c4aa86132":{"kind":"REWORK","_number":2,"created":"2023-07-27 01:39:54.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/85/889785/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/889785/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/889785/2"}}},"commit":{"parents":[{"commit":"740636b8e42bf30fb020e40d492c981162686aef","subject":"Merge \"Add more unit tests for ranged GETs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/740636b8e42bf30fb020e40d492c981162686aef"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-24 17:45:21.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-27 01:39:46.000000000","tz":-420},"subject":"Improve coverage for FakeSwift tests","message":"Improve coverage for FakeSwift tests\n\nChange-Id: Idb4020fdeee87a9164312dc9647ab0820b098ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2a728a5a480939af18e207eca3bf39c4aa86132"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2a728a5a480939af18e207eca3bf39c4aa86132"}]},"branch":"refs/heads/master"},"b2fb5c90303d8376c3b2b36fe4d53aa3d76b6677":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-07-27 01:49:41.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/85/889785/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/889785/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/889785/3"}}},"commit":{"parents":[{"commit":"167aff4b84b3f52b9c7148392e27b0d4e79767f0","subject":"Merge \"versioning: Only list versions container if it seems to exist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/167aff4b84b3f52b9c7148392e27b0d4e79767f0"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-24 17:45:21.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-07-27 01:49:41.000000000","tz":0},"subject":"Improve coverage for FakeSwift tests","message":"Improve coverage for FakeSwift tests\n\nChange-Id: Idb4020fdeee87a9164312dc9647ab0820b098ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b2fb5c90303d8376c3b2b36fe4d53aa3d76b6677"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b2fb5c90303d8376c3b2b36fe4d53aa3d76b6677"}]},"branch":"refs/heads/master","description":"Rebase"},"23f392e780914905e16d8e3b42d2e074ef1b976f":{"kind":"REWORK","_number":4,"created":"2023-07-28 18:54:17.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/85/889785/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/889785/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/889785/4"}}},"commit":{"parents":[{"commit":"167aff4b84b3f52b9c7148392e27b0d4e79767f0","subject":"Merge \"versioning: Only list versions container if it seems to exist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/167aff4b84b3f52b9c7148392e27b0d4e79767f0"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-24 17:45:21.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-28 18:53:29.000000000","tz":0},"subject":"Improve coverage for FakeSwift tests","message":"Improve coverage for FakeSwift tests\n\nChange-Id: Idb4020fdeee87a9164312dc9647ab0820b098ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/23f392e780914905e16d8e3b42d2e074ef1b976f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/23f392e780914905e16d8e3b42d2e074ef1b976f"}]},"branch":"refs/heads/master"},"22ea7ccefba330d38b0dfde8b81e8adcc8b5bf52":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2023-07-28 19:06:11.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/85/889785/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/889785/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/889785/5"}}},"commit":{"parents":[{"commit":"167aff4b84b3f52b9c7148392e27b0d4e79767f0","subject":"Merge \"versioning: Only list versions container if it seems to exist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/167aff4b84b3f52b9c7148392e27b0d4e79767f0"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-24 17:45:21.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-07-28 19:06:11.000000000","tz":0},"subject":"Improve coverage for FakeSwift tests","message":"Improve coverage for FakeSwift tests\n\nWe have to improve our unit-test coverage in swift since we detected that we couldn\u0027t issue a HEAD request with a query param to FakeSwift app\n\nChange-Id: Idb4020fdeee87a9164312dc9647ab0820b098ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22ea7ccefba330d38b0dfde8b81e8adcc8b5bf52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22ea7ccefba330d38b0dfde8b81e8adcc8b5bf52"}]},"branch":"refs/heads/master","description":"Edit commit message"},"53193c2aee24cc904c658aca4992c73f38953b19":{"kind":"REWORK","_number":6,"created":"2023-08-01 17:43:24.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/85/889785/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/889785/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/889785/6"}}},"commit":{"parents":[{"commit":"167aff4b84b3f52b9c7148392e27b0d4e79767f0","subject":"Merge \"versioning: Only list versions container if it seems to exist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/167aff4b84b3f52b9c7148392e27b0d4e79767f0"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-24 17:45:21.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-08-01 17:43:16.000000000","tz":-420},"subject":"Improve coverage for FakeSwift tests","message":"Improve coverage for FakeSwift tests\n\nChange-Id: Idb4020fdeee87a9164312dc9647ab0820b098ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/53193c2aee24cc904c658aca4992c73f38953b19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/53193c2aee24cc904c658aca4992c73f38953b19"}]},"branch":"refs/heads/master"},"5f9b1165ce9d455562aa5b9fecb79f2fca3e9812":{"kind":"REWORK","_number":7,"created":"2023-08-01 20:15:16.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/85/889785/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/889785/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/889785/7"}}},"commit":{"parents":[{"commit":"167aff4b84b3f52b9c7148392e27b0d4e79767f0","subject":"Merge \"versioning: Only list versions container if it seems to exist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/167aff4b84b3f52b9c7148392e27b0d4e79767f0"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-24 17:45:21.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-08-01 20:15:09.000000000","tz":-420},"subject":"Improve coverage for FakeSwift tests","message":"Improve coverage for FakeSwift tests\n\nChange-Id: Idb4020fdeee87a9164312dc9647ab0820b098ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f9b1165ce9d455562aa5b9fecb79f2fca3e9812"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f9b1165ce9d455562aa5b9fecb79f2fca3e9812"}]},"branch":"refs/heads/master"},"a211db60d97bbe172e37d669a1e3b9508c9097e4":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-08-01 21:22:03.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/85/889785/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/889785/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/889785/8"}}},"commit":{"parents":[{"commit":"8a5baf8e1316f39b169b3cb60cd243994f335a3b","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8a5baf8e1316f39b169b3cb60cd243994f335a3b"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-24 17:45:21.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-08-01 21:22:03.000000000","tz":0},"subject":"Improve coverage for FakeSwift tests","message":"Improve coverage for FakeSwift tests\n\nChange-Id: Idb4020fdeee87a9164312dc9647ab0820b098ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a211db60d97bbe172e37d669a1e3b9508c9097e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a211db60d97bbe172e37d669a1e3b9508c9097e4"}]},"branch":"refs/heads/master","description":"Rebase"},"8f80a8471ba648b3dad716b7377c43d2440f9f13":{"kind":"REWORK","_number":9,"created":"2023-08-01 21:28:53.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/85/889785/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/889785/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/889785/9"}}},"commit":{"parents":[{"commit":"8a5baf8e1316f39b169b3cb60cd243994f335a3b","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8a5baf8e1316f39b169b3cb60cd243994f335a3b"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-24 17:45:21.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-08-01 21:28:44.000000000","tz":-420},"subject":"Improve coverage for FakeSwift tests","message":"Improve coverage for FakeSwift tests\n\nChange-Id: Idb4020fdeee87a9164312dc9647ab0820b098ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f80a8471ba648b3dad716b7377c43d2440f9f13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f80a8471ba648b3dad716b7377c43d2440f9f13"}]},"branch":"refs/heads/master"},"ba6bd30dfb3705889d8761ae2abeadaf74f0ad10":{"kind":"REWORK","_number":10,"created":"2023-08-02 21:31:30.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/85/889785/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/889785/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/889785/10"}}},"commit":{"parents":[{"commit":"8a5baf8e1316f39b169b3cb60cd243994f335a3b","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8a5baf8e1316f39b169b3cb60cd243994f335a3b"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-24 17:45:21.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-08-02 21:30:28.000000000","tz":-420},"subject":"Improve coverage for FakeSwift tests","message":"Improve coverage for FakeSwift tests\n\nChange-Id: Idb4020fdeee87a9164312dc9647ab0820b098ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba6bd30dfb3705889d8761ae2abeadaf74f0ad10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba6bd30dfb3705889d8761ae2abeadaf74f0ad10"}]},"branch":"refs/heads/master"},"f8faf3c21820c0d7dbc9f786c26ebf71074aa968":{"kind":"REWORK","_number":11,"created":"2023-08-02 22:33:58.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/85/889785/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/889785/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/889785/11"}}},"commit":{"parents":[{"commit":"8a5baf8e1316f39b169b3cb60cd243994f335a3b","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8a5baf8e1316f39b169b3cb60cd243994f335a3b"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-24 17:45:21.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-08-02 22:33:51.000000000","tz":-420},"subject":"Improve coverage for FakeSwift tests","message":"Improve coverage for FakeSwift tests\n\nChange-Id: Idb4020fdeee87a9164312dc9647ab0820b098ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8faf3c21820c0d7dbc9f786c26ebf71074aa968"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8faf3c21820c0d7dbc9f786c26ebf71074aa968"}]},"branch":"refs/heads/master"},"9eac2ab6530de3c6679287f8b85c6e97665a7e8e":{"kind":"REWORK","_number":12,"created":"2023-08-02 22:35:06.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/85/889785/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/889785/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/889785/12"}}},"commit":{"parents":[{"commit":"8a5baf8e1316f39b169b3cb60cd243994f335a3b","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8a5baf8e1316f39b169b3cb60cd243994f335a3b"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-24 17:45:21.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-08-02 22:35:01.000000000","tz":-420},"subject":"Improve coverage for FakeSwift tests","message":"Improve coverage for FakeSwift tests\n\nChange-Id: Idb4020fdeee87a9164312dc9647ab0820b098ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9eac2ab6530de3c6679287f8b85c6e97665a7e8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9eac2ab6530de3c6679287f8b85c6e97665a7e8e"}]},"branch":"refs/heads/master"},"c76474ec0ee505211ef954d6b598df7f3b1e5f85":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2023-08-07 16:17:10.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/85/889785/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/889785/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/889785/13"}}},"commit":{"parents":[{"commit":"8a5baf8e1316f39b169b3cb60cd243994f335a3b","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8a5baf8e1316f39b169b3cb60cd243994f335a3b"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-24 17:45:21.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-08-07 16:17:10.000000000","tz":0},"subject":"unit tests: improve FakeSwift handling of HEAD with query param","message":"unit tests: improve FakeSwift handling of HEAD with query param\n\nWe have to improve our unit-test coverage in swift since we detected that we couldn\u0027t issue a HEAD request with a partNum query param to our FakeSwift app without generating a KeyError since it wasn\u0027t included in the acceptable dict of responses.\n\nChange-Id: Idb4020fdeee87a9164312dc9647ab0820b098ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c76474ec0ee505211ef954d6b598df7f3b1e5f85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c76474ec0ee505211ef954d6b598df7f3b1e5f85"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a2a5c5c0aeaa9fb777e5d28ea2d052382932510b":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2023-08-07 16:17:21.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/85/889785/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/889785/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/889785/14"}}},"commit":{"parents":[{"commit":"e71102c5bdc65358e0ea7766b296967ce5bb9ec2","subject":"Merge \"docs: Clean up proxy logging docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e71102c5bdc65358e0ea7766b296967ce5bb9ec2"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-24 17:45:21.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-08-07 16:17:21.000000000","tz":0},"subject":"unit tests: improve FakeSwift handling of HEAD with query param","message":"unit tests: improve FakeSwift handling of HEAD with query param\n\nWe have to improve our unit-test coverage in swift since we detected that we couldn\u0027t issue a HEAD request with a partNum query param to our FakeSwift app without generating a KeyError since it wasn\u0027t included in the acceptable dict of responses.\n\nChange-Id: Idb4020fdeee87a9164312dc9647ab0820b098ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2a5c5c0aeaa9fb777e5d28ea2d052382932510b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2a5c5c0aeaa9fb777e5d28ea2d052382932510b"}]},"branch":"refs/heads/master","description":"Rebase"},"32d9a818348974c24344f62a8668af481586eced":{"kind":"REWORK","_number":15,"created":"2023-08-07 16:29:02.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/85/889785/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/889785/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/889785/15"}}},"commit":{"parents":[{"commit":"e71102c5bdc65358e0ea7766b296967ce5bb9ec2","subject":"Merge \"docs: Clean up proxy logging docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e71102c5bdc65358e0ea7766b296967ce5bb9ec2"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-24 17:45:21.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-08-07 16:28:25.000000000","tz":-420},"subject":"unit tests: improve FakeSwift handling of HEAD with query param","message":"unit tests: improve FakeSwift handling of HEAD with query param\n\nWe have to improve our unit-test coverage in swift since we detected that we couldn\u0027t issue a HEAD request with a partNum query param to our FakeSwift app without generating a KeyError since it wasn\u0027t included in the acceptable dict of responses.\n\nChange-Id: Idb4020fdeee87a9164312dc9647ab0820b098ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32d9a818348974c24344f62a8668af481586eced"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32d9a818348974c24344f62a8668af481586eced"}]},"branch":"refs/heads/master"},"48e331bb16e0e0db801aaa830cc72b83890eda55":{"kind":"REWORK","_number":16,"created":"2023-08-07 16:59:19.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/85/889785/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/889785/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/889785/16"}}},"commit":{"parents":[{"commit":"e71102c5bdc65358e0ea7766b296967ce5bb9ec2","subject":"Merge \"docs: Clean up proxy logging docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e71102c5bdc65358e0ea7766b296967ce5bb9ec2"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-24 17:45:21.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-08-07 16:59:12.000000000","tz":-420},"subject":"unit tests: improve FakeSwift handling of HEAD with query param","message":"unit tests: improve FakeSwift handling of HEAD with query param\n\nWe have to improve our unit-test coverage in swift since we detected that we couldn\u0027t issue a HEAD request with a partNum query param to our FakeSwift app without generating a KeyError since it wasn\u0027t included in the acceptable dict of responses.\n\nChange-Id: Idb4020fdeee87a9164312dc9647ab0820b098ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48e331bb16e0e0db801aaa830cc72b83890eda55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48e331bb16e0e0db801aaa830cc72b83890eda55"}]},"branch":"refs/heads/master"},"4dde712977972c336bf7e2e62e18b6a7c5766b1c":{"kind":"REWORK","_number":17,"created":"2023-08-09 17:15:10.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/85/889785/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/889785/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/889785/17"}}},"commit":{"parents":[{"commit":"e71102c5bdc65358e0ea7766b296967ce5bb9ec2","subject":"Merge \"docs: Clean up proxy logging docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e71102c5bdc65358e0ea7766b296967ce5bb9ec2"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-24 17:45:21.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-08-07 23:11:12.000000000","tz":-420},"subject":"unit tests: improve FakeSwift handling of HEAD with query param","message":"unit tests: improve FakeSwift handling of HEAD with query param\n\nWe have to improve our unit-test coverage in swift since we detected that we couldn\u0027t issue a HEAD request with a partNum query param to our FakeSwift app without generating a KeyError since it wasn\u0027t included in the acceptable dict of responses.\n\nChange-Id: Idb4020fdeee87a9164312dc9647ab0820b098ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4dde712977972c336bf7e2e62e18b6a7c5766b1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4dde712977972c336bf7e2e62e18b6a7c5766b1c"}]},"branch":"refs/heads/master"},"4ec3022389a5400815cf7cae31c346e2e34d28d3":{"kind":"REWORK","_number":18,"created":"2023-08-09 19:55:57.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/85/889785/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/889785/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/889785/18"}}},"commit":{"parents":[{"commit":"e71102c5bdc65358e0ea7766b296967ce5bb9ec2","subject":"Merge \"docs: Clean up proxy logging docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e71102c5bdc65358e0ea7766b296967ce5bb9ec2"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-24 17:45:21.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-08-09 19:55:50.000000000","tz":-420},"subject":"unit tests: improve FakeSwift handling of HEAD with query param","message":"unit tests: improve FakeSwift handling of HEAD with query param\n\nWe have to improve our unit-test coverage in swift since we detected that we couldn\u0027t issue a HEAD request with a partNum query param to our FakeSwift app without generating a KeyError since it wasn\u0027t included in the acceptable dict of responses.\n\nChange-Id: Idb4020fdeee87a9164312dc9647ab0820b098ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ec3022389a5400815cf7cae31c346e2e34d28d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ec3022389a5400815cf7cae31c346e2e34d28d3"}]},"branch":"refs/heads/master"},"238aefa14eb56586fa3723fca6d75158d8b8b2a1":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2023-08-09 19:56:08.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/85/889785/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/889785/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/889785/19"}}},"commit":{"parents":[{"commit":"dce403f0fb0c7ba4873dc28e04de40b79845b342","subject":"Merge \"proxy-server: fix AttributeError in exception handler\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dce403f0fb0c7ba4873dc28e04de40b79845b342"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-24 17:45:21.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-08-09 19:56:02.000000000","tz":-420},"subject":"unit tests: improve FakeSwift handling of HEAD with query param","message":"unit tests: improve FakeSwift handling of HEAD with query param\n\nWe have to improve our unit-test coverage in swift since we detected that we couldn\u0027t issue a HEAD request with a partNum query param to our FakeSwift app without generating a KeyError since it wasn\u0027t included in the acceptable dict of responses.\n\nChange-Id: Idb4020fdeee87a9164312dc9647ab0820b098ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/238aefa14eb56586fa3723fca6d75158d8b8b2a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/238aefa14eb56586fa3723fca6d75158d8b8b2a1"}]},"branch":"refs/heads/master"},"84784d19c19c187231ad3e6ef37ef5a512507aec":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2023-08-10 20:04:47.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/85/889785/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/889785/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/889785/20"}}},"commit":{"parents":[{"commit":"6444ef9be054973d3a695c3a86a0251ffc9bf22a","subject":"ContainerBroker: clarify get_shard_ranges docstring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6444ef9be054973d3a695c3a86a0251ffc9bf22a"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-24 17:45:21.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-08-10 20:04:47.000000000","tz":0},"subject":"unit tests: improve FakeSwift handling of HEAD with query param","message":"unit tests: improve FakeSwift handling of HEAD with query param\n\nWe have to improve our unit-test coverage in swift since we detected that we couldn\u0027t issue a HEAD request with a partNum query param to our FakeSwift app without generating a KeyError since it wasn\u0027t included in the acceptable dict of responses.\n\nChange-Id: Idb4020fdeee87a9164312dc9647ab0820b098ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/84784d19c19c187231ad3e6ef37ef5a512507aec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/84784d19c19c187231ad3e6ef37ef5a512507aec"}]},"branch":"refs/heads/master","description":"Rebase"},"53978bc573b31d35153d04466e087e0872b3cb18":{"kind":"REWORK","_number":21,"created":"2023-08-16 02:04:22.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/85/889785/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/889785/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/889785/21"}}},"commit":{"parents":[{"commit":"7ee91ba1bff33d89dab4b87e907d0fd30ce50bfe","subject":"fix swob HEAD handling for leak tracking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7ee91ba1bff33d89dab4b87e907d0fd30ce50bfe"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-24 17:45:21.000000000","tz":-420},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-08-16 01:58:19.000000000","tz":-300},"subject":"tests: fix FakeSwift HEAD with query param","message":"tests: fix FakeSwift HEAD with query param\n\nThe existing FakeSwift implementation already supports using registered\nGET responses for GET requests with a query param.  It also supports\nusing registered GET responses for HEAD requests (if they either both\nhad the exact SAME matching query params, or both did not have ANY query\nparams).  But it did not support using registered GET responses w/o\nquery params for HEAD requests with a query param, even though a GET\nwith the same query param would work.\n\nThis change makes it a little more consistent: any client or test that\nmakes a GET request, should be able to make a similar HEAD request and\nexpect consistent response headers and status.\n\nThis test infra improvement is needed as we\u0027re going to be extending\ntest_slo with a bunch of tests that assert consistent response headers\nfor both GET and HEAD requests w/ new query params.\n\nDrive-By: normalize FakeSwift.uploaded on raw_path, for some small\nadditional sanity when FakeSwift is making things up.\n\nChange-Id: Idb4020fdeee87a9164312dc9647ab0820b098ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/53978bc573b31d35153d04466e087e0872b3cb18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/53978bc573b31d35153d04466e087e0872b3cb18"}]},"branch":"refs/heads/master"},"6049fc2323d9be6fa2d76163721807fa279cb351":{"kind":"NO_CHANGE","_number":22,"created":"2023-08-17 15:18:38.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/85/889785/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/889785/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/889785/22"}}},"commit":{"parents":[{"commit":"9e065e2d23947f9f87fb7b68b5c3925419b8e16e","subject":"fix swob HEAD handling for leak tracking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e065e2d23947f9f87fb7b68b5c3925419b8e16e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-24 17:45:21.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-08-17 15:18:38.000000000","tz":0},"subject":"tests: fix FakeSwift HEAD with query param","message":"tests: fix FakeSwift HEAD with query param\n\nThe existing FakeSwift implementation already supports using registered\nGET responses for GET requests with a query param.  It also supports\nusing registered GET responses for HEAD requests (if they either both\nhad the exact SAME matching query params, or both did not have ANY query\nparams).  But it did not support using registered GET responses w/o\nquery params for HEAD requests with a query param, even though a GET\nwith the same query param would work.\n\nThis change makes it a little more consistent: any client or test that\nmakes a GET request, should be able to make a similar HEAD request and\nexpect consistent response headers and status.\n\nThis test infra improvement is needed as we\u0027re going to be extending\ntest_slo with a bunch of tests that assert consistent response headers\nfor both GET and HEAD requests w/ new query params.\n\nDrive-By: normalize FakeSwift.uploaded on raw_path, for some small\nadditional sanity when FakeSwift is making things up.\n\nChange-Id: Idb4020fdeee87a9164312dc9647ab0820b098ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6049fc2323d9be6fa2d76163721807fa279cb351"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6049fc2323d9be6fa2d76163721807fa279cb351"}]},"branch":"refs/heads/master","description":"Rebase"},"5f2426277149e191489e23bfb5f62b86ce67aa5d":{"kind":"NO_CODE_CHANGE","_number":23,"created":"2023-08-17 15:47:24.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/85/889785/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/889785/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/889785/23"}}},"commit":{"parents":[{"commit":"9e065e2d23947f9f87fb7b68b5c3925419b8e16e","subject":"fix swob HEAD handling for leak tracking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e065e2d23947f9f87fb7b68b5c3925419b8e16e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-24 17:45:21.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-08-17 15:47:24.000000000","tz":0},"subject":"tests: fix FakeSwift HEAD with query param","message":"tests: fix FakeSwift HEAD with query param\n\nThe existing FakeSwift implementation already supports using registered\nGET responses for GET requests with a query param.  It also supports\nusing registered GET responses for HEAD requests (if they either both\nhad the exact SAME matching query params, or both did not have ANY query\nparams).  But it did not support using registered GET responses w/o\nquery params for HEAD requests with a query param, even though a GET\nwith the same query param would work.\n\nThis change makes it a little more consistent: any client or test that\nmakes a GET request, should be able to make a similar HEAD request and\nexpect consistent response headers and status.\n\nThis test infra improvement is needed as we\u0027re going to be extending\ntest_slo with a bunch of tests that assert consistent response headers\nfor both GET and HEAD requests w/ new query params.\n\nChange-Id: Idb4020fdeee87a9164312dc9647ab0820b098ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f2426277149e191489e23bfb5f62b86ce67aa5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f2426277149e191489e23bfb5f62b86ce67aa5d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"eb2925f6b855e34b5fa95237948ed641826f9b0f":{"kind":"REWORK","_number":24,"created":"2023-08-17 15:55:27.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/85/889785/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/889785/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/889785/24"}}},"commit":{"parents":[{"commit":"9e065e2d23947f9f87fb7b68b5c3925419b8e16e","subject":"fix swob HEAD handling for leak tracking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e065e2d23947f9f87fb7b68b5c3925419b8e16e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-24 17:45:21.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-08-17 15:55:18.000000000","tz":-420},"subject":"tests: fix FakeSwift HEAD with query param","message":"tests: fix FakeSwift HEAD with query param\n\nThe existing FakeSwift implementation already supports using registered\nGET responses for GET requests with a query param.  It also supports\nusing registered GET responses for HEAD requests (if they either both\nhad the exact SAME matching query params, or both did not have ANY query\nparams).  But it did not support using registered GET responses w/o\nquery params for HEAD requests with a query param, even though a GET\nwith the same query param would work.\n\nThis change makes it a little more consistent: any client or test that\nmakes a GET request, should be able to make a similar HEAD request and\nexpect consistent response headers and status.\n\nThis test infra improvement is needed as we\u0027re going to be extending\ntest_slo with a bunch of tests that assert consistent response headers\nfor both GET and HEAD requests w/ new query params.\n\nDrive-By: normalize FakeSwift.uploaded on raw_path, for some small\nadditional sanity when FakeSwift is making things up.\n\nChange-Id: Idb4020fdeee87a9164312dc9647ab0820b098ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb2925f6b855e34b5fa95237948ed641826f9b0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb2925f6b855e34b5fa95237948ed641826f9b0f"}]},"branch":"refs/heads/master"},"6d50935bdc7330502ed738737475e685791e63e1":{"kind":"REWORK","_number":25,"created":"2023-08-17 22:35:52.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/85/889785/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/889785/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/889785/25"}}},"commit":{"parents":[{"commit":"9e065e2d23947f9f87fb7b68b5c3925419b8e16e","subject":"fix swob HEAD handling for leak tracking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e065e2d23947f9f87fb7b68b5c3925419b8e16e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-24 17:45:21.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-08-17 22:35:41.000000000","tz":-420},"subject":"tests: fix FakeSwift HEAD with query param","message":"tests: fix FakeSwift HEAD with query param\n\nThe existing FakeSwift implementation already supports using registered\nGET responses for GET requests with a query param.  It also supports\nusing registered GET responses for HEAD requests (if they either both\nhad the exact SAME matching query params, or both did not have ANY query\nparams).  But it did not support using registered GET responses w/o\nquery params for HEAD requests with a query param, even though a GET\nwith the same query param would work.\n\nThis change makes it a little more consistent: any client or test that\nmakes a GET request, should be able to make a similar HEAD request and\nexpect consistent response headers and status.\n\nThis test infra improvement is needed as we\u0027re going to be extending\ntest_slo with a bunch of tests that assert consistent response headers\nfor both GET and HEAD requests w/ new query params.\n\nDrive-By: normalize FakeSwift.uploaded on raw_path, for some small\nadditional sanity when FakeSwift is making things up.\n\nChange-Id: Idb4020fdeee87a9164312dc9647ab0820b098ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6d50935bdc7330502ed738737475e685791e63e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6d50935bdc7330502ed738737475e685791e63e1"}]},"branch":"refs/heads/master"},"4a4b7c5a83c539b188e1568aeac66ed8657c04de":{"kind":"REWORK","_number":26,"created":"2023-08-17 23:22:32.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/85/889785/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/889785/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/889785/26"}}},"commit":{"parents":[{"commit":"9e065e2d23947f9f87fb7b68b5c3925419b8e16e","subject":"fix swob HEAD handling for leak tracking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e065e2d23947f9f87fb7b68b5c3925419b8e16e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-24 17:45:21.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-08-17 23:21:17.000000000","tz":-420},"subject":"tests: fix FakeSwift HEAD with query param","message":"tests: fix FakeSwift HEAD with query param\n\nThe existing FakeSwift implementation already supports using registered\nGET responses for GET requests with a query param.  It also supports\nusing registered GET responses for HEAD requests (if they either both\nhad the exact SAME matching query params, or both did not have ANY query\nparams).  But it did not support using registered GET responses w/o\nquery params for HEAD requests with a query param, even though a GET\nwith the same query param would work.\n\nThis change makes it a little more consistent: any client or test that\nmakes a GET request, should be able to make a similar HEAD request and\nexpect consistent response headers and status.\n\nThis test infra improvement is needed as we\u0027re going to be extending\ntest_slo with a bunch of tests that assert consistent response headers\nfor both GET and HEAD requests w/ new query params.\n\nDrive-By: normalize FakeSwift.uploaded on raw_path, for some small\nadditional sanity when FakeSwift is making things up.\n\nChange-Id: Idb4020fdeee87a9164312dc9647ab0820b098ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4a4b7c5a83c539b188e1568aeac66ed8657c04de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4a4b7c5a83c539b188e1568aeac66ed8657c04de"}]},"branch":"refs/heads/master"},"076e306afa12aec62b1195431f83a4dcef6dc5e0":{"kind":"REWORK","_number":27,"created":"2023-08-17 23:23:44.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/85/889785/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/889785/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/889785/27"}}},"commit":{"parents":[{"commit":"9e065e2d23947f9f87fb7b68b5c3925419b8e16e","subject":"fix swob HEAD handling for leak tracking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e065e2d23947f9f87fb7b68b5c3925419b8e16e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-24 17:45:21.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-08-17 23:23:37.000000000","tz":-420},"subject":"tests: fix FakeSwift HEAD with query param","message":"tests: fix FakeSwift HEAD with query param\n\nThe existing FakeSwift implementation already supports using registered\nGET responses for GET requests with a query param.  It also supports\nusing registered GET responses for HEAD requests (if they either both\nhad the exact SAME matching query params, or both did not have ANY query\nparams).  But it did not support using registered GET responses w/o\nquery params for HEAD requests with a query param, even though a GET\nwith the same query param would work.\n\nThis change makes it a little more consistent: any client or test that\nmakes a GET request, should be able to make a similar HEAD request and\nexpect consistent response headers and status.\n\nThis test infra improvement is needed as we\u0027re going to be extending\ntest_slo with a bunch of tests that assert consistent response headers\nfor both GET and HEAD requests w/ new query params.\n\nDrive-By: normalize FakeSwift.uploaded on raw_path, for some small\nadditional sanity when FakeSwift is making things up.\n\nChange-Id: Idb4020fdeee87a9164312dc9647ab0820b098ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/076e306afa12aec62b1195431f83a4dcef6dc5e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/076e306afa12aec62b1195431f83a4dcef6dc5e0"}]},"branch":"refs/heads/master"},"a74aff5fd4a80aef5d8e4fbf47b7a6c2a3c21e65":{"kind":"NO_CHANGE","_number":28,"created":"2023-08-17 23:27:40.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/85/889785/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/889785/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/889785/28"}}},"commit":{"parents":[{"commit":"9e065e2d23947f9f87fb7b68b5c3925419b8e16e","subject":"fix swob HEAD handling for leak tracking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e065e2d23947f9f87fb7b68b5c3925419b8e16e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-24 17:45:21.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-08-17 23:27:40.000000000","tz":0},"subject":"tests: fix FakeSwift HEAD with query param","message":"tests: fix FakeSwift HEAD with query param\n\nThe existing FakeSwift implementation already supports using registered\nGET responses for GET requests with a query param.  It also supports\nusing registered GET responses for HEAD requests (if they either both\nhad the exact SAME matching query params, or both did not have ANY query\nparams).  But it did not support using registered GET responses w/o\nquery params for HEAD requests with a query param, even though a GET\nwith the same query param would work.\n\nThis change makes it a little more consistent: any client or test that\nmakes a GET request, should be able to make a similar HEAD request and\nexpect consistent response headers and status.\n\nThis test infra improvement is needed as we\u0027re going to be extending\ntest_slo with a bunch of tests that assert consistent response headers\nfor both GET and HEAD requests w/ new query params.\n\nDrive-By: normalize FakeSwift.uploaded on raw_path, for some small\nadditional sanity when FakeSwift is making things up.\n\nChange-Id: Idb4020fdeee87a9164312dc9647ab0820b098ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a74aff5fd4a80aef5d8e4fbf47b7a6c2a3c21e65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a74aff5fd4a80aef5d8e4fbf47b7a6c2a3c21e65"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7a3eed9c4b5d0efc5ac14be07ea85a10615f5ce7":{"kind":"REWORK","_number":29,"created":"2023-08-17 23:51:02.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/85/889785/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/889785/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/889785/29"}}},"commit":{"parents":[{"commit":"9e065e2d23947f9f87fb7b68b5c3925419b8e16e","subject":"fix swob HEAD handling for leak tracking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e065e2d23947f9f87fb7b68b5c3925419b8e16e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-24 17:45:21.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-08-17 23:50:50.000000000","tz":-420},"subject":"tests: fix FakeSwift HEAD with query param","message":"tests: fix FakeSwift HEAD with query param\n\nThe existing FakeSwift implementation already supports using registered\nGET responses for GET requests with a query param.  It also supports\nusing registered GET responses for HEAD requests (if they either both\nhad the exact SAME matching query params, or both did not have ANY query\nparams).  But it did not support using registered GET responses w/o\nquery params for HEAD requests with a query param, even though a GET\nwith the same query param would work.\n\nThis change makes it a little more consistent: any client or test that\nmakes a GET request, should be able to make a similar HEAD request and\nexpect consistent response headers and status.\n\nThis test infra improvement is needed as we\u0027re going to be extending\ntest_slo with a bunch of tests that assert consistent response headers\nfor both GET and HEAD requests w/ new query params.\n\nDrive-By: normalize FakeSwift.uploaded on raw_path, for some small\nadditional sanity when FakeSwift is making things up.\n\nChange-Id: Idb4020fdeee87a9164312dc9647ab0820b098ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a3eed9c4b5d0efc5ac14be07ea85a10615f5ce7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a3eed9c4b5d0efc5ac14be07ea85a10615f5ce7"}]},"branch":"refs/heads/master"},"dab7192e1ed13ee8cf62970762cbb78edbd4ed17":{"kind":"REWORK","_number":30,"created":"2023-08-18 13:48:39.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/85/889785/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/889785/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/889785/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/889785/30"}}},"commit":{"parents":[{"commit":"9e065e2d23947f9f87fb7b68b5c3925419b8e16e","subject":"fix swob HEAD handling for leak tracking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e065e2d23947f9f87fb7b68b5c3925419b8e16e"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-07-24 17:45:21.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-08-18 13:28:25.000000000","tz":60},"subject":"tests: fix FakeSwift HEAD with query param","message":"tests: fix FakeSwift HEAD with query param\n\nThe existing FakeSwift implementation already supports using registered\nGET responses for GET requests with a query param.  It also supports\nusing registered GET responses for HEAD requests (if they either both\nhad the exact SAME matching query params, or both did not have ANY query\nparams).  But it did not support using registered GET responses w/o\nquery params for HEAD requests with a query param, even though a GET\nwith the same query param would work.\n\nThis change makes it a little more consistent: any client or test that\nmakes a GET request, should be able to make a similar HEAD request and\nexpect consistent response headers and status.\n\nThis test infra improvement is needed as we\u0027re going to be extending\ntest_slo with a bunch of tests that assert consistent response headers\nfor both GET and HEAD requests w/ new query params.\n\nChange-Id: Idb4020fdeee87a9164312dc9647ab0820b098ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dab7192e1ed13ee8cf62970762cbb78edbd4ed17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dab7192e1ed13ee8cf62970762cbb78edbd4ed17"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
