)]}'
{"id":"openstack%2Fswift~890325","triplet_id":"openstack%2Fswift~master~I47ed1a292708ae95af948f7d25ee7113df7a3ba7","project":"openstack/swift","branch":"master","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-08-02 11:49:22.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I47ed1a292708ae95af948f7d25ee7113df7a3ba7","subject":"proxy-server: clarify variables in get_account_info()","status":"NEW","created":"2023-08-02 10:33:10.000000000","updated":"2023-08-02 11:49:22.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"36f1659733131eb26f85304b0ad76895db7a0ace","_number":890325,"virtual_id_number":890325,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-08-02 11:49:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-02 11:23:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-02 11:49:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"04fb22318395602b17db4df9f387d75b2eb87e6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-02 10:33:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c68bf4aa284cc1876ce1ec478e92846757766220","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 11:23:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbc3e006984140daa651c273b27e3338\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8b7764f71fb34f5494a9e49b5abf3ddf : SUCCESS in 48m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/48f037578af94b3f829a2513ae6fbca8 : SUCCESS in 16m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f09341022694bdca9ca475eabf8671b : SUCCESS in 40m 58s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7a597b78d5294ac186683bc4d97c9937 : SUCCESS in 17m 05s","accounts_in_message":[],"_revision_number":1},{"id":"36f1659733131eb26f85304b0ad76895db7a0ace","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 11:49:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2413e7f8324a4779b7127f5f9a7de15a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/925d4219575041c49e4ccb4cdbe216c0 : SUCCESS in 8m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/9e27937ecf6845c58a721e630d5b2231 : SUCCESS in 1h 00m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/38048049726d4a21bdda60ada6278372 : SUCCESS in 45m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/110e0ddc2c0c4494b4b764515ca9585f : FAILURE in 1h 07m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/27150b2ad3564de097fbe44a8f2b2adf : SUCCESS in 17m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b393adbdbec9461888e2186672b84f76 : SUCCESS in 25m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8b7da9522c67494cb994b27ac48ae47c : SUCCESS in 28m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/144d5063520348ee828a55fbb396ef4c : SUCCESS in 9m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ecde84a6100449c9b9722fe61bcc4b61 : SUCCESS in 5m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/31ce8f9043e840029d93683ffce77eb9 : SUCCESS in 14m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d55d544d7a494f1b82e90991e49b345c : SUCCESS in 16m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/05fd276e3dd34eee98bf2554f87f4542 : SUCCESS in 14m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2f88a31e8d754eb2a73a6feab4c44e03 : SUCCESS in 14m 08s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2c6d6aa8280e4f218f958c013abf2985 : SUCCESS in 14m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cc732edf0d0c480dbd406d5bb38e3633 : SUCCESS in 16m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/49a79f898f6e41baa14622615cdb53fc : SUCCESS in 26m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aeaed2a28f3a4226a809eb02291c8145 : SUCCESS in 16m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/724ae3468e8d4c0a8346ec8fb71fc5f2 : SUCCESS in 28m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a9f6d5ef8a8f47e3bb136294c7750f24 : SUCCESS in 24m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/530fce9f4d7a41a89d43d7f2978138d3 : SUCCESS in 15m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c96a189489874bd9b9337a38edc7d14c : SUCCESS in 7m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7aaf36a36fe74fb49812ddc827cd2574 : SUCCESS in 26m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3e6299862d9f4e918ed0d74559b26e41 : SUCCESS in 8m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f36fde4b3fbb41ff82498f707739a12d : SUCCESS in 56m 25s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4653f08768094ae3bc15316052e5c076 : SUCCESS in 1h 11m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7f93f8a1afbf4d3297c0dc3274cc1232 : SUCCESS in 51m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c60f727de49649469147a32a0a9d811d : SUCCESS in 45m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1111b82170904f6abee3652ee8a42b11 : SUCCESS in 16m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0eb8fbce09c49d881c3e25079b981ba : SUCCESS in 4m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7348310fbdb744d19a78908f53d6ada3 : SUCCESS in 25m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65622e0e385946fe8b20fbb203c3ebaa : SUCCESS in 56m 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"18d70ade73c300916269c8aa0fe9f0b6da7b58b9","revisions":{"18d70ade73c300916269c8aa0fe9f0b6da7b58b9":{"kind":"REWORK","_number":1,"created":"2023-08-02 10:33:10.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/25/890325/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/890325/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/890325/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/890325/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/890325/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/890325/1"}}},"commit":{"parents":[{"commit":"c51e81f640274db31bf45b5253d8e99d19a7e3d6","subject":"proxy: Bring back logging/metrics for get_*_info requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c51e81f640274db31bf45b5253d8e99d19a7e3d6"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-08-02 10:31:06.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-08-02 10:31:06.000000000","tz":60},"subject":"proxy-server: clarify variables in get_account_info()","message":"proxy-server: clarify variables in get_account_info()\n\nSplit out separate variables for logging_app and proxy_app in\nget_account_info(), for clarity and to be consistent with\nget_container_info().\n\nChange-Id: I47ed1a292708ae95af948f7d25ee7113df7a3ba7\nRelated-Change: I3f531f904340e4c8407185ed64b41d7d614a308a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/18d70ade73c300916269c8aa0fe9f0b6da7b58b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/18d70ade73c300916269c8aa0fe9f0b6da7b58b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
