)]}'
{"id":"openstack%2Fswift~890527","triplet_id":"openstack%2Fswift~master~I938ff7ed7bcbfb363f55821cccd92763dfce449d","project":"openstack/swift","branch":"master","topic":"p-get-path-cleanup","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-09-27 20:17:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I938ff7ed7bcbfb363f55821cccd92763dfce449d","subject":"proxy-server: fix node error limiting","status":"MERGED","created":"2023-08-04 14:03:43.000000000","updated":"2023-09-27 20:18:58.000000000","submitted":"2023-09-27 20:17:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"890527-p-get-path-cleanup","meta_rev_id":"e5c0ebb1daabe5e1224880cc93def8371a1a8602","_number":890527,"virtual_id_number":890527,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-27 20:17:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":1,"date":"2023-09-20 04:51:03.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":1,"date":"2023-09-25 06:50:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2023-09-05 17:03:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2023-09-27 16:39:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}],"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-04 14:28:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-04 15:28:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-04 15:50:06.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"},{"updated":"2023-08-31 05:08:25.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"CC"},{"updated":"2023-09-05 17:03:30.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2023-09-20 04:51:03.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2023-09-25 06:50:55.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"6a8e16b7920a52c74dcb51cd75664e578b3f3163","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-04 14:03:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"493162c7616f1b9ae83a3ebdcdeb35f4bdca1c87","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 14:28:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f826c7d9270f44f3b73bc4065a6191e2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b424316d48214da0b0b598e114fe2e6e : SUCCESS in 18m 20s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c1643b9e1f6d4ebdb333643118e34d9c : SUCCESS in 19m 37s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b3f8f4691b4c4da8ae06ecebf5119e76 : SUCCESS in 22m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ffc96b804084411d964aee84de3c3e48 : SUCCESS in 23m 05s","accounts_in_message":[],"_revision_number":1},{"id":"582dff050cd0b9cc6e2f661bca1ae038b043e5c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 15:28:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd7fdc1e20854813a780f4853a17e3cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/696e89cc518a4d25afc11775bca4a54f : SUCCESS in 10m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/c7b5d51486a84ae19f0fe4c074f42ed1 : FAILURE in 55m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d3cb91be3ba345aabb467aa4db43266c : SUCCESS in 1h 06m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/331a2d7b9c8545b3b189b0309d5be0bd : SUCCESS in 1h 15m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6749619c669248afada8b5c2cc36cf53 : SUCCESS in 18m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f9467240013f4398b9d269d649737ee5 : SUCCESS in 24m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3f18b249e95e4c108bb486a1b8bd9f29 : SUCCESS in 25m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3ca347ac5be34cbab2d86e23c680846b : SUCCESS in 10m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c82e1c3f4b224ade878b632787deab96 : SUCCESS in 7m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2c6f79aa0c374a73a13d02442324b4f3 : SUCCESS in 25m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/906ed4702213450394bfe0666748c3f9 : SUCCESS in 23m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/307a283652b9417eb99625076f7c3b0e : SUCCESS in 18m 46s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/77c53d2162bb41498499e3f5509fba81 : SUCCESS in 16m 48s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3ef096ab30844dc591b8eea3ceab8d1f : SUCCESS in 23m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/db2f47b55d8f45a78d6c8307381b01f2 : SUCCESS in 23m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2642b14563554406b26797e00c7a493c : SUCCESS in 16m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4aac0113b9aa464ca76bdcb150d33949 : SUCCESS in 17m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a6d6db9214334d4d934b1814657e8620 : SUCCESS in 17m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f56761c44e93411bbb336b8d31f58b0a : SUCCESS in 17m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bccf997e64db44749128b54e7ca601de : SUCCESS in 17m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dcbd57aa310a49cfa6f8a16b3c777131 : SUCCESS in 6m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dbee6b81278a4aa4970100f4780bb0c5 : SUCCESS in 16m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/12b0a22b55cf42098218f1a5b9353f31 : SUCCESS in 6m 37s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9e9bbac8d2d94a719b26edf350d6bd0e : SUCCESS in 1h 12m 59s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e966d56f17e04cf2b067edda58c356fa : SUCCESS in 1h 02m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/03014184aea64c438023fc61c5d60795 : SUCCESS in 40m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a3dc1d1a9af24f4da603fb59ab0b958b : SUCCESS in 55m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da907564693e4e57a33d6f0716313442 : SUCCESS in 25m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/704c1adeb79d4c8eb316c42f8b1381cb : SUCCESS in 4m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e38f1ab0e80f4a7184816672a7a3ceae : SUCCESS in 20m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d7caf2bbd7a499785a686c901a904cb : SUCCESS in 50m 57s","accounts_in_message":[],"_revision_number":1},{"id":"4fe390136663cd67e8d19d6a3c9c9cb964f6d379","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-08-04 15:50:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6461c09814ed88f6cc8759b5c7e5167144b0a43d","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-08-04 15:53:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4b5ff4e43c352568bc37daf9979ada3b52b6d588","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-04 16:58:26.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"40df303474e39838ca27e626385eb363c571d4f3","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-04 16:59:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3e0754251d14ef3c87efc12369350547f471fb3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 18:09:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/093052f64f714519a6c71ab6198d5838\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16c3ed018078420b976d4b2d7cfc62ca : SUCCESS in 8m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/2b3a0ca7444d4a3b9fe9e936322bc01b : SUCCESS in 59m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6fd32c104d6347598882f025cecd3eb6 : SUCCESS in 37m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d085ed7e00f4af59452919385035ea5 : SUCCESS in 53m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eb9942f64f1a4d9c94b36954a47e2191 : SUCCESS in 19m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d257849c65374bb489020c0abbd80cb0 : SUCCESS in 24m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fdd668b915aa4756aba071cd789d28d7 : SUCCESS in 18m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d7775ead2fee4e7ba4f8e8034b527999 : SUCCESS in 10m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e0bba494f08f449580fc0d4d99c466f7 : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/65e54674f46c4faea9a9cb8c752fcd7c : SUCCESS in 15m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d3449331a6b444a695a78e8f38c1b878 : SUCCESS in 17m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2c3d95510733404ca9089c9aac9f7cb7 : SUCCESS in 12m 45s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2b641d07e40941e1a4469fe95907c99e : SUCCESS in 22m 51s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/22b5eb11fce84744888809a3081d352f : SUCCESS in 16m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c6643683929540adaaf5d05b20670b81 : SUCCESS in 15m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/472aa931b3b047798c23a3de881cee6e : SUCCESS in 15m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8a3c2599907a43038f6eddb286a7346d : SUCCESS in 13m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9a482a3b8e4645f38ee04c0a330ac725 : SUCCESS in 15m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0cab350c1bee4882996d7060a556bf16 : SUCCESS in 15m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c9451d1d76ad43729bf816781bb92baf : SUCCESS in 17m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/76ffb20e3dc143a7ad8090ad85f9f992 : SUCCESS in 7m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d7f693de991e4bf1b4836ac98d749464 : SUCCESS in 18m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4698dfa32b574257875a57aede4c3455 : SUCCESS in 8m 05s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/88f4b7d4b97d42cdb3a900d6bb1b0bb8 : SUCCESS in 54m 45s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ecf5938d088e4ba2b776ffd7c5ec2391 : SUCCESS in 58m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/45a3ccd1933349c2aeab66c2c67dadd0 : SUCCESS in 1h 03m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2a7a301ac21648c7ac97800f91502041 : SUCCESS in 36m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d56a010bc0884bcdac761e4517bd9cec : SUCCESS in 18m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a68ff6dcf9724ebe9b2d06587938df49 : SUCCESS in 4m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dcf5ccf45b3443deb0638ed4844ac582 : SUCCESS in 20m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed19728079664a4baee35b4bd31f3925 : SUCCESS in 44m 17s","accounts_in_message":[],"_revision_number":2},{"id":"85828c7c414afafe4135d205181463b714576616","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 18:29:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d80890387a7445909016b0e733d069d0\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/018625ac5bec4ba0af3178466d832d01 : SUCCESS in 16m 48s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fa33a961183b4fa2b3da3629195ed474 : SUCCESS in 17m 04s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab2eba8edccb46e183b37d0aa3277001 : SUCCESS in 20m 47s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/314cc2d28c7c402e834bccbae95a0e9f : SUCCESS in 1h 28m 57s","accounts_in_message":[],"_revision_number":2},{"id":"9b998596dfac1b73d4556e2f726c2bcf6e6bc1e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-07 10:19:42.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"1815bfb3bebbc8fa33eda1434be58b3fc66c812e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 11:53:19.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89ffa3a541a14b7ea01153d0a1581bb7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cea7549c57b74c1eb3e11c69bd838210 : SUCCESS in 15m 41s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a14729874e4d47a6bb66b3f908f32eef : TIMED_OUT in 1h 30m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6500883e46114bc091d8568b06e0249a : SUCCESS in 17m 32s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/80b2f1f1ed3345c5b1c14c67ba9bb1d1 : SUCCESS in 17m 15s","accounts_in_message":[],"_revision_number":3},{"id":"66a5fbedb4047c700cf93fc618de1311033f45da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 12:24:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a679c3d14f94c91badeed319f041ba3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e9de475495f4516b293f46dd30495d2 : SUCCESS in 8m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/91a69c37d4b64a3c99a70a1bb51951ee : SUCCESS in 53m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7dca4bcb485a42afbc5062460ec759c6 : SUCCESS in 54m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b4b8b1085cb44819ac2f2ef5e00c2785 : SUCCESS in 1h 06m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/98871dedfcd7496a837722c2a54c4e64 : SUCCESS in 17m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3df79947a21a43939c2b23f9b561c8d0 : SUCCESS in 16m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/03b1f1d23e0d4898801672919db47574 : SUCCESS in 16m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a47a33de0ed845c09e15f5ebdcfb1746 : SUCCESS in 9m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d7eaa787aeb14ddebe426c387086d0bf : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/59526c9cbae64d14bd1444bfbf6cabbb : SUCCESS in 43m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bae1750e411b42bd9d8e89af0694bac8 : SUCCESS in 16m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/14a4ace83a134dddad9e79b59e67e210 : SUCCESS in 17m 12s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/81c4f2a5ebd04bf596ee8d2521a92224 : SUCCESS in 18m 32s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/732e0a7cb7274bb28c6b5de1122bc5d3 : SUCCESS in 18m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/48d8387f71024fcc950f25973c136f19 : SUCCESS in 16m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9f19a723d59f443994163c05f061e529 : SUCCESS in 13m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/639a0ea24e6647c287b4562f744013c4 : SUCCESS in 13m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9af0176c2f134d2aa3c724b78fcfccdd : SUCCESS in 12m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1d210f56ee854b3d8a55e382b40dea8b : SUCCESS in 17m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/154a653b47e54934b09dbc5bf8ff0fdb : SUCCESS in 21m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fd07f8b72ddf44df979f594d34ad8102 : SUCCESS in 7m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/49edd00681c04b729467c97a179c8cda : SUCCESS in 16m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4a5f787fcc1c4792b4a56da9c1421aaf : SUCCESS in 7m 15s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4c818a6ebecd4caf8459670021f11633 : SUCCESS in 58m 03s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8774966c7785417289b18948e47e9d41 : SUCCESS in 56m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2dd5fe7d917f47a2b860602a277ada49 : SUCCESS in 51m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1fc25f3582ff40dca8f34b37521667e9 : TIMED_OUT in 1h 36m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f6e9cb836ff4107aed398eb7283662a : SUCCESS in 19m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbd583f99a0f441ab9da411dd994ae44 : SUCCESS in 5m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/87dd2e5420fd4a63ad9dde4cab501979 : SUCCESS in 17m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1bbde01e4f88459dbe24b82617dd5c3f : SUCCESS in 1h 08m 54s","accounts_in_message":[],"_revision_number":3},{"id":"fa1d06d2558557caab2ef4ce2f8709b22e990ea7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-09 14:48:28.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"03904b2d1e96ec53e3b7f28b32db709e317bc5b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 15:08:06.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/703f30168af247f49a2a9bc75c9a16ed\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5046bf23c28246c895e73080b5514858 : FAILURE in 14m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8d0939a4bf85433c95ffcff689787f21 : FAILURE in 14m 45s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f799e8ff708b41d1a28243ca42680e43 : SUCCESS in 17m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/067f4d62b9a84d7e91339c32789cc8be : SUCCESS in 18m 18s","accounts_in_message":[],"_revision_number":4},{"id":"3995452ca08c465c0cc74f66bf291745d29017ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-09 15:22:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d3044b63cc1633251e45f41b69f691b5dcb060e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 16:37:56.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d95bb12ee73f426e9d328e80750bd571\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/95139166590d4072bfc6dddc1d342b62 : SUCCESS in 1h 10m 31s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7d28794f35004bf2a6083fa5cb483ff0 : SUCCESS in 16m 11s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9bac845f909c4b5a800e50e3dd9e007b : SUCCESS in 1h 13m 16s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a717e00ee0474be9ade0ca02a340b212 : POST_FAILURE in 20m 02s","accounts_in_message":[],"_revision_number":5},{"id":"e1fe1ad4ff580e4f8972ad0c33bf96b99ac99feb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 16:42:43.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9dc473147265468092979620deb6b3cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1152f2b708364d35a7acb7791c999a5d : SUCCESS in 8m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/56a9ab6bcf6941069f872fb30cdf09dd : SUCCESS in 51m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/44312e3bb46d43f7afe6b29934088a92 : SUCCESS in 49m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ccfd72507bfc48b085b9fbd1a78fcd56 : SUCCESS in 51m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/23a492c7796246668724810af783e2d3 : SUCCESS in 17m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1a1ecbb8d18049fe875cf92673fdfa4c : SUCCESS in 24m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ba273e44c3ed4fb194c82eefc637e3a3 : SUCCESS in 28m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4d94cfce3c444553b0590a03913c55bb : SUCCESS in 8m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/439a1f02f5d9465fbda5867af044352f : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/17c285ba51ae493e96ae41afca32e1f6 : SUCCESS in 16m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fc39b2173afc43b991fd60c0c9198047 : SUCCESS in 14m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9f002994378946248374882b37de96a4 : SUCCESS in 15m 54s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/31cb87b59f174bb6b3a908768e1ef2e3 : SUCCESS in 15m 29s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3e73b18743f54c7f8106dc452d04a225 : POST_FAILURE in 18m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c71e54bbe26a4762b3a099885163e228 : SUCCESS in 17m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8917ae357ab54cfbab6b3d9cf48adccf : SUCCESS in 18m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7789b94c95de4876bc6d28a9a3ef3866 : SUCCESS in 19m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/44753a9e143541a596003643eff41acf : SUCCESS in 15m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4240491304ba4db28a589318209f78b8 : SUCCESS in 15m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/93675be3932e438692679d7db47cbb9c : SUCCESS in 16m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d3a4b88b13ba4315ba27a4f282895dd8 : SUCCESS in 8m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/548ab2a9625a4f2abef4cd9cb29c138a : SUCCESS in 17m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5fb1937d8fba4bec9bba86830ef42377 : SUCCESS in 6m 33s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/078176cc9ae44807832418c17cf8df24 : SUCCESS in 1h 08m 37s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9b20f3a662b84557ad5e45509e493cad : SUCCESS in 1h 06m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/90668e7fafdc481d948b2f327d7a3f69 : SUCCESS in 42m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a1a099828d5f4d16a68fae1cb44274eb : SUCCESS in 57m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de6fd14246ad474a994a8cbb1fbdb07b : SUCCESS in 14m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f64194580b1f412bad076eb184f787c7 : SUCCESS in 5m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6d7d82e68df84fd29b3e83c78703baa8 : SUCCESS in 28m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09c620e51ea54e71977e8c0f7d357d01 : SUCCESS in 1h 01m 51s","accounts_in_message":[],"_revision_number":5},{"id":"b3735d8a4ef047eee9a0de75ad1d30bdb4bf1e8d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-16 14:38:11.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"aaa5f22eb92a47eb6ba4b968b0adc5136baa68d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 15:57:42.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25e6a16d6776466186a2d56e99a3c624\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/00badca14a454c9a8e6d39c02c1f77f6 : SUCCESS in 1h 09m 17s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/32049bcd0bcf40e1b9f6a2ed5d5cccfa : SUCCESS in 1h 05m 34s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/772de038102f4519823957a3efb00fe8 : SUCCESS in 20m 17s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/abb9ea8fa464432795888bff035299de : SUCCESS in 1h 13m 29s","accounts_in_message":[],"_revision_number":5},{"id":"26b9d44aada82c4dfdf6c4a9810b9ea06a77102b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 16:15:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce52ff9bac924ca4b76fec7591e4afcd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/762f84d8cdcc4759bce9fb00b154010a : SUCCESS in 8m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/19b1eef419954817800a69810bfad048 : SUCCESS in 49m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2803233b08964285bacef872108e5db2 : SUCCESS in 48m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/678b7277c4c841bd82cd07c90761c646 : SUCCESS in 54m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a8519b3481e0431d93949de71b78f67c : SUCCESS in 20m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3bc4a3a3342d4348bcbda02de0722dfe : SUCCESS in 25m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c81ca46d48734c4cbca50174e041ff75 : SUCCESS in 18m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8f4f44c96d58458187dd18fea001a84e : SUCCESS in 9m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7c5ee5a2a8bb49f6a78cedb967b6de3d : SUCCESS in 6m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/502715773c424d99a639ce4d64c13e14 : SUCCESS in 20m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6b5f3c1e5cd54817baef5e9ec36e95e1 : SUCCESS in 24m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e75c177765034242bc4f74294f264be6 : SUCCESS in 28m 39s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0ad33ac9dd924abea5189b6c6066afab : SUCCESS in 17m 11s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/afce3b9f0ea843ac99ab4c1fc17a0442 : SUCCESS in 18m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9388de6db5374953b0620f797ee2d85d : SUCCESS in 19m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9a5e8f87912545319a7f6dde97cb7a5e : SUCCESS in 19m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0af51874e7ce4d7cba4b1b99abed707b : SUCCESS in 20m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7b3b1f6456004fd6a3e8a56ee4e36004 : SUCCESS in 13m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e5e1f720934f4e058a511b95c24a0e0a : SUCCESS in 15m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a1963697046842c5b1e418c4bd59c48b : SUCCESS in 16m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7fd79f23ffda40859f75055c76547cb6 : SUCCESS in 7m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/97651cb36ebd41d49d434063ba359f97 : SUCCESS in 14m 35s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/da028ae3a9654537ada2e9415aca095d : SUCCESS in 6m 46s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/48ed2d67073440ffb1afa3c657c89784 : SUCCESS in 59m 53s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f25cbfb1d8a946f4b48f14e6688bdf22 : SUCCESS in 1h 08m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dd9ec06ee44b48c98f4255f42fc149f5 : SUCCESS in 50m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8a21abb660334c47a65d8e943000ef20 : SUCCESS in 49m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/faac20b266414903bd09f61498eb7e5b : SUCCESS in 17m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8344e19a90074f959771b95a0fb16864 : SUCCESS in 5m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ed62c02e335f47479c351f35165db911 : SUCCESS in 23m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04d0894863d24386b939d878fb8452bc : SUCCESS in 53m 23s","accounts_in_message":[],"_revision_number":5},{"id":"ada926e9cd2742ca636b69e9a64e645d5e437dab","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-08-31 05:08:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6e7ff75c73fcaf2c681685f19d6e9365805fc404","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-09-05 17:03:30.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"802ae34f28b6c80f3aa5b8c8f261c811c8251429","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-09-06 11:12:23.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7481ec6f4239c67e4086cca3a4da0cb4289e6735","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-09-20 04:51:03.000000000","message":"Patch Set 5: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"37fe5ba2d33deddbee980024b63c031e8adb333c","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-09-25 06:50:55.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"80b919997c798788687cca4140f58224c2c287cb","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-09-27 16:39:20.000000000","message":"Patch Set 5: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f73195cb679457d3c9643db79ea5502d8f0c45d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 16:39:58.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b824807c4bcfd2598c8c7f65a0c99ae2417cf657","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 20:17:51.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7477b4b3b5d641debd64a0ba55a0f9ed\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91db886c41b44fd3ae6d9b0f28f54bcc : SUCCESS in 8m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/eeff87ee90fe43299c5b37121a426b52 : SUCCESS in 51m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d580d447741842a786013e308c1384de : SUCCESS in 31m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3755382d448d477fa60ef2eb8f40e4ad : SUCCESS in 1h 08m 37s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/32f19a1d073d4e5a8bd23acf1f23567e : SUCCESS in 9m 10s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/acb41bf0061d46d69009d68e1bfdceb8 : SUCCESS in 6m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f19c595aee6446988572f2dc71771e81 : SUCCESS in 16m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3ce276ab581440f996265ac70b2a151f : SUCCESS in 16m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1d8c97ed83b04cfc9204c1ad844dc343 : SUCCESS in 20m 37s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ef133a2bbea541c5ac4016ffa3a44dd4 : SUCCESS in 15m 02s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3972bb99ca08417b9dc50db8fa553afe : SUCCESS in 13m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8dd572aba9e8488789277ac5c49b4f82 : SUCCESS in 20m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/27a413d74a1b44c9bffab058d5f2bb87 : SUCCESS in 18m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cbd91aacafa740a3b9d824c96a8df3cd : SUCCESS in 19m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4bd1c756eda74f1f9a629af8057aad04 : SUCCESS in 17m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2a0689336fac4a9a8938994531a5893f : SUCCESS in 12m 42s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/94cf6ed3ae984216a384e7224cd310eb : SUCCESS in 16m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/aa2d83bd6f90418b9f1b0af33b7f4b65 : SUCCESS in 8m 03s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1938d350d82f4ff0954621c5946258bc : SUCCESS in 8m 19s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9b33221c096a41e287daa2e67a6d3b0b : SUCCESS in 52m 28s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/43f58b0633494b3d90459483fd75231f : SUCCESS in 50m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/114b84917d8341afa4c25e0862d798c7 : SUCCESS in 56m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e20b01fd90f145bab8177db7248b1e42 : SUCCESS in 1h 03m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/994da59663304e63a83df1c810cb699d : SUCCESS in 16m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29fa239d3485444db78b61775b7ae158 : SUCCESS in 4m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f86c8db398ef4e2fb1672ebc5084ca2f : SUCCESS in 1h 00m 48s","accounts_in_message":[],"_revision_number":5},{"id":"20aab56357a28e42369a1489e4b4db0ca002137a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 20:17:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"e5c0ebb1daabe5e1224880cc93def8371a1a8602","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 20:18:58.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0366e51eb804339a29ff344f3fb2a2d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f571913576a1484e855c8626b9f0147c : SUCCESS in 47s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/03b0743e01e5469cabd63740ad2536d8 : SUCCESS in 33s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/b948af60c2b14f3399449134e71e1c18 : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"01c7ade798b612eca440fd389942abe131574cc2","revisions":{"27c840d5d4bb932cb93e3c3e3b0e3d07e53882fe":{"kind":"REWORK","_number":1,"created":"2023-08-04 14:03:43.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/27/890527/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/890527/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/890527/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/890527/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/890527/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/890527/1"}}},"commit":{"parents":[{"commit":"24d2976f63191f778bcf351d601876bdf988dbc4","subject":"proxy: encapsulate Getter resp, node and parts_iter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24d2976f63191f778bcf351d601876bdf988dbc4"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-08-04 13:06:25.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-08-04 13:58:35.000000000","tz":60},"subject":"proxy-server: error limit nodes that cannot be replaced","message":"proxy-server: error limit nodes that cannot be replaced\n\nIf the proxy-server fails while reading data from a backend node\nduring an object GET then it will increment the node\u0027s error limiting\ncounter. Previously the error counter was only incremented if a useful\nreplacement node was found. For example, if the last useful primary\nnode failed and all handoff nodes returned 404, the last primary would\nnot have its error counter incremented.\n\nThis patch ensures that the node error limiting counter is always\nincremented after reading from a node fails, even if no more useful\nnodes are available.\n\nChange-Id: I938ff7ed7bcbfb363f55821cccd92763dfce449d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27c840d5d4bb932cb93e3c3e3b0e3d07e53882fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27c840d5d4bb932cb93e3c3e3b0e3d07e53882fe"}]},"branch":"refs/heads/master"},"f6dce2d17a45e428f3ec2aff9fa9b679209bd886":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-08-04 16:58:26.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/27/890527/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/890527/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/890527/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/890527/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/890527/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/890527/2"}}},"commit":{"parents":[{"commit":"24d2976f63191f778bcf351d601876bdf988dbc4","subject":"proxy: encapsulate Getter resp, node and parts_iter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24d2976f63191f778bcf351d601876bdf988dbc4"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-08-04 13:06:25.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-08-04 16:35:28.000000000","tz":60},"subject":"proxy-server: fix node error limiting","message":"proxy-server: fix node error limiting\n\nIf the proxy-server gets an error (e.g. timeout) while reading data\nfrom a backend node during an object GET then it will increment the\nnode\u0027s error limiting counter. If the node\u0027s error counter reaches a\nthreshold then the node is \u0027error-limited\u0027 i.e. not used for a period\nof time.\n\nWhen a read from a node fails the proxy will also attempt to resume\nthe read by searching for another useful node to replace the failed\nnode. Previously, the outcome of this search for a replacement node\nerroneously determined whether the failed node\u0027s error counter was\nincremented or not: the error counter was only incremented if a useful\nreplacement node was found.\n\nFor example, if a read from the last useful primary node failed, and\nall handoff nodes then returned 404, the last primary node\u0027s would not\nhave its error counter incremented (for no good reason). However, if a\nuseful replacement handoff node was found, then the last primary\nnode\u0027s error counter would be incremented.\n\nThis patch fixes the inconsistent and confusing handling of node\nerrors by ensuring that the node error limiting counter is *always*\nincremented when reading from a node fails, *before* the search for a\nreplacement node.\n\nNote: The Related-Change fixed a bug whereby the replacement node\u0027s\nerror counter was incremented rather than the failed node\u0027s error\ncounter. This patch addresses the inconsistent incrementing of the\nfailed node\u0027s error counter.\n\nRelated-Change: Ibd405b79820cedaf9cb02906b8c13c8d34f4d7ec\nChange-Id: I938ff7ed7bcbfb363f55821cccd92763dfce449d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6dce2d17a45e428f3ec2aff9fa9b679209bd886"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6dce2d17a45e428f3ec2aff9fa9b679209bd886"}]},"branch":"refs/heads/master"},"caccdf0fb2e16ca75b6d74eef18f6639803bd27c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-08-07 10:19:42.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/27/890527/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/890527/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/890527/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/890527/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/890527/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/890527/3"}}},"commit":{"parents":[{"commit":"f8a6dee622e47cec7111007069dd4ca74b56b49c","subject":"proxy: encapsulate Getter resp, node and parts_iter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8a6dee622e47cec7111007069dd4ca74b56b49c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-08-04 13:06:25.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-08-07 10:17:22.000000000","tz":60},"subject":"proxy-server: fix node error limiting","message":"proxy-server: fix node error limiting\n\nIf the proxy-server gets an error (e.g. timeout) while reading data\nfrom a backend node during an object GET then it will increment the\nnode\u0027s error limiting counter. If the node\u0027s error counter reaches a\nthreshold then the node is \u0027error-limited\u0027 i.e. not used for a period\nof time.\n\nWhen a read from a node fails the proxy will also attempt to resume\nthe read by searching for another useful node to replace the failed\nnode. Previously, the outcome of this search for a replacement node\nerroneously determined whether the failed node\u0027s error counter was\nincremented or not: the error counter was only incremented if a useful\nreplacement node was found.\n\nFor example, if a read from the last useful primary node failed, and\nall handoff nodes then returned 404, the last primary node\u0027s would not\nhave its error counter incremented (for no good reason). However, if a\nuseful replacement handoff node was found, then the last primary\nnode\u0027s error counter would be incremented.\n\nThis patch fixes the inconsistent and confusing handling of node\nerrors by ensuring that the node error limiting counter is *always*\nincremented when reading from a node fails, *before* the search for a\nreplacement node.\n\nNote: The Related-Change fixed a bug whereby the replacement node\u0027s\nerror counter was incremented rather than the failed node\u0027s error\ncounter. This patch addresses the inconsistent incrementing of the\nfailed node\u0027s error counter.\n\nRelated-Change: Ibd405b79820cedaf9cb02906b8c13c8d34f4d7ec\nChange-Id: I938ff7ed7bcbfb363f55821cccd92763dfce449d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/caccdf0fb2e16ca75b6d74eef18f6639803bd27c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/caccdf0fb2e16ca75b6d74eef18f6639803bd27c"}]},"branch":"refs/heads/master"},"793b05e5e7543c2af2a938147fc5950fbf52755d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-08-09 14:48:28.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/27/890527/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/890527/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/890527/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/890527/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/890527/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/890527/4"}}},"commit":{"parents":[{"commit":"dce403f0fb0c7ba4873dc28e04de40b79845b342","subject":"Merge \"proxy-server: fix AttributeError in exception handler\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dce403f0fb0c7ba4873dc28e04de40b79845b342"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-08-04 13:06:25.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-08-09 14:33:47.000000000","tz":60},"subject":"proxy-server: fix node error limiting","message":"proxy-server: fix node error limiting\n\nIf the proxy-server gets an error (e.g. timeout) while reading data\nfrom a backend node during an object GET then it will increment the\nnode\u0027s error limiting counter. If the node\u0027s error counter reaches a\nthreshold then the node is \u0027error-limited\u0027 i.e. not used for a period\nof time.\n\nWhen a read from a node fails the proxy will also attempt to resume\nthe read by searching for another useful node to replace the failed\nnode. Previously, the outcome of this search for a replacement node\nerroneously determined whether the failed node\u0027s error counter was\nincremented or not: the error counter was only incremented if a useful\nreplacement node was found.\n\nFor example, if a read from the last useful primary node failed, and\nall handoff nodes then returned 404, the last primary node\u0027s would not\nhave its error counter incremented (for no good reason). However, if a\nuseful replacement handoff node was found, then the last primary\nnode\u0027s error counter would be incremented.\n\nThis patch fixes the inconsistent and confusing handling of node\nerrors by ensuring that the node error limiting counter is *always*\nincremented when reading from a node fails, *before* the search for a\nreplacement node.\n\nNote: The Related-Change fixed a bug whereby the replacement node\u0027s\nerror counter was incremented rather than the failed node\u0027s error\ncounter. This patch addresses the inconsistent incrementing of the\nfailed node\u0027s error counter.\n\nRelated-Change: Ibd405b79820cedaf9cb02906b8c13c8d34f4d7ec\nChange-Id: I938ff7ed7bcbfb363f55821cccd92763dfce449d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/793b05e5e7543c2af2a938147fc5950fbf52755d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/793b05e5e7543c2af2a938147fc5950fbf52755d"}]},"branch":"refs/heads/master"},"01c7ade798b612eca440fd389942abe131574cc2":{"kind":"REWORK","_number":5,"created":"2023-08-09 15:22:50.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/27/890527/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/890527/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/890527/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/890527/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/890527/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/890527/5"}}},"commit":{"parents":[{"commit":"dce403f0fb0c7ba4873dc28e04de40b79845b342","subject":"Merge \"proxy-server: fix AttributeError in exception handler\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dce403f0fb0c7ba4873dc28e04de40b79845b342"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-08-04 13:06:25.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-08-09 15:16:58.000000000","tz":60},"subject":"proxy-server: fix node error limiting","message":"proxy-server: fix node error limiting\n\nIf the proxy-server gets an error (e.g. timeout) while reading data\nfrom a backend node during an object GET then it will increment the\nnode\u0027s error limiting counter. If the node\u0027s error counter reaches a\nthreshold then the node is \u0027error-limited\u0027 i.e. not used for a period\nof time.\n\nWhen a read from a node fails the proxy will also attempt to resume\nthe read by searching for another useful node to replace the failed\nnode. Previously, the outcome of this search for a replacement node\nerroneously determined whether the failed node\u0027s error counter was\nincremented or not: the error counter was only incremented if a useful\nreplacement node was found.\n\nFor example, if a read from the last useful primary node failed, and\nall handoff nodes then returned 404, the last primary node\u0027s would not\nhave its error counter incremented (for no good reason). However, if a\nuseful replacement handoff node was found, then the last primary\nnode\u0027s error counter would be incremented.\n\nThis patch fixes the inconsistent and confusing handling of node\nerrors by ensuring that the node error limiting counter is *always*\nincremented when reading from a node fails, *before* the search for a\nreplacement node.\n\nNote: The Related-Change fixed a bug whereby the replacement node\u0027s\nerror counter was incremented rather than the failed node\u0027s error\ncounter. This patch addresses the inconsistent incrementing of the\nfailed node\u0027s error counter.\n\nRelated-Change: Ibd405b79820cedaf9cb02906b8c13c8d34f4d7ec\nChange-Id: I938ff7ed7bcbfb363f55821cccd92763dfce449d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/01c7ade798b612eca440fd389942abe131574cc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/01c7ade798b612eca440fd389942abe131574cc2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
