)]}'
{"id":"openstack%2Fswift~890578","triplet_id":"openstack%2Fswift~master~I35d682e58419f141429877ac6a6964132f7e658b","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-08-08 09:11:02.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-08-08 13:11:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I35d682e58419f141429877ac6a6964132f7e658b","subject":"Save some effort when discovering used_source_etag","status":"MERGED","created":"2023-08-04 23:03:31.000000000","updated":"2023-08-08 13:12:38.000000000","submitted":"2023-08-08 13:11:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890578","meta_rev_id":"0e51efbe0381b3fbd814f3d6857815c745acd37b","_number":890578,"virtual_id_number":890578,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-08 13:11:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2023-08-07 11:18:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2023-08-07 11:18:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-04 23:24:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-05 00:16:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-07 11:18:53.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"8ebc816fe9e5ec6bfaf96735e78a50652b4ba272","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-04 23:03:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"440a0602f5da573b8d2b15fe340bb6f9e3c2f1ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 23:24:58.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f738ec83e97428e831c1560724db94c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8fe82dadc1704062a1955f10b7ae11f6 : FAILURE in 15m 23s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/199f932e4d2d4d2289a8c73375078dd8 : FAILURE in 15m 22s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/64b0365ed0ff485c9904366708b0ce09 : SUCCESS in 20m 31s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ae7d8c0e804b42ad859bd670a982a74e : SUCCESS in 18m 29s","accounts_in_message":[],"_revision_number":1},{"id":"f05266dbd12332763143691cfd180b2ea5be5a97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-05 00:16:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/baa38a4b7d3d4416806452f5136690b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/222774da4542493c897102abcf1ee9da : SUCCESS in 7m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/326004df748f4d8b9a5c15a2de3dff8b : SUCCESS in 1h 04m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a9b41fe8f3924a8ba4478e9fbbc46229 : SUCCESS in 53m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ac311ad2f34c46ba953e138eac0debad : SUCCESS in 58m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/14853dd2d7814206bfaf85694a447154 : SUCCESS in 17m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c1fb577128234ff7ba85b213c81b751d : SUCCESS in 19m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5b785126a4894c34ac206c36904a6493 : SUCCESS in 20m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b9a82785b349423bb7933d5343b0c8bf : SUCCESS in 7m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d9273db27fb841b2b57f4f65c031e8cc : SUCCESS in 4m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a1b4d5261969432a95a06d35b67ac4a4 : FAILURE in 18m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cb983a77b09d4cee9d7702584ab9282f : FAILURE in 17m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/85c3f07841c7472b848cc42a6469a223 : FAILURE in 57m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/319a4e86d1824642966b498f8e676408 : FAILURE in 17m 59s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f576aec64085434695463a2ecedc6090 : FAILURE in 18m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/90bb71ced03348eebe88657537c30ceb : SUCCESS in 18m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8955370cb5dd48ad9955fc197ec054ac : SUCCESS in 18m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/399282064ffc4de2b904114eb55675c8 : SUCCESS in 17m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bbcc372a146642c38cc0bff31ab84277 : SUCCESS in 16m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a2bfe78bd8ac41d4874ab99ad62d31fb : SUCCESS in 16m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4b0b47f96f5e423581f9d2b4ac49a15d : SUCCESS in 16m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c487cf9c06294ad59acfec7539b1d8f4 : SUCCESS in 6m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/282b61186f6b4f16a3bec8bc3b6b35b4 : SUCCESS in 15m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/13b0390bee7647789326d16e831dc59c : SUCCESS in 7m 17s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6a6752e7192b4e05a8399a9a748d717b : SUCCESS in 1h 03m 01s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3fc9db49658b410cad4155e240d4bd35 : SUCCESS in 1h 00m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/76ea9c8ebcb148a3bf22c262cac6ba84 : SUCCESS in 56m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cfeac085c497406992ac1051c8f34333 : SUCCESS in 54m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/88fa1a4dc6e740b687db64a22acb18c0 : FAILURE in 17m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e11744ac219c426fa2b22d88f9275dd0 : SUCCESS in 4m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6a10c82531a34a7b911b289adf931122 : SUCCESS in 20m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a236c959b484ba68fbeb8a4c89015c6 : SUCCESS in 1h 01m 57s","accounts_in_message":[],"_revision_number":1},{"id":"75176de6d9e00ff1e0feaa794a8fd5f0a918ef8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-07 11:17:50.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"31b98589f6789db6a994732200f410665bc9680b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-07 11:18:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3c4106b45374e5c646102698b5d23b28544b1e3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 11:37:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09d11f784f9346f2bfb3fd4c59380104\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/49a5c33e0ae842c59a261aab219478ab : SUCCESS in 16m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/861e56d0b1814f8891a162a47c9cb24c : SUCCESS in 16m 16s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3779ebabd9d94da4abfb514c5bb3450a : SUCCESS in 18m 42s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8fecb6cd5cd84153bd09e2297f9a3384 : SUCCESS in 18m 35s","accounts_in_message":[],"_revision_number":2},{"id":"7442b9f93c285bba07c2d9a4d97d0e2a82938b96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 12:27:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba32132e054c4c7e92060e0c49ebd0f4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0483e370df944a5ab61a5681e558c57 : SUCCESS in 8m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/80168ec05f2b49bcb2589fce98684676 : SUCCESS in 59m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9296860810554703baf9a9bacc83ace0 : SUCCESS in 44m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52d74bb18af2496288ce3f42b082acf9 : SUCCESS in 42m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/da6c4d79be8745e99bdb9376e1e7175c : SUCCESS in 19m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e674494601574967adf769608976eea9 : SUCCESS in 19m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2f9d1fee8b3d483c99f19f0fe007af58 : SUCCESS in 20m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fe8071e936c942dcaca8f3b8271a5b7a : SUCCESS in 7m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/370a9bfc512748a18540692b21d573be : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3d3d5806c9e44fce8227d4eb362048ef : SUCCESS in 18m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/07a4dfff9bbb40ee825e0c61d30f63e7 : SUCCESS in 16m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bdcdbea71dc04b79a9e22410229b5f03 : SUCCESS in 16m 14s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8bf7ec998523481d854b7dd522eb9a44 : SUCCESS in 16m 08s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0232efadcf6049649acbf4fc63068310 : SUCCESS in 18m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/160e647fa3244407824d5454ca1fcfe5 : SUCCESS in 16m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5bd345abdc5c4b3880498c56e1a080cb : SUCCESS in 15m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9797d4855a3c4436a7670c7e501aaf5b : SUCCESS in 18m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4cae6564a07a4f76b32296f6774971f2 : SUCCESS in 14m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e897c3f66b124afb95def6b81bc4037f : SUCCESS in 12m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ec546d05583e4f6cbb8f7cbd117d625e : SUCCESS in 15m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3d5bd61b61b84c1687b073a4017f65c3 : SUCCESS in 7m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bf8fca1f399e4651aa4cba83f46d0acd : SUCCESS in 13m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/62a71fc3170e48b5802b2ef8f5d82494 : SUCCESS in 8m 32s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/864d3d48165c496e93b613a9a89f41e0 : SUCCESS in 58m 19s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6a52e3b968384b3abe9e44c100eb27d5 : FAILURE in 1h 01m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b31931a498a7432eb9ec58f2c5781cea : SUCCESS in 47m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/796c1668c1704e7bb967b0534457b356 : SUCCESS in 1h 01m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a82f9de699f4a0c9382fcb35d6fa4c0 : SUCCESS in 14m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9051da56df154fe3a3a0a2aac2988efd : SUCCESS in 5m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ef8871aee7bc408f8863989a86cb3480 : SUCCESS in 22m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f65392a74f144cdaf44d30a3d26b526 : SUCCESS in 42m 19s","accounts_in_message":[],"_revision_number":2},{"id":"a679f37dde89cb7b3903cb8ed8db9967171eae3e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-07 16:48:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cd1f420e1adfa778fe8b2fa0f52209269f9f5faa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 17:52:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e553da82e1a2434890ba629a761d83df\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ceab3894d7742f8b1559ae4c69d814d : SUCCESS in 8m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/750c6a8a204d443a8d0fab2fa0c76216 : SUCCESS in 58m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8620e1e4fe6d43148beab3ad5f7930c6 : SUCCESS in 40m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d832a0fbda5e4dc5825412249f46b4d7 : SUCCESS in 48m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b777866941bc47aabcf9da76cbb99efd : SUCCESS in 17m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f387aa247e334f1a89cca0d7af8dd79b : SUCCESS in 20m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/419a6f786c0644e1bc05c5c84498ea8f : SUCCESS in 26m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b4c79fdcedf848238a99ac6a1ee52901 : SUCCESS in 9m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/eaecceff21e94034a27cd1065177a9fb : SUCCESS in 5m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/42563de459494e218f2c76b14df96b5c : SUCCESS in 20m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c222e456e2114d88bbf22a93912c4fc1 : SUCCESS in 18m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9ff9fa7992ca41fb874117c73667b182 : SUCCESS in 17m 20s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6d2f2dd3220941758d05a3e8162c353b : SUCCESS in 17m 39s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f6ec478a5b31484fa1fdbb11c871a4f4 : SUCCESS in 20m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f1d959baf90d417bb4315ee42865d278 : SUCCESS in 17m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c03bd7dcb808438ea8c3c3d259fd105f : SUCCESS in 19m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d132a17a3c554700a32efdea2fc1f181 : SUCCESS in 17m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e5c82f2cbfe647cfb20627845e846411 : SUCCESS in 14m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ce8b6a33289146e1ae3e93a49e4e6139 : SUCCESS in 22m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b86f753522ba4a1680fa95c6b87eedcc : SUCCESS in 17m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/76e6037103394dc8b4fd910125386969 : SUCCESS in 6m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/92a1c29dcc904813a942bfae77392ec3 : SUCCESS in 17m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4eff74fc040348708234a927230c2e1a : SUCCESS in 9m 15s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b659ed4ab6564e04908105e3c20a4102 : SUCCESS in 1h 00m 36s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/2cca9096a1614630b3ced13abe25faf4 : SUCCESS in 1h 01m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c8d0bffe1a2642268f1500b82d1fe15b : SUCCESS in 46m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/14503658adf043da90248d408519e20a : SUCCESS in 45m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/200d5e78fdc1421cbf90b72b9e37ac8f : SUCCESS in 18m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/230e670e56f14550a33ab6a5d450e02b : SUCCESS in 4m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/da33f0dda6f54baf8705b0d788973ca7 : SUCCESS in 18m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f71224fbd08945a5987450b39a01db69 : SUCCESS in 54m 05s","accounts_in_message":[],"_revision_number":2},{"id":"aa74cca0834630bfce5d654eb100026bbc1eef12","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 18:21:24.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70c8de2bf5d345cb8906e791dc3f1892\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bc20a3130c6442f0bbfa19cc8d76e082 : SUCCESS in 14m 34s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f1d917f5119c469ebf28c995e0286b99 : SUCCESS in 15m 44s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f1f310d0fc2a4fd7a89387b5ba2b98b6 : TIMED_OUT in 1h 30m 52s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9c46ffe10acb4da59a3497a5e5b96c9f : SUCCESS in 16m 59s","accounts_in_message":[],"_revision_number":2},{"id":"87310655ab2cbb69c5d2a62dbe9e1f4b0c314090","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 19:43:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a15cd2881c7629f66203797f19911efbcaabee64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 21:28:56.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/886994 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"d5529c9253e1c911a4f9e576f0321fdec8d55bc5","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-08 09:11:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0e24c32512f9d6ace2064c8290106805a1d928aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 10:36:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7bce737add74d6da52a55c20877ad5a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ac59a3786554c5e97b580a050bb0c13 : SUCCESS in 18m 24s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5f8540fa8f834c45b797024002e61218 : SUCCESS in 1h 23m 40s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/542583ed597e4572a4353a29a2de1053 : SUCCESS in 23m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d0f9e9f2ea904160b03cc7e42dce9c91 : SUCCESS in 1h 07m 40s","accounts_in_message":[],"_revision_number":2},{"id":"40b9c69280a330025f368c4b1427a1848ad1a26c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 11:00:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98d1e3aabef740d3bd6ff4473fed8f56\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4baa19253b61478ea019911131f70927 : SUCCESS in 7m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/661f969694f84a73806cbd22dbe26bf5 : SUCCESS in 50m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e1465ab266614e8e9d6e0eb298e17dfb : SUCCESS in 51m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7cd74ec56d83446d8e5914a2913b5151 : SUCCESS in 48m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e4ec7e8fa9fa49ab8c8d143f8330d9bb : SUCCESS in 19m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8509c4d631b94347a7161f6970a33270 : SUCCESS in 18m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/35cd5b5810164c228c10b44b124016f2 : SUCCESS in 22m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4c8547eea57d45b8a3c6f6f8b662a680 : SUCCESS in 9m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5083cb561fe64c719328f36d73dc363e : SUCCESS in 5m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/846b0eb4ad034f7c8315200fc0b77d9d : SUCCESS in 17m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/95cbc24b25d94774a4d809a3ac51df06 : SUCCESS in 15m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6d26d681a1e546b2862348793df2baca : SUCCESS in 16m 48s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8a3bcdf6df7d421a9fc85c90c1f21759 : SUCCESS in 16m 26s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5459079eff614456902dbbcc05b0d461 : SUCCESS in 16m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ac450dd6289d4ccc95727894f3bb2b89 : SUCCESS in 18m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/67c7b46d2c42452ab779d2ef81857da3 : SUCCESS in 17m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/900a43a5681b40b79053e5707956275a : SUCCESS in 18m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2ee8b238dba74fd29ed1a989d6d3fba0 : SUCCESS in 11m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e267a8368aa242b2b814c52f4f860103 : SUCCESS in 14m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d3422bea4f104ebb8d280a83aa2b6e4c : SUCCESS in 14m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/74da09ee2d2846e5b8f39b16730940e5 : SUCCESS in 8m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ae40ef495ebb4d0aa42859e002ff1510 : SUCCESS in 19m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8031aa831a69421eaa1f51fb344c9dc5 : SUCCESS in 9m 16s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a303051e9bec47dcbc06a3ac126451b9 : SUCCESS in 1h 02m 25s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/454964b0949844a4813b30454b9cfcff : SUCCESS in 1h 04m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f9254fe2cfec403b8fb4ceac118c55a0 : SUCCESS in 50m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/983cf7b127e04827a3d6ad4630146476 : SUCCESS in 37m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78ebdc765f3f430db42b1829a2e3e844 : SUCCESS in 15m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f61f2a2850a04014acd37ecd4b1663f3 : SUCCESS in 4m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dfcf82c223f14848ad04a0dbed146f41 : SUCCESS in 26m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba0c9c3cc8e2489caa5b3d6268c6fec0 : SUCCESS in 1h 03m 47s","accounts_in_message":[],"_revision_number":2},{"id":"e314f22b14787c715a5d9b14fd430a34755c3106","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 11:01:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"aa0ee11fbf8c2ad4a4a31958c479dd8b5f392b97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 13:11:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef8b695d22ca402daee7d74eee8774e3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebac909e1f19444e8746588bd00220dc : SUCCESS in 8m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/c4ed9bd1de8f45d3af86898db1a5fc9e : SUCCESS in 54m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/87c685cf8e90412e87ee0b8bf0eb99b9 : SUCCESS in 37m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26325b7474c848838b5380cba8a011e0 : SUCCESS in 57m 08s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/7aaae85e603549e6bc73f47c6779800b : SUCCESS in 8m 47s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/6915f9f9add749468cffbad463eb4ec2 : SUCCESS in 6m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b3fc318836154bbe932151e318cd310a : SUCCESS in 20m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4ffb553f02144f12a268f5fa8f0480f1 : SUCCESS in 16m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a2c521b52dde4d5da918fa9097f38a9e : SUCCESS in 12m 54s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/01911b2b05184cb797776b1ff042f0fb : SUCCESS in 24m 23s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/990d6895de0247d9b01e244fb395b2d9 : SUCCESS in 17m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b244556a5dea4f029f917d0bb01e3dfd : SUCCESS in 20m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/669e95600ce748d6b7e0bf12faca6eaa : SUCCESS in 36m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0ea9734681934599872f1ce13d68dc57 : SUCCESS in 18m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ddb79aa0dd5f4caa9909045166b13e50 : SUCCESS in 15m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3463f2918e2b4c01aa8a1d9c1ac1c4b3 : SUCCESS in 16m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/07b78f64e08c4aab88e18c261fce9750 : SUCCESS in 16m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/eac2be5abe514915ae35e6fba8995f41 : SUCCESS in 7m 27s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c5f436461202402f9d969705a99f3426 : SUCCESS in 8m 05s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/170fae63ab29402bb8aa9ce851bf5199 : SUCCESS in 48m 57s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/da24ddba9dd347beb248eb4e685f783d : SUCCESS in 1h 21m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2cec32daaea54c6a9273e1117d42c939 : SUCCESS in 33m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dcd0cdb820a84402b06ef87bea0df93c : SUCCESS in 1h 01m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37bd6ddf15994676b8d2783ecac57751 : SUCCESS in 18m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a9e0ac56a344064846daecf100f9062 : SUCCESS in 4m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5cfa8f3e66d241f499c043735a6abaac : SUCCESS in 54m 07s","accounts_in_message":[],"_revision_number":2},{"id":"b432c73f84345eb9760e2d5c58e2f6bdf575cee8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 13:11:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0e51efbe0381b3fbd814f3d6857815c745acd37b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 13:12:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b4ed5f57b29472cb640532e7e7feaf3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2407585b55c4c58aab1188cdcc74cc9 : SUCCESS in 48s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/25e33fc3f50d43f98c1efd7f9937aa10 : SUCCESS in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/7a1dbcfa4d1f43ce97611f35bafca9cf : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fe09ef258157f00183eda593559dd728ea0e1676","revisions":{"c82c2dc9235621b536da59d9c51cc8d5fb25f988":{"kind":"REWORK","_number":1,"created":"2023-08-04 23:03:31.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/78/890578/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/890578/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/890578/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/890578/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/890578/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/890578/1"}}},"commit":{"parents":[{"commit":"24d2976f63191f778bcf351d601876bdf988dbc4","subject":"proxy: encapsulate Getter resp, node and parts_iter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24d2976f63191f778bcf351d601876bdf988dbc4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-08-04 23:01:32.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-08-04 23:03:12.000000000","tz":-420},"subject":"Save some effort when discovering used_source_etag","message":"Save some effort when discovering used_source_etag\n\nChange-Id: I35d682e58419f141429877ac6a6964132f7e658b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c82c2dc9235621b536da59d9c51cc8d5fb25f988"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c82c2dc9235621b536da59d9c51cc8d5fb25f988"}]},"branch":"refs/heads/master"},"fe09ef258157f00183eda593559dd728ea0e1676":{"kind":"REWORK","_number":2,"created":"2023-08-07 11:17:50.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/78/890578/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/890578/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/890578/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/890578/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/890578/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/890578/2"}}},"commit":{"parents":[{"commit":"f8a6dee622e47cec7111007069dd4ca74b56b49c","subject":"proxy: encapsulate Getter resp, node and parts_iter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8a6dee622e47cec7111007069dd4ca74b56b49c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-08-04 23:01:32.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-08-07 11:16:32.000000000","tz":60},"subject":"Save some effort when discovering used_source_etag","message":"Save some effort when discovering used_source_etag\n\nChange-Id: I35d682e58419f141429877ac6a6964132f7e658b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe09ef258157f00183eda593559dd728ea0e1676"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe09ef258157f00183eda593559dd728ea0e1676"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
