)]}'
{"id":"openstack%2Fswift~890805","triplet_id":"openstack%2Fswift~master~Icf77b9d6b8065814033059e33aca6e98a633dbd4","project":"openstack/swift","branch":"master","topic":"p-fix-attr-error","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-08-09 09:31:46.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"Icf77b9d6b8065814033059e33aca6e98a633dbd4","subject":"proxy-server: fix AttributeError in exception handler","status":"MERGED","created":"2023-08-08 15:26:44.000000000","updated":"2023-08-09 12:13:02.000000000","submitted":"2023-08-09 12:11:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"890805-p-fix-attr-error","meta_rev_id":"6a66d9996e0bc8555d5462d97c4dd3ba0e61534c","_number":890805,"virtual_id_number":890805,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-09 12:11:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-08 16:47:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-08 16:47:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-08 16:47:23.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2023-08-08 16:51:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-08 17:01:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f70765a69b8d77b6ee0e29ddd8d489c3d8a72528","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-08 15:26:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58abf09a648d80b519fa1235011b5c0065f9cd83","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-08 16:47:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3dc043b695a8b1b54de80a2c4bbff58daff8c50d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 16:51:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d5e5a47aba94db28e8cc86edcce0d72\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7bbbfb04170b485f95a8735ef7d464a7 : SUCCESS in 1h 22m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d59d406f9da64a0581b09ea884512b1b : SUCCESS in 17m 19s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/69d526e5dde14715a900d367f301fd7d : SUCCESS in 1h 23m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/75da9a949b434987bd2aceb755022d78 : SUCCESS in 20m 58s","accounts_in_message":[],"_revision_number":1},{"id":"dc9787ff913d68493c36c1379fa4fef0b1a47f35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 17:01:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3bb550c079341fd86f4c0791eca0232\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4153b2b259ce451681295f6bbcb9ab46 : SUCCESS in 8m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/6c69b0bd30724f29b9ed6a4f025bbc52 : SUCCESS in 57m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f45c6e327f294a5a86dcf4b745c0afac : SUCCESS in 46m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e62b794b62e34581ae76604c0102813f : SUCCESS in 1h 23m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cff3af9766f6478d9f6fbeac7ada561f : SUCCESS in 16m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c5c6d978be514a079aac6e6ad39b9fb2 : SUCCESS in 18m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/71cabb389693471e98e00c0db83a3d03 : SUCCESS in 33m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/df567492434146a0834f105ea2d94daa : SUCCESS in 8m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/304f59c77f4c4412b60dca9a94882fe3 : SUCCESS in 6m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/aeb6e8fe73034a87bcfe0f8ecc832d49 : SUCCESS in 16m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3c6b13ff6f954e539939dc5530777348 : SUCCESS in 24m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/da3589dafd2b4f22870868b870cce144 : SUCCESS in 17m 08s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a65a4112e4e44b84afd6a40b395e2b28 : FAILURE in 1h 26m 00s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/214f70f9d8db4c0489692ade9bb0c046 : SUCCESS in 17m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0aea0cbf4d9641aa960ff11f9596898c : SUCCESS in 18m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/144f4b2ee51e4e0685f6bb58eb56d697 : FAILURE in 16m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e3a6945603be45688a8ab3b55f3ebec3 : SUCCESS in 17m 17s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c0e1c8fa34f046daa22a6766f0eb1223 : SUCCESS in 15m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0ac33cc8b43a4d0ab7e376d6cac7ccf6 : SUCCESS in 1h 27m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e6d7c0b9c7d24c2db4ca09f31cbe7ce0 : SUCCESS in 16m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/079904bd920442b4b5108ef4f77268e9 : SUCCESS in 6m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2f22b6b3027a451698d29ba175629563 : SUCCESS in 20m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/da8442391c2448dfbee4376e739943d9 : SUCCESS in 6m 58s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/95a7c0a64c3c41dda6b7a5d34e4c3659 : SUCCESS in 1h 05m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d8e34f46f318444e80c777e6f2d008b6 : SUCCESS in 1h 25m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0fe146ed167d4cfbb67b25237fc46764 : SUCCESS in 1h 02m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/19a4af762a104c96918add33bead3961 : SUCCESS in 1h 06m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d33f033d817c4ae1b157dc98b29cecaf : SUCCESS in 19m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85b31d5c80ca44368afd3be960837583 : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f931b198c1d24847af2e968ea58179a9 : SUCCESS in 21m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ccb7fda499624e579aebd3900779cb88 : SUCCESS in 1h 02m 44s","accounts_in_message":[],"_revision_number":1},{"id":"52db9da39f18731318276bc340c1c93be6aff75b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-08 18:06:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3e244808c0a2359f0425996e0b29ece4bd7d8da0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 19:22:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5502bf1e0b9b48d88fc39c4ebc477847\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f864152c641422081b0edb63337b03b : SUCCESS in 9m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/fa1b27baa61c402888c58fe5a3193587 : SUCCESS in 52m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/95b1bd6ef6494d598b8e376f568e55ae : SUCCESS in 42m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/11ed17d8e99b4fba90a936573d19b8b6 : SUCCESS in 55m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/82b23743a8a04249aa1c6963434f0093 : SUCCESS in 15m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e36b128fadcd4ac6a51d395e95bfc49e : SUCCESS in 15m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/175152b884ca49ec9fc1239f451a3544 : SUCCESS in 25m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/94aa5eff4c66410ab3c86b1a92dc52d8 : SUCCESS in 8m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a22ea96d4e6f4973ac80daa168a00e26 : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/74e1ced34734406884b48d3e38585369 : SUCCESS in 17m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dc9812b87d044bebb63a01bb5c341b6f : SUCCESS in 12m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bf8b54d5026e4ae1812d3bec65a4f69d : SUCCESS in 13m 10s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/13895af4d9434a518f9c2adc406ebc16 : SUCCESS in 14m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/80c03c6132674d23b2ddbec1c6f5f45a : SUCCESS in 11m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/61906281694440b28e719e4791d4a9dc : SUCCESS in 14m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/363ce797532c4b38b0774d08ef695743 : SUCCESS in 15m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9ea8942960bd4161984b0934e20d0ff3 : SUCCESS in 14m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bf838b7d7b444663a86c27ff731f89ce : SUCCESS in 14m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/66e4358521ca47dc8a39c4dafdb7a010 : SUCCESS in 15m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/771f887996e44859bb1fb9d4435d8cbb : SUCCESS in 15m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/291422ee49964a8f9084395c964108fa : SUCCESS in 6m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7e301dae010044f7bef0283faa18a0fb : SUCCESS in 17m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4c173bee4d8a49fca4b239ba3ba8901a : SUCCESS in 7m 24s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d221f592db0247b1b796f71005aa8950 : SUCCESS in 1h 14m 45s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4c48995b5f52468a9a61fa95f99e3825 : SUCCESS in 1h 00m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/38495710478049f4bb3811470ec44e91 : SUCCESS in 50m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/199cdc82ffe6465da428cf8d5e916248 : SUCCESS in 49m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c047f63bb5c543068c61eb348f16f052 : SUCCESS in 16m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a53c06f45334df4996adb44caea34f2 : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ab211f945ea04a06b07a7b1592134d3f : SUCCESS in 19m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3eac3566b0a34759ad4663970862cf4f : FAILURE in 48m 17s","accounts_in_message":[],"_revision_number":1},{"id":"579e2e2a3c7b1e95c3a53d1c339a3d15e0f03b31","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 19:39:20.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40ef6e5ebb214d10a7fe7fec212319cf\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e5076ea3995145b4b339fd3dc0c6eb2e : TIMED_OUT in 1h 31m 21s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/677063dcf0764602a8cf66098b0885cb : SUCCESS in 13m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/db3d5140f208436e99f66a0c754096bb : TIMED_OUT in 1h 31m 00s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f48b8fc81f94776aad997cfd89512da : SUCCESS in 15m 50s","accounts_in_message":[],"_revision_number":1},{"id":"ced462235f3725f22a2566ee8c8344ebedd48403","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-08 23:47:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ed9a6160fa1c2aef8cdea93b37d725f1a2a8b6c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 01:05:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29f409dfce69454089a7878dce3c6b72\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9ad7c64487fb4e4686f9351efc0d3eb8 : SUCCESS in 16m 20s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a2c526de0f054fb2b2724e80821783ba : SUCCESS in 1h 15m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7bdd52cb2f8d4c3c9707288b11177eec : SUCCESS in 16m 21s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/55959a5d9d5e4ce28f7f66977d4a2057 : SUCCESS in 1h 13m 10s","accounts_in_message":[],"_revision_number":1},{"id":"2cde7696a7c1cd85ed5e867c2c7bb1df6194d135","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 01:12:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d6833a596264f79916ad135f28019ae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/328bc392aec549989998d61048e14da0 : SUCCESS in 8m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/328ab277e21e4f1e874293373628c1de : FAILURE in 48m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d7956b8bf1f745e2bbec0fef12af94b6 : SUCCESS in 1h 02m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/804e6ec6c81c4fde84e1c8c9258b79f8 : SUCCESS in 43m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/581ddae22ae14540a389468c03066047 : SUCCESS in 17m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/600d2b0fdc8446338420e359b9a65400 : SUCCESS in 18m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c55b00d5fee84bc8b1ab4d5c973b8333 : SUCCESS in 18m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a31476d5b26448dab440677d152cdba7 : SUCCESS in 8m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ecd18919bdfc4b97b189e37ce6707dc1 : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/15928d09028e458fb7c978fdae36b6cb : SUCCESS in 18m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7ce97e66b47542eba042fcd91e1d0a48 : SUCCESS in 18m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3ae75cf93b424c6d9d94e75ce112fd92 : SUCCESS in 14m 39s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5440d63ac63c4343be27b8feee88c3c4 : SUCCESS in 17m 19s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/44f89a38dc5d407fb3a585d639c35ce8 : SUCCESS in 14m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8158de806845491db2660dba7b0f3805 : SUCCESS in 17m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/236c354e744a4493894b7d3567ad9c39 : SUCCESS in 15m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2abd0af858af4fa0a7d3c5581ac88063 : SUCCESS in 18m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bbdfa2fa808a4d98bcf72ad1aaaa4244 : SUCCESS in 12m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/28ad304918b54b11b02d48ed43754533 : SUCCESS in 14m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ccc54486e2c7425283b7287c2ea3e5ea : SUCCESS in 18m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/20bdb93314a4429e958bfb42fcccd1f1 : SUCCESS in 8m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5114eed3011645f99e8edfad4e2a080b : SUCCESS in 18m 27s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cfe92db4cd5d421a80dd107e71058ab4 : SUCCESS in 7m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/37db383ab23b4c31b95b7dc5dd0127b5 : SUCCESS in 54m 09s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9c962f7776514d969ed0360656f9b412 : SUCCESS in 1h 05m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fb49139455c345bf873a60324dee8f33 : SUCCESS in 57m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c9849c08e61543de9452ba5523738bf1 : SUCCESS in 56m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65f28d3bf78c4dc1a841c881eb77182b : SUCCESS in 18m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/506701a928c54671ba9d817f1dc1c48f : SUCCESS in 5m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/49210ddbdd18415eae6e1c4e0e292eef : SUCCESS in 20m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e94c858522d4e52aac81faad75952fc : SUCCESS in 41m 24s","accounts_in_message":[],"_revision_number":1},{"id":"a6f28fd3fa974dd65a088841d7819c7c99674402","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-09 09:31:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8e832b4c024c170b8e84d0d7e817ea0f942b68b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 10:35:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c5daf30c9fe45ccb4948c5839a9a801\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/491e8e563e2b422ab0eabfd9f97f2ee0 : SUCCESS in 8m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/af2dfa81d8c34247b5a9c16c341d74d1 : SUCCESS in 52m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/218315fd07e64816a504c27d0bf1e16c : SUCCESS in 51m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab7def114de048319147026e11b2759d : SUCCESS in 50m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7b379d1964f5442bb41c0e8679172f8f : SUCCESS in 18m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4bb91a06701b4122a6bcc8bc860323cb : SUCCESS in 14m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d17871c28b1b4de4acf0b3f925089f11 : SUCCESS in 17m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0f934e92770b4c0ca583d46a3c95f57b : SUCCESS in 7m 06s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e9f37901331c4bbfb2841e9ec2483af2 : SUCCESS in 5m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3d806a3d367a4a0dbeb5d1711cdff50a : SUCCESS in 17m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3299dbcf44794085a64ffbc4ba3e815d : SUCCESS in 15m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ae7c4736394746d1be2ec3db464acf5b : SUCCESS in 15m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0d08bf9514bd4f0faf785597fc35e0a3 : SUCCESS in 26m 27s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9b851dd0e4e7428ab197869549adbd32 : SUCCESS in 25m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/43632da7aaf44983b825865efe3c14c6 : SUCCESS in 16m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9232d8c65c8b4270a7dc4550f2c52560 : SUCCESS in 18m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f0de0aeaf2ec4f82b49edc6c78e3b3a5 : SUCCESS in 27m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b02abc1b139a464c83eb59a7c72436d2 : SUCCESS in 14m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a6a0a6b3b92e49f1b9a6a67e47236197 : SUCCESS in 14m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/458cd775b0744d2690cbcadc3b5450c8 : SUCCESS in 15m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e34217a585f84044bb523a5d0cffd592 : SUCCESS in 6m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/df5fb897157c4fa2bdb1d1818cb9b71d : SUCCESS in 16m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a4b8e5cba92340ec83f31ac160bb7642 : SUCCESS in 7m 25s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7fd02c642e044ebc9137e163165b4a5e : SUCCESS in 54m 43s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/98b3358b62624615b2fb54eeb532edfc : SUCCESS in 1h 01m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/612bfecb7bff4c71bee57b85952072fd : SUCCESS in 55m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6e04530d43c54f788916c9754234a10e : SUCCESS in 47m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92806bdaf2c44a8bb5466ff326fcc133 : SUCCESS in 17m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62ef252cff7047afa63feaaff1d31b57 : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bd658c00ca29458ba93a74486a6f215f : SUCCESS in 23m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/34015967aea4458795ea9809b0752034 : SUCCESS in 56m 09s","accounts_in_message":[],"_revision_number":1},{"id":"d094ee9cab0780bb435f6ba1762ce14e13a5c34c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 10:35:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5245afcf31af858d1021b9bac2bc083f72a1fdf8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 11:04:29.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d89100cbc4384e3abd2b86a0b1166f59\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b164c24e86e4346a617e5047f20ec6a : SUCCESS in 15m 37s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7b8ad3694c6d4b0e836eb7aac8b6ae81 : SUCCESS in 15m 39s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/311efc2408204be8961ecab1f8256aaa : TIMED_OUT in 1h 31m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e522a414e04c4abd92f88033877a7c75 : SUCCESS in 17m 44s","accounts_in_message":[],"_revision_number":1},{"id":"67fdcc007db5157cd5ab5bf4be5d00990ff5b4d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 12:11:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b957ba1cdb384114b927af0bb6154dde\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/611e09e6018c417fa4b3516c43a3970f : SUCCESS in 8m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/909e9ca238694dcba32ad9e9e6b665e1 : SUCCESS in 53m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/19cceb88f55d4f27be02df5f54090eed : SUCCESS in 51m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/abab4589d111479b84a4c62a35a8f61a : SUCCESS in 47m 13s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/63bc262f2c024dde8e248bd9178de7f4 : SUCCESS in 8m 18s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/fbb465f68d454d9eaf1f4e8f4488cbae : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/132d4555ebaa4348ac00a41d08075d79 : SUCCESS in 16m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c3b4a500944a461f9d9571eac0e365bb : SUCCESS in 15m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9b59ae4454fb43ce81a117fdaf3b708f : SUCCESS in 19m 53s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ed052f20494646e98df6a83e7fbdc7ab : SUCCESS in 24m 51s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3ca5c053cedc4a04bbbd93a8a7674f62 : SUCCESS in 17m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c591d6a07ae54baaa5fbccec304dcd99 : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3b9de9382b414a3a9aeab3ee8d57dac8 : SUCCESS in 26m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/97d611c66bf4498ab4448d7aa51abecb : SUCCESS in 18m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0479c10561f94018ab15cf88ad0df328 : SUCCESS in 14m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/db07bfe7ace44afe832ea0e4544bfdd7 : SUCCESS in 24m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1d16dc06585f4a1cbdf8fce86afc80f3 : SUCCESS in 15m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f05818189e28469aa6896d084619dfff : SUCCESS in 6m 47s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/207d8b13c2b0494bb698f247e914771a : SUCCESS in 7m 54s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c8323acdf47848519c3c45ceb29b3d69 : SUCCESS in 55m 57s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/49b2e626f9fb43c8ad5524039eb4b32c : SUCCESS in 1h 35m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ede996d4051f4ddbbbab64645ac72fd9 : SUCCESS in 51m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/96b4d9d040364fa8bce1a3783b149ef5 : SUCCESS in 52m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1ffc5b77de74dbdb0d170ec903ba791 : SUCCESS in 29m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37d56cc893a74f0e82be415ffb4a29e9 : SUCCESS in 5m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e9208e40fea44e68843a539ce6baf50 : SUCCESS in 1h 05m 41s","accounts_in_message":[],"_revision_number":1},{"id":"e2bd6dc03097979d445527889ff707e8a5562813","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 12:11:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6a66d9996e0bc8555d5462d97c4dd3ba0e61534c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 12:13:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60bbe5880ee3405d852a3a59adbf08f8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3986cb73d9c4f8592960ce0b805e019 : SUCCESS in 52s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e560d4ea44ef475985de40598477f3be : SUCCESS in 34s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/890c1b1bb9d14b4bba1e66a1a4fc9f06 : SUCCESS in 31s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4a8879a3641540d10121569063aba806abb580e6","revisions":{"4a8879a3641540d10121569063aba806abb580e6":{"kind":"REWORK","_number":1,"created":"2023-08-08 15:26:44.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/05/890805/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/890805/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/890805/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/890805/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/890805/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/890805/1"}}},"commit":{"parents":[{"commit":"54f8dd0ff854c9cba511e81fe4e60bc98d8e99c9","subject":"Merge \"proxy-server: de-duplicate implementation of is_good_source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/54f8dd0ff854c9cba511e81fe4e60bc98d8e99c9"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-08-08 15:21:28.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-08-08 15:26:01.000000000","tz":60},"subject":"proxy-server: fix AttributeError in exception handler","message":"proxy-server: fix AttributeError in exception handler\n\nFix an AttributeError that may be raised during exception handling in\nthe proxy object GET path. GetterBase.node was removed by the\nRelatedChange.\n\nChange-Id: Icf77b9d6b8065814033059e33aca6e98a633dbd4\nRelatedChange: I76ea042ef4b3f5cc1caa4774e35d6191f4ca548e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4a8879a3641540d10121569063aba806abb580e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4a8879a3641540d10121569063aba806abb580e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
