)]}'
{"id":"openstack%2Fswift~891141","triplet_id":"openstack%2Fswift~master~I5c5d37ec1e0909befed6b9f7948ac5768bbcf1fa","project":"openstack/swift","branch":"master","topic":"get-and-head-part","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-08-21 16:25:18.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5c5d37ec1e0909befed6b9f7948ac5768bbcf1fa","subject":"wip: functests for swift part-number api","status":"ABANDONED","created":"2023-08-11 01:00:05.000000000","updated":"2023-08-21 16:25:18.000000000","total_comment_count":8,"unresolved_comment_count":7,"has_review_started":true,"meta_rev_id":"ba65e592794f0098b0822c1ed98e3c078da17ae9","_number":891141,"virtual_id_number":891141,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-08-11 02:10:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":-1,"date":"2023-08-11 01:09:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-11 02:10:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d983b32c41a7fc7ddb165311c9f3185bcf13d6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-08-11 01:00:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b7f02b8267bcec6ca16d5a0ebf430062172a3687","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-08-11 01:09:49.000000000","message":"Patch Set 1: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7f10f8b0ec2b5228124b7a3c4f12e0a2d2d9d641","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 02:10:11.000000000","message":"Patch Set 1: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/84ac17dfd2ce42549a932381e63a9f02\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/754366f860cf485aa3f65512a9ceb886 : SUCCESS in 7m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/f05e51f025d84f1499f74dc20d99fa95 : SUCCESS in 39m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c3b3d55e72e446009124b6323f718963 : SUCCESS in 34m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca9b5bb52f164889b0b72941ac4bd8b3 : SUCCESS in 41m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/68a8e116eec9431c88c6008926393590 : FAILURE in 16m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1cca245bede54f7a84916f9ccc06166e : FAILURE in 18m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/112dbb4c90524fff8a5b2e81667a1b96 : FAILURE in 18m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/285a4d354e41441eb35677ace392d47e : SUCCESS in 7m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0e77ba64e0654b23a883ae56de32adcc : SUCCESS in 9m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c34b86ceb30f4e0f8354affdb48a9fea : FAILURE in 16m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ed9b2ab7166847d993f621b739440a13 : FAILURE in 15m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3b23e8ef83d24f05b573d1885dce6311 : FAILURE in 13m 26s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f999388eb0204998916748982dab57d9 : FAILURE in 16m 29s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ce4b4fd6345b44808efe5210aeee313d : FAILURE in 15m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5f1a4996b143437b8960a01c41aa6478 : FAILURE in 16m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/652fd465f1fb40fb99adc13a31a8b68d : FAILURE in 17m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fd9d2b106e8444ae984debc161965378 : FAILURE in 15m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9aee05dc4e524bf3b9a2e86cbe539346 : FAILURE in 15m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/17cb9783be1540729e9fc3a3780244d1 : FAILURE in 13m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/77bd9573b6974ef09688d5fdb77ca95f : FAILURE in 17m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f9f9e44223af415493e13e2504154e25 : SUCCESS in 5m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f9a901a3de5d48ccaf263f261d5416ac : SUCCESS in 17m 15s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/34999e7e3ad444fa890fe1195806f6dd : SUCCESS in 7m 04s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/18342b247958445d91fc1807834882a6 : SUCCESS in 43m 21s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9c3e41e36b8f4af8a7e59f0a4145f9e6 : SUCCESS in 50m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/56ef4ea79a574a3ba73c00c1c987efe8 : FAILURE in 20m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4fbc962a81dd450da2f144ea1aa9d054 : FAILURE in 32m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c8326c5a2c04b9f81620adf579381fe : FAILURE in 17m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22b6a533ee10436dabd5c14d6c1e3d39 : FAILURE in 3m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2d7f9fcaece64f428dd7ff87c00fef7d : SUCCESS in 21m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/534ccb2efa7b4e29b38b736ea6ebb1d6 : SUCCESS in 1h 04m 08s","accounts_in_message":[],"_revision_number":1},{"id":"29ec94db9a89dae2c9651da5eb706681fefe1afc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 02:32:59.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf5e475b6def44e4a8ea38267ba49aef\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2b0fd623c4e0407b9111550f74c2505a : FAILURE in 14m 45s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f9be945457ce4fa3918ec891523b43d1 : TIMED_OUT in 1h 30m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/76a6fab80fab4e32b7f7ac81df331a83 : FAILURE in 17m 33s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ada5bf36f547480597857d8044343e27 : FAILURE in 1h 15m 42s","accounts_in_message":[],"_revision_number":1},{"id":"ba65e592794f0098b0822c1ed98e3c078da17ae9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-08-21 16:25:18.000000000","message":"Abandoned\n\nsquashed https://review.opendev.org/c/openstack/swift/+/888954/35/test/functional/test_slo.py","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"622c4429dfadebb398de0a1ad2ffdc1b80aafea9","revisions":{"622c4429dfadebb398de0a1ad2ffdc1b80aafea9":{"kind":"REWORK","_number":1,"created":"2023-08-11 01:00:05.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/41/891141/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/41/891141/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/891141/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/891141/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/891141/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/41/891141/1"}}},"commit":{"parents":[{"commit":"e2ca3c48a7bcf03ffc98ea81c3025db2d3972c51","subject":"slo: Support GET/HEAD request with PartNumber","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2ca3c48a7bcf03ffc98ea81c3025db2d3972c51"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-08-11 00:59:56.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-08-11 00:59:56.000000000","tz":-300},"subject":"wip: functests for swift part-number api","message":"wip: functests for swift part-number api\n\nChange-Id: I5c5d37ec1e0909befed6b9f7948ac5768bbcf1fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/622c4429dfadebb398de0a1ad2ffdc1b80aafea9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/622c4429dfadebb398de0a1ad2ffdc1b80aafea9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
