)]}'
{"id":"openstack%2Fswift~891242","triplet_id":"openstack%2Fswift~master~I6b06d4a4e4b49504f6378d4ac73844fa2faf62ff","project":"openstack/swift","branch":"master","topic":"fix-manifest-test","hashtags":[],"change_id":"I6b06d4a4e4b49504f6378d4ac73844fa2faf62ff","subject":"drive-by: symlink metadata does not leak to target response","status":"NEW","created":"2023-08-11 23:02:38.000000000","updated":"2023-08-12 00:22:00.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ed0e9ef8c3aeffe3fc0b59dade5846badd6c0360","_number":891242,"virtual_id_number":891242,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-08-12 00:08:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-12 00:08:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aae4bc57a61a5745c6097d9f03b5a7da7d724a97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-08-11 23:02:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca8d28ae5900314800f191c24486482df90c935d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-12 00:08:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88e270c213a941e8b6f8b8533a710f21\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad67c58c824343b89e5ff9d97beb4d31 : SUCCESS in 9m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d500491e841f4006a2f380135ca309cf : SUCCESS in 18m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8e5ef58187084e9b9ab17825885aef6f : SUCCESS in 18m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a8473258d5b743ac9da03249f8781702 : SUCCESS in 19m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/07fb04f1a87740b4bf19c6ac352a59ed : SUCCESS in 8m 33s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ec6b0c131eef4291b3ea6ea4b36273b7 : SUCCESS in 6m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6377301926ea421592ebac7e9a8df9df : SUCCESS in 13m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2cc859cb7b0b4dee9d9aff434769b97b : SUCCESS in 13m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/276e76aba8d64821a5ca9f7dc6a9633f : SUCCESS in 11m 03s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/507c5b74aa7d4348a7b69ba6ca501486 : SUCCESS in 14m 19s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9fd367f951794ace90806d294df68b5a : SUCCESS in 11m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/66126191fa844a179a32f914b6c73fd2 : SUCCESS in 16m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/da58ae5e5c7d4749b03641ab660d9158 : SUCCESS in 13m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3e1197c3e1974c83a692f178e2e53607 : SUCCESS in 16m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3d34126df1af40b7b38620c04db576be : SUCCESS in 13m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2c8b7f90c24746be9c9b24c6585a9f13 : SUCCESS in 15m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e3a7677b31d44ddcbaec30e9cf80bbd8 : SUCCESS in 15m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7ffe774455a9452c8800915c5e211c84 : SUCCESS in 39m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b4eed1bdb27e49f5823856c23b0778df : SUCCESS in 58m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08df22cc6ac5487480a929dfe3e6027f : SUCCESS in 16m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9994cb11f82f41ed94dd51b97f2154a0 : SUCCESS in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b6be092cd8254aea8e187f0917481c29 : SUCCESS in 21m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ed0e9ef8c3aeffe3fc0b59dade5846badd6c0360","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-12 00:22:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/745fda3040ed461999bcfd516a54b9e4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8fd4cdb150314e7db921868a8a9864e9 : SUCCESS in 1h 16m 36s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e7e6f71b168f4be587200fac56f3d24d : SUCCESS in 1h 13m 31s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b5528c30f7e14fedbac8c2c9f12d0a5d : SUCCESS in 16m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f6e7be6fc4294ed1915a611b4a011dfe : SUCCESS in 1h 02m 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d5b0bf388784d8914d245f22c029fdfe34a012e1","revisions":{"d5b0bf388784d8914d245f22c029fdfe34a012e1":{"kind":"REWORK","_number":1,"created":"2023-08-11 23:02:38.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/42/891242/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/891242/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/891242/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/891242/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/891242/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/891242/1"}}},"commit":{"parents":[{"commit":"a8b752fabe4f3db69e5d304b71f793aa10035b48","subject":"squash: some cleanup and more tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8b752fabe4f3db69e5d304b71f793aa10035b48"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-08-11 23:02:29.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-08-11 23:02:29.000000000","tz":-300},"subject":"drive-by: symlink metadata does not leak to target response","message":"drive-by: symlink metadata does not leak to target response\n\nChange-Id: I6b06d4a4e4b49504f6378d4ac73844fa2faf62ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d5b0bf388784d8914d245f22c029fdfe34a012e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d5b0bf388784d8914d245f22c029fdfe34a012e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
