)]}'
{"id":"openstack%2Fswift~891526","triplet_id":"openstack%2Fswift~master~I079315ebeb772eaad1bf190fefd132b4087b4897","project":"openstack/swift","branch":"master","topic":"fix-leak-tracking","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-08-17 12:08:36.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-08-17 13:31:28.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-08-17 13:31:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I079315ebeb772eaad1bf190fefd132b4087b4897","subject":"fix swob HEAD handling for leak tracking","status":"MERGED","created":"2023-08-15 19:43:07.000000000","updated":"2023-08-17 13:32:47.000000000","submitted":"2023-08-17 13:31:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"891526-fix-leak-tracking","meta_rev_id":"e2415db85464e5b23e342b6f5a5d7b80d000b0d4","_number":891526,"virtual_id_number":891526,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-17 13:31:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2023-08-17 12:08:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2023-08-17 12:08:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-15 20:47:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-15 20:59:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-16 12:21:57.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2023-08-16 12:23:09.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2023-08-17 04:39:47.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"9f6ebff30c66a1ceff5ecda89de1c818ba4e6105","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-08-15 19:43:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ad2dc6899a2af077076bcdf2ba1853ef00fcdd0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 20:47:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd3977378b244243ac424f81384f2ffe\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/04b297beb30544ab8844fa5a0d2a2020 : SUCCESS in 50m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cf24d8565b8b414a950fb1c8bc5724c9 : SUCCESS in 14m 08s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/663bfacae45846739d2bef79421c15e0 : SUCCESS in 1h 02m 32s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/950706d895d4445ea4745d37292fdcaa : SUCCESS in 17m 11s","accounts_in_message":[],"_revision_number":1},{"id":"667170fc4954cbbac2adfa8aaf3dcb18402bb65a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 20:59:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edc8b2a3ee1e4b9eab554bd8d969f4a5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50937496a8fd491bb47c362c165117a7 : SUCCESS in 8m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/ae28d988cb434254b5cf57ecb9128299 : SUCCESS in 53m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f4e23e76772e4b3aaecf18f41568449d : SUCCESS in 36m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/55fd611b545d402fae44794917933e3c : SUCCESS in 54m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f23ec4584e13435c82df048ae4febada : SUCCESS in 15m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7c9cccdf27eb4db6b2118b8c66ba25c8 : SUCCESS in 18m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/49c15ccdc53f4e2c9f099424ff314d6b : SUCCESS in 18m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d1b713292d7249788c7eee710b76205a : SUCCESS in 10m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0d9ce00b96bc4d0d80d07ffadb3ac950 : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/59659dd138a64fdabe8eb4b70d020b12 : SUCCESS in 17m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/01d66a2909ff4d2caae624690a722beb : SUCCESS in 16m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8feb9bd336294bce9240d19fb09e7d7d : SUCCESS in 16m 19s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d491a7d0500143e4ac108948eea6005b : SUCCESS in 16m 51s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f66c4e7ba39c4771bda86f4445a4b5b3 : SUCCESS in 13m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/76df6e203a844feca3e4a07e82cccf1a : SUCCESS in 16m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/33aaf60bdb524233a4e00f762c29d417 : SUCCESS in 18m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/75c78c18cb2a4c9d9c2c27b3f2f52b9c : SUCCESS in 22m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b4d8a6d9aed640e497497f39daab7af6 : SUCCESS in 14m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/59c228945c98456f91d82ec93d4e87f5 : SUCCESS in 14m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b54e20591a334bbabb6770a06c70f131 : SUCCESS in 19m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/403c50eefcb14c78832284d8dcdeab19 : SUCCESS in 8m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ce434d46d4864cf8959e7b297e9b6f19 : SUCCESS in 15m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2e09dc58a5d5459c8618d676b5c89851 : SUCCESS in 7m 21s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c159c7d07b72420597b2caa15eb5756f : SUCCESS in 56m 12s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9651a9f7588d4bbaabac6dff0f79e826 : SUCCESS in 48m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ac0cd756fd1a4fac9bd0745cbd932a71 : SUCCESS in 34m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/098c281768c64f6ca5bc3f1038073aa7 : SUCCESS in 50m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1938f393675049ee9b586738e9c4fa76 : SUCCESS in 14m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53d396d666a240c39dea0b30442241b3 : SUCCESS in 4m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/10f7526e5c214aa8bf86a3fae956b217 : SUCCESS in 17m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0767f594c35e4c05977cf607c2d2d9cb : SUCCESS in 44m 29s","accounts_in_message":[],"_revision_number":1},{"id":"54e259288a3bc48e8e339427220d8bb6aa2badc7","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-16 12:21:57.000000000","message":"Patch Set 1:\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8891ea78eb094f3648a3dbd51c695d63aba4f2e3","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-16 12:23:09.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"39c02895811b1a43ace030154f500ab90fc5854a","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-08-16 23:54:58.000000000","message":"Patch Set 1:\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a5d8396c12389b5340b0f383e0784855dda4c68a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-08-16 23:55:05.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f4b7c2c57e6d83fc8b385474601163baeaf73f21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 01:01:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f76aabf6acca4099a6031decf7cbd149\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35f8546de22a4ee1a723c0d409d4db6e : SUCCESS in 8m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/cb780665a48949e5b6b4a7cec6f3bead : SUCCESS in 59m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7ab47149afcd4502b99b397590b04936 : SUCCESS in 33m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1cd10af5a4e6476996d9904461b964fe : SUCCESS in 46m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6698c30443b34f7c95ee8871eb4c376e : SUCCESS in 17m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fd128eda6e874102bd245c6a8865c1d6 : SUCCESS in 17m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a8be323adb304cddb64d2fae34b6a772 : SUCCESS in 19m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/84b784dac2b14864aa8af90790b50c8b : SUCCESS in 7m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/20b30b3c820a4d47a8f969ad1817d6ab : SUCCESS in 4m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7915eb0f68a346ac94835cc0e5226d42 : SUCCESS in 21m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c52a607a90a64d6faaabb9e025153f7d : SUCCESS in 17m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/952c0e8fff124f6a860e8a520cd2fee4 : SUCCESS in 17m 50s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e44a05cdf02a4d70868544c8bffe79e2 : SUCCESS in 21m 01s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/dfc77b4475694d65b23de27f4e95a568 : SUCCESS in 18m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f9e88d1ee9354a2493f65ae060cb200c : SUCCESS in 19m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ba595eb79b5d4317b25b8c006bca930f : SUCCESS in 19m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/54dc8d7f4f8e44ee8054c1c479861ff9 : SUCCESS in 17m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/250fe82bed624abcb8fe4135192b8841 : SUCCESS in 17m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/56a2624a51d94a1a882a2f36c9dc950f : SUCCESS in 15m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c54df236e85d4e348cfa8f72c9f9eaaf : SUCCESS in 19m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e0936266eae2405a9ed70b2641a5a3a6 : SUCCESS in 7m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5abc6a7a82954b5da63c635acfad7df3 : SUCCESS in 15m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8d4c787a35424b2288a7b862f7805674 : SUCCESS in 7m 24s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fdc13b6b8b044036a4b99787098d0136 : SUCCESS in 53m 46s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/843f79007b04481d9796f10ae3f6a9ea : SUCCESS in 49m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ea327efe8b964e719e8f8fe39cc80897 : SUCCESS in 38m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/de4ffe6d078c4bc7a5aeb3ac91c7932f : SUCCESS in 52m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c25469ecaba74a899fb99ec55b8ef1bd : SUCCESS in 18m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dca9198e22d4dbc9e6a977f878249fe : SUCCESS in 3m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b8619017c83745018df0d814d4c710a1 : SUCCESS in 18m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bbedd94a83da40d49d7be0e4e7a2016e : SUCCESS in 1h 00m 45s","accounts_in_message":[],"_revision_number":2},{"id":"20e03ec27f80d9fdcc3a91551ef825a89a553dd8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 01:14:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56c8b021116d410280e8273b4b01fbc1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b10006895334698a470703c2c7325a4 : SUCCESS in 13m 29s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/830e90aca59e4c7a8c9523dc73dde566 : SUCCESS in 14m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e2ae244cdeb48f3800accf8c144388b : SUCCESS in 1h 18m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2fc3ca7a7c8641faa82d6c927ffa15d1 : SUCCESS in 17m 00s","accounts_in_message":[],"_revision_number":2},{"id":"189b1c0462f7771bedfd95a8c0f559c736feae51","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-17 04:39:47.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ae557fe9a268f902623c0d1f6bdcdcff6d69175b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-17 12:08:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a5428cd8cba19a6af48ee531c6f5086e6d077035","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 12:09:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1da824368e7440f77d846b1ffe5bc504d258a739","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 13:31:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81547e86ac81458089d5ac3f40a43212\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ea5d04f0d2c4e2d97fb6ab3a203cab6 : SUCCESS in 9m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/8c6e3d5ab9124a62842c22752ab1357c : SUCCESS in 1h 11m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a9aee61e46694f92946da5d153075086 : SUCCESS in 55m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eacb880319ee4f66a10b03cf830a4f4c : SUCCESS in 1h 06m 37s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/3c27bb6793244dfa80ccf40a7abe7d3e : SUCCESS in 11m 09s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/e653ce8aa7df41f083aee420d1c06ead : SUCCESS in 7m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8ef843070d95413b9a6ddab510986e85 : SUCCESS in 17m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b7f133ec03ba471baabb2ca22ec4afa1 : SUCCESS in 16m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3d401bbe344840429e3961d79434448d : SUCCESS in 19m 12s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/604190167dc746ea864154342e8b0a40 : SUCCESS in 18m 39s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9f7ffa8831744156b36ba1f7514f46f7 : SUCCESS in 13m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0ad9741dc0154cf18d803c5d71f4ceaa : SUCCESS in 17m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/92dfc6a77ccf4dd982f6837cfe883e44 : SUCCESS in 30m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9c8445d0b1444d06b6cc87b0f86265ca : SUCCESS in 19m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3428267e6ac0456194eb4a4bfe25ec63 : SUCCESS in 14m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7b666ff1c638443aa59def928b5e665d : SUCCESS in 14m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1fedad94326741c7a5ab64c1858200ee : SUCCESS in 18m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bb3d48668ba7410e91081acc6b294f19 : SUCCESS in 9m 10s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2d5abe6d05074874be5f66af9778f520 : SUCCESS in 8m 33s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/41200639639342d281727733df92e34d : SUCCESS in 58m 50s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/bf7dcf6902f143bb90d871c1db04d15a : SUCCESS in 1h 01m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/372095861a6b4056a413caa1596c0e6d : SUCCESS in 48m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fdc348cf2ebc4885b96bec5360723a80 : SUCCESS in 53m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23045d942b134553b1e0715d46644cb3 : SUCCESS in 31m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e16c909782da4da8ac837748da0e67a3 : SUCCESS in 6m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74566ce630a54ce3869e566455ae84c2 : SUCCESS in 1h 04m 32s","accounts_in_message":[],"_revision_number":2},{"id":"6fd435f06e6ddf4bd4a2049d3da828d740a97ab6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 13:31:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e2415db85464e5b23e342b6f5a5d7b80d000b0d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 13:32:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cfe4bce57384f62a7e1d526c9f556e2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2d8348ddc704fed9952e5b1321ab7a7 : SUCCESS in 56s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/cd372e4296934b4a941c6115436ebd5c : SUCCESS in 33s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/c6da8ec688d64e32b1c6f6f69ebc1ff2 : SUCCESS in 41s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9e065e2d23947f9f87fb7b68b5c3925419b8e16e","revisions":{"7ee91ba1bff33d89dab4b87e907d0fd30ce50bfe":{"kind":"REWORK","_number":1,"created":"2023-08-15 19:43:07.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/26/891526/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/26/891526/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/891526/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/891526/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/891526/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/26/891526/1"}}},"commit":{"parents":[{"commit":"cc59929412f2fa0271c2940afaa3fa3c6e9aff99","subject":"Merge \"s3api: extra unit test assertion for statsd metric prefix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc59929412f2fa0271c2940afaa3fa3c6e9aff99"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-08-15 17:55:21.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-08-15 19:32:22.000000000","tz":-300},"subject":"fix swob HEAD handling for leak tracking","message":"fix swob HEAD handling for leak tracking\n\nWe added a \"safety net\" in swob that won\u0027t let HEAD responses return a\nbody.  It seems like we don\u0027t currently have many unittests that both\nenforce leak tracking and ALSO make HEAD requests.  The way swob works\ncurrently it\u0027s not possible for even a \"well behaved client\" to get a\nclean assert with FakeSwift\u0027s leak tracking because the leak tracking\niterator isn\u0027t returned to the client to close and swob doesn\u0027t do\nanything to close down and consume the HEAD response iterator it\u0027s\nthrowing away.\n\nChange-Id: I079315ebeb772eaad1bf190fefd132b4087b4897\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7ee91ba1bff33d89dab4b87e907d0fd30ce50bfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7ee91ba1bff33d89dab4b87e907d0fd30ce50bfe"}]},"branch":"refs/heads/master"},"9e065e2d23947f9f87fb7b68b5c3925419b8e16e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-08-16 23:55:05.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/26/891526/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/26/891526/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/891526/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/891526/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/891526/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/26/891526/2"}}},"commit":{"parents":[{"commit":"cc59929412f2fa0271c2940afaa3fa3c6e9aff99","subject":"Merge \"s3api: extra unit test assertion for statsd metric prefix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc59929412f2fa0271c2940afaa3fa3c6e9aff99"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-08-15 17:55:21.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-08-16 23:49:35.000000000","tz":0},"subject":"fix swob HEAD handling for leak tracking","message":"fix swob HEAD handling for leak tracking\n\nWe added a \"safety net\" in swob that won\u0027t let HEAD responses return a\nbody.  It seems like we don\u0027t currently have many unittests that both\nenforce leak tracking and ALSO make HEAD requests.  The way swob works\ncurrently it\u0027s not possible for even a \"well behaved client\" to get a\nclean assert with FakeSwift\u0027s leak tracking because the leak tracking\niterator isn\u0027t returned to the client to close and swob doesn\u0027t do\nanything to close down and consume the HEAD response iterator it\u0027s\nthrowing away.\n\nRelatedChange: I168e147aae7c1728e7e3fdabb7fba6f2d747d937\nChange-Id: I079315ebeb772eaad1bf190fefd132b4087b4897\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e065e2d23947f9f87fb7b68b5c3925419b8e16e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e065e2d23947f9f87fb7b68b5c3925419b8e16e"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
