)]}'
{"id":"openstack%2Fswift~891878","triplet_id":"openstack%2Fswift~master~Ifed13765e00ea760a403a0fd11ef54aa3c962e5c","project":"openstack/swift","branch":"master","topic":"p-drain-and-close-cap","hashtags":[],"change_id":"Ifed13765e00ea760a403a0fd11ef54aa3c962e5c","subject":"Add min_bytes arg to drain_and_close","status":"NEW","created":"2023-08-17 17:52:26.000000000","updated":"2023-08-17 20:19:38.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":6,"unresolved_comment_count":6,"has_review_started":true,"meta_rev_id":"4d75798936e4b4e6cf8577374a8d456fe436ec34","_number":891878,"virtual_id_number":891878,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-08-17 20:19:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-1,"date":"2023-08-17 17:58:16.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-17 19:31:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-17 20:19:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5803ca52c2d1a6ccdf4950dcb39bfc77e9b2b545","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-17 17:52:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30783254d32ca753d6e1aa8aaf3429f730998abc","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-17 17:58:16.000000000","message":"Patch Set 1: Workflow-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7e3aac4b51ac5c6d98f1a83f943bfcce0103d841","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-17 17:59:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e533dbbc15528ef8f741a7158a7af4c2517eb95","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 19:31:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/160786317f72450693f3beb73e3dc88e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/80d9b07078ba471d8f3c23460b306641 : SUCCESS in 1h 01m 18s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/75862be3834d4c0197aa41501ac4a4a9 : SUCCESS in 1h 29m 44s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b20e2849119d4fa1ab6afdb8130391e0 : SUCCESS in 1h 16m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/be8917bb9a3d435ebe8f5ce7754a8508 : SUCCESS in 1h 21m 25s","accounts_in_message":[],"_revision_number":1},{"id":"4d75798936e4b4e6cf8577374a8d456fe436ec34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 20:19:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9089700b40c44120b78deff52dc01267\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/becb98fa9c234e018c0140dd7303b026 : SUCCESS in 8m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/68b52328bf1b49889f892e239878a6c1 : SUCCESS in 1h 12m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9e037e48e231449e924c59beb2d11859 : SUCCESS in 48m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c5aa186cba34cd389ca047d73b43c7a : SUCCESS in 1h 12m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e265a64fe1494a6d9e6d41423b2d36ea : SUCCESS in 16m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4f3eb40017fd4cf8a3f34fa2a4e46ab2 : SUCCESS in 17m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/971929a6e73948e489c57a728a831d16 : SUCCESS in 17m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e9b31c1132cc47f894f9b98eb0ac42bf : SUCCESS in 8m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/91818986656b45a785fff523d0add499 : SUCCESS in 5m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/748ae738967046519e0658e13474422d : SUCCESS in 18m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/86bffb37676043c8843f18520c29497c : SUCCESS in 27m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a2a909b3103e4cd7a2da84464078fee2 : SUCCESS in 27m 11s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/823bfb5c3cf04edb945b34f72b301a4b : SUCCESS in 16m 25s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/385e062514ca45fea4865a23e2d36b58 : SUCCESS in 14m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/94de8dc80cd94a0687cb9fa10972e38c : SUCCESS in 16m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4a6aff9e234e4e01b04473869e8923a5 : SUCCESS in 15m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/67c11af16c194e10ba13cfdbdfeaef0d : SUCCESS in 17m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cc23b3bcc14d45ba982b47fae79f8d78 : SUCCESS in 23m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5286d12c0ec74cd987a186dc39359105 : SUCCESS in 25m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ed0eac02a9534c5fa024b9b90572eacf : SUCCESS in 24m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1ee3f28535f246e891ad685a3680c098 : SUCCESS in 7m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7b3356ef18774f11b32189505519c3c5 : SUCCESS in 24m 40s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/27937b48d8dd4d2aba13eaee631945a5 : SUCCESS in 7m 48s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2752b03722044db7b1d7e30584f020f4 : SUCCESS in 59m 44s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/cb74b4c0dd634b95ac67a90e9301fee1 : SUCCESS in 1h 04m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/887a917fe9cf4311931f5d27e857e4d0 : SUCCESS in 44m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/431197e75557468e8b93f1080e1cae00 : SUCCESS in 52m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2839bf338bf74610adca9a1e6f2ef537 : SUCCESS in 18m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6bdd3fe1f694e519f12c60087d660d9 : SUCCESS in 4m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/68a461245a944bbf82976e30b7c0144d : SUCCESS in 28m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/28d6f22ea53f49ce98eb3e02fc60b885 : SUCCESS in 59m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"30764add008d96b61844699da5773182affa13e3","revisions":{"30764add008d96b61844699da5773182affa13e3":{"kind":"REWORK","_number":1,"created":"2023-08-17 17:52:26.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/78/891878/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/891878/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/891878/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/891878/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/891878/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/891878/1"}}},"commit":{"parents":[{"commit":"9e065e2d23947f9f87fb7b68b5c3925419b8e16e","subject":"fix swob HEAD handling for leak tracking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e065e2d23947f9f87fb7b68b5c3925419b8e16e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-08-17 17:51:19.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-08-17 17:51:19.000000000","tz":60},"subject":"Add min_bytes arg to drain_and_close","message":"Add min_bytes arg to drain_and_close\n\nWe may want to drain_and_close an iter, but with a cap on the number\nof bytes drained.\n\nChange-Id: Ifed13765e00ea760a403a0fd11ef54aa3c962e5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/30764add008d96b61844699da5773182affa13e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/30764add008d96b61844699da5773182affa13e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
