)]}'
{"id":"openstack%2Fswift~891977","triplet_id":"openstack%2Fswift~master~If17e10d309826b815d2b5b470a8bff071f5c4e87","project":"openstack/swift","branch":"master","topic":"fix-manifest-test","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-08-21 23:00:35.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-08-18 19:24:51.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2023-08-21 23:00:35.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-08-21 23:00:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If17e10d309826b815d2b5b470a8bff071f5c4e87","subject":"FakeSwift: use env[\u0027PATH_INFO\u0027] to index uploaded objects","status":"MERGED","created":"2023-08-18 12:35:15.000000000","updated":"2023-08-21 23:01:32.000000000","submitted":"2023-08-21 23:00:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"891977-fix-manifest-test","meta_rev_id":"aacf9a99c9ba3b188c6f05e59f454ac074e0772d","_number":891977,"virtual_id_number":891977,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-21 23:00:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2023-08-18 19:24:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":1,"date":"2023-08-18 19:10:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2023-08-18 19:24:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-18 13:02:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-18 13:40:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-18 19:10:26.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"REVIEWER"},{"updated":"2023-08-18 19:10:39.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2023-08-18 19:10:39.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"8b170e49dbcffed331479d0a582d841d56281826","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-18 12:35:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"25c9ff0a7c4deefdaf802c8dbcfc373ec2d81d48","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-18 12:36:39.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"54122fff168d3ebb0f29d3045a4279e9f819cddf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 13:02:58.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a626dfa08715486b8188057d11955f1d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e9b88a29658b4ceba496d0519e83642c : SUCCESS in 20m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4fc7473e0bf24878a256976b04d1b3f5 : FAILURE in 24m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/844ac15869634a52a459b1b993365df4 : SUCCESS in 26m 22s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7d106a814ef0448a99e7c93631147167 : SUCCESS in 24m 09s","accounts_in_message":[],"_revision_number":1},{"id":"143ff42536faf9b7e9484205ed0bfbd813188ac5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 13:40:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0537f3f7978c4ab0b83342a45e43a884\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab05aadb6c5b44a58cda4bf20b92be30 : SUCCESS in 8m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/98bc4f0bec0d4932b9f8683823d9e9a6 : SUCCESS in 27m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/db6f4966168e4f5e8f05d76cb2e48de9 : SUCCESS in 18m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d90c5668fad949dcbadcffd2f767227b : SUCCESS in 18m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c3dd88049e3b48bda9125608add2bcac : SUCCESS in 9m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8a483f23216e4849afb150c9bd2e1cce : SUCCESS in 5m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/65b9c23908c24711bb09ebe4ab60bd76 : SUCCESS in 13m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9a528773042c4fac86927d02cdaa6bb6 : SUCCESS in 17m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d47d3e078cb14428ba7a7267e6a5fb96 : SUCCESS in 17m 44s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2d4a2ea3bfbf4739a814ebfc4f07d96e : SUCCESS in 17m 01s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/830e12b2787a439ca04cee837672212b : SUCCESS in 29m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e44a4955fa264507bc4a9549f479b122 : SUCCESS in 17m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f86e214433ef4a6e965ab8c6496360b9 : SUCCESS in 16m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cfac3b04c8b9475f884fb6b79e6c536c : SUCCESS in 19m 17s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/566365a40eee4fa78fbccb4095d71b44 : SUCCESS in 19m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3709b6213f01429ca4068d82af7539a1 : SUCCESS in 11m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ccebe0eff5a041b0aacdf348207fd53e : SUCCESS in 18m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9ef2a13b11df412d9c9d86bca19094c5 : SUCCESS in 51m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6a3b88bbf86c4b27a36f7d3feacf0f83 : SUCCESS in 52m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82e5100de4cb4726a12314de00782fa4 : SUCCESS in 30m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c43469046a554a8ab0325a4d7f4a1adc : SUCCESS in 4m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ea4833951cbc495ca67b5e4fc5dc055c : SUCCESS in 32m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e9c31633bb12460545f64c63d42ab8ceb615e2b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-18 13:48:39.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d675cdf347d61ac11d3029f3ad11441bb582355e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 15:00:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec0281945ed94d6bbb059a06d2cd4dbd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90a52ba75b844b4e82056b727719072e : SUCCESS in 9m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/af73d62f293640c9b8a050527ca829fe : SUCCESS in 17m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6bc34c57a5394d83b8e881bf18b56716 : SUCCESS in 19m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/46c438d805464ff9971eeda50e8518a3 : SUCCESS in 19m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5d195db155a344e2bf31719fb618363e : SUCCESS in 9m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f28ed2c7af2b47dbad98851c54e44835 : SUCCESS in 5m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4d0bd10573354424b206788379434f6f : SUCCESS in 18m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/43dd7f749b79428581edb74cbab9d2b3 : SUCCESS in 17m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8a20d77a97644aefb83a4de63e538a5e : SUCCESS in 17m 14s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/34551fef07a1494fa15f45bc338db8a2 : SUCCESS in 16m 36s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/57046be4f29640dbb5c350e7564c75a0 : SUCCESS in 17m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3aa4187cb63a452093f87af92861bed8 : SUCCESS in 17m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/91cb88480ca747609a09ac9957080f71 : SUCCESS in 17m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2c68869767e14a8e80ff81ecc25d65e5 : SUCCESS in 17m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1dfe7d4790d84cc9890c7e2ea675ddf3 : SUCCESS in 13m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/908ce4d315a44e08badfcf3c3912ea10 : SUCCESS in 15m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/18f10c6224c649d482a8dd2a26b03d67 : SUCCESS in 14m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9d001b3ff76c4cab98e828bda3cf0743 : SUCCESS in 54m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c064db24ed6c49338dc8d87d9c599023 : SUCCESS in 1h 03m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7dd4e7b62a7437d841b26adba8ac17f : SUCCESS in 14m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a7cccbac5114766bf48d90e268ac70c : SUCCESS in 5m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fb0c729daaf84457a9660904f2021a98 : SUCCESS in 26m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"86a0f5b13e4ee08214a8a09b8d0eff78bee749ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 15:47:57.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/546c07cb9a464ba394c69904ba5e4d84\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/92dd2476fc2d4f40a6c596e5d330af2c : TIMED_OUT in 1h 31m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a5c1a6a812794c0d9ce6cb6bba162955 : SUCCESS in 1h 10m 08s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b2c9bb241bce485e9ca020521257e929 : TIMED_OUT in 1h 30m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/356868462aea4f8bb4449c21cc03dc72 : SUCCESS in 1h 04m 43s","accounts_in_message":[],"_revision_number":2},{"id":"b427ea845bd2ed17ad7490fb6dbe15da742c6dd9","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-18 19:10:26.000000000","message":"Patch Set 2: Code-Review+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e8d321b55fed738541e008fe70a27e0a1a8d0dc2","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-08-18 19:24:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ba948490e02097dd0a9aafa342807047726d028c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 21:57:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"513e5bb8c64e01d24ea608ae204f89f08be61f1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 23:00:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a1cae25517a4d5b85102e0edb0f7d06\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32af0c24c9554c1e9757840b979dfeb6 : SUCCESS in 9m 34s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/cf64a49ae8764279a47ea0487f3b7ec0 : SUCCESS in 8m 34s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/8f2fb51e6ccc4810b154e5b29af77a3e : SUCCESS in 6m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/50d1e52137e649f3b755baf49b4a06d2 : SUCCESS in 21m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c682092267ef41a085ef5a2bdeec8e2f : SUCCESS in 16m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6451f6aa36ec4242852b2fa3d0ff56d0 : SUCCESS in 16m 45s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/51b7b711374e498a81f7752c074b2f73 : SUCCESS in 17m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/add482ca14c9498e90a39b27c6c49b99 : SUCCESS in 15m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/388c2a3c45e64a69983a854dd1161d55 : SUCCESS in 13m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6d1034626ec3444ba2a496b22ae5c5f4 : SUCCESS in 19m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/15f2f7e3589e4ad2adf4f02d4518a9f7 : SUCCESS in 16m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/07bb8c8151c141fcb25d40c4911d8d21 : SUCCESS in 14m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/be8008177b7f48fdacd361a6edc52f08 : SUCCESS in 14m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3bd92c6c8bab491da2a4d1fe38445807 : SUCCESS in 12m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/daeb2af3996046f7bbdde161b01c3d63 : SUCCESS in 8m 02s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/57ba5c08b3144f77a65ca0ab285d6d16 : SUCCESS in 6m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/11f5ea9196174955b34e009520709c47 : SUCCESS in 56m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8e3d3d7d9f194d59b4ef2daff0cc495f : SUCCESS in 53m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd4ac5a611cf4322b80bf8673dde442a : SUCCESS in 22m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9ff9af827ab481d99c78ddb7791823c : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":2},{"id":"d4f857e0456e7749b5f129c2e23dfb7ed55bfa8d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 23:00:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"aacf9a99c9ba3b188c6f05e59f454ac074e0772d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 23:01:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f13726295df4943aad26bd51cf37c38\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/682d75ac8c014476990648bd21ee0c52 : SUCCESS in 38s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f3b49da2037446c49ebc835e06af4a67 : SUCCESS in 34s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/ce928cbe0e8d4df7b00b14462d6fb81e : SUCCESS in 27s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8f85e27c27302ae623e4c0811515858f8d340603","revisions":{"7f435863659ad045d224f19288883b919f2df83d":{"kind":"REWORK","_number":1,"created":"2023-08-18 12:35:15.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/77/891977/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/891977/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/891977/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/891977/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/891977/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/891977/1"}}},"commit":{"parents":[{"commit":"7a3eed9c4b5d0efc5ac14be07ea85a10615f5ce7","subject":"tests: fix FakeSwift HEAD with query param","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a3eed9c4b5d0efc5ac14be07ea85a10615f5ce7"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-08-18 12:31:01.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-08-18 12:31:01.000000000","tz":60},"subject":"sq: FakeSwift fixups","message":"sq: FakeSwift fixups\n\nChange-Id: If17e10d309826b815d2b5b470a8bff071f5c4e87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7f435863659ad045d224f19288883b919f2df83d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7f435863659ad045d224f19288883b919f2df83d"}]},"branch":"refs/heads/master"},"8f85e27c27302ae623e4c0811515858f8d340603":{"kind":"REWORK","_number":2,"created":"2023-08-18 13:48:39.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/77/891977/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/891977/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/891977/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/891977/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/891977/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/891977/2"}}},"commit":{"parents":[{"commit":"746daad3376f43d66c8a48ca1079086555f82136","subject":"FakeSwift: add tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/746daad3376f43d66c8a48ca1079086555f82136"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-08-18 12:31:01.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-08-18 13:48:17.000000000","tz":60},"subject":"FakeSwift: use env[\u0027PATH_INFO\u0027] to index uploaded objects","message":"FakeSwift: use env[\u0027PATH_INFO\u0027] to index uploaded objects\n\nChange-Id: If17e10d309826b815d2b5b470a8bff071f5c4e87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f85e27c27302ae623e4c0811515858f8d340603"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f85e27c27302ae623e4c0811515858f8d340603"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
