)]}'
{"id":"openstack%2Fswift~892361","triplet_id":"openstack%2Fswift~master~I9a22fc824c29b5518dee3f55d2f8fa822311fc02","project":"openstack/swift","branch":"master","topic":"slo-part-num-api","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-09-05 16:13:02.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I9a22fc824c29b5518dee3f55d2f8fa822311fc02","subject":"fix legacy manifest HEAD and simplify HEAD part-num","status":"ABANDONED","created":"2023-08-22 14:40:16.000000000","updated":"2023-09-05 16:13:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"effbd44d8cdea9a499a400ad6cceb4d38fd9f7e0","_number":892361,"virtual_id_number":892361,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-08-22 15:53:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-22 15:50:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-22 15:53:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ca28bc89205e430713bc7ef9d710f86d8fd7e417","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-08-22 14:40:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2aa5d7aba2c44146beb45ff4736cb98480b77ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 15:50:02.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01979eacfa7948b1bccdbe5f46abcfd1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d615ff403f3547aa9a4a056196254f36 : SUCCESS in 1h 02m 39s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e581c0ff5d224814a97f628553dc6dc1 : SUCCESS in 17m 36s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1d8472947a0240599ec19f9beb29b856 : FAILURE in 1h 07m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/426b0c10a1fb4f518c2fbc85f3f63ec7 : FAILURE in 19m 07s","accounts_in_message":[],"_revision_number":1},{"id":"7c2de09057d8e3f8d254274e69bdfaeeb2b3e4ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 15:53:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f4dc1e6c8e4422eac514b30567548cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05acc969a844453882778326aa34fe0e : SUCCESS in 9m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/3734fae93e80430388f7ecca87922e84 : SUCCESS in 49m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/053087f7b5914bb0a59a9e35bf75b04f : SUCCESS in 53m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c91c29d2990b4a65849b758ce8ad9ddd : SUCCESS in 1h 04m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1b5629cc994f4564b4cfa259c59f55ca : FAILURE in 25m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3e73acaac32143ff98166b1857f3d3e4 : FAILURE in 18m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7bbd5a5284004373abe7f0bf36d261d8 : FAILURE in 18m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c47a7c64305c492285c9a7890cb9487f : SUCCESS in 8m 43s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d697b15b1e644ae99987933a2e9eebdc : SUCCESS in 8m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0a20a8d2306f420da6dfc0ed35b52cc4 : SUCCESS in 23m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2493c913d41d486fb751e32874e3fe9a : SUCCESS in 19m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f3f7fa8e99164a0a8cc3fc7eade932f6 : SUCCESS in 14m 56s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ec032ccd184b4233bb937f633262eba1 : SUCCESS in 18m 31s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/09f21b94bf4446f6bd52a6dd231bd138 : SUCCESS in 15m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a8a55a161057414a96a66af8038577a4 : FAILURE in 16m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/51c6b24719ef4470a43b84fb7337e84e : FAILURE in 17m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/428325e8c2e446a6bc14852786171010 : FAILURE in 17m 17s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f7ee52c2e7d94e6bb8c2477b0745f19d : FAILURE in 15m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/89cfbb54f00c419bab4536f51f52c08a : FAILURE in 16m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/01bbb823ea6546bb8d124ed3251ffba5 : FAILURE in 17m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4038fa4482314dc0a7e7598f0a79ba8d : SUCCESS in 10m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9e4e743eef6e43af808aca5bd9cd54b5 : SUCCESS in 18m 48s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7f174a99381d42f692037aae5220f67f : SUCCESS in 9m 24s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/50d1de3a80f64d359fadbe8d5de3acae : SUCCESS in 1h 05m 52s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0da6a262020d4014a60c31b9a7ba0bd1 : SUCCESS in 1h 05m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/781a292f9e214788b0038e63db72e4ae : FAILURE in 27m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1bde889b14024f0eb6052bb03b8a3f3d : FAILURE in 29m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ac1b7e940f194f749ac12a2aa9cc4b12 : SUCCESS in 18m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d89f177c68b4261adcbb2a5edc2f02a : SUCCESS in 5m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e026c77448a04616aa98c00fd910bd13 : SUCCESS in 20m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ba5ca9958964466a5abc71c4dc5b286 : SUCCESS in 49m 00s","accounts_in_message":[],"_revision_number":1},{"id":"effbd44d8cdea9a499a400ad6cceb4d38fd9f7e0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-09-05 16:13:02.000000000","message":"Abandoned\n\nsquashed into https://review.opendev.org/c/openstack/swift/+/888954/47","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"34e6b6a39686f074ea1661994bfd685adbf5b21f","revisions":{"34e6b6a39686f074ea1661994bfd685adbf5b21f":{"kind":"REWORK","_number":1,"created":"2023-08-22 14:40:16.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/61/892361/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/892361/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/892361/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/892361/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/892361/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/892361/1"}}},"commit":{"parents":[{"commit":"6791e8420c07a129040a3186ea5f609af1782af0","subject":"slo: Support GET/HEAD request with PartNumber","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6791e8420c07a129040a3186ea5f609af1782af0"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-08-22 14:33:37.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-08-22 14:33:37.000000000","tz":-300},"subject":"fix legacy manifest HEAD and simplify HEAD part-num","message":"fix legacy manifest HEAD and simplify HEAD part-num\n\nChange-Id: I9a22fc824c29b5518dee3f55d2f8fa822311fc02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/34e6b6a39686f074ea1661994bfd685adbf5b21f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/34e6b6a39686f074ea1661994bfd685adbf5b21f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
