)]}'
{"id":"openstack%2Fswift~892717","triplet_id":"openstack%2Fswift~master~Id59846e508545caf56091ebe30557a44adff1d56","project":"openstack/swift","branch":"master","topic":"release-2.32.0","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-08-24 23:18:41.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Id59846e508545caf56091ebe30557a44adff1d56","subject":"Add rolling-upgrade job from Antelope","status":"MERGED","created":"2023-08-24 20:35:56.000000000","updated":"2023-08-25 02:21:47.000000000","submitted":"2023-08-25 02:19:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892717-release-2.32.0","meta_rev_id":"d4ddf79057ecec7c572d8c19b3dfac5b30adad9b","_number":892717,"virtual_id_number":892717,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-25 02:19:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-24 23:18:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-24 23:18:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-24 21:01:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-24 22:22:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9bdf3e37299f58c6a6338b757231b85f72b73580","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-24 20:35:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23ba7c8e348f4776bf8157c616b97665dce3b2b8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-24 20:36:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8d819da782f482dfcdff7e8457692f421aeea62e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 21:01:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d501f95843f4d4a8ef24866a3b59b0b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c417a02c5136449db1bbb2a8e4e3ad40 : SUCCESS in 19m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7c9080e385c04c37bc9591cce81f452c : SUCCESS in 20m 25s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c4269d43c7d34803bc2c18d4b68c1928 : SUCCESS in 23m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/68a289dc859747e5970fcfa5e711028a : SUCCESS in 23m 35s","accounts_in_message":[],"_revision_number":1},{"id":"726e7afae7ecb1fcbe4c21cbeb02ce776a6685d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-24 21:05:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f205cd5acb2fbaba04ab71a5f43ca66560001f42","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-24 21:06:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c956cc8ec46cfcd7f13341d5495eb9c6ea946db4","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 21:48:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab51ae8f222f4e959b4dba6cd604de20\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2e95d857c2884bd1a6d70f7bd1f3c099 : SUCCESS in 18m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/35bc50ec1e464eefb7249b8ff35b30cd : SUCCESS in 14m 21s\n- swift-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/97c20dda4d1c45d895935bfc1d6e5e8c : SUCCESS in 21m 05s\n- swift-tox-func-py27-centos-7 https://zuul.opendev.org/t/openstack/build/d8fec1999da5482cb133a8c13c468dd2 : SUCCESS in 23m 05s\n- swift-tox-func-encryption-py27-centos-7 https://zuul.opendev.org/t/openstack/build/8b393a8ff8a04df88a617f7c011783eb : SUCCESS in 23m 31s\n- swift-tox-func-ec-py27-centos-7 https://zuul.opendev.org/t/openstack/build/c020b595874a495480100ab63a8b3e78 : SUCCESS in 22m 59s\n- swift-tox-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/a47667ec2b794e92b31f513da41d2ce9 : SUCCESS in 18m 37s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/880f1c4714a84eb9af6e8d5401550969 : SUCCESS in 15m 16s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/58ec09f36a554a63a6b0e7ae19107a18 : SUCCESS in 15m 44s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/52932db24f2a4e4faf8e923eade5e329 : SUCCESS in 13m 40s\n- swift-multinode-rolling-upgrade-train https://zuul.opendev.org/t/openstack/build/ed4f5839f8454c9fb1b9163812483768 : SUCCESS in 17m 11s\n- swift-multinode-rolling-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/b3e8ae1e303a4f3985b22ba4d0861dd8 : SUCCESS in 20m 29s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/2cdac5349db949798194ee8649e2c554 : SUCCESS in 21m 38s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/abb20ec9cd5a4155a2f6887aa5370831 : SUCCESS in 22m 56s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/223b26f43ee54ffa932f1328ab073fd6 : SUCCESS in 15m 31s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/c2b8fc1251414b0c8537bc0b1a52beb5 : SUCCESS in 21m 30s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/4e3dda6d1759486a85ff8776da7335d5 : SUCCESS in 21m 41s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/6cdb2bbbbd824942aeeec2331fffc111 : SUCCESS in 22m 41s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/e995ebe84fae44ac86ff16768ab075f8 : SUCCESS in 21m 28s","accounts_in_message":[],"_revision_number":2},{"id":"3da24a770f52e0bf0971bd6c50ea5571bc00be52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 22:22:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6bd3a11d0b5f45ffa08e2dda5054d3be\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb3261a32827423882601c5f403644f2 : SUCCESS in 8m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/beb40a2de29b440b98591771318a95a9 : SUCCESS in 1h 00m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5af4f4bf4d8f4c85bb04965fe3e78c12 : SUCCESS in 51m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7fb7bd5181924650bb8b5477dcdc7d4b : SUCCESS in 1h 07m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cb941d105bd04cfd81304a0185fc5a30 : SUCCESS in 16m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/707cf34b06d64a3aab5bca33a00cbfdb : SUCCESS in 17m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8877359d06f145caa126338012717ee2 : SUCCESS in 18m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7e15181fd31c431690276fc114c4b245 : SUCCESS in 9m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bd684b862d19495b8237e6c9d499c15a : SUCCESS in 5m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e41f5fb1342b4ec18e6ae844e39a3ffe : SUCCESS in 20m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f188814796954b8db8ae620285cc303f : SUCCESS in 14m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e322161ac7ee44bdb45a93ce09e24f14 : SUCCESS in 18m 56s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/dd0b911670394d7eb438d3b382011a7c : SUCCESS in 17m 35s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3c7ca9ae599a453c87fad729369ae991 : SUCCESS in 15m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3fe8d8c5659a4b039f3f8329a65f7685 : SUCCESS in 19m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b47be6c457f64245a3488f6ad548b355 : SUCCESS in 13m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2d413ee718634100b370c111d4383e40 : SUCCESS in 18m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0f282330947f40369b38df2663d990c1 : SUCCESS in 17m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4d54c4ed674c47b29b989b84882cd045 : SUCCESS in 14m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4811d8741ce745049ac2602b47c1fb2d : SUCCESS in 11m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/528a9f59447c451cba0831a9a946470a : SUCCESS in 6m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b4b9ff714929400397c6991036cae875 : SUCCESS in 18m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e18d694995b14e118f8c68e9e13e40e2 : SUCCESS in 6m 49s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/55af06926759431d8e2d727266a74937 : SUCCESS in 1h 11m 02s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/2de24fbeab3540c5a82fd46e41f34ddf : SUCCESS in 47m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5a424211d4cf4513b110473901d41c45 : SUCCESS in 37m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/faee16d3ae2c496583d94ddf87f43fd9 : POST_FAILURE in 56m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b07611299c0f4563914119a0c9921ee8 : SUCCESS in 17m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8403d634edc54b4d87d893a58288e36b : SUCCESS in 5m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cbf3d11087324df081601708902d29bb : SUCCESS in 21m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b5825e437dc407882f776a73d899421 : SUCCESS in 1h 09m 42s","accounts_in_message":[],"_revision_number":2},{"id":"68fc4600cad0b76a2abf7036ab9cee859ce9e9e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 22:25:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e19465555664fa884c50b43c0c60d21\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2832f1b0674245638d1ba826a8e41b5c : SUCCESS in 16m 19s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4886b7e7d1d64d889ec4bf6b26ac35af : SUCCESS in 15m 36s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3071fc8360f4446fb0ff01034e8b3453 : SUCCESS in 18m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/03db255c8d8649ddb3aab1f497b3ddf6 : SUCCESS in 1h 16m 48s","accounts_in_message":[],"_revision_number":2},{"id":"da46ddaee7ab0eb551400e915400bf45de7b3099","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-24 23:18:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"44ed2d9ca2aba33d6a043247bc6bc02d0868bc90","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 23:38:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b5c4d784d99494cbfea7512456766b6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f1121b3d10704634af1c863c4d9060e2 : SUCCESS in 14m 50s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3b1c5b25f655416d8aa45e384503bbdf : SUCCESS in 16m 22s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/bba3870917844d6c95a734d4becdd345 : SUCCESS in 18m 41s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/75da3a20fb1b461881e7ef520359a17c : SUCCESS in 18m 02s","accounts_in_message":[],"_revision_number":2},{"id":"2a66357d59a990a6d8454a9af0548d72ad73e75c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 00:38:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c626d3868824b4e8f1ddf87daadd537\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01891ef774a349a99b888e5865415931 : SUCCESS in 9m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/2207ec3bd02d4e73972570ef3693a54c : SUCCESS in 1h 13m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1f475d9a7f04443ebb71914ee19162de : SUCCESS in 53m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9f2e150e8f1488886da6e3b92849a7d : SUCCESS in 47m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/da3fc314bfae4c4cb1e2524d58779878 : SUCCESS in 20m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e75ecb842f4742668c1e03e2edd07a69 : SUCCESS in 19m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/57b56852b74d4dfa8ad3bd9ca2fd92b7 : SUCCESS in 18m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/922f39ee64b54e70acc2476312cf08a7 : SUCCESS in 8m 43s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2414a83f5fd847bf9dc45a3fdb3f05e5 : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5065a1cab47f454b8ced9396232acda4 : SUCCESS in 17m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/916c22dd347d41bf82275e3c26259e4f : SUCCESS in 18m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d6f361c93160488f9ec0d41e8b5cbcd4 : SUCCESS in 18m 18s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/dfed41b890c4412faf6d81927872802b : SUCCESS in 18m 02s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0f217da6cbca43b4b5ce2bd6bcda5d4d : SUCCESS in 13m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d4dd88a6219e41d59648a67014ac09fb : SUCCESS in 17m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6bdf466f1f6b4724af4e056c85b5fcc9 : SUCCESS in 18m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/08ee9c284593427db9d4651b487b8ffc : SUCCESS in 15m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0fe5c1a4205047cfba2c6073bd3c53e3 : SUCCESS in 17m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/bcc0bc415d654222940fb00cc575c5e7 : SUCCESS in 14m 07s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/eec4b4b63a864c038e970768efe1652b : SUCCESS in 18m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b0b5d0bd858c411e84c593ae3c3a2845 : SUCCESS in 8m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6983c038f5694f22b7cb3d101e7af88d : SUCCESS in 16m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5827565b55904091b0159dd018e1d1b8 : SUCCESS in 7m 59s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/abf4fe7f848c4f1986becca42c06ef7a : SUCCESS in 49m 47s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a52e7848a3d42048349135ebf0fc499 : SUCCESS in 1h 03m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/398e0fbe3e25460c9deaae45c8da7666 : SUCCESS in 59m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ff231864d8b64fa79e16fa7443c2f6e7 : SUCCESS in 59m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6cc89626e624acc8efdd727d5dccfa2 : SUCCESS in 17m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/222f7c28b83e42f1b15e9920b3fbc097 : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/35c8430bb6cf4c94b19dd1d7107ae2c8 : SUCCESS in 20m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c238558ab264488aab08ce1e38ea57c9 : SUCCESS in 1h 11m 03s","accounts_in_message":[],"_revision_number":2},{"id":"53a699b03f9ee86c122824a29283069dd30e9790","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 00:39:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3b89fd68140d396266e66dc6ea4f6fc0c8ad79b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 02:19:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e30ce9eb6b174a0b9b51241a90d0440c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/260a1384b8d94e109364d280c0e30af7 : SUCCESS in 8m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/d944332774504c648fc5af8a17b8f887 : SUCCESS in 1h 03m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d387218b20b8412c923516bbe44a4d2b : SUCCESS in 51m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b46187b1bed6487ba6be533583b47008 : SUCCESS in 49m 23s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c6c1c41a380b447b9f2a69130d160e10 : SUCCESS in 9m 05s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/e67202a262f94f11b801f76a2ab9f4d9 : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/18507a2ae126445db6494658a1bc3cea : SUCCESS in 17m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dc7fe39e0e21403b96b12ec039b2b584 : SUCCESS in 17m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9bb5c2459aa0404192664873fc6f1758 : SUCCESS in 13m 30s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9f98e2bee95b4dc8b386fec2bade93fa : SUCCESS in 18m 58s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/010f879444354c4aa86c8fbd9cc2f6a7 : SUCCESS in 16m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/43599ee5c1d44445a15fc9014ac82a32 : SUCCESS in 16m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4d49969ad5ff4dd5b64f65cc141eff82 : SUCCESS in 17m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2fe50a86015745809eb3d36eb4ab737d : SUCCESS in 13m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0022c9805eea4c7090a69b9b62eaa596 : SUCCESS in 18m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0a700de4e2c9406b8a1f24194e7b0b20 : SUCCESS in 16m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4eec5dcca47c45afb6589442b9622274 : SUCCESS in 12m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6dd524cac4864cd8bc707c398f71dcc3 : SUCCESS in 6m 24s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/defa2065476e4fcf83f3fe88b6d06190 : SUCCESS in 7m 08s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0a6a2a9816cd4c9eb5ec8f1b8cb54041 : SUCCESS in 1h 00m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7043b4ad73c6462ea95aa5c2624143e6 : SUCCESS in 46m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9e733122ebc944238022f4635cc501a0 : SUCCESS in 52m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8974c274b1bc47d198a63f5d6f026987 : SUCCESS in 48m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fedb266192874bf48e834bd15966720f : SUCCESS in 13m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/047305a3db0749f185bc444f0b9e545f : SUCCESS in 3m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0cd12f95e3f4232ad3af0913219efd2 : SUCCESS in 38m 31s","accounts_in_message":[],"_revision_number":2},{"id":"c9a2307f3780c3385c8e20d31a5cec78ceb957ba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 02:19:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d4ddf79057ecec7c572d8c19b3dfac5b30adad9b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 02:21:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4edb39fb7dc6448289d0732ffd8f767d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a28d9fdbfe9492d94ebb6516dbc2ec6 : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/6b3372b33fc441c4b3cc473f3660c3c9 : SUCCESS in 42s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/731786e50d3241d69870ff258a188dd1 : SUCCESS in 38s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f06e5369579599648cc78e4b556887bc6d978c2b","revisions":{"a47daa6e56288be6118c414a26cdcd7170b2e0d4":{"kind":"REWORK","_number":1,"created":"2023-08-24 20:35:56.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/17/892717/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/892717/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/892717/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/892717/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/892717/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/892717/1"}}},"commit":{"parents":[{"commit":"db5c3aeb56134ed220fd141b0415ab37e80b899c","subject":"Merge \"FakeSwift: use env[\u0027PATH_INFO\u0027] to index uploaded objects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/db5c3aeb56134ed220fd141b0415ab37e80b899c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-08-24 20:35:07.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-08-24 20:35:24.000000000","tz":-420},"subject":"Add rolling-upgrade job from Antelope","message":"Add rolling-upgrade job from Antelope\n\nChange-Id: Id59846e508545caf56091ebe30557a44adff1d56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a47daa6e56288be6118c414a26cdcd7170b2e0d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a47daa6e56288be6118c414a26cdcd7170b2e0d4"}]},"branch":"refs/heads/master"},"f06e5369579599648cc78e4b556887bc6d978c2b":{"kind":"REWORK","_number":2,"created":"2023-08-24 21:05:55.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/17/892717/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/892717/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/892717/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/892717/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/892717/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/892717/2"}}},"commit":{"parents":[{"commit":"db5c3aeb56134ed220fd141b0415ab37e80b899c","subject":"Merge \"FakeSwift: use env[\u0027PATH_INFO\u0027] to index uploaded objects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/db5c3aeb56134ed220fd141b0415ab37e80b899c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-08-24 20:35:07.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-08-24 21:05:48.000000000","tz":-420},"subject":"Add rolling-upgrade job from Antelope","message":"Add rolling-upgrade job from Antelope\n\nChange-Id: Id59846e508545caf56091ebe30557a44adff1d56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f06e5369579599648cc78e4b556887bc6d978c2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f06e5369579599648cc78e4b556887bc6d978c2b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
