)]}'
{"id":"openstack%2Fswift~892718","triplet_id":"openstack%2Fswift~master~Idbca73a0b8441c27c241dfb790bc9622eab89d9a","project":"openstack/swift","branch":"master","topic":"release-2.32.0","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2023-08-29 05:40:56.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-08-29 05:40:56.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-08-29 04:53:47.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Idbca73a0b8441c27c241dfb790bc9622eab89d9a","subject":"Authors/ChangeLog for 2.32.0","status":"MERGED","created":"2023-08-24 20:35:56.000000000","updated":"2023-08-29 06:26:46.000000000","submitted":"2023-08-29 05:40:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892718-release-2.32.0","meta_rev_id":"492dbae11392eb1e87ef759639cfc4e1d90a143c","_number":892718,"virtual_id_number":892718,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-29 05:40:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-28 19:11:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-28 19:11:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-24 20:38:38.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2023-08-24 21:00:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-25 13:00:02.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"5ac4abc919005832dfa88e23d93a01a497beb678","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-24 20:35:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dc4515ed378ba4d14a2383b8555521485092129f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-24 20:38:38.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fe1e95925cfa35ac9385ddbf06899f8072f018c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 21:00:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03204e33a0c54457b2c27d19671fff4b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff3ce195f7cb414ea01ff997be4030de : SUCCESS in 8m 51s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5c3dd9a874674e3a89e8fa8683f10d49 : SUCCESS in 9m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/23e8428c45e9457d8c8526c849b0454d : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2d822a706a35486a82ac9086a1d802e1 : SUCCESS in 18m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/35b71ebadfe14dc49bbf518c8f6cb528 : SUCCESS in 16m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e1769280e31b4cb6becb4cc8e87a83ad : SUCCESS in 16m 42s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1d192f50d340469a93ac07e0e1d25f9d : SUCCESS in 14m 05s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ffaa9325bfd049cfac79c274894fe4ca : SUCCESS in 18m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7009c560ae1e4919939d47d035002d09 : SUCCESS in 12m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ff0eb934201478fb65486fd51e2280b : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"1a289aa141f00ac21fcd5b399836747ba2d63324","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-24 21:05:55.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d0e1e0451bfe5e79f482cb0f9506675b0eb59686","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 21:37:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61a39a62691d480ca736cab791423344\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ed511ae90e64de994a443746470d09d : SUCCESS in 8m 42s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/622b811cdb5248a08bc20aefe40fb2ce : SUCCESS in 9m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/33808147a3b249fd9c69ebb5b7b1b887 : SUCCESS in 5m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e48e44596e014fc9be8b3d6e46e45c51 : SUCCESS in 16m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/eac87808f9b448b99557c70951c8a81d : SUCCESS in 14m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f1042da384fe4454abb25a3759e66d87 : SUCCESS in 16m 08s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ae43c42bb0f5481eb03e30544f92bc76 : SUCCESS in 15m 37s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c8aeacfead954811a757024dd234050c : SUCCESS in 16m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81bce60f20c34060b8647d81d534bb74 : SUCCESS in 17m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55d8999f196547b08d87569ce06f590e : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":2},{"id":"067d8f9f1db0f820802a8868c8eac11ff80929ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 22:15:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1c578a3af734187bbd3b0e939cef766\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7b868ac348144b6da0ea9026c1edeeaa : SUCCESS in 1h 05m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a0843818609e4b4e86371cc2edd6514b : SUCCESS in 16m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/da95536efd704e78818ce386895cf99f : SUCCESS in 21m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0ea4465ca33e48e29b15a8f7c4f8ac65 : SUCCESS in 1h 07m 10s","accounts_in_message":[],"_revision_number":2},{"id":"46dba0a76fc1ccbe60011e68c5a2910ac4c3ac78","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-08-25 13:00:02.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0b236bcea42f9e5332973b62c6daca2fa80d2b29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-28 17:49:58.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"e7ec9ee31cec6c74735d087ff375e78bd14766a0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-28 17:57:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7efc33acc2474142e54556a1a52dc8b35a6d8f81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 18:16:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4a7059fde044d0785e43fd923d10915\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f28434e18ebf400db15466c6033379c9 : SUCCESS in 8m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2877f0b36eb9496ea6e967df4af2007f : SUCCESS in 4m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/94f47972c266403b8417cba0da519b64 : SUCCESS in 9m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/dcf34915b14a4023bcf419b2aad039a9 : SUCCESS in 6m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2d2d643ebcf7419cad77a978bb05b17d : SUCCESS in 25m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9aa69c72a06349fd9350bc0b7c855864 : SUCCESS in 16m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c04025129b154bce87d10bbcd7997732 : SUCCESS in 18m 02s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/244b4d9919a04b6d9489aa0fc210ce59 : SUCCESS in 23m 11s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c79ed556f2e74aa5b6404aec5dc1ba72 : SUCCESS in 16m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/664ba511523b48cb8c0e8c579031bc95 : SUCCESS in 19m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78db4a9614b14d1ba15b1ce94d840bbc : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":3},{"id":"6e27c90be33ab988332161793e7dabfdb48557df","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 18:54:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8c3d39605764469ac56a9b6190e15d5\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dc7ee2c0578342659a1fc35d28220a02 : SUCCESS in 17m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c248aaee42bb43b79ab2db698b82d07b : SUCCESS in 22m 35s\n- swift-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/5490668ecd9346e3b5c8d4be317fed82 : SUCCESS in 23m 49s\n- swift-tox-func-py27-centos-7 https://zuul.opendev.org/t/openstack/build/b9a9453f3f6549129adf65c9d7847f8a : SUCCESS in 20m 26s\n- swift-tox-func-encryption-py27-centos-7 https://zuul.opendev.org/t/openstack/build/6349cda4e2204c3da6c5bd7e31f398e2 : SUCCESS in 26m 09s\n- swift-tox-func-ec-py27-centos-7 https://zuul.opendev.org/t/openstack/build/91664de2cf064485ae5b908e64ce8a10 : SUCCESS in 20m 26s\n- swift-tox-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/92918139e0dc4ec08f40e3aafd42e102 : SUCCESS in 24m 10s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/d52d499f322a4b438f84b29ffbf779d7 : SUCCESS in 17m 07s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/d109a830fc7f434aab9cbd214caed3af : SUCCESS in 17m 58s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/d6b58f4759584882bb274c102298f4c9 : SUCCESS in 17m 06s\n- swift-multinode-rolling-upgrade-train https://zuul.opendev.org/t/openstack/build/a46db1393cab459795509484cc411960 : SUCCESS in 16m 32s\n- swift-multinode-rolling-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/4af8e2331e1b4b43a18d63621a376c5c : SUCCESS in 19m 41s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/d651293e7e184938949205b63ea083bf : SUCCESS in 22m 08s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/ee5c946dabe44637b513d0fa8152eed8 : SUCCESS in 20m 54s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/e3d84044ba5846a095d84062f76f7289 : SUCCESS in 20m 30s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/18ed9a0c73444a82a5df88c80d183323 : SUCCESS in 28m 28s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/fde739493e624567867056c76632caa8 : SUCCESS in 27m 35s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/57dceec29f5b49bb82e4d0c22df05e2c : SUCCESS in 19m 47s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/24d8e11c87dd4ada83a0574eaf46b188 : SUCCESS in 21m 02s","accounts_in_message":[],"_revision_number":3},{"id":"3c6b1c05cf2359a1deaab84398a14a88ad753db6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 18:58:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a08aafba4504deeb7d48d6f608df0f7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9bb0c67efa764c88a6f70c49c378bb19 : SUCCESS in 13m 31s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/06f9e7ac72a64322b180ca9aff1299dd : SUCCESS in 57m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/08c4c2b1b16c4d5baf4de634f92cd435 : SUCCESS in 15m 48s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0336726e44d5418ebc376b8ed9fb0024 : SUCCESS in 1h 06m 23s","accounts_in_message":[],"_revision_number":3},{"id":"986463b52de7084f09466983d67bf8eb7c762242","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-28 19:10:41.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"b8d169040cbb46ba480a9fa92d628ec45de46b14","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-28 19:11:07.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"abf09c8e9603512bbf15a3c98409a0e923563941","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 19:32:54.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/201ab8e5d1e24f20a624934e7592c9ce\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e44d490d3b934bf18cb15e27ec6904a2 : SUCCESS in 13m 41s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4d7b911cf2df4e5bab543238be403fe1 : SUCCESS in 14m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab8a00727c1c486b863fbd012377f74e : SUCCESS in 17m 48s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/db90478ecd164cbe97c33c51c4c714ff : SUCCESS in 17m 20s","accounts_in_message":[],"_revision_number":4},{"id":"6d31846fe111607ec3cde17cd2f368e6a7ea9b0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 19:41:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a67b2a3a9baf4b6f97f27f6b2ebcefec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efbf69951e2b42d29f41f73046f6b716 : SUCCESS in 9m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a1166d48bc6464da208c642c4751f3f : SUCCESS in 6m 16s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d7a40dea946c473099e19e5df7ff7dc0 : SUCCESS in 9m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a076b83066c34158803faaca6e2143c2 : SUCCESS in 6m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e689e9996da84cb4950a798e32f33dae : SUCCESS in 18m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bd9d86136faf4be887b7b44c5783b1a5 : SUCCESS in 17m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/82deeca434ef4e80b551ae5b3aca6bef : SUCCESS in 15m 39s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2881b2bde2b140738d899f61e7b8e25d : SUCCESS in 16m 10s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b0e81ae4c915423c91a8c087d01d5430 : SUCCESS in 29m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/337e238f013f4b34914ca303b1694b92 : SUCCESS in 17m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b34515a2198425ea7f356eac7156a6e : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":4},{"id":"f0735ed852a0fbde4103662d8ebd638f6bcae0b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 19:57:20.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3574774189b62b2c71af58b616907ca7aa7a9caf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 23:15:44.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/df48b23147374287abf4acad0fbd22df\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57e0257f3fed4b45bfcc02fdfa21bb1c : SUCCESS in 9m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92499c40099a42e281503122b2af8dbf : SUCCESS in 4m 50s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/426d23022c384c10ae5be46d7324c67c : SUCCESS in 9m 19s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/08f6b4eb041a4d5ab57a1e55d383897f : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4ccf9beac40143e3926d6d617a69dcaa : SUCCESS in 14m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b97992b351ea4349aa8bc1c7a06a2190 : SUCCESS in 17m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8e304bbeeae24aac86979470892aa21f : SUCCESS in 19m 17s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9696a045fa214bf39c8b7edb40ffcd21 : SUCCESS in 18m 35s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/4184903605284ec49ed1e53d66cb831e : POST_FAILURE in 16m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/17d2315117c7447eaf4cd637157dbef8 : SUCCESS in 19m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a3887a3e21394da88ffe00a687a9a62b : SUCCESS in 20m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/247a99d0f2984e6396e500f4060d9407 : SUCCESS in 20m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/931c0b36d9954ec89a79f3dac0603b90 : SUCCESS in 14m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ea379c65814449c4b4879a50829ec746 : SUCCESS in 14m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f8eaaa94ac6d4f0baf89757b513f9dba : SUCCESS in 15m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f4891bca14444cd3b80653b139c74410 : SUCCESS in 9m 22s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/466a4e065aed45ca9e53b53229c42a5b : SUCCESS in 9m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ee01152105c4d69ac42b89fa4592f19 : SUCCESS in 18m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24bc36e8a5e649cba648180ab206b143 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":4},{"id":"2a3689072a87435b2b6d391bf952eda8a25d74ba","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-29 04:53:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6667dba3586c171ddf1c964a0e483ad17593bbe8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 05:16:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d530c30bf3c5477482caec4109caeb28\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13d14c04a6ab47b08afb50ac14b5b0da : SUCCESS in 8m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f60d10dc56504070a9283570626dda74 : SUCCESS in 4m 40s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c5724b61013c4371aa1fbf1d4b7d92ac : SUCCESS in 9m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/015ca3a31da244ac9942ac3a33a1885d : SUCCESS in 5m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8fbff896beb04a2b8ff43a86c720b6a6 : SUCCESS in 17m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6bb6442202014162a6e95f98a0fd6c53 : SUCCESS in 17m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3e678b80a31c4d0b9236f3e1d3a99f9d : SUCCESS in 13m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e4485bfc07774c2d9a8c224ad2356a07 : SUCCESS in 12m 33s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/302c09fe7d344a81a26f254cd334cbbd : SUCCESS in 11m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82d3edf39b0e4a21b671fe2d5100cc67 : SUCCESS in 16m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f84d173122e4670af5ca3a125a7981d : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":4},{"id":"b5fa32af287f54ebddad66b7c0cafbf46d9151d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 05:17:01.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a4243e7afb50a7b68a988bbd35c16a72dbf1a8ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 05:40:56.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62d98dc1158f47978d021d573ee4e907\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6bd65592563471bb7eddcfb080f1c68 : SUCCESS in 8m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43f96bc5e5044f5b8713d46d3be2f39c : SUCCESS in 4m 22s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b0c32ddbd28c4c10be0de1b9974ac0e8 : SUCCESS in 7m 39s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/cdfd258289994552bea092acb1dfa0ec : SUCCESS in 5m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f5b8b4a7fb6445b4924f9e4ea0a08172 : SUCCESS in 15m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/00c78cc7e9534c1e9d7a3cdf1b681ec9 : SUCCESS in 12m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3a002f5935dd4339a5fb219376ebe35d : SUCCESS in 16m 15s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b572f0fe919c412482d9d0a767a79ecd : SUCCESS in 16m 35s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0462386d26b340968ede9d6b175e374b : SUCCESS in 15m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c2b5182f2e9e407e819f5fd4ad12a2a3 : SUCCESS in 16m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/44fa2b7fe8f944fab72c8721672cc4e5 : SUCCESS in 17m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a3377bfecbe548f2a5a5f36d726e9f3e : SUCCESS in 17m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/81a3b1d73681443a99d1c9acb2a93c16 : SUCCESS in 14m 50s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a6c702f547b74a6193112768fe4e3903 : SUCCESS in 15m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/619c3158ab6c4c25af1af82507d04354 : SUCCESS in 15m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5746cf34d28c410e8ff43219b0fbcaf7 : SUCCESS in 5m 51s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5ca242e5e04b4f03898a2c523b7f888d : SUCCESS in 7m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95457cfe1518482ca49d1b55b379e464 : SUCCESS in 17m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4255483c96514f47b0520d8fe9bfa9ee : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":4},{"id":"f3914dc1fe981b94f43a1f9b16c9992958b40992","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 05:40:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"582fa95bcbca347f68ced396df0ee16dff9f9848","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 05:41:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d37c1d21372047409c291966868049c1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c74ea0bfb154b1d9cbbda7adee47ab8 : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/982d324b22ff49f3925a37dae3020f04 : SUCCESS in 43s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d900f8c224d64465a078d060ca8fe0bd : SUCCESS in 29s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/50504f507e424ffc992023bc04723ea8 : SUCCESS in 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"492dbae11392eb1e87ef759639cfc4e1d90a143c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 06:26:46.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0f0cf3c7b394144894ad7275ab1e5d7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5a83fd2d0b854dc390cc197b0676c39f : SUCCESS in 12m 50s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7fa60cd94a8c4467803e157c3d2d258c : SUCCESS in 13m 49s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f8ce827825bb4caf9b66788c8844b15b : SUCCESS in 17m 05s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c39ad53b9214d8e87cc8a33457a0bfc : TIMED_OUT in 1h 31m 08s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"191b627b856bc2e8eaa4049521e3476bf29ed40c","revisions":{"b0dd9d781278a26873ca6a84836a3a16926c02a6":{"kind":"REWORK","_number":1,"created":"2023-08-24 20:35:56.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/18/892718/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/892718/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/892718/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/892718/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/892718/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/892718/1"}}},"commit":{"parents":[{"commit":"a47daa6e56288be6118c414a26cdcd7170b2e0d4","subject":"Add rolling-upgrade job from Antelope","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a47daa6e56288be6118c414a26cdcd7170b2e0d4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-08-24 20:13:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-08-24 20:35:24.000000000","tz":-420},"subject":"Authors/ChangeLog for 2.32.0","message":"Authors/ChangeLog for 2.32.0\n\nChange-Id: Idbca73a0b8441c27c241dfb790bc9622eab89d9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b0dd9d781278a26873ca6a84836a3a16926c02a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b0dd9d781278a26873ca6a84836a3a16926c02a6"}]},"branch":"refs/heads/master"},"c9fe1c6224c7890996f85d15bc79f9dee3b86f94":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-08-24 21:05:55.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/18/892718/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/892718/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/892718/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/892718/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/892718/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/892718/2"}}},"commit":{"parents":[{"commit":"f06e5369579599648cc78e4b556887bc6d978c2b","subject":"Add rolling-upgrade job from Antelope","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f06e5369579599648cc78e4b556887bc6d978c2b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-08-24 20:13:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-08-24 21:05:48.000000000","tz":-420},"subject":"Authors/ChangeLog for 2.32.0","message":"Authors/ChangeLog for 2.32.0\n\nChange-Id: Idbca73a0b8441c27c241dfb790bc9622eab89d9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9fe1c6224c7890996f85d15bc79f9dee3b86f94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9fe1c6224c7890996f85d15bc79f9dee3b86f94"}]},"branch":"refs/heads/master"},"728275664be07c0c6696b664fe8f71faff4ba860":{"kind":"REWORK","_number":3,"created":"2023-08-28 17:49:58.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/18/892718/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/892718/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/892718/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/892718/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/892718/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/892718/3"}}},"commit":{"parents":[{"commit":"f06e5369579599648cc78e4b556887bc6d978c2b","subject":"Add rolling-upgrade job from Antelope","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f06e5369579599648cc78e4b556887bc6d978c2b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-08-24 20:13:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-08-28 17:49:17.000000000","tz":-420},"subject":"Authors/ChangeLog for 2.32.0","message":"Authors/ChangeLog for 2.32.0\n\nChange-Id: Idbca73a0b8441c27c241dfb790bc9622eab89d9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/728275664be07c0c6696b664fe8f71faff4ba860"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/728275664be07c0c6696b664fe8f71faff4ba860"}]},"branch":"refs/heads/master"},"191b627b856bc2e8eaa4049521e3476bf29ed40c":{"kind":"REWORK","_number":4,"created":"2023-08-28 19:10:41.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/18/892718/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/892718/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/892718/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/892718/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/892718/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/892718/4"}}},"commit":{"parents":[{"commit":"f06e5369579599648cc78e4b556887bc6d978c2b","subject":"Add rolling-upgrade job from Antelope","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f06e5369579599648cc78e4b556887bc6d978c2b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-08-24 20:13:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-08-28 18:07:42.000000000","tz":-420},"subject":"Authors/ChangeLog for 2.32.0","message":"Authors/ChangeLog for 2.32.0\n\nChange-Id: Idbca73a0b8441c27c241dfb790bc9622eab89d9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/191b627b856bc2e8eaa4049521e3476bf29ed40c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/191b627b856bc2e8eaa4049521e3476bf29ed40c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
