)]}'
{"id":"openstack%2Fswift~893516","triplet_id":"openstack%2Fswift~stable%2F2023.2~I7b04e128e93b8674141ec9f29b5176e07c9b5262","project":"openstack/swift","branch":"stable/2023.2","topic":"create-2023.2","hashtags":[],"change_id":"I7b04e128e93b8674141ec9f29b5176e07c9b5262","subject":"Update .gitreview for stable/2023.2","status":"MERGED","created":"2023-09-01 13:36:51.000000000","updated":"2023-09-01 19:08:21.000000000","submitted":"2023-09-01 19:07:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893516-create-2023.2","meta_rev_id":"8b515c0e08aa9714f3eb71c9e70590e63f9fd2f8","_number":893516,"virtual_id_number":893516,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-01 19:07:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-01 18:44:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-01 18:44:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-01 14:05:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-01 18:44:29.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"eb371809131aa9485441b4e29e05d3e6942290ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2023-09-01 13:36:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"436fe4c2fda4128d9e3f7183386e686b27424219","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 14:05:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9f8f1151c134551903b4d0e52ce0407\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c246859c45214b9a8c2e5b98cb35af9e : SUCCESS in 8m 06s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2b85fbbcf92c42f091ef17bf5a2ba246 : SUCCESS in 8m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a8486586f591428d8f53938bc320ec7e : SUCCESS in 5m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f59841500b77413b8a468ac9f5bd21ca : SUCCESS in 26m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a5ed9db903a94be395551748ce0fdafa : SUCCESS in 15m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bdac64b8b69c4c1498d93002f2bf9e1c : SUCCESS in 14m 59s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1fb9aef1ca774f808454877f88872995 : SUCCESS in 15m 08s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/665e8f259858427c8f4ac7cd567e6aec : SUCCESS in 14m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ac0b60c5c6943a7a63b443cbd4ed62d : SUCCESS in 18m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f097cb25047d4b2bac02f70e55cc9d1e : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"d052cafd19580edb3b48c64b968caa6b0d46e1f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 14:43:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/275673d2064d41879e39fd8563ddd8fa\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2f98e4856fcd4417a7459a5be6a4494d : SUCCESS in 16m 29s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a29068f939c3454c8fd57d723f2e4216 : SUCCESS in 15m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4bc1b219e4314577b3fc9cf9c8744ff8 : SUCCESS in 22m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/31307ac545c347159db8db6a7398e7ea : SUCCESS in 1h 02m 57s","accounts_in_message":[],"_revision_number":1},{"id":"1b9a320e0db36a24a8cd3ef5933866be24bd4f37","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-01 18:44:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aad6ed378bb7e50b21260b6632fbf2836689f194","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 18:44:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3b9fba13d24defb51d0c2ad3468758fffc7cb8fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 19:07:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82aff3633d13419e9a499795c6596ad1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d34364ee32848188fc6bbd2d73aee30 : SUCCESS in 7m 15s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/28ac885f196348118989c4fcbd999689 : SUCCESS in 7m 22s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/3e4adf184cf64d968b4420733b0e04a1 : SUCCESS in 5m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dacd90bb35ab4ad1b208f8407f924806 : SUCCESS in 16m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d9231cbd4bbd441f935afa641a79e102 : SUCCESS in 15m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/124c123f183149fd8819a16123009c91 : SUCCESS in 15m 16s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6e2d6b9a33f84f3bb995164f1b209351 : SUCCESS in 15m 47s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/585e22525c6644eb8c876dc3427536af : SUCCESS in 15m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6b3d541baf874aa2b8d33186ea7a1f08 : SUCCESS in 15m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/73a57db2cb734eafa3cb2119f12d34ae : SUCCESS in 16m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5593f2d612fc4764b1b70d7746cb3845 : SUCCESS in 17m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f45b040a68fc4b6a98e6c1ed409b4adf : SUCCESS in 14m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/62f525f83b1047f68b529e1063b26816 : SUCCESS in 15m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1ca99cac2dc74b91b4c50de77cdc06fe : SUCCESS in 16m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/158c4527fccb4aacba6f8c24da83ff55 : SUCCESS in 6m 20s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a64a3b0e3a8f482bb1e10376751b7a2e : SUCCESS in 6m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1205cac4a1114d2dbae7986c56fc558b : SUCCESS in 15m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d192f1c5b7ff48898d59fba16a9485ff : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"a9c9918101f50318631ab8aa458f8e12997b4ec6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 19:07:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8b515c0e08aa9714f3eb71c9e70590e63f9fd2f8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 19:08:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dfc703eadaf434ca2551c3033593d65\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d893bfd2af74cc693ab2d9e07932a7c : SUCCESS in 41s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/4330817b7ef54bdaa11758e6ede09984 : SUCCESS in 38s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/0837fa14020d4041a656eac87aa62f1d : SUCCESS in 28s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"468306d4f8fb12bd69cb9597f8c3b2c7b86b2942","revisions":{"468306d4f8fb12bd69cb9597f8c3b2c7b86b2942":{"kind":"REWORK","_number":1,"created":"2023-09-01 13:36:51.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/16/893516/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/893516/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/893516/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/893516/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/893516/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/893516/1"}}},"commit":{"parents":[{"commit":"191b627b856bc2e8eaa4049521e3476bf29ed40c","subject":"Authors/ChangeLog for 2.32.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/191b627b856bc2e8eaa4049521e3476bf29ed40c"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-09-01 13:36:50.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-09-01 13:36:50.000000000","tz":0},"subject":"Update .gitreview for stable/2023.2","message":"Update .gitreview for stable/2023.2\n\nChange-Id: I7b04e128e93b8674141ec9f29b5176e07c9b5262\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/468306d4f8fb12bd69cb9597f8c3b2c7b86b2942"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/468306d4f8fb12bd69cb9597f8c3b2c7b86b2942"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
