)]}'
{"id":"openstack%2Fswift~893517","triplet_id":"openstack%2Fswift~stable%2F2023.2~Ic6bb45f3f3a50daccdca186008a012e4a75854dd","project":"openstack/swift","branch":"stable/2023.2","topic":"create-2023.2","hashtags":[],"change_id":"Ic6bb45f3f3a50daccdca186008a012e4a75854dd","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.2","status":"MERGED","created":"2023-09-01 13:36:52.000000000","updated":"2023-09-30 07:24:26.000000000","submitted":"2023-09-30 07:23:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893517-create-2023.2","meta_rev_id":"d780a9b55baac192f7ad42732562db2ae840accd","_number":893517,"virtual_id_number":893517,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-30 07:23:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-30 05:55:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-30 05:55:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-01 14:48:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-01 18:48:41.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2023-09-30 05:55:14.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"ef21730eba23c861a5025aa06162436be3b63849","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2023-09-01 13:36:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32d06290b8e7948d89f4b9e7c44c18e3845b7f32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 14:48:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba5818fcac884d739c10effb0fc3c309\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f112853a9e942338e880f00d5d902f1 : SUCCESS in 9m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09c4f8e258634781b987b85e1586a995 : SUCCESS in 5m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/62b39c4c2f644139b7474fbd50e78f94 : SUCCESS in 19m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/83da3018880243f4ae95884e7e441240 : SUCCESS in 19m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7e5c55b3f59342129da7ee5d28736fdd : SUCCESS in 29m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0a7108fb20524030a7a203629925d5ef : SUCCESS in 7m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/aa7a7d1d89934141ae017ef98d120dde : SUCCESS in 5m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3b7ed6402f7344baae527ccd65ee4746 : SUCCESS in 17m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b065a37e4efb4935a864c18d6257930b : SUCCESS in 16m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/46f36a45de624752b1a9d94748dda0e6 : SUCCESS in 18m 26s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/451b8c66698f4e02a40f0355fc0d9199 : SUCCESS in 17m 31s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ca4b65cb06294a19a6c0eaf4c174a927 : POST_FAILURE in 12m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c4c22d3da0f34c7a863a96f6ac9fb8d7 : SUCCESS in 14m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2f36c7cb457c4aa88d6b39963409fce3 : SUCCESS in 19m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9aa8ab2a1c7c49d0ade381a1f03dc8b8 : SUCCESS in 17m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/35542bc6d1cf4c2e90b4d644c320356d : SUCCESS in 13m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4aef6bc4fb3a40a5bd954248b8e3a434 : SUCCESS in 14m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8841737290bb48959f04b681dd70aecc : SUCCESS in 18m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1fbe2fcb3f494e4985a22b56a8c4e12c : SUCCESS in 7m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dc924f3031f64f9ebbd6308168908fe0 : SUCCESS in 17m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/08347427461e4b1dad8e5243e3c1558b : SUCCESS in 8m 19s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9599ea792ed24d74849437a7aa41cfe3 : SUCCESS in 52m 48s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b4fe3dbb2c6c4ff0a5d73670e4fc5219 : SUCCESS in 1h 00m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11c6699abd7b4b179c5679ccf111c67e : SUCCESS in 18m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d25b0b7487d48d0814c761c3f21c725 : SUCCESS in 5m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7e77a090ab9b45369171420678c417b2 : SUCCESS in 19m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"06d5413776b8961fc3d76db04ae4ff131fdde24b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 14:54:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cad553f065bc40d3aa2650dcde325d84\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7176a70b8ca54c00ba26688da9a8d19c : SUCCESS in 17m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ca75f505b066410da9d9fa4607db47e3 : SUCCESS in 17m 43s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6592460a4a0147f8832730efb5dec3fe : SUCCESS in 1h 13m 42s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c33f387cdd64809934bc01071f0e811 : SUCCESS in 1h 01m 41s","accounts_in_message":[],"_revision_number":1},{"id":"800f57ab3b6a8e96ee81993531de5c95372c1ba4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-01 18:48:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f71460062b3f32bb16436c3d293e1fc070fa2042","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-29 22:07:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6d1b7c3522600b93ff3c618a91e9d0cb35a89479","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 23:23:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a63b877e39046eb9cf1d40720285522\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d11d5961cf24706a61789de0f15b000 : SUCCESS in 8m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/020f289042c846e1b6df487370b817cd : SUCCESS in 4m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/33c8bc489b194409993216f19efb1820 : FAILURE in 45m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a9dad2ef81304c41bf80ec4c2ff2a8f1 : SUCCESS in 1h 01m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf65c020574c4caebbabf4d5a2f77e36 : SUCCESS in 1h 02m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/486310455634416bb76bfb094fa3e1f5 : SUCCESS in 18m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9a91643282344ad68a6e8b7589fd0a97 : SUCCESS in 18m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a59afefafb09492ab2ad6e7917288847 : SUCCESS in 18m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b677b8ff364b428a9cbed6a6438fde29 : SUCCESS in 9m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/231462614ba543fd9c383b463d9e21c4 : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/315037054af44e919e855adaf771c1a0 : SUCCESS in 18m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/05dc796566bb4f45a555c9dbf355b56d : SUCCESS in 14m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f1675cabd5da43a6b730e2c9ec8ed26f : SUCCESS in 15m 24s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d482921649f4412a8ab0326e1f9047d8 : SUCCESS in 16m 34s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/619b82ef9f624d34a45ff1de9d6619bb : SUCCESS in 18m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2a3a49c908804e8a872a47968325f3c7 : SUCCESS in 14m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/639677409d684b76a441c7a123b6980a : SUCCESS in 18m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ab8124d102b34440bd1b5d2063e0e997 : SUCCESS in 18m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/dc632e4ccc024093be9d1501c0c7ef42 : SUCCESS in 14m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b03335e128dd4880886dc96f61f17a60 : SUCCESS in 14m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/09cb7586581e4d72a1a915db4fa0fe6a : SUCCESS in 12m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1c8faae226714198823a4e3f44db486e : SUCCESS in 6m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f864a94fdca84d57a7b2009e61f8772e : SUCCESS in 13m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/08d94875dee4462a9420bf613ee75a80 : SUCCESS in 6m 49s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5460a201db474af2abd231dd6341a83c : SUCCESS in 40m 42s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b20a2eff2182420092b2d2bc839ac1c6 : SUCCESS in 47m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d1cf8792195d4e2f96aaec983934674e : SUCCESS in 35m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2844de394bf54765b450c345525bd3a3 : SUCCESS in 34m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d039207598442f684f8caf7bd09a17c : SUCCESS in 13m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c5b8444e6584a7f87349aa548c975c2 : SUCCESS in 4m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/302115a98fca4bc2b2a17274327d8f1f : SUCCESS in 24m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb46dd741a254c64b90328d46dac3ed1 : SUCCESS in 1h 00m 40s","accounts_in_message":[],"_revision_number":1},{"id":"c767bfcb50a216e80f689ae8bd7f6b75a2d59ea0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-30 00:17:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c805d51494fc79be5c7fbc12b82f6e2050b5c82b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-30 00:50:44.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/634a485db58e47768370d83761aa00c2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d16b28597804898b617d4003febe07c : SUCCESS in 1h 26m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3914858399214ea59c489c637334b06d : TIMED_OUT in 1h 30m 49s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d81e7385188f474389652c3214f2f584 : TIMED_OUT in 1h 31m 07s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6f2a775b6a0e4299a6d1438d2881ff65 : TIMED_OUT in 1h 30m 59s","accounts_in_message":[],"_revision_number":1},{"id":"8e50962f2cba45d129fb44b1105b4c08f5ad01ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-30 01:29:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbd4090e8f804448806b091830585bec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81dfcb739cb34a8e86b9a4ac3690e0f4 : SUCCESS in 7m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/356ce3e83bb34da6b544f10c1810be3d : SUCCESS in 4m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/6d4d64d7de31484189a5a48e92004ba6 : SUCCESS in 39m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/36ce1b50304f4465a9049c9f4d25c5a9 : SUCCESS in 37m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c5e86ea74da487fbdc4f0065b6dfa8e : SUCCESS in 1h 03m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9d7ade91cd464c55b361a551ff9d2099 : SUCCESS in 20m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9ac78b1f864a48d5b6cc3969e0424a0c : SUCCESS in 19m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7abda53dfa3d457d85808712b5027405 : SUCCESS in 19m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/adb0a8fb3a664f13919cff40957833ac : SUCCESS in 8m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/94c89b34dab24b8b9663f84f01f7fa26 : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d0d78157c5ef4ab2a4f7b7ba71d30703 : SUCCESS in 16m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/031f459c85bb479b8253b5437c9dd3fe : SUCCESS in 18m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/16edbe277df64005b3389da2fc4af34c : SUCCESS in 15m 28s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/71f78c088d3a4435b3274054f3ea80ca : SUCCESS in 15m 58s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b45fcca2fef649188b027a87602d861a : SUCCESS in 16m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2b22f8e078e84677b263cf1a24f9074c : SUCCESS in 13m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/528538b0d6d4403390344e18ba028072 : SUCCESS in 12m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/38be05adac8f49e49e789e943e7f2714 : SUCCESS in 13m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c0d89ec034d3433b988bb52e66132d60 : SUCCESS in 18m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ca8bfcc201614e0cb9b0c694898b75eb : SUCCESS in 11m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3b5b7af3693f45f7a5c733ca7cf06c46 : SUCCESS in 19m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4eccec1ffdd74674ab70bdc56ad80c79 : SUCCESS in 6m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/20383bb7003b4fcaa9acb431a0126532 : SUCCESS in 16m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/64a9a9c1d5714649839bc8f80131cc7a : SUCCESS in 7m 52s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/44b61bcad5974677a5a54572ba5daffa : SUCCESS in 41m 43s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b0ccd4a72b854ea8ac83c908b38e9f1b : SUCCESS in 1h 02m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e7db1f3adb9f44fe8155a4cd2ece3437 : SUCCESS in 46m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a1d0b35e72fe4396bbe60c728c5626ab : SUCCESS in 54m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b8fa46ef11e48889f7a1e3772acb3fe : SUCCESS in 17m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d968965d3a9e43c09bf580e03e87bc73 : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/769228da01fd41228a75424c9bf11c10 : SUCCESS in 24m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1fc6638095a141db93edfcafb2e3b510 : SUCCESS in 1h 04m 29s","accounts_in_message":[],"_revision_number":1},{"id":"f366af8394b190615e9de425955c3845785ef25c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-30 05:55:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2f4912d12aae5405a1f74e48fc3bca7a25955558","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-30 05:55:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f7c3f30461a949d6c8fbc1ec627fba6f406b5764","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-30 07:23:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd8b4cf3de7b4cbda1d98bc0e65d0953\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76bbe3ddfcce4c58a9db754918747858 : SUCCESS in 11m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f180ac220a5a439a93689a109782b115 : SUCCESS in 4m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/f4cb4b8a3f1a419296ca696c01ba1104 : SUCCESS in 52m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b5869a3b9f404911bb5695da8ab3e8d3 : SUCCESS in 53m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b42366ffc26240e9836a12687b104513 : SUCCESS in 56m 36s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6a80358526de46f6a481c8e03e2941e0 : SUCCESS in 8m 57s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/dd3e96d2173d4dbba735b3c1bac9a2ca : SUCCESS in 7m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fa0e1c7768fb4611b9839daf63aed28e : SUCCESS in 17m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/837cf3279320449c9125e855c6c55e18 : SUCCESS in 15m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2534bc07f11c4b6f8a729c4154438cdf : SUCCESS in 17m 10s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/bd9ea138e498404ab1113272c78db4c5 : SUCCESS in 16m 27s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d9ca59a7d5f24b3cb6d93025a6efad56 : SUCCESS in 14m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1fdc72f86b5144f68253d2c60e729f79 : SUCCESS in 18m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/649c4cf57cac4140b6d433adc5c19d03 : SUCCESS in 17m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/423772cd7ed245c593f3dd5e9303f5e2 : SUCCESS in 15m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2bcb0f8146d5493db07a74922d9bf268 : SUCCESS in 15m 01s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/44de2351ad4649cfade60b2d29fb82dc : SUCCESS in 15m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cd7bb9adf06b4ce992a52e9f9a1b5a35 : SUCCESS in 19m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8b2629de50754d8cb16dae99d8f2b29b : SUCCESS in 7m 22s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/faf769e97e324075bf857f92622516f2 : SUCCESS in 6m 54s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fd5453847c8c47af9a4eeeedbf124fc7 : SUCCESS in 56m 25s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e3922cebf2a541ef9112c04a576ff01d : SUCCESS in 1h 02m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c6e3200283724dd3aef33d87d0d47b43 : SUCCESS in 52m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ab1bcb2e5eeb4684800264a87def2d8f : SUCCESS in 54m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9721dc3c76e4cefafe19794476ac927 : SUCCESS in 18m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80ee7bd6e9514c44ab1b37dff5f1945b : SUCCESS in 4m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e554b12092645c58cb00c88ea4b3c89 : SUCCESS in 1h 06m 26s","accounts_in_message":[],"_revision_number":1},{"id":"60e839438a6d114db6efbd911bdaaac4c53a35bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-30 07:23:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d780a9b55baac192f7ad42732562db2ae840accd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-30 07:24:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfdf9a6df9c14f079a4dc5ec70d99ba8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adc748064de74124af763c42af40b73a : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0e4c1b20ebd40edb97f4a27a99c53eb : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/51216ad5c8804e15a8ffe23220e06bcb : SUCCESS in 31s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/a5b63032f0384d948f87340b80a50584 : SUCCESS in 28s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e01b28b48e34d802443a96866c7f5f4bd6950d35","revisions":{"e01b28b48e34d802443a96866c7f5f4bd6950d35":{"kind":"REWORK","_number":1,"created":"2023-09-01 13:36:52.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/17/893517/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/893517/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/893517/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/893517/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/893517/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/893517/1"}}},"commit":{"parents":[{"commit":"468306d4f8fb12bd69cb9597f8c3b2c7b86b2942","subject":"Update .gitreview for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/468306d4f8fb12bd69cb9597f8c3b2c7b86b2942"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-09-01 13:36:51.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-09-01 13:36:51.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.2","message":"Update TOX_CONSTRAINTS_FILE for stable/2023.2\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/2023.2 branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Ic6bb45f3f3a50daccdca186008a012e4a75854dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e01b28b48e34d802443a96866c7f5f4bd6950d35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e01b28b48e34d802443a96866c7f5f4bd6950d35"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
