)]}'
{"id":"openstack%2Fswift~893735","triplet_id":"openstack%2Fswift~master~I2945ae725f97ee56e9c7ad580168ec8091f2049a","project":"openstack/swift","branch":"master","topic":"p-fix-test-log-notice","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-09-05 18:35:10.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-09-05 18:35:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2945ae725f97ee56e9c7ad580168ec8091f2049a","subject":"debug_logger: ensure NOTICE level is registered","status":"MERGED","created":"2023-09-05 13:54:24.000000000","updated":"2023-09-05 18:36:22.000000000","submitted":"2023-09-05 18:35:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"893735-p-fix-test-log-notice","meta_rev_id":"48ac2adeda6e07d782d7bb97b50b0ae782866ece","_number":893735,"virtual_id_number":893735,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-05 18:35:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2023-09-05 16:22:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-05 15:22:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-05 15:22:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-05 15:16:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-05 15:22:24.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2023-09-05 16:22:53.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"8157cd06ad3a51f963c4f7d893083f3f78d3fca3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-09-05 13:54:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f473fe08f34eb65139d0acdf47d8ed5c87c50e0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-09-05 14:05:22.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"92f376ac93c78dca057e025da83ddc895b1146dc","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-09-05 14:10:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b97b927159ffde95afc7a483cf1045930e63169","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 15:16:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfec5fc4ba8c4baabe1113ba4e033280\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d1ce0dd09364bf3b29a23bfc9339724 : SUCCESS in 9m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a6ee81ba7b3a4933868caa9fa0d8b716 : SUCCESS in 18m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3643c4d93eb44cf5bb4da1f38d15d47d : SUCCESS in 19m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ade1a96645ac4bdeb43cd273416717e5 : SUCCESS in 15m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ff9d98915ba7449f82f14580b4dfb2e3 : SUCCESS in 7m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e8b51e6c22244abeb81a8bd3fc34438c : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/488a95391fff42a6a5d1127fe9dfaace : SUCCESS in 13m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5f888faccd0a44ff9fe5d1ba9458410b : SUCCESS in 16m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/251370a19b0346a98a3db5657f68dfec : SUCCESS in 15m 57s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2660b151cb494b9189501193c2cc6d1b : SUCCESS in 16m 52s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/557aa430ed1e4273895c82b38c19ac67 : SUCCESS in 12m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7213c30f196d4d009d438e9fb513b7d8 : SUCCESS in 16m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/03435279a5364fec972ba4415f7bad24 : SUCCESS in 18m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/816c290140c740ffa83b91f745f26b63 : SUCCESS in 13m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/18f39646798b41d58dd72152816add40 : SUCCESS in 15m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b62c72c81fea43ec8998da8071949c80 : SUCCESS in 13m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f1797cd52bc2475dbefb8ef18e1aa341 : SUCCESS in 14m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/84a6fa8e4cf3426287d1b50df7908a29 : SUCCESS in 7m 24s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bf2d3279d4ee48f7883f784bdce5aa5d : SUCCESS in 9m 08s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/08cbdb7727d6401d8dd6ec2636237e71 : SUCCESS in 1h 10m 07s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/78c2a56714164292bc3fee0b1eba982b : SUCCESS in 58m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6aa0cd51662544bd88b17066fa1edee6 : SUCCESS in 45m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/57dacc9d7eac4eaaa6c2f334f1b932b4 : SUCCESS in 56m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e78150205fc44afa4a98e57210f33cd : SUCCESS in 20m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6791d3ed088485bac96a4f35de53ab6 : SUCCESS in 5m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/77889fbf82854347a241136283c66844 : SUCCESS in 30m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6aae5b517f926d450d4077fc9c6c9fdf91bca95e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 15:19:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82747dfd2e8846068f2a208f555d3f8a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f386990ce4234c01ba1dc2a116d9469b : SUCCESS in 1h 12m 50s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c2736db201e748a1853fac9dbd45a989 : SUCCESS in 17m 02s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2f52f0248e0240acb5b704280da41474 : SUCCESS in 1h 09m 34s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3b7bdb1c6258445cad9e01533903a2b5 : SUCCESS in 18m 17s","accounts_in_message":[],"_revision_number":2},{"id":"2681801b98f139790b7a4b9f3f5b36f5054d1844","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-05 15:22:10.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"ac787f8dda74b0925d9388f97daffe839e125f42","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-05 15:22:24.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e80f944dcf30a77dc3ccd8ca9e41353760e3651b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 15:40:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/923540239e30405fa41abc1a0ba5a7af\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/57b1ae645c454379a3a85f15f2bb86d9 : SUCCESS in 14m 50s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ed9025ea727643dc974eaf818e096d5c : SUCCESS in 14m 12s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c54ef758bbc542309c6734fd24840fb7 : SUCCESS in 17m 47s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e7db51a1f06141dfb3f53b57baea930d : SUCCESS in 17m 05s","accounts_in_message":[],"_revision_number":3},{"id":"69b2fab70fbf786a9f6aa191c2e8f73b22112465","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-09-05 16:22:53.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f1589478244c544c0a46f3c73e7c18ea8eabc873","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 16:34:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a5b9a6c93704fc2b97fd291fffb301e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0383b47da9a4c9f8791f41d0785353c : SUCCESS in 9m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6cd490a41f64618b80cea15246dfa2c : SUCCESS in 18m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b99c6b711a6e466f86c946a66dadf562 : SUCCESS in 19m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c3d79021cbc645048c762b88a6c61752 : SUCCESS in 18m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ec424b6f1c5c433fab7da579dbbe6d00 : SUCCESS in 8m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7f35e980337e4e5c8cc323b5a5ffc8f2 : SUCCESS in 5m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2d03fbcd72c84ea48f003ca94536bfd5 : SUCCESS in 17m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3cd176122a1540138a2dcfff5dcff90d : SUCCESS in 15m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9f621b167caf46cdbb264665f7923912 : SUCCESS in 12m 27s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9a9fd10d7ab74031a852d21372e19973 : SUCCESS in 14m 40s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/dc86f58ba74943ef9c2fe5309bf063d3 : SUCCESS in 18m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/643387f5e25046af93050cf3f9db568f : SUCCESS in 17m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ed0239bb0b0741b385e13d1cdd37db3b : SUCCESS in 23m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/047dabcddfa0473b82788c31874e899e : SUCCESS in 15m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8cbc9b223cd249f6bb96f4a22664c467 : SUCCESS in 12m 20s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/22ed8fffaacc4c68a729298f01974656 : SUCCESS in 20m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7fddf9df67b946a5bd5bbbb6857856e0 : SUCCESS in 13m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cad66be4484842c7b4b4ffa106317428 : SUCCESS in 6m 56s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2065c684033d41dbaad6541abb66fdba : SUCCESS in 7m 32s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ffdbca216b954fafbf785f68ee9134fb : SUCCESS in 1h 11m 00s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d09a1dfd57144e59a82fec0b8b6f6f2d : SUCCESS in 55m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/05014a3b853048ce8d70b6d61e97c6ba : SUCCESS in 56m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c218301955d141daab588302a3507de2 : SUCCESS in 55m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/563ced650277433994dd3fc86d35c629 : SUCCESS in 14m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a36f64d1989e4a7ba07c1edea67eee7b : SUCCESS in 4m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/241facf1b4d645748321d69f98aea3ff : SUCCESS in 17m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"186ed17bc2ad0093ed9c78cd9c162fedb606fd2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 16:35:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"765cbf4698404114c7e4227a346f978c7742ef89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 18:35:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81ca7cad110c49d7ac344cff15268e1d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6502019a978940f98f4a840b377ca817 : SUCCESS in 8m 42s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/be2d1c82925645219fe3f0e2850e0cd1 : SUCCESS in 7m 35s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/45554487730141cbbb83de18c6794977 : SUCCESS in 7m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/879834481fa7457c9d462076502f7312 : SUCCESS in 16m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9a6e304a307a4b36a4bc9a8ecbae8016 : SUCCESS in 20m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7a9958be62c0467b8a8660a5c0c5c05f : SUCCESS in 15m 12s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8069d86b4ee94a86ad66ea932c73a335 : SUCCESS in 16m 42s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/23f1b998599a47f8bd708ab0f6d8b4e6 : SUCCESS in 21m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d641f0103713414a8c6a4fd39f48c15f : SUCCESS in 16m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f321d5a6c03843fe98032e1fe1fcc743 : SUCCESS in 16m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f8d5afeb3b7c491881790f7008224bfd : SUCCESS in 20m 13s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b9bee4b62f1840ba826ec6a82a8d9c48 : SUCCESS in 13m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8992e8eff0154af28b909a34c064431c : SUCCESS in 21m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7a0e4930496b4508b728b73cfa11a102 : SUCCESS in 21m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9d35c4094be6456dbea0dbffa927f289 : SUCCESS in 9m 06s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c1a58adb35024071b28fcd4bec0c9e1b : SUCCESS in 8m 14s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0104ce4b4d00424d9850e6e2bc1a4aa0 : SUCCESS in 1h 08m 33s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/76a2b415990548a78589e23ffaf70cf1 : SUCCESS in 55m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5c0beba837e74a8e8015d106d04101be : SUCCESS in 56m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/240ef9d0082a4b63881bb21fa29df5e6 : SUCCESS in 57m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12f7d17ffc0846949d7bed8c95ede47c : SUCCESS in 23m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10150cd33cf34f21be865d7cfc13dd45 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":3},{"id":"7e8fbf98c05ce6b86f44bde379cea013c2b63768","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 18:35:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"48ac2adeda6e07d782d7bb97b50b0ae782866ece","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 18:36:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59516959767444e49fe5094ae63d318e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaf7fe05f18f499f9fee2e6c00263882 : SUCCESS in 53s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/0573897a20424a33b0c939520f14e85f : SUCCESS in 33s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/722863e9785b4572b9a59decfa4108e1 : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4a80963b409a5da73bc18185b4754977d1837130","revisions":{"9f0d538463a3eebedfcb8875281771760eee5169":{"kind":"REWORK","_number":1,"created":"2023-09-05 13:54:24.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/893735/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/893735/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/893735/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/893735/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/893735/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/893735/1"}}},"commit":{"parents":[{"commit":"af97975ace268bea0b0c761c2270e135c73da43b","subject":"Merge \"s3api: use string for swift info request method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af97975ace268bea0b0c761c2270e135c73da43b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-09-05 13:44:56.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-09-05 13:54:01.000000000","tz":60},"subject":"debug_logger: ensure NOTICE level is registered","message":"debug_logger: ensure NOTICE level is registered\n\nThe NOTICE log level (25) is added to logging when utils.monkey_patch\nis executed. Depending on test execution order, this may not have\nhappened when some unit tests run, so LogRecords at NOTICE level may\nhave the default level name of \u0027Level 25\u0027. FakeLogger doesn\u0027t have a\nkey in its lines dict for \u0027Level 25\u0027, so those tests would blow up.\n\nChange-Id: I2945ae725f97ee56e9c7ad580168ec8091f2049a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f0d538463a3eebedfcb8875281771760eee5169"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f0d538463a3eebedfcb8875281771760eee5169"}]},"branch":"refs/heads/master"},"7f022e9c67ba5086cf7916f81acedde26e50b4bc":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-09-05 14:05:22.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/893735/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/893735/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/893735/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/893735/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/893735/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/893735/2"}}},"commit":{"parents":[{"commit":"af97975ace268bea0b0c761c2270e135c73da43b","subject":"Merge \"s3api: use string for swift info request method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af97975ace268bea0b0c761c2270e135c73da43b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-09-05 13:44:56.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-09-05 14:04:59.000000000","tz":60},"subject":"debug_logger: ensure NOTICE level is registered","message":"debug_logger: ensure NOTICE level is registered\n\nThe NOTICE log level (25) is added to logging when utils.monkey_patch\nis executed. Depending on test execution order, this may not have\nhappened when some unit tests run, so LogRecords at NOTICE level may\nhave the default level name of \u0027Level 25\u0027. FakeLogger doesn\u0027t have a\nkey in its lines dict for \u0027Level 25\u0027, so those tests would blow up.\n\nCloses-Bug: #2034254\nChange-Id: I2945ae725f97ee56e9c7ad580168ec8091f2049a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7f022e9c67ba5086cf7916f81acedde26e50b4bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7f022e9c67ba5086cf7916f81acedde26e50b4bc"}]},"branch":"refs/heads/master"},"4a80963b409a5da73bc18185b4754977d1837130":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-09-05 15:22:10.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/893735/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/893735/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/893735/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/893735/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/893735/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/893735/3"}}},"commit":{"parents":[{"commit":"af97975ace268bea0b0c761c2270e135c73da43b","subject":"Merge \"s3api: use string for swift info request method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af97975ace268bea0b0c761c2270e135c73da43b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-09-05 13:44:56.000000000","tz":60},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2023-09-05 15:22:10.000000000","tz":0},"subject":"debug_logger: ensure NOTICE level is registered","message":"debug_logger: ensure NOTICE level is registered\n\nThe NOTICE log level (25) is added to logging when utils.monkey_patch\nis executed. Depending on test execution order, this may not have\nhappened when some unit tests run, so LogRecords at NOTICE level may\nhave the default level name of \u0027Level 25\u0027. FakeLogger doesn\u0027t have a\nkey in its lines dict for \u0027Level 25\u0027, so those tests would blow up.\n\nCloses-Bug: #2034254\nRelated-Change: I6f35ad41414898fb7dc5da422f524eb52ff2940f\nChange-Id: I2945ae725f97ee56e9c7ad580168ec8091f2049a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4a80963b409a5da73bc18185b4754977d1837130"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4a80963b409a5da73bc18185b4754977d1837130"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
