)]}'
{"id":"openstack%2Fswift~893861","triplet_id":"openstack%2Fswift~master~If6c2e767737f52e262fd2dc56fb0a0d34826f6cb","project":"openstack/swift","branch":"master","topic":"swift_expired_stats","attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2023-09-08 13:23:34.000000000","reason":"\u003cGERRIT_ACCOUNT_34175\u003e replied on the change","reason_account":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"}},"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2025-02-19 18:07:54.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},"34175":{"account":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"last_update":"2025-02-18 18:05:43.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}}},"removed_from_attention_set":{"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2023-09-08 15:44:27.000000000","reason":"\u003cGERRIT_ACCOUNT_597\u003e replied on the change","reason_account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-02-19 18:07:54.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"If6c2e767737f52e262fd2dc56fb0a0d34826f6cb","subject":"swift-expired-status: tool to inspect the expiry queue","status":"NEW","created":"2023-09-06 10:54:16.000000000","updated":"2025-06-26 21:23:48.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":24,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"1d3a9449df742baed1eab2f83d3e775be17e39ab","_number":893861,"virtual_id_number":893861,"owner":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-06-26 21:23:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":1,"date":"2025-06-26 19:57:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}],"CC":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-06 12:06:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-06 20:24:11.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2023-09-06 20:58:29.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"},{"updated":"2023-09-07 02:12:13.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2024-10-31 23:07:45.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"},{"updated":"2025-02-18 18:05:43.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"}],"messages":[{"id":"dd8c47eb4fc59af5262d193e3e8bd84fbdbec1f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"date":"2023-09-06 10:54:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0cdc3d029c9e5e01bea9639cef9f52f37a453987","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"date":"2023-09-06 10:55:04.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f91143bd7973b8a679e53b74a63acc4d95f270b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 12:06:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2066fd48a1924530ac90da0abb847ad8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1993e455229c41d786c9335925894db5 : SUCCESS in 13m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/72395a21a32b4444ae7b71571c91ea86 : SUCCESS in 1h 01m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c70f5fe7b1834ee48628f80de721ee6e : SUCCESS in 36m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0a7c5a838eb74b5fa96738f6ee544c1a : SUCCESS in 1h 00m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/01740c804c5148eca89b7a2145fc9f9a : SUCCESS in 16m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d619bdf03bc4425f9487c24722e5b68b : SUCCESS in 18m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dcdfe2f8b5044ff581e5e5725fed857b : SUCCESS in 19m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4489d20a2dcf43509d7ba1ed2a2c4b7e : SUCCESS in 8m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c3d74d86100240b3a585483e3cd3021c : SUCCESS in 6m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e71f61588f3c451abcd57ab89792243a : SUCCESS in 13m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/647ffa4db09c40d2be767b01f3795ee1 : SUCCESS in 17m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/77b04d56deee4d11abb854ceefece089 : SUCCESS in 13m 13s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/36a9913910814db18fe63c647974bf0f : SUCCESS in 16m 57s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e8fc5790140447c9a9ac594ace79d3e4 : SUCCESS in 19m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/abefb27d80e54ffd945b3cab7a338f40 : SUCCESS in 19m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fd2c5df010244c85a2adbab4f06af9bc : SUCCESS in 21m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f39007e6ae3f466aa52e58f99aea390e : SUCCESS in 16m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7e3167543d85473089a1914324c8208b : SUCCESS in 13m 06s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/76a7f7c3c63b40e2a16c88de3923f4eb : SUCCESS in 13m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/83b170ef30d74a00aac4d5a8e30f0950 : SUCCESS in 18m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c361e52a49274198bdabf1e9b4c1a35a : SUCCESS in 7m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/43c57399f4d2455280ce8323262e1af7 : SUCCESS in 17m 12s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/834e00d6aa1f4c4faab8b35566e8db5c : SUCCESS in 7m 07s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/99d82ae3f78343ef889781be6c74315b : SUCCESS in 59m 50s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/fc9c8bddff584885bc1f612a16593172 : SUCCESS in 1h 06m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/635ab556b4a54979a1279392a4713e21 : SUCCESS in 53m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a156a83b60ff44af988e3eefe53cf857 : SUCCESS in 53m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dead0ab43ee8469ca90175e0db57be30 : SUCCESS in 15m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6a40c765dc34a94ac53ada32d945c21 : SUCCESS in 5m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/36a15574191a4e59828caf583714d805 : SUCCESS in 19m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a841fb5be8b1405cb74877adcdaa9f02 : SUCCESS in 46m 03s","accounts_in_message":[],"_revision_number":2},{"id":"e358b407ba355ae75c7db212550ce30590845179","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 12:39:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fe9290909944913930b133e3f63500b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eb096b24586d4796a4a1744170ae1f98 : SUCCESS in 1h 02m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3bc369e4e5db42818ee1cd47c6a2d6e3 : SUCCESS in 1h 24m 09s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e19bfc673c9c4d559321e908dea2d777 : SUCCESS in 25m 44s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/65b73bb464774581b030afb7a3f85e8c : SUCCESS in 19m 56s","accounts_in_message":[],"_revision_number":2},{"id":"dbff7a19ae57ab01358fb363b5f327cecea88818","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-06 20:24:11.000000000","message":"Patch Set 2: Code-Review+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"32595c2a24e043c17b5d6f74efc0acfe60801687","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-09-06 20:58:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6ffd586bba753b1a81034eee1df4fc0803b9c52f","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-09-07 02:12:13.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"89acc298233604df1d3736457a2a6269b3f8a85b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"date":"2023-09-08 13:19:39.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"fefcf149bb16e09de7db1009894f7626c484cf66","author":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"date":"2023-09-08 13:23:34.000000000","message":"Patch Set 3:\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"68b60de3e4f88da49c92f33fdb575f79dcc0421f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 14:25:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2996b7d31ebf48a5a86c54c931dea192\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/152f3ce3008642c39eaa4c6f5d3005ec : SUCCESS in 8m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/11d76bb2792a4080b16716626677a9dc : SUCCESS in 52m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/09e121bcbdaa4a4b8050e976f8cbac36 : SUCCESS in 42m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72d77f4ad43249dc8ceffba24f8170df : FAILURE in 53m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a75f28067c0b4491824d0976d3529df0 : SUCCESS in 16m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/55905d1c834e4c6a8ad69162e11f6eae : SUCCESS in 19m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a2a174d9214a420c8fb15963a406ef17 : SUCCESS in 19m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fd62bd5db5034df0b4361128727e615f : SUCCESS in 9m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/373bb628736d4af9b02968d78cac57ab : SUCCESS in 6m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4a2c0af2ee074058873c9a69bf6d96bb : SUCCESS in 30m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/399bbc02cda046eb952996026145046c : SUCCESS in 28m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9a3464d2a2b1459ba34f25ee4b10a4db : SUCCESS in 24m 19s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9d129d7c4db24b54bbca4f940a97b806 : SUCCESS in 23m 53s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/775f33ca6a004a1884ae77b5c868f7a9 : SUCCESS in 24m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a3d081b66dd24d9f8f7fc14dc93538ea : SUCCESS in 27m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2f6f4d74a15741548652b1b371dd4836 : SUCCESS in 17m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3b982eafcf9947bca2161ed2c2ed36c9 : SUCCESS in 24m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3fa0dd781b1d413ab15557651ab8e9a2 : SUCCESS in 14m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c5e1e870853c43448fd38d22852693d6 : SUCCESS in 22m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e26cd515fd9f40a0a349498b93751765 : SUCCESS in 15m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4ab18675c6b04307841f682564424993 : SUCCESS in 6m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ba64346b3d6b44a7ab8156f34c55e81e : SUCCESS in 18m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4e527e55216f45bc8638a28d408d541c : SUCCESS in 8m 15s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d7c33252f369490f8ffdc39fb4c728e8 : SUCCESS in 1h 00m 21s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f342f6d809314d34aa55dee4964c0178 : SUCCESS in 1h 01m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/71d296ad93284df88ccf7724012b0e65 : SUCCESS in 53m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/085ab247f55f4a8d9287f8dba688128b : SUCCESS in 41m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/463e1345fe324874a78423d3ccc23087 : SUCCESS in 16m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e70363fce624b389412ba28639d0616 : FAILURE in 5m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0d86b6e59aba415d8f76fe2a21232d42 : SUCCESS in 23m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78076537ca454b6f93bb557e45f1ead2 : SUCCESS in 56m 56s","accounts_in_message":[],"_revision_number":3},{"id":"3bbfb92c97a0eb6d792d144a042d0fe52834182e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"date":"2023-09-08 14:33:42.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"bfe0f9272c1bb24320d3fcf5a8345b3df6c4e395","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-09-08 15:02:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ea6599c65dd0ecc2e8710a627996e7b6805b52eb","author":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"date":"2023-09-08 15:14:07.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fdb4b62fde10de2cee322d0c666c275b4d907a26","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-09-08 15:44:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"aa56403d0e78c88bbd1ab2d1b23b93f7b441ba20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 15:50:49.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/207fe89697dc4025942a5ecc8465f2ea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f89c3ec63684a44b0d4b46cebe0d21c : SUCCESS in 9m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/7f34f351c40b4d048a89764d595b3dca : SUCCESS in 1h 11m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/619bc23f221f42b985864b3472575e0b : SUCCESS in 54m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d1644c4a71494a5d8508014ce0096fb5 : SUCCESS in 56m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/85e52a6148a04006986661662d4e82ca : SUCCESS in 17m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0173258015a548a697b61e3d37189fa1 : SUCCESS in 17m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f33794ebf8a5432690af9898e8ead481 : SUCCESS in 20m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fc14fbf952f24cb79fb13743e4649fd3 : SUCCESS in 9m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/27b0598550ef4d4ab2b5ef67fc55edb5 : SUCCESS in 6m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/aff5ce7abb284bd595a6841f6c65dac8 : SUCCESS in 18m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d18d5c354a52496c9f479f015b3019aa : SUCCESS in 15m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c355520b9e344e90a7da438089f0ba67 : SUCCESS in 13m 16s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6eb55449178d42a09fe4f1012cb9d4e7 : SUCCESS in 16m 59s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/4948ec150bb1495dacf5c30521b5013e : POST_FAILURE in 18m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/eeb001a413954d7ca5ad0871f459a222 : SUCCESS in 18m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c8d5e583910d43d0b51bceae8b7bf4d6 : SUCCESS in 16m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4187ac2ff34a4e6f869abcbfd0217cd0 : SUCCESS in 18m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cc7917be93db4c53a7d471b19e50a3bb : SUCCESS in 14m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/492d068e8d344f718ae1b13d3d2b7ac2 : SUCCESS in 16m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e228db8ef8ff4e2dade231b6e17ec011 : SUCCESS in 21m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e4feae8345884d3092ba3f7dc8502222 : SUCCESS in 7m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/57579b053a334354a34e62cfdd9c05d4 : SUCCESS in 20m 58s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/27bd9e83148949ffb47102f248572ea2 : SUCCESS in 7m 41s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5276a76fbe7d442c85ab298ce466d0cf : SUCCESS in 1h 06m 12s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/929c58e8fa4440588464317d1d7eb47d : SUCCESS in 1h 05m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0e7dfa85a91c4b4c8e772a4eb77c0d8e : SUCCESS in 53m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7328aae750064247bf629798ffce23d7 : SUCCESS in 54m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8203bf92f2ad427db7894946b4dd0ef0 : SUCCESS in 16m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9385d57c673c4e449938d63bec370291 : SUCCESS in 5m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3fc555d441564a81886257c4b6a8e5a2 : SUCCESS in 23m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6d0a2ce220047118d46eb65d60c9b8a : SUCCESS in 1h 00m 59s","accounts_in_message":[],"_revision_number":4},{"id":"42276690471074731c33224839699a1209888cd9","author":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"date":"2023-09-08 15:54:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"259d529a24d969472b2a173df4f23cefbbd8c0d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 15:59:45.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7136e10d9dc4396bb4607a247b27a67\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/13f5089c927c4a71bb96162afb91862c : SUCCESS in 17m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f62ac0d6b29c4ad39ec14232de8d9c4c : SUCCESS in 37m 49s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/00389607270f4992a5017adc436ef0cc : SUCCESS in 36m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e0384acb768f45439d587c8943bb6e6e : SUCCESS in 18m 37s","accounts_in_message":[],"_revision_number":4},{"id":"1b9ae4777a67dfd396a5bc62d32136f9b6ec7c22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 17:17:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d940cbc2ad242029c27af2b5accbabe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5984736b0faa4f7ca56b0101ccd363cd : SUCCESS in 7m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/115b9f161fd74819af0ced7b3be105ec : SUCCESS in 1h 03m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ade6922d6c824cf3b8ee1af9123fd714 : SUCCESS in 53m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/009451607adb47f9b2d02a87687b740a : SUCCESS in 51m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/de8faa2adc654797ad9949cf0e8bd2bb : SUCCESS in 16m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/312666367f9047ec95c74541eb534bc8 : SUCCESS in 18m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0eb6552cd6f34feebd34257717e1b348 : SUCCESS in 20m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/380e5f7aba5f4af999e5ec37b7983931 : SUCCESS in 9m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e7872ebc37de4dca81e8055b29de1a78 : SUCCESS in 5m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/69f71a2a87754cc89ad4a3bb3811ed13 : SUCCESS in 20m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3ea73a8c57e44f0f8576eb519eb5559f : SUCCESS in 17m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c37a6a3c942b44af827c47e44aa8b1f5 : SUCCESS in 16m 13s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/82ea09f955d142b49a657afc62683a59 : SUCCESS in 17m 24s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/dc66eae0f4464e44908b5bec7a931373 : SUCCESS in 16m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c41e6bb1b3094ce2863bc1103b6d0e42 : SUCCESS in 22m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bd2225b723d3419eba7c8d4457852e48 : SUCCESS in 24m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/98d2602f5f2d485b86ac53e3612fd4aa : SUCCESS in 18m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4ca3889b51bd401fa17f4939ef4bebf4 : SUCCESS in 14m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/34faeec4ca8d49f9a6bf96c821fe5ea4 : SUCCESS in 15m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c7ec7ddc55434307a727879dea00c85d : SUCCESS in 18m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/739b0e96c21f42f296cd16d5844c91b5 : SUCCESS in 7m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fcd6adcb258e4f6ea161cf1902c5bc8f : SUCCESS in 19m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c15e3fa1940743bfb7e303a0984e402e : SUCCESS in 8m 03s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4cdf4d76d4054fe58d02a4349096dd5c : SUCCESS in 59m 53s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a608e82c02ac445a897f9a33a61205ae : SUCCESS in 57m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2274b8cd88a049278f43f85414741132 : SUCCESS in 50m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/08f73684bd8046ffbbebebd6bb5f762c : SUCCESS in 50m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e6cf48285b8747eb83f8cf5d5913cf3a : SUCCESS in 16m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22f36c70f247495b833b42a44525936d : SUCCESS in 4m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bb362f1cd0b14876b15122e6ffe5d536 : SUCCESS in 20m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a303fb33f364a2b99ea22f20f22b3f6 : SUCCESS in 1h 03m 06s","accounts_in_message":[],"_revision_number":4},{"id":"977dc2f5d4107c207eef32933fcd1c1b7e20b70e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-04-10 13:43:46.000000000","message":"Patch Set 5: Patch Set 4 was rebased on behalf of \u003cGERRIT_ACCOUNT_34175\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"}],"_revision_number":5},{"id":"6b03d02775713b890c6dabc2b26870423f46acd8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 14:34:00.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d308facc7cd8401c9a2dc6e3ae2a4b45\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/70d19f7b6b5c4af9831937c2421d17b5 : SUCCESS in 14m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/58e356a17b324241aec71fcb957044fb : SUCCESS in 39m 52s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4c448299ae314f7594ff640a3c7798c4 : SUCCESS in 19m 27s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/be9b68132e16409c88581e0a23aa7f20 : SUCCESS in 48m 35s","accounts_in_message":[],"_revision_number":5},{"id":"eda341420a66a79496c517d73860ed0e2e1d2ca8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 14:53:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/385b5a93d0cd4aa2ab68681241dd047e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41802dad4c744046ba1a4d2c42123d3e : SUCCESS in 8m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/b70f2dd9b58c44c69eb274e85f4faf87 : SUCCESS in 45m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/366cff94bf8b44c4a4d728a639f47230 : SUCCESS in 56m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b717409212e3487fb5e43c270ecdfb0b : SUCCESS in 1h 03m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/78613b47091f4f1c9c82780571e04b51 : SUCCESS in 17m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6abf2879dd384faea01018475c44fc22 : SUCCESS in 21m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1b3cc7e44c4e4a51b74860ad6009055b : SUCCESS in 19m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/495b8c4e97384deb89bb03f5440f4dfe : SUCCESS in 11m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/327e024d555d47e7a6fb1b1f91381e86 : SUCCESS in 4m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/76eaa3925c9f4e02bfd843f43fe50f66 : SUCCESS in 20m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/843e99bb8ec342cbb94cf6f0ba8f9965 : SUCCESS in 21m 17s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3ce85d2804f84a09be60524d2027c2c6 : SUCCESS in 17m 38s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/60ad4fb4037240cfa93cce4af0e4d944 : SUCCESS in 18m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/da75fd4c2da144fb86367b7af7727cca : SUCCESS in 19m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8aa59fbe457349e2a58d73d7a517c852 : SUCCESS in 22m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b11bf869b54647acb8cb2269c559c397 : SUCCESS in 19m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/51e75fd332414685aac1a3d8400b54c6 : SUCCESS in 16m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9184d9c32986446e95080cd1b7025cb5 : SUCCESS in 16m 09s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bb0b35e520b54895a60d6695c31393ba : SUCCESS in 22m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1491dc5e6c014233a320242a8112f778 : SUCCESS in 7m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8c140f0e51b4443a82d3663e17798a65 : SUCCESS in 16m 16s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a1d8110ef1444bc8bfb02e79b3540b83 : SUCCESS in 6m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/dd3c31b1ad704521b67c48eb6e67e756 : SUCCESS in 44m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d841ebf49cef4003bed1381964cd1f9d : SUCCESS in 45m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a49f50b8183d4e16883f07bff583b0d6 : SUCCESS in 51m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48bd9397644d48c6b211192e5c3ea122 : SUCCESS in 28m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dccba7cd17e7405d95b838a6aba14f8b : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/50201ee96130450e9647a9ddeb7ac286 : SUCCESS in 19m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed2f81c837464b6c83c7d62a87e87791 : SUCCESS in 1h 00m 52s","accounts_in_message":[],"_revision_number":5},{"id":"c35cd49e50b0646eb712ceeeb5be1c50fbd0e522","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-02-18 18:05:43.000000000","message":"Patch Set 5: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"287cdd5112fbfb0d82cca5f642e65730d22c0fe6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-19 18:07:54.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"001b049d010d4dfa2c8f9c65f30769c8cc4426df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-04-23 20:14:37.000000000","message":"Patch Set 6: Patch Set 5 was rebased on behalf of \u003cGERRIT_ACCOUNT_34175\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"}],"_revision_number":6},{"id":"283b0bd87c28bb482cbd74f7ef51cc9490f59e3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 20:51:14.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4587a78bec34aad8be2a28c5daeb163\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eae4f44d28c8483a9a4369b29c5a4a9e : SUCCESS in 34m 51s","accounts_in_message":[],"_revision_number":6},{"id":"554f9e1a872361eeccb80b0d2ffc37dab22692a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 21:25:09.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91376d57863346e388b646ef9cfc9cb5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69c23970f99e4e5ca04146da3b3a8b23 : SUCCESS in 7m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/02be80910e564c16a9ec6757e4835e61 : SUCCESS in 49m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6f296fb378f24d9fa15dcde6bc58cea7 : SUCCESS in 43m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/91aa22aad46d4a6ca19c2d933c809963 : SUCCESS in 52m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1cf9a60aa8ed401ea3e74da075bf2968 : SUCCESS in 1h 03m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bc04329f6ebd4b01808f2408d11ca678 : SUCCESS in 17m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/02323618869343309bec1dfed3dd6588 : SUCCESS in 18m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/58d4b0837e35499a865e034e9e0aacfc : SUCCESS in 19m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/564060818be347fba3dc1064a925d4ba : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d306dcd82ec14fa498764e17890f14e0 : SUCCESS in 18m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ff85c654d4534e158790191f7edc2697 : SUCCESS in 24m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ddb712e4e3c74e2587fe0aa73258b862 : SUCCESS in 31m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/684baffdbfab40e48caa369731d491f0 : SUCCESS in 14m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/047db88446e0409491e0e0056a007467 : SUCCESS in 15m 52s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1aaee874ee7340e2be56e1994182f6de : SUCCESS in 15m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a8a2082813674764b8168ef2e61ba07e : SUCCESS in 4m 23s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/506afc4cbe6c4e0289c5b5f24fcf2faf : SUCCESS in 8m 15s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c5a19539a41f4433a460ba5a0756c2c2 : SUCCESS in 3m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3dce5bd9a09544ddab9dbc6b2c8ab02c : SUCCESS in 35m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0a4893d4629049c4a9fd9222f9fc1ece : SUCCESS in 18m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cf657f73d8ea48fb8cd2f37a0b2cc1c2 : SUCCESS in 21m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc230c5bbdb94f75a617216dff95db71 : SUCCESS in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4893b38af6654420b1010a399a7e5e9a : SUCCESS in 3m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a538c700bf84439ba2817325c46328a2 : SUCCESS in 15m 24s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37fee40bc8534dfa89b912de103a5c68 : SUCCESS in 56m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a8fdd8abf0554241a2adf7edb8102608 : SUCCESS in 58m 39s","accounts_in_message":[],"_revision_number":6},{"id":"173e9334aba12f4e9063d744030cac392354fec9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-04-29 13:35:44.000000000","message":"Patch Set 7: Patch Set 6 was rebased on behalf of \u003cGERRIT_ACCOUNT_34175\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"}],"_revision_number":7},{"id":"b80e0d478ae67c1a5eb77c117570cefe598732f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 14:17:52.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/967175d3ccd04ab8baf56b429feb0a6b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/052afa62f84845259c7cede70b291bef : SUCCESS in 40m 35s","accounts_in_message":[],"_revision_number":7},{"id":"ef8c93fc04a4b1b3a74f98d65b4f4e361fad1c50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 14:43:24.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a4dd77cf53e469b8e996cfb45d991a0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ed17ffd070244bbb7603f31b684c2c8 : SUCCESS in 4m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/1e071a6762f844bd9d0b87e403b38ca3 : SUCCESS in 27m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/113ce3d501b84e36b5f1eeadc13da118 : SUCCESS in 46m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f234c45349ce4237a7b222b8548aec92 : SUCCESS in 31m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a91775d6b0644c09aaf689f644bc5da : SUCCESS in 49m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/368e721b463d46229da8b0d94e70065c : SUCCESS in 17m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e1e8cf705917434db5f3fc1209b9f365 : SUCCESS in 18m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dd7fc5b5aa034ea9bf698549c3031e57 : SUCCESS in 19m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3bb10a5ad791427aa277ebf4a7afc735 : SUCCESS in 4m 41s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bdc733d1580845b3b89afb1006cc1b42 : SUCCESS in 16m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7d6227d3a1c74d9aa7286a3d12f26188 : SUCCESS in 8m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b0324dbc20ac456ca28dd55456bb2619 : SUCCESS in 26m 00s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d18acd44b1954fa7ac6330873248d3f8 : SUCCESS in 10m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b37928f4f33444118dc5fd747d17e3e2 : SUCCESS in 15m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b49199131b9c42a2b43723424cb7e6cd : SUCCESS in 7m 41s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0d56159b91ff4ea981b98bfc61a0b36b : SUCCESS in 14m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ad421dc9af0e4e8586777e949aac7394 : SUCCESS in 7m 11s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e1c5e09bfbb047f9922217471f7598d1 : SUCCESS in 7m 32s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1eab68bde69c4fe1a9c258a8c54bfbbc : SUCCESS in 8m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/25b8ca6a9c3e4440b6dbdcae445f9a7d : SUCCESS in 1h 02m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0fd258159efa4da9ad46c993ef553a3d : SUCCESS in 17m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8850fce569e7488587fd2dd38bea7693 : SUCCESS in 52m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50883534037440c7815fccae166d0f48 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfc2a522d7534611bec75b8017cbca8f : SUCCESS in 3m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9b287037dcee48409bdae97bf2eab7f0 : SUCCESS in 8m 57s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/956e1a8783d945ff90e594b3ee8e6f7d : SUCCESS in 51m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eb1932fb1c0a46248fd4952111a18896 : SUCCESS in 52m 23s","accounts_in_message":[],"_revision_number":7},{"id":"1a463dbbcbb08bea9b2a72bf9c99613e72b3bf58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-06-26 19:57:45.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"8c38176399689c8a3a79865101ed6d758fbf6022","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 21:15:55.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0560fa9d83ec40bf90f8e649a7f5ff37\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/67d863caf9404c3fb2c29f5e33aaae61 : SUCCESS in 1h 16m 02s","accounts_in_message":[],"_revision_number":8},{"id":"1d3a9449df742baed1eab2f83d3e775be17e39ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 21:23:48.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9418ec69d972440e88b59ecf7ca8d256\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77d5d66bf6ff4f5390dd6b5ba3a1ea72 : SUCCESS in 5m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/3721ec04da80414abe311abb08c79218 : SUCCESS in 1h 04m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bf75faed5d014696a67d022d2b735f05 : SUCCESS in 1h 02m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/38fce1639b404d90899165b3cb1cacd2 : SUCCESS in 30m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/66d1fddc271c40aca6c663f1678fd92c : SUCCESS in 1h 25m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fc2ba0ddc9e9410082011b5edcd079b5 : SUCCESS in 20m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/91865872b03740f481a4cb55cf55db42 : SUCCESS in 10m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ece620259363492892ce528fd7067bf1 : SUCCESS in 21m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b184780f32ee4e50b565172cf501f381 : SUCCESS in 4m 59s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5a8a74f606e74bcfae24d4815dba6970 : SUCCESS in 8m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/14e1a81014464fec9d7b91a87f7fad59 : SUCCESS in 21m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/fff1ea3f5f4441b89c2fc03b89547b0b : SUCCESS in 27m 11s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a32336ee5d87478482f35738e167f096 : SUCCESS in 10m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/077d9352923441c39815afb21088699b : SUCCESS in 13m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a3f1339d01054af9bd6357f2db4ce681 : SUCCESS in 15m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ec25695509c64330981fc8f440975e5b : SUCCESS in 17m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e309dbd0c7a94ea9aadad65d692402ab : SUCCESS in 6m 55s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0161c6e281f444749dd236134b012895 : SUCCESS in 17m 14s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/59fbee5175c4476191c79ba13ce51192 : SUCCESS in 9m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/803b9e90891c4c5db3e9cc26d145cdc7 : SUCCESS in 45m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c4f1888b2f714d8abe402fd8b7536fb2 : SUCCESS in 57m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a996c6c585fc4662a1f9c7f71753392f : SUCCESS in 35m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00712cfa04eb46749933c6f4513e5bc1 : SUCCESS in 21m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a19ee15b5602487e98bfc14b0e216bd3 : SUCCESS in 5m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8ce35915c9944131a725d208acd16006 : SUCCESS in 21m 27s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b56b8645d612401eb987553589261ec0 : SUCCESS in 51m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1defb6e289574677a219703c49c218e3 : SUCCESS in 53m 22s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"ca9c450b028a84b5b59f7f47864adb0e3a1c9982","revisions":{"321d2631e57f6019ed6dbd4dac98b0624321bac0":{"kind":"REWORK","_number":1,"created":"2023-09-06 10:54:16.000000000","uploader":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"ref":"refs/changes/61/893861/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/893861/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/893861/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/893861/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/893861/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/893861/1"}}},"commit":{"parents":[{"commit":"af97975ace268bea0b0c761c2270e135c73da43b","subject":"Merge \"s3api: use string for swift info request method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af97975ace268bea0b0c761c2270e135c73da43b"}]}],"author":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2023-09-06 10:50:01.000000000","tz":60},"committer":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2023-09-06 10:50:01.000000000","tz":60},"subject":"swift-expired-status CLI tool to inspect the expiry queue","message":"swift-expired-status CLI tool to inspect the expiry queue\n\nThis is a tool for inspecting the state of the expiry queue. It\noutputs the total number of entries in the queue, as well as the\nnumber of pending and stale entries.\n\nBased on a gist by Clay Gerrard:\nhttps://gist.github.com/clayg/7f66eab2a61c77869e1e84ac4ed6f1df\n\nSubsequently tidied up, modernised, and a logic error fixed by Matthew\nVernon; we think this is a useful enough tool for Swift administrators\nto make it more widely available.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nSigned-off-by: Matthew Vernon \u003cmvernon@wikimedia.org\u003e\nChange-Id: If6c2e767737f52e262fd2dc56fb0a0d34826f6cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/321d2631e57f6019ed6dbd4dac98b0624321bac0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/321d2631e57f6019ed6dbd4dac98b0624321bac0"}]},"branch":"refs/heads/master"},"5a0cf11cf51c4e69307243db42b7708a51440c8e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-09-06 10:55:04.000000000","uploader":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"ref":"refs/changes/61/893861/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/893861/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/893861/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/893861/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/893861/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/893861/2"}}},"commit":{"parents":[{"commit":"af97975ace268bea0b0c761c2270e135c73da43b","subject":"Merge \"s3api: use string for swift info request method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af97975ace268bea0b0c761c2270e135c73da43b"}]}],"author":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2023-09-06 10:50:01.000000000","tz":60},"committer":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2023-09-06 10:54:47.000000000","tz":60},"subject":"swift-expired-status: tool to inspect the expiry queue","message":"swift-expired-status: tool to inspect the expiry queue\n\nThis is a tool for inspecting the state of the expiry queue. It\noutputs the total number of entries in the queue, as well as the\nnumber of pending and stale entries.\n\nBased on a gist by Clay Gerrard:\nhttps://gist.github.com/clayg/7f66eab2a61c77869e1e84ac4ed6f1df\n\nSubsequently tidied up, modernised, and a logic error fixed by Matthew\nVernon; we think this is a useful enough tool for Swift administrators\nto make it more widely available.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nSigned-off-by: Matthew Vernon \u003cmvernon@wikimedia.org\u003e\nChange-Id: If6c2e767737f52e262fd2dc56fb0a0d34826f6cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a0cf11cf51c4e69307243db42b7708a51440c8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a0cf11cf51c4e69307243db42b7708a51440c8e"}]},"branch":"refs/heads/master"},"a2fd891b8cc2a8ab00dfc0e7237a3247f841274d":{"kind":"REWORK","_number":3,"created":"2023-09-08 13:19:39.000000000","uploader":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"ref":"refs/changes/61/893861/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/893861/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/893861/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/893861/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/893861/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/893861/3"}}},"commit":{"parents":[{"commit":"af97975ace268bea0b0c761c2270e135c73da43b","subject":"Merge \"s3api: use string for swift info request method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af97975ace268bea0b0c761c2270e135c73da43b"}]}],"author":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2023-09-06 10:50:01.000000000","tz":60},"committer":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2023-09-08 13:19:26.000000000","tz":60},"subject":"swift-expired-status: tool to inspect the expiry queue","message":"swift-expired-status: tool to inspect the expiry queue\n\nThis is a tool for inspecting the state of the expiry queue. It\noutputs the total number of entries in the queue, as well as the\nnumber of pending and stale entries.\n\nBased on a gist by Clay Gerrard:\nhttps://gist.github.com/clayg/7f66eab2a61c77869e1e84ac4ed6f1df\n\nSubsequently tidied up, modernised, and a logic error fixed by Matthew\nVernon; we think this is a useful enough tool for Swift administrators\nto make it more widely available.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nSigned-off-by: Matthew Vernon \u003cmvernon@wikimedia.org\u003e\nChange-Id: If6c2e767737f52e262fd2dc56fb0a0d34826f6cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2fd891b8cc2a8ab00dfc0e7237a3247f841274d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2fd891b8cc2a8ab00dfc0e7237a3247f841274d"}]},"branch":"refs/heads/master"},"c88992460a846bc1c100f38652468b301889514d":{"kind":"REWORK","_number":4,"created":"2023-09-08 14:33:42.000000000","uploader":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"ref":"refs/changes/61/893861/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/893861/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/893861/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/893861/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/893861/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/893861/4"}}},"commit":{"parents":[{"commit":"af97975ace268bea0b0c761c2270e135c73da43b","subject":"Merge \"s3api: use string for swift info request method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af97975ace268bea0b0c761c2270e135c73da43b"}]}],"author":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2023-09-06 10:50:01.000000000","tz":60},"committer":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2023-09-08 14:33:32.000000000","tz":60},"subject":"swift-expired-status: tool to inspect the expiry queue","message":"swift-expired-status: tool to inspect the expiry queue\n\nThis is a tool for inspecting the state of the expiry queue. It\noutputs the total number of entries in the queue, as well as the\nnumber of pending and stale entries.\n\nBased on a gist by Clay Gerrard:\nhttps://gist.github.com/clayg/7f66eab2a61c77869e1e84ac4ed6f1df\n\nSubsequently tidied up, modernised, and a logic error fixed by Matthew\nVernon; we think this is a useful enough tool for Swift administrators\nto make it more widely available.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nSigned-off-by: Matthew Vernon \u003cmvernon@wikimedia.org\u003e\nChange-Id: If6c2e767737f52e262fd2dc56fb0a0d34826f6cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c88992460a846bc1c100f38652468b301889514d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c88992460a846bc1c100f38652468b301889514d"}]},"branch":"refs/heads/master"},"ba9b462dd10295026bbdf4df6f81d156d196b482":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-04-10 13:43:46.000000000","uploader":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/61/893861/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/893861/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/893861/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/893861/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/893861/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/893861/5"}}},"commit":{"parents":[{"commit":"6387949a540e871e19062de8131503a8ac2ceef7","subject":"lower-constraints: update to mock 3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6387949a540e871e19062de8131503a8ac2ceef7"}]}],"author":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2023-09-06 10:50:01.000000000","tz":60},"committer":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2024-04-10 13:43:46.000000000","tz":0},"subject":"swift-expired-status: tool to inspect the expiry queue","message":"swift-expired-status: tool to inspect the expiry queue\n\nThis is a tool for inspecting the state of the expiry queue. It\noutputs the total number of entries in the queue, as well as the\nnumber of pending and stale entries.\n\nBased on a gist by Clay Gerrard:\nhttps://gist.github.com/clayg/7f66eab2a61c77869e1e84ac4ed6f1df\n\nSubsequently tidied up, modernised, and a logic error fixed by Matthew\nVernon; we think this is a useful enough tool for Swift administrators\nto make it more widely available.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nSigned-off-by: Matthew Vernon \u003cmvernon@wikimedia.org\u003e\nChange-Id: If6c2e767737f52e262fd2dc56fb0a0d34826f6cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba9b462dd10295026bbdf4df6f81d156d196b482"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba9b462dd10295026bbdf4df6f81d156d196b482"}]},"branch":"refs/heads/master","description":"Rebase"},"6f2ef63214a6be385c567a8829dc5cbc429f34a0":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-04-23 20:14:37.000000000","uploader":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/61/893861/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/893861/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/893861/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/893861/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/893861/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/893861/6"}}},"commit":{"parents":[{"commit":"1ca073ce1dac2d0e3554399fc8e8a9e0af5e4c06","subject":"Let WSGI know the length","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ca073ce1dac2d0e3554399fc8e8a9e0af5e4c06"}]}],"author":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2023-09-06 10:50:01.000000000","tz":60},"committer":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2025-04-23 20:14:37.000000000","tz":0},"subject":"swift-expired-status: tool to inspect the expiry queue","message":"swift-expired-status: tool to inspect the expiry queue\n\nThis is a tool for inspecting the state of the expiry queue. It\noutputs the total number of entries in the queue, as well as the\nnumber of pending and stale entries.\n\nBased on a gist by Clay Gerrard:\nhttps://gist.github.com/clayg/7f66eab2a61c77869e1e84ac4ed6f1df\n\nSubsequently tidied up, modernised, and a logic error fixed by Matthew\nVernon; we think this is a useful enough tool for Swift administrators\nto make it more widely available.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nSigned-off-by: Matthew Vernon \u003cmvernon@wikimedia.org\u003e\nChange-Id: If6c2e767737f52e262fd2dc56fb0a0d34826f6cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f2ef63214a6be385c567a8829dc5cbc429f34a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f2ef63214a6be385c567a8829dc5cbc429f34a0"}]},"branch":"refs/heads/master","description":"Rebase"},"a034a5d377d829728671241db932677e6cce86b4":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-04-29 13:35:44.000000000","uploader":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/61/893861/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/893861/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/893861/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/893861/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/893861/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/893861/7"}}},"commit":{"parents":[{"commit":"f882ba76dda5e6f0df374d8a628accc1f19c6e93","subject":"Merge \"Remove tags from README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f882ba76dda5e6f0df374d8a628accc1f19c6e93"}]}],"author":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2023-09-06 10:50:01.000000000","tz":60},"committer":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2025-04-29 13:35:44.000000000","tz":0},"subject":"swift-expired-status: tool to inspect the expiry queue","message":"swift-expired-status: tool to inspect the expiry queue\n\nThis is a tool for inspecting the state of the expiry queue. It\noutputs the total number of entries in the queue, as well as the\nnumber of pending and stale entries.\n\nBased on a gist by Clay Gerrard:\nhttps://gist.github.com/clayg/7f66eab2a61c77869e1e84ac4ed6f1df\n\nSubsequently tidied up, modernised, and a logic error fixed by Matthew\nVernon; we think this is a useful enough tool for Swift administrators\nto make it more widely available.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nSigned-off-by: Matthew Vernon \u003cmvernon@wikimedia.org\u003e\nChange-Id: If6c2e767737f52e262fd2dc56fb0a0d34826f6cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a034a5d377d829728671241db932677e6cce86b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a034a5d377d829728671241db932677e6cce86b4"}]},"branch":"refs/heads/master","description":"Rebase"},"ca9c450b028a84b5b59f7f47864adb0e3a1c9982":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-06-26 19:57:45.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/61/893861/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/893861/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/893861/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/893861/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/893861/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/893861/8"}}},"commit":{"parents":[{"commit":"1cd20f87dec469dc4f7c9ac27d99ffec0b859301","subject":"Merge \"Add labeled metrics to proxy-logging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1cd20f87dec469dc4f7c9ac27d99ffec0b859301"}]}],"author":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2023-09-06 10:50:01.000000000","tz":60},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-05-22 20:45:20.000000000","tz":-420},"subject":"swift-expired-status: tool to inspect the expiry queue","message":"swift-expired-status: tool to inspect the expiry queue\n\nThis is a tool for inspecting the state of the expiry queue. It\noutputs the total number of entries in the queue, as well as the\nnumber of pending and stale entries.\n\nBased on a gist by Clay Gerrard:\nhttps://gist.github.com/clayg/7f66eab2a61c77869e1e84ac4ed6f1df\n\nSubsequently tidied up, modernised, and a logic error fixed by Matthew\nVernon; we think this is a useful enough tool for Swift administrators\nto make it more widely available.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nSigned-off-by: Matthew Vernon \u003cmvernon@wikimedia.org\u003e\nChange-Id: If6c2e767737f52e262fd2dc56fb0a0d34826f6cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca9c450b028a84b5b59f7f47864adb0e3a1c9982"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca9c450b028a84b5b59f7f47864adb0e3a1c9982"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
