)]}'
{"id":"openstack%2Fswift~893991","triplet_id":"openstack%2Fswift~master~I71b00c888e7fd4d7eaf85a46cfee762268910a0e","project":"openstack/swift","branch":"master","topic":"unraisablehook","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-09-08 16:21:40.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I71b00c888e7fd4d7eaf85a46cfee762268910a0e","subject":"Add our own sys.unraisablehook","status":"NEW","created":"2023-09-06 22:19:20.000000000","updated":"2023-09-08 17:15:10.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"9489ee51462ccc2803fac7180abdd2017f17a6d0","_number":893991,"virtual_id_number":893991,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-09-08 17:15:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-06 23:27:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7700e6dc4cfeedda1c31f0e9c8f3c86340cc0d21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-06 22:19:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35a1f55c58086f90cce8e512bb94d6d61bc4c8ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 23:27:38.000000000","message":"Patch Set 1: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ede9f7202abe42d2a941d135857d7764\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e07eefe6bef441fa7aa249710e6e5dd : SUCCESS in 9m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/2bddc852ee4f488c8f80a01ccb30d545 : SUCCESS in 1h 02m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/712163cf4f1b4c8380b0424288efc9f7 : SUCCESS in 40m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f69220831c740ffa2106df60ea92f6f : SUCCESS in 46m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7305d01acf274507bc539fd28f90cd49 : SUCCESS in 18m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/63eba1d66efd45b3bfb985883e7d977b : SUCCESS in 17m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b429dab3698742a1adfa7d8ff4e28275 : SUCCESS in 18m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e278a6dca0b64652ac0a21f367c34e56 : SUCCESS in 7m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/71fb74a3664a4c7c8eaef71ce8dee0bf : SUCCESS in 6m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/00e7df1657e84b86b21437bba6989efe : FAILURE in 9m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5ff0cb392d5940bd8c75da24b97b01c2 : FAILURE in 9m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/40ad2bdeaafa4eeb9aa8ade791d62bff : FAILURE in 15m 48s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/657f34551edf43b08612208755b92e41 : FAILURE in 18m 53s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/09effb1e80784530be185ccbfc1b4787 : FAILURE in 20m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/54dbdcc056cf4acf99a3b3d0edc8b341 : SUCCESS in 17m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5838b35649914cde9c86c2a71e386b28 : SUCCESS in 16m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8a44abec23eb4e119072cd20c520eb6e : SUCCESS in 17m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/31919129d8754f0cb028cf3b97fe1ba8 : SUCCESS in 13m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8a955f4529e741adbb70107e315c690f : SUCCESS in 18m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/67ca99aa218640d9ba87b0f719ebfbd0 : SUCCESS in 11m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a3a7af5dbbf24f3a899f31a50676a846 : SUCCESS in 6m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/93efa7de4de04ef392ae40143aff86c8 : SUCCESS in 18m 03s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f8aa97fafd884a878c272d8e9840e8e2 : SUCCESS in 7m 51s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7cb407d19964442ca07bf4aabe74533e : SUCCESS in 49m 51s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a639eee82ad434691bd776baeb4a393 : SUCCESS in 56m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/093d7a28c3864e3abdea28b76c37c1e3 : SUCCESS in 42m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a4d73f8f485f4554b8afca0f8b2e1698 : SUCCESS in 42m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39d8c942b71d418f9424a2c1014e8cca : FAILURE in 10m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95d9229659e1451290a54cb5e9bc0db5 : FAILURE in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/86dae1fa9c824e78bc2233e73684a4f4 : SUCCESS in 20m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f9d49be8aa34ea3892802e0cf21feb7 : SUCCESS in 45m 36s","accounts_in_message":[],"_revision_number":1},{"id":"d6b6f3c6149cd7584277690063954e181c73b98f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 23:33:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96a9e51e2b5642f28ff954582c5a2de5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b6755047c71f4055a637e1a4449776fc : FAILURE in 1h 11m 48s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/44e07fd91a294483a3e5e45038c136e5 : FAILURE in 1h 06m 50s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/47a264b532aa446a86978ff3803b2791 : SUCCESS in 20m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d3414d8dbeb4094824cb50cdadfb1dc : SUCCESS in 19m 04s","accounts_in_message":[],"_revision_number":1},{"id":"e04bb6c76db2e986cabaf9301566032bac0b359f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-07 18:32:28.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1bcca300f2496ec17e1af7569cc90e6ca27080fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-07 23:34:29.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"fd167e16d0974ee9646aa0acd6580213afd4a842","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 00:22:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a7a30ee76854213af8a706db1d27266\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ee4ebf657eda402fbee165202e758c1e : SUCCESS in 13m 12s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/679a08fd018e4ee59d4e2134f0daf343 : SUCCESS in 46m 45s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ee96c299a5374c0ab50f426ef6c53df5 : SUCCESS in 18m 42s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/69881d6023b44bf78f5e9afc115084d4 : SUCCESS in 43m 20s","accounts_in_message":[],"_revision_number":2},{"id":"64a8cee679750cee213638342b5cfea74840ebbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 00:43:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f258278c145420f8eae188f84f3001c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/300cb51647ac423d8c51e407ef4ab234 : SUCCESS in 10m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/0b0dc7f1e51049b0ab72688c71918acf : SUCCESS in 1h 03m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b5da96614145453e94f4ff67c7716234 : SUCCESS in 47m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9b6690917e1401d8b5d3f3769c58291 : SUCCESS in 1h 03m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/098ddbfd82d64495bc94919dadbec73c : SUCCESS in 18m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/deb283b5b2354ce8b289f7ee205718f3 : SUCCESS in 18m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ce7089f4c5944e43b14a54f0483ae372 : SUCCESS in 18m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/19bcf446b22248a7ad41ae544f860307 : SUCCESS in 8m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/90113e9718b04d9eabe25faf1d3199fd : SUCCESS in 5m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d1974e6bca46462eb948102bf7b1afe6 : FAILURE in 20m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/17ddfefc62444e0c9d3867d014e39022 : SUCCESS in 17m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/58b0acdce1c5464187e7836ce9fafc19 : SUCCESS in 11m 24s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/83aec34a6f5a4a8d8477ce519ebdf9d4 : SUCCESS in 17m 42s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b2186eee584a4763b1e4d3c5186d1658 : SUCCESS in 16m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d95d3e664f34490a93108e10e3ef1851 : SUCCESS in 18m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0916ab3613e8481ea710aa8deea16eed : SUCCESS in 14m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8e320fafd44747929fe753b7afa8e2c9 : SUCCESS in 19m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/42c5fdbb59164e77b20cc2f748c0dc93 : SUCCESS in 16m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4d45e11086da43dfaaa20c1d42bc698c : SUCCESS in 17m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/dfb447e50c204dbb924a4dfc3d8081cd : SUCCESS in 14m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/27a888bfad6e42c79cef3c9326c05c43 : SUCCESS in 7m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b59a916525754f9e81a7b43327978d44 : SUCCESS in 19m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7db08af6ff534ec9b7fcc945b1e305eb : SUCCESS in 9m 03s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ef3736ae7d924432bc2ababd24cbc32d : SUCCESS in 50m 50s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7a443891127248d6946a493cbcedb499 : SUCCESS in 1h 00m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5431d3484200492eb6ec203d25d8fcc0 : SUCCESS in 55m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/61efda8a6f864ff9a90b971572dd485f : SUCCESS in 47m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5c0ded1ee314bd8a712326c4ae3885a : FAILURE in 18m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8ea2c918d2248fdbe880c9d4405a0c8 : SUCCESS in 5m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ff2e68f8424a4b459379bd5b424776c1 : SUCCESS in 19m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c56e10ab966740718b0fed0c24a66c3c : SUCCESS in 1h 02m 44s","accounts_in_message":[],"_revision_number":2},{"id":"834b21d3f1ce139f26996d5e9185df4b2e1ffacb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-08 15:48:32.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"f37a7fc321828735469f55e858f3cb5f08f5f952","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-08 16:21:40.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d9951b21b0466a5ba74f53207b6ac340a31c9096","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 16:47:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0717613a4ab447c4a43258e380484d63\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc28d38456394ce094c1dc8f3ef20f99 : SUCCESS in 19m 05s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/da0eb730269443168e9a443ea25560c0 : SUCCESS in 56m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3e7fcc994cde4cee8af10e34e50c6938 : SUCCESS in 45m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e5113e7d4034492e88bff357118d4822 : SUCCESS in 44m 29s","accounts_in_message":[],"_revision_number":3},{"id":"9489ee51462ccc2803fac7180abdd2017f17a6d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 17:15:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50bb0a0257344dd18136ca70ac67448d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39e2f7a5b6e84405bb7d4aa58e1c29c7 : SUCCESS in 9m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/7f6bc59cd0304b698c78f0f9ff6715d1 : SUCCESS in 1h 17m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f78b8973e71445b3a13046812a53351a : SUCCESS in 42m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fdd395f0b73c4a2d822a53b24f48b184 : SUCCESS in 1h 05m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dfec5f276293446ea2e636f2f9a28723 : SUCCESS in 26m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fe31ed5087b548eeb2ef21749bb7bf32 : SUCCESS in 21m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2c9b629770c648aab05f1d3bb58aa3a0 : SUCCESS in 28m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/697d0431c3bf41f5bad6551cc8e96c13 : SUCCESS in 9m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/609a7ddf85aa476b9c75ae11608ba07f : SUCCESS in 7m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2e789251dd96468f9bb40a0dec34cb26 : SUCCESS in 17m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a16d6b275b174365b236c2987c7b4b12 : SUCCESS in 15m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8ab774d2ebb74a109b33681cf51acc74 : SUCCESS in 17m 30s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b5183725ab02406d9be7fc2b6cc45b11 : SUCCESS in 19m 08s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/6d80a6209d424242b30dcf2568537c13 : SUCCESS in 17m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3c2fb7b8a52348d78da6396b211212f3 : SUCCESS in 17m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9592c56cdf9246e4856f617db62c10f1 : SUCCESS in 16m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3d7501212e12449aac05cf1534d46fb5 : SUCCESS in 13m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/95f0967ba9f34dd69b08aa6279c3bc21 : SUCCESS in 16m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f4bc9c45c24d4da8950eaf08dfa04b85 : SUCCESS in 17m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ca4a7cd4df214a0a8b3c753d5e21d313 : SUCCESS in 15m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0a9ea8965445489a956bef963aef3a2c : SUCCESS in 6m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0a608ccf456645ef8e3574bc98bac2c9 : SUCCESS in 16m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/145d59f23359418780eafbc1738836e6 : SUCCESS in 7m 34s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4db69ba2ecd5463ebbf43caebb4aa8bb : SUCCESS in 48m 56s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b63c0ea51d6e4fdc875e7e9b0664b934 : SUCCESS in 1h 00m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ee91074bf9514ef2b197e71c6c77d48e : SUCCESS in 53m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1025f544178244d1879693dfb5391182 : SUCCESS in 50m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22c0502534bd4eeb871d9bbddc2ec6ae : SUCCESS in 19m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8fd149439064e35919611d3c2582a7b : SUCCESS in 5m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5835ed443379409585fe47ac8e1bc801 : SUCCESS in 27m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/becadb006f074fe88c639acf64a527b0 : SUCCESS in 1h 09m 24s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9213ec8d3a3a79b64a019e2b08e6e138338af9cd","revisions":{"73c4e36dd04ab9758a65d5bc79fcc742cd56b22b":{"kind":"REWORK","_number":1,"created":"2023-09-06 22:19:20.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/91/893991/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/893991/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/893991/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/893991/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/893991/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/893991/1"}}},"commit":{"parents":[{"commit":"957743f25f957e6329c8b2abf1009f6f1d460f0a","subject":"Merge \"slo: 500 if we can\u0027t load the manifest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/957743f25f957e6329c8b2abf1009f6f1d460f0a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-09-06 18:44:16.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-09-06 21:10:02.000000000","tz":-420},"subject":"Add our own sys.unraisablehook","message":"Add our own sys.unraisablehook\n\nThis gives us an opportunity to buffer the whole traceback before\nlogging on py38+, rather than logging each piece as it\u0027s gradually\nconstructed.\n\nChange-Id: I71b00c888e7fd4d7eaf85a46cfee762268910a0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73c4e36dd04ab9758a65d5bc79fcc742cd56b22b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73c4e36dd04ab9758a65d5bc79fcc742cd56b22b"}]},"branch":"refs/heads/master"},"7702ae6e3bf2fb01b08b6ad3cccc721240be7e0f":{"kind":"REWORK","_number":2,"created":"2023-09-07 23:34:29.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/91/893991/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/893991/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/893991/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/893991/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/893991/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/893991/2"}}},"commit":{"parents":[{"commit":"957743f25f957e6329c8b2abf1009f6f1d460f0a","subject":"Merge \"slo: 500 if we can\u0027t load the manifest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/957743f25f957e6329c8b2abf1009f6f1d460f0a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-09-06 18:44:16.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-09-07 23:34:08.000000000","tz":-420},"subject":"Add our own sys.unraisablehook","message":"Add our own sys.unraisablehook\n\nThis gives us an opportunity to buffer the whole traceback before\nlogging on py38+, rather than logging each piece as it\u0027s gradually\nconstructed.\n\nChange-Id: I71b00c888e7fd4d7eaf85a46cfee762268910a0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7702ae6e3bf2fb01b08b6ad3cccc721240be7e0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7702ae6e3bf2fb01b08b6ad3cccc721240be7e0f"}]},"branch":"refs/heads/master"},"9213ec8d3a3a79b64a019e2b08e6e138338af9cd":{"kind":"REWORK","_number":3,"created":"2023-09-08 15:48:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/91/893991/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/893991/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/893991/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/893991/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/893991/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/893991/3"}}},"commit":{"parents":[{"commit":"957743f25f957e6329c8b2abf1009f6f1d460f0a","subject":"Merge \"slo: 500 if we can\u0027t load the manifest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/957743f25f957e6329c8b2abf1009f6f1d460f0a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-09-06 18:44:16.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-09-08 15:47:54.000000000","tz":-420},"subject":"Add our own sys.unraisablehook","message":"Add our own sys.unraisablehook\n\nThis gives us an opportunity to buffer the whole traceback before\nlogging on py38+, rather than logging each piece as it\u0027s gradually\nconstructed.\n\nChange-Id: I71b00c888e7fd4d7eaf85a46cfee762268910a0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9213ec8d3a3a79b64a019e2b08e6e138338af9cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9213ec8d3a3a79b64a019e2b08e6e138338af9cd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
