)]}'
{"id":"openstack%2Fswift~894215","triplet_id":"openstack%2Fswift~master~Ia94d4b63e5c97e75656cdf7be044b3ec2c7813e7","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ia94d4b63e5c97e75656cdf7be044b3ec2c7813e7","subject":"Add rolling upgrade job coming from bobcat","status":"MERGED","created":"2023-09-07 17:34:30.000000000","updated":"2023-09-08 00:08:55.000000000","submitted":"2023-09-08 00:06:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894215","meta_rev_id":"d55bb1667811277c74a89aa4160446c4d7e921ab","_number":894215,"virtual_id_number":894215,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-08 00:06:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-07 23:01:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-07 23:01:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-07 18:15:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-07 18:42:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"acc6cf160ed82553de08fb7a93a2a985054ed084","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-07 17:34:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f873fb0528673a1899584f70b8268260762cf26","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-07 17:40:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d58d3613e51f87ab079ccdb6bea313263784e9b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 18:15:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67b314f8f29044b9809d9607fe166cc9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/142638f0c3ca4f80b3b7f0724dee6fe5 : SUCCESS in 13m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2f1c3e5e26f6462a99e27f03e43b3500 : SUCCESS in 13m 37s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/98f51fa088a64ed8984e543f1802d0b0 : SUCCESS in 39m 14s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/25491ade85f742c1a240b33fd7549a4e : SUCCESS in 17m 15s","accounts_in_message":[],"_revision_number":1},{"id":"8c908876e3aa09494c86fef7993bbd08b5b937cd","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 18:40:45.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0c9cf3fdfe64101a6d17d1e0da4302a\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/acd1899dbf4f4604a1c9ff1379934eb7 : SUCCESS in 16m 20s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/57dd60ff5d9c46bdb3cbde4242da7220 : SUCCESS in 19m 17s\n- swift-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/fab18b6077574bbb85034cd7a0e7ea13 : SUCCESS in 21m 49s\n- swift-tox-func-py27-centos-7 https://zuul.opendev.org/t/openstack/build/5240f8bb21904534906236389a1d8787 : SUCCESS in 17m 40s\n- swift-tox-func-encryption-py27-centos-7 https://zuul.opendev.org/t/openstack/build/5e821c82a5a74ca78ed9f6098ea48405 : SUCCESS in 23m 13s\n- swift-tox-func-ec-py27-centos-7 https://zuul.opendev.org/t/openstack/build/c5d696263e7d4385a177bbcc17fd0ad5 : SUCCESS in 21m 42s\n- swift-tox-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/e85c4f1900f945cd8476a3813bf69009 : SUCCESS in 19m 55s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/80023af3e0504232b789293e5d4c078f : SUCCESS in 18m 51s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/697b6c6d3b184e4f8f283d422288d400 : SUCCESS in 16m 34s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/64631d8dfa4d4087af722ec27ddaf3fb : SUCCESS in 16m 20s\n- swift-multinode-rolling-upgrade-train https://zuul.opendev.org/t/openstack/build/9b0d129128874fe8bee91554072ad481 : SUCCESS in 16m 06s\n- swift-multinode-rolling-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/d48353e698db4da3b7e306dd226b8589 : FAILURE in 23m 52s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/48afe75eaa4c47438aa81fb35871aab2 : SUCCESS in 26m 36s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/eae9484082a941febfcb44c92326d87e : SUCCESS in 21m 07s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/d35b7b4c7c6c4be7b7cb2b451c78c638 : SUCCESS in 20m 57s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/edcca28ce56c48b5ae17c120b12a4699 : SUCCESS in 24m 34s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/505c2d7fe8324e10ad5233a87a29a15a : SUCCESS in 17m 31s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/c31f04e3cd114c8182d7f13d2fe73b18 : SUCCESS in 20m 39s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/464595fda3914999ba7a1c3f2458db62 : SUCCESS in 23m 17s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/4136cc15b88840ccb5ef072b5dc69b30 : SUCCESS in 26m 51s","accounts_in_message":[],"_revision_number":1},{"id":"64cd574d4445b1b4f5e8cb7b8b37e1ed6335119f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 18:42:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0d064ef3c734b4fa6b3627a03f6a40c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7946703bc2d47ef90fb1c1cba968ab1 : SUCCESS in 8m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/512dad61af0e4f35ada449b9f8cdd852 : SUCCESS in 47m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/198291da728642f1b100b75fe72a2911 : SUCCESS in 39m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d67bb102c1440bd9ea05b19c3186fce : SUCCESS in 45m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/969805580c7849d18921b589a3584819 : SUCCESS in 18m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bd4fbe297dc44808a47d0b0789ca8413 : SUCCESS in 17m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07aac0aded204c649b7790fe838b1778 : SUCCESS in 19m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9f8e0d85eddf44efa351370e48810b51 : SUCCESS in 9m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/30ab231c8d144e829b4af43675892a45 : SUCCESS in 5m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fa786ba252c948e3b24f20319f2d17cb : SUCCESS in 18m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/469900fd71ee45e79ca4a3a387fd88d3 : SUCCESS in 16m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0141b4194c0d42cfbb4137ee1914ff90 : SUCCESS in 16m 46s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b0a6e25c660444ab80a3d900b4167cf1 : SUCCESS in 12m 28s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/7d0f189017c9417582109ef20813beb6 : SUCCESS in 16m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9d3609a550c94b96b788adad4cd359ca : SUCCESS in 13m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cce48cef8d96475099510f7790e96af8 : SUCCESS in 17m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/427f927565874bd08cbf1ec6b6af8e88 : SUCCESS in 17m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5fbdc159ac1b42fa99a1831c63b303d2 : SUCCESS in 15m 50s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4816d272c7fe456382b5c99153bc5ae5 : SUCCESS in 17m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/743767ece9d54d0e9a91bddd2f97527b : SUCCESS in 15m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5085288ea59c48cb84c26735620f8cf8 : SUCCESS in 7m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/af05aa31399b46baa9974b151babd209 : SUCCESS in 16m 46s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f54273770fe1448a90c6498c1a641d32 : SUCCESS in 6m 34s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d9cad81c69d04873ac14b7e9c13530ee : SUCCESS in 55m 06s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/203c47307dca40a2a4775f9ac03c079e : SUCCESS in 1h 01m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/664d9c4f62824f7b9ed46b6b96f4cf9d : SUCCESS in 54m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c4e226c6b4e54bd3b70e60d59a5a7d7e : SUCCESS in 40m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e19bba5a521c42fa9e34ccca0ea9f8b6 : SUCCESS in 19m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0aa36fc10b59441b9dd2c71886444c85 : SUCCESS in 5m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4611acbd12a94ea9b3670c251e0db4a1 : SUCCESS in 20m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a75bc805608463bb43e4a580bdade37 : SUCCESS in 42m 27s","accounts_in_message":[],"_revision_number":1},{"id":"4e2ea59cd43742b74df16e61a059e05d39f957e8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-07 23:01:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c952b586cc2e5f4384dd4f889def35a8ae82b64e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 23:02:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3542bc31ac4cceb8c69483dff88917df6692593d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 00:06:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cd4eb70b26f42a0a29e093bc4636e9c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4f2e35685f34fdbac515af414faba94 : SUCCESS in 8m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/aec7b354f34943528ab76df721f74bf7 : SUCCESS in 42m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b8439f7bbb2840eea81b55cc9a6c8253 : SUCCESS in 36m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fde9f5ced4a44bca8d000b7820a95b14 : SUCCESS in 41m 51s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a4c5335d6b3c418fa8022ce7826a7dd9 : SUCCESS in 9m 37s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/89deea2d44a84e65adba681567a79f65 : SUCCESS in 6m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b3a44c88ab2f464990b670eee9ea838d : SUCCESS in 17m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/83120622d7514615971162d9f82120a7 : SUCCESS in 16m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a4efc7c68ec5460696f2ea646d84e4cb : SUCCESS in 15m 54s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e6cad3ed9e104fd5836432567cfb6a89 : SUCCESS in 15m 45s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/acd20c5c95c64c1388a489b49caf892d : SUCCESS in 16m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f83129caa59f4f4ebfa6c77575e7f440 : SUCCESS in 17m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4f352539ec13468ba4900c2387812c01 : SUCCESS in 17m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3849da7d987749aea2a32e7dd097d73b : SUCCESS in 17m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9f2b78e06bda4f9f92db78f6ab8abc20 : SUCCESS in 14m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fc6c9a7b53984048955fe857d779bab6 : SUCCESS in 12m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4bc5482e917b4657b087e38c17345b6d : SUCCESS in 17m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d8bb54ae064641bbb79e614d60b0841e : SUCCESS in 6m 50s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ba0598eb599f4d6991e028ce8848bffd : SUCCESS in 6m 52s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f2f9efa872fa48f58ba379209f838fa4 : SUCCESS in 57m 45s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/53a38cb2b4434f4da167ae7844ef52cd : SUCCESS in 58m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/74ed4e173c6d442082dc74381c0ddb26 : SUCCESS in 33m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/10d8b90971ad4b2eb01ed699fc210530 : SUCCESS in 40m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/abfb0522094c42b48055a3a26c9d7689 : SUCCESS in 12m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69ab2784f87b46c6ade63d5d00496ce8 : SUCCESS in 5m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/31eb26e586dd4ed7b0768a5d1d4ee227 : SUCCESS in 57m 42s","accounts_in_message":[],"_revision_number":1},{"id":"107301f32c1e2af4f1f0027cc5c1b9682c97c171","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 00:06:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d55bb1667811277c74a89aa4160446c4d7e921ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 00:08:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/380c564d9655403babe63257e056f1e2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f409ebfb13514ed099a90cfe9c03671a : SUCCESS in 1m 02s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e5db7dbca02e420684af821c4ffa01a6 : SUCCESS in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/916778c9ea5e4fb8a2c539e118959d98 : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"370c15c039575298d25142e214cae679183b67b3","revisions":{"370c15c039575298d25142e214cae679183b67b3":{"kind":"REWORK","_number":1,"created":"2023-09-07 17:34:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/15/894215/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/894215/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/894215/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/894215/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/894215/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/894215/1"}}},"commit":{"parents":[{"commit":"957743f25f957e6329c8b2abf1009f6f1d460f0a","subject":"Merge \"slo: 500 if we can\u0027t load the manifest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/957743f25f957e6329c8b2abf1009f6f1d460f0a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-09-07 17:29:36.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-09-07 17:33:44.000000000","tz":-420},"subject":"Add rolling upgrade job coming from bobcat","message":"Add rolling upgrade job coming from bobcat\n\nAlso clean up a couple jobs for rocky and stein; those were dropped\na while ago; it\u0027s not clear why it didn\u0027t cause some sort of error\nbefore.\n\nRelated-Change: Ibe46aecf0f4461be59eb206bfe9063cc1bfff706\nChange-Id: Ia94d4b63e5c97e75656cdf7be044b3ec2c7813e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/370c15c039575298d25142e214cae679183b67b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/370c15c039575298d25142e214cae679183b67b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
