)]}'
{"id":"openstack%2Fswift~894403","triplet_id":"openstack%2Fswift~master~Ic6f04a5a6af8a3e65b226cff2ed6c9fce8ce1fa2","project":"openstack/swift","branch":"master","topic":"refactor-slo","attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-09-25 22:39:48.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-09-19 18:15:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic6f04a5a6af8a3e65b226cff2ed6c9fce8ce1fa2","subject":"tests: Add test(s) for MPU part copy from range","status":"MERGED","created":"2023-09-08 23:32:09.000000000","updated":"2023-09-25 22:39:48.000000000","submitted":"2023-09-19 18:15:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"894403-refactor-slo","meta_rev_id":"da3137d1dc1fdb6c9fb3aa2a71ceefcb06ae85cc","_number":894403,"virtual_id_number":894403,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-19 18:15:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2023-09-18 22:46:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-18 23:31:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-18 23:31:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-09 00:12:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-12 01:45:50.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"c69857d3c39c7ba75eab52ab4122ac98ec527f9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-08 23:32:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c750f2c78ba46ee2d3e13c7e3368b32920246ca5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-09 00:12:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6962ef586e464eb9ae0b0fe35b2f4ef1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcd74df4e8e843a7bcc0433c31c66cf7 : SUCCESS in 7m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fdc683926d634bfe92dfb83dd4171d3c : SUCCESS in 17m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4f9e9f66c4c04f9998b2f158930adbf1 : SUCCESS in 19m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f9338397156041c7ac7f5bf9b36cc9af : SUCCESS in 16m 53s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0c077009fd314d8898fcafd28b4ab999 : SUCCESS in 13m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/097fa50cb1a745b4b92f91eb6f9b0315 : SUCCESS in 16m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/89bd7142c4564bfbae65673be1976461 : SUCCESS in 16m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e3c4250f108f457291ff42059df4106d : SUCCESS in 10m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1a2abe3a8d5e4a139a7b4a5759f75cc5 : SUCCESS in 14m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/39376efe656a435ab92663939b913e9c : SUCCESS in 12m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/060ac63467ea44f0af1aea0d3d16806c : SUCCESS in 38m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/26f90fc9207947f793992f57983d5ab8 : SUCCESS in 33m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fa46370a012448e97a1120827a4fc11 : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dccbc9752b3e476bb7a6840d2cca6652 : SUCCESS in 18m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"36463ac2a8e3de2646a115669b75014a55b0b788","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-09 00:32:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/246aeb2200b940caa513a8f496ded18e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e10eb32e62d4b3a98f2d7185f96d42b : SUCCESS in 15m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b898f4ef758d45b3be93d7eb1079b21c : SUCCESS in 14m 45s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4fb382dcd5354865a8915be174a0585b : SUCCESS in 47m 17s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1a5f895b79af48adb9d8abe790398c33 : SUCCESS in 57m 59s","accounts_in_message":[],"_revision_number":1},{"id":"b01f668cefebca1ba41a91a8b53244bf92133fec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-09-12 01:31:26.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"5bb53ba0e761f37f50dd8517ddbe0c18a9294524","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-09-12 01:45:50.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d91b753caba444f0e682aba99d17a13ddcf8fde0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 02:43:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7deefb1555214ba29b9989e0ea3bfc3c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc0bf42b7b66422482015115d4179d89 : SUCCESS in 9m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7888dbd9ffc04f909019ccedde41ee80 : SUCCESS in 13m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/def45ccf9bd648368ace71bcc7f5eb76 : SUCCESS in 20m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/96e1bc0f0d3142e49627b1d46b0861da : SUCCESS in 14m 24s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6d45f394bb1f4d13bcb4e0e8cacdb536 : SUCCESS in 16m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/adb2cb2ec0834f299605d2d60b47c0e3 : SUCCESS in 18m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c3be145d39684598927fdf687b6cb4ac : SUCCESS in 14m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/77648a4f0e2f438ca35eb2294f7d5db3 : SUCCESS in 15m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/42c2dc6ae5e544a88128f3fefa474621 : SUCCESS in 12m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/82e68bacccfd40368a289435b6bec3eb : SUCCESS in 12m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ee0cb5a048b94ec9b761c4b47e332766 : SUCCESS in 1h 06m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e070c3020f92432f83dc8a0bec077077 : SUCCESS in 38m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90d4058279ae4c15baadd9285d4f0389 : SUCCESS in 5m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ebe123a424c7417390c4b43402228750 : SUCCESS in 21m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4735860ee74f56a1e608aaebca60b6810a65d384","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 03:05:02.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0844a5f50e52407cbc252743846d443f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/02e7c03ea2654a9cb95b234fb0f15920 : SUCCESS in 1h 22m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/69fb901bf3a84ea7a743996af1420738 : TIMED_OUT in 1h 31m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d1fb5f6404c64efea6c0d30a2f8b7e9a : TIMED_OUT in 1h 31m 00s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/05fee7c29b1a410a846e5e0bd0eeeb54 : SUCCESS in 1h 08m 04s","accounts_in_message":[],"_revision_number":2},{"id":"3ddbc02376894951fdbad0233c87a677c1c91992","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-12 19:26:24.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dac3e7dd76113b058de64b1e76dad9ca8495c320","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-09-17 20:11:48.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"313c11c2c1d1f0b6e4d0e97d51ffd8697334bc49","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-09-17 20:12:29.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"95ae1fb562c85ab8fae87c6d21aa29a24f411f05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-17 20:58:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed7c05ee5b8348a481c8ea6204e08e04\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c502d31d6c554e4c87dc6438b1e674d2 : SUCCESS in 9m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8775e4877e8d4f3d9cd0636052dabe3f : FAILURE in 19m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/695cebab8a4b4e7ca85f1ed0e3e4bba7 : FAILURE in 15m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/620f9b9bcd254ad788a7bfe9ab375e52 : FAILURE in 20m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/500ceef8592141498c0f30bdc51fb64f : SUCCESS in 9m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/068ec16f6b5a4f6598ec72039ca923ec : SUCCESS in 6m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/24aea26d919b414fab268a4f204073da : SUCCESS in 15m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b15a5e3bb51647c3b6f13264623dd730 : SUCCESS in 15m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2dd271a80d9e44978fd503b49b781845 : SUCCESS in 15m 51s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a475eff772bc4faab7af3903526d090b : SUCCESS in 16m 06s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/fbfe8274d0814a4f9f6dca3a8cd8c891 : SUCCESS in 17m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ebd5fbd1f13c4c97b6c8433a8f2e732f : FAILURE in 18m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/32bcbf09fbf8489ca22833b5618e0c79 : FAILURE in 17m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2862c9cdc82c48ba85d59d97175696e0 : FAILURE in 19m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/92cde585f0be4974bd400cd488245911 : FAILURE in 18m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/bb00e48e51ed4694b016ba67b1862268 : FAILURE in 18m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b67afc544cee40ef9424f853bbcaf1b2 : FAILURE in 19m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2b209bd66d5a456d8d67469d7052b729 : FAILURE in 37m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/18de4ad0f6314ee9b8064e64fb9fe43f : FAILURE in 23m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/711f09250bf447658f5a7d54b50c8b47 : SUCCESS in 17m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/151e26f3791248fca827d96ee32fcd5c : SUCCESS in 4m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/940913995f424fe28f516fa38f51e0e5 : SUCCESS in 19m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8324a56d15b9fc63e5315d09c4d175afa47bb4c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-17 21:33:32.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79966383ff714a0d8a12279273575e12\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e36e0f6b49324bfc9536664e11435e5d : SUCCESS in 1h 18m 31s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fcfe3b56a0d543feb619cc1099eb60ef : SUCCESS in 1h 10m 38s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8443fd1b0a2045b9bfe9500b2eb4a332 : FAILURE in 1h 02m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/86fb78b152a24b32b8774dee3c1fcfcb : FAILURE in 1h 19m 42s","accounts_in_message":[],"_revision_number":3},{"id":"74c7218b7a4005e630c124eaa3a2fefdfaa76967","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-18 16:37:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"27914650b5926381a3cf61d837a30b2bdd9fbf9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-09-18 18:23:02.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"a95f16e821bf20fbb889fc62a1ec445c8ae84dba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 19:40:28.000000000","message":"Patch Set 4: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/00a2b4fe9a53471e858ba505d14935e3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47697516671e4e9d8ce8b3e5ef8e2dca : SUCCESS in 9m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c52f0162a4cc49a59f9cb828a134e3f8 : SUCCESS in 22m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed76126937bc4ef5a7d6319bff656830 : SUCCESS in 20m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/094a002a40f14a12934c3f85fe8ac4a1 : SUCCESS in 23m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a29337d9d346491e8b29c05d8d79aa68 : SUCCESS in 10m 21s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d98739879e4d4c53bbb83fc47dbfad4b : SUCCESS in 6m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/06043871e4464e5896c56f588f3f5b55 : SUCCESS in 20m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4c76373b43114ff4bfbc9501a3e79d6a : SUCCESS in 17m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bdda1c76378a43c18f0e521f6e51a823 : SUCCESS in 23m 24s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/271fdc3e76e94dc5a7ddba23f4148a22 : SUCCESS in 20m 05s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0005b994ab3c41948249312800ccab18 : SUCCESS in 15m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/96f57c143f8c434aaefae78e525c9931 : SUCCESS in 17m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/290af5004c574a41882041045c2dce49 : SUCCESS in 15m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/779c73e7e9d842d789b478da7c3a6e6d : SUCCESS in 15m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e37145e376a44d9db07b25aea8b3b822 : SUCCESS in 15m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/41fd94a4fb9045179058e2c349a6b1cb : SUCCESS in 15m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/48b33189a6194ed79dbb3fceb0c3c8b3 : SUCCESS in 18m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/065795800c0049b483a707d59ff1ae19 : SUCCESS in 40m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c364db0f61f849bf8654ff8c70740c66 : SUCCESS in 1h 08m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b4b5d600e2b43ae9f9f832c54335e8f : SUCCESS in 25m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a6cafa03a3148f7bc9e9c1ad587b9fc : FAILURE in 4m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/de885647d23e4625ab01d28f6735968f : SUCCESS in 20m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c31b3a46a2af2dfffaf555ec046308e70fc592b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 19:53:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89d5ba1fa2b14d98bbf66b2af2c91efb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e93efff9bfc5417e85d2c1fc12427f7a : SUCCESS in 1h 02m 28s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a00741ff3928410ca088be582ae0da29 : SUCCESS in 1h 14m 28s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8931481baa80477b811e094de91820a4 : SUCCESS in 1h 13m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/801dfea7b0a244a2adb88d64f345562c : SUCCESS in 1h 28m 31s","accounts_in_message":[],"_revision_number":4},{"id":"a8f256458c5b6d428f410adfa4221e0f5b300e53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-09-18 20:26:31.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"2b6437197c45869dcecce0473c9a8897109516d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 21:36:20.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06df096b4ab647e28a95959b461a9926\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b1ac2e64cba0499c827bc6f4789c1f10 : SUCCESS in 1h 06m 20s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1b747ae0d0c3441ea6a5526dc484e6ab : SUCCESS in 1h 08m 06s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9317c566da434053aa250e2c8048dd9b : SUCCESS in 52m 21s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f9a902aba14c4da9a602489a4dbad6b1 : SUCCESS in 1h 04m 08s","accounts_in_message":[],"_revision_number":5},{"id":"0a173e56637e9944eb481000d125aca836566d9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 21:43:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ccd175b221348cda26203038cd0ef5c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40ba5ca343c440f999cab4c84d1e0dba : SUCCESS in 9m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d01b20248c024be5b10a7811e7a0b34b : SUCCESS in 23m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/32d1ef292c604e119e69233ed5aab348 : SUCCESS in 19m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/feecc594fc9145bfbdb73b2b23cea63e : SUCCESS in 24m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/951fcf7d202f461287fe41a078c5e290 : SUCCESS in 8m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/db6a2521da2d4da88f9ab82797316193 : SUCCESS in 6m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e29f6eab63754d28993690d1f3498f73 : SUCCESS in 21m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2ecf3d7d3faf45c3a290f2ce4f3f3c85 : SUCCESS in 15m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1076148a90e949b085e6e3cc535241e0 : SUCCESS in 21m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c563f948c27041cab046ce81f0e47c15 : SUCCESS in 16m 10s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/383d1fdc99654172b8ad4427707e0b7c : SUCCESS in 19m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f0e5fe74698b4d4abc1aef55712db14c : SUCCESS in 21m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/18fb71a05c40420898c161c1621bc242 : SUCCESS in 17m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/109df511a1914fad874395a1ac21f784 : SUCCESS in 17m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7dc4bf4d7120472d935e19b8577d418e : SUCCESS in 15m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/369715ff11024af89f6f9417395e408e : SUCCESS in 22m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9ac53f154e1e444d8e783dac56bfcd98 : SUCCESS in 16m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9e57c30ec2a54b87a4722ecc5bf4db51 : SUCCESS in 1h 08m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ef65940dbf904ba69aa07b72a119ff2d : SUCCESS in 1h 08m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a611003751fd4d5281159d2e4dff50e2 : SUCCESS in 21m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6724fcb7f5a04456b8c2179e15910897 : SUCCESS in 5m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/177f6d533cfb425db5ca6b51a1d4950f : SUCCESS in 20m 16s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"43baee9bbf05c0722fd3ba264ec3767550c827fc","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-09-18 22:46:22.000000000","message":"Patch Set 5: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"eb5a765ca3aef1e1d22bd8230df8ebe352f51948","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-18 23:31:23.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"3658f4a293a65cf6870da5abbae6576c4fb8d28d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 23:31:46.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"15d6cac52e9b27fcd53576c0cd49bf0eef965ce6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 00:40:23.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d0c6590cde8d4b0d9f24c80fa8e3051b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04775cb2442e4decbcbb58ade4408eca : SUCCESS in 7m 04s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ad2e23d16d3e4410b79d66ab862e02b6 : SUCCESS in 8m 25s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/2b676a520d184aad9a7fd7e6367624ec : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e5151cf983314025bc67b8c40f79c673 : SUCCESS in 12m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ed1376f2110a421386afd5fc8c8476fd : SUCCESS in 13m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/331986f1c99a4da6b62c1b191763792f : SUCCESS in 11m 17s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/541c74cb805f4a91b7782a4ea5db76a1 : SUCCESS in 15m 49s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/66b8c1711c104e488ca1209f317e0f53 : POST_FAILURE in 17m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0c915aa95d964570a08fee814717f49d : SUCCESS in 18m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f1d403af1b5942699ee182f57b494eac : SUCCESS in 16m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/528106a2f4384863a9c04d8a5fb51cb1 : SUCCESS in 17m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6fc9c54096bd4c38be115a2dd0c1d8f9 : SUCCESS in 15m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1e6966368daf4d81b0a5734cef9698ce : SUCCESS in 15m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c8a0735a179d4872adab620c73c450a0 : SUCCESS in 17m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3bffbc70554343e3b4fcdf9380a23062 : SUCCESS in 7m 08s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fe2f900abbb64495ade6437abb5cc75f : SUCCESS in 6m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/697e7daad0c9422d82fd1abb9f830bb8 : SUCCESS in 54m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e9d0b2359f384559984100f0811bf277 : SUCCESS in 36m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55ad7f1f6b484421958c9cae8e5b99b6 : SUCCESS in 16m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/982af6c8d0724f92b3281cc066eb5162 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":5},{"id":"f3dcfa47ca1f11cc7794b1bb2d1beedc3019ea26","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-19 16:26:47.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"04205dc8819f8cab172625318e3dd100ad8e3fa3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 17:09:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d40c363744034befb4e267f89f085d0a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b56224807cbc47339fede051c3fb2336 : SUCCESS in 9m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/37b98f7ba21b43628217441aedf403e0 : SUCCESS in 20m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/13b7e22a2851412c98d38525b0217c2a : SUCCESS in 20m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c1febf127b23478e8ef34e4273f24d22 : SUCCESS in 21m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f6c82dc4e5704ff890cf9c8fc33e1156 : SUCCESS in 9m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a4533e002ba24c71a774258caa5bc5ec : SUCCESS in 6m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/57a7d83e0a5649628595efba2bb667da : SUCCESS in 18m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7ccb6599871c46aebd7cd4701168530d : SUCCESS in 20m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7ee48d2536264c0ba7ebe364248f2734 : SUCCESS in 20m 39s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3f637c3b42684a2c9eac26a283819bfc : SUCCESS in 16m 41s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9a4601f45d8b4016b87c74e6e6d0d077 : SUCCESS in 17m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/141d1fc571a441c99a8884fc133ae5e8 : SUCCESS in 18m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fd43f7d6e4764525addfea830c1524df : SUCCESS in 16m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/482b27bb141f45c8b2f7a984c76b04ce : SUCCESS in 15m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1aa0038a017e48698559996ecc68b80d : SUCCESS in 19m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/05145ed8071a4ba9b802cfd97d1c0164 : SUCCESS in 18m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/875392ee7a5c42ebaa5f4b20f0274aeb : SUCCESS in 20m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/08c43d79e7d848d2aa7ad17931f500f4 : SUCCESS in 40m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f484864eea16453198d1ed9082e66c5b : SUCCESS in 41m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dde011bff0af40b1966a8095d0b79d6f : SUCCESS in 18m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf190636ce194379ab0e67d76c7ec263 : SUCCESS in 5m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/53d8b7295a824e76a5a2d1a77b144556 : SUCCESS in 21m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"481ec3c38765acac0826d9ddd2e0d9eccb8b42bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 17:10:03.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a0fc63be3d01e5139dac834acad15f70a9e948ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 18:15:55.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f39f4d9512e47dbb838e93662c79d35\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e56514b6a93460bbaa3da8fe9f11a5f : SUCCESS in 9m 58s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/189690d615cf4ce3b9c52b65d8520495 : SUCCESS in 9m 13s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/fa1ef2875f294060b35a67fcf8272881 : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2691f40b411842f6aadec1db1c589d09 : SUCCESS in 20m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1938c32512b04194ba97e5506cd11f3c : SUCCESS in 15m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a3e8a4df09284b2880205c274b69f77c : SUCCESS in 16m 42s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/923e70eaf40c4878ad36ba9ca986eac1 : SUCCESS in 15m 42s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/15988eedfea94d229e605f6158d96150 : SUCCESS in 17m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/522949807e7e4cba98e7339d636cee56 : SUCCESS in 19m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7db607fab3ad459cb4e591ac97dd05b4 : SUCCESS in 14m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7b4b206d5cda487fa23d6ca40dfa72f0 : SUCCESS in 18m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4a4f90afeed84e74ab621a66497ae0c4 : SUCCESS in 16m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6ee1eb6484e24159a41f237d16d1d5e7 : SUCCESS in 13m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/57ed1a9f023944f29ef4acf260750296 : SUCCESS in 11m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b5778dc35aa34f2a8e1f459e675fadf1 : SUCCESS in 7m 21s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/066ef8830bbf475090fd038214bfd356 : SUCCESS in 6m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f6b82c41d61b4fb4893688a70aefee8c : SUCCESS in 1h 00m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a736ca614e1248d89cc613afe3894ab4 : SUCCESS in 46m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f1046c0d92b64d1db29458dc2e1b0de7 : SUCCESS in 19m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb4c68a651124b9fbb3cb4f143366c4a : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":5},{"id":"243ea61166da45ba98a272ec33840b9b9220db94","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 18:15:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"0d13e64e7a70675a98a2f3ee5ace7e0a29934f0d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 18:17:14.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e66571d824d043e2859f02defa4dedeb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a5ad2ca191f4de08845af988c545fa8 : SUCCESS in 1m 01s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/9430520f6c54422b945da9bcd377dee9 : SUCCESS in 39s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/74eb82f83ae34bf3a00e5333d1acdf38 : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9cac2202c4d7ac6a6ed10bf051111fae429e6535","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 18:28:12.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64f1f6acff4b4eecbf3e6e51501c4134\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6868285996944d568d33737ccb27016c : TIMED_OUT in 1h 30m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/56d21de972b64491a0469a20d7bf0ebb : TIMED_OUT in 1h 30m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/414b26ff6abc40748f55b92e6a82f12a : SUCCESS in 1h 29m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/72e958dcc92741d5a5c7e8964f37f366 : TIMED_OUT in 1h 30m 51s","accounts_in_message":[],"_revision_number":5},{"id":"da3137d1dc1fdb6c9fb3aa2a71ceefcb06ae85cc","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-09-25 22:39:48.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5392a2057bdf7e91878023cfdb0b7193d677a5b2","revisions":{"60f3a44f7b60d8e625fbd612e65d4c43fcc51e62":{"kind":"REWORK","_number":1,"created":"2023-09-08 23:32:09.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/03/894403/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/894403/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/894403/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/894403/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/894403/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/894403/1"}}},"commit":{"parents":[{"commit":"957743f25f957e6329c8b2abf1009f6f1d460f0a","subject":"Merge \"slo: 500 if we can\u0027t load the manifest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/957743f25f957e6329c8b2abf1009f6f1d460f0a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-09-08 23:30:40.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-09-08 23:30:40.000000000","tz":-420},"subject":"tests: Add func test to complete an MPU with a copied part","message":"tests: Add func test to complete an MPU with a copied part\n\nChange-Id: Ic6f04a5a6af8a3e65b226cff2ed6c9fce8ce1fa2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/60f3a44f7b60d8e625fbd612e65d4c43fcc51e62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/60f3a44f7b60d8e625fbd612e65d4c43fcc51e62"}]},"branch":"refs/heads/master"},"1a960e01e886b3567277d06bcabb7bd41c178a4e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-09-12 01:31:26.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/03/894403/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/894403/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/894403/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/894403/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/894403/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/894403/2"}}},"commit":{"parents":[{"commit":"e2dca715038ddcda072841b413db73569494c31c","subject":"Merge \"Update doc for include_service_catalog with app cred\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2dca715038ddcda072841b413db73569494c31c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-09-08 23:30:40.000000000","tz":-420},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-09-12 01:31:20.000000000","tz":-300},"subject":"tests: Add func test to complete an MPU with a copied part","message":"tests: Add func test to complete an MPU with a copied part\n\nChange-Id: Ic6f04a5a6af8a3e65b226cff2ed6c9fce8ce1fa2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a960e01e886b3567277d06bcabb7bd41c178a4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a960e01e886b3567277d06bcabb7bd41c178a4e"}]},"branch":"refs/heads/master"},"3e0c4c5f9c8ffe1929252b5f46a8ccd623eacdcd":{"kind":"REWORK","_number":3,"created":"2023-09-17 20:11:48.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/03/894403/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/894403/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/894403/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/894403/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/894403/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/894403/3"}}},"commit":{"parents":[{"commit":"e2dca715038ddcda072841b413db73569494c31c","subject":"Merge \"Update doc for include_service_catalog with app cred\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2dca715038ddcda072841b413db73569494c31c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-09-08 23:30:40.000000000","tz":-420},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-09-17 20:10:30.000000000","tz":-300},"subject":"tests: Add test(s) for MPU part copy from range","message":"tests: Add test(s) for MPU part copy from range\n\nWhen using the copy-part API it is expected for s3api to write down an\nempty value for X-Object-Sysmeta-S3Api-Etag on segments.  This was\nostensibly to prevent writing down an unrelated S3Api-Etag when copying\na part from another MPU the copy transfers object sysmeta.  We should\nassume a S3Api-Etag w/o X-Static-Large-Object is non-sense, and SLO\nshould forever expect empty values for it\u0027s sysmeta.\n\nRelated-Bug: #2035158\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: Ic6f04a5a6af8a3e65b226cff2ed6c9fce8ce1fa2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e0c4c5f9c8ffe1929252b5f46a8ccd623eacdcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e0c4c5f9c8ffe1929252b5f46a8ccd623eacdcd"}]},"branch":"refs/heads/master"},"2c09f585f46c4e7bc5507612ee4b6a221ca11601":{"kind":"REWORK","_number":4,"created":"2023-09-18 18:23:02.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/03/894403/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/894403/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/894403/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/894403/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/894403/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/894403/4"}}},"commit":{"parents":[{"commit":"e2dca715038ddcda072841b413db73569494c31c","subject":"Merge \"Update doc for include_service_catalog with app cred\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2dca715038ddcda072841b413db73569494c31c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-09-08 23:30:40.000000000","tz":-420},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-09-18 18:12:15.000000000","tz":-300},"subject":"tests: Add test(s) for MPU part copy from range","message":"tests: Add test(s) for MPU part copy from range\n\nWhen using the copy-part API it is expected for s3api to write down an\nempty value for X-Object-Sysmeta-S3Api-Etag on segments.  This was\nostensibly to prevent writing down an unrelated S3Api-Etag when copying\na part from another MPU the copy transfers object sysmeta.  We should\nassume a S3Api-Etag w/o X-Static-Large-Object is non-sense, and SLO\nshould forever expect empty values for it\u0027s sysmeta.\n\nDrive-By: consolidate handling of boto2 sigv4 skips\n\nRelated-Bug: #2035158\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: Ic6f04a5a6af8a3e65b226cff2ed6c9fce8ce1fa2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c09f585f46c4e7bc5507612ee4b6a221ca11601"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c09f585f46c4e7bc5507612ee4b6a221ca11601"}]},"branch":"refs/heads/master"},"5392a2057bdf7e91878023cfdb0b7193d677a5b2":{"kind":"REWORK","_number":5,"created":"2023-09-18 20:26:31.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/03/894403/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/894403/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/894403/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/894403/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/894403/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/894403/5"}}},"commit":{"parents":[{"commit":"e2dca715038ddcda072841b413db73569494c31c","subject":"Merge \"Update doc for include_service_catalog with app cred\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2dca715038ddcda072841b413db73569494c31c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-09-08 23:30:40.000000000","tz":-420},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-09-18 20:25:46.000000000","tz":-300},"subject":"tests: Add test(s) for MPU part copy from range","message":"tests: Add test(s) for MPU part copy from range\n\nWhen using the copy-part API it is expected for s3api to write down an\nempty value for X-Object-Sysmeta-S3Api-Etag on segments.  This was\nostensibly to prevent writing down an unrelated S3Api-Etag when copying\na part from another MPU the copy transfers object sysmeta.  We should\nassume a S3Api-Etag w/o X-Static-Large-Object is non-sense, and SLO\nshould forever expect empty values for it\u0027s sysmeta.\n\nDrive-By: consolidate handling of boto2 sigv4 skips\n\nRelated-Bug: #2035158\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: Ic6f04a5a6af8a3e65b226cff2ed6c9fce8ce1fa2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5392a2057bdf7e91878023cfdb0b7193d677a5b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5392a2057bdf7e91878023cfdb0b7193d677a5b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
