)]}'
{"id":"openstack%2Fswift~896280","triplet_id":"openstack%2Fswift~stable%2F2023.1~I696fc5bbc159b8d749e9240cc14a6468949076be","project":"openstack/swift","branch":"stable/2023.1","hashtags":[],"change_id":"I696fc5bbc159b8d749e9240cc14a6468949076be","subject":"Check on the state of stable branch gates","status":"ABANDONED","created":"2023-09-22 17:43:04.000000000","updated":"2023-09-22 21:14:33.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1172d7d1aa825b4e74d5dbdc92ab81cf0a0ec092","_number":896280,"virtual_id_number":896280,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-09-22 18:52:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-22 18:52:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e576d3ebe007828935752e5560e4aeb3f695a168","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-22 17:43:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d073a443947a421cd312379d223b19ac58b1665d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 18:52:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d5d7eed88ba4642a7695fd212b1298b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d042eada69d45698f0c011dc59660ec : SUCCESS in 8m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/18830c8c8adc4fc3b80a80d64a7a16a7 : SUCCESS in 1h 01m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bea4235019ef488592925305108ba937 : SUCCESS in 32m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/32c13bb0e7d34dc4a7d020e4612d9395 : SUCCESS in 55m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/359fd5a567004cd7bfc3586b68edfcfe : SUCCESS in 19m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/33378367e8f648528e212ae50293b9fe : SUCCESS in 19m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/002ddaab33f545a5a60ee40561c917f0 : SUCCESS in 19m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/385be92d3cb440deb5e7c4b1af936437 : SUCCESS in 8m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b65c80cb508b4fbcb3c6805bc30b1d26 : SUCCESS in 6m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/52de5f2d0fdc40ca98937f32fb9b3e7b : SUCCESS in 17m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e2ea335540ad4101be742280ff38d217 : SUCCESS in 25m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0453bd069f1c4b368a68f76e040a00cf : SUCCESS in 15m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/443f7b8173324b65912ae963d29ac50a : SUCCESS in 15m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0c2f9decc320407fb5e02b17a7b1f440 : SUCCESS in 17m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0e8e773e72c94985aac6318ff61dba8d : SUCCESS in 18m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/74ddf084513045c28b849c188b8d30b8 : SUCCESS in 17m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a03eff5bbcad494b9b4ec3c8340147a1 : SUCCESS in 14m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9fb3ffe5466547829d85537e1d0e72c9 : SUCCESS in 16m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7475ec0b5ab84b5fb673297e26fcb269 : SUCCESS in 18m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d4f4300167e74e50bf7d0bd00df94e7d : SUCCESS in 7m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dd88cb1a1d514662ab884e09c2e2279f : SUCCESS in 18m 36s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f57a0aa8435b4d2887a521229f64561b : SUCCESS in 8m 03s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/881172026b6e47698bd10b9bddb8ad17 : SUCCESS in 52m 49s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f5d953a0400547b9bbb2eb82a463b006 : SUCCESS in 52m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7d9e58535a4d4eb29becbb378b92362f : SUCCESS in 47m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9cc5a92284bc44c89641ccb6b3a41652 : SUCCESS in 43m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e12da913b8cf46f8a02df1f7ef85578e : SUCCESS in 15m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44e94562241845a88460c765b8123eda : SUCCESS in 4m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b143e2028d844688816dd5f7fa7c5c25 : SUCCESS in 22m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c62e9dd3e72647cfb95514e59cd7b96f : SUCCESS in 1h 04m 05s","accounts_in_message":[],"_revision_number":1},{"id":"82cb07ccd18d0e60b053152a2c4296b7816cf72f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-22 18:53:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"161ef79eba84d7beb44b975e27fa97e693963bd5","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 19:27:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d0e84b288b146c2a53a91f9415d8ef5\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/11ca4d0522ae4a439fb434dffc4ee9d9 : SUCCESS in 16m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/565c94ecc96a441ca131b09f09715310 : SUCCESS in 15m 32s\n- swift-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/813c914b08ce495f97df4b5040d76395 : SUCCESS in 19m 55s\n- swift-tox-func-py27-centos-7 https://zuul.opendev.org/t/openstack/build/401e4a8b9d724caea6d795bf1a8acd81 : SUCCESS in 21m 45s\n- swift-tox-func-encryption-py27-centos-7 https://zuul.opendev.org/t/openstack/build/ba80cd873bdc41159b300cff9a1b2365 : SUCCESS in 21m 32s\n- swift-tox-func-ec-py27-centos-7 https://zuul.opendev.org/t/openstack/build/bad05f1222f04634bc8ca337d2a15e0d : SUCCESS in 21m 30s\n- swift-tox-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/9d9f242a1f754f899926c71db0a5a1ae : SUCCESS in 18m 33s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/7f59fafd84824b20b3cf029e7fbc3d77 : SUCCESS in 17m 32s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/c956b9018fe9479b95aeab375ca604a8 : SUCCESS in 16m 56s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/a06aa96958ea4228a5000811e606f689 : SUCCESS in 16m 13s\n- swift-multinode-rolling-upgrade-train https://zuul.opendev.org/t/openstack/build/1f746fb3662249b3bac149c8670a0704 : SUCCESS in 20m 41s\n- swift-multinode-rolling-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/00f8d241077b4dc5a02f1f8cc77ba083 : SUCCESS in 23m 40s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/82def58d57a94600ba6c2d4e4ecff160 : SUCCESS in 23m 46s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/142efd20cdd74c2ba8346a20f1e8a9eb : SUCCESS in 20m 33s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/ebb513a1f5f74b14972699c70d1fe961 : SUCCESS in 20m 20s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/57dc7540f8724d17b127498f8f391041 : SUCCESS in 23m 54s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/bf4554ad87f2460ab8d7836e9e7abfeb : SUCCESS in 20m 45s","accounts_in_message":[],"_revision_number":1},{"id":"1a8d3fd9faa4e88edff9c5fac4ce9d831e88375e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 20:40:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dee95fa6dda64e988a435fa3b4e1b07d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cd79f760ce23429db5b99635601a76db : SUCCESS in 1h 04m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/94afe085f43844e2b216d43eefb0aa4c : SUCCESS in 1h 04m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ed326355733b41edb00680620b19bf2d : SUCCESS in 1h 29m 47s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ee83f45583554762959de1a61fe61eea : TIMED_OUT in 1h 30m 46s","accounts_in_message":[],"_revision_number":1},{"id":"1172d7d1aa825b4e74d5dbdc92ab81cf0a0ec092","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-22 21:14:33.000000000","message":"Abandoned\n\nThis stable gate\u0027s still good 👍","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d4c2f29a35ff934f03a2d8122888435d8b8f69ad","revisions":{"d4c2f29a35ff934f03a2d8122888435d8b8f69ad":{"kind":"REWORK","_number":1,"created":"2023-09-22 17:43:04.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/80/896280/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/896280/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/896280/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/896280/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/896280/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/896280/1"}}},"commit":{"parents":[{"commit":"8089925e776d0dd139d4d4a6dc8b5988b258ede3","subject":"Merge \"Update url\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8089925e776d0dd139d4d4a6dc8b5988b258ede3"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-09-22 17:42:50.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-09-22 17:42:50.000000000","tz":-420},"subject":"Check on the state of stable branch gates","message":"Check on the state of stable branch gates\n\nChange-Id: I696fc5bbc159b8d749e9240cc14a6468949076be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d4c2f29a35ff934f03a2d8122888435d8b8f69ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d4c2f29a35ff934f03a2d8122888435d8b8f69ad"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
