)]}'
{"id":"openstack%2Fswift~896858","triplet_id":"openstack%2Fswift~stable%2Fvictoria~I7d667823060c978c779978272cade8479cbd81f0","project":"openstack/swift","branch":"stable/victoria","hashtags":[],"change_id":"I7d667823060c978c779978272cade8479cbd81f0","subject":"Check on the state of stable branch gates","status":"ABANDONED","created":"2023-09-29 22:09:50.000000000","updated":"2023-09-29 23:19:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":896283,"cherry_pick_of_patch_set":1,"meta_rev_id":"ea77087e6a838ce97597a5180fba356159c2746f","_number":896858,"virtual_id_number":896858,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-09-29 23:17:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-29 23:17:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2a755b279cb51ed0db81681e01f3d158b1a011f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-29 22:09:50.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"5b046565c47fa356a8e67e5ca631a862bdad8b4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 23:17:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fccc01145554145b00b8cdb08b73190\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5898ddc3a96e45a5a01aeb8c9f255dfb : SUCCESS in 6m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/ae8b74cabaa04b70b1d0be4403448fa1 : SUCCESS in 42m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9a75b4ff213c445bbc1fc94c301b1a48 : SUCCESS in 29m 43s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b1feb2aa92954ce3b74b149a16d12fce : FAILURE in 3m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/71254a8610d149c8b5f0a34eb2ac7be2 : FAILURE in 3m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3b71e3ae5d3b494b830e9ddd78e07ef0 : SUCCESS in 10m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9a262db853dc4603b9a982be28d662a6 : SUCCESS in 13m 03s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/231cd02244e2447ba7b1c2a719438278 : SUCCESS in 10m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/54b4f4347fec465685d4e629ff79e240 : SUCCESS in 13m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e6a7a876648a41dfad71413c9aa3f76c : SUCCESS in 17m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3f41a42b8ef640ada0c98d1bdbcb390b : SUCCESS in 16m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f5c7f4d7f7664f1793449f7f663e1c46 : SUCCESS in 20m 32s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/780def16130a4f468ca395664a55d673 : SUCCESS in 17m 29s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3faf7796d5044bcba437c0eb514a751c : SUCCESS in 17m 06s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/7ad08d13edc6498193e27b09d5377d38 : SUCCESS in 18m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5b4291b94b8341dfad686d03fdd753df : FAILURE in 7m 15s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1e3c6fbc6ccd4d378757ed30176caf4d : SUCCESS in 44m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/86ec55f2a5674bb399d4ea76ef9b876b : SUCCESS in 50m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7c767bd670704977af631bc3066d9f3f : SUCCESS in 35m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c70242ff86aa450e953e4f31ebaeb3f7 : SUCCESS in 13m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00e3bd0332154399908944ac40d86d3e : SUCCESS in 5m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6112f830ffa4416cb06f38219bf90dc0 : RETRY_LIMIT in 6m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65619ae4b2844e29adadd1c81445b906 : SUCCESS in 1h 02m 16s","accounts_in_message":[],"_revision_number":1},{"id":"ea77087e6a838ce97597a5180fba356159c2746f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-29 23:19:33.000000000","message":"Abandoned\n\nvictoria looks good","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f2a277133c37ba3dd84634f0a91aff0daa7751e7","revisions":{"f2a277133c37ba3dd84634f0a91aff0daa7751e7":{"kind":"REWORK","_number":1,"created":"2023-09-29 22:09:50.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/58/896858/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/896858/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/896858/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/896858/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/896858/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/896858/1"}}},"commit":{"parents":[{"commit":"f80dce1f1921ec7812893bb22243b86aa7356f93","subject":"[stable-only] Cap virtualenv/setuptools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f80dce1f1921ec7812893bb22243b86aa7356f93"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-09-22 17:43:57.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2023-09-29 22:09:50.000000000","tz":0},"subject":"Check on the state of stable branch gates","message":"Check on the state of stable branch gates\n\nChange-Id: I7d667823060c978c779978272cade8479cbd81f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2a277133c37ba3dd84634f0a91aff0daa7751e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2a277133c37ba3dd84634f0a91aff0daa7751e7"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
