)]}'
{"id":"openstack%2Fswift~897343","triplet_id":"openstack%2Fswift~master~Ie35ff306aa33bfbeb39a4c79c96d57f9d1db7cb7","project":"openstack/swift","branch":"master","topic":"p-get-path-fixes","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-10-04 16:55:57.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ie35ff306aa33bfbeb39a4c79c96d57f9d1db7cb7","subject":"WIP: proxy: stop logging backend timeouts as client disconnects","status":"NEW","created":"2023-10-04 15:35:12.000000000","updated":"2023-12-05 16:54:56.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"4318b539c4e61a00e1a5343db9cd7a5f37d2b596","_number":897343,"virtual_id_number":897343,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-12-05 16:54:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-04 16:55:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-24 16:30:39.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"}],"messages":[{"id":"d40d57bc98a88bad2bb7b00952bc85c161dd6483","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-10-04 15:35:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18ad8923c39622ca97494dafb55d6e7238a6fff0","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-10-04 15:40:28.000000000","message":"Patch Set 1: Workflow-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e39cd6c62acc90ea65d0335bc28feac55e3cff63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 16:55:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c0342896897649e9a7075dca58d1cfbf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9194d0d229f141a48b2f883e7a17fbc1 : SUCCESS in 9m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/770af46150f346daa3fa705d7a95dd67 : SUCCESS in 1h 09m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/054eed87c8c74eea8b647f2a650cb0c8 : SUCCESS in 55m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d56b963bf2e34e058ea0cea8e310cb77 : SUCCESS in 52m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/265fee84776345b6a0abfde73bf22923 : SUCCESS in 23m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2aaf8231121b414a9aa89d8c9134d4cd : SUCCESS in 19m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b32b7a24d504b28b90c2bc8880e735e : SUCCESS in 17m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/179ce80a4751498abf2e555bd5f124e9 : SUCCESS in 11m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c0617efbdf7f40038ecaf2838d0e60b7 : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c1dee69d65d94b1b971da29e5b439333 : FAILURE in 21m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fbe2336953a54937b9da547bf2553249 : SUCCESS in 16m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5c0ecc5b8ce646c1a2573f887b955af6 : SUCCESS in 14m 18s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6076fda4c9b04131a4ac95fc60ccb4ba : SUCCESS in 23m 45s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/65f0530c94cd4a8ab8ac851841ad322b : SUCCESS in 15m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ff3e1dc23c924998ab875cc8dfdbf0bf : SUCCESS in 18m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/31bc545c0df54f919342f27d2c555569 : SUCCESS in 20m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/87ba98861f664b4ea5aa50125d0bc851 : SUCCESS in 17m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/811db49ae2a847b8b11f690a20134b94 : SUCCESS in 15m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/90c373b83cc640b1b7265f9417d5cc01 : SUCCESS in 14m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/29e7d631d69341b7845e46d1aaf43157 : SUCCESS in 15m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/95d650db794a4021868dd55153111f1d : SUCCESS in 6m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/effa37b0b11749219dff106bf5460a8a : SUCCESS in 15m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/944a6ad406cd48af8b72aec7fd1d535f : SUCCESS in 7m 21s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e16d58a732cb4d2ba760587028eab3be : SUCCESS in 57m 10s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7bbb45a0fa864f79a3705f2ae7e37583 : SUCCESS in 1h 08m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ec580510d13b4a509412c43029b6d63d : SUCCESS in 54m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/77e724c468e5402b85cc4ad18b178ead : SUCCESS in 51m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bfe2bff8393042de80efe1148e9c6973 : FAILURE in 20m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e169d2996cda43d99c6e9b07f8cb8299 : SUCCESS in 5m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a3d6aa83f5614f56a1ea6151a2f13c25 : SUCCESS in 24m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50c2fcba9c5e477b90f54763d5273baf : SUCCESS in 57m 27s","accounts_in_message":[],"_revision_number":1},{"id":"9106f7ddf45aef8ec49781a7e468f74127c56cb5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 22:00:58.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48bba9cf76fe425bb96be34c761dc86d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d2cd19a4cee54a2fb2d2c5cfbc75961c : TIMED_OUT in 1h 30m 52s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8759e271667b4761906fb60bb0ea44db : SUCCESS in 1h 28m 22s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/369b59f120ff4792941d0f3802e12258 : SUCCESS in 1h 28m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/559474bee7e149e38d928d47a8f91299 : SUCCESS in 1h 08m 36s","accounts_in_message":[],"_revision_number":1},{"id":"d5b40c934894ba55e93deb5c19ff0e5b29d27c90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-10-06 15:27:05.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"cd3d7f8783761759b89641e47bdc74ac939f1ffb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 16:59:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f42b8cdc1454da88e1160e05decdebc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89087988bca84847a5362f25d9f65f2d : SUCCESS in 7m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/010696fb89884d4db35ddf08e484b56d : SUCCESS in 55m 47s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ac62e07be46b4aa7a57db888e14590fa : SUCCESS in 1h 04m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f6c59d5d83084647ae94b3d1ce7221e1 : SUCCESS in 1h 00m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd22876d3f9a4f5fb381be4f3674d29f : SUCCESS in 1h 10m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/49268317c38246e6a81b648da5bc979b : SUCCESS in 20m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad3e9472144844f6ac8ccaebd07dbe40 : SUCCESS in 19m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ace62297d52a4599a9066ed8851449af : SUCCESS in 20m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/47d8260d829f4db1bb26eed843633dde : SUCCESS in 10m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5e6a3626efca4c7f8e781b754e5c6905 : SUCCESS in 6m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4f222540a7424c639861307317fb1e7e : SUCCESS in 18m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/29056706c8904065ac0fb70a2135325b : SUCCESS in 18m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dbe7e560b38b478c9e62c8ff7c25cf36 : SUCCESS in 15m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3cca7bae64af47a1894f3b4cb3402a07 : SUCCESS in 15m 53s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c49e52f9932347ba9de5452ab303701c : SUCCESS in 19m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f75c6c550ea24d43b8975d57407b7d6d : SUCCESS in 20m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d1027a60f0424f9bbc9e62ca62252c13 : SUCCESS in 14m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c6693722807a46a2813f30647dff7016 : SUCCESS in 17m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d5b34a901ce4439d94ca326b609740fe : SUCCESS in 14m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/050f9e6acc234f2c97baa44cf858acd7 : SUCCESS in 16m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/46d541fcb0844e079339b713c32bf123 : SUCCESS in 14m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/29222082892240ab8aef391e1bd78e30 : SUCCESS in 6m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1665ec0daf8b42c29afe7c542647dabc : SUCCESS in 16m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/10f8705fa8bb49d282c1a38f6e7bb75a : SUCCESS in 7m 34s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/953f8b53904847afb3129ce74ff1ace8 : SUCCESS in 1h 01m 27s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5137164aa96a4e35b8ad6d631a548cae : SUCCESS in 57m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1a7d609ae9bd4ebf850d941261e82e8a : SUCCESS in 50m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ba9e0ceef6d6460497e285e8a1178452 : SUCCESS in 42m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13ae004092e441439a7a610863e688e4 : SUCCESS in 16m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa6ea88bbd254a149c3bc5bc069669e2 : SUCCESS in 4m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/78469d9836b94ce2b828ade85999cdd3 : SUCCESS in 18m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2325731974a64481ae658e8eae856aa4 : SUCCESS in 1h 05m 52s","accounts_in_message":[],"_revision_number":2},{"id":"c69b668feba41c42403d02eae96cdb95a0dec6d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 17:31:57.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88ef5268691b41338a229f44f397318a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b1cdb92e8764646850a6d411add9c6c : SUCCESS in 1h 08m 36s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/266556f956cc4d6eae34fb431c56f13d : SUCCESS in 52m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f70c8a7d12bd470c84ed328deb1d810d : FAILURE in 1h 01m 16s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/94465322cc4f4fe9a065cd6c15d34415 : SUCCESS in 49m 52s","accounts_in_message":[],"_revision_number":2},{"id":"98a73ad485ed76d4de3d3f84f576551d5ce4518c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-10-18 22:25:34.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"521b84da50128436f47daf480fd61168bd4e7f34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 23:42:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8dd88d29b1ba4981ae1aeab9b1f2412a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e22d66787f564fc8a261ed6b80fc5ba1 : SUCCESS in 8m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/76b92687b6934a5fb6da452c93d1a031 : SUCCESS in 1h 01m 36s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/59a93cc014ce40f48c5e7d0c620a9929 : SUCCESS in 53m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/25c1364a696848809c7320d8ea2e58d0 : SUCCESS in 33m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b0e14352ff645c38e0de31040f8a249 : SUCCESS in 47m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/efabebf72915485294bcbc769fd96c50 : SUCCESS in 18m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0167d69be3304b9fb63006889337a1a4 : SUCCESS in 20m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fbdcf686e2e34589adcbb02becb27864 : SUCCESS in 19m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a2bc4ebfeb9149a1a59059c2b9d29bae : SUCCESS in 9m 06s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/18dba965e6554996a9c2cf3346ce9177 : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f8af1c4feea64e14bb8a357482a6c280 : SUCCESS in 18m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7ffa840660d3423d9b851ff9369b3b1f : SUCCESS in 18m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1cd221da4e294442ad939726d627bfb3 : SUCCESS in 15m 59s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a83d004e41504c67a3ab83e43657fcf5 : SUCCESS in 16m 09s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/467183627365456082a33d8493398086 : SUCCESS in 16m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5aec210dcb204eecac22d0c568924125 : SUCCESS in 18m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/283ca60107b1476b9d3b6de3bd5fb81d : SUCCESS in 17m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/03a51c043a3742ccbc8ccbbaf726c6dc : SUCCESS in 18m 17s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a36d3a4e3cd64d0786827633156b5e22 : SUCCESS in 15m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/129ef355c72e45128de9bd817bd92624 : SUCCESS in 16m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/88ba6972cc1b49248d364246a19d56a9 : SUCCESS in 19m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d343a798fa964656875375e9780b0260 : SUCCESS in 6m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0ee14c1f7db046aeaffc1ddeab9f26cf : SUCCESS in 19m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d307404c5dd949e2b2653f0d9f7b9641 : SUCCESS in 8m 52s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0b4483e9df404b13b43ecfdb387e1762 : SUCCESS in 57m 55s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/af058ae4be91462f8bcdfe47f182da9c : SUCCESS in 1h 01m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/770f6f79c4cf4f2294de7c18f03699e3 : SUCCESS in 50m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9e07b02884124c07a6fb83b0f715dd14 : SUCCESS in 52m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83d631acd5d4495f808dccabab424623 : SUCCESS in 18m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72c8cbcd77b14f8a9765a4933a14a22c : SUCCESS in 5m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0d917b881a9743489642da0112949201 : SUCCESS in 21m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff913f80190148c48b8bd53eb7f1dae7 : SUCCESS in 1h 08m 31s","accounts_in_message":[],"_revision_number":3},{"id":"d8c0295084516c987d2bd5a023bc0ef099545ddf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 00:17:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/823cde27920b46c19b7426ac883254db\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4c1bddaffc404d7594f0dfa67be52570 : SUCCESS in 59m 06s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/419d3a0d1d9347b79266cdd66bc2fedb : SUCCESS in 1h 08m 33s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b2dd4501d46944cfa90389b8605cb838 : SUCCESS in 1h 16m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/900920f79e4d4939b3dcdcf9cb25f239 : SUCCESS in 58m 40s","accounts_in_message":[],"_revision_number":3},{"id":"f89e796cce32b00c2404fb7c75cb3b06a64a409e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-10-24 17:34:56.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"faff13f325e113d4c4d9ff440245fce332cad810","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 18:49:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e99f7f69aa2f4e2ebd5402fa658d8c0d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e25396a9e7b4002a2f0e47daca38597 : SUCCESS in 9m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/3f1fe9048092424f9f7bce1b38726e45 : SUCCESS in 53m 08s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/532f9940f48c4646a5be836466a4c89d : SUCCESS in 1h 04m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5178bece8942487cab0e91f55ca43331 : SUCCESS in 49m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f62969fcf8014283bacc3ca0e7b2a98a : SUCCESS in 55m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ecaad98e079343748d63af9bc111da66 : SUCCESS in 15m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2db58a23fa0d43e6b5b5c6516cfdf358 : SUCCESS in 19m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/093f52805c9043fcbeb7d165634ebdc5 : SUCCESS in 20m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/eec9965c500048ef93d6fb163f272054 : SUCCESS in 8m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/491e26969ab64348bd912930bd389896 : SUCCESS in 5m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b7ea60a25bc7470598cb4151929dc1ad : SUCCESS in 16m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/42648f7feee6452f85f74d79fe81dcbc : SUCCESS in 17m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/04a6dab947864bd79c8396a0ace526a9 : SUCCESS in 15m 31s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ede10ce6b7bb419cb2401b4ce8b0fc54 : SUCCESS in 15m 56s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/01267140637e4038b5058dd0e7d96654 : SUCCESS in 16m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3482587ff7a54ecb82e6008879e98270 : SUCCESS in 18m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fb4ba2bc56db4cc7a543cb9aaed77b92 : SUCCESS in 19m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5a7fcec24b574338aede91beb61993fd : SUCCESS in 19m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3da87b624bd041258f32031da09455a2 : SUCCESS in 19m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ec308f62e8d04311b3faa394d5a1537c : SUCCESS in 18m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3585a15176bf4c54801965c13f9efd58 : SUCCESS in 16m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/57d3f668e7d846268aa615778843865f : SUCCESS in 8m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/df35a83bc31f475aa094e4ca91d0e8f6 : SUCCESS in 15m 46s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/59e357acdc3a405c9c83e945fc7df79e : SUCCESS in 7m 49s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e7395789fcef4ed2a14308d3650932df : SUCCESS in 56m 09s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a28c9224e9d24518b3c9c01efbdb963d : SUCCESS in 1h 03m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f5525c3b38a34f3a8cb2e6889d22d2c9 : SUCCESS in 43m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/adcf18f8f4994ec8967b4e27342b301b : SUCCESS in 56m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e2b3cef5e0c44d8bd338efa42037778 : SUCCESS in 17m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d46668089714ab894d58409aa93b8ab : SUCCESS in 5m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d5918b7fae0143bda73ffdc67486e915 : SUCCESS in 18m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e0a281984884e82b1e9ae1bc9d2cd5b : SUCCESS in 1h 08m 03s","accounts_in_message":[],"_revision_number":4},{"id":"87711c5c0bd48a9bd7ff80dda74fe5110c309bfd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 19:09:10.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/453380d900be4f69a44f48a08063449b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0fc296ef23624301882fd0c854423b8e : TIMED_OUT in 1h 31m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b8fa792aa458457c8a5d74fd5d39b17e : SUCCESS in 17m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/14ae351499aa491b873e4de6b905437a : TIMED_OUT in 1h 31m 10s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6ae2a45146cb4f118bc2d28c39cac282 : SUCCESS in 24m 43s","accounts_in_message":[],"_revision_number":4},{"id":"5c0e51051ede2b78a3405f55cc77eb3fedd8e05e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-10-30 11:03:10.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"cebc33b8edfb301f57ee63dcc6588d44d191b632","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 12:18:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/870471892cf941d4b83e6cf77e85830e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c216244a62bf46678b37585d5f01c1b7 : SUCCESS in 1h 12m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/19ed8b1c4f1647a299feec17fed8ad8c : SUCCESS in 28m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/bacac4d99f7340bd8ddd816a5c20fb56 : SUCCESS in 1h 08m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9db2aba712294d92b7b94fbf3273ecb1 : SUCCESS in 38m 04s","accounts_in_message":[],"_revision_number":5},{"id":"674f66675954009dde73473c630db1c1edb24edc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 13:57:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22a201b3a7554c28a0ab960c35c850c7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c55d49eef2b4baab717d89b841f06e2 : SUCCESS in 7m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/0f9ee3175d9346478ac26fac77c447fb : SUCCESS in 47m 04s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/68074a8b3a0f487f8ddfbac522885eec : SUCCESS in 46m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d122ba4774c447dd906a858244cd9e10 : SUCCESS in 37m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5b019c7d774471a9317061930bdd530 : SUCCESS in 46m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f966aedb059846c39a16f53693521600 : SUCCESS in 16m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/506af330849d49e59633f63f51786f97 : SUCCESS in 23m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/788ade59f29949d4af5eb5402cff1483 : SUCCESS in 30m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fca080907f6d49309f5fb795c9aa74f0 : SUCCESS in 10m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ab083bfa419f448689866d5a853220ab : SUCCESS in 5m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5d62134e789e4a3b9e418071515d3060 : SUCCESS in 15m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/64b80d6aada74163af18b877d791a65c : SUCCESS in 14m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ac477ad613484d04a25d0d506e792ec6 : SUCCESS in 16m 39s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/fbab76fb8a5b4b4db7dc38dc0bc4f0c2 : SUCCESS in 13m 37s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/fad10d9a309c4745ad26b3f29f58b387 : SUCCESS in 14m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/45396ad9a74d46e998a1efacebb6496d : SUCCESS in 16m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5c4bab2c8a72415f8486aa2f1f66d487 : SUCCESS in 24m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c76ca7086712428c9776ac4687a72d96 : SUCCESS in 16m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8469bd0e7d4f491997ef92a2c8119034 : SUCCESS in 23m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cb235353c70e4b2eb77744dda8d7f226 : SUCCESS in 24m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1103fb095b154ad2860f1fc46990679f : SUCCESS in 26m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f9b2e295bff04134a7a672482f56cc3a : SUCCESS in 6m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d5d533cdc7384e04840e36bbc253b48c : SUCCESS in 18m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ad0bd73a83d346c99ca66e0c0c09bffb : SUCCESS in 9m 11s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f41649db91fa489da7989913f0be1dd0 : SUCCESS in 50m 33s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f4699674953c4b64b10503e512e24bc3 : SUCCESS in 59m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3579432508f8407ea6e62d6f887bb52b : SUCCESS in 51m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/58435911a9874911b734419685d4be98 : SUCCESS in 47m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/446ac7f9d2d44982b8bb9a79e904b468 : SUCCESS in 18m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df1f521a119448d1a810f8f32bb9f4e4 : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fb639f3db8c44048b67aa042a0536849 : SUCCESS in 25m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0b6fb5b2362451ba1d8f9beb2190d2a : SUCCESS in 55m 28s","accounts_in_message":[],"_revision_number":5},{"id":"ab001b709ac04162a302a0e3ecf03b9ce9a9a4b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-02 18:24:37.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"72cd138f621219b55c1f7774e705326cc4dcaf1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 19:43:50.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9de25bca83ab40edba5be016e0c4e24b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/958b135885c6452d91d4070a6a3fd080 : SUCCESS in 8m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/114e063c7dad4366b1178212d019a74f : SUCCESS in 1h 09m 20s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5d5e74956f5f41ec92e964ee1e8913e7 : SUCCESS in 57m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/906e44453f6545b4949c0d257d00f5e7 : SUCCESS in 45m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b071ea33283496d8cc1fb05e55d8cf0 : SUCCESS in 48m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0149ac7051104cb6a0d5e76ee5ae6e5a : SUCCESS in 18m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ef9320aeeecd48c0935481e8b0e63546 : SUCCESS in 17m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/23e6eb1e38b04cb480c06f4368460319 : SUCCESS in 19m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f1075e7dee1a4779afdfcd9bbd1335d1 : SUCCESS in 9m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/762ae6acfc31462583610edf7cb04ccb : SUCCESS in 6m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7f1595d15cd542e2955f03a405edf907 : FAILURE in 17m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d0c384fab12d437583e2b87b40bb8545 : SUCCESS in 17m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5c6adcdbc87e4d459301977e73fb3e5d : SUCCESS in 16m 32s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/776823cfe8144dc9964a16abd39acc85 : SUCCESS in 15m 52s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1598bc38b29544bca681cbfb2b8c80ca : SUCCESS in 21m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a2656ce104b047438356ca04ebbd39ab : SUCCESS in 19m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/22717d56be6243d59ccfb9a8e67a0673 : SUCCESS in 19m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d6c53b3fa3b242bab49b00c4b20e1c3d : SUCCESS in 17m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b47a454596eb4f2f8991bc69ee11d3d1 : SUCCESS in 15m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ae160d42201f4bf8bc7261033ce97187 : SUCCESS in 15m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/036aeccab3a64e95b43c078f84f51c79 : SUCCESS in 17m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2815e0044aa64f7a832d5b74d7c3e161 : SUCCESS in 8m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4df61ae31e8f468b8016f93e2a9f15ab : SUCCESS in 17m 11s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/13b3a42db2db4bbc869742cdc9461766 : SUCCESS in 7m 10s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/500209fce60b49bc9b4dae471f267ef1 : SUCCESS in 54m 21s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/dec48eb79a11468dbad489fe7a778ac7 : SUCCESS in 1h 00m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0ba5e1b9a8f54f8aa789aa7e8e8f27cd : SUCCESS in 56m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/84afbf3654aa40a48c678b53dd271281 : SUCCESS in 51m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/054d68ecae384ef2bf348406c64255b4 : FAILURE in 18m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efea750ea33443748ac26360f8b5560d : SUCCESS in 4m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/be02fac2ce754a9489fcc7f24b99fcac : SUCCESS in 21m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/40a62534597b483fa5cdb35b7fed717b : SUCCESS in 54m 27s","accounts_in_message":[],"_revision_number":6},{"id":"4b4d0a37aa8b3626e96a48c155c2d4a3a8d36abe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 20:21:12.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09cf93ef815b484991c0e2360288f828\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d694c24ae90a4d24b5d59d540342bdb4 : FAILURE in 1h 21m 38s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5a078e33846641799e80c7568c41c181 : FAILURE in 53m 16s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/da1aafce3e2e4604b26431e2ef737cf8 : SUCCESS in 1h 11m 05s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1521375f56db4da7970bafc66aa727f0 : SUCCESS in 23m 55s","accounts_in_message":[],"_revision_number":6},{"id":"13a431a245468d36970ec17a198106aa77cb74fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-03 14:30:47.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"f611a2635737787a7fd4a39ccad25d6dd2279ac7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 15:43:31.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7a0db052e7c461d885a103cee7fd126\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31713dd276e043fb9c0b50b9129c8011 : SUCCESS in 9m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/b6384b9b151b4b2cbb153efdebf1746d : SUCCESS in 1h 02m 41s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b5ac004ce7ef4491a6feb9b30c4cb16d : SUCCESS in 1h 05m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3ae6a4a6962d4fe6a8e1b6dd8604edba : SUCCESS in 52m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ae8f2e6251c4c68bc1e62aed725785d : SUCCESS in 1h 04m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c543f14c883e429187a3d35683fa72aa : SUCCESS in 17m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5b35ed0676294322a62d8e280cd36ff9 : SUCCESS in 18m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c8919d73b54e4457bc002ac3ffdee410 : SUCCESS in 19m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/366ecb07eb214bd98bf04ab5e7f8562a : SUCCESS in 9m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0e17458b6bc1469eafab51b079692901 : SUCCESS in 5m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c69a6cc61aac42a689f0d85b4d7e5b2d : FAILURE in 17m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/74357ca1345240f0abd1e5be62d763ab : SUCCESS in 16m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/39616b9f45df48658ca12a1e9f8f1eae : SUCCESS in 15m 02s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e0e90a05f5f5409da4912b74c094a1f9 : SUCCESS in 16m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0d2ab2b86d414c9a8e2085836f1c719a : FAILURE in 15m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cf2eae6e3ff24ab9b2a018ea1eb2fc71 : SUCCESS in 14m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/25dea0c28d654e2fb1c08e3fd173b498 : SUCCESS in 19m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c053fe479069488cbc3504ae40390193 : SUCCESS in 19m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1321e76e04934fc59926a33d4aeeb06a : SUCCESS in 13m 50s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/dec1ad952dd044678ae6c477bd7a3e42 : SUCCESS in 15m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bb7aa126c6f54ec0ac65e04fe21dd23f : SUCCESS in 19m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/941f901c096249118b8ddcddc922da13 : SUCCESS in 7m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4496bed4afc84946990005ba8af11f5f : SUCCESS in 15m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0835d9b02cf04539af303759c71bad9f : SUCCESS in 10m 15s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5034bea87e6e434fb7adc034f7345d28 : SUCCESS in 55m 17s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/86ccc1e53bcf490288722f5ed803b885 : SUCCESS in 1h 01m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/07b6baaecae14779a9a4a7f6cac1dbce : SUCCESS in 1h 01m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2aa8a0fc12994752a324c3c2555dc8ca : SUCCESS in 51m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03b101fafa29417a9afab944cfdf6a62 : FAILURE in 18m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e38ee59e93944f6587f7c804b6267b7e : SUCCESS in 4m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a00c29a0da294e3e98627e28adc3614d : SUCCESS in 19m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1bd99e16673c4be3b1f4c1f768d83b0c : SUCCESS in 1h 04m 47s","accounts_in_message":[],"_revision_number":7},{"id":"f1c7031a760fec893a8b2ef29713dca0b0fb4161","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 16:04:45.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/891fb51e2fa8442c9944ce8d1990ef12\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c3ae9e8de5e0436aa25e4182351c21bd : SUCCESS in 19m 18s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7b2490c617fa44f9952f9ec6c7e876c9 : FAILURE in 1h 07m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec742fc9edf24ee7b0d252e899bf48e9 : SUCCESS in 21m 19s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7387570abd7f41f28f46705a54c55643 : TIMED_OUT in 1h 31m 01s","accounts_in_message":[],"_revision_number":7},{"id":"8537d22ee96306cd465f8cfc8567462e22750d94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-04 16:15:22.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"fe9d4fb16ecd898ec400af03fa095f51f24bef07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 18:01:46.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/72a1d0529cf64b15a803236335ee986f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03133c3803f143ea83c3e7698f1aa112 : SUCCESS in 9m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/91f46994dcee4b0c95088b7d90ee1323 : SUCCESS in 1h 05m 38s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/10921cd2578e4a14a23135c7413f91bd : SUCCESS in 53m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/90bf8ece2e2544f7a9be2d08f7d7ad04 : SUCCESS in 53m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1fee29ad3a014b3ab6dc5795dbbbe5e4 : SUCCESS in 1h 07m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/249a5d7f5ab849d19d36c2a1a720979e : SUCCESS in 18m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5098a81b2934447ea701a5cddfec0f7d : SUCCESS in 19m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fc5ccf9a03d841a996fae6d66604635b : SUCCESS in 17m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c4185cdc80784a1c8837ce0ced4cb8a8 : SUCCESS in 8m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7817444d99794b26970cc965476e175f : SUCCESS in 5m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/eb47b74f772a44d7a4716010b6c31d3a : FAILURE in 19m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3a113d5996224561ad2217992ca05c21 : SUCCESS in 16m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4aeac757779f4e9aa8ac795a4e4f6045 : SUCCESS in 15m 18s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/db52b2e56dc9403aab523a64f8e2d4e9 : SUCCESS in 14m 53s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8dc2082144b848059a41a94026be990f : FAILURE in 15m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/35f1caaf86e7492095b320f67694e894 : SUCCESS in 17m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/41a1f926925446b4b03f8b1525a38704 : SUCCESS in 17m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dc156116b031476d892bfd8d32cc4ac1 : SUCCESS in 17m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6e9e328cde5f4ad193ab31a73d5d4673 : SUCCESS in 15m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/035e5c05d9604c3db4936919a67a1c8d : SUCCESS in 16m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d6a025dd88bd4bae89b6fcab2b0751d4 : SUCCESS in 19m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/558e208cebd24dbfa41a3a0689accb02 : SUCCESS in 7m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/80699cf2a8b244ea817dfb11a86274b7 : SUCCESS in 17m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/43717e5f4e30434498aa095aedcc8f7a : SUCCESS in 8m 05s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8ae049e7f5ba4a77ba921fecad71eb2c : SUCCESS in 1h 00m 04s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/93695d25ac414ffb9bbc1611b8637809 : SUCCESS in 56m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a81ed9d720904ad980011493ac7a4fd0 : SUCCESS in 43m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2f38f4a6ee5d49128013e9c54f945ed2 : SUCCESS in 58m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a91b558393ce4a969a2f362290dd985f : FAILURE in 16m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ae40c1053c94695af283e90260dbea4 : SUCCESS in 4m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f19d738094274446843faf5a3f9335b5 : SUCCESS in 22m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cdd1ed26daa541e7ad5bddb3ce6b9f41 : SUCCESS in 1h 07m 22s","accounts_in_message":[],"_revision_number":8},{"id":"93a39fe99f1c2d66a9db41bbefb846ba81951c6a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 18:12:12.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3beb6315a4747c4ab83da17376a9bbe\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1ac8c8bb8c4b431dbf4a9f38882cb36c : SUCCESS in 18m 12s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/20df89e150ed4a01a61938c7ad72191f : SUCCESS in 20m 34s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/566661bc21f247e0b2471e8ea5f4c4c0 : SUCCESS in 1h 00m 22s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6b907bfc5cab43d8a00d895083b014e8 : TIMED_OUT in 1h 31m 06s","accounts_in_message":[],"_revision_number":8},{"id":"e6e7d0e50643cc417a1e3fed8806277755deb26e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-05 15:02:24.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"babfdb8f9ff60731e9a0754368bfb423f2e5a344","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 15:32:00.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a24661dae8b142408482e6e2a870e738\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/48e775c0837f4b11a438dd13d2739a88 : SUCCESS in 18m 35s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d2eb21d6ea4b483eb10b85aa247d90e5 : SUCCESS in 18m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/72f06f224bf040deb55603232ee6e063 : SUCCESS in 26m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/90436c148d3148079556e08a8f3f7c18 : SUCCESS in 22m 52s","accounts_in_message":[],"_revision_number":9},{"id":"4318b539c4e61a00e1a5343db9cd7a5f37d2b596","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 16:54:56.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/063b5e68fe4447619aca0d7ef26c4906\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03eb819342bc4f52858957d082ffe3c4 : SUCCESS in 8m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/21241797a0b4422db229aeea3caadd54 : SUCCESS in 1h 42m 50s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/562ac61aa2cc411393cb869968f1a0fe : SUCCESS in 1h 06m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a91b9cfc2fb142a89f7dd34ed825a0e7 : SUCCESS in 56m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2cd31380cd26477391fd3657e6bfa15c : SUCCESS in 1h 10m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a09e146538ed4d718f81d3bb77ffc368 : SUCCESS in 20m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2e85bac2f5114c8e8b85a9898e37c802 : SUCCESS in 19m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b1791d0f5a1b470d829e293fbd2f3f23 : SUCCESS in 24m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8f282bf7b0f54bebb0a8de1d2dd35356 : SUCCESS in 9m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3f987a96c44d47129484d7ee845512fd : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/08b6b77413f54adc9bf146edf77972fd : FAILURE in 18m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5808d5e943264a28ba0c4bbc34eb2e4a : SUCCESS in 17m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/084e38e6467f43a3b111c7c3cc8c00d8 : FAILURE in 14m 12s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1cab3519a37a43dd8ef08846cd8b6f52 : SUCCESS in 17m 26s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2fffc30e84be4fe2a988798e1a2ad2f2 : FAILURE in 18m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/581fd177c0564d7889a9c58b1d17b0a1 : SUCCESS in 16m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b3b8832e79864ffd9a8211bcd459ebca : SUCCESS in 18m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8dae16fc6eb64559bb238f471b353186 : SUCCESS in 20m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fde65ccb6312450bb49d1d62a7495051 : SUCCESS in 15m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7fde8f2cc1b744f0ba9547c77bc646d3 : SUCCESS in 15m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/616dc465946144309e60cbe1910750ed : SUCCESS in 16m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fe9702e033f7409b898991533c59edaf : SUCCESS in 7m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1e7828db2a234734a99264670e18aff4 : SUCCESS in 15m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/35390d30b1e3485aa107aaaa8f79240b : SUCCESS in 8m 50s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f96fd833301149b1890382fefa739016 : SUCCESS in 1h 01m 14s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/da0f1f43774649a889db68308bbf1f9c : SUCCESS in 56m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/613562b51a90408582e1e8a7a9ddfe8c : SUCCESS in 55m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4c628026a05a4239be2a0aa270cac071 : SUCCESS in 55m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3804c59b1b24481da984ffe9c2762a0f : FAILURE in 16m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b00ac7a3e49b4dc3b554c6fbe074d3eb : SUCCESS in 5m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cb88a4b894d44659932c5a39db906660 : SUCCESS in 20m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4bd82e599d6f446c91c804a55c7edce5 : SUCCESS in 51m 26s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"a584e064507b495a5215e403f0faefe732d232b4","revisions":{"6c1c32a5740d13f8952841e3f0aa75e70097cd1e":{"kind":"REWORK","_number":1,"created":"2023-10-04 15:35:12.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/897343/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/897343/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/897343/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/897343/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/897343/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/897343/1"}}},"commit":{"parents":[{"commit":"0e9fdf42d2989df6f776f84cf29663f93385a690","subject":"proxy-server: de-duplicate _get_next_response_part method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e9fdf42d2989df6f776f84cf29663f93385a690"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-10-04 15:23:22.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-10-04 15:23:22.000000000","tz":60},"subject":"DNM: proxy GET - close the parts iter","message":"DNM: proxy GET - close the parts iter\n\nWe sometimes wrongly blame the client for disconnecting when a backend\nGET times out. That\u0027s because swift.non_client_disconnect is not being\nset when it should. That\u0027s a bug, but this patch isn\u0027t fixing that -\nbecause the egregious \"Client disconnected\" log message is helpful to\nreveal that an iterator is not being closed in a timely way.\n\nIn document_iters_to_http_response_body, for a single range request,\nwe take the first item of the ranges iter (there shuld only be one\nitem), and then iterate over that range\u0027s bytes iterator. We take care\nto make sure that the bytes iter is closed using a closing_if_possible\ncontext manager. But we don\u0027t close the ranges iter, which means that\nthe GetOrHeadHandler._iter_parts_from_response() does not exit and the\negregious \"Client disconnected...\" log message in its GeneratorExit\nhandler is not logged...\n\n...until the garbage collector cleans up the handler, then the\nGeneratorExit handler is reached and \"Client disconnected...\" is\nlogged.\n\nChange-Id: Ie35ff306aa33bfbeb39a4c79c96d57f9d1db7cb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c1c32a5740d13f8952841e3f0aa75e70097cd1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c1c32a5740d13f8952841e3f0aa75e70097cd1e"}]},"branch":"refs/heads/master"},"c7e51a439d4a642ea75da8929d7b94e321936e23":{"kind":"REWORK","_number":2,"created":"2023-10-06 15:27:05.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/897343/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/897343/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/897343/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/897343/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/897343/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/897343/2"}}},"commit":{"parents":[{"commit":"38b9534de2738ec94b686f974db2e4d24efc2463","subject":"proxy: handle ChunkReadTimeout in replicated GET path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38b9534de2738ec94b686f974db2e4d24efc2463"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-10-06 13:37:28.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-10-06 14:59:52.000000000","tz":60},"subject":"proxy: stop logging backend timeouts as client disconnects","message":"proxy: stop logging backend timeouts as client disconnects\n\nWhen a ChunkReadTimeout is raised while reading bytes from the\nbackend, the GetOrHeadHandler._iter_parts_from_response will catch a\nGeneratorExit. The GeneratorExit handler checks the request environ\nfor a swift.non_client_disconnect flag, and if this is not found it\nlogs a \"Client disconnected...\" message.\n\nIn the replicated single-part GET path, swift.non_client_disconnect\nwas not being set when ChunkReadTimeout was raised during backend\nreads. This led to an erroneous \"Client disconnected...\" message.\n\nWorse, the GetOrHeadHandler._iter_parts_from_response was not being\n\"closed\" in a *timely* way, so the GeneratorExit and erroneous log\nmessage might not occur until garbage collection reaped the iterator\nsome time later. This could be seen by observing the unit test log\noutput for test_base.TestGetOrHeadHandler.test_node_timeout: some info\nlines with \"Client disconnected...\" messages are emitted after the\ntest completes.\n\nThis patch therefore:\n\n  1. Ensures that swift.non_client_disconnect is set when\n     ChunkReadTimeouts are raised.\n\n  2. Ensures that the _iter_parts_from_response is \"closed\" when the\n     ChunkReadTimeout is raised.\n\nChange-Id: Ie35ff306aa33bfbeb39a4c79c96d57f9d1db7cb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7e51a439d4a642ea75da8929d7b94e321936e23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7e51a439d4a642ea75da8929d7b94e321936e23"}]},"branch":"refs/heads/master"},"42e0225284ac3c25cb4630ab9d897eeb3f1ac6db":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-10-18 22:25:34.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/43/897343/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/897343/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/897343/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/897343/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/897343/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/897343/3"}}},"commit":{"parents":[{"commit":"ef50cf4017dc174f1119f94be666d7887ff0dbaf","subject":"proxy: handle ChunkReadTimeout in replicated GET path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef50cf4017dc174f1119f94be666d7887ff0dbaf"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-10-06 13:37:28.000000000","tz":60},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-10-18 22:25:34.000000000","tz":0},"subject":"proxy: stop logging backend timeouts as client disconnects","message":"proxy: stop logging backend timeouts as client disconnects\n\nWhen a ChunkReadTimeout is raised while reading bytes from the\nbackend, the GetOrHeadHandler._iter_parts_from_response will catch a\nGeneratorExit. The GeneratorExit handler checks the request environ\nfor a swift.non_client_disconnect flag, and if this is not found it\nlogs a \"Client disconnected...\" message.\n\nIn the replicated single-part GET path, swift.non_client_disconnect\nwas not being set when ChunkReadTimeout was raised during backend\nreads. This led to an erroneous \"Client disconnected...\" message.\n\nWorse, the GetOrHeadHandler._iter_parts_from_response was not being\n\"closed\" in a *timely* way, so the GeneratorExit and erroneous log\nmessage might not occur until garbage collection reaped the iterator\nsome time later. This could be seen by observing the unit test log\noutput for test_base.TestGetOrHeadHandler.test_node_timeout: some info\nlines with \"Client disconnected...\" messages are emitted after the\ntest completes.\n\nThis patch therefore:\n\n  1. Ensures that swift.non_client_disconnect is set when\n     ChunkReadTimeouts are raised.\n\n  2. Ensures that the _iter_parts_from_response is \"closed\" when the\n     ChunkReadTimeout is raised.\n\nChange-Id: Ie35ff306aa33bfbeb39a4c79c96d57f9d1db7cb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/42e0225284ac3c25cb4630ab9d897eeb3f1ac6db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/42e0225284ac3c25cb4630ab9d897eeb3f1ac6db"}]},"branch":"refs/heads/master","description":"Rebase"},"1fe78e2f144e419281227f31d92a28afdf4a08f8":{"kind":"REWORK","_number":4,"created":"2023-10-24 17:34:56.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/897343/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/897343/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/897343/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/897343/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/897343/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/897343/4"}}},"commit":{"parents":[{"commit":"0cccc3dc931257bc3b1a0483f6ee4aeb7a51d354","subject":"proxy: log ChunkReadTimeouts in replicated GET path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0cccc3dc931257bc3b1a0483f6ee4aeb7a51d354"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-10-06 13:37:28.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-10-24 17:23:39.000000000","tz":60},"subject":"proxy: stop logging backend timeouts as client disconnects","message":"proxy: stop logging backend timeouts as client disconnects\n\nWhen a ChunkReadTimeout is raised while reading bytes from the\nbackend, the GetOrHeadHandler._iter_parts_from_response will catch a\nGeneratorExit. The GeneratorExit handler checks the request environ\nfor a swift.non_client_disconnect flag, and if this is not found it\nlogs a \"Client disconnected...\" message.\n\nIn the replicated single-part GET path, swift.non_client_disconnect\nwas not being set when ChunkReadTimeout was raised during backend\nreads. This led to an erroneous \"Client disconnected...\" message.\n\nWorse, the GetOrHeadHandler._iter_parts_from_response was not being\n\"closed\" in a *timely* way, so the GeneratorExit and erroneous log\nmessage might not occur until garbage collection reaped the iterator\nsome time later.\n\nThis patch therefore:\n\n  1. Ensures that swift.non_client_disconnect is set when\n     ChunkReadTimeouts are raised.\n\n  2. Ensures that the _iter_parts_from_response is \"closed\" when the\n     ChunkReadTimeout is raised.\n\nChange-Id: Ie35ff306aa33bfbeb39a4c79c96d57f9d1db7cb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1fe78e2f144e419281227f31d92a28afdf4a08f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1fe78e2f144e419281227f31d92a28afdf4a08f8"}]},"branch":"refs/heads/master"},"aaeef93b6c924ce880c2b5c2d79cfcacc9f71aac":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-10-30 11:03:10.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/897343/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/897343/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/897343/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/897343/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/897343/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/897343/5"}}},"commit":{"parents":[{"commit":"077cc62b82ba97dbaf4f742db6050fd3b08a047c","subject":"proxy: log ChunkReadTimeouts in replicated GET path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/077cc62b82ba97dbaf4f742db6050fd3b08a047c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-10-06 13:37:28.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-10-30 10:58:03.000000000","tz":0},"subject":"proxy: stop logging backend timeouts as client disconnects","message":"proxy: stop logging backend timeouts as client disconnects\n\nWhen a ChunkReadTimeout is raised while reading bytes from the\nbackend, the GetOrHeadHandler._iter_parts_from_response will catch a\nGeneratorExit. The GeneratorExit handler checks the request environ\nfor a swift.non_client_disconnect flag, and if this is not found it\nlogs a \"Client disconnected...\" message.\n\nIn the replicated single-part GET path, swift.non_client_disconnect\nwas not being set when ChunkReadTimeout was raised during backend\nreads. This led to an erroneous \"Client disconnected...\" message.\n\nWorse, the GetOrHeadHandler._iter_parts_from_response was not being\n\"closed\" in a *timely* way, so the GeneratorExit and erroneous log\nmessage might not occur until garbage collection reaped the iterator\nsome time later.\n\nThis patch therefore:\n\n  1. Ensures that swift.non_client_disconnect is set when\n     ChunkReadTimeouts are raised.\n\n  2. Ensures that the _iter_parts_from_response is \"closed\" when the\n     ChunkReadTimeout is raised.\n\nChange-Id: Ie35ff306aa33bfbeb39a4c79c96d57f9d1db7cb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aaeef93b6c924ce880c2b5c2d79cfcacc9f71aac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aaeef93b6c924ce880c2b5c2d79cfcacc9f71aac"}]},"branch":"refs/heads/master"},"0661cf2ea7f8c24d1727d07e478dc0ab9cbe1d2e":{"kind":"REWORK","_number":6,"created":"2023-11-02 18:24:37.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/897343/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/897343/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/897343/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/897343/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/897343/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/897343/6"}}},"commit":{"parents":[{"commit":"5de8c1ba4448f9fc2fee7a5a74863605a24c82c5","subject":"proxy: don\u0027t send multi-part terminator when no parts sent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5de8c1ba4448f9fc2fee7a5a74863605a24c82c5"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-10-06 13:37:28.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-02 18:12:48.000000000","tz":0},"subject":"WIP: proxy: stop logging backend timeouts as client disconnects","message":"WIP: proxy: stop logging backend timeouts as client disconnects\n\nTODO:\n- same fixes in EC path\n- resolve unexpected assertions\n  # TODO: not sure this is correct...\n  self.assertTrue(req.environ.get(\u0027swift.non_client_disconnect\u0027))\n\nWhen a ChunkReadTimeout is raised while reading bytes from the\nbackend, the GetOrHeadHandler._iter_parts_from_response will catch a\nGeneratorExit. The GeneratorExit handler checks the request environ\nfor a swift.non_client_disconnect flag, and if this is not found it\nlogs a \"Client disconnected...\" message.\n\nIn the replicated single-part GET path, swift.non_client_disconnect\nwas not being set when ChunkReadTimeout was raised during backend\nreads. This led to an erroneous \"Client disconnected...\" message.\n\nWorse, the GetOrHeadHandler._iter_parts_from_response was not being\n\"closed\" in a *timely* way, so the GeneratorExit and erroneous log\nmessage might not occur until garbage collection reaped the iterator\nsome time later.\n\nThis patch therefore:\n\n  1. Ensures that swift.non_client_disconnect is set when\n     ChunkReadTimeouts are raised.\n\n  2. Ensures that the _iter_parts_from_response is \"closed\" when the\n     ChunkReadTimeout is raised.\n\nChange-Id: Ie35ff306aa33bfbeb39a4c79c96d57f9d1db7cb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0661cf2ea7f8c24d1727d07e478dc0ab9cbe1d2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0661cf2ea7f8c24d1727d07e478dc0ab9cbe1d2e"}]},"branch":"refs/heads/master"},"b06615af1b16d294d3510da585f08055165c9fd2":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-11-03 14:30:47.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/897343/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/897343/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/897343/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/897343/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/897343/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/897343/7"}}},"commit":{"parents":[{"commit":"3b17b3ebf8a97501472c48963d6d984b667f58e5","subject":"proxy: don\u0027t send multi-part terminator when no parts sent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3b17b3ebf8a97501472c48963d6d984b667f58e5"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-10-06 13:37:28.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-03 14:28:53.000000000","tz":0},"subject":"WIP: proxy: stop logging backend timeouts as client disconnects","message":"WIP: proxy: stop logging backend timeouts as client disconnects\n\nTODO:\n- same fixes in EC path\n- resolve unexpected assertions\n  # TODO: not sure this is correct...\n  self.assertTrue(req.environ.get(\u0027swift.non_client_disconnect\u0027))\n\nWhen a ChunkReadTimeout is raised while reading bytes from the\nbackend, the GetOrHeadHandler._iter_parts_from_response will catch a\nGeneratorExit. The GeneratorExit handler checks the request environ\nfor a swift.non_client_disconnect flag, and if this is not found it\nlogs a \"Client disconnected...\" message.\n\nIn the replicated single-part GET path, swift.non_client_disconnect\nwas not being set when ChunkReadTimeout was raised during backend\nreads. This led to an erroneous \"Client disconnected...\" message.\n\nWorse, the GetOrHeadHandler._iter_parts_from_response was not being\n\"closed\" in a *timely* way, so the GeneratorExit and erroneous log\nmessage might not occur until garbage collection reaped the iterator\nsome time later.\n\nThis patch therefore:\n\n  1. Ensures that swift.non_client_disconnect is set when\n     ChunkReadTimeouts are raised.\n\n  2. Ensures that the _iter_parts_from_response is \"closed\" when the\n     ChunkReadTimeout is raised.\n\nChange-Id: Ie35ff306aa33bfbeb39a4c79c96d57f9d1db7cb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b06615af1b16d294d3510da585f08055165c9fd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b06615af1b16d294d3510da585f08055165c9fd2"}]},"branch":"refs/heads/master"},"35a53513ba64c82efbeb474fe4afdbfd1c1b41e7":{"kind":"REWORK","_number":8,"created":"2023-12-04 16:15:22.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/897343/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/897343/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/897343/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/897343/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/897343/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/897343/8"}}},"commit":{"parents":[{"commit":"66228d8e07122db091dbd141955fdae7b19d163b","subject":"proxy: more consistent ChunkReadTimeout handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/66228d8e07122db091dbd141955fdae7b19d163b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-10-06 13:37:28.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-04 16:08:32.000000000","tz":0},"subject":"WIP: proxy: stop logging backend timeouts as client disconnects","message":"WIP: proxy: stop logging backend timeouts as client disconnects\n\nTODO:\n- same fixes in EC path\n- resolve unexpected assertions\n  # TODO: not sure this is correct...\n  self.assertTrue(req.environ.get(\u0027swift.non_client_disconnect\u0027))\n\nWhen a ChunkReadTimeout is raised while reading bytes from the\nbackend, the GetOrHeadHandler._iter_parts_from_response will catch a\nGeneratorExit. The GeneratorExit handler checks the request environ\nfor a swift.non_client_disconnect flag, and if this is not found it\nlogs a \"Client disconnected...\" message.\n\nIn the replicated single-part GET path, swift.non_client_disconnect\nwas not being set when ChunkReadTimeout was raised during backend\nreads. This led to an erroneous \"Client disconnected...\" message.\n\nWorse, the GetOrHeadHandler._iter_parts_from_response was not being\n\"closed\" in a *timely* way, so the GeneratorExit and erroneous log\nmessage might not occur until garbage collection reaped the iterator\nsome time later.\n\nThis patch therefore:\n\n  1. Ensures that swift.non_client_disconnect is set when\n     ChunkReadTimeouts are raised.\n\n  2. Ensures that the _iter_parts_from_response is \"closed\" when the\n     ChunkReadTimeout is raised.\n\nChange-Id: Ie35ff306aa33bfbeb39a4c79c96d57f9d1db7cb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/35a53513ba64c82efbeb474fe4afdbfd1c1b41e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/35a53513ba64c82efbeb474fe4afdbfd1c1b41e7"}]},"branch":"refs/heads/master"},"a584e064507b495a5215e403f0faefe732d232b4":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2023-12-05 15:02:24.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/897343/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/897343/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/897343/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/897343/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/897343/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/897343/9"}}},"commit":{"parents":[{"commit":"e17186536e10d2b1fd9b042367abf6583658f51a","subject":"proxy: more consistent ChunkReadTimeout handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e17186536e10d2b1fd9b042367abf6583658f51a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-10-06 13:37:28.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-05 14:56:12.000000000","tz":0},"subject":"WIP: proxy: stop logging backend timeouts as client disconnects","message":"WIP: proxy: stop logging backend timeouts as client disconnects\n\nTODO:\n- same fixes in EC path\n- resolve unexpected assertions\n  # TODO: not sure this is correct...\n  self.assertTrue(req.environ.get(\u0027swift.non_client_disconnect\u0027))\n\nWhen a ChunkReadTimeout is raised while reading bytes from the\nbackend, the GetOrHeadHandler._iter_parts_from_response will catch a\nGeneratorExit. The GeneratorExit handler checks the request environ\nfor a swift.non_client_disconnect flag, and if this is not found it\nlogs a \"Client disconnected...\" message.\n\nIn the replicated single-part GET path, swift.non_client_disconnect\nwas not being set when ChunkReadTimeout was raised during backend\nreads. This led to an erroneous \"Client disconnected...\" message.\n\nWorse, the GetOrHeadHandler._iter_parts_from_response was not being\n\"closed\" in a *timely* way, so the GeneratorExit and erroneous log\nmessage might not occur until garbage collection reaped the iterator\nsome time later.\n\nThis patch therefore:\n\n  1. Ensures that swift.non_client_disconnect is set when\n     ChunkReadTimeouts are raised.\n\n  2. Ensures that the _iter_parts_from_response is \"closed\" when the\n     ChunkReadTimeout is raised.\n\nChange-Id: Ie35ff306aa33bfbeb39a4c79c96d57f9d1db7cb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a584e064507b495a5215e403f0faefe732d232b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a584e064507b495a5215e403f0faefe732d232b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
