)]}'
{"id":"openstack%2Fswift~897468","triplet_id":"openstack%2Fswift~master~I255cf44d42736c0a56db6d4b34bb6de74f033ef7","project":"openstack/swift","branch":"master","topic":"slo-part-num","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-10-11 22:45:58.000000000","reason":"Change was abandoned"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-10-11 22:45:58.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I255cf44d42736c0a56db6d4b34bb6de74f033ef7","subject":"WIP: more conditional tests","status":"ABANDONED","created":"2023-10-05 15:30:01.000000000","updated":"2023-10-11 22:45:58.000000000","total_comment_count":9,"unresolved_comment_count":8,"has_review_started":true,"meta_rev_id":"72addab6384e4cc7b03eb38952bc7db0ecaaeb06","_number":897468,"virtual_id_number":897468,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-10-10 01:34:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":-1,"date":"2023-10-10 00:33:43.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-05 16:43:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-10 00:33:43.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"ac5c3e8ab32b9d1129545c8f48e26b12400abc0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-10-05 15:30:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fca15c349a429ba8b0ec806ff352d11d09cbf726","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-10-05 16:43:36.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e4ddde8938fcec58c581f150832c60a2c5731f4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 16:43:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/09d02af4a7654bcbbf4134f33a4b7e1f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10e58ade1a72482e9e5fbe005f9ee568 : SUCCESS in 8m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/b175ab0b7ca2416d921f95293822efa1 : SUCCESS in 48m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/272f19ed0abc4a2892440f5f640ebaa7 : SUCCESS in 36m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/683ead9ee11a4c09998295e56ce542d4 : SUCCESS in 1h 03m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/21ce600417a14a58ba7caafe3d9943a8 : FAILURE in 15m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a239759b4b7340dfb1d8bed89cb1167c : FAILURE in 15m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/65b12c74ac8d46c3bfad1dcf682d774e : FAILURE in 21m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9b16f36e166e4e57948d77b75f6aafea : SUCCESS in 8m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f67913565f544bea9d0f709cc7286a96 : SUCCESS in 6m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4d4e66604c2f4156846e97fba067682d : SUCCESS in 13m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/44053f1e33b844f6bf9afccd5e127c8b : SUCCESS in 12m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e5c1cc65d93d474d8120d2ebb12dcde6 : SUCCESS in 15m 52s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b169dec9818b46d6ad36c71cc988d4c3 : SUCCESS in 14m 19s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ce3bc8461d9940dbb5f757c30207ed15 : SUCCESS in 26m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f4dd851a285e417aa22d5163a48e1efe : FAILURE in 22m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e01cd87bbeae4a2bb29ddcce317b31dd : FAILURE in 21m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c7b0d7b309ce4d579d8bd41d6f3329f2 : FAILURE in 19m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/edc735c21c724294910559a7e5a2fbca : FAILURE in 15m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/932cdf9231f9429faaef31603065811f : FAILURE in 22m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4693170c45a74314a75d42cddb3371f5 : FAILURE in 18m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/94c08097693b41cf961637af8ace7ebf : SUCCESS in 6m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f939af64f6c44f10b787d31faee3c065 : SUCCESS in 17m 24s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/56452e2ae67a4d748a1ee1b996c73719 : SUCCESS in 7m 42s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6368a9a57c2e48a18f359e9471b6f7ab : SUCCESS in 1h 07m 39s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1422643f21a74d0281e9ad8d248c6b82 : SUCCESS in 1h 01m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a6ff3ccf3e7d458abc2f5012ec64ab9d : FAILURE in 27m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4e685b8814b44b5b99396e43513a6cef : FAILURE in 27m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e2c59e8940e4b9aa265961a364ba35e : SUCCESS in 24m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf5ee54f457646349f1ed2f88a34abd2 : SUCCESS in 5m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/56ca7a3bcf5f4289a221ef4487e64885 : SUCCESS in 24m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ced501dd05e34f309e5b617d539b8a3b : SUCCESS in 1h 01m 58s","accounts_in_message":[],"_revision_number":1},{"id":"a5a6377d9c9a944e0a56942c052183481ddf4a58","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 20:00:03.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2afeb8febd07405da25b924ebe515068\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/86732bda6c8f4b928d2cb53414e0553c : SUCCESS in 1h 20m 31s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9988a643742d48a5a8c5ad4371ead6c0 : SUCCESS in 1h 26m 39s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/80c227fa621e45e1828ec6cd2aa105be : FAILURE in 1h 24m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/29627728fda0411cb0575518a50d6fc0 : TIMED_OUT in 1h 32m 16s","accounts_in_message":[],"_revision_number":1},{"id":"12fe760ff7b31d1f683553fbe1a950e0796c2835","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-10-10 00:29:32.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c3122e5d494bf50a2517f06b1ab27ad28b3da3f4","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-10-10 00:33:43.000000000","message":"Patch Set 2: Workflow-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"55f6f617f1e5d78e2bb8fb0e336a36ad95241a2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 01:34:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aad23a37fbbc4bdeac5914c349d12c9f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e81b419621da406eb54baa3721f39b7a : SUCCESS in 8m 08s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9bae16706b4947d0bb80a9595d819465 : SUCCESS in 43m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/67bf7cc50fa74766af310fd509ca5406 : SUCCESS in 19m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a8939cf024ef4dfbb6309758f849e4cd : SUCCESS in 18m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d273bfc445fd43dca2a8c5a2ef013123 : SUCCESS in 21m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/67db5ddfb46548b3a88a187834ceba34 : SUCCESS in 9m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bc34b16b9df54e9098a5eb905d3f3dcb : SUCCESS in 5m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/83feb929bdeb4314b32da2a588c35785 : FAILURE in 16m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5eedcd8a7b1c4b1bbe79943c570edd6e : FAILURE in 16m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7411c8f55935447c914408f902cfc00b : FAILURE in 17m 30s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7cd4f2656d49478092d0a084548ac7c7 : FAILURE in 16m 00s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e062efe10d984276bdb2d0d50fbee264 : FAILURE in 15m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3074f9c1c15744fe8b056e0b282a4850 : SUCCESS in 18m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/37d75cbbdd334f15b2d5e37d7fe393f6 : SUCCESS in 18m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a4f88e00a86348178e2461e329ee7739 : SUCCESS in 19m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1dc7a2568af64334a818c1f1eee9b4eb : SUCCESS in 15m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/38c361d2e9874d9cbe5f064554e0c498 : SUCCESS in 16m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6ba0d918dddd47b3aac5ef44daa90242 : SUCCESS in 16m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/390fcb466e6a4312b405ff1619eef556 : SUCCESS in 57m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/be0b5202582e4b61a3d6e29d36a1c2e5 : SUCCESS in 36m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6313300390e47b1be3ac9d6f53d3b7e : FAILURE in 14m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18ad2125140447c68fe191a95277071d : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4da5f3c047c24c6dbfd1e88d992cc1d2 : SUCCESS in 17m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bd9575b9de206c0f77b5bfbde484fd77a8dc8b75","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 02:03:15.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/075e62600a0943d8acc4d68b99383640\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d5ce055571d942548420736907e13a51 : TIMED_OUT in 1h 31m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/66401f81910a4538a92db40b8c4570c8 : FAILURE in 53m 14s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/006781923c0444c0aaa171f072076c0d : SUCCESS in 1h 00m 48s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d9cdd479e7664891b1da6b04beca351b : SUCCESS in 58m 31s","accounts_in_message":[],"_revision_number":2},{"id":"72addab6384e4cc7b03eb38952bc7db0ecaaeb06","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-10-11 22:45:58.000000000","message":"Abandoned\n\nsquashed https://review.opendev.org/c/openstack/swift/+/896466","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6cd041df0ce0e27589573f3b8828336b233d1882","revisions":{"725bac53d09d886044f10a3e5fa8e8102a9c58a4":{"kind":"REWORK","_number":1,"created":"2023-10-05 15:30:01.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/68/897468/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/897468/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/897468/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/897468/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/897468/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/897468/1"}}},"commit":{"parents":[{"commit":"827a727abc30d969721681ce8458529a19d5e34c","subject":"slo: refactor GET/HEAD response handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/827a727abc30d969721681ce8458529a19d5e34c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-10-05 15:25:26.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-10-05 15:25:26.000000000","tz":60},"subject":"sq: fix conditional non-slo after refetch","message":"sq: fix conditional non-slo after refetch\n\nA refetch is deliberately non-conditional, but may return a\nnon-slo. We need to make sure the client request\u0027s conditions are\nenforced.\n\nChange-Id: I255cf44d42736c0a56db6d4b34bb6de74f033ef7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/725bac53d09d886044f10a3e5fa8e8102a9c58a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/725bac53d09d886044f10a3e5fa8e8102a9c58a4"}]},"branch":"refs/heads/master"},"6cd041df0ce0e27589573f3b8828336b233d1882":{"kind":"REWORK","_number":2,"created":"2023-10-10 00:29:32.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/68/897468/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/897468/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/897468/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/897468/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/897468/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/897468/2"}}},"commit":{"parents":[{"commit":"80d0395514a5e45981483c84bfbdf5b40e5a58be","subject":"Refactor SLO size/etag sysmeta tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/80d0395514a5e45981483c84bfbdf5b40e5a58be"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-10-09 23:42:22.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-10-10 00:27:48.000000000","tz":-300},"subject":"WIP: more conditional tests","message":"WIP: more conditional tests\n\nChange-Id: I255cf44d42736c0a56db6d4b34bb6de74f033ef7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6cd041df0ce0e27589573f3b8828336b233d1882"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6cd041df0ce0e27589573f3b8828336b233d1882"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
