)]}'
{"id":"openstack%2Fswift~898244","triplet_id":"openstack%2Fswift~master~I76f641c6664eb79e58d6fbe3ef0e94edafbed599","project":"openstack/swift","branch":"master","topic":"container_get_namespace","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2024-01-10 15:51:58.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-01-04 12:14:10.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"I76f641c6664eb79e58d6fbe3ef0e94edafbed599","subject":"Use Namespaces when that\u0027s all we need","status":"NEW","created":"2023-10-13 17:04:42.000000000","updated":"2024-01-10 16:26:02.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8a534a963f1dfb6f8ba842a37ab897917777dffa","_number":898244,"virtual_id_number":898244,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-01-10 16:26:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-2,"date":"2024-01-10 14:52:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-13 18:19:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"13eb3532f04616ce5f938c543c644e720915177c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-10-13 17:04:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"270e0f53a0f7758199f6d4acc2160e8fc5d9a499","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 18:19:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0ee20eae3014889b9c90e5cbd207695\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/769ba435795940528db9d19aae8e19ef : SUCCESS in 8m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/c107a045ad8545cbb0d791f4df481dcc : SUCCESS in 1h 05m 06s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ddb763d1ff3844469fe452473e6c6ae3 : SUCCESS in 54m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/636bdc44a4e8499dbd8bfc2101149ed1 : SUCCESS in 43m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18c11b4bb6904610a36b40d77e4e6964 : SUCCESS in 1h 08m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6d2147d7c08848e7b2852e8916319ba6 : SUCCESS in 24m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5782e9cc178e42cbb8721a4d6ac47ba5 : SUCCESS in 25m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/953ab0382cc246f8b6d1c46b3c264d79 : SUCCESS in 20m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4da847ecb1f2452489f667fc3f344202 : SUCCESS in 8m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c36089e5948e4f819f243668376ffc6b : SUCCESS in 5m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9612a9db0965464aace9cdb1f55522f6 : SUCCESS in 15m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9a73622c74924261955ee0cef8f8cac3 : SUCCESS in 12m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/82ed3de0df9a44fc8b554ea535a19f44 : SUCCESS in 23m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/cf329690b9d34ca1808e702dcda1a7a7 : SUCCESS in 16m 39s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/601c45486d954302bf8b475af463042c : SUCCESS in 17m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e73b0f90c6fe46e8b65a1313209a9617 : SUCCESS in 18m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/88ed3871cf5643cb90846eceebbfba7f : SUCCESS in 14m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/502f00e884c7447885df6a23b7b61684 : SUCCESS in 19m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b5228708b47040af8fe8d81c471dfc84 : SUCCESS in 23m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0470028789f84cdebcd70d5158fad80c : SUCCESS in 16m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1b11bc4f3d624f4a9cbd9f44fae16f70 : SUCCESS in 16m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/982b8b46b41c44659e791f8ff689b32d : SUCCESS in 6m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b149535593aa4c5fadfbb7450cd956b5 : FAILURE in 21m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/687a8328da294c06ae5c99fbb4bdea8c : SUCCESS in 7m 18s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/88afc86a89fb4aeeb36a1f62dcee8e3e : SUCCESS in 1h 04m 39s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/802486ca540b4d5c89db734b92617955 : SUCCESS in 1h 01m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9f84544dcec64ce8a55c5d28f685b4ab : SUCCESS in 55m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8014ef398be042378d2fca484d02527c : SUCCESS in 47m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3f229ea4c754829b07cc06e5817aa9e : SUCCESS in 19m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a93b49ce5f2a47c5a031970a372cffd2 : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2c0436a923394432b344b0b793c67bbc : SUCCESS in 19m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af502001168a4875a50919aea78c7800 : SUCCESS in 1h 07m 09s","accounts_in_message":[],"_revision_number":1},{"id":"4f236a6ba6b44f93f60ece6da3fb81a0b34506b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 18:49:03.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e17b24bc0cf49a7b5c0d73699404a12\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2be5024a362e4f6584f4ce522f2c9d9c : SUCCESS in 41m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dd0565f5ee71403aae15d85cb8364650 : SUCCESS in 50m 53s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1681e7900b784a92a660261144e23208 : TIMED_OUT in 1h 31m 07s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/70ab66f279304dc59556b5fbcc9298c0 : SUCCESS in 42m 31s","accounts_in_message":[],"_revision_number":1},{"id":"72648053827c8b39a86950a6f76909684ccaba13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-10-16 00:32:26.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2ccdca68d38fbdad97562f2d549500a7f70f7820","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 01:50:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7b5482016e54a87a1fbf1cdb3750eb6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/42bba7a8ffe24a79b096513c89de0a4a : SUCCESS in 1h 02m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6cf05fe9022643fbb4a9d867ab6759cd : SUCCESS in 1h 15m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4012fd4b67aa4cf9a63ee9e7822dc4d8 : SUCCESS in 1h 11m 58s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b34c5eb3d0fb4ce2b41951fd6d4c7407 : SUCCESS in 47m 18s","accounts_in_message":[],"_revision_number":2},{"id":"837efc9c239e88dcbd59369ed305dec1a2276f7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 02:06:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cafa3541664d4317b7430d8ce0f2d6ba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6e3f687369f44578366f904221b8cb0 : SUCCESS in 8m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/14b6b1ce0bdd4a159e118c1d757cf277 : SUCCESS in 48m 10s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2f8ebc14c1334147a96fdd341d4a3330 : SUCCESS in 45m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/36b5a65cda794aa8984629eba662dfbf : SUCCESS in 57m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/535fa4d99ec844ac901c0d2646eb0776 : SUCCESS in 55m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d4243d5198574dbf87c73abf45f4c886 : SUCCESS in 18m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5131c864fcf54dc7bf705a52d7d106c8 : SUCCESS in 18m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fde68f9439d04f698c83dbba84bdd1d2 : SUCCESS in 18m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/db2f3b49ec9d426a923d3bda5486f376 : SUCCESS in 10m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/018ea2d883de4d95b6fab25a8470027c : SUCCESS in 8m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f9feb7dcbf6143f4965fe44bda41f63f : SUCCESS in 17m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4d569b5eefb4442191805b9c0f1f4320 : SUCCESS in 18m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/13e6bb6b4290439b97fb3f278d197307 : SUCCESS in 17m 16s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/efa55dc1336849febf9bef4ccf14bea7 : SUCCESS in 17m 02s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2fb9d201168a4e848a5ee28819c2c667 : SUCCESS in 20m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fde6a6c6d9fd4b4996c00326a507afeb : SUCCESS in 17m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1ff029268278490d9a9114796835851e : SUCCESS in 19m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2cd73bbff80a45f089d95e2207cce352 : SUCCESS in 19m 17s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/89f808e9307247fdbf29e6a84e2dd0b4 : SUCCESS in 15m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/53161cdd1ce04de69c61f6e42bc92672 : SUCCESS in 15m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d082a0ed3fc74755a23d8db0ff3f5ab4 : SUCCESS in 17m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8999fba25d0b46409e9f2ad1a8b5c50b : SUCCESS in 6m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a3a99d598ea141abb1be3eae78af8b45 : FAILURE in 15m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/13004a7a60cd44bba2edaae787b60903 : SUCCESS in 7m 09s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7d042e683cc549418f37ba3e30fe28e4 : FAILURE in 56m 41s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d1e047a716d64e51926254c205d9c5d0 : SUCCESS in 59m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a66e8b10763949559c8f23884b97aa9a : SUCCESS in 53m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8521ea9ced2a4d8ca8ce27b9df02e71b : SUCCESS in 39m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be90dee3bc6a4fda971c32f421e01e95 : SUCCESS in 14m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43be5929bca44b6a8020e8b241bf0ec3 : SUCCESS in 5m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/395b4b31e7e343968aed450d8dc5ad4a : SUCCESS in 20m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9eda6823096f4866a591760366b73164 : SUCCESS in 46m 37s","accounts_in_message":[],"_revision_number":2},{"id":"dac36f48baae7c122646eadd681ad307a10d16bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-12-20 17:29:36.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_7233\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}],"_revision_number":3},{"id":"400897ff78cc3e334355da69d7a3139c5dfc62e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 18:44:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a2adae3c20b440787e8033335db4f0a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c830907f5fca4d23a084a21dd0563681 : SUCCESS in 9m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/3d1369ef1e8b4c749057317838972a97 : SUCCESS in 1h 06m 52s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/fcfe390207714e60946d3aae902872d6 : SUCCESS in 1h 01m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c9cbe09a75f24c75a81d2bebb69121e1 : SUCCESS in 54m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c244ffbf50ff4bcb99c1a970bc707b17 : SUCCESS in 1h 06m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0db9ca4d45a44193b4403147b83fe2a6 : SUCCESS in 19m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0b4e42c8957e4bb49835c1559b9cb6c3 : SUCCESS in 16m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2602830c75ad4ac9a8264e3d09b28348 : SUCCESS in 21m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/75597289dfa54e5b8e41bfd351b91428 : SUCCESS in 9m 13s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/92fd67103d414d8f98a880f125584d3f : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d128afb1263649209732be09602f12c0 : SUCCESS in 20m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1be576689d374482a0f31b0ed0947303 : SUCCESS in 16m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e313c729587e4d59b039dcf0e430a2bc : SUCCESS in 13m 44s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c01b1d3999fd4ee482888ed0b3786ff2 : SUCCESS in 16m 57s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/97f0e4151db448aaa494467d50e0bbd2 : SUCCESS in 14m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3194143ae7b446d99b87548c74ee4ba5 : SUCCESS in 17m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d7e97a531c5742b58e4d4726d1008a71 : SUCCESS in 19m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9e6035afdf2a46a78ec60e649ba519f2 : SUCCESS in 20m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7f477012f0f84d59b5e10208972e73dd : SUCCESS in 15m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5e7bcd347dcb4956814b832329fcee67 : SUCCESS in 18m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/00b7c119e8864e77a08d864f01b927cf : SUCCESS in 15m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/eda9d5fc0cf3489e8a917bda5922a9ea : SUCCESS in 7m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/671c20f4946e424386ffc1f7d1f1d2d2 : SUCCESS in 18m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6cec0e056b3843faa36e70b7dc703a15 : SUCCESS in 8m 20s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/03f140d5809c4c23953bad86b93bfac8 : SUCCESS in 50m 23s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/dba7453800444bf3b9e25a314ece2499 : SUCCESS in 52m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0757d8b5f8f945b0bf4753749f32b9e0 : SUCCESS in 56m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/afe55ff8f23b403cb33f3dbf0ba19232 : SUCCESS in 56m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24c809e4387d47eea53a7fff0727a758 : SUCCESS in 20m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1413858af5d449b58fdf85a79bde4407 : SUCCESS in 5m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e7fc344e88284ce793014452208c15f9 : SUCCESS in 18m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2af2d915118c4121815e220a705017b5 : SUCCESS in 1h 02m 55s","accounts_in_message":[],"_revision_number":3},{"id":"33fbc1a4475112e9b17d5453ee8d082d9c14d076","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 18:47:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b5b0456a4654a81ab3df2d15b9e0b76\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7cfa0cbf77414dbf9a005c1eba9bb5ca : SUCCESS in 57m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bd2bf68a915f43abb183b148cd13b1e0 : SUCCESS in 53m 30s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/00a7eda8c764468f87367a8a7a847030 : SUCCESS in 24m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e22a02e3df3049b79b230a2036c3006a : SUCCESS in 23m 56s","accounts_in_message":[],"_revision_number":3},{"id":"aac194463f0d7716d949e1fe22f17267bd97c1ff","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-04 12:14:10.000000000","message":"Patch Set 3: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"07eb7654206f8f971d74b96d502f103442db4266","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-10 14:52:39.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"30b4e8dd28a1f0ca02ead954d51c35e35fc768e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 15:51:58.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e081d7e410244a0f89ce20225ebd01d2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aba07bafaa6d41778e282355fce8dffe : SUCCESS in 16m 24s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d32dafdc8b8b45e3b6dead0de47a0249 : SUCCESS in 17m 19s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2eeb6902f45e4c81ac05d3253c42f4cd : SUCCESS in 57m 27s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f25f1bf17ba341eeb54447b1a8d6250b : SUCCESS in 19m 45s","accounts_in_message":[],"_revision_number":4},{"id":"8a534a963f1dfb6f8ba842a37ab897917777dffa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 16:26:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c20516f74a434647be87f3cc409e5f81\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39a3eee9c872482b94ff358699166f19 : SUCCESS in 9m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/283dd746c8ea4c36867a915475a78cbe : SUCCESS in 1h 01m 09s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4592f135eb4642b990a38764f1445f00 : SUCCESS in 1h 01m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c2037ed8cbed45a29bcf1eabe1c29ffe : SUCCESS in 48m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d23624430894a979127658c5449a8d7 : SUCCESS in 36m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c8999228604b457f8f61b81bd1ef6500 : SUCCESS in 19m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dfb1649161824f2682322d3b2a06159b : SUCCESS in 20m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7bff713aa7e74ca98383ec9ce6e4ea08 : SUCCESS in 20m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6e4b352421b54c2cb610ee73033b9156 : SUCCESS in 10m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c0422eeb31d341bb8d520613d478f6ef : SUCCESS in 6m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6943786424e64b36a6d62eb57aab7f8b : SUCCESS in 19m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ae52a6a159a74c6f9d99b85aa01515dd : SUCCESS in 17m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/74b36fe9e8d14c9dbec5a5d088ef83d9 : SUCCESS in 17m 18s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4f56ffdbe4a9463386b8ef430ed97435 : SUCCESS in 16m 40s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/117b735efa3a482997604050e1d47fa8 : SUCCESS in 15m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e06d921e1c6e4f29a2fd895362829424 : SUCCESS in 19m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/03f9e05903334025a1b75be3facc221e : SUCCESS in 18m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ad84a099d1f347a8a9928cb128cdd0bd : SUCCESS in 22m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7e02689ca4a542208cf9bcee046f7e38 : SUCCESS in 15m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d858d07532f3416a9852aff8d3601e41 : SUCCESS in 14m 56s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4c084aafbb72419b940a0a0776a13c9d : SUCCESS in 16m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/51bcd9b49c8241729b72824319c69e8d : SUCCESS in 7m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b208327053454af6ac31b4d22ca48768 : SUCCESS in 18m 04s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/28109113cf4f45c1ac7fae97eca3a975 : SUCCESS in 8m 47s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/812c79e470724a2d9023b72744b852f8 : SUCCESS in 1h 01m 15s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0a5645913ab5442880a1b29642d7d1cd : SUCCESS in 1h 06m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/15e6c48e982648429cc9d1874254f095 : SUCCESS in 55m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cc72fdaa9f294b96993723eb72199236 : SUCCESS in 40m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3abbe121f0ac458ebde369f6d8a7cd0a : SUCCESS in 14m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4aee3bb13b5b4763a30350eabe052fdd : SUCCESS in 4m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e960de2e6aed4bce828023e8b1eda771 : SUCCESS in 23m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc209f007213463ab3db468fbcb47f3a : SUCCESS in 43m 24s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"74aca80e7a5578964b6aac834cb97016b1894624","revisions":{"ade8a8cf58567c0a4b693fbc9010b0634dd63720":{"kind":"REWORK","_number":1,"created":"2023-10-13 17:04:42.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/44/898244/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/44/898244/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/898244/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/898244/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/898244/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/44/898244/1"}}},"commit":{"parents":[{"commit":"5e0369cfd37630f116e3ea15c0aa7282f764e8d3","subject":"sq: fix ShardRange name/account/container setters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e0369cfd37630f116e3ea15c0aa7282f764e8d3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-10-13 16:58:50.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-10-13 16:58:50.000000000","tz":60},"subject":"Use Namespaces when that\u0027s all we need","message":"Use Namespaces when that\u0027s all we need\n\nThere\u0027s some places that we historically instantiated a ShardRange\nmerely to represent a range of the namespace tht isn\u0027t actually a\nshard. All we need of the ShardRange is an encapsulation of a lower\nand upper bound. Since the Namespace class now gives us that simpler\nencapsulation, use it where we can.\n\nChange-Id: I76f641c6664eb79e58d6fbe3ef0e94edafbed599\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ade8a8cf58567c0a4b693fbc9010b0634dd63720"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ade8a8cf58567c0a4b693fbc9010b0634dd63720"}]},"branch":"refs/heads/master"},"b657c9816f2656b3a76316ad2e00504ab5735adb":{"kind":"NO_CHANGE","_number":2,"created":"2023-10-16 00:32:26.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/44/898244/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/44/898244/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/898244/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/898244/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/898244/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/44/898244/2"}}},"commit":{"parents":[{"commit":"efb974f65df803d4fde17c2c4587fc3da562903b","subject":"Proxy: Use shard-format for auto cont GETs and updating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/efb974f65df803d4fde17c2c4587fc3da562903b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-10-13 16:58:50.000000000","tz":60},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-10-16 00:31:54.000000000","tz":660},"subject":"Use Namespaces when that\u0027s all we need","message":"Use Namespaces when that\u0027s all we need\n\nThere\u0027s some places that we historically instantiated a ShardRange\nmerely to represent a range of the namespace tht isn\u0027t actually a\nshard. All we need of the ShardRange is an encapsulation of a lower\nand upper bound. Since the Namespace class now gives us that simpler\nencapsulation, use it where we can.\n\nChange-Id: I76f641c6664eb79e58d6fbe3ef0e94edafbed599\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b657c9816f2656b3a76316ad2e00504ab5735adb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b657c9816f2656b3a76316ad2e00504ab5735adb"}]},"branch":"refs/heads/master"},"8a1829a17660de2615b952f76fcca527ccfe3c5d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-12-20 17:29:36.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/44/898244/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/44/898244/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/898244/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/898244/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/898244/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/44/898244/3"}}},"commit":{"parents":[{"commit":"0f48c895d4afb63ef309cf30ea4655ff1672502a","subject":"Proxy: Use shard-format for auto cont GETs and updating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f48c895d4afb63ef309cf30ea4655ff1672502a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-10-13 16:58:50.000000000","tz":60},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-12-20 17:29:36.000000000","tz":0},"subject":"Use Namespaces when that\u0027s all we need","message":"Use Namespaces when that\u0027s all we need\n\nThere\u0027s some places that we historically instantiated a ShardRange\nmerely to represent a range of the namespace tht isn\u0027t actually a\nshard. All we need of the ShardRange is an encapsulation of a lower\nand upper bound. Since the Namespace class now gives us that simpler\nencapsulation, use it where we can.\n\nChange-Id: I76f641c6664eb79e58d6fbe3ef0e94edafbed599\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8a1829a17660de2615b952f76fcca527ccfe3c5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8a1829a17660de2615b952f76fcca527ccfe3c5d"}]},"branch":"refs/heads/master","description":"Rebase"},"74aca80e7a5578964b6aac834cb97016b1894624":{"kind":"REWORK","_number":4,"created":"2024-01-10 14:52:39.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/44/898244/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/44/898244/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/898244/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/898244/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/898244/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/44/898244/4"}}},"commit":{"parents":[{"commit":"7d5c73fcdec06f982c85c5084191bc62218f0a3b","subject":"Merge \"ContainerBroker.get_shard_ranges(): states must be a list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d5c73fcdec06f982c85c5084191bc62218f0a3b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-10-13 16:58:50.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-01-10 14:52:24.000000000","tz":0},"subject":"Use Namespaces when that\u0027s all we need","message":"Use Namespaces when that\u0027s all we need\n\nThere\u0027s some places that we historically instantiated a ShardRange\nmerely to represent a range of the namespace tht isn\u0027t actually a\nshard. All we need of the ShardRange is an encapsulation of a lower\nand upper bound. Since the Namespace class now gives us that simpler\nencapsulation, use it where we can.\n\nChange-Id: I76f641c6664eb79e58d6fbe3ef0e94edafbed599\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/74aca80e7a5578964b6aac834cb97016b1894624"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/74aca80e7a5578964b6aac834cb97016b1894624"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
